)]}'
{"id":"openstack%2Foctavia~880801","triplet_id":"openstack%2Foctavia~master~I4dd4e7dabe71d0a81eaba9bdad6969911a395429","project":"openstack/octavia","branch":"master","attention_set":{},"removed_from_attention_set":{"29244":{"account":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"last_update":"2023-04-20 18:03:07.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I4dd4e7dabe71d0a81eaba9bdad6969911a395429","subject":"Fix pep8 error","status":"MERGED","created":"2023-04-19 08:57:30.000000000","updated":"2023-04-20 18:04:21.000000000","submitted":"2023-04-20 18:03:07.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":4,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"880801","meta_rev_id":"6c3cc2c60ca4f50afd7422912cdb8df24f15bc46","_number":880801,"virtual_id_number":880801,"owner":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-04-20 18:03:07.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"all":[{"value":2,"date":"2023-04-19 11:39:31.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2023-04-19 08:58:15.000000000","_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"all":[{"value":1,"date":"2023-04-19 11:39:31.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-04-19 08:58:15.000000000","updated_by":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"reviewer":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"state":"REVIEWER"},{"updated":"2023-04-19 11:07:02.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"e045b4b85ac3a17f6404cb1a30b8af077079d78f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2023-04-19 08:57:30.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"40b92d3b7d2a15d3b5c09d38f61a13751224683c","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2023-04-19 08:58:15.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"e5af26dca40670245125e8684cfd127a7183a1a4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-19 11:07:02.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1e29ac475fbd43459d23985fbd793d3e\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/27a5a3f545114b37874f8d0160366947 : SUCCESS in 10m 27s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3b32449be6534ed19d8a2833b0f8cf12 : SUCCESS in 6m 01s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/46dd4c4c97a34a2b847ed4b110db7a00 : SUCCESS in 4m 27s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/846066e7fe5f4d4e9bb0ead2674f662f : SUCCESS in 4m 20s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7efc8ad25ca54831ad9a85674d2fe5ed : SUCCESS in 9m 28s\n- octavia-tox-py39-tips https://zuul.opendev.org/t/openstack/build/dd5724bf83b847ad862e9285f092a1dc : SUCCESS in 5m 07s\n- octavia-tox-functional-py39-tips https://zuul.opendev.org/t/openstack/build/c06c11ce80224f43b4ad5cd6c038492e : SUCCESS in 7m 19s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/c664657720cf41b6b2b804b376658221 : SUCCESS in 4m 20s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/c595e7e30b9e44b8b0611ad4236ff709 : SUCCESS in 8m 12s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/af0930130dea43df830620ce503b1ba4 : SUCCESS in 7m 22s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/382649be5ce4461b89bf30805d95c229 : SUCCESS in 6m 55s (non-voting)\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/fe6e339399824cd7af4fb1ac432c167b : SUCCESS in 1h 16m 48s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/2e18c990e10244ed979f2669e4c5c4a8 : SUCCESS in 1h 44m 34s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/9c4f0e5ca4fe4e2fa543ba08a62a1de6 : SUCCESS in 35m 53s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/b99638999e5a4c2a958f9d18c58e90c0 : SUCCESS in 46m 41s\n- octavia-v2-dsvm-tls-barbican-fips https://zuul.opendev.org/t/openstack/build/f71f4d9c8dbb40b28f3528f56122d954 : SUCCESS in 1h 18m 06s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/b2fd397240514a0fa3933f5b08b7883a : SUCCESS in 34m 05s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/d8c0ad1c25db4c8d9d12af9c7ff82ce5 : SUCCESS in 2h 07m 54s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/f1cc73dee4604546a47d7ae85621247f : FAILURE in 41m 57s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/29c928f775834c499e27dffbb58f636f : SUCCESS in 1h 36m 15s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/bd23a49581d345bdaf2bf3b7040de7f8 : SUCCESS in 1h 55m 16s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/4528edf2b90b41c19f68f4712ba66d27 : SUCCESS in 1h 42m 56s (non-voting)\n- octavia-v2-dsvm-scenario-centos-9-stream https://zuul.opendev.org/t/openstack/build/80150fed5ca94bf1bd3d67ca48b6b7f3 : NODE_FAILURE Node request 300-0021008588 failed in 0s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"a990eb97b79f8e0665bde02c369bc0af96601a25","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2023-04-19 11:39:31.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"8a318b97a61227cc83981c688d5e5dde90d912ef","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-19 11:40:05.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"d2e2a51a04e6b9da5af70fdc9c228a55448a8bf4","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-19 14:14:54.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/cc864103b99b4b30aa004d9387d42f14\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6586164affa9487c943303cbb304ae0d : SUCCESS in 4m 59s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/e3789bfc57e94055954c183399215ee2 : SUCCESS in 4m 24s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/92c05907d6c64168a1da67f8ad01129a : SUCCESS in 4m 44s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5cb964eac8034ae08d4fa9333faf6c31 : SUCCESS in 10m 21s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/e44309d1ce2246b183f85cff7cfa76ab : SUCCESS in 4m 37s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/f683220ab83e49339e022f5150625051 : SUCCESS in 8m 46s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/1c37eb8fd3ed42e993bc82416b82dfef : SUCCESS in 1h 26m 05s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/334be153dbac4eb89c50c8dafd216cbf : TIMED_OUT in 2h 34m 41s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/930ef07c65994cd9b29c076bcf7ac350 : SUCCESS in 47m 28s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/76003604274a423096aef8877919db63 : SUCCESS in 44m 56s","accounts_in_message":[],"_revision_number":1},{"id":"911bcd2fda1505ae7bd15040d73cf2938eb9e101","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2023-04-19 14:23:30.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"f3feea515e372eff71d9212ef6e672af86e59431","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-19 16:51:15.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/355411edb9e247eab4906fe09ed1d663\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/941b4191adad45809e36e919fb619ab2 : SUCCESS in 10m 39s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a8e2fcb2122f4df0a940ce82ef3f6f79 : SUCCESS in 5m 14s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/db78da90932f44b485c93b0308921339 : SUCCESS in 4m 55s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/0c12ac5bc0c24005b1bbe2da5af1f0de : SUCCESS in 4m 57s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fe0cee285d944d42b559e73df38dc84c : SUCCESS in 11m 46s\n- octavia-tox-py39-tips https://zuul.opendev.org/t/openstack/build/b270dcec90f34027ad3d76fc6aed4413 : SUCCESS in 5m 04s\n- octavia-tox-functional-py39-tips https://zuul.opendev.org/t/openstack/build/ba94f1081963458ca781d0d95defe16c : SUCCESS in 7m 47s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/5cb5cdf0d7b64b2fb1b070576b7ef47c : SUCCESS in 5m 37s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/59cc00e0ed8b4e59b7e8d83bf43dc5dd : SUCCESS in 9m 03s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/4353aa4179c74e819fba392b7a7734e0 : SUCCESS in 8m 21s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/35ef7a99702546329efb960b4e689b6e : SUCCESS in 8m 38s (non-voting)\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/65b02f882d32488ca88db424ef14d2a2 : SUCCESS in 1h 33m 09s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/920f145df15d4a09bba93bb2a7ee4879 : FAILURE in 42m 36s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/d0aa75cf1acd449caf6ae43b0a23aa96 : FAILURE in 53m 09s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/53cf685c35564771b0b08ab738050bd9 : FAILURE in 5m 34s\n- octavia-v2-dsvm-tls-barbican-fips https://zuul.opendev.org/t/openstack/build/10afe62e43444d92936e67833ca08fad : SUCCESS in 1h 02m 09s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/2e00e500bb6949a99f9c787a84fce65e : FAILURE in 6m 08s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/881a3b5672f440c19ac5a8b28f0ab491 : FAILURE in 8m 29s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/a45db5b13ed44f5c96e2eef6cd750255 : FAILURE in 12m 14s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/50858375fba749c7867eb7632b3292af : SUCCESS in 2h 25m 25s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/021514c6540845a48215ce8594bb6ddd : SUCCESS in 2h 18m 31s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/db863009a5674189a86909377ef3e096 : FAILURE in 7m 51s (non-voting)\n- octavia-v2-dsvm-scenario-centos-9-stream https://zuul.opendev.org/t/openstack/build/1681d80c02f34e558c6412bd9ce538da : SUCCESS in 2h 13m 07s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"963659cd612654a149289f1a4b73188d5dcbfe50","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2023-04-20 08:23:36.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"da767b6a9272569675b09cfb875d34cba240574d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-20 11:00:15.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/e96611410f1445bcb5fbb1ae01224bbc\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/0d063d14b38842e59e41cf7a940bf34a : SUCCESS in 12m 38s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/631766c61b734b66b3d8626b54a8f411 : SUCCESS in 5m 10s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/2b99424af1ac460a9267b71a11e84ef4 : SUCCESS in 4m 46s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/79b43998e78c481482e8d1f6fea37611 : SUCCESS in 5m 00s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b37efc110ecc4fd18ac58ba4750fbce9 : SUCCESS in 12m 37s\n- octavia-tox-py39-tips https://zuul.opendev.org/t/openstack/build/1443d68b7db446199395cbea8e548f6c : SUCCESS in 4m 57s\n- octavia-tox-functional-py39-tips https://zuul.opendev.org/t/openstack/build/48c6a6cfef9b42ce88f43e6ebe35422d : SUCCESS in 7m 19s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/b8106c3ce1644471a712836877979ed9 : SUCCESS in 4m 40s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/fd9503ae0cf64a2a81da3ea869c99995 : SUCCESS in 7m 29s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/8c037bdeaf624926b69d09a57baa211b : SUCCESS in 6m 40s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/61a2f8d7f4cd4473bb0ddb96dd189ca7 : SUCCESS in 7m 29s (non-voting)\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/93ed136832394e6e9ad4ea354e8b1af0 : SUCCESS in 1h 43m 58s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/6caab1e99aac4034ac8509785a678ad9 : SUCCESS in 1h 56m 32s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/ecdd119ed1b149228bcb684617e86781 : FAILURE in 6m 39s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/fa272d3e96a64833ac520a3e8c081a8f : FAILURE in 6m 53s\n- octavia-v2-dsvm-tls-barbican-fips https://zuul.opendev.org/t/openstack/build/17da065bf84045108e8b172f43adf809 : SUCCESS in 1h 09m 36s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/a69f8246247d41ebb95866ae2a6ee3ec : FAILURE in 7m 38s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/d546e027d52b4e7e939c5cb3e6373065 : SUCCESS in 1h 33m 25s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/cadf972c63dd48389349e2024c8488c1 : SUCCESS in 2h 15m 42s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/ad9214cf0bdf43c8bb72484b3e600470 : FAILURE in 5m 01s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/0cb5fb3d110e4508a0ca7ea9652dc102 : FAILURE in 5m 37s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/b9d91085751e42c592d5281826047390 : FAILURE in 6m 32s (non-voting)\n- octavia-v2-dsvm-scenario-centos-9-stream https://zuul.opendev.org/t/openstack/build/b74a2abb13b546dab45dece38690ecd8 : TIMED_OUT in 2h 34m 39s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"02729bf1a2043295937a572998ebdf58e33d7632","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2023-04-20 13:41:01.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"8844fc20c88e0cd3ff56ef3ae02a3fe1077ef97d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-20 16:19:03.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c8b906630adb425d9e3630baa5e4ed09\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/90bf6838895f4ecb9bf1c84d11b45b24 : SUCCESS in 10m 29s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/cca654c205ca42918e8a09368cbbd10d : SUCCESS in 5m 44s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/2b953bf51dea4dc692052b6bf8667029 : SUCCESS in 4m 53s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/3184f73e90194c4298260e210d093c7c : SUCCESS in 4m 51s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/daa2fdecfce24a75ad8e6dbd61a49bb4 : SUCCESS in 12m 55s\n- octavia-tox-py39-tips https://zuul.opendev.org/t/openstack/build/f4e23bae96c742bfbf8cb95767cd2355 : SUCCESS in 4m 15s\n- octavia-tox-functional-py39-tips https://zuul.opendev.org/t/openstack/build/7c3b3a489be0433cb74850edbb98589d : SUCCESS in 7m 50s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/625eb43244604015b23e50bc879a85cd : SUCCESS in 5m 22s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/603d6bbd66bc4e26bea1606f0fa2547e : SUCCESS in 6m 51s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/2555ecc58c9347a18cb3e3e4686aa767 : SUCCESS in 6m 52s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/daf0d58e15b24b3da418990b378b31d9 : SUCCESS in 8m 54s (non-voting)\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/ad20861e027749ddac3eda4dcab42914 : SUCCESS in 1h 29m 23s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/c8a999f180b8498791e738284fb638ac : SUCCESS in 2h 26m 04s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/0077b83c7eb2495fbfd38e228a042b3e : SUCCESS in 47m 31s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/600012dea206401b95a110bd429a1176 : SUCCESS in 43m 03s\n- octavia-v2-dsvm-tls-barbican-fips https://zuul.opendev.org/t/openstack/build/8768a93ba2254b948ae8fb21decdc58d : SUCCESS in 1h 12m 10s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/597c137c7b7c4fc5ab58d337d5bfdcee : SUCCESS in 34m 30s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/908927ee08324e8780ad9f8933e118e9 : SUCCESS in 2h 27m 30s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/ac838229a8ac4e1ba987d67b4f6192e0 : SUCCESS in 2h 32m 19s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/5597f90d1dd14b6781ca9e0b96f9ee1c : TIMED_OUT in 2h 32m 40s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/fb05fbd81d924d2a88b8d501b8562e5b : SUCCESS in 2h 16m 42s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/5c07f34d8a9641c2b546f53b46c2f046 : SUCCESS in 1h 59m 47s (non-voting)\n- octavia-v2-dsvm-scenario-centos-9-stream https://zuul.opendev.org/t/openstack/build/8be2124c9cd94fc2aec9c83c5ca07d34 : SUCCESS in 2h 29m 19s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"c414ad15f1d9aeba8529fd1494fe70848c5b7a58","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-20 16:19:39.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"13629cf91a8ab1dd36cfd14d41eaa92032e8b295","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-20 18:03:07.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/087b1c5eb1054e8ea11e98d74601e86c\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9fb14c73a9234c7eaf1be3c209bca4b2 : SUCCESS in 4m 43s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/965ec6b1f038417eb7ef34b76c77e77d : SUCCESS in 4m 31s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/5fa04376ae414202b92e93469151b5f4 : SUCCESS in 4m 34s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1fa5aae2067d4bdfb03c7c1237b3d508 : SUCCESS in 11m 05s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/56308c995d6441d7b36441347081a045 : SUCCESS in 4m 08s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/e285db58d0854128b3917463867e08b2 : SUCCESS in 6m 49s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/d6dc87d332f042fa96ae8e76b831d0ac : SUCCESS in 1h 15m 59s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/6e30153094324aab8fbd6ebe2d7c897f : SUCCESS in 1h 41m 22s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/e878de3eb18d444ead25a5a8b414bbc2 : SUCCESS in 40m 30s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/a606767ca1fc4760b3977c5ba33e7d94 : SUCCESS in 42m 49s","accounts_in_message":[],"_revision_number":1},{"id":"bd3af2d585d4979ade358a3118c22e2e130ff5f2","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-20 18:03:07.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"6c3cc2c60ca4f50afd7422912cdb8df24f15bc46","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-20 18:04:21.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/993c5a84a0254c68aceb4ec1434bdb9b\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/99729569ba484a668e9021041340b934 : SUCCESS in 55s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"6e7ce65ce6b5eeb3f23d562c32ff8f3c5e00bb8d","revisions":{"6e7ce65ce6b5eeb3f23d562c32ff8f3c5e00bb8d":{"kind":"REWORK","_number":1,"created":"2023-04-19 08:57:30.000000000","uploader":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"ref":"refs/changes/01/880801/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/01/880801/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/01/880801/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/01/880801/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/01/880801/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/01/880801/1"}}},"commit":{"parents":[{"commit":"3819edb51a98a18ec1836e5d39d382281acca8a9","subject":"Merge \"Remove specific handling of amphorav1 parameters\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/3819edb51a98a18ec1836e5d39d382281acca8a9"}]}],"author":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2023-04-19 08:55:39.000000000","tz":120},"committer":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2023-04-19 08:55:39.000000000","tz":120},"subject":"Fix pep8 error","message":"Fix pep8 error\n\nnew flake8 issue:\n- E275 missing whitespace after keyword\n\nChange-Id: I4dd4e7dabe71d0a81eaba9bdad6969911a395429\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/6e7ce65ce6b5eeb3f23d562c32ff8f3c5e00bb8d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/6e7ce65ce6b5eeb3f23d562c32ff8f3c5e00bb8d"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
