)]}'
{"id":"openstack%2Foctavia~880986","triplet_id":"openstack%2Foctavia~master~I29d44660d002e8bbfb206db4ce4ebf015ac2a873","project":"openstack/octavia","branch":"master","topic":"ruff","attention_set":{},"removed_from_attention_set":{"34429":{"account":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"last_update":"2024-01-29 15:02:55.000000000","reason":"Change was abandoned"}},"hashtags":[],"change_id":"I29d44660d002e8bbfb206db4ce4ebf015ac2a873","subject":"Add ruff to pre-commit config","status":"NEW","created":"2023-04-20 15:33:06.000000000","updated":"2024-10-28 16:02:34.000000000","submit_type":"MERGE_IF_NECESSARY","mergeable":false,"submittable":false,"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"a79aedd69a48dad676a7b58271af942ef38f1e3a","_number":880986,"virtual_id_number":880986,"owner":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":1,"date":"2024-10-28 16:02:34.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-04-20 18:25:25.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"4829e2a731d70022221444bff56b6cf31c3d5b2e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2023-04-20 15:33:06.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"3b603a71169f5932eda9523c8b7bea4e5e0ee2a8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-20 18:25:25.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/bc3144e388e44baebce599e09d5daa04\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/56053386b3cd412fbd6fbd04a02f1a63 : SUCCESS in 11m 01s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a605e092f7d940008548d3baaddb10d7 : FAILURE in 4m 33s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/3401cad4a720461fa9963808f23a098b : SUCCESS in 4m 57s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/84f14ecf119b4d9f9bafb42103b9ca44 : SUCCESS in 5m 02s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8fceb827b0c7453398601b9dce7f9912 : SUCCESS in 12m 31s\n- octavia-tox-py39-tips https://zuul.opendev.org/t/openstack/build/e506135e7fc34055a6605ed576038e89 : SUCCESS in 5m 22s\n- octavia-tox-functional-py39-tips https://zuul.opendev.org/t/openstack/build/5ade057c3bde4e28b44daabb0437300d : SUCCESS in 7m 53s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/5a928382279c45e3a3d9488d0f1f992c : SUCCESS in 5m 09s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/cbac5f19bdb64dbea602cbe452348309 : SUCCESS in 7m 37s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/46e249c7c17f4c65a2f09a6776dc4424 : SUCCESS in 7m 27s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/921c17fbd12443c19b0689b677cea317 : SUCCESS in 9m 05s (non-voting)\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/cf1e960ada4547028fd93961b3267b68 : SUCCESS in 1h 46m 45s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/2b84c498f982402aa18d86722eadd082 : SUCCESS in 2h 05m 11s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/689aec5fe26f48a2b6410e8a2a47fe71 : SUCCESS in 40m 39s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/1f03603eb5a343bdb0c28f953b1413b5 : SUCCESS in 1h 02m 12s\n- octavia-v2-dsvm-tls-barbican-fips https://zuul.opendev.org/t/openstack/build/452f69b522214430a2d3006e63e4560e : SUCCESS in 1h 01m 40s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/1f6c5b49cb93495a9635772d640f3128 : SUCCESS in 36m 51s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/cc9ff965b090446cbc1ee1dfa2dd4994 : SUCCESS in 1h 55m 54s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/6156d8518cbf4c549c231ec03613c4c0 : SUCCESS in 2h 49m 19s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/5884fdb4fd734b4bbe5abe6371f90c7c : FAILURE in 22m 29s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/442f704ae28b40c18dc2328c3ff0f144 : SUCCESS in 2h 25m 12s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/045bf0574451437b97d55ebaeb9baac7 : FAILURE in 8m 09s (non-voting)\n- octavia-v2-dsvm-scenario-centos-9-stream https://zuul.opendev.org/t/openstack/build/a47d48e8d11c488e9d98faa774bc3dfb : SUCCESS in 2h 09m 18s (non-voting)\n\nWarning:\n  Comments left for invalid file octavia/hacking/checks.py","accounts_in_message":[],"_revision_number":1},{"id":"1bcf525b807cd09d873824191466eb46e340f5f8","tag":"autogenerated:gerrit:abandon","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2024-01-29 15:02:55.000000000","message":"Abandoned\n\nwas just a test","accounts_in_message":[],"_revision_number":1},{"id":"cb8c751be5b7bafd06f68c31c60de769a6c071f9","tag":"autogenerated:gerrit:restore","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2024-10-28 09:43:21.000000000","message":"Restored","accounts_in_message":[],"_revision_number":1},{"id":"76e93f12b288ed844608a83ba5d7b572b06a5940","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2024-10-28 09:44:31.000000000","message":"Patch Set 2: Commit message was updated.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"fba28befd5863e660343dab1eb8a6c95c121416d","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2024-10-28 09:44:45.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"db7e5afac8c656c0358219d29dc70a8fd15d0191","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-28 12:18:56.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2da0b765ac294cf9a4910e207a008965\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/53c0f49c51ff4a2aac024e96efec16a0 : SUCCESS in 9m 08s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/833f452988144ac1920c3172b0f190bb : SUCCESS in 5m 17s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/8ce7385261264ed98d3ba819e262150b : SUCCESS in 4m 24s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/f67fca6795b84c29a4128bfef0e256e6 : SUCCESS in 3m 53s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d7ccbb6f58f9442dbd40e1e88aa3a664 : SUCCESS in 10m 42s\n- octavia-tox-py311-tips https://zuul.opendev.org/t/openstack/build/53cbdf0b29624d439f5aa75c471d721b : SUCCESS in 3m 54s\n- octavia-tox-functional-py311-tips https://zuul.opendev.org/t/openstack/build/71c87998515a4b0ea3fc20905cd72a53 : SUCCESS in 5m 21s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/0876098b59824c4e93523ddcc6fcbe9b : SUCCESS in 3m 57s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/3ed576f601d04619a4e42e52b6bd1b41 : SUCCESS in 6m 33s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/9d1b6ed2cb744cb087c90838fb29635c : SUCCESS in 7m 39s\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/1daeacd85e514b7db84edb97c879fc94 : SUCCESS in 7m 40s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/7ebc17fddf484928aa16969d374543ef : SUCCESS in 1h 38m 59s\n- octavia-v2-dsvm-scenario-traffic-ops https://zuul.opendev.org/t/openstack/build/48cf48ad1b5f4c8ebc348daac4adef0f : SUCCESS in 1h 44m 48s\n- octavia-v2-dsvm-scenario-non-traffic-ops https://zuul.opendev.org/t/openstack/build/c3071fb15de64bf5a02b5a3092ebbbdc : SUCCESS in 51m 43s\n- octavia-v2-dsvm-scenario-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/6e85cc84658c4f1181e0e76591cf1d3b : SUCCESS in 1h 46m 31s\n- octavia-v2-dsvm-scenario-non-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/b27272258bde437f94fd61b875b95b66 : SUCCESS in 1h 05m 35s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/5e30025758034e71bb3a53d9a2fa794e : SUCCESS in 34m 07s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/0c9c93214c704ba2a9a20a84c4e26132 : SUCCESS in 47m 58s\n- octavia-v2-dsvm-tls-barbican-fips https://zuul.opendev.org/t/openstack/build/ac03296bf0f8454d91b8153f602c0de3 : SUCCESS in 49m 49s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/53f4ca9e52a1412dbf063a59017b1e80 : SUCCESS in 46m 11s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/b6e9bcce1a1340b290e58f6edcda3828 : SUCCESS in 2h 12m 30s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/4baa91cf4c9f46ef89a7447f595f788b : FAILURE in 1h 55m 39s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/44bda44b6fe145f4b4701c4ca0ed8397 : SUCCESS in 2h 18m 46s (non-voting)\n- octavia-v2-dsvm-scenario-centos-9-stream https://zuul.opendev.org/t/openstack/build/2c4f5c982a4148cebd61452ee71777bd : TIMED_OUT in 2h 33m 21s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"c9c59206029dd0107cb2c3d3abb1e85f56d33df5","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2024-10-28 12:46:00.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":3},{"id":"bab145f659ea0afe5507ceb7a5c05455caacbb63","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2024-10-28 12:57:45.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"a79aedd69a48dad676a7b58271af942ef38f1e3a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-28 16:02:34.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/16b3a480208142c49872cdba56ab4f7d\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/faf562943776453d8e250baf827a26de : SUCCESS in 9m 28s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1284ab2927654e85949af43dd88525fc : SUCCESS in 4m 27s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/8087fd12cc994cb080ec41f0fa50bcee : SUCCESS in 4m 12s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/156bd2e9b2d64bc99014c705fe46aee2 : SUCCESS in 4m 53s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7fd5a2704d98496295ab21729b44c6b3 : SUCCESS in 9m 54s\n- octavia-tox-py311-tips https://zuul.opendev.org/t/openstack/build/15f9632a0d7b45bdaab41ee9631dda70 : SUCCESS in 4m 03s\n- octavia-tox-functional-py311-tips https://zuul.opendev.org/t/openstack/build/3d210808c3814ed29bf79042b71c7e7f : SUCCESS in 6m 58s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/1faede0f479b48a088ea69f03838d773 : SUCCESS in 4m 24s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/079104be5df443d0b42eccbf3b11e33b : SUCCESS in 7m 23s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/a8a3d53df4fe4865a9f9cbb0afe28887 : SUCCESS in 7m 28s\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/fc1cb009d02b4fd5a35b16d03c128c33 : SUCCESS in 9m 53s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/7f253129c1d14f8d80a7d289e18bc5bb : SUCCESS in 1h 32m 02s\n- octavia-v2-dsvm-scenario-traffic-ops https://zuul.opendev.org/t/openstack/build/edeac658c2934b309bfcbd6b4aa05a05 : SUCCESS in 1h 37m 58s\n- octavia-v2-dsvm-scenario-non-traffic-ops https://zuul.opendev.org/t/openstack/build/128a7f0e42d147fbaa92b1f4df638394 : SUCCESS in 59m 02s\n- octavia-v2-dsvm-scenario-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/9d85aa5d92ae421695ef7c8d9e168973 : SUCCESS in 1h 40m 41s\n- octavia-v2-dsvm-scenario-non-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/6fb10f0aa57c4af68f6f8aaad93030d5 : SUCCESS in 1h 06m 54s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/ab469073bf6442fdb72cecb037ddc76e : SUCCESS in 31m 20s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/a0573445f5de41578a04caf02941ce73 : SUCCESS in 42m 19s\n- octavia-v2-dsvm-tls-barbican-fips https://zuul.opendev.org/t/openstack/build/169c9630f0b845ae9435e85340369158 : SUCCESS in 44m 36s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/b81b9f73f499422ca3f9d6433ec40cc2 : SUCCESS in 36m 13s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/9e3af5b5c0a343e3a4942292ebe154e5 : TIMED_OUT in 2h 34m 01s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/c656a87297de49519f6db8e2776b83ba : TIMED_OUT in 3h 03m 22s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/7a48b79202064bddb8b50df6546539b8 : TIMED_OUT in 2h 32m 40s (non-voting)\n- octavia-v2-dsvm-scenario-centos-9-stream https://zuul.opendev.org/t/openstack/build/8472f4b441604099a2984ef5aaccf585 : TIMED_OUT in 2h 34m 54s (non-voting)","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"35f4756a2a79ab28d18ceca5124882d09569f9b7","revisions":{"59e9456585a427a9f802b97761503b4d6e054f59":{"kind":"REWORK","_number":1,"created":"2023-04-20 15:33:06.000000000","uploader":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"ref":"refs/changes/86/880986/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/86/880986/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/86/880986/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/86/880986/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/86/880986/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/86/880986/1"}}},"commit":{"parents":[{"commit":"3819edb51a98a18ec1836e5d39d382281acca8a9","subject":"Merge \"Remove specific handling of amphorav1 parameters\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/3819edb51a98a18ec1836e5d39d382281acca8a9"}]}],"author":{"name":"Tom Weininger","email":"tweining@redhat.com","date":"2023-04-20 15:32:48.000000000","tz":120},"committer":{"name":"Tom Weininger","email":"tweining@redhat.com","date":"2023-04-20 15:32:48.000000000","tz":120},"subject":"Try out ruff","message":"Try out ruff\n\nChange-Id: I29d44660d002e8bbfb206db4ce4ebf015ac2a873\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/59e9456585a427a9f802b97761503b4d6e054f59"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/59e9456585a427a9f802b97761503b4d6e054f59"}]},"branch":"refs/heads/master"},"88db4ae494ed9178a0a621cd6042f8a9284fdf8e":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2024-10-28 09:44:31.000000000","uploader":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"ref":"refs/changes/86/880986/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/86/880986/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/86/880986/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/86/880986/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/86/880986/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/86/880986/2"}}},"commit":{"parents":[{"commit":"3819edb51a98a18ec1836e5d39d382281acca8a9","subject":"Merge \"Remove specific handling of amphorav1 parameters\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/3819edb51a98a18ec1836e5d39d382281acca8a9"}]}],"author":{"name":"Tom Weininger","email":"tweining@redhat.com","date":"2023-04-20 15:32:48.000000000","tz":120},"committer":{"name":"Tom Weininger","email":"tweining@redhat.com","date":"2024-10-28 09:44:31.000000000","tz":0},"subject":"Add ruff to pre-commit config","message":"Add ruff to pre-commit config\n\nAs agreed during the 2025.1 PTG we want to try out ruff in pre-commit.\n\nChange-Id: I29d44660d002e8bbfb206db4ce4ebf015ac2a873\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/88db4ae494ed9178a0a621cd6042f8a9284fdf8e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/88db4ae494ed9178a0a621cd6042f8a9284fdf8e"}]},"branch":"refs/heads/master","description":"Edit commit message"},"c2af5104b481051a33431ef0d5c816ef4e4ada57":{"kind":"REWORK","_number":3,"created":"2024-10-28 12:46:00.000000000","uploader":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"ref":"refs/changes/86/880986/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/86/880986/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/86/880986/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/86/880986/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/86/880986/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/86/880986/3"}}},"commit":{"parents":[{"commit":"3819edb51a98a18ec1836e5d39d382281acca8a9","subject":"Merge \"Remove specific handling of amphorav1 parameters\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/3819edb51a98a18ec1836e5d39d382281acca8a9"}]}],"author":{"name":"Tom Weininger","email":"tweining@redhat.com","date":"2023-04-20 15:32:48.000000000","tz":120},"committer":{"name":"Tom Weininger","email":"tweining@redhat.com","date":"2024-10-28 12:45:52.000000000","tz":60},"subject":"Add ruff to pre-commit config","message":"Add ruff to pre-commit config\n\nAs agreed during the 2025.1 PTG we want to try out ruff in pre-commit.\n\nChange-Id: I29d44660d002e8bbfb206db4ce4ebf015ac2a873\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/c2af5104b481051a33431ef0d5c816ef4e4ada57"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/c2af5104b481051a33431ef0d5c816ef4e4ada57"}]},"branch":"refs/heads/master"},"35f4756a2a79ab28d18ceca5124882d09569f9b7":{"kind":"REWORK","_number":4,"created":"2024-10-28 12:57:45.000000000","uploader":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"ref":"refs/changes/86/880986/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/86/880986/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/86/880986/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/86/880986/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/86/880986/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/86/880986/4"}}},"commit":{"parents":[{"commit":"b99b6f2ac1ff24d65b14e4dbcff84a3871c0a2f8","subject":"Update pre-commit pyugrade hook","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/b99b6f2ac1ff24d65b14e4dbcff84a3871c0a2f8"}]}],"author":{"name":"Tom Weininger","email":"tweining@redhat.com","date":"2023-04-20 15:32:48.000000000","tz":120},"committer":{"name":"Tom Weininger","email":"tweining@redhat.com","date":"2024-10-28 12:57:13.000000000","tz":60},"subject":"Add ruff to pre-commit config","message":"Add ruff to pre-commit config\n\nAs agreed during the 2025.1 PTG we want to try out ruff in pre-commit.\n\nChange-Id: I29d44660d002e8bbfb206db4ce4ebf015ac2a873\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/35f4756a2a79ab28d18ceca5124882d09569f9b7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/35f4756a2a79ab28d18ceca5124882d09569f9b7"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"OK","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY"},{"label":"Workflow","status":"MAY"},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
