)]}'
{"id":"openstack%2Foctavia~881719","triplet_id":"openstack%2Foctavia~master~I7f150c2c1c3da706055aba5e02b5ad7c032806f1","project":"openstack/octavia","branch":"master","attention_set":{},"removed_from_attention_set":{"11628":{"account":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"last_update":"2023-05-25 00:16:18.000000000","reason":"Change was submitted"},"29244":{"account":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"last_update":"2023-05-24 06:09:57.000000000","reason":"\u003cGERRIT_ACCOUNT_29244\u003e replied on the change","reason_account":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"}},"34429":{"account":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"last_update":"2023-05-23 14:53:52.000000000","reason":"removed on reply"}},"hashtags":[],"change_id":"I7f150c2c1c3da706055aba5e02b5ad7c032806f1","subject":"Avoid interface name collisions in the amphora","status":"MERGED","created":"2023-04-27 12:03:03.000000000","updated":"2023-05-25 00:18:21.000000000","submitted":"2023-05-25 00:16:18.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":9,"unresolved_comment_count":2,"has_review_started":true,"submission_id":"881719","meta_rev_id":"d6c75efd1b37a5cec901a773ea0ebca52e84570e","_number":881719,"virtual_id_number":881719,"owner":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-05-25 00:16:18.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2023-05-23 19:47:18.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":2,"date":"2023-05-23 14:53:52.000000000","_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2023-05-23 19:47:18.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-04-27 14:39:56.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-05-02 11:00:58.000000000","updated_by":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"reviewer":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"state":"CC"},{"updated":"2023-05-23 14:53:52.000000000","updated_by":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"reviewer":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"state":"REVIEWER"},{"updated":"2023-05-23 19:47:18.000000000","updated_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"reviewer":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"state":"REVIEWER"}],"messages":[{"id":"9a5948c2b0547abd65ed71ff49eb70ebb5d6921a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2023-04-27 12:03:03.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"3fcaad87562e47ee34f4721aa74a520771c5159a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-27 14:39:56.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/a58818c4c877491c946702c4ade50c75\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/a6a217db5bcb4b03a1ffe03d5dd124fe : FAILURE in 5m 23s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a34b07db39c743618d07f2710807855a : SUCCESS in 5m 53s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/9bfb8866e4fb4dd892df48750b1d7ace : FAILURE in 4m 54s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/f715a02a3ac749ac88ee37e91b0a874e : FAILURE in 6m 24s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/82b16a2ae38140e394ef1d55956174aa : SUCCESS in 12m 29s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/749b2c5e8b00426ca00957e508f9aca5 : SUCCESS in 6m 58s\n- octavia-tox-py39-tips https://zuul.opendev.org/t/openstack/build/0ebf0a5a036e45159289b36396e2e3e1 : FAILURE in 6m 34s\n- octavia-tox-functional-py39-tips https://zuul.opendev.org/t/openstack/build/b4295c7a81944d6a83e04497cd0ef6f2 : FAILURE in 7m 43s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/2ac53519386f4fb19de83b3b293bdb7e : SUCCESS in 4m 11s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/facdb746a1a744a99d78dfc73767cf05 : FAILURE in 9m 44s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/7663b8123cc148c2a9d543e8c0797719 : FAILURE in 8m 02s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/3f1d274ed87841f5bd098bf6a909ab0d : FAILURE in 7m 36s (non-voting)\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/4726e3ce595049b5b416402eacfac6dc : SUCCESS in 1h 52m 27s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/f33d31a278fb408ca353b31ec96198a8 : TIMED_OUT in 2h 34m 19s\n- octavia-v2-dsvm-scenario-jobboard https://zuul.opendev.org/t/openstack/build/f62fdf14177345cc9ea21c2d0ea0bb6a : TIMED_OUT in 2h 32m 57s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/7c676009ef2f4b849aa0814ea2d5e525 : SUCCESS in 48m 09s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/ab22244e3d504abbaffa28ad3f90b2b1 : SUCCESS in 38m 44s\n- octavia-v2-dsvm-tls-barbican-fips https://zuul.opendev.org/t/openstack/build/bf685585a382422da650c91f3f100681 : SUCCESS in 1h 17m 21s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/64a39e6842534d14a776623ea81b9fcc : SUCCESS in 33m 23s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/a740ca516f2f4ae5a32ba51ea5a4c0dc : SUCCESS in 2h 02m 47s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/2d4683680e174755954f186ee14b48c0 : SUCCESS in 2h 31m 23s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/926c518b66b14c79939be52451171603 : FAILURE in 6m 14s (non-voting)\n- octavia-v2-dsvm-scenario-centos-9-stream https://zuul.opendev.org/t/openstack/build/486a45af0f974efc9c617cf8163d5896 : SUCCESS in 2h 21m 42s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"c0471573e6d14b1f854df42937e6888b32ac925a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2023-04-27 15:34:44.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"5b95f7a54ed912ffa9c72397e9087e3be4d50280","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-27 18:12:20.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/cf01479fa3ae491b8ca3a32409755786\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/0a432d674b2443848a25e7a57a93afc1 : SUCCESS in 10m 45s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5a39ff9f8ffb475d8f5022099b8bb64b : SUCCESS in 5m 54s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/8dbc8d3f54e241cdb83f7cc4a0567bdc : SUCCESS in 6m 00s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/5cdde7ac8ebf46f3a665e02b7a962312 : SUCCESS in 5m 02s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a410134b5b1741c0a4f2e3f252a745c2 : SUCCESS in 11m 18s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/115dd2c0292444caa42d6bdd64ab706d : SUCCESS in 5m 02s\n- octavia-tox-py39-tips https://zuul.opendev.org/t/openstack/build/e54cd2a57b784a6d96c58b501d3f2bc9 : SUCCESS in 5m 46s\n- octavia-tox-functional-py39-tips https://zuul.opendev.org/t/openstack/build/f9efd807b7ce420493f5849796f134f0 : SUCCESS in 7m 30s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/7663268c89b349719ab2cf45b7828cbe : SUCCESS in 4m 43s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/a2b69debcec749ff96ac6d9e21b78349 : SUCCESS in 7m 45s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/2504ec8026ce4481816e0dc3119e1832 : SUCCESS in 8m 12s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/f19f516ca0104094aa694ca3f8df232a : SUCCESS in 6m 37s (non-voting)\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/22d1a7071abe4dda958c55bcad00b8fa : SUCCESS in 1h 32m 22s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/e3448f55ef214f4e9ff68e05b483284c : SUCCESS in 2h 06m 52s\n- octavia-v2-dsvm-scenario-jobboard https://zuul.opendev.org/t/openstack/build/85fbfebbb0cf4deb8130ee612be1afce : SUCCESS in 2h 27m 40s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/ae5d13059f084444a8fe6dc80725fb4c : SUCCESS in 39m 15s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/5f663035f6cb4e00ab7764897796ac44 : SUCCESS in 42m 09s\n- octavia-v2-dsvm-tls-barbican-fips https://zuul.opendev.org/t/openstack/build/0b8e1d7fff82428a86c75f5cfc7d93e2 : SUCCESS in 1h 02m 45s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/dd9008d2102540eaa2212c0b8b65a64e : FAILURE in 6m 15s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/0734b38dc17244779eb55b673ceb7c9f : SUCCESS in 1h 58m 16s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/8e52d5307d574b22b3efca5049edbadd : FAILURE in 24m 27s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/7de5e3d779eb4057b01b33f72d0b6ec0 : FAILURE in 6m 10s (non-voting)\n- octavia-v2-dsvm-scenario-centos-9-stream https://zuul.opendev.org/t/openstack/build/fa0854af38474f4699a207d72c48457a : TIMED_OUT in 2h 34m 57s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"51f9597819320f97f9a44a737931625e0c691027","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2023-05-02 11:00:58.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"f1ef3a941b0d3fd2f8f2edc8e3c9c00d7492aab3","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2023-05-02 11:04:29.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"f4d3d680942d9c266c0ff8e6b8c5003601392cfa","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2023-05-02 11:58:48.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"f9a87112a3ffce50c496e1da30592aabaeb94467","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2023-05-16 09:05:29.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":3},{"id":"506a260a332d29744c9267e9a57368678f6305e5","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2023-05-16 09:05:33.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"a4055dab66d91486c998d19375ff12ed5852cbe1","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2023-05-16 09:58:08.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"b2e727c48f7cb32926f4d5756bf26b46b595758e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-16 12:51:15.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/4983ab56d8cc48138721bddef633998f\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/5b5d39eb5cd9410d9254306f21085d9c : SUCCESS in 10m 57s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/53029c60fc8c4451ade32d650cb8e925 : SUCCESS in 5m 13s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/efe2275013d247e7ab835862b43d32f9 : SUCCESS in 4m 34s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/eb49447fd17f40d29875a03ee0c715c0 : SUCCESS in 4m 33s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/b89328b8823b4885b8fe1b19f87f62bc : SUCCESS in 5m 00s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6c1133c5eada4db28cab2a6c0dbb6b88 : SUCCESS in 10m 16s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/8ea24cb2681e4d24bf39a9b6bbd6e77b : SUCCESS in 4m 24s\n- octavia-tox-py39-tips https://zuul.opendev.org/t/openstack/build/5074d00607da4d709d008a7c94b6981f : SUCCESS in 5m 45s\n- octavia-tox-functional-py39-tips https://zuul.opendev.org/t/openstack/build/af24c7a11ad9405496d360a778eb4f23 : SUCCESS in 8m 06s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/8e55b47196a2438b9cb91e36d8471bc3 : SUCCESS in 5m 21s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/64c7b91ba2e14205b5d7ad06599feb03 : SUCCESS in 6m 47s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/f1ac8e63db094daca68ec8cb3a576dbf : SUCCESS in 7m 24s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/c1adf40da17b427f9ca7769bf05e7302 : SUCCESS in 8m 31s (non-voting)\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/e80a4cc4502e4bc191a36b2f05e46c9d : SUCCESS in 1h 36m 35s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/891bf07351034603ac2b3ff68dc6e673 : TIMED_OUT in 2h 33m 00s\n- octavia-v2-dsvm-scenario-jobboard https://zuul.opendev.org/t/openstack/build/ac3480c4e67d48fa92b6a4df5d21eff9 : TIMED_OUT in 2h 33m 56s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/2600bdf1cfe94b529c2f0eaa32902c2e : SUCCESS in 41m 11s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/514227dfeb2c48e49ef19ddb037e5340 : SUCCESS in 42m 03s\n- octavia-v2-dsvm-tls-barbican-fips https://zuul.opendev.org/t/openstack/build/551b384195f34a568926c6a890f6777a : SUCCESS in 1h 20m 05s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/d54b4e61cc17435c8aeb5b22739e5551 : SUCCESS in 32m 31s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/4974924faaf6452492f4b5801b57a56b : SUCCESS in 2h 24m 05s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/1e36563475f747459b069100e7c76f38 : SUCCESS in 2h 46m 56s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/4bea90ecbbae4039a92c56dc24de2db9 : TIMED_OUT in 2h 34m 37s (non-voting)\n- octavia-v2-dsvm-scenario-centos-9-stream https://zuul.opendev.org/t/openstack/build/2a60c442173c48e6be1346c8fbe3cfe6 : TIMED_OUT in 2h 34m 53s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"1072d490cc4ccc51f05a2b3d5f43c0b31cad55ca","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2023-05-23 07:32:15.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"e88b984ad68bf40b3802c6c80791d774f0d0fa93","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-23 10:14:35.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c250091701cc487bb52bd567136ba65e\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/0dd5e46cb9d24cef850d487d8ee4c219 : SUCCESS in 11m 05s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1ec6204dfae5471cb70100ccc63a1251 : SUCCESS in 5m 02s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/6ee10c3c458e4a31938bd076e974695e : SUCCESS in 6m 19s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/490a11a4123041a1980c49ba721d5cee : SUCCESS in 4m 27s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/d3128a3fc12841639a9f825020c1c414 : SUCCESS in 4m 45s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2afba4b579384fe985396ab05c93fc28 : SUCCESS in 11m 42s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/0000e2c1aa00451b94764a6941e1f50e : SUCCESS in 4m 48s\n- octavia-tox-py39-tips https://zuul.opendev.org/t/openstack/build/2c975a11c1d2445aad491a52186a555b : SUCCESS in 6m 47s\n- octavia-tox-functional-py39-tips https://zuul.opendev.org/t/openstack/build/1bd7257eaa2f4dbfb4868cbcf2bb8da2 : SUCCESS in 7m 54s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/159d71842fe84c77a3c31703e7ffa721 : SUCCESS in 4m 24s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/d21ff31b049b4d4da464c8891dd9514e : SUCCESS in 7m 53s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/0c18badd91374a4799072910935ab974 : SUCCESS in 8m 01s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/a2c729e8df9847ee81db1fcdb14a7d45 : SUCCESS in 8m 09s (non-voting)\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/36d3e0a0364f40ba95e19df03c12de91 : SUCCESS in 1h 21m 27s\n- octavia-v2-dsvm-scenario-traffic-ops https://zuul.opendev.org/t/openstack/build/1a8050f79f9c402db68b43cc2793ce81 : SUCCESS in 1h 57m 35s\n- octavia-v2-dsvm-scenario-non-traffic-ops https://zuul.opendev.org/t/openstack/build/b819730f99e54b6ca340c43e326e3cb4 : SUCCESS in 1h 25m 49s\n- octavia-v2-dsvm-scenario-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/2a2d8c95e1b14ababa685371573095a5 : SUCCESS in 2h 03m 57s\n- octavia-v2-dsvm-scenario-non-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/effb7bd35cf34cd385b3ad878412dfef : SUCCESS in 1h 17m 29s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/6c5a1c0fbb45459fb047e9f7b8f953d7 : SUCCESS in 46m 23s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/fdb0dbd5540f4333adbb9a21b339a83c : SUCCESS in 50m 19s\n- octavia-v2-dsvm-tls-barbican-fips https://zuul.opendev.org/t/openstack/build/d8c855446a4b4870ae74776389c21625 : FAILURE in 11m 35s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/375bf0aa39224a15b74318c1a9cc10ca : SUCCESS in 33m 01s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/e055f72558a34915a0faabaaf8dd1087 : TIMED_OUT in 2h 34m 37s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/48f15a6e9d4f4d94aa679853f63a0d12 : SUCCESS in 2h 22m 04s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/3535c95d969144daba857e8797e248ff : TIMED_OUT in 2h 32m 48s (non-voting)\n- octavia-v2-dsvm-scenario-centos-9-stream https://zuul.opendev.org/t/openstack/build/6326d788fa3e4e048ca6f2227b8d9662 : SUCCESS in 2h 29m 05s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"7287d87e5317ca0e6141b8602aa6ede1703267e8","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2023-05-23 14:53:52.000000000","message":"Patch Set 4: Code-Review+2","accounts_in_message":[],"_revision_number":4},{"id":"ba0e20761ede866a5dd2c5dcf1d4f3c8ff8cf0df","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2023-05-23 19:47:18.000000000","message":"Patch Set 4: Code-Review+2 Workflow+1\n\n(2 comments)","accounts_in_message":[],"_revision_number":4},{"id":"e0404caff4019460e175c5091cd6a5a72dd7f186","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-23 19:47:55.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"900a18e7c9de5c482b5227f7f5a4191d4dfb707b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-23 22:21:58.000000000","message":"Patch Set 4: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/7820b0785c354a7793db515a8e90854e\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/910f4d1c33eb48569f177631ab11d481 : SUCCESS in 4m 58s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/e888c8f241284f4a94044b8d5f8f20e5 : SUCCESS in 4m 21s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/8e00d00b6b7b464aa26b597bfa40b1c0 : SUCCESS in 4m 37s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/245b5dd058ae402fa732ef73e7f48e96 : SUCCESS in 5m 08s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f56e9fd93c174783aee31747d63f8f06 : SUCCESS in 9m 46s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/60dc4aaf606a4b8f9e594e51f7ce78ed : SUCCESS in 4m 24s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/4084a820ac7b46b895a84d1f4da3b242 : SUCCESS in 4m 38s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/69d8666e1a9b4cc39810f6eccbb4f8ae : SUCCESS in 7m 09s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/72f5f474a03949df82b3c130ba41a3d4 : SUCCESS in 1h 22m 11s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/8efcb7830b0549c1930341b3f5ef29d1 : TIMED_OUT in 2h 33m 06s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/db98c549dcef403998120d12a21be6d6 : SUCCESS in 41m 23s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/b1a39abf0e784e1992029f78812e268a : SUCCESS in 41m 03s","accounts_in_message":[],"_revision_number":4},{"id":"8a1e3c6a4af7b430effa80f1feb436eff69cff18","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2023-05-24 06:09:57.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"1ad5398c64dac081e93c02ca3d44b3a6ad59750a","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2023-05-24 19:12:23.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"61165732c36cbc60a59e329b5fc08d924893c0e3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-24 22:17:08.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/668072007977404e8956b2102a9578d4\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/b7aeb5e1c0c64c3db63912352f2b1615 : SUCCESS in 10m 26s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b70e9084f09240e787f4bad0aeb994a3 : SUCCESS in 4m 02s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/1a812aed6a394405a5333483a9771063 : SUCCESS in 4m 51s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/0edee6be50524210bdb11087000051da : SUCCESS in 5m 16s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/cb8f8c825fa84338a73418a8a0890c03 : SUCCESS in 4m 51s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6e83d0ff2bdc46bfbf5cf0bfbb2d0057 : SUCCESS in 12m 21s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/c3edb8374feb4afb93ea05f4f18ec523 : SUCCESS in 5m 32s\n- octavia-tox-py39-tips https://zuul.opendev.org/t/openstack/build/17f85de489f04e0caaf6a1294f0d1d3b : SUCCESS in 6m 01s\n- octavia-tox-functional-py39-tips https://zuul.opendev.org/t/openstack/build/1190675eab584b5290db8f19fe1c856a : SUCCESS in 8m 22s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/c8b239ea2eb444658ee055746a7c0060 : SUCCESS in 4m 23s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/2ab9b70336e74746b441d6d1d6682578 : SUCCESS in 6m 48s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/014203fe46534243b0e58fcd76f33e98 : SUCCESS in 9m 45s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/d8d252557b154eef808e051c54afe16e : SUCCESS in 8m 14s (non-voting)\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/7836be4bfed942229ebaaf0a5925e7f6 : SUCCESS in 1h 33m 15s\n- octavia-v2-dsvm-scenario-traffic-ops https://zuul.opendev.org/t/openstack/build/e924e2cb493046738a8a7cedfc6b32d1 : SUCCESS in 1h 55m 24s\n- octavia-v2-dsvm-scenario-non-traffic-ops https://zuul.opendev.org/t/openstack/build/1ba67a603c264526910352811afbe8f1 : SUCCESS in 1h 21m 48s\n- octavia-v2-dsvm-scenario-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/de5227d856a349f5b7710e8b8001a474 : SUCCESS in 2h 03m 13s\n- octavia-v2-dsvm-scenario-non-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/2c537fc8e8ac4300b5f72ad49891978d : SUCCESS in 1h 20m 36s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/b8acc0a224d248c5aadbdbec0f6d989f : SUCCESS in 41m 50s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/5713a58239c246d192ad11791b9b6e28 : SUCCESS in 37m 04s\n- octavia-v2-dsvm-tls-barbican-fips https://zuul.opendev.org/t/openstack/build/3e9009976a3348689710f0bc4369816e : FAILURE in 11m 28s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/73906f73234e4398937e1aa04b18a0cb : SUCCESS in 32m 35s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/9275a1becc7b48d78229b71601d5d124 : TIMED_OUT in 2h 32m 53s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/d92c2ee712a941489f95c96b3979f0f5 : TIMED_OUT in 3h 03m 19s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/33b23f420171475883751761806078d8 : TIMED_OUT in 2h 32m 51s (non-voting)\n- octavia-v2-dsvm-scenario-centos-9-stream https://zuul.opendev.org/t/openstack/build/9aefec6482cc45a08ae1ca5bd2781937 : FAILURE in 1h 25m 01s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"341fa581d10c1d6a464d221aff13875a85c838a7","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-24 22:17:45.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"72e4276a9344652e35233094c0ee046443d95f88","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-25 00:16:18.000000000","message":"Patch Set 4: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c2d598103229400ebdf12196fa3a90aa\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0045b11f78e44d3eaa7b88e465701ac9 : SUCCESS in 5m 43s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/bb317c4f3e0642ffb00d26245b697e9f : SUCCESS in 4m 36s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/9c7e02ef162b48aaa9db7105cf429830 : SUCCESS in 4m 15s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/937025f0af4349ebad8284f5e46bdb7e : SUCCESS in 4m 35s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/256c945e6fc34f7eba060386a0e8721d : SUCCESS in 11m 56s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/b3c21ee374b2478e87984c152da10b26 : SUCCESS in 5m 37s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/3188919eef4a45fea16a7390a28c96db : SUCCESS in 4m 23s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/4332e5bf997341528170a36aa0c2795a : SUCCESS in 8m 26s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/6798e6e2056f4a1aafc3595ab8367b75 : SUCCESS in 1h 27m 23s\n- octavia-v2-dsvm-scenario-traffic-ops https://zuul.opendev.org/t/openstack/build/3abfb8dc0dee4318b56fd2608c73e142 : SUCCESS in 1h 57m 15s\n- octavia-v2-dsvm-scenario-non-traffic-ops https://zuul.opendev.org/t/openstack/build/1281d0cbd8e247d790f73b58e0087369 : SUCCESS in 1h 17m 31s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/7387a99dc1464e6f8fe7f8543aec53b7 : SUCCESS in 37m 42s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/aee26118ef86448e8f245311c0f838a3 : SUCCESS in 37m 08s","accounts_in_message":[],"_revision_number":4},{"id":"bc1b71732b04b8104fdc995a1462c518cf84e9f1","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-25 00:16:18.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":4},{"id":"d6c75efd1b37a5cec901a773ea0ebca52e84570e","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-25 00:18:21.000000000","message":"Patch Set 4:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3e461ef3a02345bbbe49c7f0238891b7\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/afe0457f41004e89a82dcff850c2ef26 : SUCCESS in 1m 37s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/3b042af58d1e4310871a0b17757dfa3f : SUCCESS in 41s","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"c4dd87f9b119ccaef2d01ae35c9fa0ffbe5f23b2","revisions":{"1b8dda513ee8a66857cb5b90ad4784a30406c9d2":{"kind":"REWORK","_number":1,"created":"2023-04-27 12:03:03.000000000","uploader":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"ref":"refs/changes/19/881719/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/19/881719/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/19/881719/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/19/881719/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/19/881719/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/19/881719/1"}}},"commit":{"parents":[{"commit":"db03617acbc9b1ebf90fae6db79d5a10254eee2b","subject":"Merge \"Change Pydevd settrace configuration\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/db03617acbc9b1ebf90fae6db79d5a10254eee2b"}]}],"author":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2023-04-27 10:00:52.000000000","tz":-240},"committer":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2023-04-27 12:00:00.000000000","tz":-240},"subject":"Avoid interface name collisions in the amphora","message":"Avoid interface name collisions in the amphora\n\nAfter unpluging a network, the octavia-agent may try to plug a new\nnetwork to a new interface while another interface with the same name\nalready exists.\n\nCloses-Bug: #2017894\n\nChange-Id: I7f150c2c1c3da706055aba5e02b5ad7c032806f1\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/1b8dda513ee8a66857cb5b90ad4784a30406c9d2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/1b8dda513ee8a66857cb5b90ad4784a30406c9d2"}]},"branch":"refs/heads/master"},"10bb635488c9478516e44026bbc7f64b3023933c":{"kind":"REWORK","_number":2,"created":"2023-04-27 15:34:44.000000000","uploader":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"ref":"refs/changes/19/881719/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/19/881719/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/19/881719/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/19/881719/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/19/881719/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/19/881719/2"}}},"commit":{"parents":[{"commit":"db03617acbc9b1ebf90fae6db79d5a10254eee2b","subject":"Merge \"Change Pydevd settrace configuration\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/db03617acbc9b1ebf90fae6db79d5a10254eee2b"}]}],"author":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2023-04-27 10:00:52.000000000","tz":-240},"committer":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2023-04-27 15:31:43.000000000","tz":-240},"subject":"Avoid interface name collisions in the amphora","message":"Avoid interface name collisions in the amphora\n\nAfter unpluging a network, the octavia-agent may try to plug a new\nnetwork to a new interface while another interface with the same name\nalready exists.\n\nCloses-Bug: #2017894\n\nChange-Id: I7f150c2c1c3da706055aba5e02b5ad7c032806f1\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/10bb635488c9478516e44026bbc7f64b3023933c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/10bb635488c9478516e44026bbc7f64b3023933c"}]},"branch":"refs/heads/master"},"ed6b2263a77d0e1e401ee0a7a96e40fddee56f68":{"kind":"REWORK","_number":3,"created":"2023-05-16 09:05:29.000000000","uploader":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"ref":"refs/changes/19/881719/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/19/881719/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/19/881719/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/19/881719/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/19/881719/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/19/881719/3"}}},"commit":{"parents":[{"commit":"db03617acbc9b1ebf90fae6db79d5a10254eee2b","subject":"Merge \"Change Pydevd settrace configuration\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/db03617acbc9b1ebf90fae6db79d5a10254eee2b"}]}],"author":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2023-04-27 10:00:52.000000000","tz":-240},"committer":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2023-05-16 09:05:23.000000000","tz":120},"subject":"Avoid interface name collisions in the amphora","message":"Avoid interface name collisions in the amphora\n\nAfter unpluging a network, the octavia-agent may try to plug a new\nnetwork to a new interface while another interface with the same name\nalready exists.\n\nCloses-Bug: #2017894\n\nChange-Id: I7f150c2c1c3da706055aba5e02b5ad7c032806f1\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/ed6b2263a77d0e1e401ee0a7a96e40fddee56f68"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/ed6b2263a77d0e1e401ee0a7a96e40fddee56f68"}]},"branch":"refs/heads/master"},"c4dd87f9b119ccaef2d01ae35c9fa0ffbe5f23b2":{"kind":"REWORK","_number":4,"created":"2023-05-16 09:58:08.000000000","uploader":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"ref":"refs/changes/19/881719/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/19/881719/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/19/881719/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/19/881719/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/19/881719/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/19/881719/4"}}},"commit":{"parents":[{"commit":"db03617acbc9b1ebf90fae6db79d5a10254eee2b","subject":"Merge \"Change Pydevd settrace configuration\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/db03617acbc9b1ebf90fae6db79d5a10254eee2b"}]}],"author":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2023-04-27 10:00:52.000000000","tz":-240},"committer":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2023-05-16 09:52:54.000000000","tz":120},"subject":"Avoid interface name collisions in the amphora","message":"Avoid interface name collisions in the amphora\n\nAfter unpluging a network, the octavia-agent may try to plug a new\nnetwork to a new interface while another interface with the same name\nalready exists.\n\nCloses-Bug: #2017894\n\nChange-Id: I7f150c2c1c3da706055aba5e02b5ad7c032806f1\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/c4dd87f9b119ccaef2d01ae35c9fa0ffbe5f23b2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/c4dd87f9b119ccaef2d01ae35c9fa0ffbe5f23b2"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
