)]}'
{"id":"openstack%2Foctavia~881805","triplet_id":"openstack%2Foctavia~master~I86e50b9d3070e757f1bba8b390c8b7557e41a325","project":"openstack/octavia","branch":"master","attention_set":{"31664":{"account":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"last_update":"2024-05-21 11:25:33.000000000","reason":"Vote got outdated and was removed: Code-Review+1"},"29244":{"account":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"last_update":"2023-08-24 07:52:33.000000000","reason":"\u003cGERRIT_ACCOUNT_34429\u003e replied on the change","reason_account":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}},"34429":{"account":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"last_update":"2024-10-28 11:35:55.000000000","reason":"A robot voted negatively on a label"}},"removed_from_attention_set":{},"hashtags":[],"change_id":"I86e50b9d3070e757f1bba8b390c8b7557e41a325","subject":"Make tests work with pytest runner","status":"NEW","created":"2023-04-28 08:35:24.000000000","updated":"2025-04-02 09:50:09.000000000","submit_type":"MERGE_IF_NECESSARY","mergeable":false,"submittable":false,"total_comment_count":5,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"196359604c54b783a8f1d63572d487d6040e3bcb","_number":881805,"virtual_id_number":881805,"owner":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},{"tag":"autogenerated:zuul:check","value":1,"date":"2025-03-03 13:46:43.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-04-28 08:36:00.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-05-30 13:59:00.000000000","updated_by":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"reviewer":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"state":"CC"},{"updated":"2023-05-31 08:37:49.000000000","updated_by":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"reviewer":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"state":"REVIEWER"},{"updated":"2023-08-23 17:08:07.000000000","updated_by":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"reviewer":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"state":"REVIEWER"}],"messages":[{"id":"3bec9d28d3cabe3e7010247bfde1ef63049037b9","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2023-04-28 08:35:24.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"b86cda0db98086b459e7ae58e9c04c6652f7ef80","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-28 08:36:00.000000000","message":"Patch Set 1: Verified-1\n\nThis change depends on a change with an invalid configuration.","accounts_in_message":[],"_revision_number":1},{"id":"77a855947e0b027feb8a59bbf402fee3fb1d394d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2023-04-28 08:40:53.000000000","message":"Uploaded patch set 2: Patch Set 1 was rebased.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"ce570686f7974f27de53138340cd1b7bce6b461c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-28 08:56:33.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/05bfb25911fa404f8e7ebbff22aae0f9\n\n- octavia-pytest https://zuul.opendev.org/t/openstack/build/572ebef3724c4a7d8afca8c7612bd06a : FAILURE in 13m 53s","accounts_in_message":[],"_revision_number":2},{"id":"47317e2575878c96954df2d7a8fac20e285a5580","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2023-04-28 08:59:59.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":3},{"id":"5904cfc0bf03e0471678c4284337c55a8826e5cd","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-28 09:18:41.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/407eb873a5074e51a8a76222c9ffe909\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/f85d30fe9e1349618c22fb71195110f5 : SUCCESS in 12m 18s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8363f684177e47ba998ae9c388032ae1 : SUCCESS in 5m 11s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/829c4c46540f47f885b05d344ef892e1 : SUCCESS in 4m 42s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/bf9084b3014e4baab4ea48c053f30620 : SUCCESS in 5m 00s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0aa2620b168b4fd699f6bfd8cd8b1ed4 : SUCCESS in 11m 45s\n- octavia-tox-py39-tips https://zuul.opendev.org/t/openstack/build/82d072a0d4ff4d8f92e89a4eaf4949da : SUCCESS in 4m 40s\n- octavia-tox-functional-py39-tips https://zuul.opendev.org/t/openstack/build/3f3ac50d6635471aa117888dd8f8457d : SUCCESS in 9m 11s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/a39cf2e129b24b10b83edba54cae130a : SUCCESS in 7m 07s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/d157e4d30df34347a22ac645b922381a : SUCCESS in 7m 55s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/68b48546b7a94109875fa682ed29e43e : SUCCESS in 7m 54s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"34d420257ab86fdcdc59d7300148aba7c298e9a1","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2023-04-28 09:33:57.000000000","message":"Uploaded patch set 4.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":4},{"id":"94f440b2a8c661e3e70afb3a54636b667c461b75","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-28 09:49:02.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9f7d65f949ca4772b5db232136635ead\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/249d9b5d6e7c4d8b833afd71335f3a1f : SUCCESS in 9m 55s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/85b6ad5f01d244dea39810f72a69bcbb : SUCCESS in 5m 26s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/84dada156fe94a7084008095fe8e9dd2 : SUCCESS in 5m 31s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/c596c68d3ff04e6b9fc89cb761a31978 : SUCCESS in 4m 54s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9d4310ce12ff4406877fca526edf2d4c : SUCCESS in 10m 06s\n- octavia-tox-py39-tips https://zuul.opendev.org/t/openstack/build/e1f68d783dd1488795adbb41bfe17edf : SUCCESS in 5m 52s\n- octavia-tox-functional-py39-tips https://zuul.opendev.org/t/openstack/build/a7c4e950a21c4fdf931c33ac8ed9af24 : SUCCESS in 9m 18s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/3232f49e8a5c405c80c77b49d1e9a052 : SUCCESS in 9m 11s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/c4e33666fa7f40919c89f37251168700 : SUCCESS in 8m 01s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/74b0dc051f3d4886be45793a2e09dccf : SUCCESS in 7m 23s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"72b2a452fd0aed0192e7d4375717fccac6938b3f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2023-05-03 16:47:33.000000000","message":"Uploaded patch set 5: Patch Set 4 was rebased.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":5},{"id":"cbc12deb6e750aac5059ace58aa4b3367bcd5389","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-03 17:02:41.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/785d6525f863488db88d6db3d8e22672\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/534f9d5ceb8343c989f23fc4b5bcc8c6 : FAILURE in 5m 10s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/73203d1fb4f248a196d39a7ad8a8e8ed : SUCCESS in 5m 16s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/817495daee1044eb9a3f621d07881dbb : SUCCESS in 5m 20s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/ff10f9ae5f894ce0b9545796a45583ad : SUCCESS in 4m 25s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3d701993f3054097abe24a1393381749 : SUCCESS in 13m 28s\n- octavia-tox-py39-tips https://zuul.opendev.org/t/openstack/build/969b96bb66744f1cbf9c32b700938dd2 : SUCCESS in 5m 31s\n- octavia-tox-functional-py39-tips https://zuul.opendev.org/t/openstack/build/343605b3753647be9e20f6de82e8fa47 : FAILURE in 4m 27s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/cb6855bc6201416ab5503d562074eac4 : FAILURE in 4m 44s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/5e09fdf3b30c4bc1a2a5b62edb88c101 : FAILURE in 4m 06s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/9b64c79c84034634942fb67ff9933a94 : FAILURE in 3m 55s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"8a0456f21b2de2b9a928ee359cadb3aece73b0e7","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2023-05-04 06:07:29.000000000","message":"Uploaded patch set 6.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":6},{"id":"6c6bab84b28434ddc60ed03aa63fe768b4593366","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-04 08:49:40.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/28ee12e5514b4a69acca6844a1a1970d\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/c3996918046f43c2810fac26ab0a2c72 : SUCCESS in 5m 07s\n- requirements-check https://zuul.opendev.org/t/openstack/build/8e714b373d204627a999dc92ca8e747c : SUCCESS in 1m 58s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/6d50477da973443ea58473b3c6303998 : SUCCESS in 10m 31s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/516312d13872464ba83ad160f8836f6c : SUCCESS in 5m 14s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/248a8c714a2b43239b7a3713780927ac : SUCCESS in 4m 28s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/10cb6675b76540548932b0be491bbe70 : SUCCESS in 4m 51s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/01678086a6fd4df39d973b24ed815ee4 : SUCCESS in 10m 56s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/0eb06f41b577445b97fd82c4fb39a5ea : SUCCESS in 4m 27s\n- octavia-tox-py39-tips https://zuul.opendev.org/t/openstack/build/dba2bc33da1848e790a8d26c43da203b : SUCCESS in 5m 35s\n- octavia-tox-functional-py39-tips https://zuul.opendev.org/t/openstack/build/94f5bf80f2164813816193005bebc8e4 : SUCCESS in 7m 23s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/05cc08b144e54616935379da78846302 : SUCCESS in 5m 17s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/add682897f9e4dd8bce1cc1b7dddf12d : SUCCESS in 6m 51s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/1792ca9dc6484efea09fcb4fcd439ea9 : SUCCESS in 7m 32s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/65ef725aa0664e72bafbb57fed10cf7f : SUCCESS in 7m 20s (non-voting)\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/ae68f36114274e2aa8ed3824bd3c3a56 : SUCCESS in 1h 57m 09s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/60bd6db3e2584173bd611c64aed2e343 : SUCCESS in 2h 11m 45s\n- octavia-v2-dsvm-scenario-jobboard https://zuul.opendev.org/t/openstack/build/5de5185c68c447a382131de188eb350f : SUCCESS in 1h 58m 47s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/ed21dee6b9644a74b783b280f206b7bc : SUCCESS in 39m 48s\n- octavia-v2-dsvm-tls-barbican-fips https://zuul.opendev.org/t/openstack/build/183ce49894214f2482a243fdee4ee23f : SUCCESS in 1h 01m 15s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/dc744983ca2046b6b17d013bce398134 : SUCCESS in 32m 46s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/0f27f8a346d1475db9c430b72700037b : SUCCESS in 2h 30m 10s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/b4df059396ed4847ad1d7c89dabe4ae1 : SUCCESS in 2h 12m 59s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/30079ccd1fc74b85b485770c0a04470f : SUCCESS in 2h 28m 57s (non-voting)\n- octavia-v2-dsvm-scenario-centos-9-stream https://zuul.opendev.org/t/openstack/build/65ad47bf57414dfaa6febf9923e17bc9 : SUCCESS in 2h 07m 16s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"fb4e80b678aa6fa9475b99b0265cacad6fd5a147","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2023-05-05 06:29:28.000000000","message":"Uploaded patch set 7.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":7},{"id":"519a46b02f413bfc214484ad7a9f33d0a70eafd1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-05 06:47:49.000000000","message":"Patch Set 7: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9c54781717e74886862ffbe93babc8fd\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/65da7f9baaf54933b86bf8610135e455 : SUCCESS in 10m 47s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/047d135b7cdb4ff09426a1dad10723f6 : SUCCESS in 4m 36s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/66494d203a0649899dbd5021a5d2b38e : SUCCESS in 5m 51s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/14add097eb924100a1e26d0c5fa9df6a : SUCCESS in 4m 17s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7f74c76bc24547358a2c34db52db78da : SUCCESS in 11m 07s\n- octavia-tox-py39-tips https://zuul.opendev.org/t/openstack/build/487ce72108c44969b8d2639bfa956e2f : SUCCESS in 5m 01s\n- octavia-tox-functional-py39-tips https://zuul.opendev.org/t/openstack/build/8fd7adcbe9a64818979e87c182428723 : SUCCESS in 7m 13s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/045642c70a3d4464aa71f0f332b37e8f : SUCCESS in 6m 50s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/0fca5bc81dd44c5f81849212c8ca54b8 : SUCCESS in 9m 11s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/314eaf38d33f42c9bd282296c28f0c6c : SUCCESS in 8m 03s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"0412dab8152bd9909c5233b2b1574dac2277937e","author":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"date":"2023-05-30 13:59:00.000000000","message":"Patch Set 7:\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"be0398374a96c30d35e509fb68736889feb756de","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2023-05-30 16:28:29.000000000","message":"Uploaded patch set 8.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":8},{"id":"a249a89a31395e7d18c91114459c06d85616cd20","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2023-05-30 16:28:44.000000000","message":"Patch Set 7:\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"940379a73a6fdf8307658099c561c494fd1bc54a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-30 16:51:10.000000000","message":"Patch Set 8: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/225cfbb1d81e45aeacd01830ab200d5a\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/03f5da7d549942de95dceca92e89e181 : SUCCESS in 9m 39s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4d4a8c91faab4286986995e1804b3175 : SUCCESS in 5m 54s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/941fb7f3fc454b20a6ba6b1b55933a67 : SUCCESS in 6m 56s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/e80effa2a9d448ad81e2b95a3db8e955 : SUCCESS in 4m 45s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/f45b6855ba774ee7ac92c32152ed74c9 : SUCCESS in 4m 45s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c065a40870a64fedb0634f0944a9331f : SUCCESS in 11m 46s\n- octavia-tox-py39-tips https://zuul.opendev.org/t/openstack/build/e06695fdd43b4d1392592eb4f101325e : SUCCESS in 5m 48s\n- octavia-tox-functional-py39-tips https://zuul.opendev.org/t/openstack/build/13d6a40107334d309ab6ad737126a4c6 : SUCCESS in 8m 21s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/7d6469e3ff934ca39b6afbd0033f1fbf : SUCCESS in 8m 01s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/ee6c5274fb38404ba486147739f7eaec : SUCCESS in 10m 01s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/0646d23230164b2ab12646a921199b9c : SUCCESS in 8m 32s (non-voting)","accounts_in_message":[],"_revision_number":8},{"id":"3eb71f735950cb3720e53df1616619f59322a9e7","author":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"date":"2023-05-31 08:37:49.000000000","message":"Patch Set 8: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":8},{"id":"6203bb02b872b558528c0ecae52fb15265949b37","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2023-08-23 16:59:59.000000000","message":"Uploaded patch set 9: New patch set was added with same tree, parent tree, and commit message as Patch Set 8.\n\nCopied Votes:\n* Code-Review+1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":9},{"id":"d94ac47456575669f2057e8d4b919900fc6f30c4","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2023-08-23 17:08:07.000000000","message":"Patch Set 9: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":9},{"id":"7b55d04eaca75a2effcb289713d0426155f92879","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-23 17:16:38.000000000","message":"Patch Set 9: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9a263e6774644849a302a41dbe21a5f6\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/42152c3261a543d6b53f1791204b3f9b : SUCCESS in 11m 17s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d63d5fa411ab4ddd84ea7f87ae22b4d7 : SUCCESS in 5m 24s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/7a1c505d35f0488f9fc4a733c36969fe : SUCCESS in 4m 16s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/90ccf60e42d34c949f94831b6b42f451 : SUCCESS in 5m 15s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/9777052d18004cb983b6e6b34e8c5670 : SUCCESS in 4m 16s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/f52cd7657b254c17ac73b5d48d4444ce : SUCCESS in 4m 45s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b4c869259d8f4794a63fdaa60547ea63 : SUCCESS in 9m 59s\n- octavia-tox-py39-tips https://zuul.opendev.org/t/openstack/build/b4148b3e55b142e38711cf85e856376b : SUCCESS in 5m 16s\n- octavia-tox-functional-py39-tips https://zuul.opendev.org/t/openstack/build/6021707751cc420d921559e1fedc085b : SUCCESS in 7m 57s\n- octavia-tox-functional-py39-sqlalchemy-tips https://zuul.opendev.org/t/openstack/build/b93fa27dfae947aba59b9d2f1370234f : SUCCESS in 8m 10s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/2d667e44bea3439185abbf2304737545 : SUCCESS in 6m 45s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/815147893459408a957361537e22629d : SUCCESS in 8m 38s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/85dd1c11752d46e492430840445d7c1c : SUCCESS in 7m 11s (non-voting)","accounts_in_message":[],"_revision_number":9},{"id":"b96ab34131ff359eaa6d1c006940de79c2555446","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2023-08-24 07:52:33.000000000","message":"Patch Set 9:\n\n(1 comment)","accounts_in_message":[],"_revision_number":9},{"id":"4d55b54124a3005ab6dfc29ddc0b78b69930ee2f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2024-05-21 11:25:33.000000000","message":"Uploaded patch set 10.\n\nOutdated Votes:\n* Code-Review+1, Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":10},{"id":"ce655668e1d8374d516556c3f386782387d9df69","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-21 11:50:29.000000000","message":"Patch Set 10: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/9633633354ed4ff7907a7a9c15baa401\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/cddf8665dc4a4235988a3ad6d51cfeee : FAILURE in 5m 23s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5596ceb599ee41279f401b3eca6c9ea5 : SUCCESS in 5m 05s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/c4dda45dcf8544d49f2d0aa18929299a : FAILURE in 4m 55s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/e68575854d03439abe5944e76ac885f0 : FAILURE in 5m 02s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/98afc084666e4b7f9b90f7cfac5f4846 : FAILURE in 10m 49s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9fe9080b36d44b67b83c9f4c7913ff5b : SUCCESS in 14m 12s\n- octavia-tox-py311-tips https://zuul.opendev.org/t/openstack/build/9340d4b3c5224ec19dbc708b1bcc6b0d : FAILURE in 4m 58s\n- octavia-tox-functional-py311-tips https://zuul.opendev.org/t/openstack/build/6f453b3828e042ce8aec7f77c82d4c87 : FAILURE in 5m 30s\n- octavia-tox-functional-py311-sqlalchemy-tips https://zuul.opendev.org/t/openstack/build/6c910cc2179547729e2bffaaf7a3659b : FAILURE in 5m 44s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/262cad10ff104e2bbe1a8efa6c9fcb0c : SUCCESS in 8m 29s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/f1bc86fd859a402c876d3fe7df756d94 : SUCCESS in 7m 42s\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/9c5fea3a5d0a43f9936110dc0815555a : FAILURE in 4m 53s","accounts_in_message":[],"_revision_number":10},{"id":"f6cd1b86b954e637fcf18b085ea4b2615c06d1e4","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2024-05-21 13:56:31.000000000","message":"Uploaded patch set 11: Patch Set 10 was rebased.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":11},{"id":"eff242a5ffe6ee79f67f3074f2a1d7a45b49731e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-21 14:20:33.000000000","message":"Patch Set 11: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/03d601ad58504d48968a9e67bb87f823\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/7602fedf838c4909a7d5cdf90bd1dae2 : SUCCESS in 11m 39s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f1b3401bc701454d9c85bcda0bdbfbc0 : SUCCESS in 4m 04s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/ac7baed2e3a1457dbbc877d4e9edb3a7 : SUCCESS in 4m 46s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/966742d961dd416585b863d09a76a839 : SUCCESS in 6m 26s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/14095b532cbc49849b1e4e274283f2e5 : SUCCESS in 8m 48s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/922012ce4d07472084da2717620d0dfe : SUCCESS in 11m 33s\n- octavia-tox-py311-tips https://zuul.opendev.org/t/openstack/build/0908af7115054014aea9332dae4f3a34 : SUCCESS in 5m 27s\n- octavia-tox-functional-py311-tips https://zuul.opendev.org/t/openstack/build/0952439b1cd14dcda3f3598ae2959e3b : SUCCESS in 5m 26s\n- octavia-tox-functional-py311-sqlalchemy-tips https://zuul.opendev.org/t/openstack/build/b84d76c4e7e64d158a230ddf8d667325 : SUCCESS in 6m 58s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/5a16869b792447edafb9980fafa79c38 : SUCCESS in 8m 07s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/e56e533e6349479898981d171a65fbd5 : SUCCESS in 9m 40s\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/14ebbf3bde014453a2c505131f5243ff : SUCCESS in 5m 02s","accounts_in_message":[],"_revision_number":11},{"id":"151be16f543114bae2ffb7819d98f90c3e4ba0fc","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2024-05-21 15:43:13.000000000","message":"Uploaded patch set 12: Patch Set 11 was rebased.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":12},{"id":"a57bea9a1b50da928ed3876b490c9fde9b2c822c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-21 16:01:21.000000000","message":"Patch Set 12: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/08123c261a42490dad77011f0349012a\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/82f6c5a09f9d4dff8db37177838125e2 : SUCCESS in 8m 26s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3d5e73a0c0f64484b7082830d42f3014 : SUCCESS in 4m 37s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/183a4892157940c58e22de285497a902 : SUCCESS in 5m 17s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/f3cac411784146bfb6d88954dce1906c : SUCCESS in 4m 47s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/95edb9bfd0fe4a969214b7c2a014d57e : SUCCESS in 9m 27s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/81b63f459531449ea65cfdc8485020ac : SUCCESS in 11m 57s\n- octavia-tox-py311-tips https://zuul.opendev.org/t/openstack/build/45b1caf370584e81b28b81d53dae11ec : SUCCESS in 5m 26s\n- octavia-tox-functional-py311-tips https://zuul.opendev.org/t/openstack/build/27be1c00a5ca4d3aa89d751c13a7d17b : SUCCESS in 6m 23s\n- octavia-tox-functional-py311-sqlalchemy-tips https://zuul.opendev.org/t/openstack/build/2db350c197f74102a7b934db592f87b1 : SUCCESS in 6m 24s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/3f75c8f19d084de988b830ac66a3eb10 : SUCCESS in 6m 49s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/ed42ad2907c64675b9724e110f759f6b : SUCCESS in 9m 09s\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/44bcfc1ed18946788fb49da607697d96 : SUCCESS in 5m 07s","accounts_in_message":[],"_revision_number":12},{"id":"186426a1b752a840937426fccd2ec4d3d1caf7e5","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2024-05-21 16:06:42.000000000","message":"Uploaded patch set 13: Patch Set 12 was rebased.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":13},{"id":"3629d9d3c0a744958387fb50f91a879c1d0104ab","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-21 16:25:44.000000000","message":"Patch Set 13: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a0869cb79f6d4f5bb255504a93fc198f\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/bb656ff7677f48528d40caa3901a0d5e : SUCCESS in 12m 00s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ccb635803c1f474391b015e082003039 : SUCCESS in 6m 40s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/aca7a93a4b904251b26c869d2bbaf5dd : SUCCESS in 4m 26s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/61a0d83874d640a29e65022b0fbc7a40 : SUCCESS in 5m 28s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/d2f9d2a1cc8e49f0a0f9be680a802b00 : SUCCESS in 8m 20s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/68c4bb45e6264cf7a94eace283ad1598 : SUCCESS in 10m 39s\n- octavia-tox-py311-tips https://zuul.opendev.org/t/openstack/build/e8f4508c4de1466081f0b3804179f422 : SUCCESS in 5m 16s\n- octavia-tox-functional-py311-tips https://zuul.opendev.org/t/openstack/build/bd4b8dc8d1f04bdca32bcf8d60dc64f9 : SUCCESS in 5m 05s\n- octavia-tox-functional-py311-sqlalchemy-tips https://zuul.opendev.org/t/openstack/build/89ab7be9795b4be3b97c17716bbf2abd : SUCCESS in 6m 49s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/073526e6ad5048a599d67db6872ad960 : SUCCESS in 7m 22s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/81bf89e10e514ad9843e65528a2ace7e : SUCCESS in 7m 54s\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/7d0f5980c3b54e1bb07b7138a5ea99a6 : SUCCESS in 5m 33s","accounts_in_message":[],"_revision_number":13},{"id":"0e8a2f8a6e3483759947c17b02da7da5cee51458","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2024-05-22 17:06:25.000000000","message":"Uploaded patch set 14: Patch Set 13 was rebased.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":14},{"id":"6bea31435925210ad670994cfceb8a55b6a2b6ec","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-22 17:25:05.000000000","message":"Patch Set 14: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b45b8e1d87264800bd4ac7469f1aec77\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/26b992ed57634ad3b84b9bba9aa4ede3 : SUCCESS in 12m 36s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d49ec568e7d044b69a2b1d17492ceb8e : SUCCESS in 4m 55s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/ef6a064f60234ff6be809e9e13fbda08 : SUCCESS in 7m 02s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/6f2c6f72c41942de92f0dc479cf74d04 : SUCCESS in 4m 22s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/c0e7e1263aa24936b05907bd1804b8fa : SUCCESS in 10m 19s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7a678af6350d4b11a89e311455ca8dfe : SUCCESS in 13m 16s\n- octavia-tox-py311-tips https://zuul.opendev.org/t/openstack/build/12a1a3507f0843c9861dc0121deb9788 : SUCCESS in 4m 35s\n- octavia-tox-functional-py311-tips https://zuul.opendev.org/t/openstack/build/e361446c6cd34fa695fcc151ddebe22b : SUCCESS in 6m 59s\n- octavia-tox-functional-py311-sqlalchemy-tips https://zuul.opendev.org/t/openstack/build/492adb588fc941c5af3d6191842221c1 : SUCCESS in 8m 04s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/ee798e45d23c47b3a7f2d1f1d6b92903 : SUCCESS in 8m 12s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/c8d9240b086b4593bd0693e597e47e81 : SUCCESS in 8m 11s\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/1255fb22e18548bbbcaebbfaeebdf431 : SUCCESS in 5m 06s","accounts_in_message":[],"_revision_number":14},{"id":"a7308b710ed6877ea761a1ecace2a5bb254ea01e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2024-06-03 08:15:30.000000000","message":"Uploaded patch set 15: Patch Set 14 was rebased.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":15},{"id":"d8de7515c8b357125d306f1da02e6d5026fc3c59","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-06-03 08:36:14.000000000","message":"Patch Set 15: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5a93b3b7162a410ab47a700abd251fb5\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/2d31ba34179845f8a5591b0d7626de05 : SUCCESS in 11m 36s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/383f294814c2408cbe6ef5cd0d366d64 : SUCCESS in 4m 42s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/b0b660c79527491993ccf5e9ebbd75e8 : SUCCESS in 4m 54s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/255633c7e3284d0ca7416f47d7a61320 : SUCCESS in 4m 34s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/25a194cc36d344e6be36ae89db33fa3c : SUCCESS in 8m 05s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/75896ead12a24d9d8cb651864660c674 : SUCCESS in 11m 32s\n- octavia-tox-py311-tips https://zuul.opendev.org/t/openstack/build/284d7376bb9c453ca24080fb79105788 : SUCCESS in 4m 24s\n- octavia-tox-functional-py311-tips https://zuul.opendev.org/t/openstack/build/f34bd2b69ac94e0f8d59619da7f0753c : SUCCESS in 4m 45s\n- octavia-tox-functional-py311-sqlalchemy-tips https://zuul.opendev.org/t/openstack/build/874b22ec21b543039e8b95a629306256 : SUCCESS in 5m 43s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/268cd763dcf04bfeb8391f2d36e6819c : SUCCESS in 8m 32s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/117ee21a916c4d5b995340b2f79c7297 : SUCCESS in 7m 26s\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/9b5530ca333e42af9ded714e1cc46a84 : SUCCESS in 4m 52s","accounts_in_message":[],"_revision_number":15},{"id":"aae7f45d8b5c5a21024cc3846df0ac6fc1c1615c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2024-06-03 16:03:01.000000000","message":"Uploaded patch set 16: Patch Set 15 was rebased.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":16},{"id":"902d1a65877bd912488a0228c64d69328b87d67d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-06-03 16:30:30.000000000","message":"Patch Set 16: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/55f65aaf7a8545c78696c78f3b842f5e\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/2660e0ea4b61474abc8495df251fc813 : SUCCESS in 12m 27s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/908a91ea237040a5825db8da1f0c87a7 : SUCCESS in 6m 04s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/d1dc404d29854bc68334b2c037cc792c : SUCCESS in 6m 44s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/509bee18a1ea4504bc07e28da96672b0 : SUCCESS in 7m 19s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/4b2339f8657f452d9ced9fccf2f6cde2 : SUCCESS in 8m 40s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e94efb1f8e55496389d00b364b78c1e7 : SUCCESS in 11m 40s\n- octavia-tox-py311-tips https://zuul.opendev.org/t/openstack/build/96853d09f8c54ec79a439d703121fa4f : SUCCESS in 4m 42s\n- octavia-tox-functional-py311-tips https://zuul.opendev.org/t/openstack/build/c5384831e4424700b7a8bd7cb078e979 : SUCCESS in 6m 04s\n- octavia-tox-functional-py311-sqlalchemy-tips https://zuul.opendev.org/t/openstack/build/7b234cc148b34a409f1c2675629d9b4f : SUCCESS in 6m 39s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/8cc5fd45da3549fab7a6761f8c899995 : SUCCESS in 9m 15s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/843bab506eae46ed85595fab2a28f2b1 : SUCCESS in 6m 56s\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/0b9af22b926444429f71aad6b6fc59e7 : SUCCESS in 5m 25s","accounts_in_message":[],"_revision_number":16},{"id":"78d35d0748ef33e9a10dbd17dc158465dad4ea6f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2024-06-05 10:27:46.000000000","message":"Uploaded patch set 17: Patch Set 16 was rebased.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":17},{"id":"a9840a20c93832291c7397d65684feb7e417e137","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-06-05 10:45:35.000000000","message":"Patch Set 17: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/b5b60e73cd914e20a89c90369a40ee79\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/c31176919db84746a6a963f7dad8b28f : FAILURE in 5m 46s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/776e37b96f5944a69b478c93a378b431 : SUCCESS in 5m 26s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/ebcf8d0dcb9f499c99c64ee6fe3ce7c8 : FAILURE in 5m 55s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/e0a3220d614f42f7a9fb7d438b66a0e7 : FAILURE in 5m 44s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/1547926a92384d5e8fc1aeb625db7088 : FAILURE in 12m 00s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d6748c5355334dd7b4918b5126a1f9f4 : SUCCESS in 11m 44s\n- octavia-tox-py311-tips https://zuul.opendev.org/t/openstack/build/30e7d6107bd040a585e90bb5f60af7cf : FAILURE in 6m 10s\n- octavia-tox-functional-py311-tips https://zuul.opendev.org/t/openstack/build/412dfaf7786246f7b2c23fa37d67f5ef : FAILURE in 4m 45s\n- octavia-tox-functional-py311-sqlalchemy-tips https://zuul.opendev.org/t/openstack/build/62220f14cb4c46559b033c5afceea28a : FAILURE in 6m 52s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/9818ad3836c34e7793c7ed318383e1b7 : SUCCESS in 9m 32s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/83e7ec0ed8d343ac99615aa02b14098e : SUCCESS in 8m 46s\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/9058371d6e7a419eb3d142a2d1c4cbad : FAILURE in 5m 21s","accounts_in_message":[],"_revision_number":17},{"id":"0636fba542f163a1229777177c022334ddf9e5b6","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2024-06-05 10:46:20.000000000","message":"Uploaded patch set 18: Patch Set 17 was rebased.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":18},{"id":"52cf7a5565037d58c7e7b4538da1a389257b378a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-06-05 10:59:43.000000000","message":"Patch Set 18: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/5b3a06dbfef743ceaa619f630c9730d1\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/95a6df146ed74c76a3022a6a6e3cc66e : FAILURE in 5m 24s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/97d9de62340f4eda9b29f82b3d0667f6 : SUCCESS in 4m 35s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/2689f98e24d34f9daabdd46f2b8f9344 : FAILURE in 5m 27s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/eef39a9a16444b14b5f50710055a33cc : FAILURE in 5m 22s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/0c8f1bd0dd3b4e0e821c04e11354aad2 : FAILURE in 8m 05s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b6df4d9b5dda490798369603ba4328ce : SUCCESS in 11m 22s\n- octavia-tox-py311-tips https://zuul.opendev.org/t/openstack/build/8db7bb840bd546d2ad42d1659384eba1 : FAILURE in 6m 43s\n- octavia-tox-functional-py311-tips https://zuul.opendev.org/t/openstack/build/7203db2b36274ea7aef17a1d57a1167d : FAILURE in 5m 31s\n- octavia-tox-functional-py311-sqlalchemy-tips https://zuul.opendev.org/t/openstack/build/59abc57d909d42bf86f05be207a0f381 : FAILURE in 6m 29s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/5c590341fd1947c695366f5d5dd2f57e : SUCCESS in 7m 34s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/0f96e027173048e18d435b469caeb1c3 : SUCCESS in 7m 45s\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/b8ca0fa4ac2c4b5ca43bffb04c7b77fe : FAILURE in 4m 46s","accounts_in_message":[],"_revision_number":18},{"id":"b09beeb19b80d8b5e0cdf19c6a2f08d009f3d673","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2024-06-05 11:12:54.000000000","message":"Uploaded patch set 19: Patch Set 18 was rebased.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":19},{"id":"822ad84ba105ee8313323b9143308f83cf570265","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-06-05 11:29:37.000000000","message":"Patch Set 19: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ec652720f52a4f9c8aa5410bf4f4a417\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/59a096d014864c96b2a637116be0e54f : SUCCESS in 12m 17s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9febbd7ab5b14434888eaff35c349b4c : SUCCESS in 5m 05s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/e7546cf2d6264459bcedb3183552f124 : SUCCESS in 5m 04s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/f1e383f3ffe8496cbbe165824179b0dc : SUCCESS in 5m 25s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/804c5d8258434f5788acdc70672509a2 : SUCCESS in 8m 40s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1db9f16c6a784ed382faa2dfcf303c5f : SUCCESS in 11m 15s\n- octavia-tox-py311-tips https://zuul.opendev.org/t/openstack/build/d23646a3e4a54821a6af72dd88b17e36 : SUCCESS in 5m 29s\n- octavia-tox-functional-py311-tips https://zuul.opendev.org/t/openstack/build/4f21bc64fc1b4947b417824f7537399e : SUCCESS in 5m 13s\n- octavia-tox-functional-py311-sqlalchemy-tips https://zuul.opendev.org/t/openstack/build/a3a69b5b16dc4b0aa82d15cb8cf652d3 : SUCCESS in 6m 23s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/32706c04adbd4ddd9e96d3465411d817 : SUCCESS in 9m 29s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/fc0bc97d21cd43a2854f640f7036e4d3 : SUCCESS in 9m 37s\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/6c2f2a0a65a64455b23275a65b0a4168 : SUCCESS in 6m 07s","accounts_in_message":[],"_revision_number":19},{"id":"2982f6f8bd80b2f5c48bb0c522dafa92d5dc939d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2024-06-11 07:25:07.000000000","message":"Uploaded patch set 20: Patch Set 19 was rebased.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":20},{"id":"130d30a8e9e387f9e8a8f029e4a92fe980dec7d6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-06-11 08:05:22.000000000","message":"Patch Set 20: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/69a749ef728e4f9ea16b0f82a4fc7090\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/e6f8ab99d0444838912ff275107441ce : SUCCESS in 10m 17s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c260d73a671a4320a0710df1c8128953 : SUCCESS in 4m 35s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/bbbbc66b88db45c4807b086feabc7e3a : SUCCESS in 7m 35s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/390e5cae9a784ded9d3cd26304004e50 : SUCCESS in 5m 13s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/45e5ec00e9f644f08d8b7c30e47d525a : SUCCESS in 8m 20s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/23127fdd65394667b585941cdc7e2ddd : SUCCESS in 11m 18s\n- octavia-tox-py311-tips https://zuul.opendev.org/t/openstack/build/3c30d141e53c43f4910218fca74f98bd : SUCCESS in 5m 04s\n- octavia-tox-functional-py311-tips https://zuul.opendev.org/t/openstack/build/162936c237e34e928f87e830dc8549c7 : SUCCESS in 4m 59s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/a5bd5258088841d1ab0d1aca0d87780c : SUCCESS in 7m 52s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/5e65a4959c3c43fd97b4e105cbfed7da : SUCCESS in 7m 36s\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/76738fc55c1f4a7e9b11735c61a3b738 : SUCCESS in 4m 59s","accounts_in_message":[],"_revision_number":20},{"id":"7d7141cc3f315dbaa2fa8d99305ab33947b565e4","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2024-10-22 14:59:30.000000000","message":"Uploaded patch set 21: Patch Set 20 was rebased.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":21},{"id":"e9d9bed42259fbf8cb3cc6e4232b6869bcb3a4d0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-22 15:20:59.000000000","message":"Patch Set 21: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/cd82f26388964c8590f5e56e2022868b\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/467f78792b784d788aace2bdaf8e281e : SUCCESS in 12m 13s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d552ed3befab4c098d1fbbae1af2ed4e : SUCCESS in 4m 48s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/41e43c1ced6140d5a419726e792efd41 : SUCCESS in 4m 51s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/dc505edf3fa941a199359d662fca67fb : SUCCESS in 5m 51s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1fc704517da54d0e8830ff135ba19393 : SUCCESS in 10m 49s\n- octavia-tox-py311-tips https://zuul.opendev.org/t/openstack/build/58d513fb32424120bfb990a68eb3900b : SUCCESS in 4m 20s\n- octavia-tox-functional-py311-tips https://zuul.opendev.org/t/openstack/build/52ba83876d4c49368d014bb431072c7e : SUCCESS in 6m 03s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/775a861bde1c44aca8bb59e6f7bbb9b4 : SUCCESS in 6m 50s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/891a8ef2d79f4cf5a7fb489fef1d6d12 : SUCCESS in 7m 34s\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/91181dd7dd88420dbc1dea3b676abc00 : SUCCESS in 6m 53s","accounts_in_message":[],"_revision_number":21},{"id":"92197c3a8958552b2a8e14ab172d04727c7edd4e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2024-10-23 10:21:48.000000000","message":"Uploaded patch set 22: Patch Set 21 was rebased.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":22},{"id":"ca539cd272bd9b61bc623c07498249ba288de1ad","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-23 10:39:53.000000000","message":"Patch Set 22: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/804c2aab887c40d19a7c11965c48b511\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/3bbee422afbe4903a9c336e60dd87649 : SUCCESS in 10m 25s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d3b646f3e99349cd9458584b78110697 : SUCCESS in 4m 42s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/a9ee1bf111564cd69e7b03ff30eb9a17 : SUCCESS in 4m 30s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/b2407bc0c7ad4466a3f246a237cd2b74 : SUCCESS in 5m 59s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1975c2735046421097e6aca4f1484493 : SUCCESS in 10m 53s\n- octavia-tox-py311-tips https://zuul.opendev.org/t/openstack/build/cf59cf193aa34fefa1f25409ccdaa84f : SUCCESS in 4m 43s\n- octavia-tox-functional-py311-tips https://zuul.opendev.org/t/openstack/build/3d661529a1144fc29b08dcf3a0d55e9c : SUCCESS in 7m 30s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/985f0411421545189fb4cbcdb9e2802a : SUCCESS in 6m 26s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/f5cfe18a82184550846e71c5cb694704 : SUCCESS in 7m 51s\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/83e5c6829913456298e9163264fd5c8e : SUCCESS in 6m 22s","accounts_in_message":[],"_revision_number":22},{"id":"7926fa0dd1d71a1f0992f66e8c148d8cedd8802c","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2024-10-23 14:25:54.000000000","message":"Topic rate-limiting removed","accounts_in_message":[],"_revision_number":22},{"id":"7b6881376eb81b2080db1e15769783e881bbab06","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2024-10-28 11:14:50.000000000","message":"Uploaded patch set 23: Patch Set 22 was rebased.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":23},{"id":"53ed085a16f4464f72bdcb8e7d3dcfdb7eabd228","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-28 11:35:55.000000000","message":"Patch Set 23: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/797e3876519b46e19140d8ebc4148118\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/dd8145d561c442a8bfd76d2c10ba6f7d : FAILURE in 4m 08s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7edb1973485b469a91d2c71a69e0e9bd : SUCCESS in 4m 26s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/1faa808e905548ebbab327f437145914 : FAILURE in 4m 42s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/43a729c7d844457c9138a00f78ed5264 : FAILURE in 6m 30s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/53e98ef7906a400480a0f4a034f9d38e : SUCCESS in 10m 59s\n- octavia-tox-py311-tips https://zuul.opendev.org/t/openstack/build/83ae0c1624db40f1a1db892edb8410ab : FAILURE in 4m 51s\n- octavia-tox-functional-py311-tips https://zuul.opendev.org/t/openstack/build/42281b79d7254398bedb50eba33ec8d3 : SUCCESS in 6m 32s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/6b3b3d1106e442de8df33b2360f3f067 : SUCCESS in 7m 44s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/877c31cc64ee43478babf80356c328dc : SUCCESS in 9m 27s\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/16467f8766004f32b31696050901edd0 : SUCCESS in 6m 41s","accounts_in_message":[],"_revision_number":23},{"id":"3e0454ee076ebcf990f3bda4a89bd4880dccc27d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2024-10-28 11:45:44.000000000","message":"Uploaded patch set 24: Patch Set 23 was rebased.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":24},{"id":"82a58a268457bdf0d8251ebcee76fabc787d24a8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-28 12:10:11.000000000","message":"Patch Set 24: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/6a30f7f39aca4466ac9b57f9e5039c29\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/d78b5e2a9275410683081ff2b020da30 : FAILURE in 4m 38s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ba86eacd8017436b99a3c05fe2a6f7f0 : SUCCESS in 4m 41s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/9e4edc6b69844d5c89f89247e226f8d5 : FAILURE in 4m 26s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/c666836c9f5e412b89745556e3cb62f3 : FAILURE in 4m 02s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/68bb1ebbad4e41fb9ded111c3626ec10 : SUCCESS in 10m 44s\n- octavia-tox-py311-tips https://zuul.opendev.org/t/openstack/build/d6dcc10f795b4c018c31dcbc94db1857 : FAILURE in 4m 35s\n- octavia-tox-functional-py311-tips https://zuul.opendev.org/t/openstack/build/08676b44a02c4df79c41f83b4b6b5d95 : SUCCESS in 6m 30s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/d2303bdd9516436d9ad90253083bfaf6 : SUCCESS in 7m 08s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/91f394847715438a98a6eda941b5f30b : SUCCESS in 6m 30s\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/ee7a18ef087d4466825cb95b4a2970a8 : SUCCESS in 7m 52s","accounts_in_message":[],"_revision_number":24},{"id":"3edb13ade146d7f18f2d6335e76e23d86b5af981","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2024-10-28 13:21:24.000000000","message":"Uploaded patch set 25: Patch Set 24 was rebased.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":25},{"id":"94d4c235048cdd07e9b9ad27cf88cb60a4af35b5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-28 14:07:44.000000000","message":"Patch Set 25: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/bbf38a22e1b5461182c78a5937e4c732\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/2d5ae3b88e094a84bac7ff103c7ea0c5 : SUCCESS in 10m 00s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8313acf65a9c48e2af5ed6900adf88f0 : SUCCESS in 4m 34s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/94995b46d6d64687a97e53410d9661e8 : SUCCESS in 5m 57s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/09ca304a56a2457e821248fe7d46bb5f : SUCCESS in 5m 52s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2cd0e9aea1b04ad0a02219bbe5c358c4 : SUCCESS in 11m 40s\n- octavia-tox-py311-tips https://zuul.opendev.org/t/openstack/build/d4fbe303038c45d69a8a996b2d82cc86 : SUCCESS in 4m 59s\n- octavia-tox-functional-py311-tips https://zuul.opendev.org/t/openstack/build/0b21cd8fe56e475288478b0f61f49c6a : SUCCESS in 7m 50s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/8a9879d9340643c6a86076992aba7a3b : SUCCESS in 8m 21s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/78fab3654c17414787fe563420ec83e7 : SUCCESS in 6m 48s\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/12683be44b4049daa48a6caf464ed874 : SUCCESS in 5m 12s","accounts_in_message":[],"_revision_number":25},{"id":"03caab2d0d47f818eec2e2fd6a554ff0a5ca0ef9","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2024-11-20 12:07:07.000000000","message":"Uploaded patch set 26: Patch Set 25 was rebased.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":26},{"id":"8cdea6105bbd16fcb4ca8ce174072710deaf8930","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-20 12:52:09.000000000","message":"Patch Set 26: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/33875acaeef04855a3a5ad9618b21a43\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/455016e668ce4d1fbb720d425cd872c6 : SUCCESS in 10m 26s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b08987db960e4779ac8ec025d1c816cc : SUCCESS in 5m 40s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/fafc9e463def43878f17ac9ce08c4535 : SUCCESS in 4m 47s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/000cb28b6edd4ae08588c1cfa258f3f3 : SUCCESS in 6m 16s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e3d8a4d30c384b90931a96430779ccdc : SUCCESS in 10m 48s\n- octavia-tox-py311-tips https://zuul.opendev.org/t/openstack/build/2685a70a6f76426c9f40e841e49fbeec : SUCCESS in 4m 56s\n- octavia-tox-functional-py311-tips https://zuul.opendev.org/t/openstack/build/434a5b1afb2c43be9fe4a59bd6c9992e : SUCCESS in 7m 02s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/3ffcfa93c53d48e891b6d098f973b815 : SUCCESS in 7m 10s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/ce506bfb90a645dca88993b3598da4e0 : SUCCESS in 7m 06s\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/078de30c082744109d5d43ca41748785 : SUCCESS in 7m 30s","accounts_in_message":[],"_revision_number":26},{"id":"5e40d5116e7ac94961e2847a6e904637f810d1a6","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2025-02-10 15:47:30.000000000","message":"Uploaded patch set 27: Patch Set 26 was rebased.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":27},{"id":"801767e66da9618242c88be4e0d6033a35f8a84f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-10 17:10:56.000000000","message":"Patch Set 27: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d8ae8a394a90413c8750f679ca33d3e2\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/5a319820787f456b94802065715be464 : SUCCESS in 14m 31s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/36b202f4cc6c4a08b5440b712e1499f8 : SUCCESS in 5m 16s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/d366d8495b6c4f3fa612c504eb88c6bc : SUCCESS in 4m 39s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/2591e91eb14b4895a4d5f3b277381657 : SUCCESS in 5m 21s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/52228edfb406403693eb2cfcaea86a27 : SUCCESS in 10m 53s\n- octavia-tox-py311-tips https://zuul.opendev.org/t/openstack/build/0886e3f7f048480e896503ab3180b732 : SUCCESS in 5m 10s\n- octavia-tox-functional-py311-tips https://zuul.opendev.org/t/openstack/build/eb6180a6100545c8accce616c41ea65d : SUCCESS in 8m 47s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/9faa55b3fc704075aaa2a660a12079b4 : SUCCESS in 7m 51s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/6106cadeb88f449bb29521cc46447432 : SUCCESS in 7m 38s\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/7aec09ea67d24f9eb1ae153c39aa93d5 : SUCCESS in 6m 35s","accounts_in_message":[],"_revision_number":27},{"id":"47e3865facb6f731edfdd5246b50ea3da16f79e8","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2025-02-11 09:38:24.000000000","message":"Uploaded patch set 28: Patch Set 27 was rebased.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":28},{"id":"b78f21af5ccf03f759d4dd8b68c9ae8511416757","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-11 11:58:33.000000000","message":"Patch Set 28: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/abe29eb52cc740e3b50f349dd2dbd60f\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/8e19ea285ffb4ed5bd3f4583922e0f0e : SUCCESS in 18m 39s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/314582b28025475b9cc6c412ae7d78f9 : SUCCESS in 5m 10s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/3d9ea94fc82840099dd1c15239a3a4b7 : SUCCESS in 4m 27s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/c2ed651eda60453898e69ce19be48f2e : SUCCESS in 12m 14s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/44eb25135528445ca133618456ae2692 : SUCCESS in 10m 47s\n- octavia-tox-py311-tips https://zuul.opendev.org/t/openstack/build/963fe737250d4db0b56eb8c4ef3f9a4b : SUCCESS in 5m 17s\n- octavia-tox-functional-py311-tips https://zuul.opendev.org/t/openstack/build/2ff28df3276b49cdb1f97638f7e88433 : SUCCESS in 8m 09s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/0ef16e0d4e8544f7b178b5cf29ad04ed : SUCCESS in 8m 28s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/5a3c4bd2d1d646399dfe91f79bc00147 : SUCCESS in 7m 15s\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/604b2f8985534026ac8d8244ad544391 : SUCCESS in 7m 34s","accounts_in_message":[],"_revision_number":28},{"id":"b38cc38af741eec5c4f6eda771192a009c2b373c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2025-02-24 15:28:52.000000000","message":"Uploaded patch set 29: Patch Set 28 was rebased.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":29},{"id":"31adcb9676cd1c8a6ab7d8d012d6834ceef6f063","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-24 17:33:00.000000000","message":"Patch Set 29: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/d788d6c824464f89bd9a8b2db8c8b3e7\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/65b90424db2e454489f46c4124740374 : SUCCESS in 15m 09s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/62789825816b4cfcba589f4508e6c615 : SUCCESS in 4m 51s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/07bcfcaeac5e463eac89cd493f060ec9 : SUCCESS in 4m 40s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/2201e614fedb4362b329a97c0869e577 : SUCCESS in 4m 45s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/da01f49d13df421f847b9221cce244b6 : FAILURE in 5m 11s\n- octavia-tox-py311-tips https://zuul.opendev.org/t/openstack/build/6da21ec0c53e4ccd91e6989963443fd0 : SUCCESS in 5m 06s\n- octavia-tox-functional-py311-tips https://zuul.opendev.org/t/openstack/build/2d36c1fa289246129996f46e138e915d : SUCCESS in 8m 38s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/4a403f8c92324657b7de902e8c8a6d08 : SUCCESS in 6m 48s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/427933b31be14f1fb71d72688e4e94dd : SUCCESS in 7m 12s\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/c3b3f410b6d440bfb531e40d5ce6bd3a : SUCCESS in 9m 17s","accounts_in_message":[],"_revision_number":29},{"id":"8f3560a9166dc5286951ffbae82426c3b5f57f8e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2025-03-03 13:33:42.000000000","message":"Uploaded patch set 30: Patch Set 29 was rebased.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":30},{"id":"9f0be1f9237175b9b483a1b3309e74ff87d88dde","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-03 13:46:43.000000000","message":"Patch Set 30: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/edf2106c392a4df7834d7a50dc4fddfd\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/d5c163c3fa3f405b8a5f95b6c1779be2 : SUCCESS in 12m 26s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fe6fa90e7c044d4682812b6a167cd2b1 : SUCCESS in 5m 28s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/67eaf60d0bb646e2b40e931d40689f4d : SUCCESS in 4m 48s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/e4bbe81e36df4e1c900748b012d28a5b : SUCCESS in 5m 12s\n- octavia-tox-py311-tips https://zuul.opendev.org/t/openstack/build/63acbafb99d4411db6f63e7e050effde : SUCCESS in 4m 23s\n- octavia-tox-functional-py311-tips https://zuul.opendev.org/t/openstack/build/dec1ca2a3fbd466280a8167ec840b61c : SUCCESS in 8m 09s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/372c8f2bbf28485d9d12cdedfeecde73 : SUCCESS in 7m 00s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/0bde69853ce64027b2a48e7d79b89f0b : SUCCESS in 8m 25s\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/f7ec3d9a31b44a6bb11352cdcaeb6637 : SUCCESS in 7m 04s","accounts_in_message":[],"_revision_number":30},{"id":"196359604c54b783a8f1d63572d487d6040e3bcb","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2025-04-02 09:50:09.000000000","message":"Topic rate-limiting removed","accounts_in_message":[],"_revision_number":30}],"current_revision_number":30,"current_revision":"c4df1980f12bbf10db5d9cb92189682d9be3f6e6","revisions":{"1ff85914e5fcf92c557fd96d912e628756b9201f":{"kind":"REWORK","_number":1,"created":"2023-04-28 08:35:24.000000000","uploader":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"ref":"refs/changes/05/881805/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/05/881805/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/05/881805/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/05/881805/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/05/881805/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/05/881805/1"}}},"commit":{"parents":[{"commit":"8b29663c4ca88e3c074bd95fcde33959bfe7afb9","subject":"DNM: Test CI with pytest running","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/8b29663c4ca88e3c074bd95fcde33959bfe7afb9"}]}],"author":{"name":"Tom Weininger","email":"tweining@redhat.com","date":"2023-04-28 08:33:46.000000000","tz":120},"committer":{"name":"Tom Weininger","email":"tweining@redhat.com","date":"2023-04-28 08:35:14.000000000","tz":120},"subject":"Make tests work with pytest runner","message":"Make tests work with pytest runner\n\nChange-Id: I86e50b9d3070e757f1bba8b390c8b7557e41a325\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/1ff85914e5fcf92c557fd96d912e628756b9201f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/1ff85914e5fcf92c557fd96d912e628756b9201f"}]},"branch":"refs/heads/master"},"107135fc9459b3630a699aabee25e76b25c6149e":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2023-04-28 08:40:53.000000000","uploader":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"ref":"refs/changes/05/881805/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/05/881805/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/05/881805/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/05/881805/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/05/881805/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/05/881805/2"}}},"commit":{"parents":[{"commit":"674405063cb998c5c3a40c1547b12df6b1d09690","subject":"DNM: Test CI with pytest running","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/674405063cb998c5c3a40c1547b12df6b1d09690"}]}],"author":{"name":"Tom Weininger","email":"tweining@redhat.com","date":"2023-04-28 08:33:46.000000000","tz":120},"committer":{"name":"Tom Weininger","email":"tweining@redhat.com","date":"2023-04-28 08:40:44.000000000","tz":120},"subject":"Make tests work with pytest runner","message":"Make tests work with pytest runner\n\nChange-Id: I86e50b9d3070e757f1bba8b390c8b7557e41a325\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/107135fc9459b3630a699aabee25e76b25c6149e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/107135fc9459b3630a699aabee25e76b25c6149e"}]},"branch":"refs/heads/master"},"3e48b6477b0c48354057bc8bb06ef1010b6d6376":{"kind":"TRIVIAL_REBASE_WITH_MESSAGE_UPDATE","_number":3,"created":"2023-04-28 08:59:59.000000000","uploader":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"ref":"refs/changes/05/881805/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/05/881805/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/05/881805/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/05/881805/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/05/881805/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/05/881805/3"}}},"commit":{"parents":[{"commit":"1d6baf609b0993280e65be9135577db960f865e7","subject":"Add pytest testenv to tox.ini","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/1d6baf609b0993280e65be9135577db960f865e7"}]}],"author":{"name":"Tom Weininger","email":"tweining@redhat.com","date":"2023-04-28 08:33:46.000000000","tz":120},"committer":{"name":"Tom Weininger","email":"tweining@redhat.com","date":"2023-04-28 08:58:32.000000000","tz":120},"subject":"Make tests work with pytest runner","message":"Make tests work with pytest runner\n\nPartial-Bug: #2017973\nChange-Id: I86e50b9d3070e757f1bba8b390c8b7557e41a325\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/3e48b6477b0c48354057bc8bb06ef1010b6d6376"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/3e48b6477b0c48354057bc8bb06ef1010b6d6376"}]},"branch":"refs/heads/master"},"d4b6b6ffcb31db148bca58382590250bfad2408f":{"kind":"REWORK","_number":4,"created":"2023-04-28 09:33:57.000000000","uploader":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"ref":"refs/changes/05/881805/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/05/881805/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/05/881805/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/05/881805/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/05/881805/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/05/881805/4"}}},"commit":{"parents":[{"commit":"1d6baf609b0993280e65be9135577db960f865e7","subject":"Add pytest testenv to tox.ini","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/1d6baf609b0993280e65be9135577db960f865e7"}]}],"author":{"name":"Tom Weininger","email":"tweining@redhat.com","date":"2023-04-28 08:33:46.000000000","tz":120},"committer":{"name":"Tom Weininger","email":"tweining@redhat.com","date":"2023-04-28 09:33:48.000000000","tz":120},"subject":"Make tests work with pytest runner","message":"Make tests work with pytest runner\n\nPartial-Bug: #2017973\nChange-Id: I86e50b9d3070e757f1bba8b390c8b7557e41a325\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/d4b6b6ffcb31db148bca58382590250bfad2408f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/d4b6b6ffcb31db148bca58382590250bfad2408f"}]},"branch":"refs/heads/master"},"10174f9264d52eb0a17ac399c44adfbd4c94d076":{"kind":"TRIVIAL_REBASE","_number":5,"created":"2023-05-03 16:47:33.000000000","uploader":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"ref":"refs/changes/05/881805/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/05/881805/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/05/881805/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/05/881805/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/05/881805/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/05/881805/5"}}},"commit":{"parents":[{"commit":"70f7998f3bf5dc2fe0dcadcec2049fadbc2bdbc4","subject":"Merge \"Update amphorav2 jobs\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/70f7998f3bf5dc2fe0dcadcec2049fadbc2bdbc4"}]}],"author":{"name":"Tom Weininger","email":"tweining@redhat.com","date":"2023-04-28 08:33:46.000000000","tz":120},"committer":{"name":"Tom Weininger","email":"tweining@redhat.com","date":"2023-05-03 16:47:21.000000000","tz":120},"subject":"Make tests work with pytest runner","message":"Make tests work with pytest runner\n\nPartial-Bug: #2017973\nChange-Id: I86e50b9d3070e757f1bba8b390c8b7557e41a325\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/10174f9264d52eb0a17ac399c44adfbd4c94d076"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/10174f9264d52eb0a17ac399c44adfbd4c94d076"}]},"branch":"refs/heads/master"},"779f254a8293aa9f498d2ce5d4c3002c0ed1b0d7":{"kind":"REWORK","_number":6,"created":"2023-05-04 06:07:29.000000000","uploader":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"ref":"refs/changes/05/881805/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/05/881805/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/05/881805/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/05/881805/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/05/881805/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/05/881805/6"}}},"commit":{"parents":[{"commit":"70f7998f3bf5dc2fe0dcadcec2049fadbc2bdbc4","subject":"Merge \"Update amphorav2 jobs\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/70f7998f3bf5dc2fe0dcadcec2049fadbc2bdbc4"}]}],"author":{"name":"Tom Weininger","email":"tweining@redhat.com","date":"2023-04-28 08:33:46.000000000","tz":120},"committer":{"name":"Tom Weininger","email":"tweining@redhat.com","date":"2023-05-04 06:06:33.000000000","tz":120},"subject":"Make tests work with pytest runner and tox","message":"Make tests work with pytest runner and tox\n\n- Fix tests so they run with pytest. It picks up tests based on class\n  names and function names.\n- Add two new tox environments that use pytest as test runner.\n\nPytest is a much more advanced and easier to use than stestr. It makes\ndeveloping and debugging tests much easier.\n\nCloses-Bug: #2017973\nChange-Id: I86e50b9d3070e757f1bba8b390c8b7557e41a325\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/779f254a8293aa9f498d2ce5d4c3002c0ed1b0d7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/779f254a8293aa9f498d2ce5d4c3002c0ed1b0d7"}]},"branch":"refs/heads/master"},"9d751fa9512709b866d49465918b0e1e6b6d4fb5":{"kind":"TRIVIAL_REBASE_WITH_MESSAGE_UPDATE","_number":7,"created":"2023-05-05 06:29:28.000000000","uploader":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"ref":"refs/changes/05/881805/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/05/881805/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/05/881805/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/05/881805/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/05/881805/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/05/881805/7"}}},"commit":{"parents":[{"commit":"9b02c2fec22db6f0dd0a5017ab9b877f04bdb822","subject":"Add pytest testenv to tox.ini","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/9b02c2fec22db6f0dd0a5017ab9b877f04bdb822"}]}],"author":{"name":"Tom Weininger","email":"tweining@redhat.com","date":"2023-04-28 08:33:46.000000000","tz":120},"committer":{"name":"Tom Weininger","email":"tweining@redhat.com","date":"2023-05-03 15:02:00.000000000","tz":120},"subject":"Make tests work with pytest runner","message":"Make tests work with pytest runner\n\nPartial-Bug: #2017973\nChange-Id: I86e50b9d3070e757f1bba8b390c8b7557e41a325\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/9d751fa9512709b866d49465918b0e1e6b6d4fb5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/9d751fa9512709b866d49465918b0e1e6b6d4fb5"}]},"branch":"refs/heads/master"},"c7f3f890dfa67845c38c30e1774ea1cd4643e0f8":{"kind":"REWORK","_number":8,"created":"2023-05-30 16:28:29.000000000","uploader":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"ref":"refs/changes/05/881805/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/05/881805/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/05/881805/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/05/881805/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/05/881805/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/05/881805/8"}}},"commit":{"parents":[{"commit":"9b02c2fec22db6f0dd0a5017ab9b877f04bdb822","subject":"Add pytest testenv to tox.ini","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/9b02c2fec22db6f0dd0a5017ab9b877f04bdb822"}]}],"author":{"name":"Tom Weininger","email":"tweining@redhat.com","date":"2023-04-28 08:33:46.000000000","tz":120},"committer":{"name":"Tom Weininger","email":"tweining@redhat.com","date":"2023-05-30 16:25:44.000000000","tz":120},"subject":"Make tests work with pytest runner","message":"Make tests work with pytest runner\n\npytest considers classes with names starting with Test\nas test classes and tried to run those.\nThis patch renames base classes that\nshould not run directly.\n\nPartial-Bug: #2017973\nChange-Id: I86e50b9d3070e757f1bba8b390c8b7557e41a325\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/c7f3f890dfa67845c38c30e1774ea1cd4643e0f8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/c7f3f890dfa67845c38c30e1774ea1cd4643e0f8"}]},"branch":"refs/heads/master"},"8d4be55b9ce59f9e9264edef5a21268601372e01":{"kind":"NO_CHANGE","_number":9,"created":"2023-08-23 16:59:59.000000000","uploader":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"ref":"refs/changes/05/881805/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/05/881805/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/05/881805/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/05/881805/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/05/881805/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/05/881805/9"}}},"commit":{"parents":[{"commit":"59630d50dd94de85494adb6189f343a26647d087","subject":"Add pytest testenv to tox.ini","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/59630d50dd94de85494adb6189f343a26647d087"}]}],"author":{"name":"Tom Weininger","email":"tweining@redhat.com","date":"2023-04-28 08:33:46.000000000","tz":120},"committer":{"name":"Tom Weininger","email":"tweining@redhat.com","date":"2023-08-23 16:59:42.000000000","tz":120},"subject":"Make tests work with pytest runner","message":"Make tests work with pytest runner\n\npytest considers classes with names starting with Test\nas test classes and tried to run those.\nThis patch renames base classes that\nshould not run directly.\n\nPartial-Bug: #2017973\nChange-Id: I86e50b9d3070e757f1bba8b390c8b7557e41a325\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/8d4be55b9ce59f9e9264edef5a21268601372e01"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/8d4be55b9ce59f9e9264edef5a21268601372e01"}]},"branch":"refs/heads/master"},"14cf18a540d3efb1d27314e092182aca0752eb0e":{"kind":"REWORK","_number":10,"created":"2024-05-21 11:25:33.000000000","uploader":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"ref":"refs/changes/05/881805/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/05/881805/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/05/881805/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/05/881805/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/05/881805/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/05/881805/10"}}},"commit":{"parents":[{"commit":"09d5015bae278b24600a330db9a89634bc2974f4","subject":"Add pytest testenv to tox.ini","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/09d5015bae278b24600a330db9a89634bc2974f4"}]}],"author":{"name":"Tom Weininger","email":"tweining@redhat.com","date":"2023-04-28 08:33:46.000000000","tz":120},"committer":{"name":"Tom Weininger","email":"tweining@redhat.com","date":"2024-05-21 11:25:18.000000000","tz":120},"subject":"Make tests work with pytest runner","message":"Make tests work with pytest runner\n\npytest considers classes with names starting with Test\nas test classes and tried to run those.\nThis patch renames base classes that\nshould not run directly.\n\nPartial-Bug: #2017973\nChange-Id: I86e50b9d3070e757f1bba8b390c8b7557e41a325\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/14cf18a540d3efb1d27314e092182aca0752eb0e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/14cf18a540d3efb1d27314e092182aca0752eb0e"}]},"branch":"refs/heads/master"},"45ff6aaaed8f62f16c038702edf3c4b972d9b847":{"kind":"TRIVIAL_REBASE","_number":11,"created":"2024-05-21 13:56:31.000000000","uploader":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"ref":"refs/changes/05/881805/11","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/05/881805/11","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/05/881805/11 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/05/881805/11 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/05/881805/11 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/05/881805/11"}}},"commit":{"parents":[{"commit":"e96e3816b28840f9c76cee6e2e4a370312900e1b","subject":"Add pytest testenv to tox.ini","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/e96e3816b28840f9c76cee6e2e4a370312900e1b"}]}],"author":{"name":"Tom Weininger","email":"tweining@redhat.com","date":"2023-04-28 08:33:46.000000000","tz":120},"committer":{"name":"Tom Weininger","email":"tweining@redhat.com","date":"2024-05-21 13:56:14.000000000","tz":120},"subject":"Make tests work with pytest runner","message":"Make tests work with pytest runner\n\npytest considers classes with names starting with Test\nas test classes and tried to run those.\nThis patch renames base classes that\nshould not run directly.\n\nPartial-Bug: #2017973\nChange-Id: I86e50b9d3070e757f1bba8b390c8b7557e41a325\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/45ff6aaaed8f62f16c038702edf3c4b972d9b847"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/45ff6aaaed8f62f16c038702edf3c4b972d9b847"}]},"branch":"refs/heads/master"},"6fddee59440b74f4c6e7357ac6b3155b8c2f3d0e":{"kind":"TRIVIAL_REBASE","_number":12,"created":"2024-05-21 15:43:13.000000000","uploader":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"ref":"refs/changes/05/881805/12","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/05/881805/12","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/05/881805/12 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/05/881805/12 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/05/881805/12 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/05/881805/12"}}},"commit":{"parents":[{"commit":"10aa1b9926f2c89846c3ee468446b817aa73cc32","subject":"Add pytest testenv to tox.ini","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/10aa1b9926f2c89846c3ee468446b817aa73cc32"}]}],"author":{"name":"Tom Weininger","email":"tweining@redhat.com","date":"2023-04-28 08:33:46.000000000","tz":120},"committer":{"name":"Tom Weininger","email":"tweining@redhat.com","date":"2024-05-21 15:43:05.000000000","tz":120},"subject":"Make tests work with pytest runner","message":"Make tests work with pytest runner\n\npytest considers classes with names starting with Test\nas test classes and tried to run those.\nThis patch renames base classes that\nshould not run directly.\n\nPartial-Bug: #2017973\nChange-Id: I86e50b9d3070e757f1bba8b390c8b7557e41a325\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/6fddee59440b74f4c6e7357ac6b3155b8c2f3d0e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/6fddee59440b74f4c6e7357ac6b3155b8c2f3d0e"}]},"branch":"refs/heads/master"},"9adac61bf868606d39a5781af565171435959c77":{"kind":"TRIVIAL_REBASE","_number":13,"created":"2024-05-21 16:06:42.000000000","uploader":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"ref":"refs/changes/05/881805/13","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/05/881805/13","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/05/881805/13 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/05/881805/13 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/05/881805/13 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/05/881805/13"}}},"commit":{"parents":[{"commit":"25e1ea71b920f07d23390f6b35d0579a4f3b848a","subject":"Add pytest testenv to tox.ini","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/25e1ea71b920f07d23390f6b35d0579a4f3b848a"}]}],"author":{"name":"Tom Weininger","email":"tweining@redhat.com","date":"2023-04-28 08:33:46.000000000","tz":120},"committer":{"name":"Tom Weininger","email":"tweining@redhat.com","date":"2024-05-21 16:06:33.000000000","tz":120},"subject":"Make tests work with pytest runner","message":"Make tests work with pytest runner\n\npytest considers classes with names starting with Test\nas test classes and tried to run those.\nThis patch renames base classes that\nshould not run directly.\n\nPartial-Bug: #2017973\nChange-Id: I86e50b9d3070e757f1bba8b390c8b7557e41a325\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/9adac61bf868606d39a5781af565171435959c77"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/9adac61bf868606d39a5781af565171435959c77"}]},"branch":"refs/heads/master"},"2a268dc3e03f6dddc80bfdfe21e6eb22e02ac1d6":{"kind":"TRIVIAL_REBASE","_number":14,"created":"2024-05-22 17:06:25.000000000","uploader":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"ref":"refs/changes/05/881805/14","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/05/881805/14","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/05/881805/14 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/05/881805/14 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/05/881805/14 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/05/881805/14"}}},"commit":{"parents":[{"commit":"77171d49e8aac45e496698c8bfeae4049554592b","subject":"Add pytest testenv to tox.ini","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/77171d49e8aac45e496698c8bfeae4049554592b"}]}],"author":{"name":"Tom Weininger","email":"tweining@redhat.com","date":"2023-04-28 08:33:46.000000000","tz":120},"committer":{"name":"Tom Weininger","email":"tweining@redhat.com","date":"2024-05-22 17:06:15.000000000","tz":120},"subject":"Make tests work with pytest runner","message":"Make tests work with pytest runner\n\npytest considers classes with names starting with Test\nas test classes and tried to run those.\nThis patch renames base classes that\nshould not run directly.\n\nPartial-Bug: #2017973\nChange-Id: I86e50b9d3070e757f1bba8b390c8b7557e41a325\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/2a268dc3e03f6dddc80bfdfe21e6eb22e02ac1d6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/2a268dc3e03f6dddc80bfdfe21e6eb22e02ac1d6"}]},"branch":"refs/heads/master"},"625742eba7a3c159a0c4f3000c41635af780cd50":{"kind":"TRIVIAL_REBASE","_number":15,"created":"2024-06-03 08:15:30.000000000","uploader":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"ref":"refs/changes/05/881805/15","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/05/881805/15","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/05/881805/15 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/05/881805/15 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/05/881805/15 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/05/881805/15"}}},"commit":{"parents":[{"commit":"0237918b05df1436567a8a7acb2c40e4ba19bbf6","subject":"Add pytest testenv to tox.ini","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/0237918b05df1436567a8a7acb2c40e4ba19bbf6"}]}],"author":{"name":"Tom Weininger","email":"tweining@redhat.com","date":"2023-04-28 08:33:46.000000000","tz":120},"committer":{"name":"Tom Weininger","email":"tweining@redhat.com","date":"2024-06-03 08:15:09.000000000","tz":120},"subject":"Make tests work with pytest runner","message":"Make tests work with pytest runner\n\npytest considers classes with names starting with Test\nas test classes and tried to run those.\nThis patch renames base classes that\nshould not run directly.\n\nPartial-Bug: #2017973\nChange-Id: I86e50b9d3070e757f1bba8b390c8b7557e41a325\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/625742eba7a3c159a0c4f3000c41635af780cd50"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/625742eba7a3c159a0c4f3000c41635af780cd50"}]},"branch":"refs/heads/master"},"23bb0754cca01ef244e2872a823e0c3da4342e0b":{"kind":"TRIVIAL_REBASE","_number":16,"created":"2024-06-03 16:03:01.000000000","uploader":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"ref":"refs/changes/05/881805/16","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/05/881805/16","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/05/881805/16 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/05/881805/16 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/05/881805/16 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/05/881805/16"}}},"commit":{"parents":[{"commit":"1c15b8ca54883e2373984386a32eeaf96be60ac4","subject":"Add pytest testenv to tox.ini","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/1c15b8ca54883e2373984386a32eeaf96be60ac4"}]}],"author":{"name":"Tom Weininger","email":"tweining@redhat.com","date":"2023-04-28 08:33:46.000000000","tz":120},"committer":{"name":"Tom Weininger","email":"tweining@redhat.com","date":"2024-06-03 16:02:47.000000000","tz":120},"subject":"Make tests work with pytest runner","message":"Make tests work with pytest runner\n\npytest considers classes with names starting with Test\nas test classes and tried to run those.\nThis patch renames base classes that\nshould not run directly.\n\nPartial-Bug: #2017973\nChange-Id: I86e50b9d3070e757f1bba8b390c8b7557e41a325\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/23bb0754cca01ef244e2872a823e0c3da4342e0b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/23bb0754cca01ef244e2872a823e0c3da4342e0b"}]},"branch":"refs/heads/master"},"492fcc7e8d375ead79234b2975216df91c68d73b":{"kind":"TRIVIAL_REBASE","_number":17,"created":"2024-06-05 10:27:46.000000000","uploader":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"ref":"refs/changes/05/881805/17","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/05/881805/17","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/05/881805/17 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/05/881805/17 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/05/881805/17 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/05/881805/17"}}},"commit":{"parents":[{"commit":"c54a38c7429747addd5c30f9e59c079ac0d7aada","subject":"Add pytest testenv to tox.ini","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/c54a38c7429747addd5c30f9e59c079ac0d7aada"}]}],"author":{"name":"Tom Weininger","email":"tweining@redhat.com","date":"2023-04-28 08:33:46.000000000","tz":120},"committer":{"name":"Tom Weininger","email":"tweining@redhat.com","date":"2024-06-05 10:27:33.000000000","tz":120},"subject":"Make tests work with pytest runner","message":"Make tests work with pytest runner\n\npytest considers classes with names starting with Test\nas test classes and tried to run those.\nThis patch renames base classes that\nshould not run directly.\n\nPartial-Bug: #2017973\nChange-Id: I86e50b9d3070e757f1bba8b390c8b7557e41a325\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/492fcc7e8d375ead79234b2975216df91c68d73b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/492fcc7e8d375ead79234b2975216df91c68d73b"}]},"branch":"refs/heads/master"},"8936b7d04cf577f5746bdff51a7b8d43e3ef246e":{"kind":"TRIVIAL_REBASE","_number":18,"created":"2024-06-05 10:46:20.000000000","uploader":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"ref":"refs/changes/05/881805/18","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/05/881805/18","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/05/881805/18 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/05/881805/18 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/05/881805/18 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/05/881805/18"}}},"commit":{"parents":[{"commit":"37b4fdf3210e63d4614ea926c389c4114b84751c","subject":"Add pytest testenv to tox.ini","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/37b4fdf3210e63d4614ea926c389c4114b84751c"}]}],"author":{"name":"Tom Weininger","email":"tweining@redhat.com","date":"2023-04-28 08:33:46.000000000","tz":120},"committer":{"name":"Tom Weininger","email":"tweining@redhat.com","date":"2024-06-05 10:46:11.000000000","tz":120},"subject":"Make tests work with pytest runner","message":"Make tests work with pytest runner\n\npytest considers classes with names starting with Test\nas test classes and tried to run those.\nThis patch renames base classes that\nshould not run directly.\n\nPartial-Bug: #2017973\nChange-Id: I86e50b9d3070e757f1bba8b390c8b7557e41a325\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/8936b7d04cf577f5746bdff51a7b8d43e3ef246e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/8936b7d04cf577f5746bdff51a7b8d43e3ef246e"}]},"branch":"refs/heads/master"},"0ff3b80f3d480c7399b0ba01052b7a05403b28a3":{"kind":"TRIVIAL_REBASE","_number":19,"created":"2024-06-05 11:12:54.000000000","uploader":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"ref":"refs/changes/05/881805/19","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/05/881805/19","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/05/881805/19 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/05/881805/19 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/05/881805/19 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/05/881805/19"}}},"commit":{"parents":[{"commit":"7c117bc58e3ace5b895cca8b63f4a38217e6fd93","subject":"Add pytest testenv to tox.ini","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/7c117bc58e3ace5b895cca8b63f4a38217e6fd93"}]}],"author":{"name":"Tom Weininger","email":"tweining@redhat.com","date":"2023-04-28 08:33:46.000000000","tz":120},"committer":{"name":"Tom Weininger","email":"tweining@redhat.com","date":"2024-06-05 11:12:34.000000000","tz":120},"subject":"Make tests work with pytest runner","message":"Make tests work with pytest runner\n\npytest considers classes with names starting with Test\nas test classes and tried to run those.\nThis patch renames base classes that\nshould not run directly.\n\nPartial-Bug: #2017973\nChange-Id: I86e50b9d3070e757f1bba8b390c8b7557e41a325\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/0ff3b80f3d480c7399b0ba01052b7a05403b28a3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/0ff3b80f3d480c7399b0ba01052b7a05403b28a3"}]},"branch":"refs/heads/master"},"02375cf9bf9d8df31d402d6b2ffcd6271bd68c17":{"kind":"TRIVIAL_REBASE","_number":20,"created":"2024-06-11 07:25:07.000000000","uploader":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"ref":"refs/changes/05/881805/20","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/05/881805/20","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/05/881805/20 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/05/881805/20 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/05/881805/20 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/05/881805/20"}}},"commit":{"parents":[{"commit":"c7d147c4ee8d11196febfe66e9e2e6964d55dd10","subject":"Add pytest testenv to tox.ini","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/c7d147c4ee8d11196febfe66e9e2e6964d55dd10"}]}],"author":{"name":"Tom Weininger","email":"tweining@redhat.com","date":"2023-04-28 08:33:46.000000000","tz":120},"committer":{"name":"Tom Weininger","email":"tweining@redhat.com","date":"2024-06-11 07:24:52.000000000","tz":120},"subject":"Make tests work with pytest runner","message":"Make tests work with pytest runner\n\npytest considers classes with names starting with Test\nas test classes and tried to run those.\nThis patch renames base classes that\nshould not run directly.\n\nPartial-Bug: #2017973\nChange-Id: I86e50b9d3070e757f1bba8b390c8b7557e41a325\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/02375cf9bf9d8df31d402d6b2ffcd6271bd68c17"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/02375cf9bf9d8df31d402d6b2ffcd6271bd68c17"}]},"branch":"refs/heads/master"},"82eb824449031e75fd53f198f36880a2dd838a89":{"kind":"TRIVIAL_REBASE","_number":21,"created":"2024-10-22 14:59:30.000000000","uploader":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"ref":"refs/changes/05/881805/21","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/05/881805/21","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/05/881805/21 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/05/881805/21 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/05/881805/21 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/05/881805/21"}}},"commit":{"parents":[{"commit":"57e8396ecc50867d6b1545aa41fc7bdda7e416e2","subject":"Add pytest testenv to tox.ini","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/57e8396ecc50867d6b1545aa41fc7bdda7e416e2"}]}],"author":{"name":"Tom Weininger","email":"tweining@redhat.com","date":"2023-04-28 08:33:46.000000000","tz":120},"committer":{"name":"Tom Weininger","email":"tweining@redhat.com","date":"2024-10-22 14:59:07.000000000","tz":120},"subject":"Make tests work with pytest runner","message":"Make tests work with pytest runner\n\npytest considers classes with names starting with Test\nas test classes and tried to run those.\nThis patch renames base classes that\nshould not run directly.\n\nPartial-Bug: #2017973\nChange-Id: I86e50b9d3070e757f1bba8b390c8b7557e41a325\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/82eb824449031e75fd53f198f36880a2dd838a89"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/82eb824449031e75fd53f198f36880a2dd838a89"}]},"branch":"refs/heads/master"},"d935b9b31714e5935a9aa130fc56ad4f11646534":{"kind":"TRIVIAL_REBASE","_number":22,"created":"2024-10-23 10:21:48.000000000","uploader":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"ref":"refs/changes/05/881805/22","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/05/881805/22","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/05/881805/22 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/05/881805/22 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/05/881805/22 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/05/881805/22"}}},"commit":{"parents":[{"commit":"69b1bc038a1842c6bfe0ded8071644518713b84b","subject":"Add pytest testenv to tox.ini","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/69b1bc038a1842c6bfe0ded8071644518713b84b"}]}],"author":{"name":"Tom Weininger","email":"tweining@redhat.com","date":"2023-04-28 08:33:46.000000000","tz":120},"committer":{"name":"Tom Weininger","email":"tweining@redhat.com","date":"2024-10-23 10:21:38.000000000","tz":120},"subject":"Make tests work with pytest runner","message":"Make tests work with pytest runner\n\npytest considers classes with names starting with Test\nas test classes and tried to run those.\nThis patch renames base classes that\nshould not run directly.\n\nPartial-Bug: #2017973\nChange-Id: I86e50b9d3070e757f1bba8b390c8b7557e41a325\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/d935b9b31714e5935a9aa130fc56ad4f11646534"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/d935b9b31714e5935a9aa130fc56ad4f11646534"}]},"branch":"refs/heads/master"},"831f6b0a89b4f49332704f8c794e62ccc3bc7754":{"kind":"TRIVIAL_REBASE","_number":23,"created":"2024-10-28 11:14:50.000000000","uploader":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"ref":"refs/changes/05/881805/23","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/05/881805/23","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/05/881805/23 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/05/881805/23 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/05/881805/23 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/05/881805/23"}}},"commit":{"parents":[{"commit":"30fad8afd9cf6d90a69d0cf8b0f0e4b467ffb682","subject":"Add pytest testenv to tox.ini","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/30fad8afd9cf6d90a69d0cf8b0f0e4b467ffb682"}]}],"author":{"name":"Tom Weininger","email":"tweining@redhat.com","date":"2023-04-28 08:33:46.000000000","tz":120},"committer":{"name":"Tom Weininger","email":"tweining@redhat.com","date":"2024-10-28 11:14:24.000000000","tz":60},"subject":"Make tests work with pytest runner","message":"Make tests work with pytest runner\n\npytest considers classes with names starting with Test\nas test classes and tried to run those.\nThis patch renames base classes that\nshould not run directly.\n\nPartial-Bug: #2017973\nChange-Id: I86e50b9d3070e757f1bba8b390c8b7557e41a325\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/831f6b0a89b4f49332704f8c794e62ccc3bc7754"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/831f6b0a89b4f49332704f8c794e62ccc3bc7754"}]},"branch":"refs/heads/master"},"5a180ad5c81863128a80c17c4409f24fa01baf0e":{"kind":"TRIVIAL_REBASE","_number":24,"created":"2024-10-28 11:45:44.000000000","uploader":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"ref":"refs/changes/05/881805/24","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/05/881805/24","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/05/881805/24 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/05/881805/24 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/05/881805/24 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/05/881805/24"}}},"commit":{"parents":[{"commit":"b766af7473441853c6ac3e53541b0cedf0875896","subject":"Add pytest testenv to tox.ini","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/b766af7473441853c6ac3e53541b0cedf0875896"}]}],"author":{"name":"Tom Weininger","email":"tweining@redhat.com","date":"2023-04-28 08:33:46.000000000","tz":120},"committer":{"name":"Tom Weininger","email":"tweining@redhat.com","date":"2024-10-28 11:45:35.000000000","tz":60},"subject":"Make tests work with pytest runner","message":"Make tests work with pytest runner\n\npytest considers classes with names starting with Test\nas test classes and tried to run those.\nThis patch renames base classes that\nshould not run directly.\n\nPartial-Bug: #2017973\nChange-Id: I86e50b9d3070e757f1bba8b390c8b7557e41a325\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/5a180ad5c81863128a80c17c4409f24fa01baf0e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/5a180ad5c81863128a80c17c4409f24fa01baf0e"}]},"branch":"refs/heads/master"},"26d332a3c7eeb967bfc0caa3e00729da63825d94":{"kind":"TRIVIAL_REBASE","_number":25,"created":"2024-10-28 13:21:24.000000000","uploader":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"ref":"refs/changes/05/881805/25","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/05/881805/25","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/05/881805/25 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/05/881805/25 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/05/881805/25 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/05/881805/25"}}},"commit":{"parents":[{"commit":"fcc162af75de924846ff6473861a4a20902408fb","subject":"Add pytest testenv to tox.ini","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/fcc162af75de924846ff6473861a4a20902408fb"}]}],"author":{"name":"Tom Weininger","email":"tweining@redhat.com","date":"2023-04-28 08:33:46.000000000","tz":120},"committer":{"name":"Tom Weininger","email":"tweining@redhat.com","date":"2024-10-28 13:20:57.000000000","tz":60},"subject":"Make tests work with pytest runner","message":"Make tests work with pytest runner\n\npytest considers classes with names starting with Test\nas test classes and tried to run those.\nThis patch renames base classes that\nshould not run directly.\n\nPartial-Bug: #2017973\nChange-Id: I86e50b9d3070e757f1bba8b390c8b7557e41a325\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/26d332a3c7eeb967bfc0caa3e00729da63825d94"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/26d332a3c7eeb967bfc0caa3e00729da63825d94"}]},"branch":"refs/heads/master"},"b9dbbbd5e693062ed5e4111540bfea7146421e20":{"kind":"TRIVIAL_REBASE","_number":26,"created":"2024-11-20 12:07:07.000000000","uploader":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"ref":"refs/changes/05/881805/26","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/05/881805/26","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/05/881805/26 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/05/881805/26 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/05/881805/26 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/05/881805/26"}}},"commit":{"parents":[{"commit":"f75b08c248974258525dcf81689005b7552496ca","subject":"Add pytest testenv to tox.ini","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/f75b08c248974258525dcf81689005b7552496ca"}]}],"author":{"name":"Tom Weininger","email":"tweining@redhat.com","date":"2023-04-28 08:33:46.000000000","tz":120},"committer":{"name":"Tom Weininger","email":"tweining@redhat.com","date":"2024-11-20 12:03:27.000000000","tz":60},"subject":"Make tests work with pytest runner","message":"Make tests work with pytest runner\n\npytest considers classes with names starting with Test\nas test classes and tried to run those.\nThis patch renames base classes that\nshould not run directly.\n\nPartial-Bug: #2017973\nChange-Id: I86e50b9d3070e757f1bba8b390c8b7557e41a325\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/b9dbbbd5e693062ed5e4111540bfea7146421e20"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/b9dbbbd5e693062ed5e4111540bfea7146421e20"}]},"branch":"refs/heads/master"},"eeee8b19da1a6af82103db37e5dd450e1d52814c":{"kind":"TRIVIAL_REBASE","_number":27,"created":"2025-02-10 15:47:30.000000000","uploader":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"ref":"refs/changes/05/881805/27","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/05/881805/27","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/05/881805/27 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/05/881805/27 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/05/881805/27 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/05/881805/27"}}},"commit":{"parents":[{"commit":"27ea2add835ae63d64921a4992b9c054a58c7a71","subject":"Add pytest testenv to tox.ini","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/27ea2add835ae63d64921a4992b9c054a58c7a71"}]}],"author":{"name":"Tom Weininger","email":"tweining@redhat.com","date":"2023-04-28 08:33:46.000000000","tz":120},"committer":{"name":"Tom Weininger","email":"tweining@redhat.com","date":"2025-02-10 15:47:11.000000000","tz":60},"subject":"Make tests work with pytest runner","message":"Make tests work with pytest runner\n\npytest considers classes with names starting with Test\nas test classes and tried to run those.\nThis patch renames base classes that\nshould not run directly.\n\nPartial-Bug: #2017973\nChange-Id: I86e50b9d3070e757f1bba8b390c8b7557e41a325\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/eeee8b19da1a6af82103db37e5dd450e1d52814c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/eeee8b19da1a6af82103db37e5dd450e1d52814c"}]},"branch":"refs/heads/master"},"91bfd0b41c2383ef347f722b24ba0f9d5b55aa86":{"kind":"TRIVIAL_REBASE","_number":28,"created":"2025-02-11 09:38:24.000000000","uploader":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"ref":"refs/changes/05/881805/28","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/05/881805/28","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/05/881805/28 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/05/881805/28 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/05/881805/28 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/05/881805/28"}}},"commit":{"parents":[{"commit":"29372ce0676fe67920badc8f355aa42f55b522f7","subject":"Add pytest testenv to tox.ini","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/29372ce0676fe67920badc8f355aa42f55b522f7"}]}],"author":{"name":"Tom Weininger","email":"tweining@redhat.com","date":"2023-04-28 08:33:46.000000000","tz":120},"committer":{"name":"Tom Weininger","email":"tweining@redhat.com","date":"2025-02-11 09:38:13.000000000","tz":60},"subject":"Make tests work with pytest runner","message":"Make tests work with pytest runner\n\npytest considers classes with names starting with Test\nas test classes and tried to run those.\nThis patch renames base classes that\nshould not run directly.\n\nPartial-Bug: #2017973\nChange-Id: I86e50b9d3070e757f1bba8b390c8b7557e41a325\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/91bfd0b41c2383ef347f722b24ba0f9d5b55aa86"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/91bfd0b41c2383ef347f722b24ba0f9d5b55aa86"}]},"branch":"refs/heads/master"},"434cde623ec09c0bdc0bebb3e08ce5057f6c9fbe":{"kind":"TRIVIAL_REBASE","_number":29,"created":"2025-02-24 15:28:52.000000000","uploader":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"ref":"refs/changes/05/881805/29","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/05/881805/29","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/05/881805/29 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/05/881805/29 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/05/881805/29 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/05/881805/29"}}},"commit":{"parents":[{"commit":"b8cad01e0abd68ae952cb1fa1336ec9af6796f8e","subject":"Add pytest testenv to tox.ini","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/b8cad01e0abd68ae952cb1fa1336ec9af6796f8e"}]}],"author":{"name":"Tom Weininger","email":"tweining@redhat.com","date":"2023-04-28 08:33:46.000000000","tz":120},"committer":{"name":"Tom Weininger","email":"tweining@redhat.com","date":"2025-02-24 15:24:58.000000000","tz":60},"subject":"Make tests work with pytest runner","message":"Make tests work with pytest runner\n\npytest considers classes with names starting with Test\nas test classes and tried to run those.\nThis patch renames base classes that\nshould not run directly.\n\nPartial-Bug: #2017973\nChange-Id: I86e50b9d3070e757f1bba8b390c8b7557e41a325\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/434cde623ec09c0bdc0bebb3e08ce5057f6c9fbe"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/434cde623ec09c0bdc0bebb3e08ce5057f6c9fbe"}]},"branch":"refs/heads/master"},"c4df1980f12bbf10db5d9cb92189682d9be3f6e6":{"kind":"TRIVIAL_REBASE","_number":30,"created":"2025-03-03 13:33:42.000000000","uploader":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"ref":"refs/changes/05/881805/30","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/05/881805/30","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/05/881805/30 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/05/881805/30 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/05/881805/30 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/05/881805/30"}}},"commit":{"parents":[{"commit":"c1b5d820dc49ad6a9b4d78146d81cbbaa758e3f2","subject":"Add pytest testenv to tox.ini","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/c1b5d820dc49ad6a9b4d78146d81cbbaa758e3f2"}]}],"author":{"name":"Tom Weininger","email":"tweining@redhat.com","date":"2023-04-28 08:33:46.000000000","tz":120},"committer":{"name":"Tom Weininger","email":"tweining@redhat.com","date":"2025-03-03 13:23:11.000000000","tz":60},"subject":"Make tests work with pytest runner","message":"Make tests work with pytest runner\n\npytest considers classes with names starting with Test\nas test classes and tried to run those.\nThis patch renames base classes that\nshould not run directly.\n\nPartial-Bug: #2017973\nChange-Id: I86e50b9d3070e757f1bba8b390c8b7557e41a325\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/c4df1980f12bbf10db5d9cb92189682d9be3f6e6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/c4df1980f12bbf10db5d9cb92189682d9be3f6e6"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"OK","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY"},{"label":"Workflow","status":"MAY"},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
