)]}'
{"id":"openstack%2Foctavia~882117","triplet_id":"openstack%2Foctavia~master~Ib8b87505aaec8533bb12057c94af5eddf515e91c","project":"openstack/octavia","branch":"master","topic":"sqlalchemy2","attention_set":{},"removed_from_attention_set":{"29244":{"account":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"last_update":"2023-08-22 14:02:41.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Ib8b87505aaec8533bb12057c94af5eddf515e91c","subject":"Fix pool_timeout for tests with sqlite file backend","status":"MERGED","created":"2023-05-03 07:07:19.000000000","updated":"2023-08-22 14:05:38.000000000","submitted":"2023-08-22 14:02:41.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"882117-sqlalchemy2","meta_rev_id":"853205735edd5132290e24afe98eeda9b3f5cee1","_number":882117,"virtual_id_number":882117,"owner":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-08-22 14:02:41.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2023-08-16 20:48:20.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":2,"date":"2023-08-22 12:18:56.000000000","_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":1,"date":"2023-08-22 12:18:56.000000000","_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-05-03 07:30:29.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-08-16 20:48:20.000000000","updated_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"reviewer":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"state":"REVIEWER"},{"updated":"2023-08-22 12:18:56.000000000","updated_by":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"reviewer":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"state":"REVIEWER"}],"messages":[{"id":"c22f8aed7d0d6a58018d1d24973bf35855b55f88","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2023-05-03 07:07:19.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"398ce8ab59bdff648ae03ea3bccbf35a5e1ea80b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-03 07:30:29.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/65092e7393704209af654b7c7620f379\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/e522ae95a7cc462290c3774edf7645e5 : FAILURE in 4m 41s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b80e5b92b40748659af1f100114fb354 : SUCCESS in 5m 05s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/8d897d863c074420b8528c002179d18c : FAILURE in 4m 54s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/e927b507a11d4741aaaf9ec585ff3473 : FAILURE in 4m 47s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cb628ffe1b8443cfbc275426938848a6 : SUCCESS in 11m 35s\n- octavia-tox-py39-tips https://zuul.opendev.org/t/openstack/build/1a916aa054fd45e7a039954b500e6924 : FAILURE in 4m 32s\n- octavia-tox-functional-py39-tips https://zuul.opendev.org/t/openstack/build/2e3794d6ce824a83bb584646b16bc86a : SUCCESS in 8m 31s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/c7b73126eb304dee8582a0497e77692e : SUCCESS in 7m 21s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/b5736f7e18e34dd4adf626c388926a55 : SUCCESS in 7m 29s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/44a356261c4c4c8abd9569a8516ac749 : SUCCESS in 7m 30s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"4f88cc411893e56589dd82c9c157ac57fd01fde0","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2023-07-28 07:25:38.000000000","message":"Patch Set 2: Patch Set 1 was rebased\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"392c5490d020aaa1f5ea7c79268e9a49bf173aad","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-28 07:50:07.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/11d3e0f3135647b3928301cd030c7930\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/e8e092b62da944709a186e19ff2e7863 : FAILURE in 4m 56s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/007266ce52b24a71bbc6a318dbd3de86 : SUCCESS in 4m 24s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/b3070463229b4262a7f175bac2f20605 : FAILURE in 6m 00s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/33d696d8fe824e5296cb25f5bb0a8a39 : FAILURE in 9m 53s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/885874b3989641bd8b4037002fcf4749 : FAILURE in 5m 46s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f445abbf55f74b87938192675c4c3803 : SUCCESS in 9m 30s\n- octavia-tox-py39-tips https://zuul.opendev.org/t/openstack/build/4d0a88963f1e4d9f866aec5a35944365 : FAILURE in 5m 27s\n- octavia-tox-functional-py39-tips https://zuul.opendev.org/t/openstack/build/68460719472548a68e300bf27ac14d83 : SUCCESS in 7m 46s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/061b0ffefdeb45c0bcd59c93cf0f70f7 : SUCCESS in 8m 38s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/ea55b9be2fc4426b81710532fca6ac8e : SUCCESS in 8m 04s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/2c09b06a87464a55993a4a0cdf505194 : SUCCESS in 6m 53s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"6d13305927a6f28eb08d392e86f45acec23dc20e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2023-07-28 12:37:04.000000000","message":"Patch Set 3: Patch Set 2 was rebased\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":3},{"id":"9440e7bcb2d1687aad2d2af7f522307bcb48a5e0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-28 13:27:32.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6c67d49702234e0aac0c4ba1addc7990\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/13d629bb3eba474c8a7d208fa958d446 : SUCCESS in 12m 03s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/84099ac1e687480db0e2a03e83e8f128 : SUCCESS in 4m 51s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/115a852ee29f418a97b5e876025de181 : SUCCESS in 4m 56s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/ebe4d09337554425af5dd689b36f7e2b : SUCCESS in 5m 35s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/de26ae70dca14f49af5ecc9c3c3de343 : SUCCESS in 5m 03s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f0b49eb1ca5d422a998da66b684df914 : SUCCESS in 11m 08s\n- octavia-tox-py39-tips https://zuul.opendev.org/t/openstack/build/c88630186fa74fb78905cf49f1cd376c : SUCCESS in 5m 23s\n- octavia-tox-functional-py39-tips https://zuul.opendev.org/t/openstack/build/b2b83e6bd40e4fb7947088c439e3bc28 : SUCCESS in 9m 23s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/451b4d9e9c444938ae2f07ce0831723b : SUCCESS in 7m 14s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/e1ed39ebc369434aa78146123a3422d1 : SUCCESS in 10m 41s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/911c547ad700420d825d7807b88c0bd4 : SUCCESS in 8m 54s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"be1639630471b3864efc37c573ee845052314a79","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2023-08-16 20:48:20.000000000","message":"Patch Set 3: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"7293291e1dc8f92a06e3ec52bb6af1f25b85fc11","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2023-08-22 12:18:56.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"815621a5525d72b96033742bd1a943aa8499f8af","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-22 12:19:40.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"88b797fab546ecd9a28608658a27d9145f368e6d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-22 14:02:41.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/84f8efc110d44a61bd3ddef9a841d4f9\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ba7c151c3b67478f8fc658d7b8699e0c : SUCCESS in 5m 44s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/cdc32d2b339047a3b57baabd8181def8 : SUCCESS in 5m 37s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/1e30847e19964b0bbb50da71ac32cf8b : SUCCESS in 6m 09s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/50a5232af43147c49bb3c982441f17a3 : SUCCESS in 5m 00s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2ba99cd2f93c4b04adab0e994bf01475 : SUCCESS in 11m 33s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/d28982c0549545e89e0242c170d848f5 : SUCCESS in 8m 35s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/a69eece9680641d98317dc2c57b94b45 : SUCCESS in 1h 16m 20s\n- octavia-v2-dsvm-scenario-traffic-ops https://zuul.opendev.org/t/openstack/build/c97ccaf3c32f4397af6eb6db802a0f43 : SUCCESS in 1h 37m 49s\n- octavia-v2-dsvm-scenario-non-traffic-ops https://zuul.opendev.org/t/openstack/build/cebcc5b3d2964b6da4cc6d136daf22f9 : SUCCESS in 1h 08m 00s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/1b1cf20905fc46748d5ea40e4426977f : SUCCESS in 37m 21s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/28e6def4906848ae99b413c1a248f174 : SUCCESS in 54m 31s","accounts_in_message":[],"_revision_number":3},{"id":"c9d4f4e97511bb9a17a11a7bc41d1d723bf868cf","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-22 14:02:41.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":3},{"id":"853205735edd5132290e24afe98eeda9b3f5cee1","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-22 14:05:38.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a8e065e42d1d414985ad11df6c54487f\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ac012e072d60438a8d455c543c85c32e : SUCCESS in 1m 09s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"390ba39638522b1bef94fc48c7530dfce99da8d5","revisions":{"6794cf09e92544d895ee4d9b5b0607f967f4b382":{"kind":"REWORK","_number":1,"created":"2023-05-03 07:07:19.000000000","uploader":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"ref":"refs/changes/17/882117/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/17/882117/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/17/882117/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/17/882117/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/17/882117/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/17/882117/1"}}},"commit":{"parents":[{"commit":"fd233a40561c9d7de29c7e84426fc9fa084ab5f7","subject":"Fix sqlalchemy declarative_base import","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/fd233a40561c9d7de29c7e84426fc9fa084ab5f7"}]}],"author":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2023-05-03 07:01:35.000000000","tz":-240},"committer":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2023-05-03 07:03:49.000000000","tz":-240},"subject":"Fix pool_timeout for tests with sqlite file backend","message":"Fix pool_timeout for tests with sqlite file backend\n\npool_timeout is not supported when using files with the sqlite backend.\n\nChange-Id: Ib8b87505aaec8533bb12057c94af5eddf515e91c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/6794cf09e92544d895ee4d9b5b0607f967f4b382"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/6794cf09e92544d895ee4d9b5b0607f967f4b382"}]},"branch":"refs/heads/master"},"fcdefa64e0ef21b3078e68dc6e51735de5a16c3d":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2023-07-28 07:25:38.000000000","uploader":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"ref":"refs/changes/17/882117/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/17/882117/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/17/882117/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/17/882117/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/17/882117/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/17/882117/2"}}},"commit":{"parents":[{"commit":"258ef4b2b4d62e2d1b27584b94d61fc636864722","subject":"Fix sqlalchemy declarative_base import","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/258ef4b2b4d62e2d1b27584b94d61fc636864722"}]}],"author":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2023-05-03 07:01:35.000000000","tz":-240},"committer":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2023-07-28 07:25:38.000000000","tz":0},"subject":"Fix pool_timeout for tests with sqlite file backend","message":"Fix pool_timeout for tests with sqlite file backend\n\npool_timeout is not supported when using files with the sqlite backend.\n\nChange-Id: Ib8b87505aaec8533bb12057c94af5eddf515e91c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/fcdefa64e0ef21b3078e68dc6e51735de5a16c3d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/fcdefa64e0ef21b3078e68dc6e51735de5a16c3d"}]},"branch":"refs/heads/master","description":"Rebase"},"390ba39638522b1bef94fc48c7530dfce99da8d5":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2023-07-28 12:37:04.000000000","uploader":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"ref":"refs/changes/17/882117/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/17/882117/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/17/882117/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/17/882117/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/17/882117/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/17/882117/3"}}},"commit":{"parents":[{"commit":"e59a555becbece5c776b0876cacb0e120fcd174f","subject":"Fix sqlalchemy declarative_base import","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/e59a555becbece5c776b0876cacb0e120fcd174f"}]}],"author":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2023-05-03 07:01:35.000000000","tz":-240},"committer":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2023-07-28 12:37:04.000000000","tz":0},"subject":"Fix pool_timeout for tests with sqlite file backend","message":"Fix pool_timeout for tests with sqlite file backend\n\npool_timeout is not supported when using files with the sqlite backend.\n\nChange-Id: Ib8b87505aaec8533bb12057c94af5eddf515e91c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/390ba39638522b1bef94fc48c7530dfce99da8d5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/390ba39638522b1bef94fc48c7530dfce99da8d5"}]},"branch":"refs/heads/master","description":"Rebase"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
