)]}'
{"id":"openstack%2Foctavia~882276","triplet_id":"openstack%2Foctavia~stable%2Fzed~Ie5fb25e292461baca38499091e8f9b9a53e3ddcd","project":"openstack/octavia","branch":"stable/zed","topic":"fix_pool_create_single_lb_create_call","attention_set":{},"removed_from_attention_set":{"31664":{"account":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"last_update":"2023-06-28 00:09:48.000000000","reason":"Change was submitted"},"7249":{"account":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"last_update":"2023-06-21 08:39:56.000000000","reason":"\u003cGERRIT_ACCOUNT_7249\u003e replied on the change","reason_account":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"}},"29244":{"account":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"last_update":"2023-06-26 13:52:45.000000000","reason":"\u003cGERRIT_ACCOUNT_29244\u003e replied on the change","reason_account":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"}},"34429":{"account":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"last_update":"2023-06-28 00:09:48.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Ie5fb25e292461baca38499091e8f9b9a53e3ddcd","subject":"Fix pool creation with single LB create call","status":"MERGED","created":"2023-05-04 11:36:03.000000000","updated":"2023-06-28 00:11:42.000000000","submitted":"2023-06-28 00:09:48.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"882276-fix_pool_create_single_lb_create_call","meta_rev_id":"a975832e90ff6380ad8d156d55f9f47b8d959afa","_number":882276,"virtual_id_number":882276,"owner":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"value":0,"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-06-28 00:09:48.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"all":[{"value":2,"date":"2023-06-21 08:39:56.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"value":1,"date":"2023-06-23 12:16:13.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},{"value":1,"date":"2023-05-30 20:46:27.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":2,"date":"2023-06-26 13:52:45.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"all":[{"value":0,"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"value":0,"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":1,"date":"2023-06-26 13:52:45.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"value":0,"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}],"reviewers":{"REVIEWER":[{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-05-04 14:58:51.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-05-30 20:46:27.000000000","updated_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"reviewer":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"state":"REVIEWER"},{"updated":"2023-06-21 08:16:08.000000000","updated_by":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"reviewer":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"state":"REVIEWER"},{"updated":"2023-06-21 08:16:08.000000000","updated_by":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"reviewer":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"state":"REVIEWER"},{"updated":"2023-06-21 08:16:08.000000000","updated_by":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"reviewer":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"state":"REVIEWER"},{"updated":"2023-06-23 12:16:13.000000000","updated_by":{"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},"reviewer":{"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},"state":"REVIEWER"}],"messages":[{"id":"3c5c178442334acad44d266767f60135df5a7965","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"date":"2023-05-04 11:36:03.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"f0d01617e84a4a24562726d976860c9eeb69a553","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-04 14:58:51.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/72c79a56603845789bed42e5d5c1d3c8\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/674f9a159cc048dea2be4f4f06c12910 : SUCCESS in 11m 37s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/00ae1f671c83484492a4ef4bc0146864 : FAILURE in 4m 20s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/de68e81824de4b6f93c942ac0c6e230c : SUCCESS in 4m 38s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/059f6afb249945ec8340686970fad2ea : SUCCESS in 4m 56s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/9407627be451473e96e59e46f51d9cec : SUCCESS in 5m 06s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ac2c8b98743c4cf08de22590cb7a343c : SUCCESS in 11m 25s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/79963013f85649cc891f9e0f4e478642 : SUCCESS in 5m 52s\n- octavia-tox-py39-tips https://zuul.opendev.org/t/openstack/build/cdcb540990944973a712b8a2048c7e34 : SUCCESS in 5m 13s\n- octavia-tox-functional-py39-tips https://zuul.opendev.org/t/openstack/build/5518e5c346e24b1eaa484b6b946f2df2 : SUCCESS in 7m 23s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/f87d86760c8b483d9d235d6f395504f3 : SUCCESS in 4m 32s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/5d8ab2b9e4654f5cafa2ea38c5259c3a : SUCCESS in 7m 40s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/7346f3b3a9f54978af0923a25b9d5de0 : SUCCESS in 8m 29s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/d8bac44eef68408a85ea690bf7d888f5 : SUCCESS in 8m 26s (non-voting)\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/f979b2eba2ac40a48009043338a60b18 : SUCCESS in 1h 49m 38s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/748f87e9f5c24d8ba392f2b03984a84a : SUCCESS in 2h 11m 17s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/45a0afd5e5ab4e9e82838d84b903aed1 : SUCCESS in 43m 41s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/309fec9d5a0548f982f6abd560e61d74 : SUCCESS in 53m 30s\n- octavia-v2-dsvm-tls-barbican-fips https://zuul.opendev.org/t/openstack/build/9c7b2b7294de44f2a3c631e23623b1d1 : SUCCESS in 1h 07m 09s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/2e2312e96dd54cccabe3bfc3050958df : SUCCESS in 46m 03s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/983b0b5b94a04ee5a1f61e9e9ec84a38 : SUCCESS in 2h 23m 38s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/17900b8516614630868b5f6a654d86c3 : SUCCESS in 3h 00m 59s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/3f3d7c5a6e984946b1e4266affa1cb47 : TIMED_OUT in 2h 33m 49s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/5ad1435927114ae49f36264a6792f98a : SUCCESS in 2h 14m 01s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/ea6cb2a3c8984611a26189a031466e3c : SUCCESS in 2h 25m 46s (non-voting)\n- octavia-v2-dsvm-scenario-centos-9-stream https://zuul.opendev.org/t/openstack/build/b7a48e19604a45c08d8dab0a9fba52cb : SUCCESS in 2h 23m 27s (non-voting)\n\nWarning:\n  Comments left for invalid file octavia/hacking/checks.py","accounts_in_message":[],"_revision_number":1},{"id":"1a2bb076f20cfae248bd1111683940ca9a7bdb85","author":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"date":"2023-05-08 09:15:35.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"80a8bc124f07c0f8e3d6f1a5bb3fa27b67111a94","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-08 12:27:23.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d8249c057fc14b6eb6e91c5589fcc589\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/ba2e9b57464e4f53b1a8b89bbb2b485d : SUCCESS in 10m 20s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1b44a4376cf443ffb76577d03ace99e8 : SUCCESS in 7m 48s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/f5a0e27ca9dc4f09ad240c28729169fc : SUCCESS in 4m 57s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/7867d7f6a5e74aa9805d4374c6179e47 : SUCCESS in 4m 18s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/bb3da7172c5e4926a77e27233c9c5662 : SUCCESS in 5m 39s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7ee0ceb953704243bee076e7d57550ef : SUCCESS in 11m 21s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/9c6598a1da0d487294851d44c4a3ac50 : SUCCESS in 6m 08s\n- octavia-tox-py39-tips https://zuul.opendev.org/t/openstack/build/7d933e70638d4d819d07833e3414f653 : SUCCESS in 5m 29s\n- octavia-tox-functional-py39-tips https://zuul.opendev.org/t/openstack/build/53b2099844984f35b47f163c04b2755e : SUCCESS in 7m 58s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/6b59883c3bae40e396cacf144b1731ce : SUCCESS in 4m 17s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/00372ac3be3d41aabe9748d87514c40e : SUCCESS in 6m 49s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/df10aebd172d499a82af92dc1e78db56 : SUCCESS in 7m 31s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/e71ddd7b11aa4c6e8512d3647a5b0a46 : SUCCESS in 6m 45s (non-voting)\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/8ec4bca9798b402eadc734a06a4c34d4 : SUCCESS in 1h 18m 25s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/9a1f8fde9d3b408e928aec88d6312e34 : SUCCESS in 2h 07m 31s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/1fc054826c3744aea63059199c286e37 : SUCCESS in 44m 36s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/e054d76eff29430ba61e736b37ad21b5 : SUCCESS in 50m 50s\n- octavia-v2-dsvm-tls-barbican-fips https://zuul.opendev.org/t/openstack/build/40de3303267949d9b3cb4793f5caee71 : SUCCESS in 1h 04m 33s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/7d237af13eea4efb80d6a65ded65fdcb : SUCCESS in 41m 35s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/704c430561144711a725b57785a3f697 : TIMED_OUT in 2h 34m 19s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/e070866fcf994206ad4212406613e732 : SUCCESS in 2h 24m 50s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/2c0e50b8bd0d4f3d8d20698cc5be4b2d : SUCCESS in 2h 31m 03s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/fce5bb66746342849c334f67b4301db6 : SUCCESS in 2h 10m 25s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/85c08355f23f423fb7f1c9d1894f892b : SUCCESS in 2h 21m 05s (non-voting)\n- octavia-v2-dsvm-scenario-centos-9-stream https://zuul.opendev.org/t/openstack/build/5bd907816cc44e7e941ac4d62f44c025 : TIMED_OUT in 2h 33m 34s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"0bd28b90a616a1fb51e6b4b1aca6c32cda2dcc1b","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2023-05-30 20:46:27.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"cb2c49fc5d75126c9a89b00c6576500ab93c004b","author":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"date":"2023-06-21 08:39:56.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"c7bb81700201039df7b8f13871d1bad1afe6242b","author":{"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},"date":"2023-06-23 12:16:13.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"70aeca7a2c858c6fe5f6f557464d420d8f031044","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2023-06-26 13:52:45.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"813b38a7145d71550b873e8f30385eab8a369a29","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-26 13:53:21.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"5e25bdb6bf8fc613ac09c52fdd5d7d30744e5132","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-26 14:47:00.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/048cd491d7354af9b40fdcb672642f9a\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6ac815e3ad764c1f8a41e54979fd56cd : SUCCESS in 6m 27s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/232fe2b6887e4fa293d844b1d867cc33 : SUCCESS in 6m 04s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/f3b393e798114643bed957f133a89dda : SUCCESS in 5m 37s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/90759960dc3d48028eadcfbbe4ba2a26 : SUCCESS in 11m 58s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/60e65d04dd71462191b59c00a40fa40d : SUCCESS in 5m 45s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/6a1ba23014934814b452d3999c8fb4ce : SUCCESS in 5m 12s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/6e07d09275cf4d9091860297048dbda0 : SUCCESS in 8m 02s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/25a0be8333d141f99703e521587ec9e3 : CANCELED\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/c815358ef0b54ed9b476d59dd1100829 : CANCELED\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/6c06f24f783f4e63ba8b8d3700a0f4d5 : FAILURE in 29m 23s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/13ce7f2a707844b68b02641f16528efe : CANCELED","accounts_in_message":[],"_revision_number":1},{"id":"1a07b147a1c3f1e90f3768355f21e29a1f9187fe","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2023-06-27 19:53:11.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"722fcfcc515ab523a71597030754b7161fc5d1e5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-27 22:27:56.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/bfc96fb10c94415c9732c35011757109\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/848acff656d745cc9647aa8802dc5bb0 : SUCCESS in 11m 22s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f0ba1f1e103749509288597de8f3b494 : SUCCESS in 7m 14s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/e8a28b9a11c0407ab1211684f5f5116c : SUCCESS in 5m 26s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/1cf0a424580c416a8028c5a43f49586c : SUCCESS in 5m 25s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/a55941716bbe494996c308628e20fcf9 : SUCCESS in 4m 53s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a351b1bd1ab34f6397e99d5b4ae2b200 : SUCCESS in 12m 46s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/277c57aa84fa4b5094817707cee48036 : SUCCESS in 4m 18s\n- octavia-tox-py39-tips https://zuul.opendev.org/t/openstack/build/7bcf266e8700427fbf91d14676dd3c0c : SUCCESS in 5m 54s\n- octavia-tox-functional-py39-tips https://zuul.opendev.org/t/openstack/build/ab8e810aa3a84a70bdf796b3c5b36285 : SUCCESS in 7m 57s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/47c322e6d40244688482b3b5371f4125 : SUCCESS in 4m 56s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/57111bbe7efa481a88e6fdf1b89d94c4 : SUCCESS in 7m 24s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/88f22ce0cd4a4e549bb3314d878417c6 : SUCCESS in 8m 22s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/1fee360d73d04064bdbacf7dce59c228 : SUCCESS in 8m 15s (non-voting)\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/4197b4ce8af64786b398ca5f7e0694d8 : SUCCESS in 1h 36m 11s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/43066cd81c5f4c119193f870d6f5e265 : SUCCESS in 1h 28m 47s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/76e9a9d792db4b84bd14c60dff1ea126 : SUCCESS in 34m 41s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/0ccd89377db6472ab15713126b041803 : SUCCESS in 36m 27s\n- octavia-v2-dsvm-tls-barbican-fips https://zuul.opendev.org/t/openstack/build/229e3f0f14bc4039aa7263e4a4a2f7b7 : FAILURE in 9m 59s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/af83f4dd644440a18d07b4510c45a980 : SUCCESS in 32m 10s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/1d4f732944d343899d730448693199f0 : SUCCESS in 1h 40m 55s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/c82ebd4a81d148cf980e015d2e8c0323 : SUCCESS in 2h 21m 01s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/15e43c3af7e841c891d1b067bb390790 : SUCCESS in 2h 33m 04s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/39652ef7ddd04fb395fe1e1ab713b183 : SUCCESS in 2h 11m 38s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/04c2aa20f0ea4d75bde29ddf38a5d419 : SUCCESS in 1h 29m 07s (non-voting)\n- octavia-v2-dsvm-scenario-centos-9-stream https://zuul.opendev.org/t/openstack/build/65791dcdf098456fbf9b831053f9cc9b : SUCCESS in 2h 23m 01s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"d3ad650b2f4b608e8a6a8be96bf98e4e2b07e182","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-27 22:28:34.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"ce7ba20d36be396adbfe44a440570bdd99dc772d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-28 00:09:48.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0f0c585d6b9a4f3b807ac64dfc806fe8\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/56d5a9eabea1447d8769527bb50c85ff : SUCCESS in 6m 50s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/94d8c4e419014b15a479554e38986d4d : SUCCESS in 5m 11s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/8ae8eb8cddd1416f84d683ab39171254 : SUCCESS in 5m 25s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ebdc272c1b194338bb59ee1365a3c932 : SUCCESS in 12m 12s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/20b6a98a5dd8412f8e8c422c3ef0a3fc : SUCCESS in 10m 22s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/6dbfc541417349239f530cadb977f575 : SUCCESS in 8m 50s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/22ed6b4002084675bf8ed566d3ecfb26 : SUCCESS in 6m 49s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/fbf356beccbf4cc0a2b479c69feb87b0 : SUCCESS in 1h 01m 16s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/ee665c8e527a4ec2ae54f78ea2d25ea2 : SUCCESS in 1h 40m 45s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/758b3a519f674045827b9b5bc8db133a : SUCCESS in 35m 15s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/055c3f85c5544e14836b3787e1b46e16 : SUCCESS in 46m 34s","accounts_in_message":[],"_revision_number":1},{"id":"7175e296fc5205c5a645278c6e91aec36c9a9684","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-28 00:09:48.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"a975832e90ff6380ad8d156d55f9f47b8d959afa","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-28 00:11:42.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d7fca04f55924c2f91fc30735c288bd8\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e936e477a16b4c98b91997f0044dc1a5 : SUCCESS in 1m 36s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/d7deb21a61144564bc59b785b8d7b923 : SUCCESS in 43s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"396785dadd0e1c66fcc845799cdfc814faec7d84","revisions":{"396785dadd0e1c66fcc845799cdfc814faec7d84":{"kind":"REWORK","_number":1,"created":"2023-05-04 11:36:03.000000000","uploader":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"ref":"refs/changes/76/882276/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/76/882276/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/76/882276/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/76/882276/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/76/882276/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/76/882276/1"}}},"commit":{"parents":[{"commit":"332d7deeba740b9fd8a67065f5818010a5d2990b","subject":"Fix octavia to accept [ipv6]:port","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/332d7deeba740b9fd8a67065f5818010a5d2990b"}]}],"author":{"name":"Omer","email":"oschwart@redhat.com","date":"2022-11-08 16:32:17.000000000","tz":60},"committer":{"name":"Omer","email":"oschwart@redhat.com","date":"2023-05-04 11:35:55.000000000","tz":120},"subject":"Fix pool creation with single LB create call","message":"Fix pool creation with single LB create call\n\nSo far, creating a populated LB (or an LB tree) with non attached pools,\nleft the pools on PENDING_CREATE provisioning_state.\n\nThis patch adds the option of marking non attached pools to be active on\nthe MarkLBActiveInDB task.\n\nStory 2010426\nTask 46823\n\nConflicts:\n\toctavia/controller/worker/v1/flows/load_balancer_flows.py\n\toctavia/controller/worker/v2/flows/load_balancer_flows.py\n\toctavia/tests/unit/controller/worker/v1/tasks/test_database_tasks.py\n\toctavia/tests/unit/controller/worker/v2/tasks/test_database_tasks.py\n\nChange-Id: Ie5fb25e292461baca38499091e8f9b9a53e3ddcd\n(cherry picked from commit 3a157c0216a4b9eca75978ce1e32790aeaa4ccab)\n(cherry picked from commit 1f2aaaa7372f7527910615653c86b50165ff4317)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/396785dadd0e1c66fcc845799cdfc814faec7d84"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/396785dadd0e1c66fcc845799cdfc814faec7d84"}]},"branch":"refs/heads/stable/zed"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
