)]}'
{"id":"openstack%2Foctavia~882278","triplet_id":"openstack%2Foctavia~stable%2Fwallaby~Ie5fb25e292461baca38499091e8f9b9a53e3ddcd","project":"openstack/octavia","branch":"stable/wallaby","topic":"fix_pool_create_single_lb_create_call","attention_set":{},"removed_from_attention_set":{"31664":{"account":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"last_update":"2023-06-28 05:42:53.000000000","reason":"Change was submitted"},"7249":{"account":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"last_update":"2023-06-21 08:40:16.000000000","reason":"\u003cGERRIT_ACCOUNT_7249\u003e replied on the change","reason_account":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"}},"29244":{"account":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"last_update":"2023-06-26 13:56:01.000000000","reason":"\u003cGERRIT_ACCOUNT_29244\u003e replied on the change","reason_account":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"}},"34429":{"account":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"last_update":"2023-06-28 05:42:53.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Ie5fb25e292461baca38499091e8f9b9a53e3ddcd","subject":"Fix pool creation with single LB create call","status":"MERGED","created":"2023-05-04 11:40:49.000000000","updated":"2023-06-28 05:44:03.000000000","submitted":"2023-06-28 05:42:53.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":3,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"882278-fix_pool_create_single_lb_create_call","meta_rev_id":"d8b4635902fc4dd71bb6025bad3629615bb14a85","_number":882278,"virtual_id_number":882278,"owner":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"value":0,"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-06-28 05:42:52.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"all":[{"value":2,"date":"2023-06-21 08:40:16.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"value":1,"date":"2023-06-23 12:16:51.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},{"value":1,"date":"2023-05-31 21:19:07.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2023-06-26 13:56:01.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"all":[{"value":0,"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"value":0,"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2023-06-26 13:56:01.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"value":0,"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}],"reviewers":{"REVIEWER":[{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-05-04 14:36:40.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-05-31 18:31:07.000000000","updated_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"reviewer":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"state":"CC"},{"updated":"2023-05-31 21:19:07.000000000","updated_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"reviewer":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"state":"REVIEWER"},{"updated":"2023-06-21 08:15:26.000000000","updated_by":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"reviewer":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"state":"REVIEWER"},{"updated":"2023-06-21 08:15:26.000000000","updated_by":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"reviewer":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"state":"REVIEWER"},{"updated":"2023-06-21 08:15:26.000000000","updated_by":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"reviewer":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"state":"REVIEWER"},{"updated":"2023-06-23 12:16:51.000000000","updated_by":{"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},"reviewer":{"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},"state":"REVIEWER"}],"messages":[{"id":"58c55e5b758c6042680ba794c3c52b43a7005612","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"date":"2023-05-04 11:40:49.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"bbf29d5d0e109b1d3734ed7128706cefa046229a","tag":"autogenerated:gerrit:abandon","author":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"date":"2023-05-04 11:42:10.000000000","message":"Abandoned\n\nwrong u/s branch","accounts_in_message":[],"_revision_number":1},{"id":"49c0165dab98ea9f0537a85840c01302e798689c","tag":"autogenerated:gerrit:restore","author":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"date":"2023-05-04 11:53:28.000000000","message":"Restored","accounts_in_message":[],"_revision_number":1},{"id":"04f8307a701c112bf1082b1ff2e825d87597a54e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"date":"2023-05-04 11:54:42.000000000","message":"Uploaded patch set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"9b43dab3b71b254df0e69cfb082e350c323977ff","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-04 14:36:40.000000000","message":"Patch Set 2: Verified-1\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/3d51a5ed374542058209fae9f10e2e37\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/ae66167737ce402ca61a56850919d303 : FAILURE in 4m 41s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/2710f7f23c8b4b14ad489afbbc5138a5 : FAILURE in 5m 32s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a868dbce7ada459e8cf620819d95f85d : FAILURE in 4m 01s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/64b83d629a3f4fe690d29307979a2ac7 : FAILURE in 5m 22s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/39e80b32e53243679287d05e4397d7d8 : FAILURE in 4m 35s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/2bc44eb8ab854156bb96a43b80904706 : FAILURE in 5m 25s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/692408b3951543ce9bd70961b477b3f1 : FAILURE in 5m 13s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/e74dbaa8adb94f0e9673b86fd9028ae7 : SUCCESS in 5m 06s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/370baa6f41204760ac5370cedd2bbf76 : FAILURE in 5m 18s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/40fd485975454a3eaef2f0dc79f545ad : SUCCESS in 6m 36s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/3bb74d2ae59d4932b499862de306cc26 : SUCCESS in 5m 38s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/c0fcf40954704fc6aee0892133c2c082 : SUCCESS in 7m 21s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/4b7a5b2d90244e609871088e4b6ce89b : SUCCESS in 1h 02m 12s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/09375ff842054f87bd0c5cf9dab95dd0 : SUCCESS in 1h 51m 36s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/215daa37e29d472d82ac900e4329a3bf : SUCCESS in 40m 13s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/175e67121e0844f8bea6f722f567557f : SUCCESS in 36m 09s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/8d91feaa14a0438b9e270f0da5f7871a : SUCCESS in 36m 42s\n- octavia-v2-dsvm-tls-barbican-fips https://zuul.opendev.org/t/openstack/build/b0392eaff43d471e93440049da4cbdfa : FAILURE in 10m 44s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/3608fddac1f14097bca8d2d56c8b280c : SUCCESS in 1h 54m 46s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/5bbd9e3e801e4702bc1a016bfe5c60d8 : FAILURE in 38m 07s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/ca8745ae737940a093c45fc9e9073e3f : SUCCESS in 2h 07m 43s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/b695897d9f544e699457ac393c8588d0 : FAILURE in 2h 35m 57s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/d3431b29b4a142c49438a0edd867b426 : TIMED_OUT in 2h 33m 25s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8 https://zuul.opendev.org/t/openstack/build/0a4b53abdaef44c7a11d9f1dd25d8113 : NODE_FAILURE Node request 300-0021133598 failed in 0s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"9efa4cd2370e19f0a0c3f9a6c782fc0ab2c1dc91","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"date":"2023-05-08 11:16:53.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":3},{"id":"5d051beb48cd78e7dc0df229034128a3092f1656","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-08 12:35:28.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/95076335d3f04e479f08585fc075be28\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/bdfa8ac203fb44a2aca2012e98a9bf4a : SUCCESS in 8m 04s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/ad0cffa6131e4cf8be98c7d8ef58f640 : SUCCESS in 7m 42s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6babb499da144fe2b8b258a54071e09a : SUCCESS in 5m 40s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/006be2675bb74b8099b3219a8b77a3db : SUCCESS in 7m 40s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/3e2f3f47d0074aff9cc2942b294fa604 : SUCCESS in 5m 28s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/b1c6592f9aa7404bb9c63ac70e34dc72 : SUCCESS in 4m 39s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1d736fbef2d344059bc7c4311a6bf2a6 : SUCCESS in 10m 34s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/ea26172cf8e642c5b02abbecc2d4a2b6 : SUCCESS in 4m 50s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/98e129a43a4445deb20d5fa794fcb07d : SUCCESS in 5m 47s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/dc7a05838dc3437a823ca07064069a43 : SUCCESS in 7m 32s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/a129f5d5afee434faad324eab5987c02 : SUCCESS in 4m 42s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/676b9844904a41c3aa216068f73f8537 : SUCCESS in 6m 39s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/af552f600c3e4021b6288ef43c10443c : SUCCESS in 1h 09m 41s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/14f0ff36724e488a89f50a5dbe255723 : FAILURE in 5m 41s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/c6593bcde8eb4ef8882839b196f4f879 : FAILURE in 5m 51s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/8e53393aac0f4107a0e1b268ae94cb7e : FAILURE in 7m 29s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/2669bc09a9684c7bb8cff450b2c0e4e2 : FAILURE in 5m 24s\n- octavia-v2-dsvm-tls-barbican-fips https://zuul.opendev.org/t/openstack/build/041b81bc80c047f19a88f36bf35d236e : FAILURE in 10m 32s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/5f157a65d4b94dcc94e6736d6bd2664f : FAILURE in 6m 14s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/b2c12c13c72646a1a14ba39266ced1b7 : FAILURE in 9m 18s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/f1cde85096124b0e8e123133e02e3afe : FAILURE in 7m 39s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/9bafae94801f4b099b8ff9a31adc0342 : FAILURE in 5m 18s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/1a13d1cbcbbc4179a1fb465e702c563c : FAILURE in 5m 14s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8 https://zuul.opendev.org/t/openstack/build/784429ec0eaf446abdc4128e14d3b4c0 : NODE_FAILURE Node request 300-0021163052 failed in 0s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"ba524eefd4337f64c286db361b30ad251e916d0c","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2023-05-31 18:31:07.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"a196d1021eadc12f3cdc4fa9bbafff79cd390b2f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-31 21:13:45.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9152f08aeaee4a93b7f3a7bb77b1e72a\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/b851cb7292434facadcd9fccd1c327ba : SUCCESS in 10m 25s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/f1ddc72a6dc24b4a95973c0fed64a8be : SUCCESS in 9m 16s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0db4ce54306e4c59a0a1d5fc09680951 : SUCCESS in 5m 35s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/4d26e6adb88d4a6aa462dbface42bff1 : SUCCESS in 5m 20s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/f59b647f886547f5a994cc415deabe25 : SUCCESS in 5m 24s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/c25604572c4745dc9f93ebf2534965a7 : SUCCESS in 4m 52s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/74c3886954144e178439c7da4dde28ba : SUCCESS in 14m 12s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/e5fcdd122c294a399439d67cb8a9c91c : SUCCESS in 8m 09s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/67e63814f7e14b279ae67b9dac205e43 : SUCCESS in 8m 27s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/1be5e6aa585e4de19c14a2492a393859 : SUCCESS in 7m 04s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/7aed6441dea34f9fac8a979c04282156 : SUCCESS in 4m 57s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/8b58d21239714e8fb6122f4e467735c7 : SUCCESS in 8m 11s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/144087ba02014b8aa5669a56a312d76c : SUCCESS in 1h 25m 35s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/b7852ea687914749b309a43785fa3449 : SUCCESS in 2h 07m 47s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/9183908c85b74674b05962ce5e773bdf : SUCCESS in 45m 48s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/9c3fdc44be9b42fc8829318f86dc4d6f : SUCCESS in 35m 51s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/d17431970ca1407aab38abf9345da3ce : SUCCESS in 39m 55s\n- octavia-v2-dsvm-tls-barbican-fips https://zuul.opendev.org/t/openstack/build/b08c907d0b644b1faea4fd0a6cce7e5b : FAILURE in 9m 20s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/57a01b55dc574a7aaa3bef76742c7e63 : SUCCESS in 1h 56m 19s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/f7a06056e6d84d76ab9bc8c194876453 : FAILURE in 18m 47s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/1226d34c53a248a89d7adf16adc0d259 : FAILURE in 8m 22s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/1b2f88598eda4b6fbfa70acb5b48f350 : TIMED_OUT in 2h 33m 43s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/1b68e2f9470c427f8410e0255e195d35 : SUCCESS in 2h 23m 42s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8 https://zuul.opendev.org/t/openstack/build/b1f8313458fa49b98d72d6c763452372 : NODE_FAILURE Node request 300-0021364058 failed in 0s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"3c20ab8cd1120c204e10a391e691abca4789dc48","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2023-05-31 21:19:07.000000000","message":"Patch Set 3: Code-Review+1","accounts_in_message":[],"_revision_number":3},{"id":"ad1831138cc1f47406a315c278601bd956ea5550","author":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"date":"2023-06-21 08:40:16.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"4b301d3dbeed21e62418e3fea0dcb9547fb2e6f8","author":{"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},"date":"2023-06-23 12:16:51.000000000","message":"Patch Set 3: Code-Review+1","accounts_in_message":[],"_revision_number":3},{"id":"988b90320af4b7f0015adfd6598379c5a21501ef","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2023-06-26 13:56:01.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"ce97cfe0e403c36d6459ce383cf99f7ba08ed303","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-26 13:56:46.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"eed264e2c99a0542fa16e435095aeb99caa44fea","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-26 16:12:52.000000000","message":"Patch Set 3: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/00be123077a640eca9a5546cd2e0541f\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/a105197b30634d7fa8cc9438be83f33b : SUCCESS in 8m 41s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0fec4b018ca248888853536aaf2f98fb : SUCCESS in 5m 28s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/3681b223e8ae40b7979d4af30653266a : SUCCESS in 5m 20s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/4704d2dca4ba49d1bb3ed73800b2e953 : SUCCESS in 10m 22s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ccdf5e53157844d583d90ecff3c097a8 : SUCCESS in 11m 35s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/3a7e3ce670894048addf5d4935024f4d : SUCCESS in 4m 42s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/069289a04f224ce987bd056cb91cb6a3 : SUCCESS in 5m 51s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/5ff279334111420c9f843d0a5db61f9e : SUCCESS in 6m 03s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/2cb00dfb515d402e8b33b42f4be7310d : CANCELED\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/fe290fa62f174627b75a68d1f237584d : CANCELED\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/9e834c358d574067a2f89e19961d119a : CANCELED\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/75a18a0a955740c9b2da96815e1299bf : CANCELED\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/cf564a44fc4849f3832f2b9edb760d38 : FAILURE in 27m 27s","accounts_in_message":[],"_revision_number":3},{"id":"0ca30ec5265878d29e655eec7ed4dd3999092528","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2023-06-27 19:52:53.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"67cc47455f5e4cf6b7879b25414012d08e8e9ee5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-27 22:29:28.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/4c6db5c6104942a29c53640f94c44e54\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/ce0b7cb31df340aab868493372992380 : SUCCESS in 7m 34s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/b84080d5e9f4488ca241c8082c19d9e1 : SUCCESS in 8m 12s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4b09e06114e342508ae1a3a32bd33e12 : SUCCESS in 4m 17s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/5dc548e705684d0d94b930639c796d7c : SUCCESS in 4m 52s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/6119062476654fb98c5ddd24e75654a7 : SUCCESS in 5m 07s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/5b8b3970084f4942b950552514e0ca48 : SUCCESS in 5m 00s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6077d290cdcb4655aa63c9653699ac4c : SUCCESS in 11m 55s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/08dbbce3180142879c09546227e70d92 : SUCCESS in 4m 51s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/91592f25d21f4fd88969662da9077cde : SUCCESS in 4m 56s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/1fda6a362d21484896045b8e5041bdef : FAILURE in 7m 02s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/d26ca59cc54c40289a992d35c6f8b883 : SUCCESS in 5m 18s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/d3ba40691e8241afb6f3ef2cd60075be : SUCCESS in 7m 05s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/353a7950d46048818eae7943cc991b32 : SUCCESS in 1h 22m 29s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/eba05c2cb4ac4b5199fd90b28c79a5d8 : SUCCESS in 1h 19m 39s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/f606c94b0ac34f579a0747325abc3efd : SUCCESS in 35m 44s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/2fc0d0ccb15e4f72b274c627bfe4e51e : SUCCESS in 40m 37s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/8e457369a9c34957a264afe6f36fc562 : SUCCESS in 33m 35s\n- octavia-v2-dsvm-tls-barbican-fips https://zuul.opendev.org/t/openstack/build/67d22e06ca5b4ba198d076d677f5c589 : FAILURE in 8m 25s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/f1a1fdeb3f4d45b4b39d2f7644b73239 : SUCCESS in 1h 31m 55s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/683270db4dc34187ab15a9b32b76e655 : FAILURE in 11m 06s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/319f936dccea4d1ba30940f7f7646ce8 : FAILURE in 5m 37s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/a660ebaf5bf543c0bf9a71d4ceaf6577 : TIMED_OUT in 2h 34m 13s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/aebedac2aa2041ec8cdd333f8ac7ad0d : SUCCESS in 2h 08m 23s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8 https://zuul.opendev.org/t/openstack/build/e868e50dab12450693af6ff3cd7833da : NODE_FAILURE Node request 300-0021557268 failed in 0s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"567867555130e4e1fa04db7618234244ebeb50d8","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2023-06-28 01:32:23.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"8f0200446406f372956aa3d960d27e1e771cab7b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-28 04:09:36.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/20e34975ef134e9cac2cd4c9ccea5afa\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/c299cf77171b425ebbf0502c406a1cff : SUCCESS in 7m 22s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/097483d355364f9eb08ca0be2f92821b : SUCCESS in 6m 38s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ce6880bae7ae42c2a48eeddaf6c60de1 : SUCCESS in 4m 40s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/419ea3ac6a664500b86396550ee9cb39 : SUCCESS in 4m 35s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/8b9a8fdfaab746a2875dc4ccaff424b0 : SUCCESS in 5m 55s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/c1f4bdcfd4e346728dd462dfb12ff69a : SUCCESS in 5m 17s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/12db5a357ed04fa58245e705ec3afe31 : SUCCESS in 9m 21s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/412c70c34eb646ac8e1900fdfe3e2b8a : SUCCESS in 6m 11s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/7e5c73e7f12a4e1385b603c4ca6c77fb : SUCCESS in 5m 27s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/79d1e91e59e24e8f81d1b4117afae616 : SUCCESS in 8m 12s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/3a33a034ffb94fdb805418d48327158a : SUCCESS in 4m 07s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/18d78ef87fe6485eb0002d967822c883 : SUCCESS in 6m 45s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/1e71016174a54ec4a434e4e971b95f51 : SUCCESS in 1h 05m 54s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/0500f0a4ef7c42188f3dc77f6f7d1570 : SUCCESS in 2h 16m 48s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/9e558d3a12154665a280125f402dcd66 : SUCCESS in 39m 03s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/74a4c5c3c3ae4fec9e71035bc6806479 : SUCCESS in 32m 40s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/bcb1253802504880a2d42437e8960355 : SUCCESS in 31m 42s\n- octavia-v2-dsvm-tls-barbican-fips https://zuul.opendev.org/t/openstack/build/6abcfdb2fe7b4b9f82886fd12172b5e6 : FAILURE in 8m 44s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/21210a0b832a4d86a961f9c21ee1a79c : SUCCESS in 2h 28m 24s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/82321597029943969f2015d940c8b17d : FAILURE in 9m 07s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/7c456a1fb6d540bc91052d4082409970 : FAILURE in 4m 46s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/94df1afd84f645e7b84daefcbfdf40a8 : TIMED_OUT in 2h 33m 38s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/c5de231065e941c49278d61340edfe14 : SUCCESS in 2h 17m 00s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8 https://zuul.opendev.org/t/openstack/build/a1d0ffcbd40a45d584a0884c65e0877a : NODE_FAILURE Node request 300-0021559096 failed in 0s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"287496a00a8cd835538f9d7969d4ee190e68dc8c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-28 04:10:10.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"9fad89166924be2e6a20286812f856a0a08159bc","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-28 05:42:52.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c02ecb7ddce7420b9a4660813cc4c2be\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/e51a76b836c34f198a2dbd1b64c12ddc : SUCCESS in 7m 45s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1785b1e6fee0414193786abc0e5364e4 : SUCCESS in 4m 36s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/295ae8c762ab44d78bcbd8f657b13712 : SUCCESS in 5m 06s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/4f633ed1d63c41d0b21aafce927d5236 : SUCCESS in 4m 25s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2199a71768c442b399692147af3f9e03 : SUCCESS in 20m 39s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/4487d6cfc0d94b6e82302a1f30b553b3 : SUCCESS in 11m 29s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/dbbef606da4d4750822a2dadd9f4cc83 : SUCCESS in 12m 01s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/47fe1a8bc791429a8b612a248ddc1a51 : SUCCESS in 15m 38s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/0a730683ebc44d4a9f176d2ee6ec4863 : SUCCESS in 1h 04m 43s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/1baf9e3dabdb491089824ff20ddb396c : SUCCESS in 1h 32m 10s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/fe35621576834e81a5b73b2be0edad00 : SUCCESS in 36m 31s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/35706c3fb5db4f3ba5c0c9a3c59fd7cc : SUCCESS in 33m 16s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/e922a88f5b364e17bcd0d40ca6e6a3be : SUCCESS in 35m 34s","accounts_in_message":[],"_revision_number":3},{"id":"de9e95fef6db02596d839a03b61e149e82e42ce2","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-28 05:42:53.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":3},{"id":"d8b4635902fc4dd71bb6025bad3629615bb14a85","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-28 05:44:03.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2f4f439540564d5c9533f85f7ee1d622\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/13bd1b7e66b845df831fd4ccc74be6f1 : SUCCESS in 53s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/54a14e81ac34447dab534c8349cbaf17 : SUCCESS in 38s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"152832af17e3b158279f88a39c801f998e0193f3","revisions":{"6e8e67ec7992adc251864dde5dbaa715dbda385d":{"kind":"REWORK","_number":1,"created":"2023-05-04 11:40:49.000000000","uploader":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"ref":"refs/changes/78/882278/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/78/882278/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/78/882278/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/78/882278/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/78/882278/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/78/882278/1"}}},"commit":{"parents":[{"commit":"99d7e2b9f2a99259d075bda462f593fe91d32442","subject":"Merge \"Prevent incorrect reschedule of resumed tasks with jobboard\" into stable/wallaby","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/99d7e2b9f2a99259d075bda462f593fe91d32442"}]}],"author":{"name":"Omer","email":"oschwart@redhat.com","date":"2022-11-08 16:32:17.000000000","tz":60},"committer":{"name":"Omer","email":"oschwart@redhat.com","date":"2023-05-04 11:40:36.000000000","tz":120},"subject":"Fix pool creation with single LB create call","message":"Fix pool creation with single LB create call\n\nSo far, creating a populated LB (or an LB tree) with non attached pools,\nleft the pools on PENDING_CREATE provisioning_state.\n\nThis patch adds the option of marking non attached pools to be active on\nthe MarkLBActiveInDB task.\n\nStory 2010426\nTask 46823\n\nConflicts:\n\toctavia/controller/worker/v1/flows/load_balancer_flows.py\n\toctavia/controller/worker/v2/flows/load_balancer_flows.py\n\toctavia/tests/unit/controller/worker/v1/tasks/test_database_tasks.py\n\toctavia/tests/unit/controller/worker/v2/tasks/test_database_tasks.py\n\nChange-Id: Ie5fb25e292461baca38499091e8f9b9a53e3ddcd\n(cherry picked from commit 3a157c0216a4b9eca75978ce1e32790aeaa4ccab)\n(cherry picked from commit 1f2aaaa7372f7527910615653c86b50165ff4317)\n(cherry picked from commit 396785dadd0e1c66fcc845799cdfc814faec7d84)\n(cherry picked from commit 9a40fa7e3c8b211812a04b3b02cc637cb0fa548f)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/6e8e67ec7992adc251864dde5dbaa715dbda385d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/6e8e67ec7992adc251864dde5dbaa715dbda385d"}]},"branch":"refs/heads/stable/wallaby"},"077276469a1888a9ffe36b2720eeba2c67b78e52":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2023-05-04 11:54:42.000000000","uploader":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"ref":"refs/changes/78/882278/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/78/882278/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/78/882278/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/78/882278/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/78/882278/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/78/882278/2"}}},"commit":{"parents":[{"commit":"99d7e2b9f2a99259d075bda462f593fe91d32442","subject":"Merge \"Prevent incorrect reschedule of resumed tasks with jobboard\" into stable/wallaby","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/99d7e2b9f2a99259d075bda462f593fe91d32442"}]}],"author":{"name":"Omer","email":"oschwart@redhat.com","date":"2022-11-08 16:32:17.000000000","tz":60},"committer":{"name":"Omer","email":"oschwart@redhat.com","date":"2023-05-04 11:54:31.000000000","tz":120},"subject":"Fix pool creation with single LB create call","message":"Fix pool creation with single LB create call\n\nSo far, creating a populated LB (or an LB tree) with non attached pools,\nleft the pools on PENDING_CREATE provisioning_state.\n\nThis patch adds the option of marking non attached pools to be active on\nthe MarkLBActiveInDB task.\n\nStory 2010426\nTask 46823\n\nConflicts:\n\toctavia/controller/worker/v1/flows/load_balancer_flows.py\n\toctavia/controller/worker/v2/flows/load_balancer_flows.py\n\toctavia/tests/unit/controller/worker/v1/tasks/test_database_tasks.py\n\toctavia/tests/unit/controller/worker/v2/tasks/test_database_tasks.py\n\nChange-Id: Ie5fb25e292461baca38499091e8f9b9a53e3ddcd\n(cherry picked from commit 3a157c0216a4b9eca75978ce1e32790aeaa4ccab)\n(cherry picked from commit 1f2aaaa7372f7527910615653c86b50165ff4317)\n(cherry picked from commit 396785dadd0e1c66fcc845799cdfc814faec7d84)\n(cherry picked from commit 9a40fa7e3c8b211812a04b3b02cc637cb0fa548f)\n(cherry picked from commit f0df69df73b41440b9b3e6da3b094e87bbb154b6)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/077276469a1888a9ffe36b2720eeba2c67b78e52"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/077276469a1888a9ffe36b2720eeba2c67b78e52"}]},"branch":"refs/heads/stable/wallaby"},"152832af17e3b158279f88a39c801f998e0193f3":{"kind":"REWORK","_number":3,"created":"2023-05-08 11:16:53.000000000","uploader":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"ref":"refs/changes/78/882278/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/78/882278/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/78/882278/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/78/882278/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/78/882278/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/78/882278/3"}}},"commit":{"parents":[{"commit":"99d7e2b9f2a99259d075bda462f593fe91d32442","subject":"Merge \"Prevent incorrect reschedule of resumed tasks with jobboard\" into stable/wallaby","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/99d7e2b9f2a99259d075bda462f593fe91d32442"}]}],"author":{"name":"Omer","email":"oschwart@redhat.com","date":"2022-11-08 16:32:17.000000000","tz":60},"committer":{"name":"Omer","email":"oschwart@redhat.com","date":"2023-05-08 11:16:44.000000000","tz":120},"subject":"Fix pool creation with single LB create call","message":"Fix pool creation with single LB create call\n\nSo far, creating a populated LB (or an LB tree) with non attached pools,\nleft the pools on PENDING_CREATE provisioning_state.\n\nThis patch adds the option of marking non attached pools to be active on\nthe MarkLBActiveInDB task.\n\nStory 2010426\nTask 46823\n\nConflicts:\n\toctavia/controller/worker/v1/flows/load_balancer_flows.py\n\toctavia/controller/worker/v2/flows/load_balancer_flows.py\n\toctavia/tests/unit/controller/worker/v1/tasks/test_database_tasks.py\n\toctavia/tests/unit/controller/worker/v2/tasks/test_database_tasks.py\n\nChange-Id: Ie5fb25e292461baca38499091e8f9b9a53e3ddcd\n(cherry picked from commit 3a157c0216a4b9eca75978ce1e32790aeaa4ccab)\n(cherry picked from commit 1f2aaaa7372f7527910615653c86b50165ff4317)\n(cherry picked from commit 396785dadd0e1c66fcc845799cdfc814faec7d84)\n(cherry picked from commit 9a40fa7e3c8b211812a04b3b02cc637cb0fa548f)\n(cherry picked from commit f0df69df73b41440b9b3e6da3b094e87bbb154b6)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/152832af17e3b158279f88a39c801f998e0193f3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/152832af17e3b158279f88a39c801f998e0193f3"}]},"branch":"refs/heads/stable/wallaby"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
