)]}'
{"id":"openstack%2Foctavia~883063","triplet_id":"openstack%2Foctavia~master~I35797e252cb48eb23d82e1f89c88c6f1ae5fb02a","project":"openstack/octavia","branch":"master","attention_set":{},"removed_from_attention_set":{"6577":{"account":{"_account_id":6577,"name":"Sergey Kraynev","email":"sergejyit@gmail.com","username":"skraynev"},"last_update":"2024-11-08 16:02:53.000000000","reason":"Change was submitted"},"11628":{"account":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"last_update":"2024-08-27 02:56:49.000000000","reason":"\u003cGERRIT_ACCOUNT_11628\u003e replied on the change","reason_account":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}},"29244":{"account":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"last_update":"2024-05-21 12:04:29.000000000","reason":"Removed by \u003cGERRIT_ACCOUNT_29244\u003e by clicking the attention icon","reason_account":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"}},"34429":{"account":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"last_update":"2024-06-12 16:14:31.000000000","reason":"removed on reply"}},"hashtags":[],"change_id":"I35797e252cb48eb23d82e1f89c88c6f1ae5fb02a","subject":"Add option to limit recursion in to_data_model","status":"MERGED","created":"2023-05-12 12:29:26.000000000","updated":"2024-11-08 16:04:04.000000000","submitted":"2024-11-08 16:02:53.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":42,"unresolved_comment_count":1,"has_review_started":true,"submission_id":"883063","meta_rev_id":"898b07ea98d372b3522b983011be554c7a935b37","_number":883063,"virtual_id_number":883063,"owner":{"_account_id":6577,"name":"Sergey Kraynev","email":"sergejyit@gmail.com","username":"skraynev"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-11-08 16:02:53.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"all":[{"value":2,"date":"2024-10-23 17:36:41.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":2,"date":"2024-11-08 13:58:00.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2024-08-23 05:23:06.000000000","_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"all":[{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":1,"date":"2024-11-08 13:58:00.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-05-12 12:30:02.000000000","updated_by":{"_account_id":6577,"name":"Sergey Kraynev","email":"sergejyit@gmail.com","username":"skraynev"},"reviewer":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"state":"REVIEWER"},{"updated":"2023-05-12 15:03:43.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-05-15 06:39:53.000000000","updated_by":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"reviewer":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"state":"CC"},{"updated":"2023-09-05 18:33:08.000000000","updated_by":{"_account_id":6577,"name":"Sergey Kraynev","email":"sergejyit@gmail.com","username":"skraynev"},"reviewer":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"state":"REVIEWER"},{"updated":"2023-09-05 18:33:08.000000000","updated_by":{"_account_id":6577,"name":"Sergey Kraynev","email":"sergejyit@gmail.com","username":"skraynev"},"reviewer":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"state":"REVIEWER"}],"messages":[{"id":"612d737f0825510045fcdfc928835fc8b71d164f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":6577,"name":"Sergey Kraynev","email":"sergejyit@gmail.com","username":"skraynev"},"date":"2023-05-12 12:29:26.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"1027c6c97e9584d0ce21e7742c4b5cd8970cffdb","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-12 15:03:43.000000000","message":"Patch Set 1: Verified-1\n\n(7 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/ee0aacb5426a4f84964cdd09072403ab\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/0e6bc76d910c4d568c229404f007e654 : FAILURE in 6m 19s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2f15c7ad0a8e4b9f906bf8fd9ffdf85d : FAILURE in 3m 46s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/de5a70ef14424aa89c2ce2e016852e0f : SUCCESS in 4m 32s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/472a20011ddd49eba9e389b7779c1ac5 : SUCCESS in 4m 19s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/6bf39b34b5454685b08291858791033d : SUCCESS in 4m 22s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/56a48d2449d44360ba5144eb673abaaf : SUCCESS in 10m 59s\n- octavia-tox-py39-tips https://zuul.opendev.org/t/openstack/build/439f8748e8294e3788792dac8dad668a : SUCCESS in 5m 08s\n- octavia-tox-functional-py39-tips https://zuul.opendev.org/t/openstack/build/c8ebd45156f643fc9a0df519f60509b0 : FAILURE in 4m 54s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/8b67315ba301499bbb4934552cc680b4 : SUCCESS in 4m 07s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/641cc7385265421e9edae49708e33395 : FAILURE in 5m 48s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/931ab2f03dbf461394c560b0ef682ec2 : FAILURE in 5m 02s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/c77df9d6a1994efea3f9dfe094ef1b76 : FAILURE in 5m 13s (non-voting)\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/3ee9d4f7bbd6457a99612837180fa120 : FAILURE in 1h 44m 41s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/1dc5f380bd294b919c6835050df199f8 : TIMED_OUT in 2h 32m 42s\n- octavia-v2-dsvm-scenario-jobboard https://zuul.opendev.org/t/openstack/build/4de3420039ee447b85adfc2555417274 : TIMED_OUT in 2h 32m 53s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/394a89e882ef4044b6d35d1003b963ca : SUCCESS in 38m 03s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/c72d2f8fc6f3476cb99c15b31845e180 : FAILURE in 33m 36s\n- octavia-v2-dsvm-tls-barbican-fips https://zuul.opendev.org/t/openstack/build/fcea04785fb34e29b369fdebdfc14921 : SUCCESS in 57m 14s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/56bdd652d01d4b7998272f8b1d3d188c : FAILURE in 31m 35s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/df51661f773b456e9e43a1dfdd97b987 : TIMED_OUT in 2h 32m 47s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/e49a806d5d274882883754763e3b97bd : FAILURE in 2h 19m 29s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/969cdb4f4b1f4c2588525d1aeb18406f : FAILURE in 2h 01m 16s (non-voting)\n- octavia-v2-dsvm-scenario-centos-9-stream https://zuul.opendev.org/t/openstack/build/51c39376d4b0403fbc86de9be5ebfb31 : FAILURE in 1h 52m 52s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"7582e9b8ee870c0b5596889d39d3e55966a9cfad","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":6577,"name":"Sergey Kraynev","email":"sergejyit@gmail.com","username":"skraynev"},"date":"2023-05-12 15:08:43.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"e7236ff2e80b16a8078f3ae646bb20eef7380ce1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-12 17:55:09.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/4db009c8558f4465a0385fc35a9d2ee5\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/03c52c0222234193a9af334947d36050 : SUCCESS in 11m 31s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5afaa961b3c74dba9c2ceac5e13755bc : SUCCESS in 4m 00s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/2d3f3346175b4da39f584381f56df42f : SUCCESS in 4m 10s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/c46e4e4d8a5c4ad091e4c2840930bdf3 : SUCCESS in 4m 30s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/caae247cdbda4347aa511063d99506b5 : SUCCESS in 4m 43s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c034051e81924d6cbb6a4e77e2066ada : SUCCESS in 11m 45s\n- octavia-tox-py39-tips https://zuul.opendev.org/t/openstack/build/dfee9f9fa41a41768cec0c308c74489a : SUCCESS in 5m 05s\n- octavia-tox-functional-py39-tips https://zuul.opendev.org/t/openstack/build/afcaa6777f434722af3530e942f79938 : SUCCESS in 7m 27s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/d3014f48c5dc419e95ab74aa2b81de99 : SUCCESS in 4m 10s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/b2d35691c9c24a0088e9bde1bcf32949 : SUCCESS in 6m 59s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/c6c3370ceae042a6a343bfc0d1b79123 : SUCCESS in 7m 42s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/e5d009fcdb0140a08e6cb71210cef96f : SUCCESS in 6m 56s (non-voting)\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/6fe3631d883843e78f61bc4bd49749e8 : SUCCESS in 1h 19m 12s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/d022b993be49489fa9627c08c7fbc603 : TIMED_OUT in 2h 32m 45s\n- octavia-v2-dsvm-scenario-jobboard https://zuul.opendev.org/t/openstack/build/e8e55e84095a4aeaace15302dcd02d9a : TIMED_OUT in 2h 32m 45s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/7f08fc8e8f4a428ca0f6abc2268fffed : SUCCESS in 36m 47s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/4bf781ea748f4cbe9510bfa1f09edea1 : SUCCESS in 37m 16s\n- octavia-v2-dsvm-tls-barbican-fips https://zuul.opendev.org/t/openstack/build/16a92f5d59a04c8daf565ffbbd2942d9 : SUCCESS in 1h 00m 11s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/e1c7af9277574891860a2c15ec6ef6f3 : SUCCESS in 33m 53s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/581c219bf91a4a318a818066e3efed3c : TIMED_OUT in 2h 32m 52s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/eea5fc7f30dd4847994d03aee55fc9f6 : SUCCESS in 2h 45m 09s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/097a222414c74817a93bec4e69e22aed : TIMED_OUT in 2h 32m 30s (non-voting)\n- octavia-v2-dsvm-scenario-centos-9-stream https://zuul.opendev.org/t/openstack/build/ce1d15e8d3af40689bcc46697d1ab282 : SUCCESS in 1h 59m 55s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"088ed5348455753858a9fc54a06f962049b9c387","author":{"_account_id":6577,"name":"Sergey Kraynev","email":"sergejyit@gmail.com","username":"skraynev"},"date":"2023-05-13 06:16:33.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"0b7e7eadd04789e79a375e10455917c8b5787240","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-13 08:57:09.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/0e70a47a45f54c06bd09b10a1718622b\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/a3f3c6e5a68e4c87af24f596200ade98 : SUCCESS in 10m 23s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3589a3d55e6f40238ae536b619b2cc1f : SUCCESS in 4m 50s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/5bcf3ccf31654731a3f9b33868423bc9 : SUCCESS in 4m 18s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/eba63d139ba44437b49b17bb292bc521 : SUCCESS in 4m 33s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/768c936710fa4fe58404011bdf213608 : SUCCESS in 4m 33s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b632dcbd390a4d5798926a9a49461c18 : SUCCESS in 10m 31s\n- octavia-tox-py39-tips https://zuul.opendev.org/t/openstack/build/adc1e13b9c6d4e2aa6bda171b2e31185 : SUCCESS in 5m 44s\n- octavia-tox-functional-py39-tips https://zuul.opendev.org/t/openstack/build/d286a9044fe9494ebc5fc83a67ba31be : SUCCESS in 8m 07s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/33fdd2645c9c49589f768a2dc48c2fb5 : SUCCESS in 3m 38s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/9c3d75382713422abf8d98ead0c8e528 : SUCCESS in 5m 39s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/176a0758912c4272b9e60c5437c7a8c9 : SUCCESS in 5m 37s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/f1909dfb7e1742839fb8e7c28dd1fcc6 : SUCCESS in 7m 41s (non-voting)\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/2b6674815d9b4378a8e0909855367ca3 : SUCCESS in 1h 31m 45s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/763eaff3bc00431a82ee4fe64cca08b8 : TIMED_OUT in 2h 32m 44s\n- octavia-v2-dsvm-scenario-jobboard https://zuul.opendev.org/t/openstack/build/ae20f9925f4b44a6b0f25e9b1770366e : TIMED_OUT in 2h 32m 42s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/eb57eab3830147738c4fbd80bad4d4fd : SUCCESS in 45m 20s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/7a61b1d43c884c299b11d7221510d75d : SUCCESS in 37m 53s\n- octavia-v2-dsvm-tls-barbican-fips https://zuul.opendev.org/t/openstack/build/af4e111c768948319527254e2d4648db : SUCCESS in 52m 25s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/88c2760ba89d40b585563344350eb8bb : SUCCESS in 42m 24s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/b826e3cdc3474892952852413fd42e24 : SUCCESS in 1h 59m 19s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/cb8543355f2e497fa7f7ef6bf2c2d88a : SUCCESS in 2h 38m 26s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/327ab409e4e549778019c29aef8bb8dd : TIMED_OUT in 2h 32m 42s (non-voting)\n- octavia-v2-dsvm-scenario-centos-9-stream https://zuul.opendev.org/t/openstack/build/441c5ce0000a418eae9954d691283a3d : SUCCESS in 2h 04m 36s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"94551e3a810b9c4523c635144cda3ef5b9ba2498","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2023-05-15 06:39:53.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"4ad223f9318ebe666d1a8591601f98efbe395563","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-15 09:28:02.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/5a4df5706803439cbb68fa2dddd587c3\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/1954e4a9ab684ba1b05b6fb28f78ab93 : SUCCESS in 9m 26s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c10af847aa694a43b2650f1f20f0370f : SUCCESS in 4m 46s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/1e69ce0908f14313b8199d32678d0174 : SUCCESS in 4m 24s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/13d633ca50d74a309700cfba320c0beb : SUCCESS in 4m 16s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/6cbbe75f894b4c55b0aa06f6d0967f93 : SUCCESS in 4m 19s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0acb214c39df48ab939b0ca3b6458efb : SUCCESS in 10m 56s\n- octavia-tox-py39-tips https://zuul.opendev.org/t/openstack/build/70091326c7c94bccba186f43f2c8a867 : SUCCESS in 4m 37s\n- octavia-tox-functional-py39-tips https://zuul.opendev.org/t/openstack/build/ab387978943d41aea5c484a6085de699 : SUCCESS in 7m 11s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/f06ecd73b0d94327b7eb1add5d001ed5 : SUCCESS in 4m 37s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/e64b296ad027418bb7a953a6d2cb8bb9 : SUCCESS in 7m 25s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/4ebea9062831474aaab84bccef302756 : SUCCESS in 7m 39s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/56038e14f911435b95db2aa16f47f3af : SUCCESS in 7m 08s (non-voting)\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/071864e2e38845369b3650bd913cde39 : SUCCESS in 1h 32m 28s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/586b91939b424c558545de391d3a8555 : TIMED_OUT in 2h 32m 51s\n- octavia-v2-dsvm-scenario-jobboard https://zuul.opendev.org/t/openstack/build/ef8628bd88a94383971c595becafbc52 : TIMED_OUT in 2h 32m 49s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/48924e25e7e3497aa81c63e895b9e87a : SUCCESS in 40m 58s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/85b22c0615c34d058edaefbeb3eff52f : SUCCESS in 41m 48s\n- octavia-v2-dsvm-tls-barbican-fips https://zuul.opendev.org/t/openstack/build/fbf51f362ab843ac9ee51607c274ce7f : SUCCESS in 1h 10m 39s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/6e5158b3adef4ff5b73444e7669fb66e : SUCCESS in 33m 01s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/8add45fee26f47d7930b42233dfc4dc2 : TIMED_OUT in 2h 32m 54s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/b1ac84db71da44ebbb717a1e19abea35 : SUCCESS in 2h 46m 29s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/4ec0e7d9e48d42789ff082a67715c802 : TIMED_OUT in 2h 32m 40s (non-voting)\n- octavia-v2-dsvm-scenario-centos-9-stream https://zuul.opendev.org/t/openstack/build/2b6a7c656696451ab427d628081a0a66 : NODE_FAILURE Node request 300-0021211075 failed in 0s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"12366ad3d855fde1d461186a049ca6cb1d02b4ff","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2023-07-27 11:48:47.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"751f63d30ba63b8349dfe211ae89ac25800f2f81","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-27 14:59:18.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/74ab5a76a1c24d66a8f4a9cd1d668958\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/66dedf3830634f79a36b285c5c9ed4b5 : SUCCESS in 10m 41s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c2d4e2346ead4448a89da15106fb7b39 : SUCCESS in 6m 08s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/44dc7417619a4d65a0fd3cd826507781 : SUCCESS in 4m 50s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/ea1ff7da8dbd43fc98ca473f03032c91 : SUCCESS in 7m 00s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/e89e556da70a41d88a56bb3f6223591a : SUCCESS in 4m 07s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/778053d7a25e4c139fe85e40ef134f9f : SUCCESS in 12m 09s\n- octavia-tox-py39-tips https://zuul.opendev.org/t/openstack/build/9d8e29647fcf4d3fba852af9ce68d4f0 : SUCCESS in 5m 54s\n- octavia-tox-functional-py39-tips https://zuul.opendev.org/t/openstack/build/ed7f5437aa1c460cbc6a80c1aac0c74c : SUCCESS in 6m 31s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/58e926a7921c4176903782bc7958c7d8 : SUCCESS in 4m 27s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/9dd8db1f57fa4c1f8e6bb27ea548a9d9 : SUCCESS in 8m 03s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/676a63469b614d648e8b9090ff48867a : SUCCESS in 7m 59s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/384297901a6247fb96e5ec05b27aa178 : SUCCESS in 7m 44s (non-voting)\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/42aa702fa07f4c738bb087afddca38c1 : SUCCESS in 1h 29m 39s\n- octavia-v2-dsvm-scenario-traffic-ops https://zuul.opendev.org/t/openstack/build/de50953b18b248fcbd7c6a761d9bc1ae : SUCCESS in 1h 59m 41s\n- octavia-v2-dsvm-scenario-non-traffic-ops https://zuul.opendev.org/t/openstack/build/eccca8fd709c4c3980a65935399b9169 : SUCCESS in 1h 24m 39s\n- octavia-v2-dsvm-scenario-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/cf35789bb0ed46d291783fb251a2461a : SUCCESS in 2h 07m 15s\n- octavia-v2-dsvm-scenario-non-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/0b4bb65dadce40949213750a5ab4c190 : SUCCESS in 1h 37m 59s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/979cfa4814d04391a83e8581e644b66d : SUCCESS in 45m 46s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/1f0d4a69a0604eabb02717ec39e5da42 : SUCCESS in 39m 17s\n- octavia-v2-dsvm-tls-barbican-fips https://zuul.opendev.org/t/openstack/build/dbd45573ab2848308e459eb629dbde19 : SUCCESS in 1h 09m 24s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/54ab3fee8e7641e49034603ddbf09655 : SUCCESS in 34m 57s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/37767811cc924949b0bff7f890a14055 : TIMED_OUT in 2h 32m 54s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/b890483fe8014d46a8f8aa7e8b6286bf : TIMED_OUT in 3h 02m 55s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/7ecc38664bfa494fa47d5f83a12e9e3c : SUCCESS in 2h 30m 21s (non-voting)\n- octavia-v2-dsvm-scenario-centos-9-stream https://zuul.opendev.org/t/openstack/build/7d69d90b24c64cf6888748560b289877 : SUCCESS in 2h 20m 11s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"12214025f5f43aea6a61b1a4ded90b1ff3f59823","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2023-08-30 08:12:47.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"552fc89fe91258eb2d2fa7fa7d5a1e7ef2896006","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2023-08-30 08:15:39.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":3},{"id":"bbbb9bc5c0bd9c85d5aae5bda0e03764eacc66ce","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2023-08-30 08:15:50.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"0d961796e86ee934e6889491769adc3c3eb1fc75","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-30 10:58:01.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/adbf2c8428534b75ae68e6c57489533b\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/408415e4d0a546b881ec9bf8663859b4 : SUCCESS in 10m 01s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/33179c7dfcc84fb8a35fab5dc826bdb8 : SUCCESS in 4m 56s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/d1c769d20f7045e48f43137c26636047 : SUCCESS in 4m 50s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/93578fed1cb1418885d5eb926e1a28da : SUCCESS in 5m 07s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/64e6a28f754a4c55a06f0e66fd4ae45c : SUCCESS in 4m 46s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/eaabdda9992648ee948593a0af2d54ab : SUCCESS in 4m 41s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5a2e84b4ff804b92a6aa7928d44416dc : SUCCESS in 10m 57s\n- octavia-tox-py39-tips https://zuul.opendev.org/t/openstack/build/02072fef12d0416aa5f98f554210dcab : SUCCESS in 5m 23s\n- octavia-tox-functional-py39-tips https://zuul.opendev.org/t/openstack/build/b02fa4cb7a5b4358be998a0a1407fceb : SUCCESS in 7m 24s\n- octavia-tox-functional-py39-sqlalchemy-tips https://zuul.opendev.org/t/openstack/build/bc093ca9e4be4cb4a25dda287df57284 : SUCCESS in 9m 19s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/a9a8669ce5684f2daad9c5250c37d9d3 : SUCCESS in 5m 42s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/5b11291c361f479288db3ee90f370517 : SUCCESS in 8m 01s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/4c617a4976f74e65851b14bc30bdb801 : SUCCESS in 8m 29s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/61866b5122204495938626fbed9bfef6 : SUCCESS in 8m 04s (non-voting)\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/27b3139c1f834dc6b47c7427f7ae8424 : SUCCESS in 55m 22s\n- octavia-v2-dsvm-scenario-traffic-ops https://zuul.opendev.org/t/openstack/build/44d93a98175d4c26af37cfd856084edc : SUCCESS in 1h 58m 57s\n- octavia-v2-dsvm-scenario-non-traffic-ops https://zuul.opendev.org/t/openstack/build/064511d8fc69410797d0a330c253bffb : SUCCESS in 1h 26m 37s\n- octavia-v2-dsvm-scenario-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/fabc313f00154511aeecb23c860527fb : SUCCESS in 1h 33m 17s\n- octavia-v2-dsvm-scenario-non-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/f108517c03254636be6543951d54207c : SUCCESS in 1h 24m 46s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/8ef035d579ae42c7864c9d510d7b30fe : SUCCESS in 40m 33s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/d9febdf8e9d54e27abd021763429623a : SUCCESS in 39m 28s\n- octavia-v2-dsvm-tls-barbican-fips https://zuul.opendev.org/t/openstack/build/7238a5d00fae4299bfbfdde07b959417 : SUCCESS in 1h 02m 49s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/4b1c8ea9d2c544259b4aa8be84af900a : SUCCESS in 36m 11s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/2fe14277d1974baa8098a5b21690e264 : TIMED_OUT in 2h 33m 35s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/db1d74baa1514912a69730a75bea040d : SUCCESS in 2h 34m 14s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/142de71176d34f428fe037f56d33a89f : SUCCESS in 1h 51m 19s (non-voting)\n- octavia-v2-dsvm-scenario-centos-9-stream https://zuul.opendev.org/t/openstack/build/718df6ac24484739b23d4f7a85e01333 : SUCCESS in 2h 35m 54s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"727a90be2de4aeb2adec4271835e9a03f7abd37b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":6577,"name":"Sergey Kraynev","email":"sergejyit@gmail.com","username":"skraynev"},"date":"2023-09-03 11:15:47.000000000","message":"Uploaded patch set 4.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":4},{"id":"118efde8ac799ce781755732ec3f9e4705828658","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":6577,"name":"Sergey Kraynev","email":"sergejyit@gmail.com","username":"skraynev"},"date":"2023-09-03 11:16:40.000000000","message":"Uploaded patch set 5: Patch Set 4 was rebased.","accounts_in_message":[],"_revision_number":5},{"id":"94de68a562ff50ce23836796a24e4dddc7ae4f61","author":{"_account_id":6577,"name":"Sergey Kraynev","email":"sergejyit@gmail.com","username":"skraynev"},"date":"2023-09-03 11:17:18.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"d5af02de86660862d17a2c8372f1c7cb16f843ed","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-03 13:54:25.000000000","message":"Patch Set 5: Verified-1\n\n(4 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/28b75eb6f41647349361c6922587ddba\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/c010e6ce83f64cc99542bf663e667a98 : SUCCESS in 7m 38s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a1230b5add7a4fe3ba213a3908f20d61 : FAILURE in 3m 21s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/ce6acaf7a1d243a292b05a8a4e09c4e1 : SUCCESS in 4m 36s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/3d14aff62b7d433f8eb2fb23ee4596e1 : SUCCESS in 5m 47s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/034e878ed4124af1b404d545ff212e26 : SUCCESS in 4m 54s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/e967ba42da8f4583bbed8872685856bf : SUCCESS in 4m 41s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9d57ce6b39db4e6987d9a7c3c4208f74 : SUCCESS in 11m 45s\n- octavia-tox-py39-tips https://zuul.opendev.org/t/openstack/build/03504e46a6b346d286b17fc3e53a58ae : SUCCESS in 4m 51s\n- octavia-tox-functional-py39-tips https://zuul.opendev.org/t/openstack/build/f28295d3e34a4d81aeab94dfbc9de8b9 : SUCCESS in 7m 16s\n- octavia-tox-functional-py39-sqlalchemy-tips https://zuul.opendev.org/t/openstack/build/ea85ba9f65d540e28128d10fbc8eba20 : SUCCESS in 8m 10s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/d91068440c174afc8f9807501e4266c0 : SUCCESS in 4m 11s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/58cb927226694f54b406582fee3b3da8 : SUCCESS in 9m 19s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/987eafe347534ff2ba9a899cc6d5a18f : SUCCESS in 6m 43s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/cef1da4b32104682b6b852dd151e3c78 : SUCCESS in 7m 42s (non-voting)\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/88b5cb2dc8e749eb9afdb1d2e4827844 : SUCCESS in 52m 28s\n- octavia-v2-dsvm-scenario-traffic-ops https://zuul.opendev.org/t/openstack/build/4dd73bd6bd63497da4c6bd486b28b33b : SUCCESS in 1h 32m 07s\n- octavia-v2-dsvm-scenario-non-traffic-ops https://zuul.opendev.org/t/openstack/build/5c7ca6edc7fa4e19a83fa6b44a0ea121 : SUCCESS in 1h 14m 03s\n- octavia-v2-dsvm-scenario-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/c7689a3637db4d7d8372ff3397df45bb : SUCCESS in 2h 14m 05s\n- octavia-v2-dsvm-scenario-non-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/164128e41c564d76860e039733c22373 : SUCCESS in 59m 22s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/e045d50f013a4663a0b9d15486c6c3f9 : SUCCESS in 32m 53s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/7efafe4ba9cc43f0a6d0f4b2d87648da : FAILURE in 36m 45s\n- octavia-v2-dsvm-tls-barbican-fips https://zuul.opendev.org/t/openstack/build/31e080a9179e42ae9a3c197acc468698 : SUCCESS in 1h 02m 39s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/21ac8811c3c54c5a860748608454c747 : SUCCESS in 35m 33s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/e4ac21c807164236afe87f6578ff6685 : SUCCESS in 2h 02m 36s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/4b3fa528a157463983f442ed2cb4e6d5 : SUCCESS in 2h 36m 21s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/dd7d2b8d0fac4391a53754d56f403bdd : SUCCESS in 1h 29m 15s (non-voting)\n- octavia-v2-dsvm-scenario-centos-9-stream https://zuul.opendev.org/t/openstack/build/97a713ffd6614c5d8c1244f5c7527f71 : SUCCESS in 2h 00m 44s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"b39dd93a36da24a569a20f34a0c898336cd1c32d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":6577,"name":"Sergey Kraynev","email":"sergejyit@gmail.com","username":"skraynev"},"date":"2023-09-03 15:22:57.000000000","message":"Uploaded patch set 6.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":6},{"id":"974de1e6b9a072795584a9bbc954781a2bb0c29d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-03 18:27:11.000000000","message":"Patch Set 6: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/40d488bde300483eb753279ae3cbc860\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/f7c835b26998455ba7a9718c1b62034f : SUCCESS in 10m 43s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f13503d2b03b4d8a8f0df43a26d3b81c : SUCCESS in 5m 33s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/690c17edd33c405688d4249e15e4117f : SUCCESS in 3m 45s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/85976103f08c4f47ab265d772da1ff46 : SUCCESS in 5m 17s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/321f5a5bf411400bae1790f26efeb6e6 : SUCCESS in 5m 20s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/c1fdac53f4974805abe7f92f2340eca0 : SUCCESS in 4m 51s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b4c086ba2608498bbf3b003ecac35692 : SUCCESS in 12m 50s\n- octavia-tox-py39-tips https://zuul.opendev.org/t/openstack/build/79c77de24a8448aeb70ba16eb7cdafe8 : SUCCESS in 4m 42s\n- octavia-tox-functional-py39-tips https://zuul.opendev.org/t/openstack/build/81c42fe409764848b1fdb6b6ecac54b3 : SUCCESS in 6m 45s\n- octavia-tox-functional-py39-sqlalchemy-tips https://zuul.opendev.org/t/openstack/build/a0fb10ffdd534b1388dea92dc97a536f : SUCCESS in 8m 01s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/c82dd70fae514eccad6f3a3bc1675fb3 : SUCCESS in 3m 46s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/77e7b1278c694cd981fdeac1ee8f8c03 : SUCCESS in 5m 52s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/0bed3ea3a92840c694a16427716cd710 : SUCCESS in 6m 12s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/75f922d1db494dcf805fbad5d307c88c : SUCCESS in 5m 39s (non-voting)\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/67c1a16c9f914efbb9521e712508bfcd : SUCCESS in 50m 40s\n- octavia-v2-dsvm-scenario-traffic-ops https://zuul.opendev.org/t/openstack/build/bac3f949ee6b4d9ca4957f5b82d36805 : SUCCESS in 1h 58m 21s\n- octavia-v2-dsvm-scenario-non-traffic-ops https://zuul.opendev.org/t/openstack/build/715c6e67a1344cb6a894a732469440ba : SUCCESS in 1h 24m 47s\n- octavia-v2-dsvm-scenario-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/2da8f9f210014b66911cc1aba4e4780f : SUCCESS in 2h 00m 27s\n- octavia-v2-dsvm-scenario-non-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/cfeda725f6f94bb3a2409f78ad50d0b1 : SUCCESS in 1h 16m 31s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/1ba390b556a34e3aa360a9b1cd456ef4 : SUCCESS in 41m 56s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/7a49a35820c044918b6a48c1c6860e19 : FAILURE in 35m 42s\n- octavia-v2-dsvm-tls-barbican-fips https://zuul.opendev.org/t/openstack/build/ae0efb89a91d495fa7ba86995ec850c6 : SUCCESS in 54m 39s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/ec60743057b04a01944c19ce01e55fcf : SUCCESS in 37m 30s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/5ea892efba4d4999bf987510b3b1ee7c : SUCCESS in 1h 58m 30s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/0dff18a70c9c4563b6582e9fb886ad2b : TIMED_OUT in 3h 02m 45s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/fb7ea0d99b0b44fea7ffc9c8ba3b41a4 : SUCCESS in 1h 33m 09s (non-voting)\n- octavia-v2-dsvm-scenario-centos-9-stream https://zuul.opendev.org/t/openstack/build/d4c119b6d51d4bb2a9bbaeeccf4c9fd6 : SUCCESS in 2h 07m 36s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"2cb770c385a3058ff49cb15dcbf89574707bb61d","author":{"_account_id":6577,"name":"Sergey Kraynev","email":"sergejyit@gmail.com","username":"skraynev"},"date":"2023-09-04 04:29:03.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"d94607a2499c739a36b29444fe84c2919178d543","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-04 07:21:09.000000000","message":"Patch Set 6:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/1bd04bf7459a47439a46394de53b2087\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/a173aa68c3bb4d2885f74ebbfc522a83 : SUCCESS in 9m 52s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d1d117594a8d4190ad31a7080511f55c : SUCCESS in 4m 28s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/68aee9f88a2c47f287f98351981f617f : SUCCESS in 4m 36s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/20863450ec564cb8b9f67b0e11e59f07 : SUCCESS in 4m 45s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/78e23b871fef40dfb256a6ae5555cca6 : SUCCESS in 4m 15s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/964ef616e85b439fb63bf2f32f6c7612 : SUCCESS in 5m 19s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b003575789774c3f8c077461209f2af3 : SUCCESS in 10m 30s\n- octavia-tox-py39-tips https://zuul.opendev.org/t/openstack/build/acf5478735084d9abc397a094d130148 : SUCCESS in 4m 25s\n- octavia-tox-functional-py39-tips https://zuul.opendev.org/t/openstack/build/0fef7da365a34c51beb6c3a80e7452cd : SUCCESS in 7m 43s\n- octavia-tox-functional-py39-sqlalchemy-tips https://zuul.opendev.org/t/openstack/build/4c7243b45b8e47e8a52fc8bc7c4ce32c : SUCCESS in 10m 58s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/7f66cc549c764b2695aaf7742213526c : SUCCESS in 3m 42s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/1460b7f3ec6d465fa9dc5fa80cb722ac : SUCCESS in 7m 47s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/b3d9a6d13a954b09bb78cd55db432419 : SUCCESS in 8m 55s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/7f9a18c12db5456a8f82e5f68e02ac71 : SUCCESS in 7m 11s (non-voting)\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/7e8108d211dc4da78b92588bb3ebd6a3 : SUCCESS in 1h 09m 45s\n- octavia-v2-dsvm-scenario-traffic-ops https://zuul.opendev.org/t/openstack/build/d880c59043cb4bc8a556939daf6ba23e : SUCCESS in 1h 36m 39s\n- octavia-v2-dsvm-scenario-non-traffic-ops https://zuul.opendev.org/t/openstack/build/e69f9621220e47beb39fc61f2e9ca357 : SUCCESS in 1h 17m 23s\n- octavia-v2-dsvm-scenario-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/d0080ff0cd0946889a745788a1e29bae : SUCCESS in 2h 02m 43s\n- octavia-v2-dsvm-scenario-non-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/a4350a06136542c087de0417aa6d9d0b : SUCCESS in 1h 18m 44s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/ab08417a4bef4974809ab19bd4bd629f : SUCCESS in 39m 59s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/28e595e2c29f4fe894b088a0ffa7546a : FAILURE in 39m 14s\n- octavia-v2-dsvm-tls-barbican-fips https://zuul.opendev.org/t/openstack/build/de3c8084356d456695e3d32f2d70fd08 : SUCCESS in 57m 53s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/c5f3b0f20d2f401cb07b77fef0d7b1d3 : SUCCESS in 31m 34s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/afd35f1f36cc4bc1bd801bf301cb65a1 : SUCCESS in 2h 00m 40s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/5e11b1b05b4543aca133d2e411000c30 : SUCCESS in 2h 49m 02s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/47bf2f166f66461281037ada3a4c7501 : SUCCESS in 2h 03m 58s (non-voting)\n- octavia-v2-dsvm-scenario-centos-9-stream https://zuul.opendev.org/t/openstack/build/4e53bbb3d14044cb8a46f83b43928254 : SUCCESS in 1h 53m 10s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"94f00bdf6afda782129f6d106cf4c08ea25bb3c9","author":{"_account_id":6577,"name":"Sergey Kraynev","email":"sergejyit@gmail.com","username":"skraynev"},"date":"2023-09-04 07:28:49.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"d72edd3e0959befd7023861ae57ffa756258e104","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-04 10:28:49.000000000","message":"Patch Set 6:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/d08d21b609414962a345f76f57ea960e\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/90431c97c00048b7bae3f2061e3351b8 : SUCCESS in 10m 51s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4c8b638202ea4b89aac7f1390b00922c : SUCCESS in 6m 14s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/9ddcb9c09dba49d8b2d4ff0197ab7b03 : SUCCESS in 5m 01s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/10c7dfc6b7144e7fb22fbaeb993f78f0 : SUCCESS in 4m 39s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/089247983b4a40b5bf8d0ff2f9d4fb0f : SUCCESS in 4m 55s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/2cd5f6ac116f4a02b014ca3fdc3d4cc7 : SUCCESS in 5m 04s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7dcebd2e974b491b9465a9ea377dd420 : SUCCESS in 13m 29s\n- octavia-tox-py39-tips https://zuul.opendev.org/t/openstack/build/7a2dd6ba844a4a2eb266948628957bc7 : SUCCESS in 5m 07s\n- octavia-tox-functional-py39-tips https://zuul.opendev.org/t/openstack/build/5e76b66c417d456897e6f99d5e107d2f : SUCCESS in 7m 30s\n- octavia-tox-functional-py39-sqlalchemy-tips https://zuul.opendev.org/t/openstack/build/f89a7128eadc4204a18a42a6232f5ea6 : SUCCESS in 10m 11s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/6e394a88a931458dabf93424ee3bbcfd : SUCCESS in 5m 07s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/37a133a4d4064f51ac2f09ae96eeb329 : SUCCESS in 7m 38s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/339ac94c2b8f4ab59a36c5972dfa4f13 : SUCCESS in 7m 22s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/66926f19278f4cbb92481ad4c22199ff : SUCCESS in 8m 17s (non-voting)\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/1e17d4c6ac93480bb624b2a54b1cf5fc : SUCCESS in 1h 05m 44s\n- octavia-v2-dsvm-scenario-traffic-ops https://zuul.opendev.org/t/openstack/build/be0320b2f96f45bd8b78bb53a3909cdc : SUCCESS in 1h 23m 59s\n- octavia-v2-dsvm-scenario-non-traffic-ops https://zuul.opendev.org/t/openstack/build/6cec8412226843879efef6222dbdc4a3 : SUCCESS in 1h 09m 47s\n- octavia-v2-dsvm-scenario-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/5246eb562c5b4621a9ca17bd2c51df43 : SUCCESS in 1h 31m 56s\n- octavia-v2-dsvm-scenario-non-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/e2ff19dd43d64bdfbec69bfef7052955 : SUCCESS in 1h 29m 34s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/dba53d1d78744a209f6d7d733548808c : SUCCESS in 35m 21s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/e95824a40bb44b0c9d70f67f454c98e1 : FAILURE in 44m 00s\n- octavia-v2-dsvm-tls-barbican-fips https://zuul.opendev.org/t/openstack/build/1ceffeb65a164873ad6d197a69cfd1ac : SUCCESS in 59m 35s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/0b020032a78b4f5fb2fd22a677ac166e : SUCCESS in 36m 06s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/944279a9dcd24f40b3495ff49ee6a17a : TIMED_OUT in 2h 33m 32s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/b57bcfc180264842a43dcc9e45b9e603 : SUCCESS in 2h 58m 17s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/3c49670a10ae49349ccea1ee6d26553f : SUCCESS in 2h 31m 49s (non-voting)\n- octavia-v2-dsvm-scenario-centos-9-stream https://zuul.opendev.org/t/openstack/build/4fd713bbcfe74d38a0be95285aa160b0 : SUCCESS in 2h 31m 09s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"0493ca27f9e716595d2a5bc9be6ce04b3d7c784e","author":{"_account_id":6577,"name":"Sergey Kraynev","email":"sergejyit@gmail.com","username":"skraynev"},"date":"2023-09-05 09:29:28.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"52601e78094ad494d9bb6b5057c704aa12a00c1b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-05 12:34:46.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/90a61a0dbcd54601a32ff01f44b9decf\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/2d83fddc2bde4f39a11832bd66c6af43 : SUCCESS in 9m 48s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2dbabeffb5b444e1a25a25e36f63ad28 : SUCCESS in 5m 24s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/6be5ee4e51e4490da40e1bf92e5d9486 : SUCCESS in 4m 29s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/8b78f740690b414393751ed3fa3dd119 : SUCCESS in 5m 19s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/57f233c6dd3145d28e42aac4e7dc75d1 : SUCCESS in 4m 19s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/6bae1cd7b2e44229a37d0d76b65a51e5 : SUCCESS in 5m 14s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3f8cc483ab6a49319a99a9c7963afefe : SUCCESS in 11m 42s\n- octavia-tox-py39-tips https://zuul.opendev.org/t/openstack/build/75b8fb1ad70643eebd8261ce054eef46 : SUCCESS in 4m 44s\n- octavia-tox-functional-py39-tips https://zuul.opendev.org/t/openstack/build/9b3f690b7fce4d489f54adc57c8f1939 : SUCCESS in 6m 41s\n- octavia-tox-functional-py39-sqlalchemy-tips https://zuul.opendev.org/t/openstack/build/432e72fe74bb4792bd75e569ea4b4613 : SUCCESS in 9m 49s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/ccdc494e5ab94d1aa68c0e6072a67e25 : SUCCESS in 3m 52s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/c15c526359994eada0085ecac23b7e8a : SUCCESS in 6m 36s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/4153fea345604428bfd6e3f39c56684a : SUCCESS in 7m 29s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/1d1b95fdffa64553b80d12eb09b2cd02 : SUCCESS in 7m 21s (non-voting)\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/6a793f428c9c4b8a9d8f7c69be764843 : SUCCESS in 1h 01m 25s\n- octavia-v2-dsvm-scenario-traffic-ops https://zuul.opendev.org/t/openstack/build/dd2bc428fac640e5b02df11d433ca0b3 : SUCCESS in 2h 06m 25s\n- octavia-v2-dsvm-scenario-non-traffic-ops https://zuul.opendev.org/t/openstack/build/c5a6880d41064e5dbdbfaf1ee1a79b96 : SUCCESS in 1h 24m 41s\n- octavia-v2-dsvm-scenario-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/0af88415a8fb4576990dd29349f3e90a : SUCCESS in 2h 16m 58s\n- octavia-v2-dsvm-scenario-non-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/dbe5c8c7063a451bac38db4735ca9e87 : SUCCESS in 1h 29m 27s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/4fc6eea7af524673aa9ae37b1819ed33 : SUCCESS in 38m 40s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/aea83a67bb4049799cc44ef4a48d6517 : SUCCESS in 52m 18s\n- octavia-v2-dsvm-tls-barbican-fips https://zuul.opendev.org/t/openstack/build/68c99e720c444a09aa39aa54d6ca136e : SUCCESS in 58m 31s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/eee74c4e226641299ca41c66d5114c5f : SUCCESS in 31m 55s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/45e4cc6ab3ee4d07b615e5b61651f6fc : TIMED_OUT in 2h 33m 16s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/4c80472897c346ddaf071cee4b2508d8 : SUCCESS in 3h 02m 35s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/2a195cb102214971b9bf5edfc8566774 : TIMED_OUT in 2h 32m 57s (non-voting)\n- octavia-v2-dsvm-scenario-centos-9-stream https://zuul.opendev.org/t/openstack/build/f7bad65eb5f84daf8ce97e45eb028bcf : SUCCESS in 1h 52m 01s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"40b6f588f235ab33b15b747e5aaf9d572f904d1f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":6577,"name":"Sergey Kraynev","email":"sergejyit@gmail.com","username":"skraynev"},"date":"2023-09-18 14:31:59.000000000","message":"Patch Set 7: Patch Set 6 was rebased\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":7},{"id":"bd78df5cf73d7041edc7d3bdb601dd25e177c249","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-18 17:36:53.000000000","message":"Patch Set 7: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/e934726be0bf49c7a43a40e49ef19952\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/c90b793b97bd48c2bd9d7eaea5f46501 : SUCCESS in 11m 06s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b57c06723fe74efd83fc49be15f5e8cf : SUCCESS in 5m 39s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/03b5fba2be194bd9b0b85e285ed4788d : SUCCESS in 4m 59s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/dc8da06a683f4868980875e984660dc1 : SUCCESS in 5m 00s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/a5610b1cf453484589c5b2a19dbb324b : SUCCESS in 6m 20s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/ca820f06400a49b084795038abf82442 : SUCCESS in 5m 12s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cf6c68245aaa4f60907546dda4fa4d71 : SUCCESS in 14m 11s\n- octavia-tox-py39-tips https://zuul.opendev.org/t/openstack/build/bfff860e0ef6483db99f96489c99778b : SUCCESS in 6m 52s\n- octavia-tox-functional-py39-tips https://zuul.opendev.org/t/openstack/build/e13276b0fc174f03901cf7ee8395fb6f : SUCCESS in 8m 14s\n- octavia-tox-functional-py39-sqlalchemy-tips https://zuul.opendev.org/t/openstack/build/5abee40180f847adb4cdb9ddfa733fcd : SUCCESS in 10m 06s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/1fdaa10a588841ccbf841cd7ff725127 : SUCCESS in 5m 59s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/107d483f2ffd4c92b3302b37954b1a5d : SUCCESS in 7m 16s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/675244a628ef4bfaa1c289743adcd1e6 : SUCCESS in 8m 53s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/24ece48cc4bc4f66b333b2b845ebbabc : SUCCESS in 7m 26s (non-voting)\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/2d253961ad41454496bc820d7a88dcd7 : SUCCESS in 1h 00m 45s\n- octavia-v2-dsvm-scenario-traffic-ops https://zuul.opendev.org/t/openstack/build/1a849a4c5db54bfea83b36e868793c75 : SUCCESS in 1h 43m 12s\n- octavia-v2-dsvm-scenario-non-traffic-ops https://zuul.opendev.org/t/openstack/build/1faa5b736f404203a3167f567ff3cbf3 : FAILURE in 1h 20m 31s\n- octavia-v2-dsvm-scenario-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/2c2599aae62e447fa51628340be9b67e : SUCCESS in 1h 20m 22s\n- octavia-v2-dsvm-scenario-non-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/b82f7cab31734272b30a570944b85130 : SUCCESS in 1h 18m 38s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/6c726011c804441196c87e20c3717470 : SUCCESS in 38m 34s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/bc001a31bf7b4f7d9e928948417d9957 : SUCCESS in 38m 41s\n- octavia-v2-dsvm-tls-barbican-fips https://zuul.opendev.org/t/openstack/build/60ebc53b957645ebbe68d9ec8e183f62 : SUCCESS in 1h 03m 17s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/05967be9da7d4c4a84d390efa756e0ee : SUCCESS in 34m 44s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/3f5d39da2549472ba40c3735bad9b892 : SUCCESS in 2h 11m 08s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/b3c2fb15de91470ea05facf372f29075 : TIMED_OUT in 3h 03m 17s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/29bfaff32deb44dbb98528e0d6fec696 : TIMED_OUT in 2h 33m 05s (non-voting)\n- octavia-v2-dsvm-scenario-centos-9-stream https://zuul.opendev.org/t/openstack/build/b65cc0f3b15f44e0addd77f84d25609b : SUCCESS in 2h 00m 47s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"a7920c357a5934592dd7a6d8e2eb996e23521903","author":{"_account_id":6577,"name":"Sergey Kraynev","email":"sergejyit@gmail.com","username":"skraynev"},"date":"2023-09-19 07:31:52.000000000","message":"Patch Set 7:\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"1f511e67a6ce15d5f5eb982e52d6af5bf66d0b9b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-19 09:48:29.000000000","message":"Patch Set 7:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/42bd203b17b045ee954ed7148ce9f9d6\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/3f3a632b07c245fc924907aa7804543d : SUCCESS in 11m 29s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/37d559d924db49c3add3334c98a26d93 : SUCCESS in 5m 35s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/38082fb0393d486c8a8f3092861628c6 : SUCCESS in 5m 33s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/2ce41d6a11f349fa81deb7ea5729dcf8 : SUCCESS in 5m 40s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/cc65c7a2263848ffbe70658c26d411e4 : SUCCESS in 4m 21s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/93299f4ed40a4cc9acfa064059fa0640 : SUCCESS in 4m 12s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3cde157847564b06913566f819a23ce0 : SUCCESS in 9m 58s\n- octavia-tox-py39-tips https://zuul.opendev.org/t/openstack/build/e523a228ba9a41f1a7ba8808d43618ee : SUCCESS in 4m 43s\n- octavia-tox-functional-py39-tips https://zuul.opendev.org/t/openstack/build/ef6d3702c77e4e59aef963922a89452a : SUCCESS in 6m 08s\n- octavia-tox-functional-py39-sqlalchemy-tips https://zuul.opendev.org/t/openstack/build/40ed07c11167438e82f0fe64d904817e : SUCCESS in 8m 39s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/602560d6bfd8403aa465786a7e4d15f6 : SUCCESS in 5m 23s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/1d5e4790cc4b4c51a4b391e7a187d781 : SUCCESS in 7m 34s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/73688d49a03b4c359b3bddc6c8001cbe : SUCCESS in 8m 50s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/72b19a546bc543bb8d5c90a7b96a5cd9 : SUCCESS in 7m 51s (non-voting)\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/f2822f25a2014e848d21cc94e1fcd5d8 : SUCCESS in 1h 17m 59s\n- octavia-v2-dsvm-scenario-traffic-ops https://zuul.opendev.org/t/openstack/build/789cf36eb2934918abe64c5781af502a : SUCCESS in 1h 19m 45s\n- octavia-v2-dsvm-scenario-non-traffic-ops https://zuul.opendev.org/t/openstack/build/6548beb6d47f4eb6a1590f9def2728a9 : FAILURE in 1h 04m 12s\n- octavia-v2-dsvm-scenario-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/15b018e330824d01be17eeb78a4cac47 : SUCCESS in 1h 41m 05s\n- octavia-v2-dsvm-scenario-non-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/f807d4c22c974754a130cc25b66a7dc4 : SUCCESS in 1h 26m 26s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/ec48ee5265184bbfb0a4216472c3687a : SUCCESS in 44m 12s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/ab0fff6858cf4112aae6b8bcc8549774 : SUCCESS in 51m 11s\n- octavia-v2-dsvm-tls-barbican-fips https://zuul.opendev.org/t/openstack/build/29e7338654c84117ad0693cc5d2b9c6b : SUCCESS in 1h 10m 51s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/28fa5c5b5fb64d50a1c6646fd2b5b129 : SUCCESS in 40m 04s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/8b41987895cc4582a954d9cb6387a341 : SUCCESS in 2h 06m 24s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/2f0a801622af4c0da70bf979fbb243c4 : SUCCESS in 2h 14m 26s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/d33adb974d864603b075da487e9d7b87 : SUCCESS in 2h 13m 47s (non-voting)\n- octavia-v2-dsvm-scenario-centos-9-stream https://zuul.opendev.org/t/openstack/build/939029a00ac44a88b55e9935c1d1afdd : SUCCESS in 2h 09m 42s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"f6d0409a4ab8cd0761e0ce5e326ef8aa55cfc9ab","author":{"_account_id":6577,"name":"Sergey Kraynev","email":"sergejyit@gmail.com","username":"skraynev"},"date":"2023-09-20 13:12:27.000000000","message":"Patch Set 7:\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"e1b9aa5d99bfc42bac2809a2c942b031bda1d330","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-20 15:47:16.000000000","message":"Patch Set 7: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/11837a4d79ba47f88ee0a99d165026a0\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/ffa47cdde8d144bd891cda28842335d1 : SUCCESS in 10m 01s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/532373e4dae146ffacbdb2f1f88f7818 : SUCCESS in 4m 58s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/27d523b89c574004813dfb2d59a43b42 : SUCCESS in 4m 51s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/0f2618464f684ce4be8e4d089526d800 : SUCCESS in 4m 31s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/6f811249c1d54633a2ce3a1437cbab17 : SUCCESS in 5m 06s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/79df6114967a470eaa7c83895f089adb : SUCCESS in 5m 19s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/44085d0390754b2cb7bc0a15df484029 : SUCCESS in 11m 54s\n- octavia-tox-py39-tips https://zuul.opendev.org/t/openstack/build/c3e92924ec50424b8aacd8409b504fbc : SUCCESS in 5m 22s\n- octavia-tox-functional-py39-tips https://zuul.opendev.org/t/openstack/build/e8354b7252b845f2ab92b8a1c19c794f : SUCCESS in 7m 08s\n- octavia-tox-functional-py39-sqlalchemy-tips https://zuul.opendev.org/t/openstack/build/f83ea18e6cbb49779168ce270425d80d : SUCCESS in 9m 03s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/2ffdc75a21a843b99f8947292ec41bae : SUCCESS in 5m 00s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/12d3be08564d4237b2a1c039a76ebca3 : SUCCESS in 6m 38s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/775c8b7c41c94fb4948481e0e0eac8d7 : SUCCESS in 8m 06s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/cc56eba1e1cb4ed98a2ffc7502d0d5d1 : SUCCESS in 6m 43s (non-voting)\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/27df100585cf43c3b7604efbc1c4f640 : SUCCESS in 1h 29m 16s\n- octavia-v2-dsvm-scenario-traffic-ops https://zuul.opendev.org/t/openstack/build/166efe3acbdd49ce88474618e5fbc8cb : SUCCESS in 2h 03m 59s\n- octavia-v2-dsvm-scenario-non-traffic-ops https://zuul.opendev.org/t/openstack/build/b3d4e47503474f558127cec0fa66b36d : SUCCESS in 1h 16m 32s\n- octavia-v2-dsvm-scenario-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/b102da67893a4d59aed5bd6c055dc361 : SUCCESS in 2h 03m 17s\n- octavia-v2-dsvm-scenario-non-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/f04f7a56411f49dcb4b1149a82c08741 : SUCCESS in 1h 37m 37s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/34a1f98096464487a4dddc87561cf8bb : SUCCESS in 43m 15s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/7b74b9d51a95478db0a86fe4117832f1 : SUCCESS in 44m 11s\n- octavia-v2-dsvm-tls-barbican-fips https://zuul.opendev.org/t/openstack/build/a0150691a42548269237f16bdf1891b0 : SUCCESS in 1h 07m 35s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/073bcb511999413791d5b21052d7fa3f : SUCCESS in 36m 58s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/01109cbb3c294d3e99ebe6aa2ef42088 : TIMED_OUT in 2h 33m 17s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/14f8f6b6172247cc936aba60ce7c4be2 : SUCCESS in 2h 21m 19s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/2a7a9d6dd1fc455abb130afb32149915 : TIMED_OUT in 2h 33m 02s (non-voting)\n- octavia-v2-dsvm-scenario-centos-9-stream https://zuul.opendev.org/t/openstack/build/9a2d793b1cee4af5b3ac93efbc410314 : SUCCESS in 2h 02m 54s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"338c3104d92d8bc1c31688077687a6b51d7731d5","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":6577,"name":"Sergey Kraynev","email":"sergejyit@gmail.com","username":"skraynev"},"date":"2023-09-25 09:34:47.000000000","message":"Patch Set 8: Patch Set 7 was rebased\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":8},{"id":"92fee8f1ce5d6cbb570fd09c650a7357c293512b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-25 12:25:15.000000000","message":"Patch Set 8: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/31f8d8e26ba74e238e9f7acfead6d18e\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/cf8e2e91915b453b8972023a04b1395f : SUCCESS in 10m 58s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e515d3a7aad14d07bbd3ceb1f1e8aa9d : SUCCESS in 4m 48s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/899c747f2a7a45d6be96b6258485f45f : SUCCESS in 5m 02s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/5c10ffde4f1248d485b9718a54bcd19f : SUCCESS in 4m 35s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/30c322a4f73d44bbb8ba3a789f1684bd : SUCCESS in 5m 17s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/c7c6db157dc74cbc9b2e440c70398731 : SUCCESS in 6m 01s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d69838813a3e4d30adc524a0d7d6bc1c : SUCCESS in 12m 22s\n- octavia-tox-py39-tips https://zuul.opendev.org/t/openstack/build/d7eaa58afb744880af56e57ea3d42456 : SUCCESS in 4m 52s\n- octavia-tox-functional-py39-tips https://zuul.opendev.org/t/openstack/build/7b8a1ee830dc4afe8827457d30957541 : SUCCESS in 7m 50s\n- octavia-tox-functional-py39-sqlalchemy-tips https://zuul.opendev.org/t/openstack/build/bc382607e5394482a27b9e454c6c1bea : SUCCESS in 9m 38s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/437f346a9d5646eba6f66074c7b92849 : SUCCESS in 4m 38s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/74cc491bc0f84085aacf1183585f5e15 : SUCCESS in 6m 48s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/6ab333605ddb46fcae1e26b0398aa4db : SUCCESS in 7m 28s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/96efb1a78c1243de8aab1231d246d1f6 : SUCCESS in 7m 30s (non-voting)\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/7a0012078c714a3aa269b14f2e7ca521 : SUCCESS in 1h 05m 40s\n- octavia-v2-dsvm-scenario-traffic-ops https://zuul.opendev.org/t/openstack/build/7bd2910746134b2f85771e873ec737db : SUCCESS in 1h 33m 58s\n- octavia-v2-dsvm-scenario-non-traffic-ops https://zuul.opendev.org/t/openstack/build/2d75b376f8a94a15bdefc97e715972e2 : SUCCESS in 1h 16m 06s\n- octavia-v2-dsvm-scenario-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/e74f2193f4744cd3831cabc23d276919 : SUCCESS in 1h 27m 26s\n- octavia-v2-dsvm-scenario-non-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/36122bf19d284edd8c78fd5359f1c9bb : SUCCESS in 1h 22m 30s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/7db9e0da06884ba585f54d6066eb32f6 : SUCCESS in 41m 56s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/2658b3bd3c4f48bc8d376e7cdf998431 : SUCCESS in 44m 16s\n- octavia-v2-dsvm-tls-barbican-fips https://zuul.opendev.org/t/openstack/build/8e4dfc6b717849218c6acf93ebece102 : SUCCESS in 1h 10m 46s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/f1b02c12043d424d955f3e869478bc48 : SUCCESS in 37m 01s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/5573c61843a14ad394b1ff5a459005ed : TIMED_OUT in 2h 33m 55s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/955bf9419d4e427eb311338ede662bfd : SUCCESS in 2h 48m 48s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/6fe2213002f047adbcd863d6ad2f0e8d : SUCCESS in 2h 23m 29s (non-voting)\n- octavia-v2-dsvm-scenario-centos-9-stream https://zuul.opendev.org/t/openstack/build/e73599ab2b9d4fbe9eb93d58fe4ab8b6 : SUCCESS in 2h 29m 05s (non-voting)","accounts_in_message":[],"_revision_number":8},{"id":"b64088dd9f8c26a25c633e9fb725738b3d4a99bd","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2023-09-26 08:48:25.000000000","message":"Patch Set 8: Code-Review-1\n\n(5 comments)","accounts_in_message":[],"_revision_number":8},{"id":"077f7ec560b0ff889cd4a2d79bcbc899a59d9aa2","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":6577,"name":"Sergey Kraynev","email":"sergejyit@gmail.com","username":"skraynev"},"date":"2023-09-26 18:31:22.000000000","message":"Uploaded patch set 9.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":9},{"id":"05760ca3fff1ecb7f01030cc4910cbf2b55a1949","author":{"_account_id":6577,"name":"Sergey Kraynev","email":"sergejyit@gmail.com","username":"skraynev"},"date":"2023-09-26 18:31:58.000000000","message":"Patch Set 8:\n\n(5 comments)","accounts_in_message":[],"_revision_number":8},{"id":"70735a6a4d958dd1208d619f0ea43392d787f0de","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-26 21:00:09.000000000","message":"Patch Set 9: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b39a7db578004c37836f9206fa6930a1\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/cbfd1c6f770c426298336b3181399dc0 : SUCCESS in 9m 47s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5b3b230349cf4fd186ca97e65fd75e61 : SUCCESS in 5m 32s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/7e739e921f604f159b8f7519e72023ea : SUCCESS in 4m 20s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/14348781bc614476901532d54ca92fc1 : SUCCESS in 5m 24s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/e78948cbb9694fcb880cba02226e7d28 : SUCCESS in 4m 30s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/791175633a6b4447bd14f39acced905f : SUCCESS in 5m 32s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/74de0e348793452e8f35f8f0838c3b2d : SUCCESS in 10m 10s\n- octavia-tox-py39-tips https://zuul.opendev.org/t/openstack/build/c7a267dfb7094225ba124e6084e5f789 : SUCCESS in 4m 40s\n- octavia-tox-functional-py39-tips https://zuul.opendev.org/t/openstack/build/97332f3cb7174d7b9d1cd01b9c406f7e : SUCCESS in 7m 58s\n- octavia-tox-functional-py39-sqlalchemy-tips https://zuul.opendev.org/t/openstack/build/ceeb24099e69467882d9bfe87e3b2e21 : SUCCESS in 7m 23s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/097188d6108443df953d6fa470bf4abd : SUCCESS in 4m 42s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/2c5a6f2103504e9a8496ed8403baae5c : SUCCESS in 9m 13s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/0598d334f0cd48e8abd71ea91d1369a4 : SUCCESS in 7m 12s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/409550f1085a4524a9bcbf41d33443e3 : SUCCESS in 7m 56s (non-voting)\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/654b0ac1f5404c72824dd4c01df086f2 : SUCCESS in 1h 26m 36s\n- octavia-v2-dsvm-scenario-traffic-ops https://zuul.opendev.org/t/openstack/build/0590825d97294e4f97b4727c5aad3ee5 : SUCCESS in 1h 54m 49s\n- octavia-v2-dsvm-scenario-non-traffic-ops https://zuul.opendev.org/t/openstack/build/a7740da1cd3241c1907f922d3d0f473a : SUCCESS in 1h 12m 52s\n- octavia-v2-dsvm-scenario-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/f9b9f70c4cf14f969e33f5e5703fb158 : SUCCESS in 1h 27m 21s\n- octavia-v2-dsvm-scenario-non-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/4a1e0bb011d241a29c24278a4be896d0 : SUCCESS in 1h 31m 45s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/67cd1d56cc354aaba6c64d4ac515fa1b : SUCCESS in 38m 32s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/0c1cb9a2795e407cb525c7352ca49f6f : SUCCESS in 43m 10s\n- octavia-v2-dsvm-tls-barbican-fips https://zuul.opendev.org/t/openstack/build/5c46a11fc9774d89aa61538ba6dcb94d : SUCCESS in 53m 57s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/230d8849d70647d39fe93aef7e8fd9be : SUCCESS in 32m 39s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/ae9b5ff35b7f48b08a1f97d3106da6e0 : SUCCESS in 1h 46m 10s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/5c0dba1fe8c14c0f8f24b77809ec9201 : SUCCESS in 2h 27m 21s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/a3b5e4ad0c8a4b8180a75c95e7638b88 : SUCCESS in 1h 54m 05s (non-voting)\n- octavia-v2-dsvm-scenario-centos-9-stream https://zuul.opendev.org/t/openstack/build/dc745825c9fc44dcb71a5757f65ea210 : SUCCESS in 2h 08m 57s (non-voting)","accounts_in_message":[],"_revision_number":9},{"id":"761a3ff518008dd3c0169cf9a1e176b65d21d5d3","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2023-09-27 09:39:26.000000000","message":"Patch Set 9: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":9},{"id":"e3922d1f4afe31a0cadc2cd863aa311fdb4dd8dd","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2023-09-27 09:45:59.000000000","message":"Patch Set 9: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":9},{"id":"2990568251c6f39f132097c9836a01cdf325d36e","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2023-09-27 16:36:26.000000000","message":"Patch Set 9: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":9},{"id":"104efa04ca3c4f8657f72397dd77ae37bcdc25dd","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2023-09-28 06:57:28.000000000","message":"Patch Set 9: Code-Review-1\n\n(2 comments)","accounts_in_message":[],"_revision_number":9},{"id":"2949e6e0f928cd5e11018ca4b96c81786ddd3661","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":6577,"name":"Sergey Kraynev","email":"sergejyit@gmail.com","username":"skraynev"},"date":"2023-09-30 11:19:01.000000000","message":"Uploaded patch set 10.\n\nOutdated Votes:\n* Code-Review+2, Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":10},{"id":"d6cfe914515226b46379aaec996820a897121a98","author":{"_account_id":6577,"name":"Sergey Kraynev","email":"sergejyit@gmail.com","username":"skraynev"},"date":"2023-09-30 11:19:14.000000000","message":"Patch Set 9:\n\n(2 comments)","accounts_in_message":[],"_revision_number":9},{"id":"36b96c54302c53bd46ebdb593555d38ab7eb0852","author":{"_account_id":6577,"name":"Sergey Kraynev","email":"sergejyit@gmail.com","username":"skraynev"},"date":"2023-09-30 11:19:25.000000000","message":"Patch Set 10:\n\n(1 comment)","accounts_in_message":[],"_revision_number":10},{"id":"91789349e0d8f954f2490a05e2a0919a320c2ba1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-30 14:24:30.000000000","message":"Patch Set 10: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/81445935b04e414d87d13dd2a78770dc\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/0a7fcda89035407baded071a143e308c : SUCCESS in 7m 43s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/001d10ef19754a3d94ec0267a710b793 : SUCCESS in 4m 18s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/b0d0dea3febd46d6b33ee7c419e45c22 : SUCCESS in 4m 38s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/1302fff66b8f49bf991a152e360f4d05 : SUCCESS in 4m 23s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/15109a7ba2f040feb947f34251e6b01b : SUCCESS in 4m 40s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/a6a7bfd974ad410cb478c865543c8492 : SUCCESS in 4m 35s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9be3a817ee8c4634b1c888740758a55f : SUCCESS in 12m 23s\n- octavia-tox-py39-tips https://zuul.opendev.org/t/openstack/build/facd0662b7534f86a5b96105541fccab : SUCCESS in 4m 31s\n- octavia-tox-functional-py39-tips https://zuul.opendev.org/t/openstack/build/8b0e4f9434c548c28ab5db4c66f1cb7d : SUCCESS in 6m 16s\n- octavia-tox-functional-py39-sqlalchemy-tips https://zuul.opendev.org/t/openstack/build/f6112c1637bd4782af8fdf2aceb01907 : SUCCESS in 7m 53s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/4400f9273cc04422a51586063f048354 : SUCCESS in 4m 18s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/c042a570615b45888287e58262cc604e : SUCCESS in 6m 50s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/41950eca599146e094950ce1130c14bc : SUCCESS in 6m 52s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/a115e6e1f7d44a438fb8b77f277ed44e : SUCCESS in 6m 39s (non-voting)\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/07b840eee40f43c6b8fa2414e96fdb34 : SUCCESS in 51m 18s\n- octavia-v2-dsvm-scenario-traffic-ops https://zuul.opendev.org/t/openstack/build/4a397b37245443da9a0b83ad1797f0b1 : SUCCESS in 2h 12m 56s\n- octavia-v2-dsvm-scenario-non-traffic-ops https://zuul.opendev.org/t/openstack/build/b1af805416394cc9bc651bd214125a5d : SUCCESS in 1h 33m 05s\n- octavia-v2-dsvm-scenario-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/d576297f23024c6bbd9bca3618679519 : SUCCESS in 2h 19m 21s\n- octavia-v2-dsvm-scenario-non-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/cfd2f841da1241bb9204779453686d12 : SUCCESS in 1h 39m 14s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/4c09906f073f443b8549b5c6520a80df : SUCCESS in 39m 41s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/2b3c02c7a01a4359b4bec5bf45e1bfec : SUCCESS in 57m 36s\n- octavia-v2-dsvm-tls-barbican-fips https://zuul.opendev.org/t/openstack/build/697af7fbbdcd429b90264839a12f04bf : SUCCESS in 50m 55s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/09c9ba0cd6ff450492c493e139b5cf6e : SUCCESS in 51m 15s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/404eeba22e9c43debdace7497d545eea : TIMED_OUT in 2h 32m 57s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/2d5a52d3ce1348e6859a165a3ddabda9 : TIMED_OUT in 3h 03m 42s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/841b565e7ad1489ba17855d0cd741ce8 : SUCCESS in 1h 36m 21s (non-voting)\n- octavia-v2-dsvm-scenario-centos-9-stream https://zuul.opendev.org/t/openstack/build/011b056c1d154b21b883018c2b55b087 : SUCCESS in 1h 46m 21s (non-voting)","accounts_in_message":[],"_revision_number":10},{"id":"e9abfa47c8b7ddc37e371465560db42fb605055c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":6577,"name":"Sergey Kraynev","email":"sergejyit@gmail.com","username":"skraynev"},"date":"2023-10-09 04:57:05.000000000","message":"Patch Set 11: Patch Set 10 was rebased\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":11},{"id":"8d796361fceccdbdec4dc0044c82a43b5fcaf91d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-09 07:23:48.000000000","message":"Patch Set 11: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/23777ccf375941c09a089fd289a43bd4\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/2dd190dfcb324ac78344c8b225b66eab : SUCCESS in 11m 57s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e7a376640ca34a7daf2d6accce508ec1 : SUCCESS in 4m 58s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/46dcac0b6af94948965723e1401cb066 : SUCCESS in 4m 43s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/96e7d4bdb7fc4442ba9f39abbee679c5 : SUCCESS in 4m 55s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/ecc1f1cf92a0400096330f32d76df179 : SUCCESS in 4m 58s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/ef867517474143f191eec502f00a6e9b : SUCCESS in 5m 16s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/70ae11c6fbf74ec89148705ba348b4dd : SUCCESS in 11m 32s\n- octavia-tox-py39-tips https://zuul.opendev.org/t/openstack/build/663c45db0c0c4e3dbb36a0cdeff5fe18 : SUCCESS in 4m 36s\n- octavia-tox-functional-py39-tips https://zuul.opendev.org/t/openstack/build/9241dd83fb544ec59183d05f2cbe1381 : SUCCESS in 6m 29s\n- octavia-tox-functional-py39-sqlalchemy-tips https://zuul.opendev.org/t/openstack/build/5aef936bbbe34d628d4387bed9a23219 : SUCCESS in 8m 47s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/328fdc503e0e4f378d64babb5526e2d8 : SUCCESS in 5m 03s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/64eb688ea57b445187d151380a41faba : SUCCESS in 6m 45s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/a30a7ba8ca06479c8a5e6b2a0bd0ae35 : SUCCESS in 7m 06s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/672c2e47132d4f7a9887350d6b5ca457 : SUCCESS in 7m 03s (non-voting)\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/4feebe262eca4272840e50bd012da534 : SUCCESS in 1h 05m 30s\n- octavia-v2-dsvm-scenario-traffic-ops https://zuul.opendev.org/t/openstack/build/c832e58f49cf478598bc070e4ef64f88 : SUCCESS in 1h 31m 17s\n- octavia-v2-dsvm-scenario-non-traffic-ops https://zuul.opendev.org/t/openstack/build/8d88aed412344aad9df80f44bd7ca0b1 : SUCCESS in 1h 10m 21s\n- octavia-v2-dsvm-scenario-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/a7a4ca367f8641e0985f1e0ff35b3493 : SUCCESS in 1h 27m 54s\n- octavia-v2-dsvm-scenario-non-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/bbc9114944a14bb998e8d8d1aaec7692 : FAILURE in 1h 08m 26s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/c58aa89c06354ce0ba32cf6d7cbaafa4 : SUCCESS in 44m 12s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/f17d498b9731497eabeafb868a9daac9 : SUCCESS in 41m 10s\n- octavia-grenade-skip-level https://zuul.opendev.org/t/openstack/build/18d15a53992e40fca4f2ef7103e21caf : FAILURE in 37m 51s\n- octavia-v2-dsvm-tls-barbican-fips https://zuul.opendev.org/t/openstack/build/08b696b3ed034c94bcbd3a51da190cb6 : SUCCESS in 1h 06m 58s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/ee82880ed7664846b20bf6c44b1e6f9f : SUCCESS in 32m 50s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/e6d378c860b74075b5fdfcc75b3e854e : SUCCESS in 1h 54m 18s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/fac038d1d35e4c0eb1b438a1ebffc59f : SUCCESS in 2h 25m 14s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/89e6ed38867a40fc964a6ec10f23b1b1 : SUCCESS in 1h 58m 16s (non-voting)\n- octavia-v2-dsvm-scenario-centos-9-stream https://zuul.opendev.org/t/openstack/build/17fdb1cd90b64a94a230ad5a4c30f0c4 : SUCCESS in 1h 54m 05s (non-voting)","accounts_in_message":[],"_revision_number":11},{"id":"f8e17167c478109e02a2359ef2a052abf2878229","author":{"_account_id":6577,"name":"Sergey Kraynev","email":"sergejyit@gmail.com","username":"skraynev"},"date":"2023-10-16 16:34:07.000000000","message":"Patch Set 11:\n\n(1 comment)","accounts_in_message":[],"_revision_number":11},{"id":"1b074f1c35995434cc9c35ba1d9568880b9f9e6b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-16 19:11:35.000000000","message":"Patch Set 11: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9c7c30e362bc4be69ce2520043bb7ec6\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/ba63f9f03752454785ffd9b342f369f4 : SUCCESS in 11m 36s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/824ca67f9c0d4c36b60eaac734f9b256 : SUCCESS in 5m 07s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/3c86ed03056745499849223778686a32 : SUCCESS in 6m 39s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/2c7e3eaebe7b448c8aaa3c9177d49f8c : SUCCESS in 5m 46s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/027465ed936347bdaae3f213c978e82f : SUCCESS in 5m 11s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/bad11f380e544e77b6c514411a673b26 : SUCCESS in 6m 04s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/dd53184919d740c1916fec108e03c612 : SUCCESS in 12m 22s\n- octavia-tox-py39-tips https://zuul.opendev.org/t/openstack/build/bacb628c8bea48fc9d13124882dcff10 : SUCCESS in 5m 35s\n- octavia-tox-functional-py39-tips https://zuul.opendev.org/t/openstack/build/195c15d9782642eba630e59f8c8c8b1e : SUCCESS in 7m 53s\n- octavia-tox-functional-py39-sqlalchemy-tips https://zuul.opendev.org/t/openstack/build/d7868ac395414935a68aa566c3ee428d : SUCCESS in 10m 11s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/b52809b9e7c048e4993e5ed51c8e5d5e : SUCCESS in 4m 54s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/094a49d1d08d423b8553a84cbc9faa8a : SUCCESS in 9m 17s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/d04d2d7a0206403dad1011a94b36f788 : SUCCESS in 6m 41s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/f8c89ec3794e479cb63eb5fd617cc6ba : SUCCESS in 8m 08s (non-voting)\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/f0ca23edab8447b5bbfe33503edb1d7e : SUCCESS in 1h 17m 24s\n- octavia-v2-dsvm-scenario-traffic-ops https://zuul.opendev.org/t/openstack/build/224aedddd83340ff9c536f4711c0d3d6 : SUCCESS in 1h 53m 19s\n- octavia-v2-dsvm-scenario-non-traffic-ops https://zuul.opendev.org/t/openstack/build/d18118d2f4ed4f6bb99955613bc0ee99 : SUCCESS in 1h 15m 28s\n- octavia-v2-dsvm-scenario-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/45a5f4bfe8e9429ab7312b15b3aa7611 : SUCCESS in 1h 57m 59s\n- octavia-v2-dsvm-scenario-non-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/9742d3f263d7449994b94ef5f79b81b8 : SUCCESS in 1h 16m 53s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/a9691b9d42a34b54b76eb1134653ba1e : SUCCESS in 36m 01s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/61dbabf56e0e4667864bc09f01614388 : SUCCESS in 38m 30s\n- octavia-grenade-skip-level https://zuul.opendev.org/t/openstack/build/8bdfce6ea0ea46e5a2f7a9567c9a428a : SUCCESS in 47m 14s (non-voting)\n- octavia-v2-dsvm-tls-barbican-fips https://zuul.opendev.org/t/openstack/build/fbd862b21d0f4da0bb149cb4a977a168 : SUCCESS in 1h 10m 52s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/dc1372e9dc09477089e217fc6c937e5f : SUCCESS in 33m 27s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/c4c264e21a514df19f5077af2403ae7f : SUCCESS in 1h 58m 30s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/651fb398c0e14d0099ab6c0940545e83 : SUCCESS in 2h 32m 07s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/769dbddfd75341509c66994a76eadc63 : SUCCESS in 2h 06m 58s (non-voting)\n- octavia-v2-dsvm-scenario-centos-9-stream https://zuul.opendev.org/t/openstack/build/5f79d25ea31246b2a289ee1d895af45d : TIMED_OUT in 2h 34m 29s (non-voting)","accounts_in_message":[],"_revision_number":11},{"id":"47232410daf92a527ed7d34c905a08b8c4cbafea","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":6577,"name":"Sergey Kraynev","email":"sergejyit@gmail.com","username":"skraynev"},"date":"2024-02-13 19:17:20.000000000","message":"Uploaded patch set 12: Patch Set 11 was rebased.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":12},{"id":"dff0e47463525dc12b9b0944bffdc76a85152b10","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-13 22:04:21.000000000","message":"Patch Set 12: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/ba887cc8e02f441cbf8bd909db5908fd\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/e0162a57376f4c3a9fe3da36f09e3de3 : SUCCESS in 10m 41s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f7787f05cad44a6683a84ad03e68a2b2 : SUCCESS in 4m 35s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/6c8603bba6724d92af4caaffb45032fe : SUCCESS in 4m 32s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/0cd96ba3d3f04353be3e7854eb0b4ff0 : SUCCESS in 4m 15s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/80ca46c3fbbe487bae307f8f920c3ac9 : SUCCESS in 12m 16s\n- octavia-tox-py39-tips https://zuul.opendev.org/t/openstack/build/e938ae380b354c4f83bb7332c3312ad0 : SUCCESS in 5m 02s\n- octavia-tox-functional-py39-tips https://zuul.opendev.org/t/openstack/build/24ee97fc8de74dada9a89365ad19b533 : SUCCESS in 8m 17s\n- octavia-tox-functional-py39-sqlalchemy-tips https://zuul.opendev.org/t/openstack/build/1a88f99627f7400aa34f91a9a0b8a673 : SUCCESS in 9m 58s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/8eeda2bbf46741d39bd87c5588596dde : SUCCESS in 5m 23s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/74282f0f3528469bb74217d78a350245 : SUCCESS in 7m 39s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/83ab852ee677465db00fb1025d6e1cd2 : SUCCESS in 9m 47s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/9d9b79715e824f319b51c342126f944b : SUCCESS in 6m 50s (non-voting)\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/d6384b053a2d4ce5b94e5019dfe007c6 : SUCCESS in 1h 22m 07s\n- octavia-v2-dsvm-scenario-traffic-ops https://zuul.opendev.org/t/openstack/build/73afa8ab51c14d3697eb23ab44170927 : SUCCESS in 1h 34m 05s\n- octavia-v2-dsvm-scenario-non-traffic-ops https://zuul.opendev.org/t/openstack/build/f214233fc74444ca8a04bf1b395d3d46 : FAILURE in 1h 35m 52s\n- octavia-v2-dsvm-scenario-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/0ed47bc33b054fab9946eeb7207f9324 : SUCCESS in 1h 55m 59s\n- octavia-v2-dsvm-scenario-non-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/123fd858422e4c19b370ca013bf9770c : FAILURE in 1h 40m 34s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/75b40ae79c9e42c89e0f29ade3b1ba2d : SUCCESS in 48m 45s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/01f50e53fe4444238df35f67bca75bc9 : SUCCESS in 58m 42s\n- octavia-grenade-skip-level https://zuul.opendev.org/t/openstack/build/6366fb1913254abab3294d97022f59fc : SUCCESS in 44m 04s (non-voting)\n- octavia-v2-dsvm-tls-barbican-fips https://zuul.opendev.org/t/openstack/build/473389ed3f114fff9fc57febbfa1576e : FAILURE in 53m 36s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/7806e637f21b49c686a9b53f6570089f : SUCCESS in 37m 20s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/e8f00304252e435aa23156c8417a1f58 : SUCCESS in 2h 15m 00s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/c0e104a455ba44f8acce35b46b163c5d : SUCCESS in 2h 44m 36s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/eac93866a23849228796816a993c6d74 : FAILURE in 27m 04s (non-voting)\n- octavia-v2-dsvm-scenario-centos-9-stream https://zuul.opendev.org/t/openstack/build/9506def7e0b24d29aacc0102d90a7eb8 : FAILURE in 44m 32s (non-voting)","accounts_in_message":[],"_revision_number":12},{"id":"66038df033c065a867ab670713af62fdbc47a501","author":{"_account_id":6577,"name":"Sergey Kraynev","email":"sergejyit@gmail.com","username":"skraynev"},"date":"2024-02-14 06:29:57.000000000","message":"Patch Set 12:\n\n(1 comment)","accounts_in_message":[],"_revision_number":12},{"id":"af4de76710e838d837944828cdc4d20aa26d5c39","author":{"_account_id":6577,"name":"Sergey Kraynev","email":"sergejyit@gmail.com","username":"skraynev"},"date":"2024-02-14 06:31:02.000000000","message":"Patch Set 12:\n\n(1 comment)","accounts_in_message":[],"_revision_number":12},{"id":"12e05333379633dabc5df2018fdf57a35a2ed8ae","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-14 09:39:12.000000000","message":"Patch Set 12:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/68b9f4e9843c4da1b99fec816d03c3be\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/7e8e8ec2b006409e9bbe3bc28a90b3de : SUCCESS in 10m 51s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/26df5291c92742cca9021074954e592e : SUCCESS in 4m 42s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/5cd401a7f4dd4ca68d672e3db4c74635 : SUCCESS in 4m 32s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/6e476ceecffe4901a889d0f8c7ea78ae : SUCCESS in 7m 16s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5207bfca2a604c35bbecae3f47403f8f : SUCCESS in 9m 42s\n- octavia-tox-py39-tips https://zuul.opendev.org/t/openstack/build/3c9f99d718794f1cb7be3c55a579872f : SUCCESS in 4m 42s\n- octavia-tox-functional-py39-tips https://zuul.opendev.org/t/openstack/build/1a0df2aa11274d3dbd16e98740b91f59 : SUCCESS in 7m 11s\n- octavia-tox-functional-py39-sqlalchemy-tips https://zuul.opendev.org/t/openstack/build/75efb4cde01b44d6a74d1fbe824f2d1a : SUCCESS in 9m 57s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/bd70fb0ad6ea479cb84352c255c50c88 : SUCCESS in 4m 03s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/59526fd9539c4b71a2df9bfed2f9b005 : SUCCESS in 7m 00s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/cad42a28a0b648b483b9dbee9a01e15c : SUCCESS in 8m 13s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/6b5621c36d84458f803784eefd0a7a63 : SUCCESS in 6m 28s (non-voting)\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/f895eacbfe684715a82598b7e16425ba : SUCCESS in 1h 02m 32s\n- octavia-v2-dsvm-scenario-traffic-ops https://zuul.opendev.org/t/openstack/build/afe853c7d10e43f194bccb5fc9c0ee69 : SUCCESS in 2h 16m 27s\n- octavia-v2-dsvm-scenario-non-traffic-ops https://zuul.opendev.org/t/openstack/build/7ad6088e26ea48ee96b64f3f57dc99a7 : FAILURE in 1h 19m 20s\n- octavia-v2-dsvm-scenario-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/0d22b7ac060b4d58b73be9a8f563084c : SUCCESS in 1h 37m 21s\n- octavia-v2-dsvm-scenario-non-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/d8f5774809d545009e0213d76d3edf11 : FAILURE in 1h 32m 47s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/32f82046eb6d40af8d042e53e7040332 : SUCCESS in 33m 56s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/0979aff6141944f59cd0cc0a8bf5899b : SUCCESS in 38m 07s\n- octavia-grenade-skip-level https://zuul.opendev.org/t/openstack/build/9afdfe5c5106466f95e6de6a0f35a326 : SUCCESS in 37m 02s (non-voting)\n- octavia-v2-dsvm-tls-barbican-fips https://zuul.opendev.org/t/openstack/build/455179436eab4c529d8f949fe824eea9 : FAILURE in 49m 39s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/7c6e64f8602a43e8bfca09ccdf6e48ca : SUCCESS in 42m 16s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/3e41dccee8af4dc0aa2536d21c3452ff : TIMED_OUT in 2h 34m 03s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/91e1fc6df831477abccaf796255e4663 : TIMED_OUT in 3h 04m 32s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/d647198827914396ac609b548503c3c7 : FAILURE in 20m 33s (non-voting)\n- octavia-v2-dsvm-scenario-centos-9-stream https://zuul.opendev.org/t/openstack/build/2a2d54190ce4408fbd8f7f4bffd42c5a : FAILURE in 39m 29s (non-voting)","accounts_in_message":[],"_revision_number":12},{"id":"3952b282b0a6d0d6f87732ac12ee0c384ea547a6","author":{"_account_id":6577,"name":"Sergey Kraynev","email":"sergejyit@gmail.com","username":"skraynev"},"date":"2024-02-14 11:14:13.000000000","message":"Patch Set 12:\n\n(1 comment)","accounts_in_message":[],"_revision_number":12},{"id":"eb1017377645e1b453ed55522b96fbc649d00c76","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2024-02-14 11:35:02.000000000","message":"Patch Set 12:\n\n(1 comment)","accounts_in_message":[],"_revision_number":12},{"id":"bf4773b7eba5d2bc3b8f7b168d8d5d90ee345c35","author":{"_account_id":6577,"name":"Sergey Kraynev","email":"sergejyit@gmail.com","username":"skraynev"},"date":"2024-02-14 13:23:41.000000000","message":"Patch Set 12:\n\n(1 comment)","accounts_in_message":[],"_revision_number":12},{"id":"0f794b7d3cce9cdeb7079074ef7dad6513c19ec2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-14 13:41:38.000000000","message":"Patch Set 12: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ea0785f1f6c24a57a26da027fcb3ec5f\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/9c2c11fac697475fb14655844e08f779 : SUCCESS in 10m 06s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/04bbf144320c4aa497279a6a771c5e2a : SUCCESS in 4m 24s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/a096bd5813494431ab3b9568569553d8 : SUCCESS in 4m 33s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/c34d0fcf333c41b69f0255b0494f1063 : SUCCESS in 5m 20s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/04f217ccdd414428afacf7fca16b4d6a : SUCCESS in 15m 43s\n- octavia-tox-py39-tips https://zuul.opendev.org/t/openstack/build/9f86471d50e54bf9a340504a3dedb07c : SUCCESS in 5m 49s\n- octavia-tox-functional-py39-tips https://zuul.opendev.org/t/openstack/build/d26212adeb054d5fb080f70119e3e373 : SUCCESS in 7m 36s\n- octavia-tox-functional-py39-sqlalchemy-tips https://zuul.opendev.org/t/openstack/build/d7e4b71efda74f35b6fa777eb3dd0740 : SUCCESS in 9m 46s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/0d31d43f71e1413e8b64cdb9f57455b2 : SUCCESS in 4m 44s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/190ee3e91fec4c6084f5c707bc9e6c00 : SUCCESS in 9m 33s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/594f553b81c841e4ab37f9fe298cc8e2 : SUCCESS in 9m 05s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/dc17d5a1d8f3494a9867e16cafe01ca2 : SUCCESS in 8m 08s (non-voting)\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/0f56fdf7e01645699b35217104c420c9 : SUCCESS in 1h 27m 46s\n- octavia-v2-dsvm-scenario-traffic-ops https://zuul.opendev.org/t/openstack/build/513a5d7680b346249480a46c9e6a582d : SUCCESS in 1h 38m 37s\n- octavia-v2-dsvm-scenario-non-traffic-ops https://zuul.opendev.org/t/openstack/build/92f96aff0e754ccea74c3d5a2def9823 : SUCCESS in 1h 04m 57s\n- octavia-v2-dsvm-scenario-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/b510997e25144475b6fd3aa98f233f91 : SUCCESS in 1h 43m 17s\n- octavia-v2-dsvm-scenario-non-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/e9b91e1942bd4ae186c992bd945190ef : SUCCESS in 1h 02m 23s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/308797342cc844a483ef98b7fb3901a6 : SUCCESS in 39m 47s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/e430b4246926446ea07feac223741135 : SUCCESS in 38m 57s\n- octavia-grenade-skip-level https://zuul.opendev.org/t/openstack/build/24a4064a4def43fdbe8b2bb12e1c16c7 : SUCCESS in 39m 07s (non-voting)\n- octavia-v2-dsvm-tls-barbican-fips https://zuul.opendev.org/t/openstack/build/432d74499018410aaa29fd20e2986a06 : FAILURE in 49m 25s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/8aa87f2527724363ad3927cdd9806370 : SUCCESS in 33m 36s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/b407123176524ce0a286392dfc800f98 : SUCCESS in 1h 55m 41s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/6a778ec650d94d948cb07158bcc25308 : SUCCESS in 2h 25m 13s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/93001f049e1045f1a915403a5c8a571f : FAILURE in 22m 03s (non-voting)\n- octavia-v2-dsvm-scenario-centos-9-stream https://zuul.opendev.org/t/openstack/build/bb6f91de00614693ad4b5c5a531ad9d0 : FAILURE in 41m 14s (non-voting)","accounts_in_message":[],"_revision_number":12},{"id":"e5ce644981a1119e79720e7d4391a9b19d7e5aab","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":6577,"name":"Sergey Kraynev","email":"sergejyit@gmail.com","username":"skraynev"},"date":"2024-02-27 08:41:24.000000000","message":"Patch Set 13: Patch Set 12 was rebased\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":13},{"id":"2109ec94d64015642d82ddb914a7130b2918ad69","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-27 11:39:19.000000000","message":"Patch Set 13: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b9303c5bb4f8477287c656d78da7673f\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/e4f42230bb154cffa7e84d71192431af : SUCCESS in 12m 24s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/082eb1b4d34c4ad2898248d52045f943 : SUCCESS in 4m 59s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/f3d4b2869dcb44d4b304c74f37a33d4d : SUCCESS in 5m 07s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/c2a553f650e94306804d0f544dfb80c7 : SUCCESS in 5m 12s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/07e016f105834a328d481496a5b2f4d8 : SUCCESS in 12m 32s\n- octavia-tox-py39-tips https://zuul.opendev.org/t/openstack/build/d6a33dca69c7446a9e52f26f5b3fa33f : SUCCESS in 4m 49s\n- octavia-tox-functional-py39-tips https://zuul.opendev.org/t/openstack/build/74e6a9f2bcab4237b382eea6cf8bae40 : SUCCESS in 8m 01s\n- octavia-tox-functional-py39-sqlalchemy-tips https://zuul.opendev.org/t/openstack/build/637a48e357584fcf851f9c86ee926134 : SUCCESS in 9m 46s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/2c499a4422784a1293461a159087052f : SUCCESS in 4m 07s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/7968de2149e244f693e433634353ea1a : SUCCESS in 7m 52s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/55899ce74f614d3b83ab259ce0fc7846 : SUCCESS in 6m 36s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/d87bbe50f7e04522aded1ae6c41c0d0a : SUCCESS in 7m 31s (non-voting)\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/85936794ed2349f585e6ba47e4904a8c : SUCCESS in 1h 38m 56s\n- octavia-v2-dsvm-scenario-traffic-ops https://zuul.opendev.org/t/openstack/build/3d696102c9704557a92321148a1d8d87 : SUCCESS in 2h 11m 57s\n- octavia-v2-dsvm-scenario-non-traffic-ops https://zuul.opendev.org/t/openstack/build/092b6c42682a411faee89b6855b301e5 : SUCCESS in 1h 11m 40s\n- octavia-v2-dsvm-scenario-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/6c9d6253d3c145ee903756df123f4070 : SUCCESS in 2h 12m 47s\n- octavia-v2-dsvm-scenario-non-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/56b8aac59fa4491c95b24ffb0373314e : SUCCESS in 1h 00m 12s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/de18a96a02904953b3a3215b8322dc96 : SUCCESS in 49m 08s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/aff0349fe4344cccb64d948cf986eb72 : SUCCESS in 42m 10s\n- octavia-grenade-skip-level https://zuul.opendev.org/t/openstack/build/cbecdf90dd7843fb9914b96c8c94df80 : SUCCESS in 51m 07s (non-voting)\n- octavia-v2-dsvm-tls-barbican-fips https://zuul.opendev.org/t/openstack/build/7745f0066455450388b0def31828cf0b : SUCCESS in 1h 07m 02s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/580498ba6a534adb9abb76b8bc8aa2fd : SUCCESS in 35m 23s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/a912513d896845379f5e33d161402c0c : SUCCESS in 2h 14m 11s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/7dec06a9ddcc4cf6838b587b77833bfc : SUCCESS in 2h 56m 36s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/6d8cc2036c8a49c2b427860b34412d24 : FAILURE in 21m 43s (non-voting)\n- octavia-v2-dsvm-scenario-centos-9-stream https://zuul.opendev.org/t/openstack/build/a1b5f26034af4768a7fc5470cc777905 : FAILURE in 10m 37s (non-voting)","accounts_in_message":[],"_revision_number":13},{"id":"6cdcf188343e36424b642dc2f4cdef305a7221a7","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":6577,"name":"Sergey Kraynev","email":"sergejyit@gmail.com","username":"skraynev"},"date":"2024-03-14 17:28:30.000000000","message":"Patch Set 14: Patch Set 13 was rebased\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":14},{"id":"62e56feede5f544cc270f087e581cddc497155d0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-14 20:33:17.000000000","message":"Patch Set 14: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3a0acf57cb0c4c389bb208ebc9ecf8ee\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/12d372d093084c659a722b3973591f54 : SUCCESS in 10m 56s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f9b5dc7339ac4647be52f5cce5e59e39 : SUCCESS in 4m 50s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/b3ba649e34b845d8bdc82a6fa10c7767 : SUCCESS in 6m 11s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/6854ee29643840b1ae181bef81529300 : SUCCESS in 5m 16s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3892cc1c52804d068ff0f12d53a3b6f2 : SUCCESS in 10m 39s\n- octavia-tox-py39-tips https://zuul.opendev.org/t/openstack/build/146c309c92394313a9a2eef075fdbf8d : SUCCESS in 5m 06s\n- octavia-tox-functional-py39-tips https://zuul.opendev.org/t/openstack/build/3b48981eb69547678e68522e5fe0a589 : SUCCESS in 6m 45s\n- octavia-tox-functional-py39-sqlalchemy-tips https://zuul.opendev.org/t/openstack/build/0dc2dd9fd94a49a1b104a53638a2eab1 : SUCCESS in 12m 17s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/61e2a7cd26154cd7adbcf2782fe280c8 : SUCCESS in 4m 43s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/ff96d4775a824317972a98d7be511613 : SUCCESS in 15m 26s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/f3f1f60411ab419dbc4573543936546f : SUCCESS in 9m 50s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/2f8104a44763439b9697f6c9333dc965 : SUCCESS in 8m 26s (non-voting)\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/48cafbd0dece4324b8a31089b73fd720 : SUCCESS in 1h 23m 04s\n- octavia-v2-dsvm-scenario-traffic-ops https://zuul.opendev.org/t/openstack/build/694cfe3faca14b6f9b7d6442f159c071 : SUCCESS in 2h 23m 33s\n- octavia-v2-dsvm-scenario-non-traffic-ops https://zuul.opendev.org/t/openstack/build/a9af1b83c2804190b0fe4b640b1cce66 : SUCCESS in 59m 09s\n- octavia-v2-dsvm-scenario-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/4b659f0a1c014479afefbfad4cf98069 : SUCCESS in 2h 22m 41s\n- octavia-v2-dsvm-scenario-non-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/88170e3eca3145a392b3c2e973e79950 : SUCCESS in 58m 13s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/2946c708e52a462c9acf36ccb38a1410 : SUCCESS in 1h 00m 43s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/2fa9489542a74a4982687bbfd4e824bd : SUCCESS in 39m 45s\n- octavia-grenade-skip-level https://zuul.opendev.org/t/openstack/build/a3c66e13cf4a41b2891f42821d00f40b : SUCCESS in 41m 41s (non-voting)\n- octavia-v2-dsvm-tls-barbican-fips https://zuul.opendev.org/t/openstack/build/583e88637c944dd5af9bd0c0e0bdf2e6 : SUCCESS in 1h 07m 57s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/bea738d760234ea48f157b35e57f03b2 : SUCCESS in 34m 40s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/2aaf0c33144149fd94074eae7e96df72 : SUCCESS in 2h 14m 10s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/0d21844816b543a2a58e4fb484e70bab : SUCCESS in 3h 02m 06s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/a8e5f4ee8e58426388ea7300c687e501 : FAILURE in 31m 55s (non-voting)\n- octavia-v2-dsvm-scenario-centos-9-stream https://zuul.opendev.org/t/openstack/build/105eecef47254f57b4d57beed5b1413b : TIMED_OUT in 2h 33m 18s (non-voting)","accounts_in_message":[],"_revision_number":14},{"id":"7f7286f90982facbb0d78ef907eb3f7ae6643bd4","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":6577,"name":"Sergey Kraynev","email":"sergejyit@gmail.com","username":"skraynev"},"date":"2024-04-04 05:57:01.000000000","message":"Patch Set 15: Patch Set 14 was rebased\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":15},{"id":"d5ed528ea03eb9afc859cb6dab10f93ac4be2b78","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-04 08:46:23.000000000","message":"Patch Set 15: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/66b225c70a0a4cf191d2523030e36915\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/de4e9a4937f94f1ea5a0e0d97a53558b : SUCCESS in 9m 03s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/85eb174677e8421a89d031a863f18d84 : SUCCESS in 3m 57s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/ab463b091d2f4f1aa61a8b1c0e67ff02 : SUCCESS in 4m 20s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/1447209532224ff3a7db421c61910112 : SUCCESS in 4m 28s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/87da1c083a5547ffb1223164778dff7b : SUCCESS in 7m 59s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4ad6de53e8be447680ba80c52790f85a : SUCCESS in 11m 24s\n- octavia-tox-py39-tips https://zuul.opendev.org/t/openstack/build/bbeca9d2be3740c08dd87bd24844154b : SUCCESS in 4m 49s\n- octavia-tox-functional-py39-tips https://zuul.opendev.org/t/openstack/build/cc841543657d4ff09f6ec82e28169732 : SUCCESS in 6m 57s\n- octavia-tox-functional-py39-sqlalchemy-tips https://zuul.opendev.org/t/openstack/build/d632a29dd7b345a5b4fc94e3dcb8fdbb : SUCCESS in 7m 59s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/13bb2617de5947969290dae2710fd564 : SUCCESS in 4m 19s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/b3c773ee499a4064900c5f38edcbbfe7 : SUCCESS in 7m 00s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/14d7431c1fc344bdb1cb841bddb35097 : SUCCESS in 6m 08s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/cff8a9bef3c147029ba0ad997533474b : SUCCESS in 7m 06s (non-voting)\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/82d37d2c86a64a7f93488bdb0be7959e : SUCCESS in 1h 47m 59s\n- octavia-v2-dsvm-scenario-traffic-ops https://zuul.opendev.org/t/openstack/build/22aef4ed77e74f6e865c0495a75565f3 : SUCCESS in 1h 46m 16s\n- octavia-v2-dsvm-scenario-non-traffic-ops https://zuul.opendev.org/t/openstack/build/04aa5d87a5f7445ab0cb8b081aeab301 : SUCCESS in 1h 13m 29s\n- octavia-v2-dsvm-scenario-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/03b730f8ca2249448a3175583ab84010 : SUCCESS in 1h 49m 56s\n- octavia-v2-dsvm-scenario-non-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/aceb5f7b104848d6bed055fefb93a4a5 : SUCCESS in 1h 09m 56s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/d23d3de3d4ce4747a9b037db41880052 : SUCCESS in 45m 33s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/b8fb4f35e83c4631922d6143df8dce7f : SUCCESS in 50m 03s\n- octavia-v2-dsvm-tls-barbican-fips https://zuul.opendev.org/t/openstack/build/d1647476039d47a291828008928724a1 : SUCCESS in 59m 47s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/80a0da1edc6e410bb27419a2652bd156 : SUCCESS in 43m 46s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/e03f9d39e81f407587cd2a9795787023 : SUCCESS in 2h 31m 15s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/ffbbd15d69314b73b2ccfaf6d16a7475 : SUCCESS in 2h 46m 52s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/d598b0f935144e1db1042a812129f8cb : FAILURE in 21m 36s (non-voting)\n- octavia-v2-dsvm-scenario-centos-9-stream https://zuul.opendev.org/t/openstack/build/fc4731037b9a42a7b94c74f51f833829 : TIMED_OUT in 2h 33m 00s (non-voting)","accounts_in_message":[],"_revision_number":15},{"id":"5050e7f7a848bd229aba9f653b79cb5b6e476681","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":6577,"name":"Sergey Kraynev","email":"sergejyit@gmail.com","username":"skraynev"},"date":"2024-04-16 03:23:03.000000000","message":"Patch Set 16: Patch Set 15 was rebased\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":16},{"id":"7c8ce8c283c763d77f52f62e0d6debbbd5490726","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-16 06:11:48.000000000","message":"Patch Set 16: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3b7023ab44424af89d9f0238d285dd66\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/1dba64b130264e2a9bc4699795ab0ed0 : SUCCESS in 12m 41s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3f98cd305dd24c24af54d4a4b531158d : SUCCESS in 4m 44s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/2f92ed0681f14f44a08c854544455201 : SUCCESS in 4m 47s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/e2d7cd98c1e54e19a18fdb430c90db80 : SUCCESS in 5m 18s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/ce83b5bc9e934b6e8ff0576b4cdb20c0 : SUCCESS in 7m 50s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/30fa66d92fac402f93a6c513303dd0cd : SUCCESS in 9m 49s\n- octavia-tox-py39-tips https://zuul.opendev.org/t/openstack/build/bd0e63ca39a947e5aee67f6905b7e067 : SUCCESS in 4m 36s\n- octavia-tox-functional-py39-tips https://zuul.opendev.org/t/openstack/build/6aa3cef8334e43cdb6255985b874f741 : SUCCESS in 7m 05s\n- octavia-tox-functional-py39-sqlalchemy-tips https://zuul.opendev.org/t/openstack/build/0774030d1e264ad9a7411223ddc8bf13 : SUCCESS in 10m 24s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/71f4cb9917134d0a96de28c95314be5d : SUCCESS in 4m 36s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/823fc38e35d54cae8ad2592811986ce1 : SUCCESS in 6m 45s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/18a396b087e24c6dafe7e5cbdfbfdcc7 : SUCCESS in 6m 56s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/21be13eb870743d8a2c1276fb5a36c53 : SUCCESS in 6m 05s (non-voting)\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/78c87a30c5a842b6ac12f64c745fe717 : SUCCESS in 1h 24m 58s\n- octavia-v2-dsvm-scenario-traffic-ops https://zuul.opendev.org/t/openstack/build/5cf4102dd24c436abe5343bbf7c61c9d : SUCCESS in 1h 37m 57s\n- octavia-v2-dsvm-scenario-non-traffic-ops https://zuul.opendev.org/t/openstack/build/55ab825037e04fdfbbc2d8e44027cbce : SUCCESS in 56m 03s\n- octavia-v2-dsvm-scenario-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/ebdfb38bdec84d0a89c1dddfa9db11ec : SUCCESS in 1h 45m 24s\n- octavia-v2-dsvm-scenario-non-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/74d8287f7a9c4d2a9d8e4f4560802394 : SUCCESS in 1h 04m 02s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/8974d55506de4ee589e84d542649f952 : SUCCESS in 44m 19s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/9313dfb3c9034aa78c3053a613bc2bab : SUCCESS in 47m 48s\n- octavia-v2-dsvm-tls-barbican-fips https://zuul.opendev.org/t/openstack/build/f226f7e97fca48428b41b1c80522b24d : SUCCESS in 1h 01m 18s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/e5622cce3dee4afa9ac98ae6fe0940a7 : SUCCESS in 31m 40s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/9e50edbdfbaa4717835a448ef3260f13 : SUCCESS in 2h 31m 11s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/9078d2a85c4e4f48b45f69bb3e9de541 : SUCCESS in 2h 47m 17s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/2704a72e5a3e4313985c3c4f12bb6341 : FAILURE in 31m 14s (non-voting)\n- octavia-v2-dsvm-scenario-centos-9-stream https://zuul.opendev.org/t/openstack/build/c9dc9672d3834c4bbdc2758b57d9bec6 : SUCCESS in 2h 25m 23s (non-voting)","accounts_in_message":[],"_revision_number":16},{"id":"7013a522e0efcdb585402188e8e276e8564f62d5","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":6577,"name":"Sergey Kraynev","email":"sergejyit@gmail.com","username":"skraynev"},"date":"2024-04-18 17:49:36.000000000","message":"Patch Set 17: Patch Set 16 was rebased\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":17},{"id":"3cbcef85a27378ccd400c61ced477f0cff17d0bd","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-18 20:30:35.000000000","message":"Patch Set 17: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1543064d4fb449f1b4cbef459f2f773f\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/a85300b8c6914161bc0b617a3805dcbc : SUCCESS in 10m 05s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7c1b086cec0f43219d6fb8a636305b15 : SUCCESS in 4m 36s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/ab3aa080a68d4f988f1c6624c38a4550 : SUCCESS in 5m 33s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/e45985530d1849798fd07c19e22ffccb : SUCCESS in 5m 55s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/de0df847a47444369bb28b9a03c14240 : SUCCESS in 7m 38s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c6437e64603144fe9024954b17e89e43 : SUCCESS in 11m 26s\n- octavia-tox-py39-tips https://zuul.opendev.org/t/openstack/build/97029289c4e54dce94989a38124f8ba6 : SUCCESS in 4m 50s\n- octavia-tox-functional-py39-tips https://zuul.opendev.org/t/openstack/build/e60bbb60000a45bd91414d77450d77ff : SUCCESS in 8m 38s\n- octavia-tox-functional-py39-sqlalchemy-tips https://zuul.opendev.org/t/openstack/build/8d4848c8453647e480b953eb1f75aced : SUCCESS in 9m 35s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/bfab8d08a4a340a6b34e7255bd12a01d : SUCCESS in 4m 17s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/94eb5e1c5b1741038f51edd724865f40 : SUCCESS in 6m 46s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/9a4ec7e27058468da8ce505b85ad3fa8 : SUCCESS in 7m 51s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/b7c75c1d08f3451aaf11aaa245bf61af : SUCCESS in 6m 29s (non-voting)\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/3f1a9b42548e44af9c2487549de87317 : SUCCESS in 1h 29m 14s\n- octavia-v2-dsvm-scenario-traffic-ops https://zuul.opendev.org/t/openstack/build/af5bd500d3de45bb9a51b95945a4905a : SUCCESS in 1h 55m 08s\n- octavia-v2-dsvm-scenario-non-traffic-ops https://zuul.opendev.org/t/openstack/build/5a80da17288a45278b606b4af5383d4e : SUCCESS in 1h 29m 46s\n- octavia-v2-dsvm-scenario-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/9cfc20f22b8b47aabaa40751e73ed10a : SUCCESS in 2h 10m 02s\n- octavia-v2-dsvm-scenario-non-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/e920441a58054ebfb4889dafbd8a75b8 : SUCCESS in 55m 32s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/ac1cf51a5fb5445db4c7c49c0bf12178 : SUCCESS in 34m 59s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/74ef7963cab840f4a4cdcf92fb0a7cf1 : SUCCESS in 42m 51s\n- octavia-v2-dsvm-tls-barbican-fips https://zuul.opendev.org/t/openstack/build/dce4bfc68ba74d7591e21d6638d86361 : SUCCESS in 1h 09m 46s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/43f63103f9d04ac1b8fee2aef6651a53 : SUCCESS in 32m 47s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/5c61781ad1fe45f39e7e023f42db8db5 : TIMED_OUT in 2h 32m 50s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/15253f270fec4038b2de08a10f31815e : SUCCESS in 2h 27m 29s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/50a0ed4127714b05ab2776ec049c47a3 : FAILURE in 24m 56s (non-voting)\n- octavia-v2-dsvm-scenario-centos-9-stream https://zuul.opendev.org/t/openstack/build/12458ca12421428ca244762a68a92e70 : TIMED_OUT in 2h 34m 34s (non-voting)","accounts_in_message":[],"_revision_number":17},{"id":"4d8ad4b1f3a68d2fa9c95c310c0e8d4e32c69ef2","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":6577,"name":"Sergey Kraynev","email":"sergejyit@gmail.com","username":"skraynev"},"date":"2024-05-02 05:55:28.000000000","message":"Patch Set 18: Patch Set 17 was rebased\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":18},{"id":"6f31eefab809e4a2fe002bb3b6b069f8e7a5b708","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-02 08:33:34.000000000","message":"Patch Set 18: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3a1daf4e7751486489843050f58bf4fd\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/132b5336933741cab982813f44ecd2af : SUCCESS in 10m 02s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0cba79df4e7a4e2d99edd8a32339d729 : SUCCESS in 5m 20s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/a1eab48337424d01acf5be8574675123 : SUCCESS in 4m 25s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/c20fef0d876e415190ba7240e4187c9b : SUCCESS in 4m 35s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/7a7bcc96698b446b8aaca8af5e62f1a2 : SUCCESS in 8m 02s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f85347658c934fdfbaa64068273cce29 : SUCCESS in 12m 46s\n- octavia-tox-py311-tips https://zuul.opendev.org/t/openstack/build/91c458be767a4509b89bef5edb1f5f7f : SUCCESS in 3m 56s\n- octavia-tox-functional-py311-tips https://zuul.opendev.org/t/openstack/build/82e8a1338b754eff88ee260a587ec763 : SUCCESS in 4m 18s\n- octavia-tox-functional-py311-sqlalchemy-tips https://zuul.opendev.org/t/openstack/build/851ba57e11ac452ba43891699774d917 : SUCCESS in 5m 28s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/98e5c9a3a58c444aae317d6401ff7d66 : SUCCESS in 3m 33s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/5e3f364faa56441d9e4c378433066879 : SUCCESS in 6m 24s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/5b8c20f2fb8a46a5a956d375374c10b6 : SUCCESS in 8m 27s\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/5c1b04a940ba4e93b0fd5b397a08f5e1 : SUCCESS in 4m 58s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/3f3b613a6c34465a86018efb6b08d762 : SUCCESS in 1h 47m 42s\n- octavia-v2-dsvm-scenario-traffic-ops https://zuul.opendev.org/t/openstack/build/7c3936a7d2f0459caa7cfaf2b9ba2b23 : SUCCESS in 1h 54m 31s\n- octavia-v2-dsvm-scenario-non-traffic-ops https://zuul.opendev.org/t/openstack/build/62fa1cfbbb3343759594db8b74d23c3d : SUCCESS in 1h 05m 31s\n- octavia-v2-dsvm-scenario-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/e62782b4389949caafe73f684c3414d3 : SUCCESS in 1h 47m 32s\n- octavia-v2-dsvm-scenario-non-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/340f4142ad1a443ca3fa2a58656d33ee : SUCCESS in 1h 12m 09s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/ff6a223c7d3e44c8a8cf01393e8d385c : SUCCESS in 41m 35s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/09cb573a4e464c828453d85d29ecc5c0 : SUCCESS in 37m 45s\n- octavia-v2-dsvm-tls-barbican-fips https://zuul.opendev.org/t/openstack/build/9883bf0614e6451ab7a148cd72448e20 : SUCCESS in 57m 28s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/d24ed624fdf64f3498ffdaa530003e41 : SUCCESS in 32m 10s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/5057f9f202f8421287433f1b637b0c04 : SUCCESS in 2h 22m 34s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/9979286aa721422bb1fb24c74e1845fa : SUCCESS in 2h 36m 39s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/a4d37e6168194ce4a5eae55bf7b9a1d2 : FAILURE in 28m 30s (non-voting)\n- octavia-v2-dsvm-scenario-centos-9-stream https://zuul.opendev.org/t/openstack/build/a6e554b582dd4fe2aea0ae4cbfec106c : SUCCESS in 2h 12m 57s (non-voting)","accounts_in_message":[],"_revision_number":18},{"id":"a00631876a54455aa2e697e2a126321c77693f03","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":6577,"name":"Sergey Kraynev","email":"sergejyit@gmail.com","username":"skraynev"},"date":"2024-05-20 07:24:59.000000000","message":"Patch Set 19: Patch Set 18 was rebased\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":19},{"id":"3aedbe88a64593e684c0521b970ae5fce0fbaabb","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-20 10:25:43.000000000","message":"Patch Set 19: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e121a589a0be414c8b55a8ceb128a164\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/e3750f6a6ca042a09c8eb9da6faf34b6 : SUCCESS in 9m 43s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/45d413c5fbbe4abd927f4fbd40f10200 : SUCCESS in 4m 18s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/cb10b5135e294624a80bf7dbbbf2be55 : SUCCESS in 4m 35s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/793311beaa1349c18021ebec1ea38ffe : SUCCESS in 4m 46s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/0e068bd1b5aa4db4b124a712858d9cee : SUCCESS in 8m 56s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6078be7a6f2e47fca6021ca7c273b13c : SUCCESS in 11m 23s\n- octavia-tox-py311-tips https://zuul.opendev.org/t/openstack/build/e25e5f03e39c49d3899510f8854ec992 : SUCCESS in 5m 13s\n- octavia-tox-functional-py311-tips https://zuul.opendev.org/t/openstack/build/6e3bd417c35f4dcdb17847a534b60f7e : SUCCESS in 6m 35s\n- octavia-tox-functional-py311-sqlalchemy-tips https://zuul.opendev.org/t/openstack/build/0e42f161ccb54b04a37cbb643fcbf727 : SUCCESS in 6m 06s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/e418c9b307084a57a3c3c90992dc46f8 : SUCCESS in 4m 46s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/9028497825fa44e2a6418351562e9575 : SUCCESS in 6m 59s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/a1bc344639af4b2397226b99d4be119b : SUCCESS in 7m 58s\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/56cdd5cf6f804ffd9e55681be8baee9e : SUCCESS in 5m 02s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/07d14341bac34cf8928cd79c0f99c301 : SUCCESS in 1h 41m 56s\n- octavia-v2-dsvm-scenario-traffic-ops https://zuul.opendev.org/t/openstack/build/39947ccb7731404ebb2e780f8d2771a7 : SUCCESS in 1h 41m 20s\n- octavia-v2-dsvm-scenario-non-traffic-ops https://zuul.opendev.org/t/openstack/build/ac479b6f44e94f26b9d72fd29f0ac222 : SUCCESS in 1h 21m 26s\n- octavia-v2-dsvm-scenario-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/6373ca3409d24301afdd01b914d09662 : SUCCESS in 1h 57m 19s\n- octavia-v2-dsvm-scenario-non-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/fc118ec993a5462abf3969a7789874a3 : SUCCESS in 58m 40s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/0b5a2dc983e74b74875d3920a2b08a36 : SUCCESS in 52m 39s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/683163d9df6349d9bb817fdd7827879f : SUCCESS in 39m 48s\n- octavia-v2-dsvm-tls-barbican-fips https://zuul.opendev.org/t/openstack/build/edbb0154726e48f8a8f40bc6962b3e6a : SUCCESS in 1h 07m 15s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/6bdf97bc71014ada88177874a541f38f : SUCCESS in 47m 30s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/94b3e38cc09b4bfaa6c26d0f7d9bb560 : SUCCESS in 2h 00m 40s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/7ef9a6a6a9c4478b9be02c2b5a065cdd : SUCCESS in 2h 58m 43s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/a3c2c6b64bc141cfa8905675817b0a1a : FAILURE in 23m 57s (non-voting)\n- octavia-v2-dsvm-scenario-centos-9-stream https://zuul.opendev.org/t/openstack/build/5c3fe96fbd354ae8b65a79c4506dec42 : TIMED_OUT in 2h 33m 41s (non-voting)","accounts_in_message":[],"_revision_number":19},{"id":"59bd80e6cb000123c4fd1803307260f718f84484","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2024-06-12 16:14:31.000000000","message":"Patch Set 19: Code-Review+2","accounts_in_message":[],"_revision_number":19},{"id":"e0231d1942099e16379a06ae051bb3ab5aff46c3","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":6577,"name":"Sergey Kraynev","email":"sergejyit@gmail.com","username":"skraynev"},"date":"2024-08-19 11:55:29.000000000","message":"Patch Set 20: Patch Set 19 was rebased\n\nCopied Votes:\n* Code-Review+2 (copy condition: \"**changekind:TRIVIAL_REBASE** OR is:MIN\")\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":20},{"id":"67c422ee767b545fe7506d96e5a8d66e33c3e054","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-19 14:59:54.000000000","message":"Patch Set 20: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/ecfc0972c4bc45758c7a0b863efecf80\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/3b4a565a02034ec8bb47804ae3ef5f00 : SUCCESS in 11m 39s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/830de963a91d44e0893319979b3d98bb : SUCCESS in 4m 47s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/f3e68471b3e545e49cbe3c095dd9a783 : SUCCESS in 4m 49s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/2a2fe9cee2074cdbae2bd795e6ec60ce : SUCCESS in 4m 30s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/925c19ed717f4b988683903da70e4161 : SUCCESS in 5m 59s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/844196d445d74a0dbf6ed7161261ac44 : SUCCESS in 10m 02s\n- octavia-tox-py311-tips https://zuul.opendev.org/t/openstack/build/0a30e3f65866454ca10bcc89a0df8be7 : SUCCESS in 6m 39s\n- octavia-tox-functional-py311-tips https://zuul.opendev.org/t/openstack/build/abbfc3fdfeda4278981fcb22977893e6 : FAILURE in 8m 27s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/59169ae4eddb4fbe9120a32b66c5c5a4 : SUCCESS in 5m 58s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/efb8aefad8064896aa4df0580b7fe04f : SUCCESS in 6m 49s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/c2ec50cdd85c4463ba3b3c393f0b410d : SUCCESS in 7m 20s\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/dffd552c118b4118be3c3c3e493084a1 : SUCCESS in 8m 39s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/883dde81ab4c44e9877bd947fe75f4ce : SUCCESS in 1h 47m 07s\n- octavia-v2-dsvm-scenario-traffic-ops https://zuul.opendev.org/t/openstack/build/7fad761a026146dea06826788dfb83b0 : SUCCESS in 1h 45m 49s\n- octavia-v2-dsvm-scenario-non-traffic-ops https://zuul.opendev.org/t/openstack/build/a5831a92865a45518722141e05feb63f : SUCCESS in 1h 24m 23s\n- octavia-v2-dsvm-scenario-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/77d77f87aa434acb9ba581e29f1c66a3 : SUCCESS in 1h 52m 46s\n- octavia-v2-dsvm-scenario-non-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/f82b6f57551b483696beb4d9226c142f : SUCCESS in 1h 29m 34s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/878fa0880871407fa86948a53ab8cb0e : SUCCESS in 31m 37s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/00ea87a62de14bee802a7a05aa45f77c : SUCCESS in 1h 03m 53s\n- octavia-v2-dsvm-tls-barbican-fips https://zuul.opendev.org/t/openstack/build/29ea5ea0ab5d4d64bb323a69650e81d0 : SUCCESS in 1h 04m 28s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/0f93495362464909b56b80adfa789664 : SUCCESS in 31m 07s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/2669feaf25174a49a99ec08b847ba6b0 : TIMED_OUT in 2h 33m 06s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/36319848fe4f4637aa6d44e0053a1514 : TIMED_OUT in 3h 03m 07s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/6e234a53db52438c8f2954a4540ef605 : TIMED_OUT in 2h 33m 07s (non-voting)\n- octavia-v2-dsvm-scenario-centos-9-stream https://zuul.opendev.org/t/openstack/build/18b302d25bba46ee853552c79ccf2264 : TIMED_OUT in 2h 34m 33s (non-voting)","accounts_in_message":[],"_revision_number":20},{"id":"7002b96d2f1ec19fef822bba470e377dbe71db15","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":6577,"name":"Sergey Kraynev","email":"sergejyit@gmail.com","username":"skraynev"},"date":"2024-08-23 05:23:06.000000000","message":"Patch Set 21: Patch Set 20 was rebased\n\nCopied Votes:\n* Code-Review+2 (copy condition: \"**changekind:TRIVIAL_REBASE** OR is:MIN\")\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":21},{"id":"bc8dc8ecd22befb425e60b092d33cbaeba780757","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-23 08:10:02.000000000","message":"Patch Set 21: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c47afd73977045b2a5b6fe3a956ae176\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/7d2f4edcbfcc40dab1378da309f7c61c : SUCCESS in 12m 54s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3e7a81b1a3644f9db0cfb0db0b6d1e65 : SUCCESS in 8m 58s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/4a47442f60174b66a97e80aec413ca57 : SUCCESS in 26m 20s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/ad8f1de593414459817709602acc89d2 : SUCCESS in 13m 46s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/b192fd2154e34e5da9c5b530fad49cd1 : SUCCESS in 4m 58s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/470c69fbae354421b86f789fe94bbace : SUCCESS in 13m 49s\n- octavia-tox-py311-tips https://zuul.opendev.org/t/openstack/build/8c78112bb5da4322acf18fbc9b2472a7 : SUCCESS in 4m 42s\n- octavia-tox-functional-py311-tips https://zuul.opendev.org/t/openstack/build/92d7f15caf544baba52752ac69d24a65 : SUCCESS in 7m 56s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/0024514b1b49411cac54b2e6015e80a1 : SUCCESS in 8m 37s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/6a5ef9a2525c4216ab938b36ebeff59d : SUCCESS in 8m 17s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/010f8ea985fd47d6a90b8a8134411211 : SUCCESS in 7m 08s\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/cf8e0786dc234b4db685a25967dfd19f : SUCCESS in 6m 49s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/4285cceb84f748b09d20b541acbfe505 : SUCCESS in 1h 05m 06s\n- octavia-v2-dsvm-scenario-traffic-ops https://zuul.opendev.org/t/openstack/build/2ab111692257492d9b74b66ee6a70b2b : SUCCESS in 1h 49m 11s\n- octavia-v2-dsvm-scenario-non-traffic-ops https://zuul.opendev.org/t/openstack/build/d7ff7ee8528c48d4a975ff627975785e : SUCCESS in 1h 07m 19s\n- octavia-v2-dsvm-scenario-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/e2fe56e6a7854d8d8a0e5a60d56109d6 : SUCCESS in 1h 55m 12s\n- octavia-v2-dsvm-scenario-non-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/b1ccea3d4dbb4c99b1ba6fb516a1b63d : SUCCESS in 1h 04m 17s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/52481754a0484029a688f671b305285d : SUCCESS in 37m 47s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/c4791e6321e847b5a91c1be65a7e3162 : SUCCESS in 42m 40s\n- octavia-v2-dsvm-tls-barbican-fips https://zuul.opendev.org/t/openstack/build/e3945f40b1ea4e7aaf212558156a1cd9 : SUCCESS in 54m 51s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/6c010ca3829840d1ab8d23a058c875d6 : SUCCESS in 36m 27s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/d4e7b15abde74ae8b9593f07e1797d4c : TIMED_OUT in 2h 32m 50s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/63d1a91d30bb4087a563b2d88aaf7ba0 : SUCCESS in 2h 45m 34s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/c10f009c4dd54c2eb9f2eca76b3d3ccd : SUCCESS in 2h 29m 20s (non-voting)\n- octavia-v2-dsvm-scenario-centos-9-stream https://zuul.opendev.org/t/openstack/build/c5b5f736c3684d0e8d2c2beca2567015 : TIMED_OUT in 2h 33m 29s (non-voting)","accounts_in_message":[],"_revision_number":21},{"id":"5066154addbd2d19cd97b7fbbf701a37aabdd4ba","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2024-08-27 02:56:49.000000000","message":"Patch Set 21: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":21},{"id":"50df9cd2c5a4895bda0143dfd14f163c7e300c1c","author":{"_account_id":6577,"name":"Sergey Kraynev","email":"sergejyit@gmail.com","username":"skraynev"},"date":"2024-08-27 08:37:30.000000000","message":"Patch Set 21:\n\n(1 comment)","accounts_in_message":[],"_revision_number":21},{"id":"85c3f87bc013c1c7e85c54b0862a6d5eb1e0e7a0","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2024-10-23 17:36:41.000000000","message":"Patch Set 21: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":21},{"id":"2978dd7d0dd1e52f93202b3b3e9f04cca164bd03","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2024-10-23 17:38:27.000000000","message":"Patch Set 21: Workflow-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":21},{"id":"fa20f9ca13f2ae10d6247d79b134c2b063088cff","author":{"_account_id":6577,"name":"Sergey Kraynev","email":"sergejyit@gmail.com","username":"skraynev"},"date":"2024-10-25 04:37:03.000000000","message":"Patch Set 21:\n\n(1 comment)","accounts_in_message":[],"_revision_number":21},{"id":"446632b19dcf4b12325d14751dcc7ed21dcfb0af","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2024-10-25 06:35:01.000000000","message":"Patch Set 21:\n\n(1 comment)","accounts_in_message":[],"_revision_number":21},{"id":"55caee182630702f1752ee63cab71a212aa63062","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2024-11-08 13:58:00.000000000","message":"Patch Set 21: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":21},{"id":"04f5ddaa5f330bf4cfaec8bd33adc6a3fe5cc80f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-08 13:58:36.000000000","message":"Patch Set 21: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":21},{"id":"c852f2953a1c7b2390c7f2b52ff578dc957c4513","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-08 16:02:53.000000000","message":"Patch Set 21: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/074f4baf29714f319070ae760f51824b\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/458024c2f2db49968b6488c993539431 : SUCCESS in 4m 26s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/4a9245b7523d4cb9874999a41b1adc70 : SUCCESS in 4m 37s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/2a40137c05604806957f8701554461aa : SUCCESS in 5m 28s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4531fd9ac0894ab9b140f1dac51f9df6 : SUCCESS in 11m 35s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/3f5c4d3fef7642198060b2f743807e0f : SUCCESS in 3m 11s\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/e0a75f8af6a84c33a806b0fa18c021ae : SUCCESS in 6m 14s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/ff6c70b9386141b7acf577f5fb8d65e2 : SUCCESS in 1h 28m 30s\n- octavia-v2-dsvm-scenario-traffic-ops https://zuul.opendev.org/t/openstack/build/01eb75502a784fb899667591f325f6c1 : SUCCESS in 2h 02m 53s\n- octavia-v2-dsvm-scenario-non-traffic-ops https://zuul.opendev.org/t/openstack/build/e89428f8134d41b6a315d599ce86de98 : SUCCESS in 1h 12m 57s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/050a2a0ebbf44c4a812b44346defc515 : SUCCESS in 31m 14s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/dff47ce95b694f43a9694c43085573c0 : SUCCESS in 52m 20s","accounts_in_message":[],"_revision_number":21},{"id":"6b95439b74ccbeb68b5444e0ef243d078a95896e","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-08 16:02:53.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":21},{"id":"898b07ea98d372b3522b983011be554c7a935b37","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-08 16:04:04.000000000","message":"Patch Set 21:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/bb07cc3a8b134b95a8d733cd4c1d8b32\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f20f6d3b10944f218d39b4ec62e503d4 : SUCCESS in 51s","accounts_in_message":[],"_revision_number":21}],"current_revision_number":21,"current_revision":"cb5dca9bd9c04275141c038661efa716ba1dd3a5","revisions":{"f25027b05857889e1883665285f56cd0883b9a49":{"kind":"REWORK","_number":1,"created":"2023-05-12 12:29:26.000000000","uploader":{"_account_id":6577,"name":"Sergey Kraynev","email":"sergejyit@gmail.com","username":"skraynev"},"ref":"refs/changes/63/883063/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/63/883063/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/63/883063/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/63/883063/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/63/883063/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/63/883063/1"}}},"commit":{"parents":[{"commit":"bb7c8ca2c9b0db7e65fdaecce05c694929299054","subject":"Merge \"Remove python-neutronclient\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/bb7c8ca2c9b0db7e65fdaecce05c694929299054"}]}],"author":{"name":"Sergey Kraynev","email":"krainev.s@selectel.ru","date":"2023-05-12 12:15:58.000000000","tz":240},"committer":{"name":"Sergey Kraynev","email":"sergejyit@gmail.com","date":"2023-05-12 12:26:49.000000000","tz":240},"subject":"Add option to limit recursion in to_data_model","message":"Add option to limit recursion in to_data_model\n\nPatch adds option to limit recursion depth for to_data_model.\nThis hack is useful for situations, when only one layer of attributes of\nModel are necessary.\nAs example it could be used on GET request:\n/v2/lbaas/pools/\u003cpool_id\u003e/members/\n\nWith this changes to_data_model will not iterate through all nodes in\ngraph. It will just take main node (Member) and make one recursive call\nfor its attrbutes (here for Pool). After that iteration will stop\ninstead of recursive call of all attributes of Pool model.\n\nTODO (skraynev): add unittests after agreement, that this approach is correct\n\nCloses-Bug: 2019311\n\nChange-Id: I35797e252cb48eb23d82e1f89c88c6f1ae5fb02a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/f25027b05857889e1883665285f56cd0883b9a49"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/f25027b05857889e1883665285f56cd0883b9a49"}]},"branch":"refs/heads/master"},"7c5d6aa1d997b7e2e2f703acbc0fb1fc1d12f515":{"kind":"REWORK","_number":2,"created":"2023-05-12 15:08:43.000000000","uploader":{"_account_id":6577,"name":"Sergey Kraynev","email":"sergejyit@gmail.com","username":"skraynev"},"ref":"refs/changes/63/883063/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/63/883063/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/63/883063/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/63/883063/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/63/883063/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/63/883063/2"}}},"commit":{"parents":[{"commit":"bb7c8ca2c9b0db7e65fdaecce05c694929299054","subject":"Merge \"Remove python-neutronclient\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/bb7c8ca2c9b0db7e65fdaecce05c694929299054"}]}],"author":{"name":"Sergey Kraynev","email":"krainev.s@selectel.ru","date":"2023-05-12 12:15:58.000000000","tz":240},"committer":{"name":"Sergey Kraynev","email":"sergejyit@gmail.com","date":"2023-05-12 15:08:33.000000000","tz":240},"subject":"Add option to limit recursion in to_data_model","message":"Add option to limit recursion in to_data_model\n\nPatch adds option to limit recursion depth for to_data_model.\nThis hack is useful for situations, when only one layer of attributes of\nModel are necessary.\nAs example it could be used on GET request:\n/v2/lbaas/pools/\u003cpool_id\u003e/members/\n\nWith this changes to_data_model will not iterate through all nodes in\ngraph. It will just take main node (Member) and make one recursive call\nfor its attrbutes (here for Pool). After that iteration will stop\ninstead of recursive call of all attributes of Pool model.\n\nTODO (skraynev): add unittests after agreement, that this approach is correct\n\nCloses-Bug: 2019311\n\nChange-Id: I35797e252cb48eb23d82e1f89c88c6f1ae5fb02a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/7c5d6aa1d997b7e2e2f703acbc0fb1fc1d12f515"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/7c5d6aa1d997b7e2e2f703acbc0fb1fc1d12f515"}]},"branch":"refs/heads/master"},"b68c91dbe423f675b3e8aa6f2b143c1b45c97b09":{"kind":"REWORK","_number":3,"created":"2023-08-30 08:15:39.000000000","uploader":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"ref":"refs/changes/63/883063/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/63/883063/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/63/883063/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/63/883063/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/63/883063/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/63/883063/3"}}},"commit":{"parents":[{"commit":"30d9f709b5823d7a140f1fc30285b3a3ea9d6cab","subject":"Merge \"Remove unused wait_for_port_detach code\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/30d9f709b5823d7a140f1fc30285b3a3ea9d6cab"}]}],"author":{"name":"Sergey Kraynev","email":"krainev.s@selectel.ru","date":"2023-05-12 12:15:58.000000000","tz":240},"committer":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2023-08-30 08:11:25.000000000","tz":120},"subject":"Add option to limit recursion in to_data_model","message":"Add option to limit recursion in to_data_model\n\nPatch adds option to limit recursion depth for to_data_model.\nThis hack is useful for situations, when only one layer of attributes of\nModel are necessary.\nAs example it could be used on GET request:\n/v2/lbaas/pools/\u003cpool_id\u003e/members/\n\nWith this changes to_data_model will not iterate through all nodes in\ngraph. It will just take main node (Member) and make one recursive call\nfor its attrbutes (here for Pool). After that iteration will stop\ninstead of recursive call of all attributes of Pool model.\n\nTODO (skraynev): add unittests after agreement, that this approach is correct\n\nCloses-Bug: 2019311\n\nChange-Id: I35797e252cb48eb23d82e1f89c88c6f1ae5fb02a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/b68c91dbe423f675b3e8aa6f2b143c1b45c97b09"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/b68c91dbe423f675b3e8aa6f2b143c1b45c97b09"}]},"branch":"refs/heads/master"},"d6415f627d4098915f02c0c98cfd47f0cd274cf4":{"kind":"REWORK","_number":4,"created":"2023-09-03 11:15:47.000000000","uploader":{"_account_id":6577,"name":"Sergey Kraynev","email":"sergejyit@gmail.com","username":"skraynev"},"ref":"refs/changes/63/883063/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/63/883063/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/63/883063/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/63/883063/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/63/883063/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/63/883063/4"}}},"commit":{"parents":[{"commit":"30d9f709b5823d7a140f1fc30285b3a3ea9d6cab","subject":"Merge \"Remove unused wait_for_port_detach code\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/30d9f709b5823d7a140f1fc30285b3a3ea9d6cab"}]}],"author":{"name":"Sergey Kraynev","email":"sergejyit@gmail.com","date":"2023-09-03 11:14:55.000000000","tz":240},"committer":{"name":"Sergey Kraynev","email":"sergejyit@gmail.com","date":"2023-09-03 11:14:55.000000000","tz":240},"subject":"Add option to limit recursion in to_data_model","message":"Add option to limit recursion in to_data_model\n\nPatch adds option to limit recursion depth for to_data_model.\nThis hack is useful for situations, when only one layer of attributes of\nModel are necessary.\nAs example it could be used on GET request:\n/v2/lbaas/pools/\u003cpool_id\u003e/members/\n\nWith this changes to_data_model will not iterate through all nodes in\ngraph. It will just take main node (Member) and make one recursive call\nfor its attrbutes (here for Pool). After that iteration will stop\ninstead of recursive call of all attributes of Pool model.\n\nCloses-Bug: 2019311\n\nChange-Id: I35797e252cb48eb23d82e1f89c88c6f1ae5fb02a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/d6415f627d4098915f02c0c98cfd47f0cd274cf4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/d6415f627d4098915f02c0c98cfd47f0cd274cf4"}]},"branch":"refs/heads/master"},"77d806b2aed2413ef6b73a8396055abbf3e4d6d4":{"kind":"TRIVIAL_REBASE","_number":5,"created":"2023-09-03 11:16:40.000000000","uploader":{"_account_id":6577,"name":"Sergey Kraynev","email":"sergejyit@gmail.com","username":"skraynev"},"ref":"refs/changes/63/883063/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/63/883063/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/63/883063/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/63/883063/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/63/883063/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/63/883063/5"}}},"commit":{"parents":[{"commit":"0426285a41464a5015494584f109eed35a0d44db","subject":"Merge \"Add octavia-grenade-slurp CI job\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/0426285a41464a5015494584f109eed35a0d44db"}]}],"author":{"name":"Sergey Kraynev","email":"sergejyit@gmail.com","date":"2023-09-03 11:14:55.000000000","tz":240},"committer":{"name":"Sergey Kraynev","email":"sergejyit@gmail.com","date":"2023-09-03 11:16:31.000000000","tz":240},"subject":"Add option to limit recursion in to_data_model","message":"Add option to limit recursion in to_data_model\n\nPatch adds option to limit recursion depth for to_data_model.\nThis hack is useful for situations, when only one layer of attributes of\nModel are necessary.\nAs example it could be used on GET request:\n/v2/lbaas/pools/\u003cpool_id\u003e/members/\n\nWith this changes to_data_model will not iterate through all nodes in\ngraph. It will just take main node (Member) and make one recursive call\nfor its attrbutes (here for Pool). After that iteration will stop\ninstead of recursive call of all attributes of Pool model.\n\nCloses-Bug: 2019311\n\nChange-Id: I35797e252cb48eb23d82e1f89c88c6f1ae5fb02a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/77d806b2aed2413ef6b73a8396055abbf3e4d6d4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/77d806b2aed2413ef6b73a8396055abbf3e4d6d4"}]},"branch":"refs/heads/master"},"6cb45449a35342181d8be7a4af367d528dc5c93e":{"kind":"REWORK","_number":6,"created":"2023-09-03 15:22:57.000000000","uploader":{"_account_id":6577,"name":"Sergey Kraynev","email":"sergejyit@gmail.com","username":"skraynev"},"ref":"refs/changes/63/883063/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/63/883063/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/63/883063/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/63/883063/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/63/883063/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/63/883063/6"}}},"commit":{"parents":[{"commit":"0426285a41464a5015494584f109eed35a0d44db","subject":"Merge \"Add octavia-grenade-slurp CI job\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/0426285a41464a5015494584f109eed35a0d44db"}]}],"author":{"name":"Sergey Kraynev","email":"sergejyit@gmail.com","date":"2023-09-03 11:14:55.000000000","tz":240},"committer":{"name":"Sergey Kraynev","email":"sergejyit@gmail.com","date":"2023-09-03 15:22:43.000000000","tz":240},"subject":"Add option to limit recursion in to_data_model","message":"Add option to limit recursion in to_data_model\n\nPatch adds option to limit recursion depth for to_data_model.\nThis hack is useful for situations, when only one layer of attributes of\nModel are necessary.\nAs example it could be used on GET request:\n/v2/lbaas/pools/\u003cpool_id\u003e/members/\n\nWith this changes to_data_model will not iterate through all nodes in\ngraph. It will just take main node (Member) and make one recursive call\nfor its attrbutes (here for Pool). After that iteration will stop\ninstead of recursive call of all attributes of Pool model.\n\nCloses-Bug: 2019311\n\nChange-Id: I35797e252cb48eb23d82e1f89c88c6f1ae5fb02a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/6cb45449a35342181d8be7a4af367d528dc5c93e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/6cb45449a35342181d8be7a4af367d528dc5c93e"}]},"branch":"refs/heads/master"},"837df636d020c78732f5b92a3f79cc296c5e7fd9":{"kind":"TRIVIAL_REBASE","_number":7,"created":"2023-09-18 14:31:59.000000000","uploader":{"_account_id":6577,"name":"Sergey Kraynev","email":"sergejyit@gmail.com","username":"skraynev"},"ref":"refs/changes/63/883063/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/63/883063/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/63/883063/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/63/883063/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/63/883063/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/63/883063/7"}}},"commit":{"parents":[{"commit":"d7bba6c5af3c0f6d99ea8063b54ec3e895a5eea4","subject":"Move octavia grenade job on jammy","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/d7bba6c5af3c0f6d99ea8063b54ec3e895a5eea4"}]}],"author":{"name":"Sergey Kraynev","email":"sergejyit@gmail.com","date":"2023-09-03 11:14:55.000000000","tz":240},"committer":{"name":"Sergey Kraynev","email":"sergejyit@gmail.com","date":"2023-09-18 14:31:59.000000000","tz":0},"subject":"Add option to limit recursion in to_data_model","message":"Add option to limit recursion in to_data_model\n\nPatch adds option to limit recursion depth for to_data_model.\nThis hack is useful for situations, when only one layer of attributes of\nModel are necessary.\nAs example it could be used on GET request:\n/v2/lbaas/pools/\u003cpool_id\u003e/members/\n\nWith this changes to_data_model will not iterate through all nodes in\ngraph. It will just take main node (Member) and make one recursive call\nfor its attrbutes (here for Pool). After that iteration will stop\ninstead of recursive call of all attributes of Pool model.\n\nCloses-Bug: 2019311\n\nChange-Id: I35797e252cb48eb23d82e1f89c88c6f1ae5fb02a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/837df636d020c78732f5b92a3f79cc296c5e7fd9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/837df636d020c78732f5b92a3f79cc296c5e7fd9"}]},"branch":"refs/heads/master","description":"Rebase"},"d9c6c84b5c66414ff7a40483133139b5c1bcfc6f":{"kind":"TRIVIAL_REBASE","_number":8,"created":"2023-09-25 09:34:47.000000000","uploader":{"_account_id":6577,"name":"Sergey Kraynev","email":"sergejyit@gmail.com","username":"skraynev"},"ref":"refs/changes/63/883063/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/63/883063/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/63/883063/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/63/883063/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/63/883063/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/63/883063/8"}}},"commit":{"parents":[{"commit":"33fed53043091e5016a9114dfd803092a09a718b","subject":"Merge \"Add support for Rocky Linux\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/33fed53043091e5016a9114dfd803092a09a718b"}]}],"author":{"name":"Sergey Kraynev","email":"sergejyit@gmail.com","date":"2023-09-03 11:14:55.000000000","tz":240},"committer":{"name":"Sergey Kraynev","email":"sergejyit@gmail.com","date":"2023-09-25 09:34:47.000000000","tz":0},"subject":"Add option to limit recursion in to_data_model","message":"Add option to limit recursion in to_data_model\n\nPatch adds option to limit recursion depth for to_data_model.\nThis hack is useful for situations, when only one layer of attributes of\nModel are necessary.\nAs example it could be used on GET request:\n/v2/lbaas/pools/\u003cpool_id\u003e/members/\n\nWith this changes to_data_model will not iterate through all nodes in\ngraph. It will just take main node (Member) and make one recursive call\nfor its attrbutes (here for Pool). After that iteration will stop\ninstead of recursive call of all attributes of Pool model.\n\nCloses-Bug: 2019311\n\nChange-Id: I35797e252cb48eb23d82e1f89c88c6f1ae5fb02a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/d9c6c84b5c66414ff7a40483133139b5c1bcfc6f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/d9c6c84b5c66414ff7a40483133139b5c1bcfc6f"}]},"branch":"refs/heads/master","description":"Rebase"},"8a5ad0ae05be0fe0f55655b44c2ebd3ac8ab84a0":{"kind":"REWORK","_number":9,"created":"2023-09-26 18:31:22.000000000","uploader":{"_account_id":6577,"name":"Sergey Kraynev","email":"sergejyit@gmail.com","username":"skraynev"},"ref":"refs/changes/63/883063/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/63/883063/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/63/883063/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/63/883063/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/63/883063/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/63/883063/9"}}},"commit":{"parents":[{"commit":"025bfdd758ea6521a28dfaf40d916ef6f7421752","subject":"Merge \"Fix race condition in members batch update API call\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/025bfdd758ea6521a28dfaf40d916ef6f7421752"}]}],"author":{"name":"Sergey Kraynev","email":"sergejyit@gmail.com","date":"2023-09-03 11:14:55.000000000","tz":240},"committer":{"name":"Sergey Kraynev","email":"sergejyit@gmail.com","date":"2023-09-26 18:31:13.000000000","tz":240},"subject":"Add option to limit recursion in to_data_model","message":"Add option to limit recursion in to_data_model\n\nPatch adds option to limit recursion depth for to_data_model.\nThis hack is useful for situations, when only one layer of attributes of\nModel are necessary.\nAs example it could be used on GET request:\n/v2/lbaas/pools/\u003cpool_id\u003e/members/\n\nWith this changes to_data_model will not iterate through all nodes in\ngraph. It will just take main node (Member) and make one recursive call\nfor its attrbutes (here for Pool). After that iteration will stop\ninstead of recursive call of all attributes of Pool model.\n\nCloses-Bug: 2019311\n\nChange-Id: I35797e252cb48eb23d82e1f89c88c6f1ae5fb02a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/8a5ad0ae05be0fe0f55655b44c2ebd3ac8ab84a0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/8a5ad0ae05be0fe0f55655b44c2ebd3ac8ab84a0"}]},"branch":"refs/heads/master"},"3bf5e85557132a04ccf55dccbb63b4221354626c":{"kind":"REWORK","_number":10,"created":"2023-09-30 11:19:01.000000000","uploader":{"_account_id":6577,"name":"Sergey Kraynev","email":"sergejyit@gmail.com","username":"skraynev"},"ref":"refs/changes/63/883063/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/63/883063/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/63/883063/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/63/883063/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/63/883063/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/63/883063/10"}}},"commit":{"parents":[{"commit":"025bfdd758ea6521a28dfaf40d916ef6f7421752","subject":"Merge \"Fix race condition in members batch update API call\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/025bfdd758ea6521a28dfaf40d916ef6f7421752"}]}],"author":{"name":"Sergey Kraynev","email":"sergejyit@gmail.com","date":"2023-09-03 11:14:55.000000000","tz":240},"committer":{"name":"Sergey Kraynev","email":"sergejyit@gmail.com","date":"2023-09-30 11:17:42.000000000","tz":240},"subject":"Add option to limit recursion in to_data_model","message":"Add option to limit recursion in to_data_model\n\nPatch adds option to limit recursion depth for to_data_model.\nThis hack is useful for situations, when only one layer of attributes of\nModel are necessary.\nAs example it could be used on GET request:\n/v2/lbaas/pools/\u003cpool_id\u003e/members/\n\nWith this changes to_data_model will not iterate through all nodes in\ngraph. It will just take main node (Member) and make one recursive call\nfor its attrbutes (here for Pool). After that iteration will stop\ninstead of recursive call of all attributes of Pool model.\n\nAdds the same logic to repo.get method but with different default value\n(0). So on get request only node itself be processed without related\nnodes.\n\nCloses-Bug: 2019311\n\nChange-Id: I35797e252cb48eb23d82e1f89c88c6f1ae5fb02a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/3bf5e85557132a04ccf55dccbb63b4221354626c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/3bf5e85557132a04ccf55dccbb63b4221354626c"}]},"branch":"refs/heads/master"},"aeddd319067d2c64c1a0b943f777aed1231622da":{"kind":"TRIVIAL_REBASE","_number":11,"created":"2023-10-09 04:57:05.000000000","uploader":{"_account_id":6577,"name":"Sergey Kraynev","email":"sergejyit@gmail.com","username":"skraynev"},"ref":"refs/changes/63/883063/11","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/63/883063/11","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/63/883063/11 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/63/883063/11 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/63/883063/11 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/63/883063/11"}}},"commit":{"parents":[{"commit":"8b275c9703fa610e6c9bb47645c4befbc848d6d4","subject":"Merge \"Fix Zuul warnings with regexps\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/8b275c9703fa610e6c9bb47645c4befbc848d6d4"}]}],"author":{"name":"Sergey Kraynev","email":"sergejyit@gmail.com","date":"2023-09-03 11:14:55.000000000","tz":240},"committer":{"name":"Sergey Kraynev","email":"sergejyit@gmail.com","date":"2023-10-09 04:57:05.000000000","tz":0},"subject":"Add option to limit recursion in to_data_model","message":"Add option to limit recursion in to_data_model\n\nPatch adds option to limit recursion depth for to_data_model.\nThis hack is useful for situations, when only one layer of attributes of\nModel are necessary.\nAs example it could be used on GET request:\n/v2/lbaas/pools/\u003cpool_id\u003e/members/\n\nWith this changes to_data_model will not iterate through all nodes in\ngraph. It will just take main node (Member) and make one recursive call\nfor its attrbutes (here for Pool). After that iteration will stop\ninstead of recursive call of all attributes of Pool model.\n\nAdds the same logic to repo.get method but with different default value\n(0). So on get request only node itself be processed without related\nnodes.\n\nCloses-Bug: 2019311\n\nChange-Id: I35797e252cb48eb23d82e1f89c88c6f1ae5fb02a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/aeddd319067d2c64c1a0b943f777aed1231622da"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/aeddd319067d2c64c1a0b943f777aed1231622da"}]},"branch":"refs/heads/master","description":"Rebase"},"e176c02b85e157deee9d19c0328a645c1e8f972f":{"kind":"TRIVIAL_REBASE","_number":12,"created":"2024-02-13 19:17:20.000000000","uploader":{"_account_id":6577,"name":"Sergey Kraynev","email":"sergejyit@gmail.com","username":"skraynev"},"ref":"refs/changes/63/883063/12","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/63/883063/12","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/63/883063/12 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/63/883063/12 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/63/883063/12 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/63/883063/12"}}},"commit":{"parents":[{"commit":"bbb4713464eee61add6ce32427621e36fa4086e9","subject":"Merge \"Declare we use threads and processes for coverage\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/bbb4713464eee61add6ce32427621e36fa4086e9"}]}],"author":{"name":"Sergey Kraynev","email":"sergejyit@gmail.com","date":"2023-09-03 11:14:55.000000000","tz":240},"committer":{"name":"Sergey Kraynev","email":"sergejyit@gmail.com","date":"2024-02-13 19:17:08.000000000","tz":240},"subject":"Add option to limit recursion in to_data_model","message":"Add option to limit recursion in to_data_model\n\nPatch adds option to limit recursion depth for to_data_model.\nThis hack is useful for situations, when only one layer of attributes of\nModel are necessary.\nAs example it could be used on GET request:\n/v2/lbaas/pools/\u003cpool_id\u003e/members/\n\nWith this changes to_data_model will not iterate through all nodes in\ngraph. It will just take main node (Member) and make one recursive call\nfor its attrbutes (here for Pool). After that iteration will stop\ninstead of recursive call of all attributes of Pool model.\n\nAdds the same logic to repo.get method but with different default value\n(0). So on get request only node itself be processed without related\nnodes.\n\nCloses-Bug: 2019311\n\nChange-Id: I35797e252cb48eb23d82e1f89c88c6f1ae5fb02a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/e176c02b85e157deee9d19c0328a645c1e8f972f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/e176c02b85e157deee9d19c0328a645c1e8f972f"}]},"branch":"refs/heads/master"},"85027e037402c77cd0a747e8c84c007ba1d6574f":{"kind":"TRIVIAL_REBASE","_number":13,"created":"2024-02-27 08:41:24.000000000","uploader":{"_account_id":6577,"name":"Sergey Kraynev","email":"sergejyit@gmail.com","username":"skraynev"},"ref":"refs/changes/63/883063/13","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/63/883063/13","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/63/883063/13 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/63/883063/13 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/63/883063/13 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/63/883063/13"}}},"commit":{"parents":[{"commit":"30ce858564904a6217c3a470985a72011adbcb9c","subject":"Merge \"Fix pylint error\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/30ce858564904a6217c3a470985a72011adbcb9c"}]}],"author":{"name":"Sergey Kraynev","email":"sergejyit@gmail.com","date":"2023-09-03 11:14:55.000000000","tz":240},"committer":{"name":"Sergey Kraynev","email":"sergejyit@gmail.com","date":"2024-02-27 08:41:24.000000000","tz":0},"subject":"Add option to limit recursion in to_data_model","message":"Add option to limit recursion in to_data_model\n\nPatch adds option to limit recursion depth for to_data_model.\nThis hack is useful for situations, when only one layer of attributes of\nModel are necessary.\nAs example it could be used on GET request:\n/v2/lbaas/pools/\u003cpool_id\u003e/members/\n\nWith this changes to_data_model will not iterate through all nodes in\ngraph. It will just take main node (Member) and make one recursive call\nfor its attrbutes (here for Pool). After that iteration will stop\ninstead of recursive call of all attributes of Pool model.\n\nAdds the same logic to repo.get method but with different default value\n(0). So on get request only node itself be processed without related\nnodes.\n\nCloses-Bug: 2019311\n\nChange-Id: I35797e252cb48eb23d82e1f89c88c6f1ae5fb02a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/85027e037402c77cd0a747e8c84c007ba1d6574f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/85027e037402c77cd0a747e8c84c007ba1d6574f"}]},"branch":"refs/heads/master","description":"Rebase"},"8f613336068c5b9a7ca8638d33beb093d22d3eec":{"kind":"TRIVIAL_REBASE","_number":14,"created":"2024-03-14 17:28:30.000000000","uploader":{"_account_id":6577,"name":"Sergey Kraynev","email":"sergejyit@gmail.com","username":"skraynev"},"ref":"refs/changes/63/883063/14","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/63/883063/14","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/63/883063/14 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/63/883063/14 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/63/883063/14 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/63/883063/14"}}},"commit":{"parents":[{"commit":"9f1a6e47d7eb77492bfcc68adb1392f163ce6d26","subject":"Merge \"dib: Remove Ubuntu Forcal support\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/9f1a6e47d7eb77492bfcc68adb1392f163ce6d26"}]}],"author":{"name":"Sergey Kraynev","email":"sergejyit@gmail.com","date":"2023-09-03 11:14:55.000000000","tz":240},"committer":{"name":"Sergey Kraynev","email":"sergejyit@gmail.com","date":"2024-03-14 17:28:30.000000000","tz":0},"subject":"Add option to limit recursion in to_data_model","message":"Add option to limit recursion in to_data_model\n\nPatch adds option to limit recursion depth for to_data_model.\nThis hack is useful for situations, when only one layer of attributes of\nModel are necessary.\nAs example it could be used on GET request:\n/v2/lbaas/pools/\u003cpool_id\u003e/members/\n\nWith this changes to_data_model will not iterate through all nodes in\ngraph. It will just take main node (Member) and make one recursive call\nfor its attrbutes (here for Pool). After that iteration will stop\ninstead of recursive call of all attributes of Pool model.\n\nAdds the same logic to repo.get method but with different default value\n(0). So on get request only node itself be processed without related\nnodes.\n\nCloses-Bug: 2019311\n\nChange-Id: I35797e252cb48eb23d82e1f89c88c6f1ae5fb02a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/8f613336068c5b9a7ca8638d33beb093d22d3eec"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/8f613336068c5b9a7ca8638d33beb093d22d3eec"}]},"branch":"refs/heads/master","description":"Rebase"},"f1e41afd2882ea0007c10bc217c0055cf136b543":{"kind":"TRIVIAL_REBASE","_number":15,"created":"2024-04-04 05:57:01.000000000","uploader":{"_account_id":6577,"name":"Sergey Kraynev","email":"sergejyit@gmail.com","username":"skraynev"},"ref":"refs/changes/63/883063/15","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/63/883063/15","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/63/883063/15 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/63/883063/15 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/63/883063/15 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/63/883063/15"}}},"commit":{"parents":[{"commit":"fa7cc0178783ad4ab9f3ddce893042d5952d0bd8","subject":"Merge \"Fix fully-populated API with allowed_cidrs\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/fa7cc0178783ad4ab9f3ddce893042d5952d0bd8"}]}],"author":{"name":"Sergey Kraynev","email":"sergejyit@gmail.com","date":"2023-09-03 11:14:55.000000000","tz":240},"committer":{"name":"Sergey Kraynev","email":"sergejyit@gmail.com","date":"2024-04-04 05:57:01.000000000","tz":0},"subject":"Add option to limit recursion in to_data_model","message":"Add option to limit recursion in to_data_model\n\nPatch adds option to limit recursion depth for to_data_model.\nThis hack is useful for situations, when only one layer of attributes of\nModel are necessary.\nAs example it could be used on GET request:\n/v2/lbaas/pools/\u003cpool_id\u003e/members/\n\nWith this changes to_data_model will not iterate through all nodes in\ngraph. It will just take main node (Member) and make one recursive call\nfor its attrbutes (here for Pool). After that iteration will stop\ninstead of recursive call of all attributes of Pool model.\n\nAdds the same logic to repo.get method but with different default value\n(0). So on get request only node itself be processed without related\nnodes.\n\nCloses-Bug: 2019311\n\nChange-Id: I35797e252cb48eb23d82e1f89c88c6f1ae5fb02a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/f1e41afd2882ea0007c10bc217c0055cf136b543"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/f1e41afd2882ea0007c10bc217c0055cf136b543"}]},"branch":"refs/heads/master","description":"Rebase"},"cf9e4ce32e7a3d132bf15be91c81969e9bb24bdc":{"kind":"TRIVIAL_REBASE","_number":16,"created":"2024-04-16 03:23:03.000000000","uploader":{"_account_id":6577,"name":"Sergey Kraynev","email":"sergejyit@gmail.com","username":"skraynev"},"ref":"refs/changes/63/883063/16","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/63/883063/16","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/63/883063/16 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/63/883063/16 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/63/883063/16 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/63/883063/16"}}},"commit":{"parents":[{"commit":"603d437aefc5234af5790933185c135c50553080","subject":"Merge \"Add pyupgrade hook to pre-commit config\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/603d437aefc5234af5790933185c135c50553080"}]}],"author":{"name":"Sergey Kraynev","email":"sergejyit@gmail.com","date":"2023-09-03 11:14:55.000000000","tz":240},"committer":{"name":"Sergey Kraynev","email":"sergejyit@gmail.com","date":"2024-04-16 03:23:03.000000000","tz":0},"subject":"Add option to limit recursion in to_data_model","message":"Add option to limit recursion in to_data_model\n\nPatch adds option to limit recursion depth for to_data_model.\nThis hack is useful for situations, when only one layer of attributes of\nModel are necessary.\nAs example it could be used on GET request:\n/v2/lbaas/pools/\u003cpool_id\u003e/members/\n\nWith this changes to_data_model will not iterate through all nodes in\ngraph. It will just take main node (Member) and make one recursive call\nfor its attrbutes (here for Pool). After that iteration will stop\ninstead of recursive call of all attributes of Pool model.\n\nAdds the same logic to repo.get method but with different default value\n(0). So on get request only node itself be processed without related\nnodes.\n\nCloses-Bug: 2019311\n\nChange-Id: I35797e252cb48eb23d82e1f89c88c6f1ae5fb02a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/cf9e4ce32e7a3d132bf15be91c81969e9bb24bdc"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/cf9e4ce32e7a3d132bf15be91c81969e9bb24bdc"}]},"branch":"refs/heads/master","description":"Rebase"},"3833ef8d485dd93f83a3402548d97ced50d17688":{"kind":"TRIVIAL_REBASE","_number":17,"created":"2024-04-18 17:49:36.000000000","uploader":{"_account_id":6577,"name":"Sergey Kraynev","email":"sergejyit@gmail.com","username":"skraynev"},"ref":"refs/changes/63/883063/17","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/63/883063/17","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/63/883063/17 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/63/883063/17 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/63/883063/17 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/63/883063/17"}}},"commit":{"parents":[{"commit":"93dd9916c3d9407fd6e8580a166c6a56250389ad","subject":"Merge \"devstack: Drop removed services\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/93dd9916c3d9407fd6e8580a166c6a56250389ad"}]}],"author":{"name":"Sergey Kraynev","email":"sergejyit@gmail.com","date":"2023-09-03 11:14:55.000000000","tz":240},"committer":{"name":"Sergey Kraynev","email":"sergejyit@gmail.com","date":"2024-04-18 17:49:36.000000000","tz":0},"subject":"Add option to limit recursion in to_data_model","message":"Add option to limit recursion in to_data_model\n\nPatch adds option to limit recursion depth for to_data_model.\nThis hack is useful for situations, when only one layer of attributes of\nModel are necessary.\nAs example it could be used on GET request:\n/v2/lbaas/pools/\u003cpool_id\u003e/members/\n\nWith this changes to_data_model will not iterate through all nodes in\ngraph. It will just take main node (Member) and make one recursive call\nfor its attrbutes (here for Pool). After that iteration will stop\ninstead of recursive call of all attributes of Pool model.\n\nAdds the same logic to repo.get method but with different default value\n(0). So on get request only node itself be processed without related\nnodes.\n\nCloses-Bug: 2019311\n\nChange-Id: I35797e252cb48eb23d82e1f89c88c6f1ae5fb02a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/3833ef8d485dd93f83a3402548d97ced50d17688"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/3833ef8d485dd93f83a3402548d97ced50d17688"}]},"branch":"refs/heads/master","description":"Rebase"},"26843f5ec64a13ed98f40d3d444412a0aaa7ce0f":{"kind":"TRIVIAL_REBASE","_number":18,"created":"2024-05-02 05:55:28.000000000","uploader":{"_account_id":6577,"name":"Sergey Kraynev","email":"sergejyit@gmail.com","username":"skraynev"},"ref":"refs/changes/63/883063/18","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/63/883063/18","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/63/883063/18 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/63/883063/18 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/63/883063/18 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/63/883063/18"}}},"commit":{"parents":[{"commit":"5898d54b137f5fb7a1d45bf8d620ef8ab8133259","subject":"Merge \"Drop redundant envdir\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/5898d54b137f5fb7a1d45bf8d620ef8ab8133259"}]}],"author":{"name":"Sergey Kraynev","email":"sergejyit@gmail.com","date":"2023-09-03 11:14:55.000000000","tz":240},"committer":{"name":"Sergey Kraynev","email":"sergejyit@gmail.com","date":"2024-05-02 05:55:28.000000000","tz":0},"subject":"Add option to limit recursion in to_data_model","message":"Add option to limit recursion in to_data_model\n\nPatch adds option to limit recursion depth for to_data_model.\nThis hack is useful for situations, when only one layer of attributes of\nModel are necessary.\nAs example it could be used on GET request:\n/v2/lbaas/pools/\u003cpool_id\u003e/members/\n\nWith this changes to_data_model will not iterate through all nodes in\ngraph. It will just take main node (Member) and make one recursive call\nfor its attrbutes (here for Pool). After that iteration will stop\ninstead of recursive call of all attributes of Pool model.\n\nAdds the same logic to repo.get method but with different default value\n(0). So on get request only node itself be processed without related\nnodes.\n\nCloses-Bug: 2019311\n\nChange-Id: I35797e252cb48eb23d82e1f89c88c6f1ae5fb02a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/26843f5ec64a13ed98f40d3d444412a0aaa7ce0f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/26843f5ec64a13ed98f40d3d444412a0aaa7ce0f"}]},"branch":"refs/heads/master","description":"Rebase"},"722c5007f0fbad7d02685b6af7b23c97c807141d":{"kind":"TRIVIAL_REBASE","_number":19,"created":"2024-05-20 07:24:59.000000000","uploader":{"_account_id":6577,"name":"Sergey Kraynev","email":"sergejyit@gmail.com","username":"skraynev"},"ref":"refs/changes/63/883063/19","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/63/883063/19","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/63/883063/19 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/63/883063/19 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/63/883063/19 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/63/883063/19"}}},"commit":{"parents":[{"commit":"b43620cfa8e509647f99835e6eb43416a7f214f8","subject":"Merge \"Fix listener update when using SRIOV VIP\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/b43620cfa8e509647f99835e6eb43416a7f214f8"}]}],"author":{"name":"Sergey Kraynev","email":"sergejyit@gmail.com","date":"2023-09-03 11:14:55.000000000","tz":240},"committer":{"name":"Sergey Kraynev","email":"sergejyit@gmail.com","date":"2024-05-20 07:24:59.000000000","tz":0},"subject":"Add option to limit recursion in to_data_model","message":"Add option to limit recursion in to_data_model\n\nPatch adds option to limit recursion depth for to_data_model.\nThis hack is useful for situations, when only one layer of attributes of\nModel are necessary.\nAs example it could be used on GET request:\n/v2/lbaas/pools/\u003cpool_id\u003e/members/\n\nWith this changes to_data_model will not iterate through all nodes in\ngraph. It will just take main node (Member) and make one recursive call\nfor its attrbutes (here for Pool). After that iteration will stop\ninstead of recursive call of all attributes of Pool model.\n\nAdds the same logic to repo.get method but with different default value\n(0). So on get request only node itself be processed without related\nnodes.\n\nCloses-Bug: 2019311\n\nChange-Id: I35797e252cb48eb23d82e1f89c88c6f1ae5fb02a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/722c5007f0fbad7d02685b6af7b23c97c807141d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/722c5007f0fbad7d02685b6af7b23c97c807141d"}]},"branch":"refs/heads/master","description":"Rebase"},"a75cb944ec27d40f83d4e77d8c5f0a0aa6737506":{"kind":"TRIVIAL_REBASE","_number":20,"created":"2024-08-19 11:55:29.000000000","uploader":{"_account_id":6577,"name":"Sergey Kraynev","email":"sergejyit@gmail.com","username":"skraynev"},"ref":"refs/changes/63/883063/20","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/63/883063/20","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/63/883063/20 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/63/883063/20 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/63/883063/20 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/63/883063/20"}}},"commit":{"parents":[{"commit":"f68cafbb7201522e6140e2dc05317d06225bb82a","subject":"Merge \"Remove sysvinit/upstart support\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/f68cafbb7201522e6140e2dc05317d06225bb82a"}]}],"author":{"name":"Sergey Kraynev","email":"sergejyit@gmail.com","date":"2023-09-03 11:14:55.000000000","tz":240},"committer":{"name":"Sergey Kraynev","email":"sergejyit@gmail.com","date":"2024-08-19 11:55:29.000000000","tz":0},"subject":"Add option to limit recursion in to_data_model","message":"Add option to limit recursion in to_data_model\n\nPatch adds option to limit recursion depth for to_data_model.\nThis hack is useful for situations, when only one layer of attributes of\nModel are necessary.\nAs example it could be used on GET request:\n/v2/lbaas/pools/\u003cpool_id\u003e/members/\n\nWith this changes to_data_model will not iterate through all nodes in\ngraph. It will just take main node (Member) and make one recursive call\nfor its attrbutes (here for Pool). After that iteration will stop\ninstead of recursive call of all attributes of Pool model.\n\nAdds the same logic to repo.get method but with different default value\n(0). So on get request only node itself be processed without related\nnodes.\n\nCloses-Bug: 2019311\n\nChange-Id: I35797e252cb48eb23d82e1f89c88c6f1ae5fb02a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/a75cb944ec27d40f83d4e77d8c5f0a0aa6737506"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/a75cb944ec27d40f83d4e77d8c5f0a0aa6737506"}]},"branch":"refs/heads/master","description":"Rebase"},"cb5dca9bd9c04275141c038661efa716ba1dd3a5":{"kind":"TRIVIAL_REBASE","_number":21,"created":"2024-08-23 05:23:06.000000000","uploader":{"_account_id":6577,"name":"Sergey Kraynev","email":"sergejyit@gmail.com","username":"skraynev"},"ref":"refs/changes/63/883063/21","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/63/883063/21","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/63/883063/21 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/63/883063/21 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/63/883063/21 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/63/883063/21"}}},"commit":{"parents":[{"commit":"a4317a80d61ce2d8e699ea7ca17f161ecc4db8ee","subject":"Fix failover when using SRIOV VIP","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/a4317a80d61ce2d8e699ea7ca17f161ecc4db8ee"}]}],"author":{"name":"Sergey Kraynev","email":"sergejyit@gmail.com","date":"2023-09-03 11:14:55.000000000","tz":240},"committer":{"name":"Sergey Kraynev","email":"sergejyit@gmail.com","date":"2024-08-23 05:23:06.000000000","tz":0},"subject":"Add option to limit recursion in to_data_model","message":"Add option to limit recursion in to_data_model\n\nPatch adds option to limit recursion depth for to_data_model.\nThis hack is useful for situations, when only one layer of attributes of\nModel are necessary.\nAs example it could be used on GET request:\n/v2/lbaas/pools/\u003cpool_id\u003e/members/\n\nWith this changes to_data_model will not iterate through all nodes in\ngraph. It will just take main node (Member) and make one recursive call\nfor its attrbutes (here for Pool). After that iteration will stop\ninstead of recursive call of all attributes of Pool model.\n\nAdds the same logic to repo.get method but with different default value\n(0). So on get request only node itself be processed without related\nnodes.\n\nCloses-Bug: 2019311\n\nChange-Id: I35797e252cb48eb23d82e1f89c88c6f1ae5fb02a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/cb5dca9bd9c04275141c038661efa716ba1dd3a5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/cb5dca9bd9c04275141c038661efa716ba1dd3a5"}]},"branch":"refs/heads/master","description":"Rebase"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
