)]}'
{"id":"openstack%2Foctavia~883135","triplet_id":"openstack%2Foctavia~stable%2F2023.1~I51d2c011f123e7d8b90b5cea9ba875052d7837a3","project":"openstack/octavia","branch":"stable/2023.1","attention_set":{},"removed_from_attention_set":{"29244":{"account":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"last_update":"2023-07-03 15:57:15.000000000","reason":"\u003cGERRIT_ACCOUNT_29244\u003e replied on the change","reason_account":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"}}},"hashtags":[],"change_id":"I51d2c011f123e7d8b90b5cea9ba875052d7837a3","subject":"Fix octavia-amphora-image-build periodic job","status":"MERGED","created":"2023-05-15 07:41:32.000000000","updated":"2023-07-03 20:37:18.000000000","submitted":"2023-07-03 20:35:23.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":5,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"883135","meta_rev_id":"84f6a5a8cd51a3823328c34b5ffa103b7e9e1d2c","_number":883135,"virtual_id_number":883135,"owner":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-07-03 20:35:23.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"all":[{"value":2,"date":"2023-05-23 08:01:33.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2023-05-15 09:43:30.000000000","_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"all":[{"value":1,"date":"2023-05-23 08:01:33.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}],"CC":[{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-05-15 09:43:30.000000000","updated_by":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"reviewer":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"state":"REVIEWER"},{"updated":"2023-05-15 10:15:40.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-05-29 10:42:45.000000000","updated_by":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"reviewer":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"state":"CC"}],"messages":[{"id":"ddca23b7ec86443a82a4fb440a8333cedf83d6ab","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2023-05-15 07:41:32.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"fc2b8168b229214ed3caea5fbf7afb69f776a167","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2023-05-15 09:43:30.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"cc6116d1674c01b4c2becb5d521c3abd69a79f50","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-15 10:15:40.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/bd4597d1b2e145d98728b2469ab2fdc2\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/ed4352a5bc3d499a99f1b3a86754b344 : SUCCESS in 10m 18s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5ba6b3e7b08a4b16b0147220ff364b42 : SUCCESS in 4m 35s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/4aa4f481653449d1b34982b8d3804833 : SUCCESS in 5m 39s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/d3473b92da2647209b4ddde422ddb216 : SUCCESS in 4m 36s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3f2404de87e443e995eb76646f26bc6d : SUCCESS in 12m 17s\n- octavia-tox-py39-tips https://zuul.opendev.org/t/openstack/build/d0ceedd505a949b0b7b8b755d044d819 : SUCCESS in 4m 41s\n- octavia-tox-functional-py39-tips https://zuul.opendev.org/t/openstack/build/fa6c983178ea44d8853cc4a8d9effcb6 : SUCCESS in 6m 46s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/bb4e2eb56b494b9b899dd18d7984a8eb : SUCCESS in 4m 29s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/7991f0bb273c49c5888aaec04c086bff : SUCCESS in 6m 50s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/30461b16382246f79fad4e603c110329 : SUCCESS in 7m 18s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/8df9c31f4d9f4763b563357e90b8cc03 : SUCCESS in 7m 14s (non-voting)\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/2f3c362203b44706a3490deeeaaeb4df : SUCCESS in 1h 11m 14s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/04741dc60b03402ba399a07c8f8921cd : SUCCESS in 2h 12m 51s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/4a74d575293e4a5a94bbca5bd1811eb0 : SUCCESS in 35m 52s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/158aabc931864f0fb918b5dc53d8485e : SUCCESS in 41m 51s\n- octavia-grenade-skip-level https://zuul.opendev.org/t/openstack/build/02691c986e2041589b4de98a3d3f5c86 : SUCCESS in 45m 22s\n- octavia-v2-dsvm-tls-barbican-fips https://zuul.opendev.org/t/openstack/build/334d654a9aba46af9fc3b5f3cc6cb5ea : SUCCESS in 56m 19s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/8dfd461a5f5e4f1a981fbfabacb6d5ca : SUCCESS in 35m 47s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/452f4993f5a441078b240ed723ca4ac1 : SUCCESS in 2h 20m 44s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/37f8ca6e5bce4bd49e318980cfceba59 : SUCCESS in 2h 21m 20s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/4333fdf73264410e9f2168ca33501dc0 : SUCCESS in 2h 20m 18s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/bce94d26b57d47cd923f5b02b9bba5b7 : SUCCESS in 2h 17m 49s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/57498690c70443039174a60da9737faf : TIMED_OUT in 2h 32m 55s (non-voting)\n- octavia-v2-dsvm-scenario-centos-9-stream https://zuul.opendev.org/t/openstack/build/a1e0338d8a9e4f68acf4f0d5cc119a2e : SUCCESS in 2h 23m 26s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"15c0a395cf7f9363a52c6da19808f613a673d100","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2023-05-23 08:01:33.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"9dfcf4038ad5d3fe7227d132539135ebaf65ae58","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-23 08:02:12.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"cb58e3ad9ea8a253741f7365a90cfa7f5f5df5dd","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-23 10:35:53.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/6454ed11ba594ae584d6d7619aaad1c8\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5eef834c54214658bac63b42849d10d2 : SUCCESS in 4m 22s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/737600df7d124b8aac4e1076dfc28c35 : SUCCESS in 4m 14s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/df30e3e508414efbae8ad486bc9d0a3e : SUCCESS in 4m 53s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0e644f5c1d6c4f699fe2bdffa19f789e : SUCCESS in 11m 33s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/1d6ae955e8ee446eb5d175741b23eebe : SUCCESS in 4m 45s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/e562d58fe11c4de4a5aee6d98a583b5a : SUCCESS in 6m 37s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/cbda259846f741f48c740b6c540aee55 : SUCCESS in 1h 36m 44s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/e862ca3b0ca74007b1b34b9045303866 : TIMED_OUT in 2h 33m 00s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/d90b90a4407042649a501085c29c37f0 : SUCCESS in 36m 15s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/daf73562d1f7467f877bb0fe833f8933 : SUCCESS in 38m 23s\n- octavia-grenade-skip-level https://zuul.opendev.org/t/openstack/build/3629bbc5feaa40538b64aef8cdac9f51 : SUCCESS in 40m 20s","accounts_in_message":[],"_revision_number":1},{"id":"d1b301c1bd0d747ca6088d8ff20deba613a1056f","author":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"date":"2023-05-29 10:42:45.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"8e0056acb295051a2c55b377be11e28a4cd93eca","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-29 13:17:34.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/632838008cbe48708db84259e78c3a02\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/9844bae2ca5d4ea5877b11719a786274 : SUCCESS in 10m 26s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2da320073388409eb131499478b1f829 : SUCCESS in 5m 48s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/68e272b1b260442fb921d281234d8faa : SUCCESS in 5m 09s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/680246d811fd449db841344ccc79fbff : SUCCESS in 5m 21s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4cc247374fb840628dc0a55864ca8d64 : SUCCESS in 13m 21s\n- octavia-tox-py39-tips https://zuul.opendev.org/t/openstack/build/855bb96068644565a69ef0696ec7f8c1 : SUCCESS in 5m 15s\n- octavia-tox-functional-py39-tips https://zuul.opendev.org/t/openstack/build/24190c90c893486aad3ccad4d80dc8cd : SUCCESS in 7m 39s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/41a70f4fc59140a8b6c603cd502476a3 : SUCCESS in 5m 30s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/7f347802f5564e2897d4b8dd699df590 : SUCCESS in 8m 02s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/27efbd2a730d45948936419104c22dea : SUCCESS in 9m 11s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/1d33a8d6b5f446158f90648fc629adce : SUCCESS in 7m 55s (non-voting)\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/e6e6d108d60f4748b79e22ac21bb6d42 : SUCCESS in 1h 23m 26s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/1f5d2d6101cd4672b6039d516dcb3bd7 : SUCCESS in 2h 17m 36s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/76053fc6ffae48ebbc2a7970513f97da : SUCCESS in 39m 20s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/74ac17248b7d4926a803b1ad8833be9a : SUCCESS in 59m 32s\n- octavia-grenade-skip-level https://zuul.opendev.org/t/openstack/build/ac017abbb3c44a71b283fe97e6a11705 : SUCCESS in 46m 15s\n- octavia-v2-dsvm-tls-barbican-fips https://zuul.opendev.org/t/openstack/build/da9e9c6eed9c4cf191dd8fd2578a8e4e : FAILURE in 14m 57s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/1a4a9285aa84424a87de43469d25256b : SUCCESS in 33m 54s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/b027632e484846cd98b8780bcb4264b9 : SUCCESS in 2h 30m 24s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/797bca4af64041bca8d4d7b08f7df029 : SUCCESS in 2h 28m 11s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/d6a0f2af46fb4e5697b47c477385397e : TIMED_OUT in 2h 32m 31s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/41fd3f3a279d4f9faacbd52cef06416c : SUCCESS in 2h 29m 29s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/ecc2c675b2c74ce690184990773ca747 : TIMED_OUT in 2h 32m 47s (non-voting)\n- octavia-v2-dsvm-scenario-centos-9-stream https://zuul.opendev.org/t/openstack/build/de1080d765844c0ca0614b6ba6807c21 : SUCCESS in 2h 17m 55s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"fc901a34344661d291e7a0fcf381f1e3d19fab01","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-29 13:18:19.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"e192a6ae9d29f4866f22393102d451dc4e03c70b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-29 15:51:37.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/1e0ac16d5f64419f92706d4c9e1ce801\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/473aca36b6b848a98168442343e6e0af : SUCCESS in 5m 29s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/1192da010d9647968a30a9d89ebe9c16 : SUCCESS in 5m 27s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/f30c12eb5ffa439f80ebd73366f78ecd : SUCCESS in 4m 05s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/46a8326017834caab4df45828ae93fcd : SUCCESS in 12m 25s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/77a2fdd117ac49a68b3103c579d6d714 : SUCCESS in 4m 50s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/059233f9469a443d8d298df71aa20c22 : SUCCESS in 7m 23s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/6c932c7b5f924aaa9511c62a0bbf8741 : SUCCESS in 1h 20m 04s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/2d75cf5004cf4be79ece01edcf073bc4 : TIMED_OUT in 2h 32m 46s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/a36567b78c2945139948141e04af3e8f : SUCCESS in 35m 27s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/83d153b875c8400d9f05ee4ba7262dd2 : SUCCESS in 50m 52s\n- octavia-grenade-skip-level https://zuul.opendev.org/t/openstack/build/f13ec986a7424d57bc6dd8c3a58ff93e : SUCCESS in 39m 07s","accounts_in_message":[],"_revision_number":1},{"id":"c2e4645631b99c9c41e50991a566a73f35df84e6","author":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"date":"2023-05-29 15:53:14.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"193f8d83a48cdae30635b0a34a7c086c45b07744","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-29 18:56:31.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/eddf3d6428924fd5a8955047afa48e37\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/60e122c434084846b38251dedaab069a : SUCCESS in 13m 05s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0fdd31c6344f4c29967d02ab690c3308 : SUCCESS in 5m 25s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/e8ed4cfc613f44768ac56eef0aa82f0c : SUCCESS in 6m 08s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/1491a605f62a4400a49e51c8932b5de2 : SUCCESS in 4m 43s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/db59a8b7ce854eacb97242c6b7b0c137 : SUCCESS in 10m 35s\n- octavia-tox-py39-tips https://zuul.opendev.org/t/openstack/build/4499dd515d5e471384594291d08a6c5e : SUCCESS in 5m 24s\n- octavia-tox-functional-py39-tips https://zuul.opendev.org/t/openstack/build/f4eab6d4c0a64acdaf1dc5364dec28ac : SUCCESS in 8m 37s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/6dfa3c32d4f44580be2d29c3be536b43 : SUCCESS in 6m 03s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/5b56468e696c4494b539ea00c38cd088 : SUCCESS in 7m 38s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/637f2e742d8a473f8b54e9488dd38951 : SUCCESS in 7m 13s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/1b66ab3769eb418bbd9622e0d6036aa7 : SUCCESS in 6m 29s (non-voting)\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/8d939d9e210849e5831e7e9f089ed8c6 : SUCCESS in 1h 30m 41s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/c192059e3d9c4dff8d99955271399615 : TIMED_OUT in 2h 32m 43s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/b5d2a5853d6147cf83086e57d7a689fb : SUCCESS in 39m 46s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/93c3db76da2a4e6795d55549cf4ca24d : SUCCESS in 54m 57s\n- octavia-grenade-skip-level https://zuul.opendev.org/t/openstack/build/0f69ebae378c4be395e06c5db4546763 : SUCCESS in 39m 26s\n- octavia-v2-dsvm-tls-barbican-fips https://zuul.opendev.org/t/openstack/build/57f4821199364d3aad1df0155e424095 : FAILURE in 11m 40s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/db661d1fd6ba45159fc1cc0b92ecd739 : SUCCESS in 34m 19s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/d63e82a829d94bbbac8e886670c0651c : TIMED_OUT in 2h 34m 34s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/3db3fc53ddc24dbe86e2e9925a67fe2b : SUCCESS in 3h 01m 53s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/a00d18b2f2f94b2696b08b77e9d5e6e3 : SUCCESS in 2h 05m 49s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/dd611a6bc6c44c02904dc7613d392b9f : TIMED_OUT in 2h 33m 52s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/fb9a1924d5a14951a7678f5d7ec8235c : TIMED_OUT in 2h 32m 38s (non-voting)\n- octavia-v2-dsvm-scenario-centos-9-stream https://zuul.opendev.org/t/openstack/build/48556935b20141e98bf2c70f0bc94927 : SUCCESS in 2h 32m 51s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"099e50e3d4f7fc447fcd0d537b7770617d4e5f49","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2023-07-03 06:50:34.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"515d6875700f7e97091f781a00e9a3e1263b1d57","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-03 10:03:14.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/707a640343f8482587766e2e3408cb5f\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/beed9e120502435ab78efe2814a774fc : SUCCESS in 9m 16s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6417b0198f3f46d5a4703cabda2aa115 : SUCCESS in 4m 37s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/c39f1d4d526d45438975039d488914b4 : SUCCESS in 5m 55s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/13cab98ccbc943c1b49826b223550168 : SUCCESS in 4m 14s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/20ae1d750d174fb7b99101d2eabb45f7 : SUCCESS in 11m 38s\n- octavia-tox-py39-tips https://zuul.opendev.org/t/openstack/build/04e70ff60bea49a09de2daa21d82eefc : SUCCESS in 4m 57s\n- octavia-tox-functional-py39-tips https://zuul.opendev.org/t/openstack/build/54f1621e33bb40a18bc3c796e0c72a8e : SUCCESS in 8m 42s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/3b95174b35c341858539f24b63b41f3d : SUCCESS in 3m 48s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/2421dc5c17e04cc2a3ebf7f7faee9a83 : SUCCESS in 8m 25s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/fdc0542c524d46cab3e3db63d0382112 : SUCCESS in 8m 12s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/7c297e16bc594c2d8ce6ce91c9afd13c : SUCCESS in 7m 29s (non-voting)\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/8584bf700928467cb1a728f58fa31b38 : TIMED_OUT in 3h 07m 45s\n- octavia-v2-dsvm-scenario-traffic-ops https://zuul.opendev.org/t/openstack/build/0f06ec5a246941ce8e73d31425332c5b : SUCCESS in 1h 55m 23s\n- octavia-v2-dsvm-scenario-non-traffic-ops https://zuul.opendev.org/t/openstack/build/7e3965b3030c4ac8bb36655767a1fbde : SUCCESS in 1h 02m 32s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/e74ba53327fa46afb865ab297fd48665 : SUCCESS in 37m 31s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/4497e47a512247a4ba672c033059a10e : FAILURE in 3h 02m 13s\n- octavia-grenade-skip-level https://zuul.opendev.org/t/openstack/build/7f44d165b1494f8a83511e93013f35eb : SUCCESS in 44m 35s\n- octavia-v2-dsvm-tls-barbican-fips https://zuul.opendev.org/t/openstack/build/a02776348c15450d8218b5ff4cc5b7f9 : FAILURE in 10m 04s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/ffcbaefa4b9b4cda8123da249a7ea5cb : SUCCESS in 32m 49s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/2ea44d0653af4adba21b34db2891b408 : TIMED_OUT in 2h 32m 48s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/3d3c0a6f9fa14e7ab2c288a8b82f5127 : SUCCESS in 2h 44m 35s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/07e83968942946bc9783cb496572a193 : TIMED_OUT in 2h 33m 16s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/ce089138f10e49bab2c164889fd19ab7 : TIMED_OUT in 2h 32m 34s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/3c1e050766624a20b5d6295cc7da19e2 : TIMED_OUT in 2h 32m 25s (non-voting)\n- octavia-v2-dsvm-scenario-centos-9-stream https://zuul.opendev.org/t/openstack/build/dcaaa4f3229b4b09afaa85a3d37ea058 : SUCCESS in 2h 26m 33s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"57c6dab5ac0e09f3d9102eeeec6ce0c1fa3f3e1f","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2023-07-03 15:57:15.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"f401f8a94041b04e37c5f25b5d8b179aac562c24","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-03 18:58:32.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/41463187b28c416a907f418558a4ff9d\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/20c2d9f2b1ea40d8b511e5253209e4fb : SUCCESS in 10m 28s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c8adcc19ee2e4a93882dcaee43697af3 : SUCCESS in 6m 10s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/3724abb2b72846949f3d2be7a230f404 : SUCCESS in 4m 56s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/a7e06c59793a41839ebf361b9641c072 : SUCCESS in 5m 07s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8c79934cd1854b07b357140ce47e4e45 : SUCCESS in 12m 23s\n- octavia-tox-py39-tips https://zuul.opendev.org/t/openstack/build/7bd0cf9adffe468ea4760cf9486ca9ed : SUCCESS in 5m 14s\n- octavia-tox-functional-py39-tips https://zuul.opendev.org/t/openstack/build/31e02f43d4d2419ebe377b4abbf41a72 : SUCCESS in 7m 37s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/1f579231cbf1446dbe58ecfafced37bd : SUCCESS in 4m 42s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/2eb9984be32e413d96c072eb4b40bdef : SUCCESS in 8m 09s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/42823d5e263b445eb68eb2878910ae92 : SUCCESS in 7m 43s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/a8a23680e54b4f5cab65af61cbf9928b : SUCCESS in 7m 07s (non-voting)\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/548c2f9cc1b648f4aaa010e8402c91d7 : SUCCESS in 1h 41m 33s\n- octavia-v2-dsvm-scenario-traffic-ops https://zuul.opendev.org/t/openstack/build/6ec16034891047b4aa824e0badfcda8d : SUCCESS in 1h 56m 01s\n- octavia-v2-dsvm-scenario-non-traffic-ops https://zuul.opendev.org/t/openstack/build/1221f7eb2b8d47ed89f486a5e2f04ae7 : SUCCESS in 1h 24m 23s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/157a54bab7224b6796d230de4be3b92c : SUCCESS in 46m 12s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/780611894c734275a27a25001e8af1d4 : SUCCESS in 41m 17s\n- octavia-grenade-skip-level https://zuul.opendev.org/t/openstack/build/f29c0e59b7e742e3b3461d5d34316a46 : SUCCESS in 43m 43s\n- octavia-v2-dsvm-tls-barbican-fips https://zuul.opendev.org/t/openstack/build/56033b5ccef141978843b694d79a4faf : FAILURE in 10m 42s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/2f298953619c494c8d7e12e28ccb264d : SUCCESS in 35m 47s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/c65c0839727c4b578d034ba7510fb50a : TIMED_OUT in 2h 33m 48s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/0bbb1e1655e141b8b4b86b221d83dfc6 : SUCCESS in 2h 59m 37s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/82c754c39d5b444d882b5268e2b08d01 : TIMED_OUT in 2h 32m 51s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/b03a184d97d941d4a51750461babd279 : TIMED_OUT in 2h 32m 41s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/d51c1d5a9b98412e843b1e7895d86b75 : TIMED_OUT in 2h 33m 25s (non-voting)\n- octavia-v2-dsvm-scenario-centos-9-stream https://zuul.opendev.org/t/openstack/build/505c2f3be46d47199b18d021324569e2 : SUCCESS in 2h 32m 48s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"775348326337a09ef2d2f346981c39a5179e47d9","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-03 18:59:11.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"1fcde7dcc950888f28b95ce36e36cf9a56049676","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-03 20:35:23.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a725107fbab343dba55a036699d139b0\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6b9465379eb3445496605e60c2bd2214 : SUCCESS in 5m 23s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/7695fc5c41664fc09b03a6aefade895b : SUCCESS in 5m 34s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/5256c0385ca64a8ebaaf4f1a4a740a07 : SUCCESS in 4m 59s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/aa4c09dc5e9f4f2ba3193ed1cfb862eb : SUCCESS in 10m 54s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/4bb248b0f92b40b8b19f3dadb6016502 : SUCCESS in 5m 06s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/8a20b648989a437f8e6b105b1ccbf1ee : SUCCESS in 7m 25s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/80aeb51689234e9486b20c3deb9fc20a : SUCCESS in 1h 27m 10s\n- octavia-v2-dsvm-scenario-traffic-ops https://zuul.opendev.org/t/openstack/build/b1a5ae51edd2450296b7932c03607dae : SUCCESS in 1h 34m 31s\n- octavia-v2-dsvm-scenario-non-traffic-ops https://zuul.opendev.org/t/openstack/build/ba2ee4f821a34c12a172921d9ec819f5 : SUCCESS in 1h 14m 46s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/a631736e3ac744afb44a4ca3832465e4 : SUCCESS in 36m 37s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/ef5a08ffb2ce40959b7022c985c0e1d9 : SUCCESS in 37m 44s\n- octavia-grenade-skip-level https://zuul.opendev.org/t/openstack/build/92366d36b6a0437db3eb32fbf73d7ca6 : SUCCESS in 39m 38s","accounts_in_message":[],"_revision_number":1},{"id":"5b99d31591ce07dd23478cbe114ea4b360621a4f","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-03 20:35:23.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"84f6a5a8cd51a3823328c34b5ffa103b7e9e1d2c","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-03 20:37:18.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f0853419b32148f58b978f54d1e37d3e\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bad1a91d8be5490e873db3f6804dfe10 : SUCCESS in 46s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"abeaa55051822c25907fa87ca19a7b4151db006a","revisions":{"abeaa55051822c25907fa87ca19a7b4151db006a":{"kind":"REWORK","_number":1,"created":"2023-05-15 07:41:32.000000000","uploader":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"ref":"refs/changes/35/883135/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/35/883135/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/35/883135/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/35/883135/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/35/883135/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/35/883135/1"}}},"commit":{"parents":[{"commit":"c940f7a064b2a49eab797379f548fce1f41f9e78","subject":"Fix pep8 error","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/c940f7a064b2a49eab797379f548fce1f41f9e78"}]}],"author":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2023-04-07 05:53:15.000000000","tz":120},"committer":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2023-05-15 07:41:19.000000000","tz":120},"subject":"Fix octavia-amphora-image-build periodic job","message":"Fix octavia-amphora-image-build periodic job\n\nThe octavia-amphora-image-build periodic job fails with\n\n  The task includes an option with an undefined variable. The error was:\n  \u0027amphora_os\u0027 is undefined\n\nThis issue was introduced in Ic1ec002062f91b7d3a43669e86380a2ffbdd7b18\n\nWe must keep the definition of the amphora_os and amphora_os_release\nvariables in the parent octavia-amphora-image-build job.\n\nChange-Id: I51d2c011f123e7d8b90b5cea9ba875052d7837a3\n(cherry picked from commit cd1646cec42402661451e7874e3364437249695a)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/abeaa55051822c25907fa87ca19a7b4151db006a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/abeaa55051822c25907fa87ca19a7b4151db006a"}]},"branch":"refs/heads/stable/2023.1"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
