)]}'
{"id":"openstack%2Foctavia~884702","triplet_id":"openstack%2Foctavia~stable%2Fzed~I9deaea182568a23e8c1104d806273dc87f5caa72","project":"openstack/octavia","branch":"stable/zed","topic":"fix_hm_op_status","attention_set":{},"removed_from_attention_set":{"31664":{"account":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"last_update":"2023-07-13 18:09:27.000000000","reason":"Change was submitted"},"7249":{"account":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"last_update":"2023-07-13 18:09:27.000000000","reason":"Change was submitted"},"19298":{"account":{"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},"last_update":"2023-07-13 18:09:27.000000000","reason":"Change was submitted"},"1131":{"account":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"last_update":"2023-07-13 18:09:27.000000000","reason":"Change was submitted"},"29244":{"account":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"last_update":"2023-07-07 13:09:52.000000000","reason":"removed on reply"},"34429":{"account":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"last_update":"2023-07-13 18:09:27.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I9deaea182568a23e8c1104d806273dc87f5caa72","subject":"Fix hm operating status to ONLINE in single lb call","status":"MERGED","created":"2023-05-30 13:44:32.000000000","updated":"2023-07-13 18:10:46.000000000","submitted":"2023-07-13 18:09:27.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"884702-fix_hm_op_status","meta_rev_id":"64f91e1134df120737964655739fad6121f156ce","_number":884702,"virtual_id_number":884702,"owner":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"value":0,"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-07-13 18:09:27.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"all":[{"value":0,"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"value":0,"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":2,"date":"2023-07-07 13:09:52.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":2,"date":"2023-07-13 16:22:56.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"all":[{"value":0,"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"value":0,"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":1,"date":"2023-07-13 16:22:56.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"value":0,"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}],"reviewers":{"REVIEWER":[{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-05-30 16:48:53.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-06-20 13:19:17.000000000","updated_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"reviewer":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"state":"REVIEWER"},{"updated":"2023-06-21 08:13:28.000000000","updated_by":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"reviewer":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"state":"REVIEWER"},{"updated":"2023-06-21 08:13:28.000000000","updated_by":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"reviewer":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"state":"REVIEWER"},{"updated":"2023-06-21 08:13:28.000000000","updated_by":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"reviewer":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"state":"REVIEWER"},{"updated":"2023-06-23 12:14:40.000000000","updated_by":{"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},"reviewer":{"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},"state":"REVIEWER"},{"updated":"2023-07-13 16:22:56.000000000","updated_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"reviewer":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"state":"REVIEWER"}],"messages":[{"id":"1263e7c9db903331ef255976e2061df2c9f54b0f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"date":"2023-05-30 13:44:32.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"645c4143af3b57b5bb254c6fb556fb071d7ea854","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-30 16:48:53.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4fc3b0ea11df4361978eb129bc0ec095\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/aeb09a001fa74f20aaf0d309ff4b2de7 : SUCCESS in 11m 25s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/538be0abb7234548949f22f2eab4e42d : SUCCESS in 7m 35s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/5ce6341c6f4e476781cdb52d905b0280 : SUCCESS in 6m 36s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/558a7f93a0c74177a94d954f096c12e1 : SUCCESS in 4m 26s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/86b09b957c8e4eb18de27f5c95212a7a : SUCCESS in 5m 05s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b5a724fec0604d659dc0a9538493a2b2 : SUCCESS in 12m 07s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/155ecd122e5b42a19b8287728c7201cf : SUCCESS in 4m 59s\n- octavia-tox-py39-tips https://zuul.opendev.org/t/openstack/build/178f3c6dd1ba44b4a54ac9bbf13722e3 : SUCCESS in 6m 41s\n- octavia-tox-functional-py39-tips https://zuul.opendev.org/t/openstack/build/bb201edcd6d04b4c894c580d658afc96 : SUCCESS in 8m 11s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/4a69aec807874315b78e2d2e2716f5b1 : SUCCESS in 4m 17s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/52351a5052154d64a26a1217699ef1a3 : SUCCESS in 8m 58s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/5dab408be61f40e098fdb209430212df : SUCCESS in 9m 18s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/59b3971db44e4475a86fd0100fceb06f : SUCCESS in 9m 15s (non-voting)\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/bc9d99ea54bd4cc9bdee91315476e00e : SUCCESS in 1h 31m 07s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/61cf473c39124173b5344c925cdea041 : SUCCESS in 1h 55m 48s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/c9d75b2b0e144fe1b30b125e3717856d : SUCCESS in 47m 37s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/5833ef117f42411cb4de9301cb79d314 : SUCCESS in 39m 37s\n- octavia-v2-dsvm-tls-barbican-fips https://zuul.opendev.org/t/openstack/build/64995ecc640c41a7a4a9bb581b6dc45f : FAILURE in 12m 28s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/7808ccf0ccbd4c619939799d724eb997 : SUCCESS in 33m 34s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/c21790a0c41440be94d67b45f54399ce : SUCCESS in 1h 50m 32s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/a056b6e4a973400a84040950201636d9 : TIMED_OUT in 3h 03m 04s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/79ddac6f0c884e76b3b65e7b493f2d79 : TIMED_OUT in 2h 32m 46s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/53529cf55b5541baa609279cf128464b : SUCCESS in 1h 58m 19s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/26d570f5beba4ac088bbced663f7208c : SUCCESS in 1h 45m 26s (non-voting)\n- octavia-v2-dsvm-scenario-centos-9-stream https://zuul.opendev.org/t/openstack/build/7364bec55a7a4737b51c0d039dee2566 : SUCCESS in 2h 23m 55s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"bf0d838d232ab0024c29e53012f84190ac7685d2","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2023-06-20 13:19:17.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"30287fa29bf4540a6fe51d6feba4bc2757cc484f","author":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"date":"2023-06-21 08:38:40.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"e2cd9497715e19c4d752b064b68c4ea722e85251","author":{"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},"date":"2023-06-23 12:14:40.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"937721f6ffbd07658819aa548051f6f57c55fab7","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"date":"2023-06-28 16:41:18.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Code-Review+1, Code-Review+2 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":2},{"id":"17a6106c08a2d55fc0f87dd840f5479fb8680d46","author":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"date":"2023-06-28 16:42:10.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"4e357f73238b042c737716ced9de1830bf76a411","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-28 19:33:31.000000000","message":"Patch Set 2: Verified-1\n\n(2 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/fb2c3f51f5454d04a048954fd08fa29d\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/295bed4501d04927ada1be6ae641ea49 : SUCCESS in 10m 50s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8d1baf352d87494482373510b5a988c7 : FAILURE in 4m 11s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/fec318c292cb49db904692b036d8d30f : SUCCESS in 4m 40s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/137951964aea47a6b7f4a0973f5b7d97 : SUCCESS in 12m 47s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/3008621ac3de4b23a14570b54bd52b2f : SUCCESS in 8m 31s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d43149c848164cf584b8ef5bf24cc39f : SUCCESS in 10m 50s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/574fe8d825944277beeda880f8b6d088 : SUCCESS in 5m 11s\n- octavia-tox-py39-tips https://zuul.opendev.org/t/openstack/build/fca72d47fdb94c82969c146335628782 : SUCCESS in 12m 18s\n- octavia-tox-functional-py39-tips https://zuul.opendev.org/t/openstack/build/279e4651694d418cb2247cebb61c8f25 : SUCCESS in 13m 00s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/317c9d7cb1064552b28a113e24d88c1b : SUCCESS in 8m 03s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/fd2adb89af7f470a82ea46aae290216d : SUCCESS in 9m 45s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/1448872041534051885367d2e56199d6 : SUCCESS in 12m 20s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/0b2339a2f2bc4391bb9d7e7f3c7f5aef : SUCCESS in 10m 53s (non-voting)\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/911b591588c34f699a411359c564930d : SUCCESS in 1h 26m 00s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/de9432c46bca430a99554f6891d9ea71 : SUCCESS in 1h 59m 22s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/f4c19668aa374161a9864ed366154923 : SUCCESS in 49m 35s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/7082c89cdabd472fa2927b2d0c0713e1 : SUCCESS in 53m 26s\n- octavia-v2-dsvm-tls-barbican-fips https://zuul.opendev.org/t/openstack/build/109593f1bfd5431b82a2d48a94d52cbf : FAILURE in 11m 05s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/65a95c6b115b4ed7880be64dfbb5ecfc : SUCCESS in 33m 18s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/5776281e5eab4e64baf8851a049d6565 : SUCCESS in 1h 53m 23s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/35264756af93462e9afecc0d0051c429 : SUCCESS in 2h 50m 18s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/95fc899d35f1444da296a19691f36f96 : TIMED_OUT in 2h 32m 46s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/ae47b659a7b84232ab00d7b3e63cc560 : SUCCESS in 2h 06m 47s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/7c6c83bf46864f77a4d6008533c959cf : SUCCESS in 1h 31m 35s (non-voting)\n- octavia-v2-dsvm-scenario-centos-9-stream https://zuul.opendev.org/t/openstack/build/31b3d6a892374084832a3ad9fb4d58a3 : SUCCESS in 2h 05m 42s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"2496e644812522d03ee6ce44765e1700a2359e5e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"date":"2023-06-29 08:47:11.000000000","message":"Patch Set 3: Published edit on patch set 2.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":3},{"id":"1b4de5bd81594e1d63d6875d800fd79135d90ae0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-29 11:54:03.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3a3c079335e54db5abb8ed5278554189\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/3ac2d1c376c04557b7e7cd36ce8f7898 : SUCCESS in 11m 28s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/15e73a83f70442a8bdec3ef69d671271 : SUCCESS in 6m 23s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/c5e45daade77436e87ae245c9280aac8 : SUCCESS in 5m 38s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/b4eaee62d6254b0e9f6bb42126d2b2cc : SUCCESS in 4m 48s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/bf8837b725d247bea431f17bb4e8e7c3 : SUCCESS in 5m 22s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a7ab778a74a745909431901f6ffb67f3 : SUCCESS in 11m 14s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/cb7fe03e4003448aa265709b41535f89 : SUCCESS in 6m 00s\n- octavia-tox-py39-tips https://zuul.opendev.org/t/openstack/build/5f28a8e7ed4d45e7afac94526dff76d1 : SUCCESS in 5m 55s\n- octavia-tox-functional-py39-tips https://zuul.opendev.org/t/openstack/build/0f5c7b83299a44858615aad650f550e6 : SUCCESS in 9m 16s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/a776a11e271b43aa8d4e3d6602babf6d : SUCCESS in 4m 38s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/1f78ad9da925441e8f45b6d91c778deb : SUCCESS in 7m 43s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/9e72ddd8fafc423d82d2e7187b173986 : SUCCESS in 8m 29s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/6ca6c13cf9644809bd99dfe26125ad5f : SUCCESS in 8m 20s (non-voting)\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/caeab82acc5a4372b81982656f1e7d18 : SUCCESS in 1h 23m 48s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/d4632e6b98b0437a85a89ca146ded7dd : SUCCESS in 1h 48m 12s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/617cfa53b7dc43e5aba377875ae5d83d : SUCCESS in 46m 10s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/61ec11d24fba47f7acdf1f9735df9b1f : SUCCESS in 47m 30s\n- octavia-v2-dsvm-tls-barbican-fips https://zuul.opendev.org/t/openstack/build/f9306e2a359a4f268be7f83e19e006fc : FAILURE in 11m 34s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/cfe6f98f92f54af2bfcd1466b57e113a : SUCCESS in 35m 44s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/f4de99bfc02045eca2b429011ece1494 : SUCCESS in 2h 04m 40s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/0d772e2d13714dc588fef065db62623d : TIMED_OUT in 3h 03m 55s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/fec9e1f3d2e541ee8a0ef1f8d7d6bf2f : SUCCESS in 2h 30m 43s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/25e664c808784abeb37dfc2a2a334d05 : SUCCESS in 1h 55m 36s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/9eb9eea346c54a278e2970dd36400b15 : SUCCESS in 1h 39m 38s (non-voting)\n- octavia-v2-dsvm-scenario-centos-9-stream https://zuul.opendev.org/t/openstack/build/e7895c507105445d969ed42240bde882 : SUCCESS in 2h 05m 04s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"9f0c745fbd8e194fea4f79d00a7a763bc185e118","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2023-07-07 13:09:52.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"12d04beaa7856cd657dc77561ebbdc22d969beec","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2023-07-13 16:22:56.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"05d4f1ce59d7ccbd7ac227bc5167e3de05d53e42","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-13 16:23:38.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"d11670bbb30471c82539dd48a71ea544471a9081","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-13 18:09:27.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b62f5755cd0c42ca9c37d6151811fd2a\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/cc01dc77ed2646d6aed08d8107ecb37a : SUCCESS in 6m 34s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/30c4116a7d16410091f0d4d8e08891d8 : SUCCESS in 6m 08s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/c9434a393ff144a1b81d6f7c3023e4ef : SUCCESS in 5m 45s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/191e64dda110406aa4e1aaabde5c25f8 : SUCCESS in 10m 02s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/f3bf56c3ad3e439ebd6b0a5eee5ddc73 : SUCCESS in 5m 10s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/e5afbddfefc64573a50472797e842cdd : SUCCESS in 5m 09s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/e6ff3a9c4d414402b525904dcadb3596 : SUCCESS in 8m 51s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/f7643178f1ea4851bbf2ac90bfa13552 : SUCCESS in 1h 21m 38s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/22f51cb966e24a2b89bfa7dee2fbf43c : SUCCESS in 1h 38m 21s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/04a5ff0b36844bb2b6f6dba14a27ab58 : SUCCESS in 35m 46s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/351b3dc87a3a4c7ca52289fa7ebb1b47 : SUCCESS in 38m 46s","accounts_in_message":[],"_revision_number":3},{"id":"0f9ef5102f695236f70edd4934457842c091b637","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-13 18:09:27.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":3},{"id":"64f91e1134df120737964655739fad6121f156ce","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-13 18:10:46.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4057975b70604279ad03a0d7c6d3e256\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/16f38d445e9a4ea591c438656cfe1b63 : SUCCESS in 59s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/3412ef892fe34dac97938c0f846badaf : SUCCESS in 42s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"2e034c1eab22617ef949d81c8c81ea9ce700c750","revisions":{"713edbc792796eaa963a30487fdff4183ce4f58a":{"kind":"REWORK","_number":1,"created":"2023-05-30 13:44:32.000000000","uploader":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"ref":"refs/changes/02/884702/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/02/884702/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/02/884702/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/02/884702/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/02/884702/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/02/884702/1"}}},"commit":{"parents":[{"commit":"ba0c244df25d408f6a4581217ce5ef5c8ab46d6c","subject":"Fix pep8 error","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/ba0c244df25d408f6a4581217ce5ef5c8ab46d6c"}]}],"author":{"name":"Omer","email":"oschwart@redhat.com","date":"2022-12-20 15:58:23.000000000","tz":60},"committer":{"name":"Omer","email":"oschwart@redhat.com","date":"2023-05-30 13:44:03.000000000","tz":120},"subject":"Fix hm operating status to ONLINE in single lb call","message":"Fix hm operating status to ONLINE in single lb call\n\nSo far, Health Monitor was not set to ONLINE when single-call-creates\nwere used.\n\nThis patch adds a database task which sets the operating status of all\nthe enabled Health Monitors of a LB to ONLINE.\n\nStory 2008400\nTask 41333\n\nChange-Id: I9deaea182568a23e8c1104d806273dc87f5caa72\n(cherry picked from commit e20ceebcf245f3bb74b29287b138a3e42cc72a6c)\n(cherry picked from commit 0681d6f9f59d7a836a9fa3dcdac34d40f869e34e)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/713edbc792796eaa963a30487fdff4183ce4f58a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/713edbc792796eaa963a30487fdff4183ce4f58a"}]},"branch":"refs/heads/stable/zed"},"16594d4e84e798fdbdc037b4a6822dad4cb4652a":{"kind":"REWORK","_number":2,"created":"2023-06-28 16:41:18.000000000","uploader":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"ref":"refs/changes/02/884702/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/02/884702/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/02/884702/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/02/884702/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/02/884702/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/02/884702/2"}}},"commit":{"parents":[{"commit":"ba0c244df25d408f6a4581217ce5ef5c8ab46d6c","subject":"Fix pep8 error","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/ba0c244df25d408f6a4581217ce5ef5c8ab46d6c"}]}],"author":{"name":"Omer","email":"oschwart@redhat.com","date":"2022-12-20 15:58:23.000000000","tz":60},"committer":{"name":"Omer","email":"oschwart@redhat.com","date":"2023-06-28 16:41:08.000000000","tz":120},"subject":"Fix hm operating status to ONLINE in single lb call","message":"Fix hm operating status to ONLINE in single lb call\n\nSo far, Health Monitor was not set to ONLINE when single-call-creates\nwere used.\n\nThis patch adds a database task which sets the operating status of all\nthe enabled Health Monitors of a LB to ONLINE.\n\nStory 2008400\nTask 41333\n\nChange-Id: I9deaea182568a23e8c1104d806273dc87f5caa72\n(cherry picked from commit e20ceebcf245f3bb74b29287b138a3e42cc72a6c)\n(cherry picked from commit 0681d6f9f59d7a836a9fa3dcdac34d40f869e34e)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/16594d4e84e798fdbdc037b4a6822dad4cb4652a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/16594d4e84e798fdbdc037b4a6822dad4cb4652a"}]},"branch":"refs/heads/stable/zed"},"2e034c1eab22617ef949d81c8c81ea9ce700c750":{"kind":"REWORK","_number":3,"created":"2023-06-29 08:47:11.000000000","uploader":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"ref":"refs/changes/02/884702/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/02/884702/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/02/884702/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/02/884702/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/02/884702/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/02/884702/3"}}},"commit":{"parents":[{"commit":"ba0c244df25d408f6a4581217ce5ef5c8ab46d6c","subject":"Fix pep8 error","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/ba0c244df25d408f6a4581217ce5ef5c8ab46d6c"}]}],"author":{"name":"Omer","email":"oschwart@redhat.com","date":"2022-12-20 15:58:23.000000000","tz":60},"committer":{"name":"Omer Schwartz","email":"oschwart@redhat.com","date":"2023-06-29 08:47:11.000000000","tz":0},"subject":"Fix hm operating status to ONLINE in single lb call","message":"Fix hm operating status to ONLINE in single lb call\n\nSo far, Health Monitor was not set to ONLINE when single-call-creates\nwere used.\n\nThis patch adds a database task which sets the operating status of all\nthe enabled Health Monitors of a LB to ONLINE.\n\nStory 2008400\nTask 41333\n\nChange-Id: I9deaea182568a23e8c1104d806273dc87f5caa72\n(cherry picked from commit e20ceebcf245f3bb74b29287b138a3e42cc72a6c)\n(cherry picked from commit 0681d6f9f59d7a836a9fa3dcdac34d40f869e34e)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/2e034c1eab22617ef949d81c8c81ea9ce700c750"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/2e034c1eab22617ef949d81c8c81ea9ce700c750"}]},"branch":"refs/heads/stable/zed"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
