)]}'
{"id":"openstack%2Foctavia~884705","triplet_id":"openstack%2Foctavia~stable%2Fwallaby~I9deaea182568a23e8c1104d806273dc87f5caa72","project":"openstack/octavia","branch":"stable/wallaby","topic":"fix_hm_op_status","attention_set":{},"removed_from_attention_set":{"31664":{"account":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"last_update":"2023-07-13 18:21:49.000000000","reason":"Change was submitted"},"7249":{"account":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"last_update":"2023-07-13 18:21:49.000000000","reason":"Change was submitted"},"19298":{"account":{"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},"last_update":"2023-07-13 18:21:49.000000000","reason":"Change was submitted"},"1131":{"account":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"last_update":"2023-07-13 18:21:49.000000000","reason":"Change was submitted"},"29244":{"account":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"last_update":"2023-07-13 18:21:49.000000000","reason":"Change was submitted"},"34429":{"account":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"last_update":"2023-07-13 18:21:49.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I9deaea182568a23e8c1104d806273dc87f5caa72","subject":"Fix hm operating status to ONLINE in single lb call","status":"MERGED","created":"2023-05-30 13:47:52.000000000","updated":"2023-07-13 18:23:00.000000000","submitted":"2023-07-13 18:21:49.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"884705-fix_hm_op_status","meta_rev_id":"e790bcc4b3f033c536aa104b0de2db5f1ef3d298","_number":884705,"virtual_id_number":884705,"owner":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"value":0,"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-07-13 18:21:49.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"all":[{"value":0,"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"value":0,"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2023-07-07 13:10:20.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":2,"date":"2023-07-13 16:23:40.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"all":[{"value":0,"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"value":0,"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":1,"date":"2023-07-13 16:23:40.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"value":0,"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}],"reviewers":{"REVIEWER":[{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-05-30 15:49:32.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-06-20 13:20:23.000000000","updated_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"reviewer":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"state":"REVIEWER"},{"updated":"2023-06-21 08:12:34.000000000","updated_by":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"reviewer":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"state":"REVIEWER"},{"updated":"2023-06-21 08:12:34.000000000","updated_by":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"reviewer":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"state":"REVIEWER"},{"updated":"2023-06-21 08:12:34.000000000","updated_by":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"reviewer":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"state":"REVIEWER"},{"updated":"2023-06-23 12:15:24.000000000","updated_by":{"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},"reviewer":{"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},"state":"REVIEWER"},{"updated":"2023-07-13 16:23:40.000000000","updated_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"reviewer":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"state":"REVIEWER"}],"messages":[{"id":"8f61f16f240a71cf74d290bb9ebb3fc864abd319","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"date":"2023-05-30 13:47:52.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"5acdc2cbe11fb6b2bd0f3a0b1c1da3b841dd4b50","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-30 15:49:32.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/dfbe7b06d4b84f1eb3e362c83b5994a5\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/7f4dcc03b76f4b82af8a7a3336ea4815 : SUCCESS in 8m 15s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/df1d1e2e8faf4cc98bf500772f24d4c4 : SUCCESS in 7m 28s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d9f3b41e5fcf45eaacb14d2ac7e6e6b9 : SUCCESS in 4m 34s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/1639d5cb9d634e2a93cbf1dcd990f28e : SUCCESS in 6m 30s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/55d234dc73de4f4381bbbfd30eb53376 : SUCCESS in 4m 44s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/ce3778e22041409a9f528fa2884305a7 : SUCCESS in 6m 36s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9838ec44e66642abb1039a8957f9b27d : SUCCESS in 11m 29s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/0c572e524df04749b57a2703adf7500a : SUCCESS in 4m 56s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/62b1564957614bff82c3fe1c9be5b6fc : SUCCESS in 6m 30s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/ca9cb899220040378ebf9f97f957fb8e : SUCCESS in 6m 47s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/e9c9b023d61543b8acdcbcd50ea2afac : SUCCESS in 4m 44s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/64757552184646aebe8bd9ed424031cb : SUCCESS in 8m 31s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/15188dcf08874d73addd1be655c7040f : SUCCESS in 1h 16m 48s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/527f8931848641f69c08d0f22f1d72aa : SUCCESS in 1h 30m 51s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/be97d58adebb45128207a2e47fde6840 : SUCCESS in 34m 17s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/04535b12e3684a888a2b0a9de06bfde7 : SUCCESS in 30m 59s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/170a583dff1240a4b0d1d8444d2c26c5 : SUCCESS in 35m 48s\n- octavia-v2-dsvm-tls-barbican-fips https://zuul.opendev.org/t/openstack/build/38d7c090ce934ecfa1242aa057f1208d : FAILURE in 9m 49s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/6e235b6f9b6c4e16904461c692f00e2b : SUCCESS in 1h 35m 59s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/465ac277baf94769926da64b556dc6ca : FAILURE in 12m 46s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/1901a8f79fa9431d85a39deaf3bcb8a1 : FAILURE in 5m 23s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/3392a938ed8d4ad8bd6b2a47c2636c48 : SUCCESS in 1h 40m 16s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/ab68d0bd0fb344b991642ed6cca6bd41 : SUCCESS in 1h 56m 30s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8 https://zuul.opendev.org/t/openstack/build/0aab270ba8d0471d98dcb67ce46f3967 : NODE_FAILURE Node request 300-0021347260 failed in 0s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"90258bbe6319242d3bb849e50a5907418a351eb7","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2023-06-20 13:20:23.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"454c0033cce525c61fe6d03ea4089a6b26db88d3","author":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"date":"2023-06-21 08:39:04.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"d451b69f15e83614858d5b4491de99a2e8291212","author":{"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},"date":"2023-06-23 12:15:24.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"b1cbac9a363c968b19d5a07fc1115cbd84f47953","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"date":"2023-06-28 16:53:04.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Code-Review+1, Code-Review+2 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":2},{"id":"760419f928c016704a29506a5f411dc8198e9bbc","author":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"date":"2023-06-28 16:53:26.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"cde432937d4f4615a175c38a9b23695c64f6e1b0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-28 18:46:44.000000000","message":"Patch Set 2: Verified-1\n\n(2 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/e94f36eb4f3f4ecc8a199c71f58e0300\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/f17fc246ef46428b8ac337bdd736c789 : SUCCESS in 13m 00s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/de75e47230b84bc3b95b018ec1c64690 : SUCCESS in 8m 00s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0e688ae0ae7047bab6c88cd583bd1f00 : FAILURE in 4m 12s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/1c003e688d784a6983b090bbaab1de74 : SUCCESS in 4m 59s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/91e3ed16967649448f0c702df84e7b23 : SUCCESS in 6m 29s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/6ea487c637434c4b8399727919571439 : SUCCESS in 5m 18s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6a9515573ce34e37b8d395a8236ca8e5 : SUCCESS in 19m 59s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/47260a7fa5d0437ea2cc175c1ad3d4a5 : SUCCESS in 7m 44s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/736c647ad4414afabcdc6aa2f6c63b49 : SUCCESS in 4m 55s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/aba2616d39594db884f33d8ea943443e : SUCCESS in 12m 28s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/fe5344751a234c2296038c7a841b540a : SUCCESS in 7m 11s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/01ae31b6b2ff472cbfcf342c4f1ce246 : SUCCESS in 6m 12s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/6c3071c5d63c4785bfb2666ac0457e91 : SUCCESS in 1h 09m 21s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/67ead7e9dc8046bb89de22020dd0f64e : SUCCESS in 1h 30m 22s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/f566e11cc9034ea39de607a08dfa92f3 : SUCCESS in 35m 08s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/986967c7f13c4e14b30084d274e40d1c : SUCCESS in 31m 45s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/fa2a638b16234ca9b177c5a3dc260368 : SUCCESS in 31m 25s\n- octavia-v2-dsvm-tls-barbican-fips https://zuul.opendev.org/t/openstack/build/fead330353a340698cf337d8e5880fd9 : FAILURE in 8m 45s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/52ca10470a844f2aafa70568845d8504 : SUCCESS in 1h 27m 59s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/d18cc015ca4846d2b2cec7860dcf090c : FAILURE in 10m 22s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/23caf454428d4bbc921e1df620b21326 : FAILURE in 6m 09s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/4f8310200385489cb80a792c83d9afb5 : SUCCESS in 1h 41m 18s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/24be7899e7cd451fbc85c47f44056492 : SUCCESS in 1h 41m 57s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8 https://zuul.opendev.org/t/openstack/build/10d4910dd66f424f99e24fd047e4f2ff : NODE_FAILURE Node request 300-0021566998 failed in 0s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"ae6cb848513ba33237b2a9ada25f950c73b9a05c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"date":"2023-06-29 08:48:16.000000000","message":"Patch Set 3: Published edit on patch set 2.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":3},{"id":"61077fd8c22032006167c45ae4fc352156a14b7e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-29 10:44:38.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/519a50dc9860482badb3f11859bf212e\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/4e2043c8dca54293b0a82ab5d24f99fd : SUCCESS in 9m 04s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/93874d03f7014993b87e959a24f95422 : SUCCESS in 8m 29s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/449ffbb9183947b08b6f284df27ba8d1 : SUCCESS in 4m 24s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/3b43f0e267ee415aab3d9c7e01f392c2 : SUCCESS in 6m 17s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/14c1a87fa0fa43dd8b361ed1d6fa0b15 : SUCCESS in 4m 48s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/4c407be19f1b4ba4b112e62455af1966 : SUCCESS in 5m 43s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f24495898fae4094a0a246ae5efeee26 : SUCCESS in 19m 35s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/8852548f85384545bb337b51821f5763 : SUCCESS in 4m 49s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/748b5fc1fdd34f1db1b787c5b6ca129e : SUCCESS in 6m 18s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/06276e062fd240e38630e84232c7dd2f : SUCCESS in 7m 41s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/b00a7cfa633341aebd0b1e2591c64e38 : SUCCESS in 5m 03s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/f4bd9c4f761844e99ac98db059525201 : SUCCESS in 10m 16s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/1bbd71718c494ea0af2bf8ee3708fd94 : SUCCESS in 1h 25m 03s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/cfc0fe9022184eb4a7cf308a2688768e : SUCCESS in 1h 26m 52s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/d2ca6490f7c64ebe94385ba9a0a7921f : SUCCESS in 36m 07s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/f374d0f8f0514d8895767a616c6fde8d : SUCCESS in 31m 49s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/a5f3982df5174351bad8220e9b3f4d86 : SUCCESS in 33m 23s\n- octavia-v2-dsvm-tls-barbican-fips https://zuul.opendev.org/t/openstack/build/dc9c6e0610884e70826dad1c0d5d7200 : FAILURE in 9m 40s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/8999ff4892e54ee6985711e665c681b4 : SUCCESS in 1h 38m 09s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/8e0ec58455de49c88064a4c864a11e9c : FAILURE in 11m 25s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/b79294becd634a7a9ea646109d37e721 : FAILURE in 5m 11s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/0ad5e45ec89f4826baccd38ab4ab3ef7 : SUCCESS in 1h 54m 05s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/26b386b377e24e10b305e0c260603334 : SUCCESS in 1h 40m 39s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8 https://zuul.opendev.org/t/openstack/build/0bf88b54df024834a0144df04918ebef : NODE_FAILURE Node request 300-0021574221 failed in 0s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"93cf306a592e60b8a681dd4457a51e5d3143e1a1","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2023-07-07 13:10:20.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"72cb5eb8febde0bbbd19289b17ec10b26e57f898","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2023-07-13 16:23:40.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"b5702a1637631faa71e65281548778c154db5c4c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-13 16:24:31.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"0dff60f797363b6ebaac61a68c4e8c1ecc3fd1d4","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-13 18:21:49.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b807ae44259e493f9e2c98795388d4f3\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/f384e44a4a664c3a8733f3714d8bdca8 : SUCCESS in 7m 58s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b5def7962be5491aa9783be5a6c24979 : SUCCESS in 4m 24s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/61b0a546c8324f32b005734d53f479bc : SUCCESS in 5m 00s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/28dde08c52174edd82e6be443c97c15c : SUCCESS in 5m 09s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/333f5ac334b14d6dad80f90e631af9b2 : SUCCESS in 13m 17s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/b7be27be17184c4b94128ecb991dc508 : SUCCESS in 6m 53s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/2c7a9596b64c4d669933b15e030c1bab : SUCCESS in 5m 38s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/1370ddd63808457890fb9ea903fbb289 : SUCCESS in 8m 01s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/b3f4c08a33064b4f8cbd67b3abfa2043 : SUCCESS in 1h 09m 33s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/9a668d47b04f46319fcb5bf45ae20230 : SUCCESS in 1h 47m 16s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/33d50b60d8464b59a6dfe29b5a949516 : SUCCESS in 34m 22s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/c0f54318cdb04a279608193db51aa8a9 : SUCCESS in 32m 32s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/75086d43753e431bb21b48c76be2915f : SUCCESS in 31m 57s","accounts_in_message":[],"_revision_number":3},{"id":"ee51670f24c9457fb3129e94d647aa89c736c27f","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-13 18:21:49.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":3},{"id":"e790bcc4b3f033c536aa104b0de2db5f1ef3d298","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-13 18:23:00.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/092d97035d994c0287415b5bc81a0ebb\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3a17144dadee47a0be5dd1d24c7c58c5 : SUCCESS in 49s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/bdfeb1b2709640179752651c3c4fcdcd : SUCCESS in 36s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"9b7dd2c52b5c29382337d494a65b5d33a48450f8","revisions":{"7806ec4e2d63d0ad68bcf4322700f2018a565201":{"kind":"REWORK","_number":1,"created":"2023-05-30 13:47:52.000000000","uploader":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"ref":"refs/changes/05/884705/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/05/884705/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/05/884705/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/05/884705/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/05/884705/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/05/884705/1"}}},"commit":{"parents":[{"commit":"99d7e2b9f2a99259d075bda462f593fe91d32442","subject":"Merge \"Prevent incorrect reschedule of resumed tasks with jobboard\" into stable/wallaby","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/99d7e2b9f2a99259d075bda462f593fe91d32442"}]}],"author":{"name":"Omer","email":"oschwart@redhat.com","date":"2022-12-20 15:58:23.000000000","tz":60},"committer":{"name":"Omer","email":"oschwart@redhat.com","date":"2023-05-30 13:47:42.000000000","tz":120},"subject":"Fix hm operating status to ONLINE in single lb call","message":"Fix hm operating status to ONLINE in single lb call\n\nSo far, Health Monitor was not set to ONLINE when single-call-creates\nwere used.\n\nThis patch adds a database task which sets the operating status of all\nthe enabled Health Monitors of a LB to ONLINE.\n\nStory 2008400\nTask 41333\n\nChange-Id: I9deaea182568a23e8c1104d806273dc87f5caa72\n(cherry picked from commit e20ceebcf245f3bb74b29287b138a3e42cc72a6c)\n(cherry picked from commit 0681d6f9f59d7a836a9fa3dcdac34d40f869e34e)\n(cherry picked from commit 713edbc792796eaa963a30487fdff4183ce4f58a)\n(cherry picked from commit 8345e19d4cfaeb093ff97e566b6a0f222cfe9a00)\n(cherry picked from commit 6a711e8fa117b123a69e2c52581aed0670e780c1)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/7806ec4e2d63d0ad68bcf4322700f2018a565201"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/7806ec4e2d63d0ad68bcf4322700f2018a565201"}]},"branch":"refs/heads/stable/wallaby"},"d8b8f7bea1f17342112517d6de29fc44457f5710":{"kind":"REWORK","_number":2,"created":"2023-06-28 16:53:04.000000000","uploader":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"ref":"refs/changes/05/884705/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/05/884705/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/05/884705/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/05/884705/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/05/884705/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/05/884705/2"}}},"commit":{"parents":[{"commit":"99d7e2b9f2a99259d075bda462f593fe91d32442","subject":"Merge \"Prevent incorrect reschedule of resumed tasks with jobboard\" into stable/wallaby","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/99d7e2b9f2a99259d075bda462f593fe91d32442"}]}],"author":{"name":"Omer","email":"oschwart@redhat.com","date":"2022-12-20 15:58:23.000000000","tz":60},"committer":{"name":"Omer","email":"oschwart@redhat.com","date":"2023-06-28 16:52:57.000000000","tz":120},"subject":"Fix hm operating status to ONLINE in single lb call","message":"Fix hm operating status to ONLINE in single lb call\n\nSo far, Health Monitor was not set to ONLINE when single-call-creates\nwere used.\n\nThis patch adds a database task which sets the operating status of all\nthe enabled Health Monitors of a LB to ONLINE.\n\nStory 2008400\nTask 41333\n\nChange-Id: I9deaea182568a23e8c1104d806273dc87f5caa72\n(cherry picked from commit e20ceebcf245f3bb74b29287b138a3e42cc72a6c)\n(cherry picked from commit 0681d6f9f59d7a836a9fa3dcdac34d40f869e34e)\n(cherry picked from commit 713edbc792796eaa963a30487fdff4183ce4f58a)\n(cherry picked from commit 8345e19d4cfaeb093ff97e566b6a0f222cfe9a00)\n(cherry picked from commit 6a711e8fa117b123a69e2c52581aed0670e780c1)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/d8b8f7bea1f17342112517d6de29fc44457f5710"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/d8b8f7bea1f17342112517d6de29fc44457f5710"}]},"branch":"refs/heads/stable/wallaby"},"9b7dd2c52b5c29382337d494a65b5d33a48450f8":{"kind":"REWORK","_number":3,"created":"2023-06-29 08:48:16.000000000","uploader":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"ref":"refs/changes/05/884705/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/05/884705/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/05/884705/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/05/884705/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/05/884705/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/05/884705/3"}}},"commit":{"parents":[{"commit":"99d7e2b9f2a99259d075bda462f593fe91d32442","subject":"Merge \"Prevent incorrect reschedule of resumed tasks with jobboard\" into stable/wallaby","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/99d7e2b9f2a99259d075bda462f593fe91d32442"}]}],"author":{"name":"Omer","email":"oschwart@redhat.com","date":"2022-12-20 15:58:23.000000000","tz":60},"committer":{"name":"Omer Schwartz","email":"oschwart@redhat.com","date":"2023-06-29 08:48:16.000000000","tz":0},"subject":"Fix hm operating status to ONLINE in single lb call","message":"Fix hm operating status to ONLINE in single lb call\n\nSo far, Health Monitor was not set to ONLINE when single-call-creates\nwere used.\n\nThis patch adds a database task which sets the operating status of all\nthe enabled Health Monitors of a LB to ONLINE.\n\nStory 2008400\nTask 41333\n\nChange-Id: I9deaea182568a23e8c1104d806273dc87f5caa72\n(cherry picked from commit e20ceebcf245f3bb74b29287b138a3e42cc72a6c)\n(cherry picked from commit 0681d6f9f59d7a836a9fa3dcdac34d40f869e34e)\n(cherry picked from commit 713edbc792796eaa963a30487fdff4183ce4f58a)\n(cherry picked from commit 8345e19d4cfaeb093ff97e566b6a0f222cfe9a00)\n(cherry picked from commit 6a711e8fa117b123a69e2c52581aed0670e780c1)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/9b7dd2c52b5c29382337d494a65b5d33a48450f8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/9b7dd2c52b5c29382337d494a65b5d33a48450f8"}]},"branch":"refs/heads/stable/wallaby"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
