)]}'
{"id":"openstack%2Foctavia~885490","triplet_id":"openstack%2Foctavia~master~I9c3499b703bf0beab305cc071c74ae3acbcc4c24","project":"openstack/octavia","branch":"master","attention_set":{},"removed_from_attention_set":{"33408":{"account":{"_account_id":33408,"name":"Pierre-Yves Jourel","email":"pierre-yves.jourel@ovhcloud.com","username":"PYJ"},"last_update":"2024-08-01 19:15:34.000000000","reason":"Change was submitted"},"31664":{"account":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"last_update":"2024-08-01 19:15:34.000000000","reason":"Change was submitted"},"27442":{"account":{"_account_id":27442,"name":"Quentin GROLLEAU","email":"quentin.grolleau@corp.ovh.com","username":"QG"},"last_update":"2024-07-03 12:25:33.000000000","reason":"\u003cGERRIT_ACCOUNT_27442\u003e replied on the change","reason_account":{"_account_id":27442,"name":"Quentin GROLLEAU","email":"quentin.grolleau@corp.ovh.com","username":"QG"}},"16707":{"account":{"_account_id":16707,"name":"Mohankumar","email":"nmohankumar1011@gmail.com","username":"mohankumar_n"},"last_update":"2024-08-01 19:15:34.000000000","reason":"Change was submitted"},"32761":{"account":{"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"},"last_update":"2024-08-01 19:15:34.000000000","reason":"Change was submitted"},"29244":{"account":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"last_update":"2024-08-01 19:15:34.000000000","reason":"Change was submitted"},"11628":{"account":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"last_update":"2024-08-01 19:03:41.000000000","reason":"\u003cGERRIT_ACCOUNT_11628\u003e replied on the change","reason_account":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}},"35916":{"account":{"_account_id":35916,"name":"Pham Le Gia Dai","display_name":"Dai","email":"daipham.3213@gmail.com","username":"daiplg","status":"Full Stack Dev"},"last_update":"2024-05-29 08:56:23.000000000","reason":"\u003cGERRIT_ACCOUNT_35916\u003e replied on the change","reason_account":{"_account_id":35916,"name":"Pham Le Gia Dai","display_name":"Dai","email":"daipham.3213@gmail.com","username":"daiplg","status":"Full Stack Dev"}},"34429":{"account":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"last_update":"2024-07-03 07:48:09.000000000","reason":"\u003cGERRIT_ACCOUNT_34429\u003e replied on the change","reason_account":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}}},"hashtags":[],"change_id":"I9c3499b703bf0beab305cc071c74ae3acbcc4c24","subject":"Add specs to resize a load balancer","status":"MERGED","created":"2023-06-07 13:39:55.000000000","updated":"2024-08-01 19:16:51.000000000","submitted":"2024-08-01 19:15:34.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":76,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"885490","meta_rev_id":"b9db70081f716ce6755db38fda61db53011138c7","_number":885490,"virtual_id_number":885490,"owner":{"_account_id":33408,"name":"Pierre-Yves Jourel","email":"pierre-yves.jourel@ovhcloud.com","username":"PYJ"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},{"value":0,"_account_id":27442,"name":"Quentin GROLLEAU","email":"quentin.grolleau@corp.ovh.com","username":"QG"},{"value":0,"_account_id":16707,"name":"Mohankumar","email":"nmohankumar1011@gmail.com","username":"mohankumar_n"},{"value":0,"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-08-01 19:15:34.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":0,"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"recommended":{"_account_id":27442,"name":"Quentin GROLLEAU","email":"quentin.grolleau@corp.ovh.com","username":"QG"},"all":[{"value":0,"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},{"value":1,"date":"2024-07-03 12:25:33.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":27442,"name":"Quentin GROLLEAU","email":"quentin.grolleau@corp.ovh.com","username":"QG"},{"value":0,"_account_id":16707,"name":"Mohankumar","email":"nmohankumar1011@gmail.com","username":"mohankumar_n"},{"value":0,"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2024-08-01 19:03:41.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":2,"date":"2024-07-03 07:48:09.000000000","_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"all":[{"value":0,"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},{"value":0,"_account_id":27442,"name":"Quentin GROLLEAU","email":"quentin.grolleau@corp.ovh.com","username":"QG"},{"value":0,"_account_id":16707,"name":"Mohankumar","email":"nmohankumar1011@gmail.com","username":"mohankumar_n"},{"value":0,"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2024-08-01 19:03:41.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":0,"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},{"value":0,"_account_id":27442,"name":"Quentin GROLLEAU","email":"quentin.grolleau@corp.ovh.com","username":"QG"},{"value":0,"_account_id":16707,"name":"Mohankumar","email":"nmohankumar1011@gmail.com","username":"mohankumar_n"},{"value":0,"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":0,"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":32755,"name":"Christian Rohmann","email":"christian.rohmann@inovex.de","username":"frittentheke"},{"_account_id":35916,"name":"Pham Le Gia Dai","display_name":"Dai","email":"daipham.3213@gmail.com","username":"daiplg","status":"Full Stack Dev"}],"REVIEWER":[{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"_account_id":16707,"name":"Mohankumar","email":"nmohankumar1011@gmail.com","username":"mohankumar_n"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":27442,"name":"Quentin GROLLEAU","email":"quentin.grolleau@corp.ovh.com","username":"QG"},{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},{"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"},{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-06-07 13:53:19.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-06-07 16:13:25.000000000","updated_by":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"reviewer":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"state":"REVIEWER"},{"updated":"2023-06-08 14:55:43.000000000","updated_by":{"_account_id":27442,"name":"Quentin GROLLEAU","email":"quentin.grolleau@corp.ovh.com","username":"QG"},"reviewer":{"_account_id":27442,"name":"Quentin GROLLEAU","email":"quentin.grolleau@corp.ovh.com","username":"QG"},"state":"REVIEWER"},{"updated":"2023-06-09 10:10:08.000000000","updated_by":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"reviewer":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"state":"CC"},{"updated":"2023-08-01 16:00:24.000000000","updated_by":{"_account_id":16707,"name":"Mohankumar","email":"nmohankumar1011@gmail.com","username":"mohankumar_n"},"reviewer":{"_account_id":16707,"name":"Mohankumar","email":"nmohankumar1011@gmail.com","username":"mohankumar_n"},"state":"REVIEWER"},{"updated":"2023-08-02 20:15:09.000000000","updated_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"reviewer":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"state":"REVIEWER"},{"updated":"2023-10-25 15:12:41.000000000","updated_by":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"reviewer":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"state":"CC"},{"updated":"2023-11-03 14:12:53.000000000","updated_by":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"reviewer":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"state":"REVIEWER"},{"updated":"2023-11-22 11:58:32.000000000","updated_by":{"_account_id":32755,"name":"Christian Rohmann","email":"christian.rohmann@inovex.de","username":"frittentheke"},"reviewer":{"_account_id":32755,"name":"Christian Rohmann","email":"christian.rohmann@inovex.de","username":"frittentheke"},"state":"CC"},{"updated":"2023-11-27 13:44:04.000000000","updated_by":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"reviewer":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"state":"REVIEWER"},{"updated":"2023-11-29 06:07:30.000000000","updated_by":{"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"},"reviewer":{"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"},"state":"REVIEWER"},{"updated":"2024-05-23 07:53:46.000000000","updated_by":{"_account_id":35916,"name":"Pham Le Gia Dai","display_name":"Dai","email":"daipham.3213@gmail.com","username":"daiplg","status":"Full Stack Dev"},"reviewer":{"_account_id":35916,"name":"Pham Le Gia Dai","display_name":"Dai","email":"daipham.3213@gmail.com","username":"daiplg","status":"Full Stack Dev"},"state":"CC"}],"messages":[{"id":"0e2f7324f6cf8beb349faf450b4540f28917a0f2","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":33408,"name":"Pierre-Yves Jourel","email":"pierre-yves.jourel@ovhcloud.com","username":"PYJ"},"date":"2023-06-07 13:39:55.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"f55641bfd3d8e5d66eb4b77a7933003bb72bcc1c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-07 13:53:19.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/de69719fa78a47f1ad021f9956c2e52b\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/84403a6211ce4d1db4159e855f60bc2e : FAILURE in 4m 03s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e256cc51b4954ac6a541b5b90da94303 : FAILURE in 6m 16s\n\nWarning:\n  Job openstack-tox-docs: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path doc/source/contributor/specs/version1.1/resize.rst in the commit\u0027\n  Comments left for invalid file doc/source/contributor/specs/version1.1/resize.rst","accounts_in_message":[],"_revision_number":1},{"id":"5598c4f2c7b7db58a26d398a8445915cadc399c8","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":33408,"name":"Pierre-Yves Jourel","email":"pierre-yves.jourel@ovhcloud.com","username":"PYJ"},"date":"2023-06-07 14:38:08.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"f522bddf78aeb25fd6f46e121afdb2bd80e6a39c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-07 14:49:05.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/53926da1543643ceaded448c8b5e30da\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8262fa2359e84c15b4f25a8da54b9083 : SUCCESS in 5m 03s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8d580ac658014e56ae3f3a4b98c47d49 : SUCCESS in 10m 04s","accounts_in_message":[],"_revision_number":2},{"id":"7c6bfcab4efcd546bfaf5312133a240a84cc9106","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2023-06-07 19:36:12.000000000","message":"Patch Set 2:\n\n(3 comments)","accounts_in_message":[],"_revision_number":2},{"id":"ae41df2dc315c33907ba14266e989baf6f1d6bfb","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2023-06-08 09:53:49.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"127077992ff5eff89d74d35edd36a2d3ae367a43","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2023-06-09 10:10:08.000000000","message":"Patch Set 2:\n\n(3 comments)","accounts_in_message":[],"_revision_number":2},{"id":"32445f1b340cffdf445e9a5005708ca8032df9cd","author":{"_account_id":33408,"name":"Pierre-Yves Jourel","email":"pierre-yves.jourel@ovhcloud.com","username":"PYJ"},"date":"2023-06-09 12:48:18.000000000","message":"Patch Set 2:\n\n(5 comments)","accounts_in_message":[],"_revision_number":2},{"id":"9f7c00403959195f507253bb2c144a6ba30bc1ca","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":33408,"name":"Pierre-Yves Jourel","email":"pierre-yves.jourel@ovhcloud.com","username":"PYJ"},"date":"2023-06-09 14:59:39.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":3},{"id":"3d8309294fa6fbeb866dcb33218a28010d095b7b","author":{"_account_id":33408,"name":"Pierre-Yves Jourel","email":"pierre-yves.jourel@ovhcloud.com","username":"PYJ"},"date":"2023-06-09 15:00:38.000000000","message":"Patch Set 3:\n\n(4 comments)","accounts_in_message":[],"_revision_number":3},{"id":"acfa85db8f894a53557b412035850415f90e02bd","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-09 15:13:29.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7ec69abb90d746d692feb22695e56484\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e429702d8fd04a24b0dd3fe4b36ae7a2 : SUCCESS in 5m 40s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/37d368526c8b4e0b842be3d4c5949cca : SUCCESS in 11m 53s","accounts_in_message":[],"_revision_number":3},{"id":"97955a11359fe3d5277b2113e748eb4caf48ac9e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":33408,"name":"Pierre-Yves Jourel","email":"pierre-yves.jourel@ovhcloud.com","username":"PYJ"},"date":"2023-06-21 13:45:30.000000000","message":"Uploaded patch set 4.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":4},{"id":"012ea3493cee82ac24928cf348034023c1491e50","author":{"_account_id":33408,"name":"Pierre-Yves Jourel","email":"pierre-yves.jourel@ovhcloud.com","username":"PYJ"},"date":"2023-06-21 13:45:55.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"62f5806b7ddb0a49eb15ae813a4cb30c9685a071","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-21 15:44:40.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/256791e4d20b4d0782f50295d6ce2822\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/192c0a687209493a9b657815ce998c16 : SUCCESS in 5m 44s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c9ef21636e6c46dfa2b490c1b68c69b0 : SUCCESS in 12m 13s","accounts_in_message":[],"_revision_number":4},{"id":"1669dd754bb0e2e1e96fb7a6c4abcdc516129c34","author":{"_account_id":27442,"name":"Quentin GROLLEAU","email":"quentin.grolleau@corp.ovh.com","username":"QG"},"date":"2023-06-21 16:13:54.000000000","message":"Patch Set 4: Code-Review+1","accounts_in_message":[],"_revision_number":4},{"id":"796dd8c9accbd95e750692f3cf6f8c0268249760","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2023-06-22 13:45:13.000000000","message":"Patch Set 4:\n\n(2 comments)","accounts_in_message":[],"_revision_number":4},{"id":"207e3789b6a432fd32e295da233b9a1b2da0efac","author":{"_account_id":33408,"name":"Pierre-Yves Jourel","email":"pierre-yves.jourel@ovhcloud.com","username":"PYJ"},"date":"2023-06-22 13:52:51.000000000","message":"Patch Set 4:\n\n(2 comments)","accounts_in_message":[],"_revision_number":4},{"id":"356e92dd71789954a5998c60cc458033c3531ee9","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2023-06-22 13:55:07.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"9538aa1da5a7e39e756fcedd215b2c1e19436006","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2023-06-27 10:54:07.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"822413999d342619cc31a2af9bf8eda14ab571f9","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":33408,"name":"Pierre-Yves Jourel","email":"pierre-yves.jourel@ovhcloud.com","username":"PYJ"},"date":"2023-06-29 12:51:41.000000000","message":"Uploaded patch set 5.\n\nOutdated Votes:\n* Code-Review+1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":5},{"id":"1805c16b2814b09b8f1d1619304aa21c7de6daf0","author":{"_account_id":33408,"name":"Pierre-Yves Jourel","email":"pierre-yves.jourel@ovhcloud.com","username":"PYJ"},"date":"2023-06-29 12:52:50.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"dc0641ac384b645b017cc09e74acc100c9688296","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":33408,"name":"Pierre-Yves Jourel","email":"pierre-yves.jourel@ovhcloud.com","username":"PYJ"},"date":"2023-06-29 12:57:54.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"c85970b056637312378f391fd8e0b72d43621604","author":{"_account_id":33408,"name":"Pierre-Yves Jourel","email":"pierre-yves.jourel@ovhcloud.com","username":"PYJ"},"date":"2023-06-29 12:58:34.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"cc63e03e1e8e0907b7d31677ede3110497a37ce2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-29 13:11:01.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/85d13e38403540c59ddb9d892fe38ec1\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/18132bcc13fa4c8b99c8152b62186bbb : SUCCESS in 4m 44s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b7783809ce314061a4ea77212eb819c1 : SUCCESS in 10m 36s","accounts_in_message":[],"_revision_number":6},{"id":"47545e2c9898e589003929babb4e2b0bb8763a96","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2023-07-07 13:54:10.000000000","message":"Patch Set 6: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"d124e4ba348f750b8c0f1109054c6f202ca421a9","author":{"_account_id":27442,"name":"Quentin GROLLEAU","email":"quentin.grolleau@corp.ovh.com","username":"QG"},"date":"2023-07-07 15:12:24.000000000","message":"Patch Set 6: Code-Review+1","accounts_in_message":[],"_revision_number":6},{"id":"f7212efa94ada79f440d123b5e1a002258d011a6","author":{"_account_id":16707,"name":"Mohankumar","email":"nmohankumar1011@gmail.com","username":"mohankumar_n"},"date":"2023-08-01 16:00:24.000000000","message":"Patch Set 6: Code-Review+1","accounts_in_message":[],"_revision_number":6},{"id":"662df7893c7f8f759ee4f08d4a442a2caca8b51d","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2023-08-02 20:15:09.000000000","message":"Patch Set 6: Code-Review-1\n\n(11 comments)","accounts_in_message":[],"_revision_number":6},{"id":"036e4a4b989e7c793ecf3f236c3454008922940a","author":{"_account_id":33408,"name":"Pierre-Yves Jourel","email":"pierre-yves.jourel@ovhcloud.com","username":"PYJ"},"date":"2023-08-11 13:05:04.000000000","message":"Patch Set 6:\n\n(11 comments)","accounts_in_message":[],"_revision_number":6},{"id":"ec4b39793382c0e0619fa767e8e1841c53d634ea","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":33408,"name":"Pierre-Yves Jourel","email":"pierre-yves.jourel@ovhcloud.com","username":"PYJ"},"date":"2023-08-11 13:21:13.000000000","message":"Uploaded patch set 7.\n\nOutdated Votes:\n* Code-Review+1, Code-Review+2, Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":7},{"id":"1b0d73c82b5ba246683034d893447a33f78029ca","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-11 13:34:21.000000000","message":"Patch Set 7: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/7f4df81c27ad4401b65a64f8f19931fb\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/aa16fa474e564d8fb633f610e9e8f56c : FAILURE in 4m 40s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/04591d507f594831ab02095cb10e3899 : SUCCESS in 11m 26s","accounts_in_message":[],"_revision_number":7},{"id":"e2433241de2c9bacdec756989f727ed90d8fc22a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":33408,"name":"Pierre-Yves Jourel","email":"pierre-yves.jourel@ovhcloud.com","username":"PYJ"},"date":"2023-08-14 14:04:29.000000000","message":"Uploaded patch set 8.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":8},{"id":"a4da0c474fcb0856e014ad081cb346f80032df22","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-14 14:15:00.000000000","message":"Patch Set 8: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/53316d14abf94fb094b34ed390c52850\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/51f305e0c0af4f6bac4b2aa758d9036b : SUCCESS in 4m 39s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c4b30fa5d3e24405a9abfdcfdbee995a : SUCCESS in 9m 13s","accounts_in_message":[],"_revision_number":8},{"id":"440406310b7bc8a345d702d59f773173e774a8fd","author":{"_account_id":27442,"name":"Quentin GROLLEAU","email":"quentin.grolleau@corp.ovh.com","username":"QG"},"date":"2023-10-04 12:21:06.000000000","message":"Patch Set 8: Code-Review+1","accounts_in_message":[],"_revision_number":8},{"id":"03989b3058f6406957e8e9853c8e6806fe7dcf60","author":{"_account_id":33408,"name":"Pierre-Yves Jourel","email":"pierre-yves.jourel@ovhcloud.com","username":"PYJ"},"date":"2023-10-04 12:21:13.000000000","message":"Patch Set 8:\n\n(11 comments)","accounts_in_message":[],"_revision_number":8},{"id":"77b0cf4164fb2356ddd4a64e9ba9f116c268bcca","author":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"date":"2023-10-25 15:12:41.000000000","message":"Patch Set 8:\n\n(1 comment)","accounts_in_message":[],"_revision_number":8},{"id":"ed9d88d28ebffc478fe2ec4620c9e35df5308da9","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":33408,"name":"Pierre-Yves Jourel","email":"pierre-yves.jourel@ovhcloud.com","username":"PYJ"},"date":"2023-11-03 14:03:55.000000000","message":"Uploaded patch set 9.\n\nOutdated Votes:\n* Code-Review+1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":9},{"id":"91ff6f28106c708671dd6692dd4298e0b80b8e85","author":{"_account_id":33408,"name":"Pierre-Yves Jourel","email":"pierre-yves.jourel@ovhcloud.com","username":"PYJ"},"date":"2023-11-03 14:04:10.000000000","message":"Patch Set 9:\n\n(1 comment)","accounts_in_message":[],"_revision_number":9},{"id":"49d281bca8dcfb1ef8a511101d041a6c1b5b64cd","author":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"date":"2023-11-03 14:12:53.000000000","message":"Patch Set 9: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":9},{"id":"eaf4331a37cd9fb5b542b2ea5244c499cae5c1f3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-03 14:21:03.000000000","message":"Patch Set 9: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/dfee8f383cb046d7b06323f4d515cb6a\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ca5f234d05004d9a93115986018d14b2 : SUCCESS in 5m 09s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5772c12d09414385871acb9db448ad94 : SUCCESS in 11m 56s","accounts_in_message":[],"_revision_number":9},{"id":"4ef14881e2b656432c443f19c72e09481bef57ef","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":33408,"name":"Pierre-Yves Jourel","email":"pierre-yves.jourel@ovhcloud.com","username":"PYJ"},"date":"2023-11-03 15:23:19.000000000","message":"Uploaded patch set 10.\n\nOutdated Votes:\n* Code-Review+1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":10},{"id":"4b18e79cf9ec08ea7576b7cbf7808f1574826a93","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-03 15:42:43.000000000","message":"Patch Set 10: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ec2c5c13b9b846c6af0c0fabacb02955\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5bf9004098e34819a03a2de3151faee5 : SUCCESS in 5m 47s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/dea3d839a8a24fb98d67c0598e306841 : SUCCESS in 10m 55s","accounts_in_message":[],"_revision_number":10},{"id":"3fa4847af10e66cd1c55e74582d9648ed466a5ef","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2023-11-17 09:40:18.000000000","message":"Patch Set 10:\n\n(1 comment)","accounts_in_message":[],"_revision_number":10},{"id":"ca716cd0cb9c9dbc22a6f6cac74c3c2bec92dc9f","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2023-11-17 09:42:03.000000000","message":"Patch Set 10:\n\n(1 comment)","accounts_in_message":[],"_revision_number":10},{"id":"c2bf02f0196fda4ba5e2096d0758e48e61c93575","author":{"_account_id":33408,"name":"Pierre-Yves Jourel","email":"pierre-yves.jourel@ovhcloud.com","username":"PYJ"},"date":"2023-11-22 14:48:56.000000000","message":"Patch Set 10:\n\n(1 comment)","accounts_in_message":[],"_revision_number":10},{"id":"205e9b80610226c2b35199c741c2f84b41c10101","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2023-11-27 13:44:04.000000000","message":"Patch Set 10: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":10},{"id":"3deaaf397281f58fd3a6d0db912b5ea97a0ce4c9","author":{"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"},"date":"2023-11-29 06:07:30.000000000","message":"Patch Set 10: Code-Review+1","accounts_in_message":[],"_revision_number":10},{"id":"f1b808a83b26b49fde07054cc0b055c99ef0e1d9","author":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"date":"2023-11-29 14:09:10.000000000","message":"Patch Set 10: Code-Review+1","accounts_in_message":[],"_revision_number":10},{"id":"696d3d30a64911cc714ea7babb0fe0415b9322c5","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2024-02-06 21:13:52.000000000","message":"Patch Set 10: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":10},{"id":"d86f2186135f20a7497549cd730433fede061043","author":{"_account_id":33408,"name":"Pierre-Yves Jourel","email":"pierre-yves.jourel@ovhcloud.com","username":"PYJ"},"date":"2024-02-07 14:14:26.000000000","message":"Patch Set 10:\n\n(1 comment)","accounts_in_message":[],"_revision_number":10},{"id":"3c41f70843b5b42b66b9a5352878348cbc3e8110","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":33408,"name":"Pierre-Yves Jourel","email":"pierre-yves.jourel@ovhcloud.com","username":"PYJ"},"date":"2024-02-08 14:03:40.000000000","message":"Uploaded patch set 11.\n\nOutdated Votes:\n* Code-Review+1, Code-Review+2, Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":11},{"id":"ed11ef7b83f2a892e1c3d971e8cc67bcc8368bd5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-08 14:25:02.000000000","message":"Patch Set 11: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/8539c51224f746349ae098a895c3c22f\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8521c95798f54da5b164d12e445a3a9b : FAILURE in 4m 05s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a9113b1a00ad446b9b88ef902acf2ae0 : SUCCESS in 12m 00s","accounts_in_message":[],"_revision_number":11},{"id":"a10e18886944d0ee23a916609e107f97b02bb49e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":33408,"name":"Pierre-Yves Jourel","email":"pierre-yves.jourel@ovhcloud.com","username":"PYJ"},"date":"2024-03-13 17:21:35.000000000","message":"Uploaded patch set 12.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":12},{"id":"b54d16143245463f63adb8412cf39dac5d01db3d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-13 17:32:27.000000000","message":"Patch Set 12: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/5f9634a23e4a4b6ab3bc4481219c006f\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/34ac2baa52c5481983cc7fb4646701b4 : FAILURE in 4m 18s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/13367a9d9ce3486da407f2cfde77e0e6 : SUCCESS in 9m 55s","accounts_in_message":[],"_revision_number":12},{"id":"e10407845e45b7517418063079705e6bc3cce37b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":33408,"name":"Pierre-Yves Jourel","email":"pierre-yves.jourel@ovhcloud.com","username":"PYJ"},"date":"2024-03-13 17:41:27.000000000","message":"Uploaded patch set 13.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":13},{"id":"1d05ad22eed6d8d9eca779b86cc5b66754216747","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-13 17:57:39.000000000","message":"Patch Set 13: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/561581fb237f4d14aa83a67fae91489c\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/479a8ad54fad4f6ebd89abcc0f51d8d2 : SUCCESS in 5m 50s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/91aefd77dd26442793a1f2bbf49e61c0 : SUCCESS in 14m 32s","accounts_in_message":[],"_revision_number":13},{"id":"5e1dae1e5d994f725ae112118352c1a491f4785e","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2024-04-10 09:13:24.000000000","message":"Patch Set 13: Code-Review+2","accounts_in_message":[],"_revision_number":13},{"id":"e78fa2df4443c28e37e1a9d3efbf48fa11158c42","author":{"_account_id":27442,"name":"Quentin GROLLEAU","email":"quentin.grolleau@corp.ovh.com","username":"QG"},"date":"2024-05-20 13:39:43.000000000","message":"Patch Set 13: Code-Review+1","accounts_in_message":[],"_revision_number":13},{"id":"5be0e72925f793732d2e9d9bfbe9a18f9dc2b1d9","author":{"_account_id":35916,"name":"Pham Le Gia Dai","display_name":"Dai","email":"daipham.3213@gmail.com","username":"daiplg","status":"Full Stack Dev"},"date":"2024-05-23 07:53:46.000000000","message":"Patch Set 13:\n\n(1 comment)","accounts_in_message":[],"_revision_number":13},{"id":"1274446f9fc9d79c2f49318f72f0d2574595ec6b","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2024-05-28 07:58:13.000000000","message":"Patch Set 13:\n\n(1 comment)","accounts_in_message":[],"_revision_number":13},{"id":"528588407a1920c3d08a193c73fd1a50841d280e","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2024-05-28 08:06:14.000000000","message":"Patch Set 13:\n\n(1 comment)","accounts_in_message":[],"_revision_number":13},{"id":"1b7d072bbe09a75811a4b86718269915edcdd051","author":{"_account_id":35916,"name":"Pham Le Gia Dai","display_name":"Dai","email":"daipham.3213@gmail.com","username":"daiplg","status":"Full Stack Dev"},"date":"2024-05-29 08:56:23.000000000","message":"Patch Set 13:\n\n(1 comment)","accounts_in_message":[],"_revision_number":13},{"id":"e381daf138b14f3781768fd017cb4dd79d49b240","author":{"_account_id":33408,"name":"Pierre-Yves Jourel","email":"pierre-yves.jourel@ovhcloud.com","username":"PYJ"},"date":"2024-05-30 12:38:42.000000000","message":"Patch Set 13:\n\n(1 comment)","accounts_in_message":[],"_revision_number":13},{"id":"22c710c3175814b3657e0382826688343f017a13","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2024-07-02 10:43:10.000000000","message":"Patch Set 13: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":13},{"id":"fbd102ff0ee6cf2ccc28520ba79559f610b26310","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":33408,"name":"Pierre-Yves Jourel","email":"pierre-yves.jourel@ovhcloud.com","username":"PYJ"},"date":"2024-07-02 12:37:25.000000000","message":"Uploaded patch set 14.\n\nOutdated Votes:\n* Code-Review+1, Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":14},{"id":"04d0d60c5b62c37a540b75f50a0dec8a6b366bbd","author":{"_account_id":33408,"name":"Pierre-Yves Jourel","email":"pierre-yves.jourel@ovhcloud.com","username":"PYJ"},"date":"2024-07-02 12:41:59.000000000","message":"Patch Set 14:\n\n(1 comment)","accounts_in_message":[],"_revision_number":14},{"id":"b9b8336bf21ab17910bc8b6abe2c342c31fbcf42","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-02 12:45:04.000000000","message":"Patch Set 14: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/9cb78380278148b2a99e4c89d3d81de3\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a3cbac244239405e9d6864a7eb12ae13 : SUCCESS in 5m 47s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2515aead52df496ea1afbd0d8acfe854 : FAILURE in 5m 19s","accounts_in_message":[],"_revision_number":14},{"id":"b94629eb7ab2e47ac241a801d99267f1edf2a11d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":33408,"name":"Pierre-Yves Jourel","email":"pierre-yves.jourel@ovhcloud.com","username":"PYJ"},"date":"2024-07-02 13:20:38.000000000","message":"Uploaded patch set 15.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":15},{"id":"c3cd3c7173471ee308322f771af81dae42ba7c78","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-02 13:28:09.000000000","message":"Patch Set 15: Verified-1\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/711797bd9e864b55aeb1763b825b74c6\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/86a2a9c6380743f1a280c5a5a0ea8673 : SUCCESS in 4m 52s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4b6c06a0832d490f89a40f6e573353db : FAILURE in 5m 17s","accounts_in_message":[],"_revision_number":15},{"id":"d3ec30efb4679b59222b0ac11f5d988f5555fb5e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":33408,"name":"Pierre-Yves Jourel","email":"pierre-yves.jourel@ovhcloud.com","username":"PYJ"},"date":"2024-07-02 13:33:21.000000000","message":"Uploaded patch set 16.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":16},{"id":"cadad28dfb6b888cee19524a2d438a842ec5a496","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-02 13:48:10.000000000","message":"Patch Set 16: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3da11236d61940c982ac086f3ddfdd1f\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fabd67d6a0b2422790b3b94c3d3dcf17 : SUCCESS in 5m 28s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2ce2f37d24f9444b91272870544f7319 : SUCCESS in 14m 26s","accounts_in_message":[],"_revision_number":16},{"id":"f845b6b047d520a3191d3eb44dfbb416ae8fde8b","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2024-07-03 07:48:09.000000000","message":"Patch Set 16: Code-Review+2","accounts_in_message":[],"_revision_number":16},{"id":"c5a3bdedb8f890ce6bff0b050f81d64bddf21191","author":{"_account_id":27442,"name":"Quentin GROLLEAU","email":"quentin.grolleau@corp.ovh.com","username":"QG"},"date":"2024-07-03 12:25:33.000000000","message":"Patch Set 16: Code-Review+1","accounts_in_message":[],"_revision_number":16},{"id":"3661b3e8b8350c4bae00f1db635872b38182241e","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2024-08-01 19:03:41.000000000","message":"Patch Set 16: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":16},{"id":"597750ab80517e72cfc1228df81c84622f826a16","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-01 19:03:59.000000000","message":"Patch Set 16: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":16},{"id":"6f3fd02448f748c34d9240bd85fe17b1ccf31778","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-01 19:15:34.000000000","message":"Patch Set 16: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5896784eb1ff4112abfc324d634b4d64\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4c89ea77d7944d88bf18f00cfdaa238a : SUCCESS in 4m 55s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e5d6b64d83b548e29885a66f7f81073b : SUCCESS in 11m 30s","accounts_in_message":[],"_revision_number":16},{"id":"add072afa5f6d37ee66984fe106f4e148ce2617c","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-01 19:15:34.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":16},{"id":"b9db70081f716ce6755db38fda61db53011138c7","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-01 19:16:51.000000000","message":"Patch Set 16:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/43644a1cb9334547a5457803c666086a\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b9aeba30878940ecb4a44359d470dd90 : SUCCESS in 57s","accounts_in_message":[],"_revision_number":16}],"current_revision_number":16,"current_revision":"29855ddc694fa1060b228507ff2134f2b9b12416","revisions":{"ef4665b196cad60360f03bbdede1e651b2e2ce41":{"kind":"REWORK","_number":1,"created":"2023-06-07 13:39:55.000000000","uploader":{"_account_id":33408,"name":"Pierre-Yves Jourel","email":"pierre-yves.jourel@ovhcloud.com","username":"PYJ"},"ref":"refs/changes/90/885490/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/90/885490/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/90/885490/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/90/885490/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/90/885490/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/90/885490/1"}}},"commit":{"parents":[{"commit":"e2bc07222f40ce5c6ba73a0e4bf79af9f5ebde11","subject":"Merge \"Fix hm operating status to ONLINE in single lb call\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/e2bc07222f40ce5c6ba73a0e4bf79af9f5ebde11"}]}],"author":{"name":"Pierre-Yves Jourel","email":"pierre-yves.jourel@ovhcloud.com","date":"2023-06-07 13:36:00.000000000","tz":-240},"committer":{"name":"Pierre-Yves Jourel","email":"pierre-yves.jourel@ovhcloud.com","date":"2023-06-07 13:36:00.000000000","tz":-240},"subject":"Add specs to resize a load balancer","message":"Add specs to resize a load balancer\n\nThis specification explains how the resizing mechanism can be implemented.\n\nChange-Id: I9c3499b703bf0beab305cc071c74ae3acbcc4c24\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/ef4665b196cad60360f03bbdede1e651b2e2ce41"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/ef4665b196cad60360f03bbdede1e651b2e2ce41"}]},"branch":"refs/heads/master"},"d3a5b17d51d180cb8b1151e38615634594b310fd":{"kind":"REWORK","_number":2,"created":"2023-06-07 14:38:08.000000000","uploader":{"_account_id":33408,"name":"Pierre-Yves Jourel","email":"pierre-yves.jourel@ovhcloud.com","username":"PYJ"},"ref":"refs/changes/90/885490/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/90/885490/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/90/885490/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/90/885490/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/90/885490/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/90/885490/2"}}},"commit":{"parents":[{"commit":"e2bc07222f40ce5c6ba73a0e4bf79af9f5ebde11","subject":"Merge \"Fix hm operating status to ONLINE in single lb call\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/e2bc07222f40ce5c6ba73a0e4bf79af9f5ebde11"}]}],"author":{"name":"Pierre-Yves Jourel","email":"pierre-yves.jourel@ovhcloud.com","date":"2023-06-07 13:36:00.000000000","tz":-240},"committer":{"name":"Pierre-Yves Jourel","email":"pierre-yves.jourel@ovhcloud.com","date":"2023-06-07 14:38:06.000000000","tz":-240},"subject":"Add specs to resize a load balancer","message":"Add specs to resize a load balancer\n\nThis specification explains how the resizing mechanism can be implemented.\n\nChange-Id: I9c3499b703bf0beab305cc071c74ae3acbcc4c24\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/d3a5b17d51d180cb8b1151e38615634594b310fd"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/d3a5b17d51d180cb8b1151e38615634594b310fd"}]},"branch":"refs/heads/master"},"094ebc1b6f73e0fcd6a145db911b436405ef8253":{"kind":"REWORK","_number":3,"created":"2023-06-09 14:59:39.000000000","uploader":{"_account_id":33408,"name":"Pierre-Yves Jourel","email":"pierre-yves.jourel@ovhcloud.com","username":"PYJ"},"ref":"refs/changes/90/885490/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/90/885490/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/90/885490/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/90/885490/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/90/885490/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/90/885490/3"}}},"commit":{"parents":[{"commit":"e2bc07222f40ce5c6ba73a0e4bf79af9f5ebde11","subject":"Merge \"Fix hm operating status to ONLINE in single lb call\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/e2bc07222f40ce5c6ba73a0e4bf79af9f5ebde11"}]}],"author":{"name":"Pierre-Yves Jourel","email":"pierre-yves.jourel@ovhcloud.com","date":"2023-06-07 13:36:00.000000000","tz":-240},"committer":{"name":"Pierre-Yves Jourel","email":"pierre-yves.jourel@ovhcloud.com","date":"2023-06-09 14:59:36.000000000","tz":-240},"subject":"Add specs to resize a load balancer","message":"Add specs to resize a load balancer\n\nThis specification explains how the resizing mechanism can be implemented.\n\nChange-Id: I9c3499b703bf0beab305cc071c74ae3acbcc4c24\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/094ebc1b6f73e0fcd6a145db911b436405ef8253"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/094ebc1b6f73e0fcd6a145db911b436405ef8253"}]},"branch":"refs/heads/master"},"d2f63af4bfcc0e6edf684f3a9880acf0aa01e14e":{"kind":"REWORK","_number":4,"created":"2023-06-21 13:45:30.000000000","uploader":{"_account_id":33408,"name":"Pierre-Yves Jourel","email":"pierre-yves.jourel@ovhcloud.com","username":"PYJ"},"ref":"refs/changes/90/885490/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/90/885490/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/90/885490/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/90/885490/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/90/885490/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/90/885490/4"}}},"commit":{"parents":[{"commit":"e2bc07222f40ce5c6ba73a0e4bf79af9f5ebde11","subject":"Merge \"Fix hm operating status to ONLINE in single lb call\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/e2bc07222f40ce5c6ba73a0e4bf79af9f5ebde11"}]}],"author":{"name":"Pierre-Yves Jourel","email":"pierre-yves.jourel@ovhcloud.com","date":"2023-06-07 13:36:00.000000000","tz":-240},"committer":{"name":"Pierre-Yves Jourel","email":"pierre-yves.jourel@ovhcloud.com","date":"2023-06-21 13:45:28.000000000","tz":-240},"subject":"Add specs to resize a load balancer","message":"Add specs to resize a load balancer\n\nThis specification explains how the resizing mechanism can be implemented.\n\nChange-Id: I9c3499b703bf0beab305cc071c74ae3acbcc4c24\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/d2f63af4bfcc0e6edf684f3a9880acf0aa01e14e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/d2f63af4bfcc0e6edf684f3a9880acf0aa01e14e"}]},"branch":"refs/heads/master"},"0c95fde2edef8a05057d44b1f80c320f0f02d7e4":{"kind":"REWORK","_number":5,"created":"2023-06-29 12:51:41.000000000","uploader":{"_account_id":33408,"name":"Pierre-Yves Jourel","email":"pierre-yves.jourel@ovhcloud.com","username":"PYJ"},"ref":"refs/changes/90/885490/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/90/885490/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/90/885490/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/90/885490/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/90/885490/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/90/885490/5"}}},"commit":{"parents":[{"commit":"e2bc07222f40ce5c6ba73a0e4bf79af9f5ebde11","subject":"Merge \"Fix hm operating status to ONLINE in single lb call\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/e2bc07222f40ce5c6ba73a0e4bf79af9f5ebde11"}]}],"author":{"name":"Pierre-Yves Jourel","email":"pierre-yves.jourel@ovhcloud.com","date":"2023-06-07 13:36:00.000000000","tz":-240},"committer":{"name":"Pierre-Yves Jourel","email":"pierre-yves.jourel@ovhcloud.com","date":"2023-06-29 12:51:35.000000000","tz":-240},"subject":"Add specs to resize a load balancer","message":"Add specs to resize a load balancer\n\nThis specification explains how the resizing mechanism can be implemented.\n\nChange-Id: I9c3499b703bf0beab305cc071c74ae3acbcc4c24\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/0c95fde2edef8a05057d44b1f80c320f0f02d7e4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/0c95fde2edef8a05057d44b1f80c320f0f02d7e4"}]},"branch":"refs/heads/master"},"d8cae3913163bd641fe2853582feb8bd1ecf3c4a":{"kind":"REWORK","_number":6,"created":"2023-06-29 12:57:54.000000000","uploader":{"_account_id":33408,"name":"Pierre-Yves Jourel","email":"pierre-yves.jourel@ovhcloud.com","username":"PYJ"},"ref":"refs/changes/90/885490/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/90/885490/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/90/885490/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/90/885490/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/90/885490/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/90/885490/6"}}},"commit":{"parents":[{"commit":"e2bc07222f40ce5c6ba73a0e4bf79af9f5ebde11","subject":"Merge \"Fix hm operating status to ONLINE in single lb call\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/e2bc07222f40ce5c6ba73a0e4bf79af9f5ebde11"}]}],"author":{"name":"Pierre-Yves Jourel","email":"pierre-yves.jourel@ovhcloud.com","date":"2023-06-07 13:36:00.000000000","tz":-240},"committer":{"name":"Pierre-Yves Jourel","email":"pierre-yves.jourel@ovhcloud.com","date":"2023-06-29 12:57:49.000000000","tz":-240},"subject":"Add specs to resize a load balancer","message":"Add specs to resize a load balancer\n\nThis specification explains how the resizing mechanism can be implemented.\n\nChange-Id: I9c3499b703bf0beab305cc071c74ae3acbcc4c24\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/d8cae3913163bd641fe2853582feb8bd1ecf3c4a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/d8cae3913163bd641fe2853582feb8bd1ecf3c4a"}]},"branch":"refs/heads/master"},"f9ff9fb1e928124597f8d04b0f22586590268333":{"kind":"REWORK","_number":7,"created":"2023-08-11 13:21:13.000000000","uploader":{"_account_id":33408,"name":"Pierre-Yves Jourel","email":"pierre-yves.jourel@ovhcloud.com","username":"PYJ"},"ref":"refs/changes/90/885490/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/90/885490/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/90/885490/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/90/885490/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/90/885490/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/90/885490/7"}}},"commit":{"parents":[{"commit":"e2bc07222f40ce5c6ba73a0e4bf79af9f5ebde11","subject":"Merge \"Fix hm operating status to ONLINE in single lb call\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/e2bc07222f40ce5c6ba73a0e4bf79af9f5ebde11"}]}],"author":{"name":"Pierre-Yves Jourel","email":"pierre-yves.jourel@ovhcloud.com","date":"2023-06-07 13:36:00.000000000","tz":-240},"committer":{"name":"Pierre-Yves Jourel","email":"pierre-yves.jourel@ovhcloud.com","date":"2023-08-11 13:21:14.000000000","tz":-240},"subject":"Add specs to resize a load balancer","message":"Add specs to resize a load balancer\n\nThis specification explains how the resizing mechanism can be implemented.\n\nChange-Id: I9c3499b703bf0beab305cc071c74ae3acbcc4c24\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/f9ff9fb1e928124597f8d04b0f22586590268333"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/f9ff9fb1e928124597f8d04b0f22586590268333"}]},"branch":"refs/heads/master"},"fecf51c6aaff3c7dc85b39c9789e3389ae2d3f4f":{"kind":"REWORK","_number":8,"created":"2023-08-14 14:04:29.000000000","uploader":{"_account_id":33408,"name":"Pierre-Yves Jourel","email":"pierre-yves.jourel@ovhcloud.com","username":"PYJ"},"ref":"refs/changes/90/885490/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/90/885490/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/90/885490/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/90/885490/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/90/885490/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/90/885490/8"}}},"commit":{"parents":[{"commit":"e2bc07222f40ce5c6ba73a0e4bf79af9f5ebde11","subject":"Merge \"Fix hm operating status to ONLINE in single lb call\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/e2bc07222f40ce5c6ba73a0e4bf79af9f5ebde11"}]}],"author":{"name":"Pierre-Yves Jourel","email":"pierre-yves.jourel@ovhcloud.com","date":"2023-06-07 13:36:00.000000000","tz":-240},"committer":{"name":"Pierre-Yves Jourel","email":"pierre-yves.jourel@ovhcloud.com","date":"2023-08-14 14:04:27.000000000","tz":-240},"subject":"Add specs to resize a load balancer","message":"Add specs to resize a load balancer\n\nThis specification explains how the resizing mechanism can be implemented.\n\nChange-Id: I9c3499b703bf0beab305cc071c74ae3acbcc4c24\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/fecf51c6aaff3c7dc85b39c9789e3389ae2d3f4f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/fecf51c6aaff3c7dc85b39c9789e3389ae2d3f4f"}]},"branch":"refs/heads/master"},"79bc44676d2c00c746333d8204088bff1125c9b3":{"kind":"REWORK","_number":9,"created":"2023-11-03 14:03:55.000000000","uploader":{"_account_id":33408,"name":"Pierre-Yves Jourel","email":"pierre-yves.jourel@ovhcloud.com","username":"PYJ"},"ref":"refs/changes/90/885490/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/90/885490/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/90/885490/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/90/885490/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/90/885490/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/90/885490/9"}}},"commit":{"parents":[{"commit":"e2bc07222f40ce5c6ba73a0e4bf79af9f5ebde11","subject":"Merge \"Fix hm operating status to ONLINE in single lb call\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/e2bc07222f40ce5c6ba73a0e4bf79af9f5ebde11"}]}],"author":{"name":"Pierre-Yves Jourel","email":"pierre-yves.jourel@ovhcloud.com","date":"2023-06-07 13:36:00.000000000","tz":-240},"committer":{"name":"Pierre-Yves Jourel","email":"pierre-yves.jourel@ovhcloud.com","date":"2023-11-03 14:03:50.000000000","tz":-240},"subject":"Add specs to resize a load balancer","message":"Add specs to resize a load balancer\n\nThis specification explains how the resizing mechanism can be implemented.\n\nChange-Id: I9c3499b703bf0beab305cc071c74ae3acbcc4c24\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/79bc44676d2c00c746333d8204088bff1125c9b3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/79bc44676d2c00c746333d8204088bff1125c9b3"}]},"branch":"refs/heads/master"},"b57745510bb3e910bc8173330da37d04af249d6f":{"kind":"REWORK","_number":10,"created":"2023-11-03 15:23:19.000000000","uploader":{"_account_id":33408,"name":"Pierre-Yves Jourel","email":"pierre-yves.jourel@ovhcloud.com","username":"PYJ"},"ref":"refs/changes/90/885490/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/90/885490/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/90/885490/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/90/885490/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/90/885490/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/90/885490/10"}}},"commit":{"parents":[{"commit":"e2bc07222f40ce5c6ba73a0e4bf79af9f5ebde11","subject":"Merge \"Fix hm operating status to ONLINE in single lb call\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/e2bc07222f40ce5c6ba73a0e4bf79af9f5ebde11"}]}],"author":{"name":"Pierre-Yves Jourel","email":"pierre-yves.jourel@ovhcloud.com","date":"2023-06-07 13:36:00.000000000","tz":-240},"committer":{"name":"Pierre-Yves Jourel","email":"pierre-yves.jourel@ovhcloud.com","date":"2023-11-03 15:23:15.000000000","tz":-240},"subject":"Add specs to resize a load balancer","message":"Add specs to resize a load balancer\n\nThis specification explains how the resizing mechanism can be implemented.\n\nChange-Id: I9c3499b703bf0beab305cc071c74ae3acbcc4c24\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/b57745510bb3e910bc8173330da37d04af249d6f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/b57745510bb3e910bc8173330da37d04af249d6f"}]},"branch":"refs/heads/master"},"591d71d4b01f7ef7d0f0909066b3896518c3ae32":{"kind":"REWORK","_number":11,"created":"2024-02-08 14:03:40.000000000","uploader":{"_account_id":33408,"name":"Pierre-Yves Jourel","email":"pierre-yves.jourel@ovhcloud.com","username":"PYJ"},"ref":"refs/changes/90/885490/11","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/90/885490/11","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/90/885490/11 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/90/885490/11 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/90/885490/11 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/90/885490/11"}}},"commit":{"parents":[{"commit":"e2bc07222f40ce5c6ba73a0e4bf79af9f5ebde11","subject":"Merge \"Fix hm operating status to ONLINE in single lb call\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/e2bc07222f40ce5c6ba73a0e4bf79af9f5ebde11"}]}],"author":{"name":"Pierre-Yves Jourel","email":"pierre-yves.jourel@ovhcloud.com","date":"2023-06-07 13:36:00.000000000","tz":-240},"committer":{"name":"Pierre-Yves Jourel","email":"pierre-yves.jourel@ovhcloud.com","date":"2024-02-08 14:02:56.000000000","tz":-300},"subject":"Add specs to resize a load balancer","message":"Add specs to resize a load balancer\n\nThis specification explains how the resizing mechanism can be implemented.\n\nChange-Id: I9c3499b703bf0beab305cc071c74ae3acbcc4c24\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/591d71d4b01f7ef7d0f0909066b3896518c3ae32"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/591d71d4b01f7ef7d0f0909066b3896518c3ae32"}]},"branch":"refs/heads/master"},"50351d65a3eedec3e47b32d2dbbabfedd9a089b7":{"kind":"REWORK","_number":12,"created":"2024-03-13 17:21:35.000000000","uploader":{"_account_id":33408,"name":"Pierre-Yves Jourel","email":"pierre-yves.jourel@ovhcloud.com","username":"PYJ"},"ref":"refs/changes/90/885490/12","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/90/885490/12","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/90/885490/12 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/90/885490/12 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/90/885490/12 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/90/885490/12"}}},"commit":{"parents":[{"commit":"e2bc07222f40ce5c6ba73a0e4bf79af9f5ebde11","subject":"Merge \"Fix hm operating status to ONLINE in single lb call\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/e2bc07222f40ce5c6ba73a0e4bf79af9f5ebde11"}]}],"author":{"name":"Pierre-Yves Jourel","email":"pierre-yves.jourel@ovhcloud.com","date":"2023-06-07 13:36:00.000000000","tz":-240},"committer":{"name":"Pierre-Yves Jourel","email":"pierre-yves.jourel@ovhcloud.com","date":"2024-03-13 17:21:24.000000000","tz":-240},"subject":"Add specs to resize a load balancer","message":"Add specs to resize a load balancer\n\nThis specification explains how the resizing mechanism can be implemented.\n\nChange-Id: I9c3499b703bf0beab305cc071c74ae3acbcc4c24\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/50351d65a3eedec3e47b32d2dbbabfedd9a089b7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/50351d65a3eedec3e47b32d2dbbabfedd9a089b7"}]},"branch":"refs/heads/master"},"6f22cba7c29d5d7912908ff5e73b804cc02a8e4e":{"kind":"REWORK","_number":13,"created":"2024-03-13 17:41:27.000000000","uploader":{"_account_id":33408,"name":"Pierre-Yves Jourel","email":"pierre-yves.jourel@ovhcloud.com","username":"PYJ"},"ref":"refs/changes/90/885490/13","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/90/885490/13","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/90/885490/13 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/90/885490/13 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/90/885490/13 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/90/885490/13"}}},"commit":{"parents":[{"commit":"e2bc07222f40ce5c6ba73a0e4bf79af9f5ebde11","subject":"Merge \"Fix hm operating status to ONLINE in single lb call\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/e2bc07222f40ce5c6ba73a0e4bf79af9f5ebde11"}]}],"author":{"name":"Pierre-Yves Jourel","email":"pierre-yves.jourel@ovhcloud.com","date":"2023-06-07 13:36:00.000000000","tz":-240},"committer":{"name":"Pierre-Yves Jourel","email":"pierre-yves.jourel@ovhcloud.com","date":"2024-03-13 17:41:24.000000000","tz":-240},"subject":"Add specs to resize a load balancer","message":"Add specs to resize a load balancer\n\nThis specification explains how the resizing mechanism can be implemented.\n\nChange-Id: I9c3499b703bf0beab305cc071c74ae3acbcc4c24\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/6f22cba7c29d5d7912908ff5e73b804cc02a8e4e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/6f22cba7c29d5d7912908ff5e73b804cc02a8e4e"}]},"branch":"refs/heads/master"},"15f16367940baca91b21d0003979cf230dbdf7cc":{"kind":"REWORK","_number":14,"created":"2024-07-02 12:37:25.000000000","uploader":{"_account_id":33408,"name":"Pierre-Yves Jourel","email":"pierre-yves.jourel@ovhcloud.com","username":"PYJ"},"ref":"refs/changes/90/885490/14","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/90/885490/14","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/90/885490/14 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/90/885490/14 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/90/885490/14 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/90/885490/14"}}},"commit":{"parents":[{"commit":"e2bc07222f40ce5c6ba73a0e4bf79af9f5ebde11","subject":"Merge \"Fix hm operating status to ONLINE in single lb call\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/e2bc07222f40ce5c6ba73a0e4bf79af9f5ebde11"}]}],"author":{"name":"Pierre-Yves Jourel","email":"pierre-yves.jourel@ovhcloud.com","date":"2023-06-07 13:36:00.000000000","tz":-240},"committer":{"name":"Pierre-Yves Jourel","email":"pierre-yves.jourel@ovhcloud.com","date":"2024-07-02 12:37:19.000000000","tz":-240},"subject":"Add specs to resize a load balancer","message":"Add specs to resize a load balancer\n\nThis specification explains how the resizing mechanism can be implemented.\n\nChange-Id: I9c3499b703bf0beab305cc071c74ae3acbcc4c24\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/15f16367940baca91b21d0003979cf230dbdf7cc"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/15f16367940baca91b21d0003979cf230dbdf7cc"}]},"branch":"refs/heads/master"},"8427c060b43ba55642250a4915f141fe8f906981":{"kind":"REWORK","_number":15,"created":"2024-07-02 13:20:38.000000000","uploader":{"_account_id":33408,"name":"Pierre-Yves Jourel","email":"pierre-yves.jourel@ovhcloud.com","username":"PYJ"},"ref":"refs/changes/90/885490/15","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/90/885490/15","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/90/885490/15 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/90/885490/15 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/90/885490/15 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/90/885490/15"}}},"commit":{"parents":[{"commit":"7376a4ad0c305113a2aabc091f1bb761a532b14d","subject":"Fix functional-py31{1,2} tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/7376a4ad0c305113a2aabc091f1bb761a532b14d"}]}],"author":{"name":"Pierre-Yves Jourel","email":"pierre-yves.jourel@ovhcloud.com","date":"2023-06-07 13:36:00.000000000","tz":-240},"committer":{"name":"Pierre-Yves Jourel","email":"pierre-yves.jourel@ovhcloud.com","date":"2024-07-02 13:20:34.000000000","tz":-240},"subject":"Add specs to resize a load balancer","message":"Add specs to resize a load balancer\n\nThis specification explains how the resizing mechanism can be implemented.\n\nChange-Id: I9c3499b703bf0beab305cc071c74ae3acbcc4c24\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/8427c060b43ba55642250a4915f141fe8f906981"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/8427c060b43ba55642250a4915f141fe8f906981"}]},"branch":"refs/heads/master"},"29855ddc694fa1060b228507ff2134f2b9b12416":{"kind":"REWORK","_number":16,"created":"2024-07-02 13:33:21.000000000","uploader":{"_account_id":33408,"name":"Pierre-Yves Jourel","email":"pierre-yves.jourel@ovhcloud.com","username":"PYJ"},"ref":"refs/changes/90/885490/16","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/90/885490/16","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/90/885490/16 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/90/885490/16 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/90/885490/16 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/90/885490/16"}}},"commit":{"parents":[{"commit":"7376a4ad0c305113a2aabc091f1bb761a532b14d","subject":"Fix functional-py31{1,2} tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/7376a4ad0c305113a2aabc091f1bb761a532b14d"}]}],"author":{"name":"Pierre-Yves Jourel","email":"pierre-yves.jourel@ovhcloud.com","date":"2023-06-07 13:36:00.000000000","tz":-240},"committer":{"name":"Pierre-Yves Jourel","email":"pierre-yves.jourel@ovhcloud.com","date":"2024-07-02 13:33:17.000000000","tz":-240},"subject":"Add specs to resize a load balancer","message":"Add specs to resize a load balancer\n\nThis specification explains how the resizing mechanism can be implemented.\n\nChange-Id: I9c3499b703bf0beab305cc071c74ae3acbcc4c24\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/29855ddc694fa1060b228507ff2134f2b9b12416"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/29855ddc694fa1060b228507ff2134f2b9b12416"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
