)]}'
{"id":"openstack%2Foctavia~885573","triplet_id":"openstack%2Foctavia~stable%2Fzed~I7f150c2c1c3da706055aba5e02b5ad7c032806f1","project":"openstack/octavia","branch":"stable/zed","attention_set":{},"removed_from_attention_set":{"29244":{"account":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"last_update":"2023-06-20 05:35:28.000000000","reason":"\u003cGERRIT_ACCOUNT_29244\u003e replied on the change","reason_account":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"}}},"hashtags":[],"change_id":"I7f150c2c1c3da706055aba5e02b5ad7c032806f1","subject":"Avoid interface name collisions in the amphora","status":"MERGED","created":"2023-06-08 08:26:23.000000000","updated":"2023-06-20 10:10:16.000000000","submitted":"2023-06-20 10:08:53.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"885573","meta_rev_id":"0ac29c04e17c19a2c1dcd8fa7b8a89eb7fdc2962","_number":885573,"virtual_id_number":885573,"owner":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-06-20 10:08:52.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"all":[{"value":2,"date":"2023-06-15 16:02:53.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2023-06-09 11:20:17.000000000","_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"all":[{"value":1,"date":"2023-06-15 16:02:53.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-06-08 11:32:40.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-06-09 11:20:17.000000000","updated_by":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"reviewer":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"state":"REVIEWER"},{"updated":"2023-06-15 16:02:53.000000000","updated_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"reviewer":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"state":"REVIEWER"}],"messages":[{"id":"83c73ae4efcff950724668b19a91338bf4d2446a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2023-06-08 08:26:23.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"fd7e736731b5a0defbee362b4aa0e3bbaa5c14aa","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-08 11:32:40.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e6b07363621c4467a95a490afe7a4114\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/d19bb694965d4508b18bbfcf211d2cb5 : SUCCESS in 9m 22s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/aaad6e78b2f24d17a8e90ee79017ab0c : SUCCESS in 6m 15s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/79a01337c9134c0993b7442db9549245 : SUCCESS in 4m 45s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/1d6cde7c45ce48b5853e04fb76aff300 : SUCCESS in 4m 45s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/aac5ac2d4357487cbe60cdc3eab604d7 : SUCCESS in 4m 11s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f268798d27704eb38033d5215c6161a5 : SUCCESS in 11m 44s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/d0178bab027d4a6583fe186e02b58f23 : SUCCESS in 4m 37s\n- octavia-tox-py39-tips https://zuul.opendev.org/t/openstack/build/ef82fd85355142838fc48a3b4baa32f4 : SUCCESS in 5m 16s\n- octavia-tox-functional-py39-tips https://zuul.opendev.org/t/openstack/build/320cffb68615434fa77a3aa1d5d43eef : SUCCESS in 7m 30s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/749ff5e39fee4976ab0d628be1861711 : SUCCESS in 4m 11s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/661570acd4c44c01affaae01ba4a9be7 : SUCCESS in 7m 19s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/797b1bb092bf40cabab2d11ee901c706 : SUCCESS in 6m 29s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/1ae0e0388dd245a4b5e55b65c708dcc4 : SUCCESS in 6m 16s (non-voting)\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/0d13f34eec7445b98977b849802c74f6 : SUCCESS in 1h 15m 45s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/f4800e9cc88245b983567b0b138f945c : SUCCESS in 1h 41m 14s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/af2dbdaa904c4f7f89e5f720e17a2690 : SUCCESS in 46m 27s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/a241db13372646aaa665b9e0cdcadc68 : SUCCESS in 37m 06s\n- octavia-v2-dsvm-tls-barbican-fips https://zuul.opendev.org/t/openstack/build/0baf4b7a49e442dba28b7ec738abc418 : FAILURE in 14m 32s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/f2dc979c483e4868b8f560c66e1a0a27 : SUCCESS in 39m 09s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/441b2524936c42acbcf514d086592607 : SUCCESS in 2h 04m 58s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/d59894b1ad224aa9856537c2f6a44419 : SUCCESS in 2h 51m 16s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/e52fe0d243bb4f50be3990c7aa308299 : TIMED_OUT in 2h 33m 45s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/eadfac167cf94fda8b7081aaf2d1f088 : SUCCESS in 1h 46m 42s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/c580740f1ca24096b158c6f8e7824f40 : SUCCESS in 2h 04m 33s (non-voting)\n- octavia-v2-dsvm-scenario-centos-9-stream https://zuul.opendev.org/t/openstack/build/9560cdcbba0f4d25bbb99c0d3eaa8276 : TIMED_OUT in 2h 34m 05s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"0e6b6db28f41f0c904147dc827f6f29000068c78","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2023-06-09 11:20:17.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"b0202832e19d7752effe29468893375226ded9cb","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2023-06-15 16:02:53.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"3f682789004d7580383a623c12562e8e72d48a75","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-15 16:03:27.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"2187a8b37b584eb1954c8e7a90882ce1f36646e9","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-15 17:42:17.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/4a9b555372dc41ea9629396252690276\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/30f404a174fc4ebea13a4939bded5e83 : TIMED_OUT in 30m 38s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/9542bd26046b4516b6fe284c658f4387 : SUCCESS in 5m 48s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/06f7d52e68c14009a5ab377788fd4d65 : SUCCESS in 5m 52s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d834b59a8f38483eb3b90b16502a4e24 : SUCCESS in 13m 10s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/1d20686dd3484ae99a49ff6b88b13ad7 : SUCCESS in 6m 12s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/b00f4ab024bb422ba0fa17d6bb995a4f : SUCCESS in 5m 50s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/f6b1037e64ba48b9b94b209576ec3f85 : SUCCESS in 8m 49s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/5e32a68783b54da4ac8cd86a9b5eaf05 : CANCELED in 35m 18s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/704b4d0a3af3424dbb88ddb6610196c7 : CANCELED in 39m 25s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/6c5b62dee737415a8f29b9e56ac82f09 : SUCCESS in 39m 08s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/0987fc41683c47c2976bb31c5129b638 : CANCELED in 38m 51s","accounts_in_message":[],"_revision_number":1},{"id":"ad51fc85d608ecf791eff33a7482af4f0da61856","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2023-06-20 05:35:28.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"8d62835b3c6b785552a63d57ebc42f2a9b220a1e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-20 08:14:56.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3903e00dd052456883c72e358c19eedc\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/a3c1d1c7661f4d02bfeeb9fd71b1a208 : SUCCESS in 7m 25s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/722b82d83bf34af8be383ad8e8f51bce : SUCCESS in 6m 57s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/478352e1cc3145d083f9cde352fcf530 : SUCCESS in 4m 19s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/fcb83b375d25440b899f7e8616ec45de : SUCCESS in 5m 37s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/ed849b258c644171bbecc91d9939a958 : SUCCESS in 4m 19s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/878ef8cb48a44cf08028e0442d7de67a : SUCCESS in 8m 56s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/9d7476a776b849bfaac852500a16b9b1 : SUCCESS in 4m 36s\n- octavia-tox-py39-tips https://zuul.opendev.org/t/openstack/build/a58d4abdcc2141da982c8d000f62e502 : SUCCESS in 4m 35s\n- octavia-tox-functional-py39-tips https://zuul.opendev.org/t/openstack/build/b32a4a1f5d8c43eb863a8f7bdf5a9371 : SUCCESS in 8m 23s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/af5fec7d847843e0a1428bcc2a7b0186 : SUCCESS in 4m 49s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/e11aa474988d475196bff32d93af64ad : SUCCESS in 7m 57s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/c864cb0d129b4442971ec7507fd923d7 : SUCCESS in 6m 50s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/6e843a033da0492c9470ff97ada8be08 : SUCCESS in 8m 30s (non-voting)\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/63c42831cfd2417a96d2a2a7616f3b98 : SUCCESS in 1h 07m 21s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/7c3075f0e89a405ab86098b45478d5d6 : SUCCESS in 1h 44m 06s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/96f7243a3fc44d93b52f04e3a7dcf44d : SUCCESS in 37m 07s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/5b6c96279d4546449361e414881ee37a : SUCCESS in 44m 44s\n- octavia-v2-dsvm-tls-barbican-fips https://zuul.opendev.org/t/openstack/build/36ffd3c1dc8c453c8d492d6d364df432 : FAILURE in 10m 57s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/b0c8a9c628b7417e9c35a049ac83511d : SUCCESS in 37m 53s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/8396cb99243f4da5b8ff270e23bd7421 : SUCCESS in 1h 46m 16s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/10344224448f4960957a16968b3952af : SUCCESS in 2h 35m 59s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/9faf412d2cc348a1861e77d4498c1ff5 : TIMED_OUT in 2h 32m 39s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/5c83611c7fa942b09f1bda079d31f45e : SUCCESS in 1h 49m 52s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/a9bb41f5b78d469e98b48e40732e82d9 : SUCCESS in 1h 42m 38s (non-voting)\n- octavia-v2-dsvm-scenario-centos-9-stream https://zuul.opendev.org/t/openstack/build/16c1e401ee9049fd8ab9653c5c7a40b8 : SUCCESS in 2h 23m 43s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"d21d447ee94c73812e1878531f16b698fdfe1204","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-20 08:15:31.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"546e43332cf97afd3152c61e43bdc01fcfda679c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-20 10:08:52.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/33c195548d2c4d8bbaec89589ea917dc\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a61f99003fcd42bba7231cb56ce82969 : SUCCESS in 5m 09s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/3cc6a5335b164b5d91a0f712a4028411 : SUCCESS in 5m 08s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/e91fdc4d71974c75a9e2dc29292e8eef : SUCCESS in 4m 45s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f5666c32830241b8b59c7494ec9b3c4c : SUCCESS in 9m 44s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/24708ef58341475187ab13c2b46fb6a3 : SUCCESS in 5m 06s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/3f503d2bc4c74fc188d66dd6fd107c59 : SUCCESS in 4m 43s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/9017eceece2b4eee858e8debe8adbd2c : SUCCESS in 7m 04s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/f900feffae65474d9c1809ce8ae26ddb : SUCCESS in 1h 26m 10s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/5cf8ea09db9c49c8bd72279b02fb19b5 : SUCCESS in 1h 52m 50s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/f3b226d21a6e4a3388a78dca83f7f657 : SUCCESS in 40m 42s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/88f76ade7f0d423e80f869ee8cd0ee90 : SUCCESS in 38m 36s","accounts_in_message":[],"_revision_number":1},{"id":"18f35c9a1accf8618e3c35fb88c44d2b55fc1808","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-20 10:08:53.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"0ac29c04e17c19a2c1dcd8fa7b8a89eb7fdc2962","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-20 10:10:16.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/68d58af7870d41d5b9f07af391182792\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b8c77e1c1e54443886465d1803b9201a : SUCCESS in 1m 02s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/7ea2df6d90544a0e8bd876f8975fff9e : SUCCESS in 38s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"05b3343473fa3d2df14eb6d4e44a01c9e83d3042","revisions":{"05b3343473fa3d2df14eb6d4e44a01c9e83d3042":{"kind":"REWORK","_number":1,"created":"2023-06-08 08:26:23.000000000","uploader":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"ref":"refs/changes/73/885573/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/73/885573/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/73/885573/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/73/885573/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/73/885573/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/73/885573/1"}}},"commit":{"parents":[{"commit":"03c049a976301a8219d820c00366d063efd0dd65","subject":"Merge \"Send IP advertisements when plugging a new member subnet\" into stable/zed","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/03c049a976301a8219d820c00366d063efd0dd65"}]}],"author":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2023-04-27 10:00:52.000000000","tz":-240},"committer":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2023-06-08 08:25:33.000000000","tz":120},"subject":"Avoid interface name collisions in the amphora","message":"Avoid interface name collisions in the amphora\n\nAfter unpluging a network, the octavia-agent may try to plug a new\nnetwork to a new interface while another interface with the same name\nalready exists.\n\nCloses-Bug: #2017894\n\nChange-Id: I7f150c2c1c3da706055aba5e02b5ad7c032806f1\n(cherry picked from commit c4dd87f9b119ccaef2d01ae35c9fa0ffbe5f23b2)\n(cherry picked from commit 1b17529b134456980e302239225514f8f876aa36)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/05b3343473fa3d2df14eb6d4e44a01c9e83d3042"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/05b3343473fa3d2df14eb6d4e44a01c9e83d3042"}]},"branch":"refs/heads/stable/zed"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
