)]}'
{"id":"openstack%2Foctavia~886046","triplet_id":"openstack%2Foctavia~master~Ie611ba9fde7b399989d847dd0c61dd3a158652bc","project":"openstack/octavia","branch":"master","attention_set":{},"removed_from_attention_set":{"29244":{"account":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"last_update":"2023-06-22 13:56:59.000000000","reason":"\u003cGERRIT_ACCOUNT_29244\u003e replied on the change","reason_account":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"}}},"hashtags":[],"change_id":"Ie611ba9fde7b399989d847dd0c61dd3a158652bc","subject":"Fix TCP HMs on UDP pools with SELinux","status":"MERGED","created":"2023-06-14 08:41:56.000000000","updated":"2023-06-22 18:55:57.000000000","submitted":"2023-06-22 18:54:13.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":4,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"886046","meta_rev_id":"5d7ce8d63dfe20d40ea884477a1ed0205b0f9a5e","_number":886046,"virtual_id_number":886046,"owner":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-06-22 18:54:13.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"all":[{"value":2,"date":"2023-06-21 08:41:40.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2023-06-20 14:39:57.000000000","_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"all":[{"value":1,"date":"2023-06-21 08:41:40.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}],"CC":[{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-06-14 11:41:49.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-06-20 14:39:57.000000000","updated_by":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"reviewer":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"state":"REVIEWER"},{"updated":"2023-06-21 08:41:40.000000000","updated_by":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"reviewer":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"state":"REVIEWER"},{"updated":"2023-06-22 10:21:31.000000000","updated_by":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"reviewer":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"state":"CC"}],"messages":[{"id":"8d2f65ad4c04bc59b3c56b925b6795cac27c681a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2023-06-14 08:41:56.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"e677addc75c132c382d20ac0e7ce8d334cc5195b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-14 11:41:49.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/897e2411ad2942c3bf8440dea80afd19\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/51892638cd6e436e84813b751222e2bb : SUCCESS in 8m 59s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ad542500750e4fc9adfe3a553b2a70de : SUCCESS in 5m 03s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/f3dd0bc74f2f4545a2b263200cc77538 : SUCCESS in 5m 03s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/9a08f22d62ac4a409a456f9cb3ff9083 : SUCCESS in 4m 47s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/9c23e3cd3e6b470787d73a78f4a7969f : SUCCESS in 4m 48s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d0f69b8f496c4b1aab41c6641880002c : SUCCESS in 9m 58s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/4613bf1b188243b9b4be2e60d2159dba : SUCCESS in 4m 57s\n- octavia-tox-py39-tips https://zuul.opendev.org/t/openstack/build/8d661e1adc404eabbba362dd06f05023 : SUCCESS in 4m 55s\n- octavia-tox-functional-py39-tips https://zuul.opendev.org/t/openstack/build/a7f567c484ec456f91269cb56bd75e77 : SUCCESS in 6m 58s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/dbfedf129599493faf5f073cb708866c : SUCCESS in 4m 20s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/3ab8d50fc62e4bfdb82a7c221b914960 : SUCCESS in 6m 59s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/f1939eaee152471a80cda627d607f4e4 : SUCCESS in 6m 56s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/4208c9dc5aa946238773135ef6c5a214 : SUCCESS in 7m 52s (non-voting)\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/a47dfd17937f4ee0aa98b06416564b46 : SUCCESS in 1h 16m 54s\n- octavia-v2-dsvm-scenario-traffic-ops https://zuul.opendev.org/t/openstack/build/46fc73fc1fbd4b29b872eac0fbdf5445 : SUCCESS in 1h 53m 48s\n- octavia-v2-dsvm-scenario-non-traffic-ops https://zuul.opendev.org/t/openstack/build/ad0501a68a074e5191dbd6fbe97106d0 : SUCCESS in 1h 16m 22s\n- octavia-v2-dsvm-scenario-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/f51a3a4249ae47969a785cbac17b5c74 : SUCCESS in 2h 00m 55s\n- octavia-v2-dsvm-scenario-non-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/f4f6323ae3ba4e7a8a5e6f859323919f : SUCCESS in 1h 10m 20s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/c369eb8163d449ed8a29b3b632d7aa91 : SUCCESS in 37m 13s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/028426f1b83e4a8694f60e943207cb11 : SUCCESS in 39m 23s\n- octavia-v2-dsvm-tls-barbican-fips https://zuul.opendev.org/t/openstack/build/226b57327810488286f24575ef2962dd : FAILURE in 59m 01s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/2b09784b509b44e3840260063a3b42c5 : SUCCESS in 32m 48s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/7428459dcb864ae0b86e99d65027f5f4 : TIMED_OUT in 2h 33m 45s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/e65d86fa76d94b25bc3a33cdbc1e7c5b : SUCCESS in 2h 58m 43s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/9b9b2ca1092b438a9e1adeed5257703d : FAILURE in 1h 23m 03s (non-voting)\n- octavia-v2-dsvm-scenario-centos-9-stream https://zuul.opendev.org/t/openstack/build/1b65bfe91a70404fb86a1e02102180f7 : FAILURE in 1h 16m 35s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"b02a374f58b76d42cec7c83793074d7b68ce82c2","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2023-06-20 14:39:57.000000000","message":"Patch Set 1: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"989229e616610cd7fa410d5399eaebf73c20940e","author":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"date":"2023-06-21 08:41:40.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"15a129ac5772bae9583c9bb88882a6554b4b0853","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-21 08:42:15.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"c7d13408d0a3b2e885e503ce114904ea56c727cf","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-21 09:37:45.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/3e883d296bb445e59199d7d2921d24bd\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9774274631a3403598fa9b0c87a01c57 : SUCCESS in 15m 59s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/6111bb11352f48af8e079ad032e0cbc8 : SUCCESS in 14m 56s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/2388313f66cf4cd2a500860db09525cc : SUCCESS in 17m 20s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/82438c770e1c428980def17f26bd5a22 : SUCCESS in 14m 30s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/019f2c58e225431280219554bfd742d4 : SUCCESS in 11m 52s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/8232ea36ed2843feafdcfa64a33ab1e6 : SUCCESS in 11m 36s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/feb0e33504bb40ec9e44fb3715904408 : SUCCESS in 18m 14s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/1eaf1dfbdc004942a19f16a931e17ee9 : SUCCESS in 17m 52s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/af6eef09dd474078b5523456ec87b884 : CANCELED\n- octavia-v2-dsvm-scenario-traffic-ops https://zuul.opendev.org/t/openstack/build/1ecd7fae04f24e1c80822879cce5a1c8 : CANCELED\n- octavia-v2-dsvm-scenario-non-traffic-ops https://zuul.opendev.org/t/openstack/build/110ae82222344ef8b835f88b812ebf7f : CANCELED\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/0cc2921c79364e31880ce674559cb6d0 : FAILURE in 49m 18s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/084c81d77ddc472ba646d6ca21027e2b : SUCCESS in 41m 18s","accounts_in_message":[],"_revision_number":1},{"id":"cfd95069cca7c8d55d6e297a6bf7c12337f2b09a","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2023-06-21 11:26:50.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"ff3c979fec053fcea1ab302ba62f3ef05c35add4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-21 17:37:11.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/37ec5935b0cb411c860396af89069e1e\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/c8dbee531d2649aeba5f5a87df76f565 : SUCCESS in 11m 12s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/69fdbfb5de134d84849dfdff6e64ca27 : SUCCESS in 5m 49s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/c939d091948e4ae88b243e3a6325a5f9 : SUCCESS in 5m 21s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/951ad4463aa344eca445c6be806034a0 : SUCCESS in 5m 01s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/769e5a143fd547539af57b28f93258bc : SUCCESS in 5m 15s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5969e218cae94a31b2f781feb2b7521c : SUCCESS in 11m 55s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/fdceff70cf974b81be81ab58cb194790 : SUCCESS in 4m 57s\n- octavia-tox-py39-tips https://zuul.opendev.org/t/openstack/build/6a8cc2c8047d4765a871cd5dcbee9b15 : SUCCESS in 5m 02s\n- octavia-tox-functional-py39-tips https://zuul.opendev.org/t/openstack/build/fb9fc013136749488cbd88f9d386778f : SUCCESS in 8m 43s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/178ec0ec9d3c433aaaba4d2520db3a3e : SUCCESS in 4m 21s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/45fe255a354d4e7d9712349550ce90f8 : SUCCESS in 8m 27s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/85c19cbc5e88484a98a9d386bdb3f81f : SUCCESS in 7m 13s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/da3393ab29d24837a0de0c3c736b4122 : SUCCESS in 10m 04s (non-voting)\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/6ffe8b233b734cbba6210473cc55cda6 : SUCCESS in 1h 37m 26s\n- octavia-v2-dsvm-scenario-traffic-ops https://zuul.opendev.org/t/openstack/build/798c57a8c1e849748150ecc4d901f34f : SUCCESS in 2h 04m 28s\n- octavia-v2-dsvm-scenario-non-traffic-ops https://zuul.opendev.org/t/openstack/build/88658db1019d46a7adcc3433e89eb45c : SUCCESS in 1h 20m 35s\n- octavia-v2-dsvm-scenario-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/72712d731b914cf9b1b91ccf4aa018b3 : SUCCESS in 2h 04m 05s\n- octavia-v2-dsvm-scenario-non-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/0e3e6f40da8e4d7bb46dca822f3c6042 : SUCCESS in 1h 32m 59s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/5cb897aba2ee4d748475f186a76a03f0 : SUCCESS in 36m 51s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/65d2f4fa0af4409faedfaf563c1773bc : SUCCESS in 36m 31s\n- octavia-v2-dsvm-tls-barbican-fips https://zuul.opendev.org/t/openstack/build/b03474642fe245d1adf2819e36ac1942 : NODE_FAILURE Node request 300-0021506591 failed in 0s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/f188e35d045b4e3ebb10b332bcb1a375 : SUCCESS in 32m 19s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/46ba92c835ce41bbaccf5965e2a1334f : TIMED_OUT in 2h 32m 57s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/915b657a6811447888976cc12fe96a31 : TIMED_OUT in 3h 03m 53s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/d862a2395dda47d1bf5a256c0fdafffa : SUCCESS in 2h 31m 18s (non-voting)\n- octavia-v2-dsvm-scenario-centos-9-stream https://zuul.opendev.org/t/openstack/build/5853615efd3948a08ad6a2eca2f45785 : NODE_FAILURE Node request 300-0021506596 failed in 0s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"a5cbe1f0c58ba6180c65c68d89f979fdd827c996","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-21 17:37:55.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"bbd6f2b4ecbea30a212330f4168608b17bcfd4c4","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-21 19:30:07.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/f4ffc9dda1ac4ca4b2d968d3edee6829\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9c0740537c5e42dab6445a96851ad47c : SUCCESS in 4m 54s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/e14064902736404394c7fec5a8b24637 : SUCCESS in 5m 33s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/b69135a6fcd341e396ab62e31c101015 : SUCCESS in 4m 50s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/f4f70e4a778e4ff69cc6909b653d08a3 : SUCCESS in 6m 38s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/315fbfe072f4412683b6c04a59e0b018 : TIMED_OUT in 33m 04s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/f2fd14b04f724b3ca4f58accab367427 : SUCCESS in 4m 31s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/cf1617718a5848babe915c2c699bf081 : SUCCESS in 5m 45s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/b299e7a9bda749388f94ec90b72d187c : SUCCESS in 7m 54s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/c5c48a83ddcb4bcb9626f39920c7b602 : CANCELED\n- octavia-v2-dsvm-scenario-traffic-ops https://zuul.opendev.org/t/openstack/build/d36414cf586e401293552014d8ac355b : CANCELED\n- octavia-v2-dsvm-scenario-non-traffic-ops https://zuul.opendev.org/t/openstack/build/8d98ce8fc2924bf9b0d3f19688a8ca46 : CANCELED\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/777dbfa78d704cb08cdfe89c19b540a7 : CANCELED\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/ce75e13ed6d24203a6bbfa2afd769465 : CANCELED","accounts_in_message":[],"_revision_number":1},{"id":"efef7982a31a660efa74263359e4216a0fcff6f2","author":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"date":"2023-06-22 10:21:31.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"b2a0465e26e41e22fb98a2e85b8028135b9f6c8c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-22 13:51:32.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/5a7bab73c6bc4d5fa314f822d3178d55\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/c7cb09c129c343c2ba728f1ad9f245dc : SUCCESS in 10m 27s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/44739208ac714fb78c5e9b25d11ae839 : SUCCESS in 7m 36s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/515895cfe88f47a69768f93ace019573 : SUCCESS in 7m 19s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/43c9ac9382f24618b00c22d6551206f3 : SUCCESS in 6m 19s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/a113246c28714315a4015ab972787135 : SUCCESS in 4m 39s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c1570115848f453ebfe4bf7293ec327f : SUCCESS in 16m 27s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/ec70c65f38ba4f56bbec1d38306dc911 : SUCCESS in 5m 42s\n- octavia-tox-py39-tips https://zuul.opendev.org/t/openstack/build/52c522dd34cf4895a9f9e2d815d5de56 : SUCCESS in 5m 21s\n- octavia-tox-functional-py39-tips https://zuul.opendev.org/t/openstack/build/2bc5125ba3ba4a009b0a96fb849503b7 : SUCCESS in 8m 26s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/bc49e16365014861b2ae387598d5ae48 : SUCCESS in 7m 35s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/83b8c77e58044c0fa941bc5d32925c87 : SUCCESS in 9m 04s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/6b7cacc9e1604b94a468ee65329ba6e8 : SUCCESS in 7m 52s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/cad86a5cdd5d4379bfe5aa494259dffa : SUCCESS in 11m 31s (non-voting)\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/7ba2842a09c44cf4b235231808703f60 : SUCCESS in 2h 22m 45s\n- octavia-v2-dsvm-scenario-traffic-ops https://zuul.opendev.org/t/openstack/build/f9b35a2737a6458d8590e11b1b0670e1 : FAILURE in 1h 58m 36s\n- octavia-v2-dsvm-scenario-non-traffic-ops https://zuul.opendev.org/t/openstack/build/e3fb20e053fd48ba928f716b1bd4a0fe : SUCCESS in 1h 17m 23s\n- octavia-v2-dsvm-scenario-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/4c93efd2d6d044ac83fa22fa10c6b9b7 : SUCCESS in 2h 07m 15s\n- octavia-v2-dsvm-scenario-non-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/961d00d01de24166af3f1a7a883921f6 : SUCCESS in 1h 25m 15s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/d7ae3cd24b484dc7837e999034d4c509 : SUCCESS in 36m 16s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/d6ae1d354bf0475bb3d4a87120dac85e : SUCCESS in 36m 21s\n- octavia-v2-dsvm-tls-barbican-fips https://zuul.opendev.org/t/openstack/build/76858a4f4b6941ff9262696ec77a5806 : NODE_FAILURE Node request 300-0021516621 failed in 0s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/83edc6c0b82a41ed9cf4eb59d41c2546 : SUCCESS in 34m 44s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/1aa958fc6f2a477eb8771e0d02bc5b0f : TIMED_OUT in 2h 33m 01s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/37d69c1615a040ff861722d3cd195e89 : SUCCESS in 3h 01m 54s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/62b51dd79f0e4aefbf3d30970da76c46 : TIMED_OUT in 2h 32m 47s (non-voting)\n- octavia-v2-dsvm-scenario-centos-9-stream https://zuul.opendev.org/t/openstack/build/b639c0ee00ab469282d11fb2df8c6318 : NODE_FAILURE Node request 300-0021516626 failed in 0s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"33df695ee1b9dffe85b7461a15a494484078be9c","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2023-06-22 13:56:59.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"159bcb32934bce9a29bd18304cdd17d90619bc19","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-22 16:54:08.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/eea9b4d26f764ce9b80738d8237973c1\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/0a10c2fc0afa42648e8b7c59891af7ee : SUCCESS in 10m 15s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/41fbc785b1ae4b85a96c9bbb2a2764ea : SUCCESS in 19m 40s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/8dbc878b3eff4c068962ebec5f84a89d : SUCCESS in 5m 17s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/84970919f9a84a029a9140deae7068ed : SUCCESS in 6m 30s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/0d54fc6e65c44404b636730c962db8db : SUCCESS in 6m 31s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/088d956b9d4b4548abfd42f0a75ee5a0 : SUCCESS in 14m 38s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/373ed91b2b634591a2afcfd1e4b7366c : SUCCESS in 6m 53s\n- octavia-tox-py39-tips https://zuul.opendev.org/t/openstack/build/84cccb3fd045409abf226837e4456b81 : SUCCESS in 5m 09s\n- octavia-tox-functional-py39-tips https://zuul.opendev.org/t/openstack/build/72eca2aba9c64377bf1922c4a9f3b1d9 : SUCCESS in 7m 33s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/85854b06917f48da9463a7470c1791e6 : SUCCESS in 4m 26s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/796daee9976a4435b77d3482e6a1a4ff : SUCCESS in 9m 47s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/ba466dc1aaef49c89c4e81d76743f6c4 : SUCCESS in 7m 29s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/ad7dc5abfc5a4c829205e08dcb8d42e8 : SUCCESS in 6m 53s (non-voting)\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/ee64e13906814a6da4e13325024331b9 : SUCCESS in 1h 41m 08s\n- octavia-v2-dsvm-scenario-traffic-ops https://zuul.opendev.org/t/openstack/build/22e9679142124357b4e675391c6dcdbd : SUCCESS in 2h 05m 21s\n- octavia-v2-dsvm-scenario-non-traffic-ops https://zuul.opendev.org/t/openstack/build/1ba005a72c684a7badf62c6e5e39b63b : SUCCESS in 1h 16m 10s\n- octavia-v2-dsvm-scenario-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/dce1ae3ac9e24b62b89edb65d080901a : SUCCESS in 2h 21m 30s\n- octavia-v2-dsvm-scenario-non-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/536af5acdab54c3b8cd4216459b1f84f : SUCCESS in 1h 35m 12s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/97c4b7323cbe498aac2a595c0c4d053f : SUCCESS in 40m 54s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/1613f75562134d548e8bdbf07ecbfda0 : SUCCESS in 38m 53s\n- octavia-v2-dsvm-tls-barbican-fips https://zuul.opendev.org/t/openstack/build/5d8f03fe447e432980a9e212fd1f7d4e : NODE_FAILURE Node request 300-0021518251 failed in 0s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/4c5934b07f8c4c9ca2184382807a3bfb : SUCCESS in 34m 10s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/4086935a1c9d4bb399613c4f51742b85 : TIMED_OUT in 2h 32m 58s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/ab79ba5fb8d34b4cad28c1a45f769992 : SUCCESS in 2h 55m 40s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/356e6aec0fc4433996e51392fb6cb61c : TIMED_OUT in 2h 32m 50s (non-voting)\n- octavia-v2-dsvm-scenario-centos-9-stream https://zuul.opendev.org/t/openstack/build/e80f87ed90ef45a583f44e1d08aeb96d : NODE_FAILURE Node request 300-0021518256 failed in 0s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"0a7701fe17b51da056eb2ad3af505bfc56874cd8","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-22 16:54:47.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"1d72c2bce7c8f32f9530ed8efafb815e8750770d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-22 18:54:13.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/968165299b144088a8ab92d902b412d0\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/099d97104c1c4bf78264e3ac06de9e2d : SUCCESS in 7m 16s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/b038407b0a0c41038242d6b1b7ddc332 : SUCCESS in 5m 08s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/fd8f208ce0f443dc8f219be1ec47e29e : SUCCESS in 6m 14s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/bb274bd5747541db99c30f5de61af1a8 : SUCCESS in 4m 35s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/41303d3139584f2b86e278dd9d10935b : SUCCESS in 15m 10s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/69696a884d7547bc978a956bce7e6fc2 : SUCCESS in 4m 42s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/98a5189369294d339c437c854387dcd6 : SUCCESS in 4m 38s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/acb2a0cf218a4a51830e71aa280479b7 : SUCCESS in 7m 06s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/befda0bc6f464c348d67467465b2b472 : SUCCESS in 1h 44m 08s\n- octavia-v2-dsvm-scenario-traffic-ops https://zuul.opendev.org/t/openstack/build/661bea1a172e41758f853ef76915d186 : SUCCESS in 1h 58m 56s\n- octavia-v2-dsvm-scenario-non-traffic-ops https://zuul.opendev.org/t/openstack/build/af27c961aab34291bdf7428af51abb8c : SUCCESS in 1h 15m 28s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/a7c011197d6247eaaf37f190b38cb3c3 : SUCCESS in 39m 56s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/426f8f361a904ad68aec1394f641f984 : SUCCESS in 56m 59s","accounts_in_message":[],"_revision_number":1},{"id":"e86c0bc5cd016b403b99bb192a76bd81c6d392d1","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-22 18:54:13.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"5d7ce8d63dfe20d40ea884477a1ed0205b0f9a5e","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-22 18:55:57.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d49af7070acd491a8ecf8f8e424cce3e\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1a2bcf4971724f8899d2cc437666bf06 : SUCCESS in 1m 24s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/56c14128b2844fa793ee6c7fdc58e0d9 : SUCCESS in 1m 06s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"294bd406f312984ee3029b301727d78caf7aea1d","revisions":{"294bd406f312984ee3029b301727d78caf7aea1d":{"kind":"REWORK","_number":1,"created":"2023-06-14 08:41:56.000000000","uploader":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"ref":"refs/changes/46/886046/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/46/886046/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/46/886046/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/46/886046/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/46/886046/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/46/886046/1"}}},"commit":{"parents":[{"commit":"e2bc07222f40ce5c6ba73a0e4bf79af9f5ebde11","subject":"Merge \"Fix hm operating status to ONLINE in single lb call\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/e2bc07222f40ce5c6ba73a0e4bf79af9f5ebde11"}]}],"author":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2023-06-14 08:32:08.000000000","tz":-240},"committer":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2023-06-14 08:32:08.000000000","tz":-240},"subject":"Fix TCP HMs on UDP pools with SELinux","message":"Fix TCP HMs on UDP pools with SELinux\n\nSELinux denied some specific TCP ports when using TCP-based HMs in UDP\npools (keepalived).\nEnable a SELinux boolean keepalived_connect_any which allows keepalived\nto connect to any port.\n\nCloses-Bug: #2023751\nChange-Id: Ie611ba9fde7b399989d847dd0c61dd3a158652bc\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/294bd406f312984ee3029b301727d78caf7aea1d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/294bd406f312984ee3029b301727d78caf7aea1d"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
