)]}'
{"id":"openstack%2Foctavia~887480","triplet_id":"openstack%2Foctavia~stable%2Fyoga~Ie611ba9fde7b399989d847dd0c61dd3a158652bc","project":"openstack/octavia","branch":"stable/yoga","attention_set":{},"removed_from_attention_set":{"29244":{"account":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"last_update":"2023-07-12 17:17:26.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Ie611ba9fde7b399989d847dd0c61dd3a158652bc","subject":"Fix TCP HMs on UDP pools with SELinux","status":"MERGED","created":"2023-07-03 08:26:57.000000000","updated":"2023-07-12 17:18:49.000000000","submitted":"2023-07-12 17:17:26.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"887480","meta_rev_id":"5310071d5b1e3990607944a39c79856048761ee4","_number":887480,"virtual_id_number":887480,"owner":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-07-12 17:17:25.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"all":[{"value":2,"date":"2023-07-12 15:18:24.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"all":[{"value":1,"date":"2023-07-12 15:18:24.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-07-03 10:31:08.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-07-12 15:18:24.000000000","updated_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"reviewer":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"state":"REVIEWER"}],"messages":[{"id":"07b604eac856d81a8d520d29088a191fa1eb8deb","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2023-07-03 08:26:57.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"2ad21d68e200f94ec6bd2aa36209c66e2f9e29f5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-03 10:31:08.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5885527b63e54ae8b84c1f3d9ec3e3f5\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/1293ec25933849ea8eccc282b74e04f8 : SUCCESS in 11m 34s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/5c56d4d94b6e4e12a7410d991fe87788 : SUCCESS in 8m 41s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3e2e46ce8c874b14b57a96e273593f43 : SUCCESS in 6m 39s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/744ecd617b1b43d592ca5a3f32c690bf : SUCCESS in 5m 16s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/2d013e8bcdbf428b971719dbe5fd8467 : SUCCESS in 5m 11s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a1dfd2e8bd544b5c8b4b01ada81cba29 : SUCCESS in 11m 38s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/9a1b341550284b3fa5f8bfced3c5e8e0 : SUCCESS in 5m 25s\n- octavia-tox-py39-tips https://zuul.opendev.org/t/openstack/build/e2fda18b3671416793109ef1ec7b2729 : SUCCESS in 6m 05s\n- octavia-tox-functional-py39-tips https://zuul.opendev.org/t/openstack/build/7f4609e9b8cf4f3eaf2158b8e0b1808a : SUCCESS in 5m 35s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/c7fd0957c11b489bb6e48a16813fcc9d : SUCCESS in 5m 09s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/5dcbdc2afa734cbfa034b098fb375734 : SUCCESS in 7m 15s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/315c38265955445ca6e51b3c880db9cc : SUCCESS in 1h 27m 02s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/ada22b39d98f4d47a75aa63199d56017 : SUCCESS in 1h 45m 10s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/c3ab94f45ba14a658ce5cbf6a54ed709 : SUCCESS in 33m 16s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/c93cee6afc954639b41160ad7cf7180c : SUCCESS in 36m 53s\n- octavia-v2-dsvm-tls-barbican-fips https://zuul.opendev.org/t/openstack/build/627a57f975fc47bba19dbecee51710cf : NODE_FAILURE Node request 300-0021600866 failed in 0s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/bc1008527d0048efada7892a283507ae : SUCCESS in 30m 18s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/a06a0206c056420e9e6f33cbc73b5dd9 : SUCCESS in 1h 48m 55s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/945f244d803c4488b077d99ed2fe52fb : FAILURE in 10m 13s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/04d68ff0aca14db28533f6920ec29df2 : FAILURE in 5m 56s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/a56a7a76b4f04da68d4cfe91d48460d9 : SUCCESS in 1h 45m 19s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/36cd49abc2904a81ae4a1ca4b014db26 : SUCCESS in 1h 35m 16s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8-stream https://zuul.opendev.org/t/openstack/build/22331bce29b64884a9c0bc61381fa967 : NODE_FAILURE Node request 300-0021600873 failed in 0s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"4684f7d403fcbdaeada245b86261a27b0c59abf1","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2023-07-12 15:18:24.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"28eaaa3e11ce7dfac60dc8f17ad6f4879aec0385","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-12 15:19:06.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"397ce0a693bef4f3bdddf4ed0fc054c73fcf3008","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-12 17:17:25.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8a456520a1ae4d6a8569f2b003919401\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/dba73aca2a014803b826b73bf46afc05 : SUCCESS in 7m 38s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f65d4f4e83f143859307492bc22f7d6c : SUCCESS in 7m 31s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/64b3c9c7aade445dafe797009e6721d4 : SUCCESS in 5m 31s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/6c10ffdbe8ef4c4a9806f88f1ad16a4d : SUCCESS in 4m 59s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c770b5162ec344a89da46a85f1b9d9de : SUCCESS in 12m 25s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/87bef3545aeb46709c67b70357eb24d2 : SUCCESS in 5m 14s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/84664949109f40be9b9df163ab877f83 : SUCCESS in 4m 42s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/b79dabb60fb4496aaba60caddec8586c : SUCCESS in 8m 33s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/9b33004c5eb944fbb790bc4699215150 : SUCCESS in 1h 33m 29s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/5411adb6fed0401ba42ea4eadaa48cd7 : SUCCESS in 1h 37m 35s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/a8cccf66b43440e6831baf6e8e4fff55 : SUCCESS in 35m 55s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/6ce364878b8d4efc88edf0b9ab6201d4 : SUCCESS in 35m 35s","accounts_in_message":[],"_revision_number":1},{"id":"ad1b47625d9038ad99eebb8bc6fc8d2f4b755023","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-12 17:17:26.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"5310071d5b1e3990607944a39c79856048761ee4","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-12 17:18:49.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9fd3939bc6de4b7ca54f6e71b68e9b6b\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2a4511c502b84e858677ae1731aca8c4 : SUCCESS in 57s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/fbc679820a6c4da09c5b95ddc0631366 : SUCCESS in 47s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"da9dc1230ea5eee46a13a0367bdf53ab1f34f917","revisions":{"da9dc1230ea5eee46a13a0367bdf53ab1f34f917":{"kind":"REWORK","_number":1,"created":"2023-07-03 08:26:57.000000000","uploader":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"ref":"refs/changes/80/887480/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/80/887480/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/80/887480/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/80/887480/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/80/887480/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/80/887480/1"}}},"commit":{"parents":[{"commit":"50a4e1d4c7d2c63b59a948bbf68da0f0f4170357","subject":"Merge \"Fix pool creation with single LB create call\" into stable/yoga","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/50a4e1d4c7d2c63b59a948bbf68da0f0f4170357"}]}],"author":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2023-06-14 08:32:08.000000000","tz":-240},"committer":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2023-07-03 08:26:51.000000000","tz":120},"subject":"Fix TCP HMs on UDP pools with SELinux","message":"Fix TCP HMs on UDP pools with SELinux\n\nSELinux denied some specific TCP ports when using TCP-based HMs in UDP\npools (keepalived).\nEnable a SELinux boolean keepalived_connect_any which allows keepalived\nto connect to any port.\n\nCloses-Bug: #2023751\nChange-Id: Ie611ba9fde7b399989d847dd0c61dd3a158652bc\n(cherry picked from commit 294bd406f312984ee3029b301727d78caf7aea1d)\n(cherry picked from commit c0ceebebbfcf254e5f7b58f18208392260795259)\n(cherry picked from commit 4d52ce9c5c82c57690fdeacc44462e4822b80aea)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/da9dc1230ea5eee46a13a0367bdf53ab1f34f917"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/da9dc1230ea5eee46a13a0367bdf53ab1f34f917"}]},"branch":"refs/heads/stable/yoga"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
