)]}'
{"id":"openstack%2Foctavia~887593","triplet_id":"openstack%2Foctavia~master~Iec6732187561f767924fd192aba4efdb8af2deea","project":"openstack/octavia","branch":"master","topic":"add_lb_single_create_validation","attention_set":{"31664":{"account":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"last_update":"2026-02-05 21:15:53.000000000","reason":"\u003cGERRIT_ACCOUNT_11628\u003e replied on the change","reason_account":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}},"30996":{"account":{"_account_id":30996,"name":"Weronika Sikora","email":"weronika.sikora@ovhcloud.com","username":"weronika.sikora"},"last_update":"2026-02-05 21:15:53.000000000","reason":"\u003cGERRIT_ACCOUNT_11628\u003e replied on the change","reason_account":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}},"11628":{"account":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"last_update":"2026-02-06 08:01:25.000000000","reason":"\u003cGERRIT_ACCOUNT_29244\u003e replied on the change","reason_account":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"}}},"removed_from_attention_set":{},"hashtags":[],"change_id":"Iec6732187561f767924fd192aba4efdb8af2deea","subject":"WIP Fix vip ports leak on fully-populated load balancer create","status":"NEW","created":"2023-07-04 11:06:14.000000000","updated":"2026-02-06 14:06:25.000000000","submit_type":"MERGE_IF_NECESSARY","mergeable":false,"submittable":false,"total_comment_count":3,"unresolved_comment_count":0,"work_in_progress":true,"has_review_started":true,"meta_rev_id":"61da2d6a326460282cb3601ffd323af16c046496","_number":887593,"virtual_id_number":887593,"owner":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":30996,"name":"Weronika Sikora","email":"weronika.sikora@ovhcloud.com","username":"weronika.sikora"},{"tag":"autogenerated:zuul:check","value":-1,"date":"2026-02-06 14:06:25.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":30996,"name":"Weronika Sikora","email":"weronika.sikora@ovhcloud.com","username":"weronika.sikora"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":30996,"name":"Weronika Sikora","email":"weronika.sikora@ovhcloud.com","username":"weronika.sikora"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":30996,"name":"Weronika Sikora","email":"weronika.sikora@ovhcloud.com","username":"weronika.sikora"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"}],"REVIEWER":[{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":30996,"name":"Weronika Sikora","email":"weronika.sikora@ovhcloud.com","username":"weronika.sikora"}]},"pending_reviewers":{"CC":[{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"}],"REVIEWER":[{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":30996,"name":"Weronika Sikora","email":"weronika.sikora@ovhcloud.com","username":"weronika.sikora"}]},"reviewer_updates":[{"updated":"2023-07-04 14:11:25.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-02-05 21:11:04.000000000","updated_by":{"_account_id":30996,"name":"Weronika Sikora","email":"weronika.sikora@ovhcloud.com","username":"weronika.sikora"},"reviewer":{"_account_id":30996,"name":"Weronika Sikora","email":"weronika.sikora@ovhcloud.com","username":"weronika.sikora"},"state":"REVIEWER"},{"updated":"2026-02-05 21:15:53.000000000","updated_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"reviewer":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"state":"REVIEWER"},{"updated":"2026-02-06 08:01:25.000000000","updated_by":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"reviewer":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"state":"CC"}],"messages":[{"id":"e81bfecba3ff2915e54c1502cc5a231ff0b8c188","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"date":"2023-07-04 11:06:14.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"92c0ea4edea0130e929824c3196de349dbc736be","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-04 14:11:25.000000000","message":"Patch Set 1: Verified-1\n\n(5 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/eb1430533b89490eb7975b588a6114ed\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/1ffd1c5498ca46d3b926602f4e336f3f : FAILURE in 8m 43s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/531d37c318004387a802a91f695f0ebe : FAILURE in 3m 43s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/5f9f2a5b9eab48679d2f41962809e541 : SUCCESS in 4m 24s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/bca2778d30cf4ed69c03525eddef1779 : SUCCESS in 4m 36s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/03d98ce3c36f49faa2398343f33de27f : SUCCESS in 5m 57s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/147583d248db4635a86be6946cf76b16 : SUCCESS in 13m 30s\n- octavia-tox-py39-tips https://zuul.opendev.org/t/openstack/build/c177cdd29f4b4024b86d3c092e49f356 : SUCCESS in 4m 55s\n- octavia-tox-functional-py39-tips https://zuul.opendev.org/t/openstack/build/0cdf3cfc9a4c4163aa2e5e9c61063789 : FAILURE in 8m 17s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/fe19a1ccec86448480cdfa50e1ed073c : SUCCESS in 5m 01s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/6adef03fe41b40dfbb39529e3dc64d16 : FAILURE in 8m 08s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/37efd8e874094f7eb19c5b7221b39387 : FAILURE in 9m 30s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/3ca23e83f44648eb9771620cd722336a : FAILURE in 6m 02s (non-voting)\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/9e109373ae364e16bfab6da69c5c09b3 : SUCCESS in 1h 05m 04s\n- octavia-v2-dsvm-scenario-traffic-ops https://zuul.opendev.org/t/openstack/build/dedeefce9b2b42a1bda1276d264271d7 : SUCCESS in 2h 05m 30s\n- octavia-v2-dsvm-scenario-non-traffic-ops https://zuul.opendev.org/t/openstack/build/6362dd568ace4b0abd3ce2be44e5e95d : SUCCESS in 1h 37m 38s\n- octavia-v2-dsvm-scenario-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/3226d12a794849bcbc1d48143da2999f : SUCCESS in 2h 03m 07s\n- octavia-v2-dsvm-scenario-non-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/5dad88d1d01f43689a13249e2df681b3 : SUCCESS in 1h 18m 14s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/ead52a805802485d8d9368173b917949 : SUCCESS in 45m 25s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/f7c14628b55c47a29d076beb0a9f921d : SUCCESS in 46m 14s\n- octavia-v2-dsvm-tls-barbican-fips https://zuul.opendev.org/t/openstack/build/7c207831931f47be80c75a6249f08056 : SUCCESS in 1h 23m 13s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/c727370d37204bc88d5a9ea55f14e4a6 : SUCCESS in 33m 47s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/f0fbe56b5d964dab84490645e1a94bc7 : TIMED_OUT in 2h 32m 42s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/3ac5e7b7ff224f19851200ce866c895d : TIMED_OUT in 3h 03m 19s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/6dc0520722814648a3b40c68d2c02fd1 : SUCCESS in 2h 17m 21s (non-voting)\n- octavia-v2-dsvm-scenario-centos-9-stream https://zuul.opendev.org/t/openstack/build/b58d4773ca5e431889f99f161531c8cf : SUCCESS in 2h 21m 04s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"2548db3a810c8a912608858e94f2f0075a50b606","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"date":"2023-07-06 13:55:41.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"bb261f686a9fa84d0bb503476c39b116315a978b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"date":"2023-07-06 13:58:24.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"b35005a61376b4fccbd3c50061e498210e2e2eea","tag":"autogenerated:gerrit:setWorkInProgress","author":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"date":"2023-07-06 13:58:44.000000000","message":"Set Work In Progress","accounts_in_message":[],"_revision_number":3},{"id":"c94c5346cf2878680c75d4f12f3e549595638d85","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-06 16:53:57.000000000","message":"Patch Set 3: Verified-1\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/b10b6a7eebe64c89a309b2f6d0afe736\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/f24e8c0bfbb94cb2b5302486d67bd5e9 : FAILURE in 11m 52s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d4b2d92bf1e341d3b423f9a495455d82 : FAILURE in 10m 44s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/130e3ec423214f83b08903a9be62e3ad : SUCCESS in 4m 25s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/0ca0672e50fd41f3a5a0152568f5a6e0 : SUCCESS in 6m 06s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/8db13813074c42e4ba5f69e98134c785 : SUCCESS in 5m 25s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7182fb0e9acb4f26bb95c29c052f7d55 : SUCCESS in 13m 40s\n- octavia-tox-py39-tips https://zuul.opendev.org/t/openstack/build/fdb8320b610d44e883b17959aa3b2148 : SUCCESS in 5m 15s\n- octavia-tox-functional-py39-tips https://zuul.opendev.org/t/openstack/build/5a854ebf8c8a400dbb0cc4de1af47586 : FAILURE in 13m 04s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/e6c1be6ba26a406696239070504b9fb1 : SUCCESS in 4m 42s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/e422699bb46e436ea3d3d9a8adc74775 : FAILURE in 6m 58s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/24243ad500fe4652a26733911c9d6885 : FAILURE in 7m 52s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/7a8eaae5b1c94550acfef2cc669c7835 : FAILURE in 7m 33s (non-voting)\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/1e0f0a3266d940aa9415a770eabd7de5 : SUCCESS in 1h 08m 04s\n- octavia-v2-dsvm-scenario-traffic-ops https://zuul.opendev.org/t/openstack/build/f0a7a3bcb66c493c9c7674e79869446d : SUCCESS in 1h 58m 31s\n- octavia-v2-dsvm-scenario-non-traffic-ops https://zuul.opendev.org/t/openstack/build/9a12814c107b4b3587e3aa24b7669c6d : SUCCESS in 1h 23m 25s\n- octavia-v2-dsvm-scenario-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/b8b78775d87742b6956badd361226e5c : SUCCESS in 2h 17m 53s\n- octavia-v2-dsvm-scenario-non-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/ec19a887cd814b62b927d2f8adfee5b0 : SUCCESS in 1h 33m 40s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/2806f004ac764eacaf144e966e20fa80 : SUCCESS in 47m 31s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/299f4ebb8a1a4d93b504e22bb16b8312 : SUCCESS in 46m 27s\n- octavia-v2-dsvm-tls-barbican-fips https://zuul.opendev.org/t/openstack/build/5efa6e9c8dc84fe0a6665518f1a5ef57 : NODE_FAILURE Node request 300-0021628431 failed in 0s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/3fca9a9ec6d446f6845a3503d8de51c9 : SUCCESS in 34m 29s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/800ad439d0bc42f28c3310e2df2f1855 : TIMED_OUT in 2h 32m 46s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/0c63e2368f114b00810c6e345052ada9 : SUCCESS in 2h 54m 13s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/5d83b8222ab24612962b9f2a9389fb0a : TIMED_OUT in 2h 33m 10s (non-voting)\n- octavia-v2-dsvm-scenario-centos-9-stream https://zuul.opendev.org/t/openstack/build/cb9ac1311f6442e2808f63386f38185c : SUCCESS in 2h 31m 00s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"9eb776eb50de0041dd4ebf8182fc7699dcba95c7","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"date":"2023-07-19 16:03:24.000000000","message":"Patch Set 4: Published edit on patch set 3.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":4},{"id":"0f562e1cb3e700ee47298150108ef8821d583380","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-19 18:59:10.000000000","message":"Patch Set 4: Verified-1\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/521acf51ba7446a991f8ee7af6ef4443\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/7711b58b15674dc9b27cb7fb03c0eec0 : FAILURE in 9m 12s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/74dcae41d46e498cb3af9f8db0042627 : FAILURE in 4m 57s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/6b31562f5e63497eb7bd4f9f6c03540e : SUCCESS in 4m 53s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/8afeb1d83d1a4ada9203533e64b886ce : SUCCESS in 5m 21s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/e5bfadd9bcb5497e9a9e4e6b671da13e : SUCCESS in 5m 01s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bce5684931fa467d912d951d751b1b88 : SUCCESS in 10m 41s\n- octavia-tox-py39-tips https://zuul.opendev.org/t/openstack/build/0a97907627cd48ddac3f0ece755ed9ec : SUCCESS in 5m 48s\n- octavia-tox-functional-py39-tips https://zuul.opendev.org/t/openstack/build/182f9a10ad9d414490f1de8cf37bdfe2 : FAILURE in 9m 28s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/719e9d3bea684b7186c7d726787a9025 : SUCCESS in 4m 48s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/c9c76cf666df4a8e92e18c556e3bc41a : FAILURE in 7m 32s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/a613d5ea5b4c435da6943c7f847b1d0d : FAILURE in 8m 47s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/e02c6275339b4af99139e6a519c9584c : FAILURE in 7m 59s (non-voting)\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/5418b746a0e9418483d4e5f585b57c5d : SUCCESS in 1h 55m 13s\n- octavia-v2-dsvm-scenario-traffic-ops https://zuul.opendev.org/t/openstack/build/304cbcc1b5c34aa190ebe30caa8d7349 : SUCCESS in 2h 10m 19s\n- octavia-v2-dsvm-scenario-non-traffic-ops https://zuul.opendev.org/t/openstack/build/7d21fa393c254dc8b9e6fb5746105dba : SUCCESS in 1h 16m 27s\n- octavia-v2-dsvm-scenario-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/20d37b874aae4b35bc6d3e2fb4e8dbf2 : SUCCESS in 1h 56m 32s\n- octavia-v2-dsvm-scenario-non-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/d3ea17375886417ea4d927e9946b25ee : SUCCESS in 1h 27m 25s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/0186913ed941487989ac297f6059468b : SUCCESS in 52m 23s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/869ea8c6c8cf4300829677b44637e241 : SUCCESS in 46m 00s\n- octavia-v2-dsvm-tls-barbican-fips https://zuul.opendev.org/t/openstack/build/51d761010e3b4c3abcae874fa4a880d6 : SUCCESS in 1h 17m 16s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/c22eeb20bfec4eb6890805e367b8bd19 : SUCCESS in 38m 34s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/ccd7be4858c64b1c83868ffd4d32d634 : TIMED_OUT in 2h 33m 29s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/05cb42e327a748acb08c28efa7873399 : SUCCESS in 2h 52m 59s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/920192e38f454fa794fb71c69949486c : SUCCESS in 2h 28m 50s (non-voting)\n- octavia-v2-dsvm-scenario-centos-9-stream https://zuul.opendev.org/t/openstack/build/7c44a298241146348b500589424bf4bb : SUCCESS in 2h 31m 18s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"ef5c9932f7dfb2c2b679774e90f217889b9d3ecf","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"date":"2023-07-20 08:29:55.000000000","message":"Patch Set 5: Published edit on patch set 4.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":5},{"id":"e8cb90cf50ce67c54c6816cb1a12dfb90d68a22b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-20 13:16:51.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/0aa4ff44d400426e936f8c6b0ba1bc22\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/1ca8a927b02047419b812bebaa049ad8 : FAILURE in 13m 59s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/291f6ce64e134f01b380fb4e7804c41d : SUCCESS in 4m 34s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/1e05edbfd5c04367b441a25f7ee189eb : SUCCESS in 5m 07s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/d5b7de5408c341ff86b2863fae82e83b : SUCCESS in 4m 51s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/0c8244469850417f86e94e9f6a58591d : SUCCESS in 5m 26s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/39ad9a8977fb4d259efababa81365c79 : SUCCESS in 10m 03s\n- octavia-tox-py39-tips https://zuul.opendev.org/t/openstack/build/bb63a2202b1446478b32b483e80a2ec0 : SUCCESS in 6m 16s\n- octavia-tox-functional-py39-tips https://zuul.opendev.org/t/openstack/build/3bbcd0f91a4f46a98cb97c62a09e89dd : FAILURE in 7m 55s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/e9b8f34be9304fde88e841f364fcc0ec : SUCCESS in 4m 16s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/dcc4f7d0947e455aa92ee65dbe1651e7 : FAILURE in 7m 45s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/7c867b39504348a8b8da10e8d35f3c63 : FAILURE in 6m 56s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/98e8c90ec5fe491daf619617d8fafbdd : FAILURE in 8m 21s (non-voting)\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/8871db91304f43d0a25742e7b902fd57 : SUCCESS in 1h 44m 36s\n- octavia-v2-dsvm-scenario-traffic-ops https://zuul.opendev.org/t/openstack/build/71242fdaaa0b4c14a79c02aae06edb79 : SUCCESS in 1h 59m 14s\n- octavia-v2-dsvm-scenario-non-traffic-ops https://zuul.opendev.org/t/openstack/build/23a32906b549483188358449fe20fbde : SUCCESS in 1h 15m 40s\n- octavia-v2-dsvm-scenario-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/2d216fb2647948eebdcbd6d2c149b6c1 : SUCCESS in 2h 10m 41s\n- octavia-v2-dsvm-scenario-non-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/00860841037340eaaf34831327406940 : SUCCESS in 1h 26m 05s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/39928f82f69141838d41cb04b7c08ad7 : SUCCESS in 38m 18s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/8261951ec23046f086bdeb6d1a54bc09 : SUCCESS in 53m 46s\n- octavia-v2-dsvm-tls-barbican-fips https://zuul.opendev.org/t/openstack/build/ad301e222bdb44d5a51a69fa7f21eab6 : SUCCESS in 1h 07m 59s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/a5f51c9504034879aa4273576e4e3ad0 : SUCCESS in 43m 34s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/4c5aa51191624adb84d8d06a0435821a : TIMED_OUT in 2h 32m 48s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/3fcef3f8015d4b49ac70cfd7f4558ed1 : TIMED_OUT in 3h 04m 07s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/ca342e495bd4457fb61f78218d44d513 : TIMED_OUT in 2h 34m 33s (non-voting)\n- octavia-v2-dsvm-scenario-centos-9-stream https://zuul.opendev.org/t/openstack/build/ddd0e556dd9242ca8bc2843159fdcd1f : TIMED_OUT in 2h 34m 46s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"b0dd576d9f8351368081a3a6a0c07b20d146d593","author":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"date":"2024-03-13 16:49:20.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"43dcaa338ef87cbd284031cbbcf5be301f46f0a0","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":30996,"name":"Weronika Sikora","email":"weronika.sikora@ovhcloud.com","username":"weronika.sikora"},"date":"2026-02-05 21:11:04.000000000","message":"Uploaded patch set 6.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":6},{"id":"d6319b1cf7cf32e8d831d1a12fd407f7a97231e5","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2026-02-05 21:15:53.000000000","message":"Patch Set 6: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"6e40997ae7b05a44687d7557d43210f77abe63b5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-05 22:46:45.000000000","message":"Patch Set 6: Verified-1\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/51d3a7b794644599824497ba1124d310\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/1e04d7c3b56b465da7265c30c92b83bc : FAILURE in 13m 22s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/24ef02b9e4414b2c890183aa4dd8689c : FAILURE in 4m 00s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/907e9839f58443c0b1620948b4d954ce : SUCCESS in 4m 50s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/0a909242600c45e3be773b4fa6c23f92 : SUCCESS in 3m 46s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d94cfa07d8f54369a34da6e03ac8d274 : SUCCESS in 14m 16s\n- octavia-tox-py313-tips https://zuul.opendev.org/t/openstack/build/a6a7231a799d45bca61201c6f4edee77 : SUCCESS in 3m 19s\n- octavia-tox-functional-py313-tips https://zuul.opendev.org/t/openstack/build/f30389250de74ac6a2d9815e4ed08b77 : FAILURE in 5m 00s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/ae846d42360b440b9ffb97065f78a144 : SUCCESS in 3m 49s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/06f7345bfbf345148f21a344308782d9 : FAILURE in 6m 50s\n- openstack-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/dd831921bc184c58b0852830aff35668 : FAILURE in 3m 46s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/6016419c754d4fd0a6bc1d78e16b2797 : SUCCESS in 1h 27m 46s\n- octavia-v2-dsvm-scenario-traffic-ops https://zuul.opendev.org/t/openstack/build/e5b75c5bfd374dabb61ee794f5451f08 : FAILURE in 19m 07s\n- octavia-v2-dsvm-scenario-non-traffic-ops https://zuul.opendev.org/t/openstack/build/7e5e52f4385b4459bd8b06e342b2b183 : FAILURE in 20m 08s\n- octavia-v2-dsvm-scenario-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/d30b52109c5b430fafcd32b91444f79e : FAILURE in 20m 10s\n- octavia-v2-dsvm-scenario-non-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/3f98632371e440f7adf15c7d31f73856 : FAILURE in 20m 11s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/8fa0f372b4eb42dca83152c000229951 : FAILURE in 20m 09s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/8dc3d18d5c434ac0aae81323b4c3054f : FAILURE in 21m 11s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/9913d84c419d447489c561d882b77d67 : FAILURE in 24m 03s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/fc77842ef28b4a649e71507501db0849 : FAILURE in 22m 06s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/57375de8f5284e32ad4186e16c80e6a3 : FAILURE in 24m 54s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/7fb4efd11e7f4caa912bb8a6808b21b4 : FAILURE in 20m 53s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"5ee8881b28b8bc6d51ed405e40f610f401b23a41","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2026-02-06 08:01:25.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"ead5ec3fce6ebfe08d47aa823a02dd9cebbd7458","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":30996,"name":"Weronika Sikora","email":"weronika.sikora@ovhcloud.com","username":"weronika.sikora"},"date":"2026-02-06 11:36:55.000000000","message":"Uploaded patch set 7.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified-1\n","accounts_in_message":[],"_revision_number":7},{"id":"61da2d6a326460282cb3601ffd323af16c046496","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-06 14:06:25.000000000","message":"Patch Set 7: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/70ed53764e754e919e3d4a1b820c063c\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/219fdd86a95f43979d0deb0481abed6e : FAILURE in 8m 16s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/eadbfe67410a4f88a659c63ca4115e16 : SUCCESS in 3m 18s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/6d348a8824cb42139526944e0998b796 : FAILURE in 3m 56s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/ca2d8dd2c576429ab6237ffb8bae5560 : SUCCESS in 4m 39s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/972620d70e724f59868a7dc90a00725e : SUCCESS in 7m 02s\n- octavia-tox-py313-tips https://zuul.opendev.org/t/openstack/build/37dae55cc03c46419f266b5aec390108 : SUCCESS in 3m 44s\n- octavia-tox-functional-py313-tips https://zuul.opendev.org/t/openstack/build/1aeeaaa43b5742beb59d9f3caca54b16 : FAILURE in 8m 33s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/1f522135d406421ea4554d1e465d3278 : SUCCESS in 4m 58s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/eaa6ddae56e54356b6ac2d1a31916a54 : FAILURE in 4m 19s\n- openstack-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/c290621c20a84ae3b1665130b0386466 : FAILURE in 9m 22s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/738e44d67f6d416f8e9d5690afa6145c : SUCCESS in 44m 10s\n- octavia-v2-dsvm-scenario-traffic-ops https://zuul.opendev.org/t/openstack/build/a9251c8867f841438f7c9562c790809b : SUCCESS in 1h 44m 35s\n- octavia-v2-dsvm-scenario-non-traffic-ops https://zuul.opendev.org/t/openstack/build/ecf1e9c9715147fd8067ad266fa27a0b : SUCCESS in 1h 03m 40s\n- octavia-v2-dsvm-scenario-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/32ef8217c6904443a3b6c830f6c103c8 : SUCCESS in 1h 50m 33s\n- octavia-v2-dsvm-scenario-non-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/1fa31d37c279480a8b20d74c11850a00 : SUCCESS in 1h 14m 51s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/0ea8c15a124949e0a03241241c5d5d43 : SUCCESS in 36m 19s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/6d98123b5e6d4e8cad3a235fe85438fa : SUCCESS in 49m 52s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/745070448e7a4325b7c7d3bf556a8ffd : SUCCESS in 35m 13s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/89509744d17144cbb303bbfbdaee490a : SUCCESS in 35m 32s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/2e4a7e54c21f4aa2b446719223145c56 : SUCCESS in 2h 27m 20s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/6e3267051fee44ddbbcff69b81268c6f : SUCCESS in 1h 34m 06s (non-voting)","accounts_in_message":[],"_revision_number":7}],"current_revision_number":7,"current_revision":"81aff65398c963bea07d9a50d08941aee224c9e3","revisions":{"e37cee76c4c4d7a0950a4b5ee7384d804ad12a5d":{"kind":"REWORK","_number":1,"created":"2023-07-04 11:06:14.000000000","uploader":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"ref":"refs/changes/93/887593/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/93/887593/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/93/887593/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/93/887593/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/93/887593/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/93/887593/1"}}},"commit":{"parents":[{"commit":"c2a1aebd35eae60fbda9986c680ceb19db1a8b7d","subject":"Merge \"Fix TCP HMs on UDP pools with SELinux\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/c2a1aebd35eae60fbda9986c680ceb19db1a8b7d"}]}],"author":{"name":"Omer","email":"oschwart@redhat.com","date":"2023-07-04 10:55:03.000000000","tz":120},"committer":{"name":"Omer","email":"oschwart@redhat.com","date":"2023-07-04 11:05:17.000000000","tz":120},"subject":"WIP Fix vip ports leak on fully-populated load balancer create","message":"WIP Fix vip ports leak on fully-populated load balancer create\n\nSo far, providing invalid listener/pool parameters for fully-populated\nload balancer create API, did not check if the parameters were valid,\nstarted the lb create flow, allocated vip and only then failed in the\nlistener/pool validation, and aborted the flow with an allocated vip.\n\nThis patch adds both listener and pool allocation before any deployment\noperation takes place.\n\nCloses-Bug: #2015320\nChange-Id: Iec6732187561f767924fd192aba4efdb8af2deea\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/e37cee76c4c4d7a0950a4b5ee7384d804ad12a5d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/e37cee76c4c4d7a0950a4b5ee7384d804ad12a5d"}]},"branch":"refs/heads/master"},"51b10bcf41a1888436f1aa7c4d6e943b1ae828bd":{"kind":"REWORK","_number":2,"created":"2023-07-06 13:55:41.000000000","uploader":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"ref":"refs/changes/93/887593/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/93/887593/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/93/887593/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/93/887593/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/93/887593/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/93/887593/2"}}},"commit":{"parents":[{"commit":"c2a1aebd35eae60fbda9986c680ceb19db1a8b7d","subject":"Merge \"Fix TCP HMs on UDP pools with SELinux\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/c2a1aebd35eae60fbda9986c680ceb19db1a8b7d"}]}],"author":{"name":"Omer","email":"oschwart@redhat.com","date":"2023-07-04 10:55:03.000000000","tz":120},"committer":{"name":"Omer","email":"oschwart@redhat.com","date":"2023-07-06 13:55:34.000000000","tz":120},"subject":"WIP Fix vip ports leak on fully-populated load balancer create","message":"WIP Fix vip ports leak on fully-populated load balancer create\n\nSo far, providing invalid listener/pool parameters for fully-populated\nload balancer create API, did not check if the parameters were valid,\nstarted the lb create flow, allocated vip and only then failed in the\nlistener/pool validation, and aborted the flow with an allocated vip.\n\nThis patch adds both listener and pool allocation before any deployment\noperation takes place.\n\nCloses-Bug: #2015320\nChange-Id: Iec6732187561f767924fd192aba4efdb8af2deea\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/51b10bcf41a1888436f1aa7c4d6e943b1ae828bd"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/51b10bcf41a1888436f1aa7c4d6e943b1ae828bd"}]},"branch":"refs/heads/master"},"8258aa62d3c1f7f6528fee4a391fd76204016b6b":{"kind":"REWORK","_number":3,"created":"2023-07-06 13:58:24.000000000","uploader":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"ref":"refs/changes/93/887593/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/93/887593/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/93/887593/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/93/887593/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/93/887593/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/93/887593/3"}}},"commit":{"parents":[{"commit":"c2a1aebd35eae60fbda9986c680ceb19db1a8b7d","subject":"Merge \"Fix TCP HMs on UDP pools with SELinux\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/c2a1aebd35eae60fbda9986c680ceb19db1a8b7d"}]}],"author":{"name":"Omer","email":"oschwart@redhat.com","date":"2023-07-04 10:55:03.000000000","tz":120},"committer":{"name":"Omer","email":"oschwart@redhat.com","date":"2023-07-06 13:58:00.000000000","tz":120},"subject":"WIP Fix vip ports leak on fully-populated load balancer create","message":"WIP Fix vip ports leak on fully-populated load balancer create\n\nSo far, providing invalid listener/pool parameters for fully-populated\nload balancer create API, did not check if the parameters were valid,\nstarted the lb create flow, allocated vip and only then failed in the\nlistener/pool validation, and aborted the flow with an allocated vip.\n\nThis patch adds both listener and pool allocation before any deployment\noperation takes place.\n\nCloses-Bug: #2015320\nChange-Id: Iec6732187561f767924fd192aba4efdb8af2deea\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/8258aa62d3c1f7f6528fee4a391fd76204016b6b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/8258aa62d3c1f7f6528fee4a391fd76204016b6b"}]},"branch":"refs/heads/master"},"83f565e455a5c561d42bfb4259cfe12fb3b95766":{"kind":"REWORK","_number":4,"created":"2023-07-19 16:03:24.000000000","uploader":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"ref":"refs/changes/93/887593/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/93/887593/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/93/887593/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/93/887593/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/93/887593/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/93/887593/4"}}},"commit":{"parents":[{"commit":"c2a1aebd35eae60fbda9986c680ceb19db1a8b7d","subject":"Merge \"Fix TCP HMs on UDP pools with SELinux\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/c2a1aebd35eae60fbda9986c680ceb19db1a8b7d"}]}],"author":{"name":"Omer","email":"oschwart@redhat.com","date":"2023-07-04 10:55:03.000000000","tz":120},"committer":{"name":"Omer Schwartz","email":"oschwart@redhat.com","date":"2023-07-19 16:03:24.000000000","tz":0},"subject":"WIP Fix vip ports leak on fully-populated load balancer create","message":"WIP Fix vip ports leak on fully-populated load balancer create\n\nSo far, providing invalid listener/pool parameters for fully-populated\nload balancer create API, did not check if the parameters were valid,\nstarted the lb create flow, allocated vip and only then failed in the\nlistener/pool validation, and aborted the flow with an allocated vip.\n\nThis patch adds both listener and pool allocation before any deployment\noperation takes place.\n\nCloses-Bug: #2015320\nChange-Id: Iec6732187561f767924fd192aba4efdb8af2deea\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/83f565e455a5c561d42bfb4259cfe12fb3b95766"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/83f565e455a5c561d42bfb4259cfe12fb3b95766"}]},"branch":"refs/heads/master"},"310fec66c572c690f56e4dcaa684038d8d4983de":{"kind":"REWORK","_number":5,"created":"2023-07-20 08:29:55.000000000","uploader":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"ref":"refs/changes/93/887593/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/93/887593/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/93/887593/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/93/887593/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/93/887593/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/93/887593/5"}}},"commit":{"parents":[{"commit":"c2a1aebd35eae60fbda9986c680ceb19db1a8b7d","subject":"Merge \"Fix TCP HMs on UDP pools with SELinux\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/c2a1aebd35eae60fbda9986c680ceb19db1a8b7d"}]}],"author":{"name":"Omer","email":"oschwart@redhat.com","date":"2023-07-04 10:55:03.000000000","tz":120},"committer":{"name":"Omer Schwartz","email":"oschwart@redhat.com","date":"2023-07-20 08:29:55.000000000","tz":0},"subject":"WIP Fix vip ports leak on fully-populated load balancer create","message":"WIP Fix vip ports leak on fully-populated load balancer create\n\nSo far, providing invalid listener/pool parameters for fully-populated\nload balancer create API, did not check if the parameters were valid,\nstarted the lb create flow, allocated vip and only then failed in the\nlistener/pool validation, and aborted the flow with an allocated vip.\n\nThis patch adds both listener and pool allocation before any deployment\noperation takes place.\n\nCloses-Bug: #2015320\nChange-Id: Iec6732187561f767924fd192aba4efdb8af2deea\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/310fec66c572c690f56e4dcaa684038d8d4983de"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/310fec66c572c690f56e4dcaa684038d8d4983de"}]},"branch":"refs/heads/master"},"ff7f644b9c826a5c3999647eb78068a99c965ea2":{"kind":"REWORK","_number":6,"created":"2026-02-05 21:11:04.000000000","uploader":{"_account_id":30996,"name":"Weronika Sikora","email":"weronika.sikora@ovhcloud.com","username":"weronika.sikora"},"ref":"refs/changes/93/887593/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/93/887593/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/93/887593/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/93/887593/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/93/887593/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/93/887593/6"}}},"commit":{"parents":[{"commit":"e805a93381530bd1656eb2d2ec535d101d6fe505","subject":"Merge \"Migrate upper functional job to Python 3.13\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/e805a93381530bd1656eb2d2ec535d101d6fe505"}]}],"author":{"name":"Omer","email":"oschwart@redhat.com","date":"2023-07-04 10:55:03.000000000","tz":120},"committer":{"name":"Weronika Sikora","email":"weronika.sikora@ovhcloud.com","date":"2026-02-05 21:02:53.000000000","tz":60},"subject":"WIP Fix vip ports leak on fully-populated load balancer create","message":"WIP Fix vip ports leak on fully-populated load balancer create\n\nSo far, providing invalid listener/pool parameters for fully-populated\nload balancer create API, did not check if the parameters were valid,\nstarted the lb create flow, allocated vip and only then failed in the\nlistener/pool validation, and aborted the flow with an allocated vip.\n\nThis patch adds both listener and pool allocation before any deployment\noperation takes place.\n\nCloses-Bug: #2015320\nChange-Id: Iec6732187561f767924fd192aba4efdb8af2deea\nSigned-off-by: Weronika Sikora \u003cweronika.sikora@ovhcloud.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/ff7f644b9c826a5c3999647eb78068a99c965ea2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/ff7f644b9c826a5c3999647eb78068a99c965ea2"}]},"branch":"refs/heads/master"},"81aff65398c963bea07d9a50d08941aee224c9e3":{"kind":"REWORK","_number":7,"created":"2026-02-06 11:36:55.000000000","uploader":{"_account_id":30996,"name":"Weronika Sikora","email":"weronika.sikora@ovhcloud.com","username":"weronika.sikora"},"ref":"refs/changes/93/887593/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/93/887593/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/93/887593/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/93/887593/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/93/887593/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/93/887593/7"}}},"commit":{"parents":[{"commit":"e805a93381530bd1656eb2d2ec535d101d6fe505","subject":"Merge \"Migrate upper functional job to Python 3.13\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/e805a93381530bd1656eb2d2ec535d101d6fe505"}]}],"author":{"name":"Omer","email":"oschwart@redhat.com","date":"2023-07-04 10:55:03.000000000","tz":120},"committer":{"name":"Weronika Sikora","email":"weronika.sikora@ovhcloud.com","date":"2026-02-06 11:36:33.000000000","tz":60},"subject":"WIP Fix vip ports leak on fully-populated load balancer create","message":"WIP Fix vip ports leak on fully-populated load balancer create\n\nSo far, providing invalid listener/pool parameters for fully-populated\nload balancer create API, did not check if the parameters were valid,\nstarted the lb create flow, allocated vip and only then failed in the\nlistener/pool validation, and aborted the flow with an allocated vip.\n\nThis patch adds both listener and pool allocation before any deployment\noperation takes place.\n\nCloses-Bug: #2015320\nChange-Id: Iec6732187561f767924fd192aba4efdb8af2deea\nSigned-off-by: Weronika Sikora \u003cweronika.sikora@ovhcloud.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/81aff65398c963bea07d9a50d08941aee224c9e3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/81aff65398c963bea07d9a50d08941aee224c9e3"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"OK","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY"},{"label":"Workflow","status":"MAY"},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
