)]}'
{"id":"openstack%2Foctavia~887922","triplet_id":"openstack%2Foctavia~stable%2Fxena~I1d1a9a4f193f3766df12ae43b2c2e54c4adf405f","project":"openstack/octavia","branch":"stable/xena","attention_set":{},"removed_from_attention_set":{"29244":{"account":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"last_update":"2023-10-11 13:01:17.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I1d1a9a4f193f3766df12ae43b2c2e54c4adf405f","subject":"fix haproxy_count to display the number of running processes","status":"MERGED","created":"2023-07-07 08:10:41.000000000","updated":"2023-10-11 13:03:33.000000000","submitted":"2023-10-11 13:01:17.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"887922","meta_rev_id":"efbf4c630341736beb4c5aceee4e794508d629bd","_number":887922,"virtual_id_number":887922,"owner":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-10-11 13:01:17.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"all":[{"value":2,"date":"2023-07-13 16:28:44.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2023-10-11 10:34:36.000000000","_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"all":[{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2023-10-11 10:34:36.000000000","_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}],"CC":[{"_account_id":34380,"name":"Maximilian Stinsky","email":"maximilian@stinsky.com","username":"mstinsky"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-07-07 08:10:41.000000000","updated_by":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"reviewer":{"_account_id":34380,"name":"Maximilian Stinsky","email":"maximilian@stinsky.com","username":"mstinsky"},"state":"CC"},{"updated":"2023-07-07 09:53:31.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-07-13 16:28:44.000000000","updated_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"reviewer":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"state":"REVIEWER"},{"updated":"2023-10-11 10:34:36.000000000","updated_by":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"reviewer":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"state":"REVIEWER"}],"messages":[{"id":"5bc7ad8f9c51bcf3d7a2ad8199eb481097e1219d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2023-07-07 08:10:41.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"f4f745bef2c184125dde9558b8bb9b72bde0e6d8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-07 09:53:31.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5ca2952ba7d745ac86da8f1391ce390e\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/b1101d20b2d848caae39f968ff1192f3 : SUCCESS in 11m 16s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/3346ce6b02ea4e828a16495dd46cd42c : SUCCESS in 8m 28s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b0cbb12a7401449cbaf713b09e9eec85 : SUCCESS in 4m 40s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/a260e7e2883a4f53a8fcd6ed74d3b995 : SUCCESS in 6m 42s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/592e058902a04fb4be7b25740b358a92 : SUCCESS in 4m 07s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/7e6097bff64946b5ac717b96babfe25b : SUCCESS in 5m 22s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2ab26fc7bb3744e7a791c33e02c516f6 : SUCCESS in 12m 35s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/a662458d406549ff9de0b0a952d7bfb3 : SUCCESS in 5m 25s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/eb9ba2b2cf3049ae93f80cff7f694322 : SUCCESS in 5m 12s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/a0421a70be6b48b5b21d780d518feb9a : SUCCESS in 9m 20s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/761a6c4b66694cd1b4b9cdc19164b21e : SUCCESS in 4m 43s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/3ed293104933422a9da6b870518c8b9d : SUCCESS in 7m 01s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/a44eeb97781f4874819e0bb37f0108dc : SUCCESS in 1h 17m 29s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/b1c01b680eb84e63b7026c231a9a094c : SUCCESS in 1h 36m 01s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/9808c72556c74052b139ea6a7258c3cb : SUCCESS in 32m 34s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/4fdd57661ffb4b2c9919f6bd2d48f14a : SUCCESS in 39m 19s\n- octavia-v2-dsvm-tls-barbican-fips https://zuul.opendev.org/t/openstack/build/8ebfa44875324709a41158ca50dfa906 : FAILURE in 10m 56s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/06720f5fa07b4d27b47b6be573b3c923 : SUCCESS in 28m 52s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/3ec64dee10aa465284e22dbc9097ae35 : SUCCESS in 1h 36m 22s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/af40b3ad33a44a3389fc2dcd87f2864f : FAILURE in 10m 27s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/55d8f5a2a86646c2ab7d07f6f1fd2421 : FAILURE in 6m 12s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/a445b0fa8daa464fa9e646ea48702097 : SUCCESS in 1h 40m 54s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/35c9ed6598b44fd986f93dce2f7cb6f4 : SUCCESS in 1h 35m 12s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8 https://zuul.opendev.org/t/openstack/build/7be418fbcea945f0aa54f26fadeb143b : NODE_FAILURE Node request 300-0021636587 failed in 0s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"da2dac7bd58adea634f9e480d2c133171ac6f379","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2023-07-13 16:28:44.000000000","message":"Patch Set 1: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"35b7260024a6898cd054795e115a48e13676167f","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2023-10-11 10:34:36.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"d969739220594adc23667a848fcdbf39e3d81a9c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-11 10:35:24.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"c7bca92fd785befce6a720387084fa59f8386d0f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-11 13:01:17.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/52dae81807134874b133885d0dcd8569\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/44a18869f03c4626b5a4e2e6ce41b23a : SUCCESS in 4m 26s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/d85228393e8f41f09c69809004532974 : SUCCESS in 5m 04s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/df398f562d3340f38e27f3fb49c82e3b : SUCCESS in 6m 04s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/15a6276c75e54177879808673ab4ba9a : SUCCESS in 13m 18s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/ee843afe54d04370b7b730c2405c2599 : SUCCESS in 6m 10s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/e6cd74f9deea4015af702ef116d0902a : SUCCESS in 5m 24s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/fb612e155670454a99c96a3dcbece201 : SUCCESS in 9m 15s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/59d22961b6e54622afd39b51a973f814 : SUCCESS in 1h 25m 50s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/b03ee73d56514dba9444cc56d8e852ee : SUCCESS in 1h 49m 51s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/f711bc4dd8a34a4d86a20af6928e0ed6 : SUCCESS in 36m 13s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/9539fa40dcd848baafcb2cf4ffe9014d : SUCCESS in 41m 32s","accounts_in_message":[],"_revision_number":1},{"id":"9dd42a4ab9386c722d055980235885f6b7c987a4","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-11 13:01:17.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"efbf4c630341736beb4c5aceee4e794508d629bd","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-11 13:03:33.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8cb421d5c1ea4d77b9e7bc16b81560d5\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4471d954cee94e69bfc56bd4f8081cc0 : SUCCESS in 1m 48s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/1eb2f23c6d734efd8524f884c3a288db : SUCCESS in 54s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"b293762ab6a9553edfeff07cfc59c2b37404de13","revisions":{"b293762ab6a9553edfeff07cfc59c2b37404de13":{"kind":"REWORK","_number":1,"created":"2023-07-07 08:10:41.000000000","uploader":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"ref":"refs/changes/22/887922/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/22/887922/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/22/887922/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/22/887922/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/22/887922/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/22/887922/1"}}},"commit":{"parents":[{"commit":"7d1df086ba45434112afa0b71675ec8cf5ff14ae","subject":"Merge \"Enable taskflow retry feature when waiting for compute\" into stable/xena","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/7d1df086ba45434112afa0b71675ec8cf5ff14ae"}]}],"author":{"name":"Maximilian Stinsky","email":"maximilian.stinsky@gec.io","date":"2021-12-21 08:38:21.000000000","tz":60},"committer":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2023-07-07 08:10:36.000000000","tz":120},"subject":"fix haproxy_count to display the number of running processes","message":"fix haproxy_count to display the number of running processes\n\nAt the moment haproxy_count always returns 0 because the _count_haproxy_processes function checks for listener id\u0027s instead of haproxy loadbalancer id\u0027s\n\nChange-Id: I1d1a9a4f193f3766df12ae43b2c2e54c4adf405f\n(cherry picked from commit 33a3d4744c5945f1f99615d85ce26fc63257e09c)\n(cherry picked from commit 24024fb26614415f1028285e6797e14ffc993889)\n(cherry picked from commit 09fb252910f98c4a68813e78c1f1f38db4a0c391)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/b293762ab6a9553edfeff07cfc59c2b37404de13"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/b293762ab6a9553edfeff07cfc59c2b37404de13"}]},"branch":"refs/heads/stable/xena"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
