)]}'
{"id":"openstack%2Foctavia~887925","triplet_id":"openstack%2Foctavia~stable%2Fyoga~Iecd466130ef3941481b4ad79ff6fa60efbd0e5ad","project":"openstack/octavia","branch":"stable/yoga","attention_set":{},"removed_from_attention_set":{"29244":{"account":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"last_update":"2024-01-23 16:03:29.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Iecd466130ef3941481b4ad79ff6fa60efbd0e5ad","subject":"Fix job book deletion issue in taskflow","status":"MERGED","created":"2023-07-07 08:11:58.000000000","updated":"2024-01-23 16:05:43.000000000","submitted":"2024-01-23 16:03:29.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"887925","meta_rev_id":"7df8c212094f5e34245c8cf0bfc2810f131da600","_number":887925,"virtual_id_number":887925,"owner":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-01-23 16:03:29.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"all":[{"value":2,"date":"2023-07-13 16:31:02.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2024-01-23 11:03:03.000000000","_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"all":[{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2024-01-23 11:03:03.000000000","_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-07-07 11:29:16.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-07-13 16:31:02.000000000","updated_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"reviewer":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"state":"REVIEWER"},{"updated":"2024-01-23 11:03:03.000000000","updated_by":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"reviewer":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"state":"REVIEWER"}],"messages":[{"id":"3c61c81286398caf41e29cc699f55e23d367fdd7","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2023-07-07 08:11:58.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"cb91c7d0ad02e7c5be52fc1bd5be7e7458d02305","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-07 11:29:16.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/ecb66ff8722443c09e1086022384bc9d\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/6d486a4b46a8478da9c6bf5474cf1d63 : SUCCESS in 11m 46s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/8707a22b431b4df59b93d30a256dda24 : SUCCESS in 8m 12s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2ac7f76f23324bc296f09cc387efa43f : SUCCESS in 6m 30s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/52c19244fa094664bf8a70a2e5c4d640 : SUCCESS in 4m 29s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/d3436187dc3a4cd880d6abdc60a0c23a : SUCCESS in 6m 10s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/31e58adc6ea64e82978bd04c28bee66c : SUCCESS in 11m 32s\n- octavia-tox-py39-tips https://zuul.opendev.org/t/openstack/build/90951cbfd0dc48bd8d08fc6eae1e6664 : SUCCESS in 5m 55s\n- octavia-tox-functional-py39-tips https://zuul.opendev.org/t/openstack/build/f4c1d2464e824e72822c2f36c8bd19fc : SUCCESS in 6m 19s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/faec917e903044299630377c0b03bfad : SUCCESS in 6m 12s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/0213abd73d524996afd4afc0b85be130 : SUCCESS in 9m 56s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/6e43322f7a114842ae048e17c2a2ffb5 : SUCCESS in 1h 22m 32s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/fc6fedc493d948ffb8de52c82f9968cf : SUCCESS in 2h 27m 58s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/a1ce3ea5287c46fda039a64dd56509fb : SUCCESS in 33m 20s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/6a64ee0073be4212a8928e4f0af7141e : TIMED_OUT in 3h 08m 33s\n- octavia-v2-dsvm-tls-barbican-fips https://zuul.opendev.org/t/openstack/build/bf62f5bf4a3e45c89c388c83651639ed : FAILURE in 13m 16s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/5187c62b51ec4653addbc0625b568d65 : SUCCESS in 29m 02s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/a0bf430bdf224e968adcfd334d2e2df7 : SUCCESS in 2h 21m 57s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/833fbd6c05bd4f4b9e9427bdc86a8879 : FAILURE in 9m 48s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/a470c248171340a1ad8f838bf7fd2c0c : FAILURE in 5m 46s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/c6afd6170d394af2b3e39a455ee887ea : TIMED_OUT in 2h 32m 39s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/fdfd6a56066f45ad880cd2ed7d337377 : SUCCESS in 1h 45m 45s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8-stream https://zuul.opendev.org/t/openstack/build/b8dc14301e764ffe86e1bb97613cc824 : NODE_FAILURE Node request 300-0021636660 failed in 0s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"ad8a91e4d0ad35579455d04984ad41f1b16f8089","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2023-07-13 16:31:02.000000000","message":"Patch Set 1: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"dc8e2d08280257f6311f165af56db4e48de22848","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-13 18:27:22.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5030e7f04d5546b3a99fad85c91428ba\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/09a95431910945c89a562d7acee00db1 : SUCCESS in 10m 22s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/975e00e2a75c46bc9ec9d4621a67f56a : SUCCESS in 7m 27s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/30db6ccf5fb24d24b0f83a2e96d30f0b : SUCCESS in 6m 34s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/f9394edb5bbb421f98bdf7a1ece7288b : SUCCESS in 5m 33s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/2b113032a2b54a7f9ce13b214c6ab75b : SUCCESS in 4m 23s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7dac350e33f2493db5c7f131f8bb3b71 : SUCCESS in 9m 49s\n- octavia-tox-py39-tips https://zuul.opendev.org/t/openstack/build/8aea64ccef634865846afe5edd4d6b67 : SUCCESS in 5m 33s\n- octavia-tox-functional-py39-tips https://zuul.opendev.org/t/openstack/build/be81a0cb17424232b420a55df4bf84ab : SUCCESS in 5m 29s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/49fd76e9fe3b41deabaeec050230a2f9 : SUCCESS in 5m 02s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/30f5eaf3199f4ab7931f168388d77851 : SUCCESS in 7m 51s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/369e77b46e454d33af9cf7e70e2b00e6 : SUCCESS in 1h 34m 29s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/c1aabac66e27465395655dc323319a24 : SUCCESS in 1h 39m 00s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/6eb0b005d2094b6da89334f0067ef2ec : SUCCESS in 39m 30s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/f07d2effadff47bd880421a4ed38cb2c : SUCCESS in 39m 29s\n- octavia-v2-dsvm-tls-barbican-fips https://zuul.opendev.org/t/openstack/build/65f49e99e2f64410af8eca02d00348c0 : FAILURE in 15m 25s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/f407a5cbb8924d94989cc5f0cfb2d278 : SUCCESS in 31m 55s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/4e0738c31e7640dbbc21413a01e053b3 : SUCCESS in 1h 53m 37s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/920657a56408407e88c9760b02ad57e5 : FAILURE in 12m 22s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/ec12099af31c49eca6b81e7cc8152731 : FAILURE in 8m 08s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/efb4735242304565829df05e48c1bc15 : SUCCESS in 1h 33m 31s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/c6c4b58677bf4c7c9b70e0ff277b0f04 : SUCCESS in 1h 37m 13s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8-stream https://zuul.opendev.org/t/openstack/build/531989dd323d48408d8bdb155d4f91fd : FAILURE in 41m 15s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"62e68f953e3ffe66175c0c893002f7ad3d9ba6de","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2024-01-23 11:03:03.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"90854b9ec239dd166e1b1dc4fde6c5d6539ad273","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-23 11:04:42.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"e05965042292a6e5abbf433262fe65d377971600","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-23 16:03:29.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f6949d461ac145a289eab6a833c0d29b\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/93791773d8a44dfaaf211af724ac2b00 : SUCCESS in 7m 39s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/3a1a1a809a824103a05663470cfa4b24 : SUCCESS in 5m 10s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/538068a91efc4903b30fbc9752968fcc : SUCCESS in 6m 04s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7db97f14666d442d939f4d050be012f6 : SUCCESS in 11m 16s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/3c08a1df0ed743c98ddb35ff3656b0d8 : SUCCESS in 5m 20s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/991815b0fcf84284a723a5cb7913c8a0 : SUCCESS in 8m 05s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/bc988687e7a4467d9051dc28dfcb1e5c : SUCCESS in 1h 19m 24s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/01d6f43d81b840578c7e3035c889fe23 : SUCCESS in 1h 33m 36s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/75303403843a4fb1a0f04d1434201be7 : SUCCESS in 41m 58s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/529ff85fbe3348ce8879fae19b181ca8 : SUCCESS in 45m 05s","accounts_in_message":[],"_revision_number":1},{"id":"96aeca71d33c2b5f8527f4d2815c7bcd585e703f","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-23 16:03:29.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"7df8c212094f5e34245c8cf0bfc2810f131da600","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-23 16:05:43.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/fcc808775a1542c8ab1fb5c7006d599f\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/27d5ac2a959548f7bab62d505d8dedf3 : SUCCESS in 50s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"d410c510e16dddd0bc57ef127d3c5986e27267dc","revisions":{"d410c510e16dddd0bc57ef127d3c5986e27267dc":{"kind":"REWORK","_number":1,"created":"2023-07-07 08:11:58.000000000","uploader":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"ref":"refs/changes/25/887925/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/25/887925/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/25/887925/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/25/887925/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/25/887925/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/25/887925/1"}}},"commit":{"parents":[{"commit":"5fc80991a4779a7474e1d64f993bbefeb5338fd1","subject":"Merge \"Remove incorrect info message\" into stable/yoga","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/5fc80991a4779a7474e1d64f993bbefeb5338fd1"}]}],"author":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2021-11-04 20:17:39.000000000","tz":60},"committer":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2023-07-07 08:11:55.000000000","tz":120},"subject":"Fix job book deletion issue in taskflow","message":"Fix job book deletion issue in taskflow\n\nSome jobs unattached to a book might have been scheduled in taskflow,\ntriggering an error. On completion, skip the logbook deletion if it\ndoesn\u0027t exist.\n\nChange-Id: Iecd466130ef3941481b4ad79ff6fa60efbd0e5ad\n(cherry picked from commit 54e1f6f118ff1523df7737ef42f29b151c313943)\n(cherry picked from commit b9f01c7dacfc0ab6076325fe2459b41b790af46b)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/d410c510e16dddd0bc57ef127d3c5986e27267dc"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/d410c510e16dddd0bc57ef127d3c5986e27267dc"}]},"branch":"refs/heads/stable/yoga"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
