)]}'
{"id":"openstack%2Foctavia~887929","triplet_id":"openstack%2Foctavia~stable%2Fyoga~Iba07010b63771aafa1db224dbd0e0c97925d6173","project":"openstack/octavia","branch":"stable/yoga","attention_set":{},"removed_from_attention_set":{"29244":{"account":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"last_update":"2023-10-17 18:15:27.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Iba07010b63771aafa1db224dbd0e0c97925d6173","subject":"Fix typo in lvs-masquerade.sh","status":"MERGED","created":"2023-07-07 08:13:03.000000000","updated":"2023-10-17 18:18:32.000000000","submitted":"2023-10-17 18:15:27.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"887929","meta_rev_id":"8d8fcb0d699f3da078351db634344d74fa7f539b","_number":887929,"virtual_id_number":887929,"owner":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-10-17 18:15:27.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"all":[{"value":2,"date":"2023-07-13 16:25:54.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2023-10-17 15:56:30.000000000","_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"all":[{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2023-10-17 15:56:30.000000000","_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-07-07 10:51:14.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-07-13 16:25:54.000000000","updated_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"reviewer":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"state":"REVIEWER"},{"updated":"2023-10-17 15:56:30.000000000","updated_by":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"reviewer":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"state":"REVIEWER"}],"messages":[{"id":"733b40f8c0ce1637743ae02569809aefd260cdba","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2023-07-07 08:13:03.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"1f846acd8e4c998c85d5340e900e80699285c335","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-07 10:51:14.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/365b505bed324623a05c5ddf66427662\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/470ac2287e004b62a387dce4ed339c4b : SUCCESS in 10m 52s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/df637eaa71244c12ac006729d5645569 : SUCCESS in 7m 52s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e1d9d4d2414a4bf3bb8d4f11cff979c1 : SUCCESS in 6m 32s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/81e0b85fdeb3404ebe1735c86f907c05 : SUCCESS in 5m 01s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/7bf85f2ad6184904ac148574f0b781f9 : SUCCESS in 5m 09s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/89be6673e5c14cf1980a4344d5c72dd8 : SUCCESS in 11m 10s\n- octavia-tox-py39-tips https://zuul.opendev.org/t/openstack/build/d7f626ae17bd4421b3a609fa526ca3a6 : SUCCESS in 5m 19s\n- octavia-tox-functional-py39-tips https://zuul.opendev.org/t/openstack/build/f68e77c194654a4b83b3ace1f2671909 : SUCCESS in 5m 19s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/3acead57d2c940c183b9fb6c6a392ebd : SUCCESS in 5m 05s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/24132158a25e4694a5fafca5f3414106 : SUCCESS in 7m 27s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/3020ba5b6cea4321baf4989ed40e31b5 : SUCCESS in 1h 40m 52s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/a6075ebe9c77464cb0726eefeaa30ab8 : SUCCESS in 2h 27m 34s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/e70babe0993048c3bda63b221e6020d4 : SUCCESS in 32m 58s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/4d874dc502bd47f0a15c4c99edfde9b9 : SUCCESS in 37m 04s\n- octavia-v2-dsvm-tls-barbican-fips https://zuul.opendev.org/t/openstack/build/d64895349d65480b93b6959956a69a39 : FAILURE in 11m 56s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/302ba6fe7d7843e2bb24e82f7d12c389 : SUCCESS in 41m 10s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/9588bdc069614ca1b020d97aedfbd5d6 : SUCCESS in 1h 49m 38s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/026cccc8cd754e379866c42486c069a6 : FAILURE in 9m 35s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/3034aa79c6094fcc8218d7a5bca8def8 : FAILURE in 9m 00s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/f9fcef38b2974f639059a71206878921 : TIMED_OUT in 2h 33m 42s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/c7d130384b1043f09ee9486f87b0f84c : SUCCESS in 1h 40m 22s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8-stream https://zuul.opendev.org/t/openstack/build/2ec6311d89404d73b9632863db2bb88e : FAILURE in 40m 38s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"0ecf31e7c2f34f209f43d5813e039517a8b79386","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2023-07-13 16:25:54.000000000","message":"Patch Set 1: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"12b32e26e11e44629ab0b41bc9492e2eb1d37f44","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2023-10-17 15:56:30.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"22004ff8750f9181a3d62e0f03cbbb4171fb8b09","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-17 15:57:44.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"2913a94aaf4165d11e6a50094e14d9c6db5febf1","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-17 18:15:27.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/db3c394a4b7c46d49a098e6f57186d17\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/06222d3a1972467eaed19e8ec48da4d6 : SUCCESS in 6m 49s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/ea8b8ad370d8472faf13979305299fc6 : SUCCESS in 6m 58s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/0db71dbdb8604d37b9cd359d1a2d385d : SUCCESS in 5m 55s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/90bc6b0fd72548c38d636dcd1ea51c38 : SUCCESS in 11m 26s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/09a5285c533846fdb0ffb34462cc874a : SUCCESS in 5m 57s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/a1002cfd6b324dc3aa73ab4d940ff01c : SUCCESS in 8m 56s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/5080d9e200164e6095878531ba88c31c : SUCCESS in 1h 24m 13s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/c65d8b6060534584902ae7fd585fa3d3 : SUCCESS in 1h 49m 27s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/57ca70e67f754729a13a13b3348bd3fc : SUCCESS in 48m 56s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/ef4a1d379c8a44a2be35a82cb893e55a : SUCCESS in 46m 21s","accounts_in_message":[],"_revision_number":1},{"id":"2ea12252b4f59a560e734842d73318f7d31efae3","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-17 18:15:27.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"8d8fcb0d699f3da078351db634344d74fa7f539b","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-17 18:18:32.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/bf372db904ae433e841bad1145b5f1fd\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8ca9a1152422405f9b025ad1e7a4ec0d : SUCCESS in 1m 13s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"16578229f470cd807614f293bf06bce4340aea7d","revisions":{"16578229f470cd807614f293bf06bce4340aea7d":{"kind":"REWORK","_number":1,"created":"2023-07-07 08:13:03.000000000","uploader":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"ref":"refs/changes/29/887929/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/29/887929/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/29/887929/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/29/887929/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/29/887929/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/29/887929/1"}}},"commit":{"parents":[{"commit":"5fc80991a4779a7474e1d64f993bbefeb5338fd1","subject":"Merge \"Remove incorrect info message\" into stable/yoga","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/5fc80991a4779a7474e1d64f993bbefeb5338fd1"}]}],"author":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2022-08-19 15:28:34.000000000","tz":120},"committer":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2023-07-07 08:13:00.000000000","tz":120},"subject":"Fix typo in lvs-masquerade.sh","message":"Fix typo in lvs-masquerade.sh\n\nA typo in the lvs-masquerade script prevented the deletion of the\n\u0027ip6-udp-masq\u0027 nftables chain when stopping a network interface.\n\nThe name of the chain should be \u0027ip6-udp-masq\u0027 instead of \u0027ip-udp-masq\u0027\n\nThe bug had no impacts on the loadbalancers but it displayed an error\nmessage in the logs.\n\nChange-Id: Iba07010b63771aafa1db224dbd0e0c97925d6173\n(cherry picked from commit 55e6001ea4db3e30dd3dec2b8b2bfb2d9b1ac103)\n(cherry picked from commit c014bb7b7e58c859bfef2dfb8c8187db9118d8f8)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/16578229f470cd807614f293bf06bce4340aea7d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/16578229f470cd807614f293bf06bce4340aea7d"}]},"branch":"refs/heads/stable/yoga"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
