)]}'
{"id":"openstack%2Foctavia~887930","triplet_id":"openstack%2Foctavia~stable%2Fxena~Iba07010b63771aafa1db224dbd0e0c97925d6173","project":"openstack/octavia","branch":"stable/xena","attention_set":{},"removed_from_attention_set":{"29244":{"account":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"last_update":"2023-10-17 18:15:24.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Iba07010b63771aafa1db224dbd0e0c97925d6173","subject":"Fix typo in lvs-masquerade.sh","status":"MERGED","created":"2023-07-07 08:13:14.000000000","updated":"2023-10-17 18:18:39.000000000","submitted":"2023-10-17 18:15:24.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"887930","meta_rev_id":"3284f47e9f8eb3c1d12663d5ca36f8f6910ee449","_number":887930,"virtual_id_number":887930,"owner":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-10-17 18:15:24.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"all":[{"value":2,"date":"2023-07-13 16:25:39.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2023-10-17 15:56:09.000000000","_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"all":[{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2023-10-17 15:56:09.000000000","_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-07-07 10:54:50.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-07-13 16:25:39.000000000","updated_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"reviewer":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"state":"REVIEWER"},{"updated":"2023-10-17 15:56:09.000000000","updated_by":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"reviewer":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"state":"REVIEWER"}],"messages":[{"id":"692fd48efe9f847bf80b857174b1eac124072bb3","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2023-07-07 08:13:14.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"6c46cffde375d7f1899cda42aa8256cd14093064","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-07 10:54:50.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5b55dbc8146c4713a2c8a90f7ccc63cc\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/d39a6ae7fb204083a1011157a08e7f37 : SUCCESS in 10m 35s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/2859ce8354324355b64314de45715c26 : SUCCESS in 8m 05s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c07d46d70683407e9e9792612248531b : SUCCESS in 5m 16s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/2d9923aac760467490d2c88d991ba15b : SUCCESS in 4m 47s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/fb20f3fa6b074dd58bc7d938ab8941b0 : SUCCESS in 4m 41s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/aa2a2adc77884b79830997c30f5e59af : SUCCESS in 5m 12s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/53e76aea3a174bd3ae15cfb95673a3f1 : SUCCESS in 13m 12s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/5b6c72e754f64167882fbbae70176a30 : SUCCESS in 5m 41s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/89f916307d38490488c52e5d32b9461f : SUCCESS in 8m 44s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/04c6b7f73b14464581b439a250c61894 : SUCCESS in 5m 01s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/e2d807dcc4d342fbbf48addcf69d670e : SUCCESS in 8m 12s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/1958a86a0b094203b0caf5db0b5acc24 : SUCCESS in 1h 29m 37s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/585ca31d227f439bb1c76d2b12c29c7c : SUCCESS in 1h 36m 05s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/3c904e41cb884efcbdecaccbb4893c55 : SUCCESS in 39m 36s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/f5b5454b4a084eacb2c7c32edf65b650 : SUCCESS in 44m 32s\n- octavia-v2-dsvm-tls-barbican-fips https://zuul.opendev.org/t/openstack/build/2dd180250c3e42c699e85f8f4b98e4a4 : FAILURE in 9m 58s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/d9e18367173f4b5cb51db09399f0f05a : SUCCESS in 35m 40s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/8c4239d32a6840368449c6edfd555856 : SUCCESS in 2h 31m 49s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/ba9bcc4a958b4ab4bc159c922854b1c5 : FAILURE in 11m 29s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/6a1996dc3af643bcbebbff168eec9c28 : FAILURE in 5m 31s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/439403375452406d8a7025260ecda9cf : TIMED_OUT in 2h 34m 41s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/cc4c663681a24e2697d7b2953cdc0050 : SUCCESS in 2h 28m 22s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8 https://zuul.opendev.org/t/openstack/build/90c89071f1084b3ea71bee96cfb02a8c : NODE_FAILURE Node request 300-0021636776 failed in 0s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"0370bea21ad5d87a78b419bb1994cca3defb7c2b","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2023-07-13 16:25:39.000000000","message":"Patch Set 1: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"14f1774df22db2542934c690da444e1ee7632b20","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2023-10-17 15:56:09.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"a467c7f9ffcd54569782d5078d6ef22576986188","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-17 15:57:13.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"c7c95f713234b2bf703b82cbed4f79ea2af17666","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-17 18:15:24.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1a00e876dd7b43c3a372a0f95f820cae\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/17cb7f823e9b42f281f61dce7d7b3d93 : SUCCESS in 6m 00s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/e6f840edcdb447d28f623b99b143fff9 : SUCCESS in 5m 08s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/252bab93d8ac44be8b208665ef08f3c8 : SUCCESS in 5m 44s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6806e8d073b34193a6eb580fd212daa8 : SUCCESS in 13m 32s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/2002a21129224b87b3054c150cbccf5a : SUCCESS in 5m 35s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/1b4dfe17da8649c4aecee265126be4e3 : SUCCESS in 8m 20s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/c659943f0406409eb60e8aad8d55d5a4 : SUCCESS in 1h 20m 28s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/ee2521df38fb4e23be7148dac0dd2079 : SUCCESS in 2h 09m 04s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/cff4c8fbaa324877adff416d8c4696c7 : SUCCESS in 39m 35s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/0974c95c0a1047aaa1e982a9f556eb4f : SUCCESS in 37m 28s","accounts_in_message":[],"_revision_number":1},{"id":"1a65bf644176bf6346c5368ed5a0a6bc8e7e34e7","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-17 18:15:24.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"3284f47e9f8eb3c1d12663d5ca36f8f6910ee449","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-17 18:18:39.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/318c1525cf0f4171b68c2c3f6fde9a21\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4310996841a9467dad35776799869529 : SUCCESS in 1m 08s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"b24f225536fade99cb7e5f0f7c5bfb37cbac366b","revisions":{"b24f225536fade99cb7e5f0f7c5bfb37cbac366b":{"kind":"REWORK","_number":1,"created":"2023-07-07 08:13:14.000000000","uploader":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"ref":"refs/changes/30/887930/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/30/887930/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/30/887930/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/30/887930/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/30/887930/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/30/887930/1"}}},"commit":{"parents":[{"commit":"7d1df086ba45434112afa0b71675ec8cf5ff14ae","subject":"Merge \"Enable taskflow retry feature when waiting for compute\" into stable/xena","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/7d1df086ba45434112afa0b71675ec8cf5ff14ae"}]}],"author":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2022-08-19 15:28:34.000000000","tz":120},"committer":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2023-07-07 08:13:10.000000000","tz":120},"subject":"Fix typo in lvs-masquerade.sh","message":"Fix typo in lvs-masquerade.sh\n\nA typo in the lvs-masquerade script prevented the deletion of the\n\u0027ip6-udp-masq\u0027 nftables chain when stopping a network interface.\n\nThe name of the chain should be \u0027ip6-udp-masq\u0027 instead of \u0027ip-udp-masq\u0027\n\nThe bug had no impacts on the loadbalancers but it displayed an error\nmessage in the logs.\n\nChange-Id: Iba07010b63771aafa1db224dbd0e0c97925d6173\n(cherry picked from commit 55e6001ea4db3e30dd3dec2b8b2bfb2d9b1ac103)\n(cherry picked from commit c014bb7b7e58c859bfef2dfb8c8187db9118d8f8)\n(cherry picked from commit 16578229f470cd807614f293bf06bce4340aea7d)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/b24f225536fade99cb7e5f0f7c5bfb37cbac366b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/b24f225536fade99cb7e5f0f7c5bfb37cbac366b"}]},"branch":"refs/heads/stable/xena"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
