)]}'
{"id":"openstack%2Foctavia~888530","triplet_id":"openstack%2Foctavia~master~I1c51a5d6d6593108f431809868dcdb3d4142df45","project":"openstack/octavia","branch":"master","attention_set":{},"removed_from_attention_set":{"34307":{"account":{"_account_id":34307,"name":"Tianhao Chen","email":"840738804@qq.com","username":"tianhao.chen"},"last_update":"2023-08-23 22:36:03.000000000","reason":"Change was submitted"},"29244":{"account":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"last_update":"2023-07-19 08:39:44.000000000","reason":"\u003cGERRIT_ACCOUNT_29244\u003e replied on the change","reason_account":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"}}},"hashtags":[],"change_id":"I1c51a5d6d6593108f431809868dcdb3d4142df45","subject":"Fix IPv6 session persistence failed","status":"MERGED","created":"2023-07-14 09:25:49.000000000","updated":"2023-08-23 22:37:23.000000000","submitted":"2023-08-23 22:36:03.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":5,"unresolved_comment_count":1,"has_review_started":true,"submission_id":"888530","meta_rev_id":"8d31f8dddf24742944b743048c1a473d430a0a06","_number":888530,"virtual_id_number":888530,"owner":{"_account_id":34307,"name":"Tianhao Chen","email":"840738804@qq.com","username":"tianhao.chen"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-08-23 22:36:03.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2023-07-19 08:39:44.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":2,"date":"2023-08-23 21:01:03.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":1,"date":"2023-08-23 21:01:03.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"approved":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2023-07-19 08:39:44.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-07-14 12:20:03.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-07-18 08:55:52.000000000","updated_by":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"reviewer":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"state":"CC"},{"updated":"2023-07-19 07:28:19.000000000","updated_by":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"reviewer":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"state":"REVIEWER"},{"updated":"2023-08-23 21:01:03.000000000","updated_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"reviewer":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"state":"REVIEWER"}],"messages":[{"id":"25c8694840b0edbc6f8592b0d6b4b90673348ca7","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34307,"name":"Tianhao Chen","email":"840738804@qq.com","username":"tianhao.chen"},"date":"2023-07-14 09:25:49.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"0ffb2df6a6a6c31ae42c13ebd203798ae49671e6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-14 12:20:03.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7c3369a14bc94d0796746173a92328b1\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/04badf1a1e2f4adaba9bf11e7523f709 : SUCCESS in 10m 52s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5d54b097e8ac411db082995c4efd5ad0 : SUCCESS in 5m 23s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/c43005ce5e6d45a58b38631fbe09e56b : SUCCESS in 4m 55s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/c603c5b0dbae4bc19619bcd4468be632 : SUCCESS in 5m 02s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/29867114935f473d9e32caf8b89c2252 : SUCCESS in 4m 50s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/75311267874b44f3a29d79a6b1a8baf5 : SUCCESS in 9m 57s\n- octavia-tox-py39-tips https://zuul.opendev.org/t/openstack/build/a7cb43c3b641431c8742a7e452893eb8 : SUCCESS in 4m 29s\n- octavia-tox-functional-py39-tips https://zuul.opendev.org/t/openstack/build/4c46cd39f3984eb98f2d9399972f4b54 : SUCCESS in 8m 22s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/9f81053abdec44c2b78223998f143fe8 : SUCCESS in 4m 47s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/740fd1e33b404104b41d24479fafbb09 : SUCCESS in 8m 58s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/4f1a8be85fc1444d9a3a14d5051da46b : SUCCESS in 6m 31s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/804c3907230e4029856fecf90729debb : SUCCESS in 7m 02s (non-voting)\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/689d544f96d94b59be780fc629c17703 : SUCCESS in 1h 38m 26s\n- octavia-v2-dsvm-scenario-traffic-ops https://zuul.opendev.org/t/openstack/build/93772e7511c241a683126ad8772c2578 : SUCCESS in 1h 45m 35s\n- octavia-v2-dsvm-scenario-non-traffic-ops https://zuul.opendev.org/t/openstack/build/eadc184606614940b636fdc1dee5b11d : SUCCESS in 1h 06m 45s\n- octavia-v2-dsvm-scenario-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/f5958699799547919c34513c2d84f558 : SUCCESS in 1h 36m 48s\n- octavia-v2-dsvm-scenario-non-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/47057840263246d3a6f61a4792779932 : SUCCESS in 1h 23m 33s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/ecead9b7e99e4cc9a20ac9aea478391b : SUCCESS in 35m 35s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/27144010fefd49dd8396a4f4d80f7a30 : SUCCESS in 43m 47s\n- octavia-v2-dsvm-tls-barbican-fips https://zuul.opendev.org/t/openstack/build/80b03d66b8d1469f965394664db366f4 : SUCCESS in 56m 07s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/f722e862806b417e80c22e7627f84a12 : SUCCESS in 32m 20s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/2065d57868214f47a3b5a6ede5d293fb : TIMED_OUT in 2h 32m 51s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/fad5da7532e24cdeaebce7e2936d96ad : SUCCESS in 2h 52m 54s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/ae520e933199437088720a951674324a : TIMED_OUT in 2h 32m 56s (non-voting)\n- octavia-v2-dsvm-scenario-centos-9-stream https://zuul.opendev.org/t/openstack/build/519072d1f351461aa0a56bab165c2064 : TIMED_OUT in 2h 34m 28s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"7594857c63600152caac650f98edfc031666c01a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34307,"name":"Tianhao Chen","email":"840738804@qq.com","username":"tianhao.chen"},"date":"2023-07-17 02:36:40.000000000","message":"Uploaded patch set 2: Commit message was updated.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":2},{"id":"4bd5ca265f35588cab73d4478e6f1ea5718f1496","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34307,"name":"Tianhao Chen","email":"840738804@qq.com","username":"tianhao.chen"},"date":"2023-07-17 03:07:33.000000000","message":"Uploaded patch set 3: Commit message was updated.","accounts_in_message":[],"_revision_number":3},{"id":"c851e4ea89867f35d06d0584de7cb85f32987b09","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-17 06:09:16.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/20bcc19025974c3da5119e3ffc8e8610\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/b427a6a4c1f5487d809fbf92a3f23ee9 : SUCCESS in 12m 07s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fc953c0213794c4fba8e0a4ebb7d8628 : SUCCESS in 5m 39s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/715f26bac171437da62abbbbb680e19c : SUCCESS in 5m 35s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/f8d4b569f88e42eb8ef41194883e43e4 : SUCCESS in 5m 23s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/2a295336c2e644189563e1e323fd979a : SUCCESS in 4m 41s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0f0aed5b88a24ef5805af7231280d616 : SUCCESS in 10m 18s\n- octavia-tox-py39-tips https://zuul.opendev.org/t/openstack/build/6232de9c32d64425a7705f42e1a57b5f : SUCCESS in 6m 24s\n- octavia-tox-functional-py39-tips https://zuul.opendev.org/t/openstack/build/4f822558ff144164aa170b043fb8e697 : SUCCESS in 7m 47s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/a3fecbc5243f4c958508ebe45f0113e5 : SUCCESS in 5m 31s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/c4d82007ba444560b4d8afd961fcae02 : SUCCESS in 8m 22s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/88767cd827e34a86bc197a36ce21c0c6 : SUCCESS in 8m 03s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/0c6c217b835a42f6aa247db699e00643 : SUCCESS in 8m 52s (non-voting)\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/8129ddfa6c9d4f7b85e80a5c9c878841 : SUCCESS in 1h 10m 33s\n- octavia-v2-dsvm-scenario-traffic-ops https://zuul.opendev.org/t/openstack/build/58492cfa10b14febb4e2edbe98f9d8ad : SUCCESS in 1h 57m 29s\n- octavia-v2-dsvm-scenario-non-traffic-ops https://zuul.opendev.org/t/openstack/build/b6e50c8dbcc44375bf52c6e9cf3ab32e : SUCCESS in 1h 32m 48s\n- octavia-v2-dsvm-scenario-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/b8102daf451947dc9a2f54d7b2a72145 : SUCCESS in 2h 08m 17s\n- octavia-v2-dsvm-scenario-non-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/ae9c3059d70f4470889ce43d69df8af5 : SUCCESS in 1h 26m 03s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/effbac65b2c44317aee874d5c47a3850 : SUCCESS in 43m 20s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/400ce701ce2f415fb01f610f8ed55c9c : SUCCESS in 38m 25s\n- octavia-v2-dsvm-tls-barbican-fips https://zuul.opendev.org/t/openstack/build/577a12ef4061401aa837552c8217f064 : SUCCESS in 1h 07m 21s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/785950074e7e4c999495344c67a59ab1 : SUCCESS in 34m 11s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/df1ce986e7f94165b717d3c8440d03bf : TIMED_OUT in 2h 32m 49s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/08e8c491245e4aa1af90832ce80b20c9 : SUCCESS in 3h 00m 15s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/211cba3beac44e3ba450ece37a52936c : TIMED_OUT in 2h 32m 57s (non-voting)\n- octavia-v2-dsvm-scenario-centos-9-stream https://zuul.opendev.org/t/openstack/build/37bd994b309a4df29cc03785ae4189c3 : SUCCESS in 2h 14m 55s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"8aecff882abf20412cc0f48eed01d2eeeabc23a2","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34307,"name":"Tianhao Chen","email":"840738804@qq.com","username":"tianhao.chen"},"date":"2023-07-17 06:31:59.000000000","message":"Uploaded patch set 4: Commit message was updated.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":4},{"id":"82b506fdd459b4544414ba4daf0520c9ed02213b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-17 09:37:00.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/244660d7b834412286992db32d89fa0a\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/60e7ae30e9234e969753db4d1a79d820 : SUCCESS in 10m 57s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bd30717f8d2941efb8598289eb0d5a3b : SUCCESS in 5m 32s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/364dd3e1197f46999c787a1a40391010 : SUCCESS in 4m 56s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/feaa2d43d721495bbd7fe4a290402267 : SUCCESS in 4m 37s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/7df5037f188347199504b094b0e2355f : SUCCESS in 4m 46s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/055ac798d5ff494897cca8e9ac128473 : SUCCESS in 8m 52s\n- octavia-tox-py39-tips https://zuul.opendev.org/t/openstack/build/d427a34116b849659060d518c3536b6c : SUCCESS in 4m 57s\n- octavia-tox-functional-py39-tips https://zuul.opendev.org/t/openstack/build/7ac64ce582e3465795000d44a59ae1e1 : SUCCESS in 7m 44s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/f1ac42063d90421bbc43c7ab3777cae7 : SUCCESS in 4m 37s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/1498ab5f4d614540b21ca7b04254b5a7 : SUCCESS in 6m 43s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/ef0510e367874067a26cc94be3761c6f : SUCCESS in 7m 23s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/6287c5b4231541f5803833aa600cd71e : SUCCESS in 8m 08s (non-voting)\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/3635561fcd744444a042035ef5b4c1f6 : SUCCESS in 1h 21m 48s\n- octavia-v2-dsvm-scenario-traffic-ops https://zuul.opendev.org/t/openstack/build/a05aca3ae5b94a27ade348ef93482f91 : SUCCESS in 1h 24m 50s\n- octavia-v2-dsvm-scenario-non-traffic-ops https://zuul.opendev.org/t/openstack/build/c951403ac3c445c4a194d8a3ad3c583c : SUCCESS in 57m 17s\n- octavia-v2-dsvm-scenario-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/f374aa4baafb4aec8fccd3a538a24720 : SUCCESS in 1h 28m 06s\n- octavia-v2-dsvm-scenario-non-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/290c19c582594722bb5f253fda258386 : SUCCESS in 1h 06m 53s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/be4ac3a1565849b0b8b2d26a5a8cc17d : SUCCESS in 35m 00s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/0652e3e1cbba477ebcc27afc7fe8ae00 : SUCCESS in 36m 26s\n- octavia-v2-dsvm-tls-barbican-fips https://zuul.opendev.org/t/openstack/build/d8d838bea79a4defbbe2b7c3e6edf02b : SUCCESS in 57m 38s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/7ed76c47ad434ea0bdecb0fd0744e3c6 : SUCCESS in 36m 37s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/5bdf3717420c4b0dbd1cf31328cdaab2 : TIMED_OUT in 2h 32m 39s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/374879967808405b872c4c15e8511286 : TIMED_OUT in 3h 02m 59s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/427aee5e718a43b78579e96466fb27d0 : TIMED_OUT in 2h 32m 55s (non-voting)\n- octavia-v2-dsvm-scenario-centos-9-stream https://zuul.opendev.org/t/openstack/build/ee03f494396c4bee9f4b3f63af950d07 : SUCCESS in 2h 17m 55s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"394bded763b10ea90b4284e6dd1b7899d4115ffe","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2023-07-18 08:55:52.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"cd5e6db525c9d4876399ddbdc9eccda5af705d52","author":{"_account_id":34307,"name":"Tianhao Chen","email":"840738804@qq.com","username":"tianhao.chen"},"date":"2023-07-18 09:09:56.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"fcb76c1af3c84da1f5b4813fdead971250fbbd1f","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2023-07-19 07:28:19.000000000","message":"Patch Set 4: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"933a4ec434e74721ebf18242977327c514580124","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34307,"name":"Tianhao Chen","email":"840738804@qq.com","username":"tianhao.chen"},"date":"2023-07-19 08:16:55.000000000","message":"Uploaded patch set 5.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":5},{"id":"0a2f5f8eeaa56e9db1d5edac66a3d8eb98dd40c0","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34307,"name":"Tianhao Chen","email":"840738804@qq.com","username":"tianhao.chen"},"date":"2023-07-19 08:30:50.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"7912bd47fc16ed7649da0459d7f954cf1ef5f703","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2023-07-19 08:39:44.000000000","message":"Patch Set 6: Backport-Candidate+2 Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"2fc5facd02b2594fb9be311ef69d5c90331a270a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-19 11:35:08.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3494ee99e95e4c2897db6807457cf147\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/abd7a3dafb20472182379eecb2ffa475 : SUCCESS in 9m 34s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5cca7bbe229c4e6cb36404ccda8808d9 : SUCCESS in 4m 52s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/db5bf020575a40a482e9a2b0d1c1118b : SUCCESS in 4m 15s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/dfa46dec76554188bc5cba3364b0ae8b : SUCCESS in 4m 52s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/26b5fcbf158c4eb9b55aecfe06f36575 : SUCCESS in 4m 46s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c8c52c5999ad4722885fe5cd192bd4fe : SUCCESS in 11m 13s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/6b78abbbc83b454384509ae996a8d1dd : SUCCESS in 5m 29s\n- octavia-tox-py39-tips https://zuul.opendev.org/t/openstack/build/4c425ec00c0248568ab14828e966ea07 : SUCCESS in 4m 31s\n- octavia-tox-functional-py39-tips https://zuul.opendev.org/t/openstack/build/58a34f0f4e9d473babafd82215e28ff4 : SUCCESS in 8m 28s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/3976c7683eb44e9ba18211c989d51ee3 : SUCCESS in 4m 49s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/60a1aaaba29443fd9972f402835f6d82 : SUCCESS in 6m 42s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/d23da5373cdb467f879eda148e8ea6b5 : SUCCESS in 8m 27s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/8f256602b2b14b9590327f4092bb78ba : SUCCESS in 6m 53s (non-voting)\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/431f1db43eb54930b40374c135304739 : SUCCESS in 1h 32m 00s\n- octavia-v2-dsvm-scenario-traffic-ops https://zuul.opendev.org/t/openstack/build/2b4a291b780746fcbf98d9cee8eed9c9 : SUCCESS in 2h 01m 35s\n- octavia-v2-dsvm-scenario-non-traffic-ops https://zuul.opendev.org/t/openstack/build/7810c48869b2448e9df1ab9948451489 : SUCCESS in 1h 18m 12s\n- octavia-v2-dsvm-scenario-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/580d947f72024c83914f080289846e8f : SUCCESS in 2h 05m 20s\n- octavia-v2-dsvm-scenario-non-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/97567ff6c4e0457daf83579c27d3e9f2 : SUCCESS in 1h 41m 37s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/f0fdf0aba5d04013836bc4d2f1f8f8ec : SUCCESS in 44m 05s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/35ba8b70401346c5a256964196b21e13 : SUCCESS in 51m 53s\n- octavia-v2-dsvm-tls-barbican-fips https://zuul.opendev.org/t/openstack/build/8939eedb8a484a339fdd57deafde01ca : SUCCESS in 1h 09m 50s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/575dda53b0fd4e4da199eb3daefa95af : SUCCESS in 41m 00s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/8fad60634e214ddaa5b21d6c4cdb0d37 : TIMED_OUT in 2h 32m 41s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/d2e7a4317558485aa5b84a4695b15b87 : SUCCESS in 3h 02m 46s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/83a2defc4fa14650ab1ea0440e133c6b : TIMED_OUT in 2h 32m 56s (non-voting)\n- octavia-v2-dsvm-scenario-centos-9-stream https://zuul.opendev.org/t/openstack/build/74766b4e62ae4923b256eb3e87aa1bfd : SUCCESS in 2h 07m 04s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"00c1441381b13c883c581ae1e9cf7fa7d3dcc239","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2023-08-23 21:01:03.000000000","message":"Patch Set 6: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"0b120f49f2e85a8b8a7cf3bd6a001cbaefa4ddd7","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-23 21:01:39.000000000","message":"Patch Set 6: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":6},{"id":"9df558d75b613abb2e01eb4c8f7b7db122741e5f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-23 22:36:03.000000000","message":"Patch Set 6: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/734ef8a1a831497e9ba5fc30b5a081fa\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/aee6de25aea249edb9d857723a9b178f : SUCCESS in 5m 13s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/26972469c1394ada897471fd85bd6154 : SUCCESS in 5m 09s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/51c6d35945fb475fae25527770ed6aaf : SUCCESS in 5m 25s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/0db149c0e87749499df2cca4c9c9017f : SUCCESS in 6m 50s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a03cb93d581e4c3cb39f0f0d8971331e : SUCCESS in 10m 45s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/095553e3da29408087ff78b41989d0e1 : SUCCESS in 5m 04s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/c861e42099254feda8382b1db8f7d626 : SUCCESS in 5m 15s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/34ca11f6d43c44a6be8b199f533831f2 : SUCCESS in 7m 32s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/3f98503cad9241caae42c760c3909306 : SUCCESS in 1h 13m 40s\n- octavia-v2-dsvm-scenario-traffic-ops https://zuul.opendev.org/t/openstack/build/fd1b24fbc8974d0b909254ed178fbcd4 : SUCCESS in 1h 28m 49s\n- octavia-v2-dsvm-scenario-non-traffic-ops https://zuul.opendev.org/t/openstack/build/9201d3958c4343929a4b70dc1f49cf1f : SUCCESS in 1h 08m 29s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/b45a465814064c1a9802365a1a0ee997 : SUCCESS in 41m 39s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/a242b2fb00914da5a1f6531eafbcf2cb : SUCCESS in 52m 58s","accounts_in_message":[],"_revision_number":6},{"id":"4c37060b5df74f254968a258fd833c29082e2b58","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-23 22:36:03.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":6},{"id":"8d31f8dddf24742944b743048c1a473d430a0a06","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-23 22:37:23.000000000","message":"Patch Set 6:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/dd45df5688894138b33c0ca159099e74\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2e9f9cec54c0447a9de3001aa23386a3 : SUCCESS in 1m 02s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/656e988daf0d49e98cda18dc1b02ffe2 : SUCCESS in 45s","accounts_in_message":[],"_revision_number":6}],"current_revision_number":6,"current_revision":"2aac2a8d78b9690a132530290667928b33b12d31","revisions":{"549848e96dd9734be45d135cd510e1105904ab7c":{"kind":"REWORK","_number":1,"created":"2023-07-14 09:25:49.000000000","uploader":{"_account_id":34307,"name":"Tianhao Chen","email":"840738804@qq.com","username":"tianhao.chen"},"ref":"refs/changes/30/888530/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/30/888530/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/30/888530/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/30/888530/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/30/888530/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/30/888530/1"}}},"commit":{"parents":[{"commit":"12d8e0de5de9ca754859542bcffd2311f5145d61","subject":"Merge \"Update doc for the OVN provider\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/12d8e0de5de9ca754859542bcffd2311f5145d61"}]}],"author":{"name":"Chen Tianhao","email":"tianhao.chen@easystack.cn","date":"2023-07-14 09:23:24.000000000","tz":480},"committer":{"name":"Chen Tianhao","email":"tianhao.chen@easystack.cn","date":"2023-07-14 09:23:24.000000000","tz":480},"subject":"Fix IPv6 session persistence failed","message":"Fix IPv6 session persistence failed\n\nChange-Id: I1c51a5d6d6593108f431809868dcdb3d4142df45\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/549848e96dd9734be45d135cd510e1105904ab7c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/549848e96dd9734be45d135cd510e1105904ab7c"}]},"branch":"refs/heads/master"},"3b387ce06a0a1499711c118bf7bb2644df88a3ee":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2023-07-17 02:36:40.000000000","uploader":{"_account_id":34307,"name":"Tianhao Chen","email":"840738804@qq.com","username":"tianhao.chen"},"ref":"refs/changes/30/888530/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/30/888530/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/30/888530/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/30/888530/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/30/888530/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/30/888530/2"}}},"commit":{"parents":[{"commit":"12d8e0de5de9ca754859542bcffd2311f5145d61","subject":"Merge \"Update doc for the OVN provider\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/12d8e0de5de9ca754859542bcffd2311f5145d61"}]}],"author":{"name":"Chen Tianhao","email":"tianhao.chen@easystack.cn","date":"2023-07-14 09:23:24.000000000","tz":480},"committer":{"name":"Chen Tianhao","email":"tianhao.chen@easystack.cn","date":"2023-07-14 09:28:29.000000000","tz":480},"subject":"Fix IPv6 session persistence failed","message":"Fix IPv6 session persistence failed\n\nWhen an IPv6 load balancer is created, and the pool is enabled with SOURCE IP\nsession persistence, it does not work because the stick table only records\nIPv4 addresses. However, an IPv6 stick table can record both IPv4 and IPv6\naddresses.\n\nChange-Id: I1c51a5d6d6593108f431809868dcdb3d4142df45\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/3b387ce06a0a1499711c118bf7bb2644df88a3ee"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/3b387ce06a0a1499711c118bf7bb2644df88a3ee"}]},"branch":"refs/heads/master"},"fb21e34c8a60c362cb6c10983c8ee0fc45771b27":{"kind":"NO_CODE_CHANGE","_number":3,"created":"2023-07-17 03:07:33.000000000","uploader":{"_account_id":34307,"name":"Tianhao Chen","email":"840738804@qq.com","username":"tianhao.chen"},"ref":"refs/changes/30/888530/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/30/888530/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/30/888530/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/30/888530/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/30/888530/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/30/888530/3"}}},"commit":{"parents":[{"commit":"12d8e0de5de9ca754859542bcffd2311f5145d61","subject":"Merge \"Update doc for the OVN provider\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/12d8e0de5de9ca754859542bcffd2311f5145d61"}]}],"author":{"name":"Chen Tianhao","email":"tianhao.chen@easystack.cn","date":"2023-07-14 09:23:24.000000000","tz":480},"committer":{"name":"Chen Tianhao","email":"tianhao.chen@easystack.cn","date":"2023-07-17 03:06:25.000000000","tz":480},"subject":"Fix IPv6 session persistence failed","message":"Fix IPv6 session persistence failed\n\nWhen an IPv6 load balancer is created, and the pool is enabled with SOURCE IP\nsession persistence, it does not work because the stick table only records\nIPv4 addresses. However, an IPv6 stick table can record both IPv4 and IPv6\naddresses.\n\nCloses-Bug: #2027934\n\nChange-Id: I1c51a5d6d6593108f431809868dcdb3d4142df45\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/fb21e34c8a60c362cb6c10983c8ee0fc45771b27"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/fb21e34c8a60c362cb6c10983c8ee0fc45771b27"}]},"branch":"refs/heads/master"},"5eda6444ad472cc5aad23257ab37c20750e9e5e2":{"kind":"NO_CODE_CHANGE","_number":4,"created":"2023-07-17 06:31:59.000000000","uploader":{"_account_id":34307,"name":"Tianhao Chen","email":"840738804@qq.com","username":"tianhao.chen"},"ref":"refs/changes/30/888530/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/30/888530/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/30/888530/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/30/888530/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/30/888530/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/30/888530/4"}}},"commit":{"parents":[{"commit":"12d8e0de5de9ca754859542bcffd2311f5145d61","subject":"Merge \"Update doc for the OVN provider\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/12d8e0de5de9ca754859542bcffd2311f5145d61"}]}],"author":{"name":"Chen Tianhao","email":"tianhao.chen@easystack.cn","date":"2023-07-14 09:23:24.000000000","tz":480},"committer":{"name":"Chen Tianhao","email":"tianhao.chen@easystack.cn","date":"2023-07-17 06:30:31.000000000","tz":480},"subject":"Fix IPv6 session persistence failed","message":"Fix IPv6 session persistence failed\n\nWhen an IPv6 load balancer is created, and the pool is enabled with\nSOURCE IP session persistence, it does not work because the stick\ntable only records IPv4 addresses. However, an IPv6 stick table can\nrecord both IPv4 and IPv6 addresses.\n\nCloses-Bug: #2027934\n\nChange-Id: I1c51a5d6d6593108f431809868dcdb3d4142df45\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/5eda6444ad472cc5aad23257ab37c20750e9e5e2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/5eda6444ad472cc5aad23257ab37c20750e9e5e2"}]},"branch":"refs/heads/master"},"24145740459d9f3254f56b84d6022a4bad56d034":{"kind":"REWORK","_number":5,"created":"2023-07-19 08:16:55.000000000","uploader":{"_account_id":34307,"name":"Tianhao Chen","email":"840738804@qq.com","username":"tianhao.chen"},"ref":"refs/changes/30/888530/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/30/888530/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/30/888530/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/30/888530/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/30/888530/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/30/888530/5"}}},"commit":{"parents":[{"commit":"12d8e0de5de9ca754859542bcffd2311f5145d61","subject":"Merge \"Update doc for the OVN provider\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/12d8e0de5de9ca754859542bcffd2311f5145d61"}]}],"author":{"name":"Chen Tianhao","email":"tianhao.chen@easystack.cn","date":"2023-07-14 09:23:24.000000000","tz":480},"committer":{"name":"Chen Tianhao","email":"tianhao.chen@easystack.cn","date":"2023-07-19 08:16:01.000000000","tz":480},"subject":"Fix IPv6 session persistence failed","message":"Fix IPv6 session persistence failed\n\nWhen an IPv6 load balancer is created, and the pool is enabled with\nSOURCE IP session persistence, it does not work because the stick\ntable only records IPv4 addresses. However, an IPv6 stick table can\nrecord both IPv4 and IPv6 addresses.\n\nCloses-Bug: #2027934\n\nChange-Id: I1c51a5d6d6593108f431809868dcdb3d4142df45\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/24145740459d9f3254f56b84d6022a4bad56d034"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/24145740459d9f3254f56b84d6022a4bad56d034"}]},"branch":"refs/heads/master"},"2aac2a8d78b9690a132530290667928b33b12d31":{"kind":"REWORK","_number":6,"created":"2023-07-19 08:30:50.000000000","uploader":{"_account_id":34307,"name":"Tianhao Chen","email":"840738804@qq.com","username":"tianhao.chen"},"ref":"refs/changes/30/888530/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/30/888530/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/30/888530/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/30/888530/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/30/888530/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/30/888530/6"}}},"commit":{"parents":[{"commit":"12d8e0de5de9ca754859542bcffd2311f5145d61","subject":"Merge \"Update doc for the OVN provider\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/12d8e0de5de9ca754859542bcffd2311f5145d61"}]}],"author":{"name":"Chen Tianhao","email":"tianhao.chen@easystack.cn","date":"2023-07-14 09:23:24.000000000","tz":480},"committer":{"name":"Chen Tianhao","email":"tianhao.chen@easystack.cn","date":"2023-07-19 08:30:39.000000000","tz":480},"subject":"Fix IPv6 session persistence failed","message":"Fix IPv6 session persistence failed\n\nWhen an IPv6 load balancer is created, and the pool is enabled with\nSOURCE IP session persistence, it does not work because the stick\ntable only records IPv4 addresses. However, an IPv6 stick table can\nrecord both IPv4 and IPv6 addresses.\n\nCloses-Bug: #2027934\n\nChange-Id: I1c51a5d6d6593108f431809868dcdb3d4142df45\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/2aac2a8d78b9690a132530290667928b33b12d31"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/2aac2a8d78b9690a132530290667928b33b12d31"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}},{"label":"Backport-Candidate","status":"MAY","applied_by":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
