)]}'
{"id":"openstack%2Foctavia~888758","triplet_id":"openstack%2Foctavia~stable%2Fyoga~I21746515e4ceb0352b4e77708f58873e16d4cdfe","project":"openstack/octavia","branch":"stable/yoga","topic":"fix_persistence_granularity","attention_set":{},"removed_from_attention_set":{"31664":{"account":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"last_update":"2023-10-18 10:25:55.000000000","reason":"Change was submitted"},"29244":{"account":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"last_update":"2023-10-18 10:25:55.000000000","reason":"Change was submitted"},"34429":{"account":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"last_update":"2023-10-18 07:13:33.000000000","reason":"\u003cGERRIT_ACCOUNT_34429\u003e replied on the change","reason_account":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}}},"hashtags":[],"change_id":"I21746515e4ceb0352b4e77708f58873e16d4cdfe","subject":"Fix persistence_granularity default value","status":"MERGED","created":"2023-07-18 09:07:44.000000000","updated":"2023-10-18 10:27:01.000000000","submitted":"2023-10-18 10:25:55.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":4,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"888758-fix_persistence_granularity","meta_rev_id":"03ebc8d455251754dc76818c36cd0d2e510a7168","_number":888758,"virtual_id_number":888758,"owner":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-10-18 10:25:55.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2023-10-12 18:04:34.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":2,"date":"2023-10-18 07:13:33.000000000","_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":1,"date":"2023-10-18 07:13:33.000000000","_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"}],"reviewers":{"REVIEWER":[{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-07-18 09:11:46.000000000","updated_by":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"reviewer":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"state":"REVIEWER"},{"updated":"2023-07-18 09:11:46.000000000","updated_by":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"reviewer":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"state":"REVIEWER"},{"updated":"2023-07-18 11:22:59.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-10-12 18:04:34.000000000","updated_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"reviewer":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"state":"REVIEWER"}],"messages":[{"id":"6d556d534feb119de52cd533488717f269e11793","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"date":"2023-07-18 09:07:44.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"f5356b920bf2931b968bb06bd8d8e106f4c5e0dd","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-18 11:22:59.000000000","message":"Patch Set 1: Verified-1\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/3d988d96fe3e4ca68324e7a838a3be6a\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/30006dc697f048c886da5052184b6d5d : FAILURE in 5m 20s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/f91cc0dd1cb447f1b72e6832ba421662 : FAILURE in 4m 21s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/eaebd31bd78945ef85a3511c364b63c1 : FAILURE in 4m 19s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/15683a03a9bb4aa9a814ac4c01905245 : FAILURE in 5m 35s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/725f7702b241453695962e6ad57bef41 : FAILURE in 5m 48s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/deb64df1ef974637808d62e5369f7676 : SUCCESS in 11m 32s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/83759f146f3641c08679c2e9af9c39f9 : SUCCESS in 4m 49s\n- octavia-tox-py39-tips https://zuul.opendev.org/t/openstack/build/a053bd17e816417c87658f95cea947b7 : FAILURE in 5m 07s\n- octavia-tox-functional-py39-tips https://zuul.opendev.org/t/openstack/build/e21979b93214405db785a42b56f7249e : FAILURE in 5m 10s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/1cb612148297469c92dc6c2f5d209b54 : SUCCESS in 5m 13s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/cb79577579ac4aa580c26d1eac47c750 : SUCCESS in 8m 51s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/bab789c99f5b496e9067176f7ebcddf4 : SUCCESS in 1h 29m 25s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/aa571254560d4f4cbd3cfd5b17a29689 : SUCCESS in 1h 44m 00s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/c5916df12ec047268273cb87becf4d42 : SUCCESS in 46m 07s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/25a4e4455d5d432c93a69ed0e8d2e496 : SUCCESS in 37m 02s\n- octavia-v2-dsvm-tls-barbican-fips https://zuul.opendev.org/t/openstack/build/9cba2d8379354900a032599006d2f7a5 : FAILURE in 15m 11s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/16372fcc71fd41b99ec0e7538253e4e6 : SUCCESS in 29m 18s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/b3c04a3a3a3e46019de7380739cf12a0 : SUCCESS in 1h 49m 25s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/9b24ff984db240698935f3618de4dd00 : FAILURE in 10m 36s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/4a3546d226bb4eb986d18640d994e1a9 : FAILURE in 6m 33s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/e8b2896b1b50415c91b429f1b8659cf8 : SUCCESS in 1h 46m 44s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/19eeb286e929453ca4d2fd0479fb7d6d : SUCCESS in 1h 36m 59s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8-stream https://zuul.opendev.org/t/openstack/build/eeb4554e9c664007a29e4db445a13c2d : FAILURE in 42m 41s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"902e736d91804ace5882fdaf8b36550a90a42748","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"date":"2023-07-19 15:35:18.000000000","message":"Patch Set 2: Published edit on patch set 1.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"a992ac315a1342bb5f1bdff2d36a0a09a1aae2dc","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-19 17:42:25.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/9adabd46ba8e4be09098c505a4dd7d53\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/d09498934074481ab097a4b66616de73 : FAILURE in 4m 50s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/5e1b158ca3e04a94b96fcd6e8feb036f : FAILURE in 4m 00s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4ec61e3ee1d84e81a4f8df5e3703cae3 : SUCCESS in 6m 29s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/b733a1c39dcf418c9f075e5f406cf8a5 : FAILURE in 5m 06s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/79d8a3f0bfd64afc99066c7f0389957b : FAILURE in 5m 24s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/917a6fd9c270421d918d560e460e3c13 : SUCCESS in 12m 33s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/e6406249e10b4a708e3ba9b3358eb854 : SUCCESS in 5m 07s\n- octavia-tox-py39-tips https://zuul.opendev.org/t/openstack/build/9d539313ad9c439fbb2df796c2b96cf9 : FAILURE in 5m 10s\n- octavia-tox-functional-py39-tips https://zuul.opendev.org/t/openstack/build/63685ead4852439eb19e5dc3127e21ca : FAILURE in 5m 32s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/771b367514c1422598f84062d4364df3 : SUCCESS in 5m 06s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/4c0df292236a46689f72bb89c4317f4e : SUCCESS in 7m 29s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/3fc1228c92cc4f23abf9f212f162f36e : SUCCESS in 1h 30m 19s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/f3b62b3b21a24c17b575a4bda9bb55dc : SUCCESS in 1h 53m 03s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/c5f9034984a346d0974b6ea245c11197 : SUCCESS in 41m 24s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/46f6199e3f7c42ec944e5aa5b02fb823 : SUCCESS in 41m 42s\n- octavia-v2-dsvm-tls-barbican-fips https://zuul.opendev.org/t/openstack/build/ed51563af0af41e3b58b5b5dd5a8f4df : FAILURE in 12m 58s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/b9c6a75892dc4ce08bece6d816f2eb22 : SUCCESS in 36m 24s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/f6054dda721c43afb82cba9a2d2e5501 : SUCCESS in 1h 54m 58s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/b093268147b04726b57e40f556557d07 : FAILURE in 12m 34s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/34c6649e33444fd194d3192e25a9665a : FAILURE in 5m 36s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/c5d8368df8a44fd28a1ba4985f158335 : SUCCESS in 2h 05m 51s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/accd4ee3ddcc4d3eab59e7c8a4315118 : SUCCESS in 1h 51m 45s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8-stream https://zuul.opendev.org/t/openstack/build/0bb21b88d1c24fe6bc5d35b14bfe8cff : FAILURE in 43m 21s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"ecd080685d438b6888226610d5e7a064dd9c1084","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"date":"2023-07-20 09:50:12.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":3},{"id":"34f70445366462f57726ae772d0c98bea73e4bd1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-20 13:44:04.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/95856db1442745849a37c477437289d1\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/43964995b35b4592a285042b17254de7 : SUCCESS in 11m 15s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/f7d3ac7e33364db0bb3416b921b394e5 : FAILURE in 4m 46s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d35d18546e8d4073b0d9c1c69d89e031 : SUCCESS in 5m 47s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/79e46494d9764cb499d017bbd629d689 : SUCCESS in 5m 23s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/136dc7820699406c80f34ca6115adb64 : SUCCESS in 4m 59s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6367abeac274471d9d1e54dc9ef60249 : SUCCESS in 12m 23s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/5f804814b46d4887bbf1f0dd96ed26aa : SUCCESS in 5m 03s\n- octavia-tox-py39-tips https://zuul.opendev.org/t/openstack/build/d394438ac2e8412bbe030a30c175d9a0 : SUCCESS in 5m 27s\n- octavia-tox-functional-py39-tips https://zuul.opendev.org/t/openstack/build/100d7415ffb44e44a29675a1a860f15b : SUCCESS in 7m 12s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/bd59ff72c1de42869e4e766868687967 : SUCCESS in 5m 03s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/c0d526dd271d4c4e939fc72538832ff8 : SUCCESS in 11m 06s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/3bbe25c983214847a8b57d7680eec787 : SUCCESS in 1h 57m 30s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/083b650af0394655b9581b097d7746bf : SUCCESS in 2h 33m 12s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/3657ea54e71c4b54874e5bb9314eda0c : SUCCESS in 33m 10s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/67cbfeb9bd5441f094328669b51084d2 : SUCCESS in 36m 26s\n- octavia-v2-dsvm-tls-barbican-fips https://zuul.opendev.org/t/openstack/build/9cb2279b3cf9404cb61ad6e62b670125 : FAILURE in 11m 50s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/db5694b63c0e41a793b45d553493d8ca : SUCCESS in 31m 48s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/6e8ac97c4f3044ed8407685fcc841907 : SUCCESS in 1h 40m 39s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/7b2c2c67c22742869ee58247afe71903 : FAILURE in 11m 09s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/1befec0c257b47d0ab9bd49120ccd41e : FAILURE in 6m 04s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/1d23494a6b48410dbfc6320094dc35d3 : SUCCESS in 1h 46m 57s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/4b2748d05267415487dc9252b03eb27b : SUCCESS in 1h 51m 57s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8-stream https://zuul.opendev.org/t/openstack/build/6c9cb06afbb64e9ba5a912765e708245 : FAILURE in 45m 07s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"f290043be7531432076c72c9704a9ff0fbbfd3e2","author":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"date":"2023-07-20 14:02:45.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"350510d65509a3daea06042c67d9197118e65ba4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-20 15:53:15.000000000","message":"Patch Set 3:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/458e87db5b8e449e94deab6b8d95a5aa\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/24f22b9e9ecf42cbbcd42f74fb51dc96 : SUCCESS in 9m 51s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/c34259c7d07a4357a2f50d8522ac3aca : FAILURE in 4m 21s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/26b9e861a7d34c99bcf89b48efe188b4 : SUCCESS in 6m 51s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/3d59dff645c241edb48fe582ef1d5457 : SUCCESS in 6m 21s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/f79eb0e820b144afbd907144909d69d0 : SUCCESS in 6m 19s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9e9c3379269d4e3aa8f37d7a5a4263ca : SUCCESS in 12m 20s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/ab28042f0b834aafa9578c1246a15b42 : SUCCESS in 5m 46s\n- octavia-tox-py39-tips https://zuul.opendev.org/t/openstack/build/cff943eb340a4446a6e73a4cd4ece8b0 : SUCCESS in 5m 48s\n- octavia-tox-functional-py39-tips https://zuul.opendev.org/t/openstack/build/2b9f13d597714c52b7b0175a44041871 : SUCCESS in 6m 17s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/d838d09d678247ce80131f9c079e37e0 : SUCCESS in 4m 52s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/d1a69255141241ae83ca427a04d1f506 : SUCCESS in 8m 37s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/9932badbcc5f4e4bb7a3cd89bde3a217 : SUCCESS in 1h 31m 32s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/0e4e95ee211640e586d42e349582d053 : SUCCESS in 1h 47m 07s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/584d151bd41a40949dfa0e4e89ecec7e : SUCCESS in 37m 04s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/560d058c936546ca8b7a6a98ca525e26 : SUCCESS in 42m 00s\n- octavia-v2-dsvm-tls-barbican-fips https://zuul.opendev.org/t/openstack/build/b75ec956135f4a90a8febcac0ce3022a : FAILURE in 13m 59s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/18b904faff4b46b79c3fe9492984ae3b : SUCCESS in 32m 23s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/e8ca9df7d60c493ba5f100810662c1e5 : SUCCESS in 1h 42m 40s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/f730ea5b9dc94380ad6282f156dc316d : FAILURE in 13m 23s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/ae994238a0f54d12b86fbf95284f2c06 : FAILURE in 7m 27s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/6b601ef4637b403f99312a8ce235b54b : SUCCESS in 1h 42m 28s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/8c0d36d037024bccae455024685f0e8b : SUCCESS in 1h 48m 58s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8-stream https://zuul.opendev.org/t/openstack/build/8e11970a1be44693a7162f994dc9ee66 : FAILURE in 34m 30s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"12fd868a120ee0922429281211d13c756985a83e","author":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"date":"2023-07-21 12:34:52.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"99de4f8fecbb85d126a79fb7e1a0c3a07c2dad46","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-21 15:10:13.000000000","message":"Patch Set 3:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/eabb38bad2b04556b761b0cb3d0a692e\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/523a90167c1344b19210eeae8fb67975 : SUCCESS in 10m 26s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/12e2400ff1364798b1a9b53240949f6f : SUCCESS in 7m 44s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/9ff470e5273c4e7eaabae34c2d5e2578 : SUCCESS in 5m 37s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/bf90eb26011445b2a5cd0f45a3f66a11 : SUCCESS in 5m 21s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/37d69238bf194dd185ce4a6243179ad4 : SUCCESS in 9m 55s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/e1c1b01ff8144429ac5c46682c656f28 : SUCCESS in 5m 10s\n- octavia-tox-py39-tips https://zuul.opendev.org/t/openstack/build/78ebbecffc7b48bcab51c1b3f1bb6797 : SUCCESS in 5m 13s\n- octavia-tox-functional-py39-tips https://zuul.opendev.org/t/openstack/build/c78103dfb61e45df890e3698ae5f6aaa : SUCCESS in 4m 28s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/6b98dd22457146a2a24e0dca07655d30 : SUCCESS in 5m 19s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/cefc33c887d3405d86efa2f4fc4fc88f : SUCCESS in 8m 36s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/e0e2eafd60424cbc88d2b71659485b98 : SUCCESS in 1h 24m 31s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/e5b6b2d38e8c48359b51b54cb8e33f5c : TIMED_OUT in 2h 33m 28s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/7c05c29a29764327b8033eed1872cbce : SUCCESS in 39m 54s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/985462a3c3984a86a07252a1aac00318 : SUCCESS in 52m 59s\n- octavia-v2-dsvm-tls-barbican-fips https://zuul.opendev.org/t/openstack/build/d178cdf77e14473f9b9791c2a85b1973 : FAILURE in 12m 31s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/b3257ad834a245889bc5e322e30bb2e4 : SUCCESS in 44m 43s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/b55627b2a4c148bdb1b7d9e0ec0c2681 : SUCCESS in 2h 16m 20s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/54381c756a874e2881195e12ea8fbb58 : FAILURE in 13m 29s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/08a411e8749f4579804b17bedd0353a2 : FAILURE in 7m 24s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/102996520ca247ba80b93dee014b20c6 : SUCCESS in 2h 20m 48s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/b6843c02ecee488aaf6c580c45962b17 : FAILURE in 18m 57s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8-stream https://zuul.opendev.org/t/openstack/build/d564af6a836e466cae56146f11f928a7 : FAILURE in 41m 19s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"b41300ed4d210ff4fdb6819e8db61c3e5581cd6c","author":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"date":"2023-07-24 09:20:02.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"ebbaf5d762cb937eeb794cda6213ac401574c258","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-24 11:24:43.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/010708d8c6a24462acfb20887ea20fc9\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/249a532ef87b487791d13f8f411a00f6 : SUCCESS in 9m 33s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7bb3f04cc7b34731b0e5a64be25bc6f6 : SUCCESS in 6m 17s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/a8b9d1e86fe44cc8968ff29600648b4d : SUCCESS in 5m 49s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/234716a87ff04bdbb33cc2521d9d1ab6 : SUCCESS in 4m 37s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8f6e07b23d384f6bba90a3d0122d7b35 : SUCCESS in 12m 50s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/eb582fb5e22b46348096786eec19434f : SUCCESS in 5m 48s\n- octavia-tox-py39-tips https://zuul.opendev.org/t/openstack/build/87a05bedcfce41a889d608f7a794c53b : SUCCESS in 5m 03s\n- octavia-tox-functional-py39-tips https://zuul.opendev.org/t/openstack/build/01b49a2406ee4a5b962cead0e6ec9db2 : SUCCESS in 5m 35s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/1e3b25b3a2484b9ba13280c90a60ed31 : SUCCESS in 5m 22s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/7c303cbc8d2d42088b372841595998dc : SUCCESS in 9m 39s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/a2ab446ee7214bf6a5dded1979469a5b : SUCCESS in 1h 31m 27s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/6703290db5984e2b97af712e4626a14d : SUCCESS in 2h 02m 29s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/bf7bc12c776d4b7eb791911eac686662 : SUCCESS in 37m 36s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/f3ce03d026534336af8779057cf8cedb : SUCCESS in 36m 32s\n- octavia-v2-dsvm-tls-barbican-fips https://zuul.opendev.org/t/openstack/build/985a594b8dae4608b6366839596b6ccc : FAILURE in 11m 13s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/9b81888e2a444612b1bf774111017eee : SUCCESS in 42m 43s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/c24cfc1465024fac86c4f45881ec4751 : SUCCESS in 1h 52m 18s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/abf9efdb979a41fa9b2619d9e9340ce6 : FAILURE in 11m 33s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/38dd562901744de8a5bfb4ca1da92da5 : FAILURE in 7m 13s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/92fc5d9d77694d0f8e53fe2099417f98 : SUCCESS in 1h 53m 14s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/0753f33f428d4e68bc846162f2423fa9 : SUCCESS in 1h 45m 51s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8-stream https://zuul.opendev.org/t/openstack/build/8f14361c2d2549c49b446009c5c9faf0 : FAILURE in 30m 49s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"ae219a6946d53c107d408b81bfa71e22580c2350","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2023-10-12 18:04:34.000000000","message":"Patch Set 3: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"60d5a04c7f323dc4030d9f3c2bf10353e52b248a","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2023-10-18 07:13:33.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"3c1af4f7563a65550650bdac8f1eacd76256a215","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-18 07:14:13.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"d718b74e69db1a90937d93b5f6c5732405e41b1c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-18 10:25:55.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e0b174f0818648008ca7c171bebb6151\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3c4d15c26caa4518bc62525979479cd4 : SUCCESS in 5m 40s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/80ccfea12dcf47c190bd154bf7ff8b64 : SUCCESS in 4m 42s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/71f8c3c7ca03441e935ec0e72778e82d : SUCCESS in 4m 45s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/06389edab6a94148b728d992e62dcb2b : SUCCESS in 11m 59s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/d69462ddf60a4bdba25ea18b354f4808 : SUCCESS in 4m 45s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/95e7b183bd1645d3b6af437eb4fd8ade : SUCCESS in 5m 32s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/4fcfc6270b654a91bb6db1343190c964 : SUCCESS in 8m 41s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/0e1b82f5649349b3b41912b3945330b3 : SUCCESS in 1h 27m 24s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/baf273d95c8747248629112e0c070917 : SUCCESS in 1h 37m 36s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/81a85c493ded49429628855a623543ef : SUCCESS in 41m 07s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/d150b1469c6a404e90d2c252f5be526e : SUCCESS in 42m 10s","accounts_in_message":[],"_revision_number":3},{"id":"72163bef8e030c8614171200c713e7dd31a40749","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-18 10:25:55.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":3},{"id":"03ebc8d455251754dc76818c36cd0d2e510a7168","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-18 10:27:01.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a96eb9741bfb467f855c80a90c6bf76d\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7a005631965a4e869f899a24e0aaceed : SUCCESS in 48s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/0b25e1dcb6fc41a083b119fdb17a1bea : SUCCESS in 47s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"61bb22bbd9bdc311063e4562d209daebfd0fab3e","revisions":{"195394d9c243884078260e80a8e86c69a93fc4d9":{"kind":"REWORK","_number":1,"created":"2023-07-18 09:07:44.000000000","uploader":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"ref":"refs/changes/58/888758/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/58/888758/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/58/888758/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/58/888758/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/58/888758/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/58/888758/1"}}},"commit":{"parents":[{"commit":"b81454a2c0928610bd5b662c236e9caaab359d54","subject":"Merge \"Fix hm operating status to ONLINE in single lb call\" into stable/yoga","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/b81454a2c0928610bd5b662c236e9caaab359d54"}]}],"author":{"name":"Omer","email":"oschwart@redhat.com","date":"2022-10-03 07:24:12.000000000","tz":180},"committer":{"name":"Omer","email":"oschwart@redhat.com","date":"2023-07-18 09:07:34.000000000","tz":120},"subject":"Fix persistence_granularity default value","message":"Fix persistence_granularity default value\n\nSo far we assiged an ipv4 default value to persistence_granularity when\nwe created a UDP pool with session persistence.\n\nThis patch adds the condition of ipv6 and assigns its matching netmask\nvalue.\n\nStory 2009922\nTask 44780\n\nChange-Id: I21746515e4ceb0352b4e77708f58873e16d4cdfe\n(cherry picked from commit dd5d3cca6cd7ec27cd388d7ce4939c9c38af8c83)\n(cherry picked from commit 3a1e35da43822362727bd611f1726fb0a8c6b2c4)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/195394d9c243884078260e80a8e86c69a93fc4d9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/195394d9c243884078260e80a8e86c69a93fc4d9"}]},"branch":"refs/heads/stable/yoga"},"83dde13d572b6abd69fec1fd807a8620385f34f8":{"kind":"REWORK","_number":2,"created":"2023-07-19 15:35:18.000000000","uploader":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"ref":"refs/changes/58/888758/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/58/888758/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/58/888758/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/58/888758/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/58/888758/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/58/888758/2"}}},"commit":{"parents":[{"commit":"b81454a2c0928610bd5b662c236e9caaab359d54","subject":"Merge \"Fix hm operating status to ONLINE in single lb call\" into stable/yoga","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/b81454a2c0928610bd5b662c236e9caaab359d54"}]}],"author":{"name":"Omer","email":"oschwart@redhat.com","date":"2022-10-03 07:24:12.000000000","tz":180},"committer":{"name":"Omer Schwartz","email":"oschwart@redhat.com","date":"2023-07-19 15:35:18.000000000","tz":0},"subject":"Fix persistence_granularity default value","message":"Fix persistence_granularity default value\n\nSo far we assiged an ipv4 default value to persistence_granularity when\nwe created a UDP pool with session persistence.\n\nThis patch adds the condition of ipv6 and assigns its matching netmask\nvalue.\n\nStory 2009922\nTask 44780\n\nChange-Id: I21746515e4ceb0352b4e77708f58873e16d4cdfe\n(cherry picked from commit dd5d3cca6cd7ec27cd388d7ce4939c9c38af8c83)\n(cherry picked from commit 3a1e35da43822362727bd611f1726fb0a8c6b2c4)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/83dde13d572b6abd69fec1fd807a8620385f34f8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/83dde13d572b6abd69fec1fd807a8620385f34f8"}]},"branch":"refs/heads/stable/yoga"},"61bb22bbd9bdc311063e4562d209daebfd0fab3e":{"kind":"REWORK","_number":3,"created":"2023-07-20 09:50:12.000000000","uploader":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"ref":"refs/changes/58/888758/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/58/888758/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/58/888758/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/58/888758/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/58/888758/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/58/888758/3"}}},"commit":{"parents":[{"commit":"b81454a2c0928610bd5b662c236e9caaab359d54","subject":"Merge \"Fix hm operating status to ONLINE in single lb call\" into stable/yoga","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/b81454a2c0928610bd5b662c236e9caaab359d54"}]}],"author":{"name":"Omer","email":"oschwart@redhat.com","date":"2022-10-03 07:24:12.000000000","tz":180},"committer":{"name":"Omer","email":"oschwart@redhat.com","date":"2023-07-20 09:50:01.000000000","tz":120},"subject":"Fix persistence_granularity default value","message":"Fix persistence_granularity default value\n\nSo far we assiged an ipv4 default value to persistence_granularity when\nwe created a UDP pool with session persistence.\n\nThis patch adds the condition of ipv6 and assigns its matching netmask\nvalue.\n\nStory 2009922\nTask 44780\n\nChange-Id: I21746515e4ceb0352b4e77708f58873e16d4cdfe\n(cherry picked from commit dd5d3cca6cd7ec27cd388d7ce4939c9c38af8c83)\n(cherry picked from commit 3a1e35da43822362727bd611f1726fb0a8c6b2c4)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/61bb22bbd9bdc311063e4562d209daebfd0fab3e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/61bb22bbd9bdc311063e4562d209daebfd0fab3e"}]},"branch":"refs/heads/stable/yoga"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
