)]}'
{"id":"openstack%2Foctavia~888986","triplet_id":"openstack%2Foctavia~stable%2Fzed~Iea0ffee1a7307b0852c5a9faf31739c96e76a7ee","project":"openstack/octavia","branch":"stable/zed","attention_set":{},"removed_from_attention_set":{"29244":{"account":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"last_update":"2023-10-17 18:15:14.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Iea0ffee1a7307b0852c5a9faf31739c96e76a7ee","subject":"Fix amphorav1 member deletion bug","status":"MERGED","created":"2023-07-20 07:30:55.000000000","updated":"2023-10-17 18:17:13.000000000","submitted":"2023-10-17 18:15:14.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"888986","meta_rev_id":"2550bfed18793a6b46c96f5ae2a8bad44a084189","_number":888986,"virtual_id_number":888986,"owner":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-10-17 18:15:13.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"all":[{"value":2,"date":"2023-10-12 18:12:31.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2023-10-17 15:49:23.000000000","_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"all":[{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2023-10-17 15:49:23.000000000","_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-07-20 10:15:38.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-10-12 18:12:31.000000000","updated_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"reviewer":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"state":"REVIEWER"},{"updated":"2023-10-17 15:49:23.000000000","updated_by":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"reviewer":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"state":"REVIEWER"}],"messages":[{"id":"50f2ea521953b20ca6e1c5dfa948080e106d677c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2023-07-20 07:30:55.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"820d31b77ae276026b926ebf9292d005a3d78e4a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-20 10:15:38.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/66b21e04946345d0abe2a98c7d1db70f\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/0bfcc98861a444fe8055768bfc389322 : SUCCESS in 10m 21s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f6979567d4b14ac2afe591d4de9fcb9e : SUCCESS in 9m 39s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/4453a3ddb0db4255ad1c759ca84a73e0 : SUCCESS in 5m 20s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/def4a90c4f1a46b9b3ad7d2324280381 : SUCCESS in 6m 02s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/f41890580b39440182d12df4eae8edb9 : SUCCESS in 5m 26s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7737cf289c9c44b9a627b597ac5f5ae2 : SUCCESS in 13m 01s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/645d814ab6cd44c699f49e74c945225e : SUCCESS in 4m 17s\n- octavia-tox-py39-tips https://zuul.opendev.org/t/openstack/build/4abacb40cd194b8f8b4706d335816e25 : SUCCESS in 5m 38s\n- octavia-tox-functional-py39-tips https://zuul.opendev.org/t/openstack/build/b62cf93ce01f40e3a24c7d3dfee7d7a2 : SUCCESS in 8m 25s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/15b98e83ab694eeb9ad6b7916d656113 : SUCCESS in 5m 13s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/86a005a1a65743aa83371ed1708c187b : SUCCESS in 6m 58s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/b31b1219539e4b5e8ed842f42dd3afb3 : SUCCESS in 8m 50s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/789cca092e2349fcbe6ecdda856ffe0e : SUCCESS in 7m 52s (non-voting)\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/d74d141709864f12be2884fbcf547090 : SUCCESS in 1h 27m 30s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/9a5489493d6b41b7915c6efb1c3a4783 : SUCCESS in 2h 11m 38s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/ce8ac9b874ce47809df2b7696e10bd91 : SUCCESS in 49m 18s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/ea3698fe67224df5a6ecda441b57c08a : SUCCESS in 47m 16s\n- octavia-v2-dsvm-tls-barbican-fips https://zuul.opendev.org/t/openstack/build/d883eb559f72475788d5eb60ae96d6b4 : FAILURE in 12m 18s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/710833783fa74b84a59fb042d27969dc : SUCCESS in 31m 57s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/1ca5a8e642914b5088feef92895744cc : SUCCESS in 1h 57m 11s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/e61b18c99d3e49e6ad56c5a9c5eca8d4 : SUCCESS in 2h 40m 12s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/25f9035d18f349179a6943c52a208ab4 : TIMED_OUT in 2h 33m 07s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/4a65464381a1408298a94d973e4fa8ef : SUCCESS in 1h 35m 23s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/0a87e4a8da0e4ae1a6e66535728208b8 : SUCCESS in 1h 44m 34s (non-voting)\n- octavia-v2-dsvm-scenario-centos-9-stream https://zuul.opendev.org/t/openstack/build/aecea11633cf4cdcb33bd5c863254763 : TIMED_OUT in 2h 35m 02s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"8691d4f3542d9a0ad4c14305aeb71fccea622da4","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2023-10-12 18:12:31.000000000","message":"Patch Set 1: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"206dfb60342d63b384b13b4692e2a0764144a2e7","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2023-10-17 15:49:23.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"40ec507cedafd2ab1f3ac1b3fd72d2d22ed5e9fa","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-17 15:50:15.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"4d6e9383581d004284859b7fee5f58c1e5ee973a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-17 18:15:13.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/796f2579fbd44ad4b4f83b398a9ac04f\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/86fe98298c6a4d119bc65ff072865803 : SUCCESS in 7m 20s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/7732c027406442bd9cd2318b97d03497 : SUCCESS in 5m 40s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/94c2a3c1b7444127994f1615052ec8cc : SUCCESS in 6m 24s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/697e658d855c46a9a05c8e7dd69367bb : SUCCESS in 12m 59s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/530c8e66d5f940158da5dbdcab898e31 : SUCCESS in 5m 38s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/6db1b62b74a74e8aab9761954e572e52 : SUCCESS in 6m 07s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/9cf4c9bb070b45f7bb602965f5a1a21d : SUCCESS in 8m 46s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/14dd2efd6f54477e964e1598f4c73062 : SUCCESS in 1h 42m 03s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/80c55bbe9823447bbea6e66f648f3ca5 : SUCCESS in 2h 02m 21s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/408e25fe72924bba882c2ce10ed9709e : SUCCESS in 45m 01s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/b0cdb2fd46f347e1b97b12e97d994b1e : SUCCESS in 39m 29s","accounts_in_message":[],"_revision_number":1},{"id":"4cfbedae2712b8aaf04cc2ecf294f03c6197e085","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-17 18:15:14.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"2550bfed18793a6b46c96f5ae2a8bad44a084189","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-17 18:17:13.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/69c4f881f3e04fb3aa77210618cf3ae3\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/24edb384dbac450fb63b85ee289bdd27 : SUCCESS in 1m 32s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/448f577b671e445783456228f7505dc4 : SUCCESS in 45s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"68a3379cda266fac649ef984404090d9cfa246ac","revisions":{"68a3379cda266fac649ef984404090d9cfa246ac":{"kind":"REWORK","_number":1,"created":"2023-07-20 07:30:55.000000000","uploader":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"ref":"refs/changes/86/888986/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/86/888986/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/86/888986/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/86/888986/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/86/888986/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/86/888986/1"}}},"commit":{"parents":[{"commit":"cde74751b5fe4016f002c3031c0c88100eb08619","subject":"Merge \"Fix hm operating status to ONLINE in single lb call\" into stable/zed","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/cde74751b5fe4016f002c3031c0c88100eb08619"}]}],"author":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2023-07-17 09:02:08.000000000","tz":120},"committer":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2023-07-20 07:30:45.000000000","tz":120},"subject":"Fix amphorav1 member deletion bug","message":"Fix amphorav1 member deletion bug\n\nWhen the worker calculates the delta for plugging/unplugging subnets,\nit need to check the provisioning_status of the members. But in\namphorav1, this status comes from the LB DB object that is passed to\ntaskflow. If the provisioning_status is modified by another task\n(MarkMemberPendingDeleteInDb), the LB object is not updated.\nWe need to reload it from the DB to get the current status of the\nmembers.\n\nNote: this patch is for stable branches only as the amphorav1 driver was\n      removed during the Bobcat cycle\nCloses-Bug: #2027967\nChange-Id: Iea0ffee1a7307b0852c5a9faf31739c96e76a7ee\n(cherry picked from commit 8b196bb3bb05c9d87dae688750457dbb944d4d1b)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/68a3379cda266fac649ef984404090d9cfa246ac"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/68a3379cda266fac649ef984404090d9cfa246ac"}]},"branch":"refs/heads/stable/zed"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
