)]}'
{"id":"openstack%2Foctavia~888987","triplet_id":"openstack%2Foctavia~stable%2Fyoga~Iea0ffee1a7307b0852c5a9faf31739c96e76a7ee","project":"openstack/octavia","branch":"stable/yoga","attention_set":{},"removed_from_attention_set":{"11628":{"account":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"last_update":"2023-10-17 18:15:15.000000000","reason":"Change was submitted"},"29244":{"account":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"last_update":"2023-10-17 18:15:15.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Iea0ffee1a7307b0852c5a9faf31739c96e76a7ee","subject":"Fix amphorav1 member deletion bug","status":"MERGED","created":"2023-07-20 07:31:06.000000000","updated":"2023-10-17 18:17:11.000000000","submitted":"2023-10-17 18:15:15.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"888987","meta_rev_id":"ddcefd46730528f6df1d0fc1b420b70c8ecfb851","_number":888987,"virtual_id_number":888987,"owner":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-10-17 18:15:15.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2023-10-12 18:13:18.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":2,"date":"2023-10-17 15:49:44.000000000","_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":1,"date":"2023-10-17 15:49:44.000000000","_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-07-20 09:46:29.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-10-12 18:13:18.000000000","updated_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"reviewer":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"state":"REVIEWER"},{"updated":"2023-10-17 15:49:44.000000000","updated_by":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"reviewer":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"state":"REVIEWER"}],"messages":[{"id":"8d6bd71a5abeac9a94af0c433688b305b0e83581","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2023-07-20 07:31:06.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"f0033aeb8c06be0e15148c13f84031a2f1e85b2a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-20 09:46:29.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/309a7e9c32b843a48254ad34696e9a2f\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/99a728f2ab7e4a3c8ba525f512ae4c30 : SUCCESS in 11m 39s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/754c6e86f88f4d93a743f6b70f3db726 : FAILURE in 4m 50s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e12c023026da4ab8b0662e18401fd021 : SUCCESS in 5m 27s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/ec173136195f4a27be47212c05a5ce02 : SUCCESS in 7m 51s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/27a67a0ae0544d9c9feb9ad0188359b1 : SUCCESS in 6m 20s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/62633b32245c4330a9043e6767d0810d : SUCCESS in 12m 55s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/bc4b64f17c714719b0566b8170ede55d : SUCCESS in 5m 16s\n- octavia-tox-py39-tips https://zuul.opendev.org/t/openstack/build/d351f9b6252d4c38ae4913db5851a1aa : SUCCESS in 5m 24s\n- octavia-tox-functional-py39-tips https://zuul.opendev.org/t/openstack/build/845e43a7000a4424a9ec5cbf06156448 : SUCCESS in 4m 54s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/26e436e937d14e24903cc8fde64dfcad : SUCCESS in 6m 06s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/0e6e853ac88445c5b333159b712af861 : SUCCESS in 11m 01s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/4ac0ff3d45bf45d2b9d924c66d4ba83c : SUCCESS in 1h 40m 40s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/9743829f4adf433a943ebcb6b5022318 : SUCCESS in 1h 42m 38s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/309b4ba1aa4d4282a11f5dcb1309847b : SUCCESS in 48m 16s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/723998c033144fcbbf71adfde9b6ebde : SUCCESS in 38m 20s\n- octavia-v2-dsvm-tls-barbican-fips https://zuul.opendev.org/t/openstack/build/30b8575cca024096aa55b4304068639b : FAILURE in 13m 42s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/872a5637874d4a74af59097a749a4f54 : SUCCESS in 36m 42s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/65dac38b94504f7e9eaf970dbda72ebb : SUCCESS in 1h 50m 55s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/45d5d513f7084fe5897f123e13bbb6f5 : FAILURE in 14m 35s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/3d7ddae430cf4a0f8bf521391c66fe13 : FAILURE in 7m 03s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/48a53906f2f24800af44096ad18967f9 : SUCCESS in 2h 00m 48s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/d4b6895df8bc47b8bb909b9cfb378055 : SUCCESS in 1h 51m 30s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8-stream https://zuul.opendev.org/t/openstack/build/d57e52e90a3c4762a73907b801385277 : FAILURE in 44m 39s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"05f8312a285522aa6220d5c9d45a62c301fc1d1d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2023-10-12 18:13:12.000000000","message":"Patch Set 2: Patch Set 1 was rebased\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"c0a65dc90a779a9767e79ff71be18e2479eaf72b","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2023-10-12 18:13:18.000000000","message":"Patch Set 2: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"c2bace1260989cc172298b4a976e6d19e32f5056","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-12 20:47:43.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5776652ff89648d795e4a4da9d9a4214\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/c2119c44a0a6444fa0c4544e39fa643e : SUCCESS in 11m 02s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ca76cf9b2178413eb88efff484b89f4e : SUCCESS in 5m 52s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/358090a1202e465ea5a57b08ae18e1e5 : SUCCESS in 5m 45s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/579d739d5b184bdb93c10f6eb956a9d5 : SUCCESS in 5m 40s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2ab6c17ab95948179f86e323abc359f4 : SUCCESS in 11m 21s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/51253f5393eb4bd9869b40110efa50f5 : SUCCESS in 5m 42s\n- octavia-tox-py39-tips https://zuul.opendev.org/t/openstack/build/670a310de4f249a6840e6ca680fd225d : SUCCESS in 5m 13s\n- octavia-tox-functional-py39-tips https://zuul.opendev.org/t/openstack/build/552957c843bf447f8a7edbe04eea9c8c : SUCCESS in 5m 12s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/656baf22b913450a9228ed6196d1d81b : SUCCESS in 5m 11s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/ae2cd97a860a4e9197d13a5a1ce05303 : SUCCESS in 6m 00s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/6cd7c479652c4ae88f207af846deebb0 : SUCCESS in 1h 02m 47s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/f6f021510ebf4281a01dc2f4170be834 : SUCCESS in 2h 07m 36s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/315f1928f1824a03b8944a3ef68499c0 : SUCCESS in 41m 06s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/31d262f3532b4b9ebb5cd9fc32ddd9c2 : SUCCESS in 43m 42s\n- octavia-v2-dsvm-tls-barbican-fips https://zuul.opendev.org/t/openstack/build/547a13a94dfa4205a3b6eeb5d35729fa : FAILURE in 56m 18s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/704dcb2e128d4b52bfb0af3975f96e1f : SUCCESS in 35m 57s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/6af46ca1aec649958cd0cf83a17aeedf : SUCCESS in 1h 59m 20s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/e6b9deeb88bf489fb32d10ec1e8bd90d : FAILURE in 12m 59s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/3f74ffb0c55c42c9a4778e8e03205eab : FAILURE in 5m 14s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/019453dac96b4ad5b66a24e3eb5ad8db : SUCCESS in 2h 05m 54s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/5877782af19b428badb87a04f554d49f : SUCCESS in 2h 31m 11s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8-stream https://zuul.opendev.org/t/openstack/build/b51eb828f11147c8b8f70d185154b3e3 : FAILURE in 43m 38s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"b8fbe4abad8f998a8f22a0c76a6a100e2a3e2db5","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2023-10-17 15:49:44.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"2765cc134caae417cbc5ecd5724efbde4714b5a2","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-17 15:50:32.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"fada572aad20151e0119242b49b16c532ce46e70","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-17 18:15:15.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ea192ea23ef741d490095aca29477a5c\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/344f39049b7c4bc89b779b79188635b6 : SUCCESS in 6m 32s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/7c16d91c633146029843703306c379be : SUCCESS in 6m 12s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/919ab70e94c543f5b57aaffb9158d6b6 : SUCCESS in 5m 31s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fa9ce588bab14cf0895d273aea80565b : SUCCESS in 15m 08s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/9f9dd2226396422399ee2d4796594783 : SUCCESS in 5m 54s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/0c0e57caf4884dbab9690d1bbde5e6d3 : SUCCESS in 5m 43s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/790f1c1b3ad94c2d88040c83ee03c968 : SUCCESS in 8m 13s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/25a4fbf5060b4711873c237cefe53c4e : SUCCESS in 1h 51m 48s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/916cb133bad14aa3bbc2581176e9bce1 : SUCCESS in 1h 40m 33s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/93f053fd63fa4a0c964ee342564a58c6 : SUCCESS in 35m 21s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/ddfbc347d8ed470fb141b4ebcc3ee526 : SUCCESS in 35m 43s","accounts_in_message":[],"_revision_number":2},{"id":"3820214663905a7bce8ed16bdfc2888f2e7fcf7d","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-17 18:15:15.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"ddcefd46730528f6df1d0fc1b420b70c8ecfb851","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-17 18:17:11.000000000","message":"Patch Set 2:\n\nBuild failed (promote pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/d953db7080514076b4b1b3e2020fbdeb\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8ff04ad0c94248a8ae09bc6fa44ecd5e : SUCCESS in 1m 28s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/c7e8592d292d4835a5c55f4c720bbd69 : FAILURE in 52s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"c8bd99e384a5400635285c5bf47a03aba105fc20","revisions":{"13070bd173138189c421c92078bf34ccff7697ab":{"kind":"REWORK","_number":1,"created":"2023-07-20 07:31:06.000000000","uploader":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"ref":"refs/changes/87/888987/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/87/888987/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/87/888987/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/87/888987/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/87/888987/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/87/888987/1"}}},"commit":{"parents":[{"commit":"b81454a2c0928610bd5b662c236e9caaab359d54","subject":"Merge \"Fix hm operating status to ONLINE in single lb call\" into stable/yoga","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/b81454a2c0928610bd5b662c236e9caaab359d54"}]}],"author":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2023-07-17 09:02:08.000000000","tz":120},"committer":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2023-07-20 07:31:02.000000000","tz":120},"subject":"Fix amphorav1 member deletion bug","message":"Fix amphorav1 member deletion bug\n\nWhen the worker calculates the delta for plugging/unplugging subnets,\nit need to check the provisioning_status of the members. But in\namphorav1, this status comes from the LB DB object that is passed to\ntaskflow. If the provisioning_status is modified by another task\n(MarkMemberPendingDeleteInDb), the LB object is not updated.\nWe need to reload it from the DB to get the current status of the\nmembers.\n\nNote: this patch is for stable branches only as the amphorav1 driver was\n      removed during the Bobcat cycle\nCloses-Bug: #2027967\nChange-Id: Iea0ffee1a7307b0852c5a9faf31739c96e76a7ee\n(cherry picked from commit 8b196bb3bb05c9d87dae688750457dbb944d4d1b)\n(cherry picked from commit 68a3379cda266fac649ef984404090d9cfa246ac)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/13070bd173138189c421c92078bf34ccff7697ab"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/13070bd173138189c421c92078bf34ccff7697ab"}]},"branch":"refs/heads/stable/yoga"},"c8bd99e384a5400635285c5bf47a03aba105fc20":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2023-10-12 18:13:12.000000000","uploader":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"ref":"refs/changes/87/888987/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/87/888987/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/87/888987/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/87/888987/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/87/888987/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/87/888987/2"}}},"commit":{"parents":[{"commit":"377dcb9f9500d7118c2f988baaa9308f9142b760","subject":"Merge \"fix haproxy_count to display the number of running processes\" into stable/yoga","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/377dcb9f9500d7118c2f988baaa9308f9142b760"}]}],"author":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2023-07-17 09:02:08.000000000","tz":120},"committer":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2023-10-12 18:13:12.000000000","tz":0},"subject":"Fix amphorav1 member deletion bug","message":"Fix amphorav1 member deletion bug\n\nWhen the worker calculates the delta for plugging/unplugging subnets,\nit need to check the provisioning_status of the members. But in\namphorav1, this status comes from the LB DB object that is passed to\ntaskflow. If the provisioning_status is modified by another task\n(MarkMemberPendingDeleteInDb), the LB object is not updated.\nWe need to reload it from the DB to get the current status of the\nmembers.\n\nNote: this patch is for stable branches only as the amphorav1 driver was\n      removed during the Bobcat cycle\nCloses-Bug: #2027967\nChange-Id: Iea0ffee1a7307b0852c5a9faf31739c96e76a7ee\n(cherry picked from commit 8b196bb3bb05c9d87dae688750457dbb944d4d1b)\n(cherry picked from commit 68a3379cda266fac649ef984404090d9cfa246ac)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/c8bd99e384a5400635285c5bf47a03aba105fc20"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/c8bd99e384a5400635285c5bf47a03aba105fc20"}]},"branch":"refs/heads/stable/yoga","description":"Rebase"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
