)]}'
{"id":"openstack%2Foctavia~888989","triplet_id":"openstack%2Foctavia~stable%2Fwallaby~Iea0ffee1a7307b0852c5a9faf31739c96e76a7ee","project":"openstack/octavia","branch":"stable/wallaby","attention_set":{},"removed_from_attention_set":{"11628":{"account":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"last_update":"2023-10-17 18:15:19.000000000","reason":"Change was submitted"},"29244":{"account":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"last_update":"2023-10-17 18:15:19.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Iea0ffee1a7307b0852c5a9faf31739c96e76a7ee","subject":"Fix amphorav1 member deletion bug","status":"MERGED","created":"2023-07-20 07:31:29.000000000","updated":"2023-10-17 18:17:23.000000000","submitted":"2023-10-17 18:15:19.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"888989","meta_rev_id":"8be5fc2d5ad74e286e39bda5698ecc170a7afdfc","_number":888989,"virtual_id_number":888989,"owner":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-10-17 18:15:19.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"all":[{"value":2,"date":"2023-10-12 18:12:21.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2023-10-17 15:50:04.000000000","_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"all":[{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2023-10-17 15:50:04.000000000","_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-07-20 10:41:25.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-10-12 18:12:21.000000000","updated_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"reviewer":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"state":"REVIEWER"},{"updated":"2023-10-17 15:50:04.000000000","updated_by":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"reviewer":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"state":"REVIEWER"}],"messages":[{"id":"d923ce42b7983fbf91a1b6e551edd0b74788392b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2023-07-20 07:31:29.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"e55df354e080259a36db8c68cef636281b982fe1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-20 10:41:25.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/123472c38d694704bb6d6bbe05b99455\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/c961f7ad2b3b4d7fad3df2e6001753a6 : SUCCESS in 9m 30s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/77621ed2a68f44f29c704b051005dee7 : FAILURE in 3m 39s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6abd963b1e204048957405511b468dff : SUCCESS in 7m 51s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/030412ae8bde48b49adf039edfd91512 : SUCCESS in 5m 11s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/79ff6d61920a4154bca36a37a9aebfff : SUCCESS in 4m 22s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/1dbc0fd9c5b943b1b0fad224e7a38bc0 : SUCCESS in 4m 20s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bb2ffe13d4c145d0a86efbf3579da4c1 : SUCCESS in 11m 30s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/4bf031dba3f54f309128985d43c1590e : SUCCESS in 5m 41s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/a13d062c40bf409494dfaa8e1bf9c6fb : SUCCESS in 6m 46s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/4ca7833fb4a54066a2d7f87339918fa3 : SUCCESS in 7m 07s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/64e26548e85442109cd2f1433db95ac8 : SUCCESS in 5m 54s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/aa4380f54d3449778cb1b1eb05ec9655 : SUCCESS in 7m 00s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/259e3424383f4bebb7e6dc53207536b6 : SUCCESS in 1h 29m 13s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/3d364dc2bba14fd2932834100290344a : SUCCESS in 1h 45m 46s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/d4ea013ca7b240df890320ed7f58922e : SUCCESS in 35m 35s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/19be6e85f15d441cafcd8315d9195ff9 : SUCCESS in 31m 51s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/e32e12258a544ad6ae0e47275d9d4900 : SUCCESS in 34m 53s\n- octavia-v2-dsvm-tls-barbican-fips https://zuul.opendev.org/t/openstack/build/8afe6b472cf149a485c84ab97d41f59f : FAILURE in 9m 33s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/24131b9d63e04c3299032646b8400d3e : SUCCESS in 1h 49m 49s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/b01819dfae104871b3604cc3fa2c1673 : FAILURE in 12m 25s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/474cb414a754442a92b499cacf5a0a31 : FAILURE in 7m 10s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/6db3b71eecba4e4ca2c9f84ee0d39e80 : SUCCESS in 2h 00m 03s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/5cb84e60993a42498a1f52812fa7c921 : TIMED_OUT in 2h 35m 02s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8 https://zuul.opendev.org/t/openstack/build/04d583a06bd4438ebbe1fb3979243d20 : NODE_FAILURE Node request 300-0021734010 failed in 0s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"1b3dcdac2e70742888f77d495183350874dd9a6c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2023-10-12 18:12:14.000000000","message":"Patch Set 2: Patch Set 1 was rebased\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"9d29b827ee42721504a09bd4df914d898e1f7c3d","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2023-10-12 18:12:21.000000000","message":"Patch Set 2: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"3629254bb3f11494828846b2732f85ab05231140","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-12 20:05:15.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/dfdd833470be4d9b8a643f52aa428b25\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/1a51df18287f49c5a86a72c66a0772cb : SUCCESS in 7m 35s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/aaf74242e9524ad99c4d28dd51304900 : SUCCESS in 4m 27s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/fd22675f47274b79a7cb0b72915a9293 : SUCCESS in 5m 17s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/888715c3731e4a69ad87689d8f405564 : SUCCESS in 4m 29s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/2ad3c6bc22d9497a9de18b2607d1b75b : SUCCESS in 4m 37s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/beeef4988f5b4e0b890506ce5bc70083 : SUCCESS in 11m 32s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/c0bd05f6c397442984e964439cf26731 : SUCCESS in 5m 24s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/bb8b9b7861fb4dfd95fa3e40560d3731 : SUCCESS in 4m 48s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/44b6180d01ed42efa4f00e4c76da1a1b : SUCCESS in 6m 09s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/c28237dabb144b80b10ac4a0cadefcd0 : SUCCESS in 5m 42s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/84d2d8b1b5354246acd640fd1c007cb0 : SUCCESS in 7m 03s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/db034050df24401eb9a7f0982abc97cc : SUCCESS in 44m 57s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/8006c5e9a3ed449c8024f624d8821671 : SUCCESS in 1h 30m 23s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/7a3324197d294b119e1c49273ce977b9 : SUCCESS in 33m 49s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/211c5dde9ae4488dafa9d5f33f7ad1e7 : SUCCESS in 36m 29s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/9ec648e786c54f12b3f184b785651191 : SUCCESS in 37m 27s\n- octavia-v2-dsvm-tls-barbican-fips https://zuul.opendev.org/t/openstack/build/bc8d14ac9c2942658ce3fc70cd80eb18 : FAILURE in 9m 17s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/d2c8cd02d7e34c1cacbaf5ab4e796a29 : SUCCESS in 1h 42m 28s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/311838579c284652bc84787efc5d02c2 : FAILURE in 12m 29s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/2500836cced042948c02373be702803a : FAILURE in 8m 23s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/f2909b6d4e81444396e04b5cda2f30a3 : FAILURE in 1h 50m 26s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/e6c58cbebf8448dea3683d58cb9eb060 : SUCCESS in 1h 45m 18s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8 https://zuul.opendev.org/t/openstack/build/f904aa771e8a402b8ea4b6988c1097ec : NODE_FAILURE Node request 300-0022457230 failed in 0s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"525b9b88fc17492480ec3d62dbe6612976000b74","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2023-10-17 15:50:04.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"5ad24c55f30d1b64b376d7cbf6f1db4ff6a1cb65","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-17 15:51:02.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"6bd7b5c24bbfd8a9747098fcd1f23c39e3a169c9","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-17 18:15:19.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f239723cd2614818bd6e5791766301c8\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ba70b0cf6d5948e6aff7d8b9a7ace4e2 : SUCCESS in 6m 15s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/a635d1801b9641aba53e5149571f0274 : SUCCESS in 5m 59s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/663f44ded9cb477dac6496da665de2c7 : SUCCESS in 5m 49s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c30b3559d7fd4524843c5edc64660949 : SUCCESS in 11m 59s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/07edcf624a5c4706928451e681394348 : SUCCESS in 4m 41s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/95631b41a02046c69a2e5a5aefed46c5 : SUCCESS in 5m 40s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/f1dc2480616e42d4bc35583961f24079 : SUCCESS in 7m 14s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/514d55647a57422f98191a8adcb6442c : SUCCESS in 1h 22m 48s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/de2cda526fb644888056873786740725 : SUCCESS in 1h 45m 13s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/803660bf709e468e999303f6371346b8 : SUCCESS in 47m 30s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/88899a1eeca044ae96c342f3f5dc8f19 : SUCCESS in 36m 03s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/42d7f38d3314406898a3b02b88026a41 : SUCCESS in 36m 00s","accounts_in_message":[],"_revision_number":2},{"id":"db8b035ef71972d4c71d1eddac3662e922398f41","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-17 18:15:19.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"8be5fc2d5ad74e286e39bda5698ecc170a7afdfc","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-17 18:17:23.000000000","message":"Patch Set 2:\n\nBuild failed (promote pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/9c02c434c8594263a8ca2f392f9976c3\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/660452d9b99e4621a70daff9e61f2a7d : SUCCESS in 1m 30s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/82ae9290ffe5427884227fa14e82130f : FAILURE in 51s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"8447d91b6582d8216e75896121d176ce312c35a8","revisions":{"ce81a12e968a91f6635b3a8b1ea5b391537d0b4b":{"kind":"REWORK","_number":1,"created":"2023-07-20 07:31:29.000000000","uploader":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"ref":"refs/changes/89/888989/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/89/888989/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/89/888989/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/89/888989/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/89/888989/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/89/888989/1"}}},"commit":{"parents":[{"commit":"8af6306df63df55d47a56c2c222eb5a254b88574","subject":"Merge \"Fix hm operating status to ONLINE in single lb call\" into stable/wallaby","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/8af6306df63df55d47a56c2c222eb5a254b88574"}]}],"author":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2023-07-17 09:02:08.000000000","tz":120},"committer":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2023-07-20 07:31:25.000000000","tz":120},"subject":"Fix amphorav1 member deletion bug","message":"Fix amphorav1 member deletion bug\n\nWhen the worker calculates the delta for plugging/unplugging subnets,\nit need to check the provisioning_status of the members. But in\namphorav1, this status comes from the LB DB object that is passed to\ntaskflow. If the provisioning_status is modified by another task\n(MarkMemberPendingDeleteInDb), the LB object is not updated.\nWe need to reload it from the DB to get the current status of the\nmembers.\n\nNote: this patch is for stable branches only as the amphorav1 driver was\n      removed during the Bobcat cycle\nCloses-Bug: #2027967\nChange-Id: Iea0ffee1a7307b0852c5a9faf31739c96e76a7ee\n(cherry picked from commit 8b196bb3bb05c9d87dae688750457dbb944d4d1b)\n(cherry picked from commit 68a3379cda266fac649ef984404090d9cfa246ac)\n(cherry picked from commit 13070bd173138189c421c92078bf34ccff7697ab)\n(cherry picked from commit 0b07b39816c5fa0fd2bfe88a3541372918722385)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/ce81a12e968a91f6635b3a8b1ea5b391537d0b4b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/ce81a12e968a91f6635b3a8b1ea5b391537d0b4b"}]},"branch":"refs/heads/stable/wallaby"},"8447d91b6582d8216e75896121d176ce312c35a8":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2023-10-12 18:12:14.000000000","uploader":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"ref":"refs/changes/89/888989/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/89/888989/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/89/888989/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/89/888989/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/89/888989/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/89/888989/2"}}},"commit":{"parents":[{"commit":"77e14a8d790bfb0facaf2ee56c1f798e58269b13","subject":"Fix race condition in members batch update API call","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/77e14a8d790bfb0facaf2ee56c1f798e58269b13"}]}],"author":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2023-07-17 09:02:08.000000000","tz":120},"committer":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2023-10-12 18:12:14.000000000","tz":0},"subject":"Fix amphorav1 member deletion bug","message":"Fix amphorav1 member deletion bug\n\nWhen the worker calculates the delta for plugging/unplugging subnets,\nit need to check the provisioning_status of the members. But in\namphorav1, this status comes from the LB DB object that is passed to\ntaskflow. If the provisioning_status is modified by another task\n(MarkMemberPendingDeleteInDb), the LB object is not updated.\nWe need to reload it from the DB to get the current status of the\nmembers.\n\nNote: this patch is for stable branches only as the amphorav1 driver was\n      removed during the Bobcat cycle\nCloses-Bug: #2027967\nChange-Id: Iea0ffee1a7307b0852c5a9faf31739c96e76a7ee\n(cherry picked from commit 8b196bb3bb05c9d87dae688750457dbb944d4d1b)\n(cherry picked from commit 68a3379cda266fac649ef984404090d9cfa246ac)\n(cherry picked from commit 13070bd173138189c421c92078bf34ccff7697ab)\n(cherry picked from commit 0b07b39816c5fa0fd2bfe88a3541372918722385)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/8447d91b6582d8216e75896121d176ce312c35a8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/8447d91b6582d8216e75896121d176ce312c35a8"}]},"branch":"refs/heads/stable/wallaby","description":"Rebase"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
