)]}'
{"id":"openstack%2Foctavia~888992","triplet_id":"openstack%2Foctavia~stable%2Fxena~I0e2d06b02c06b8c77fb1c2f94207d422066a0c6d","project":"openstack/octavia","branch":"stable/xena","topic":"tmp1","attention_set":{},"removed_from_attention_set":{"29244":{"account":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"last_update":"2023-10-24 14:31:02.000000000","reason":"Change was submitted"},"11628":{"account":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"last_update":"2023-10-23 12:54:06.000000000","reason":"Reviewer/Cc was removed"},"34429":{"account":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"last_update":"2023-10-24 12:07:33.000000000","reason":"\u003cGERRIT_ACCOUNT_34429\u003e replied on the change","reason_account":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}}},"hashtags":[],"change_id":"I0e2d06b02c06b8c77fb1c2f94207d422066a0c6d","subject":"Fix text/plain mime type with healthcheck endpoint","status":"MERGED","created":"2023-07-20 07:34:50.000000000","updated":"2023-10-24 14:32:23.000000000","submitted":"2023-10-24 14:31:02.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"888992-tmp1","meta_rev_id":"f1223361a9389be5b892b9445e4dfe5a0ca9f340","_number":888992,"virtual_id_number":888992,"owner":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-10-24 14:31:02.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2023-10-23 12:54:06.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":2,"date":"2023-10-24 12:07:33.000000000","_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":1,"date":"2023-10-24 12:07:33.000000000","_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}],"CC":[{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-07-20 07:34:50.000000000","updated_by":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"reviewer":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"state":"CC"},{"updated":"2023-07-20 10:40:16.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-10-12 18:11:11.000000000","updated_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"reviewer":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"state":"REVIEWER"},{"updated":"2023-10-17 15:54:07.000000000","updated_by":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"reviewer":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"state":"CC"},{"updated":"2023-10-18 16:10:09.000000000","updated_by":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"reviewer":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"state":"REVIEWER"},{"updated":"2023-10-23 12:54:06.000000000","updated_by":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"reviewer":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"state":"CC"}],"messages":[{"id":"0556cd901acad36563b83c8d2b65cb2bc8dfe407","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2023-07-20 07:34:50.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"fdaab72f08fd6aea6156909698d066a045c78d41","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-20 10:40:16.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/4187668dc2834bd1bf59edd6c19a5e06\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/e50b8afa577c4a19ab4034cf529a1b70 : SUCCESS in 10m 32s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/edcc91c124d141958e86692c45e2ea4a : FAILURE in 3m 52s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b5cb7bcd50a147659664bd3d8da6f4a6 : SUCCESS in 4m 58s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/0b4c0dd0cfc8480f8b62bed0c3bc8499 : SUCCESS in 4m 51s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/e6e838f2e8414f9aac21312df1f9cbd0 : SUCCESS in 6m 27s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/f8bd5e30e76b43ffaf599b287c75de05 : SUCCESS in 5m 16s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/74a7a90d051f4bda9132fab38f7a3337 : SUCCESS in 12m 09s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/70070a0b10aa4291840804126911bfc0 : SUCCESS in 4m 31s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/ccb5b37cd9ac49ca99d20b431d9adfb1 : SUCCESS in 5m 57s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/a3d317833df245568deb55373b00add0 : SUCCESS in 9m 52s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/08cd7eb2e4494c36a25e5aa468ec5cc4 : SUCCESS in 5m 21s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/92d80320a44d43a7a73e5653211945f5 : SUCCESS in 9m 32s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/496026617dbd427599d5d61dd438ccb5 : SUCCESS in 1h 47m 36s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/7f8a76edb2284ab5add3d7a0cc8e581e : SUCCESS in 2h 21m 43s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/30df43a7129a49368a84d64c181cec85 : SUCCESS in 44m 58s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/0072fe8a8c554ea48aa645d79a4ea919 : SUCCESS in 44m 36s\n- octavia-v2-dsvm-tls-barbican-fips https://zuul.opendev.org/t/openstack/build/877e77026a3543d6accb831fe733e400 : FAILURE in 12m 39s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/d0f8d1d14eb14db28dd4f732d74295c3 : SUCCESS in 42m 06s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/0b916d1d3b094fdbbf941a26cfe0b18e : SUCCESS in 2h 23m 24s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/f2f1f607e3ba4080964cd75cc4746ce6 : FAILURE in 15m 37s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/2cb6143c3bca4a038a97468b435dab24 : FAILURE in 5m 54s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/2762f8153ff741a4959ec9d5e8f88df1 : TIMED_OUT in 2h 34m 15s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/3f7a3d4dd8d74fecaf76fa3527e138d0 : SUCCESS in 2h 28m 25s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8 https://zuul.opendev.org/t/openstack/build/ecfb05a3d64741c6a8b590599b6d08ed : NODE_FAILURE Node request 300-0021734092 failed in 0s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"5e5b78f3bb5563527a6befcbfa1b9c38ef783419","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2023-10-12 18:11:04.000000000","message":"Patch Set 2: Patch Set 1 was rebased\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"1cb26745aefdc407775503a9c71a436c33942b78","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2023-10-12 18:11:11.000000000","message":"Patch Set 2: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"5479b8851dbf3cf8eda1c837e627ddffcaf3bb2b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-12 20:39:25.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/6c53f84c4ef94d318fbb56ac9fc5f6f3\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/26533bb67e1642c6a7ea16e249f18b48 : SUCCESS in 9m 06s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3a663b2fada147128ce30fa0412b8c5c : SUCCESS in 5m 43s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/fcddf52ed229431bbe302997a8378d24 : SUCCESS in 5m 34s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/c23ed383ce4d45f6b76b5d185a77eb83 : SUCCESS in 5m 24s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/bf521c1a8cc84e0483d08f9bde33ba69 : SUCCESS in 5m 59s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4f173b369e19469baab7df05e812e320 : SUCCESS in 12m 48s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/702341daa699435aaf4c9ffbe5b1e592 : SUCCESS in 4m 45s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/68c16651cec94a0d9deb699730e25080 : SUCCESS in 6m 30s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/25b0347bb56444feab585e5df443596f : SUCCESS in 8m 19s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/bfee52c6f99c40769ea3a28bd9085daf : SUCCESS in 4m 46s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/8922079058b04b36bf1e1478d9f75555 : SUCCESS in 6m 43s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/b2d15d5cb5174358ab2c3f6087d001a3 : SUCCESS in 1h 40m 51s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/854c31a84e2940a0bef2379c05c5eb84 : FAILURE in 1h 56m 35s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/53a1f5a8291346809709afafbeb8737a : SUCCESS in 37m 03s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/2d79784e386c482ebe36e89569922d54 : SUCCESS in 38m 45s\n- octavia-v2-dsvm-tls-barbican-fips https://zuul.opendev.org/t/openstack/build/9d999585e01a4897835bc99e1799e459 : FAILURE in 9m 36s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/b8885c3b144e411eaa63741d658fa947 : SUCCESS in 29m 23s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/5c784290ed1647d0b02869554ebfb4ec : SUCCESS in 1h 59m 04s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/bba29e005f7e425f9f610959d4c5f444 : FAILURE in 12m 20s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/61829d1418b549349018607b30f35ccd : FAILURE in 5m 09s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/7f5b054f2065462bb98233f020e7a7ae : SUCCESS in 1h 51m 02s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/467de0f403714f44b9213b7db0540dfb : SUCCESS in 2h 26m 12s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8 https://zuul.opendev.org/t/openstack/build/7d87dd0350f845fd831d07a4924b89ee : NODE_FAILURE Node request 300-0022457205 failed in 0s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"123105f07690752fc1d8e3c168879205a020091e","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2023-10-17 15:54:07.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"d6d2cfaf9c9784ebd299bd5b8917ac72e3d61495","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2023-10-18 15:57:34.000000000","message":"Patch Set 3: Patch Set 2 was rebased\n\nCopied Votes:\n* Code-Review+2 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":3},{"id":"3f228b98715090082d07723d95cf4b4eca07f583","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2023-10-18 16:10:09.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"890a22f1ce198aef250676d6100fc96582504c61","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2023-10-23 12:54:06.000000000","message":"Uploaded patch set 4: Patch Set 3 was rebased.\n\nCopied Votes:\n* Code-Review+2 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n\nOutdated Votes:\n* Workflow+1\n","accounts_in_message":[],"_revision_number":4},{"id":"70e539023adbca4385180f908b1ed5f7efbf645b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-23 15:14:39.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/537d7860d8104be4a62018e5656887a1\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/e7d3f1dea5ed47719bacd9f501bee6d5 : SUCCESS in 8m 32s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/122c63a5b28e41a48454dabbe446a1b6 : SUCCESS in 5m 42s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/730c845bf5c74ace86d8e3e6864d0b47 : SUCCESS in 4m 28s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/8683fd629ff1498785ab8e264c4eeee2 : SUCCESS in 4m 28s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/be3ff96481b24be18c75cd887e401abb : SUCCESS in 5m 41s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1643987062c24e319f0aba933e518993 : SUCCESS in 10m 09s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/45939fbf3dd1413984922d346124c3c8 : SUCCESS in 5m 00s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/5bcefac8e700456b8056ef49c45691dd : SUCCESS in 5m 11s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/21f879b85a644c79ad82a7ef703e9046 : SUCCESS in 9m 04s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/cafb896e1bbb4be1a1650c802181464e : SUCCESS in 5m 28s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/ff1b31558aa94cbd97b69b427bebe5f4 : SUCCESS in 6m 55s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/25f69fb517824c48a4cfa2168b19b6d6 : SUCCESS in 1h 36m 14s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/2b197db6868549a19160daad93040e64 : SUCCESS in 2h 12m 29s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/72bf3f6ba2194ee5ae26ce859a13cfe8 : SUCCESS in 38m 07s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/b5ffedac37124c60bd0a43380f7f0b8e : SUCCESS in 54m 28s\n- octavia-v2-dsvm-tls-barbican-fips https://zuul.opendev.org/t/openstack/build/f87f536027b64c9480a48d2e45955a81 : FAILURE in 9m 46s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/9f3763a7c0534e949bf22da4d79ca215 : SUCCESS in 39m 53s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/aff6943365c44ba2a1c35464ca76880c : SUCCESS in 2h 03m 27s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/4c582561fcae4cbda5c45659f628d03d : FAILURE in 12m 42s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/477066d8834c404a819c80cea9b67122 : FAILURE in 5m 18s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/7c96f719146f4e79be1eb33a2204afbb : SUCCESS in 1h 40m 56s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/0876635d32e245069450f2af22a526fd : SUCCESS in 1h 41m 55s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8 https://zuul.opendev.org/t/openstack/build/fcb94949c35240119f35d903c358401b : NODE_FAILURE Node request 300-0022536583 failed in 0s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"0f9cd571a676ae06c3b788dd619760f914999fd0","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2023-10-24 12:07:33.000000000","message":"Patch Set 4: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":4},{"id":"f85563b683a1b7c3ef97eaab80d9ede4d37dcc3a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-24 12:08:08.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"ca1e49bd5f9fbbe838b874fa17744a7cc5d65ce2","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-24 14:31:02.000000000","message":"Patch Set 4: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2af540a86d664d21be15f7ac6322cc87\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2422a9eaa0ce40dea2b21bac75782248 : SUCCESS in 4m 47s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/e923ef1176894a2bb5dbbfafaf530a89 : SUCCESS in 4m 45s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/2e7580084a964cf9bf20247bc1755da2 : SUCCESS in 4m 47s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/056da1fc6414477885f4165257e6b066 : SUCCESS in 10m 26s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/831ccc9d84664bb7a73b89038410f39d : SUCCESS in 5m 38s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/de638824abd746c9a36b3f14b71d171d : SUCCESS in 4m 45s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/c7b577a6ce8f4187a1c376bcfb38511c : SUCCESS in 7m 55s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/ecf129da44f04f00ba4b3a0bec4ccd82 : SUCCESS in 1h 02m 22s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/2dda1c0d36ec4df48febe3afdf100edd : SUCCESS in 2h 22m 49s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/34199577c595496d9b7a02e8c7677548 : SUCCESS in 43m 33s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/c3056cb62e9446049f35b21d2d930043 : SUCCESS in 54m 46s","accounts_in_message":[],"_revision_number":4},{"id":"40abe3cb2759739ee595d43cfd9a04f448ede720","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-24 14:31:02.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":4},{"id":"f1223361a9389be5b892b9445e4dfe5a0ca9f340","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-24 14:32:23.000000000","message":"Patch Set 4:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a15b1dd780534b488c8f281f5335fb53\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/074f6bcc5e0b454c979fefeee092f69b : SUCCESS in 1m 03s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/2f9b8451677f446a8b045b7ae8de39af : SUCCESS in 40s","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"8219f39e21216c97b4c12defef6e0597bbfcf658","revisions":{"972b5ad70050a2df2ebf53ca01ef8a23abccb4c0":{"kind":"REWORK","_number":1,"created":"2023-07-20 07:34:50.000000000","uploader":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"ref":"refs/changes/92/888992/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/92/888992/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/92/888992/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/92/888992/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/92/888992/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/92/888992/1"}}},"commit":{"parents":[{"commit":"c5b5bc98c1f6e780a342d03cccd8daa3a8108085","subject":"Merge \"Fix hm operating status to ONLINE in single lb call\" into stable/xena","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/c5b5bc98c1f6e780a342d03cccd8daa3a8108085"}]}],"author":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2023-02-17 00:21:19.000000000","tz":0},"committer":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2023-07-20 07:34:46.000000000","tz":120},"subject":"Fix text/plain mime type with healthcheck endpoint","message":"Fix text/plain mime type with healthcheck endpoint\n\nPreviously if you used an accept header of text/plain when querying the healthcheck endpoint, you would get a 406 Not Acceptable response.\nThis patch corrects that to allow requesting the healthcheck document in text format.\n\nChange-Id: I0e2d06b02c06b8c77fb1c2f94207d422066a0c6d\n(cherry picked from commit 8db655aed2e579ace3e778e100f1baa8cd13000c)\n(cherry picked from commit 106376d3fd9486aee9fd60a35a504d10b436a754)\n(cherry picked from commit ffdcfb486c2ca916129b62fcb5e4f18880785317)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/972b5ad70050a2df2ebf53ca01ef8a23abccb4c0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/972b5ad70050a2df2ebf53ca01ef8a23abccb4c0"}]},"branch":"refs/heads/stable/xena"},"bd1af1282a10d556a14aafd5ad44e529b9c7a3c9":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2023-10-12 18:11:04.000000000","uploader":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"ref":"refs/changes/92/888992/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/92/888992/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/92/888992/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/92/888992/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/92/888992/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/92/888992/2"}}},"commit":{"parents":[{"commit":"37e798e05a0b0e3b21ee12eea049c2eaf2e4b86a","subject":"Merge \"fix haproxy_count to display the number of running processes\" into stable/xena","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/37e798e05a0b0e3b21ee12eea049c2eaf2e4b86a"}]}],"author":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2023-02-17 00:21:19.000000000","tz":0},"committer":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2023-10-12 18:11:04.000000000","tz":0},"subject":"Fix text/plain mime type with healthcheck endpoint","message":"Fix text/plain mime type with healthcheck endpoint\n\nPreviously if you used an accept header of text/plain when querying the healthcheck endpoint, you would get a 406 Not Acceptable response.\nThis patch corrects that to allow requesting the healthcheck document in text format.\n\nChange-Id: I0e2d06b02c06b8c77fb1c2f94207d422066a0c6d\n(cherry picked from commit 8db655aed2e579ace3e778e100f1baa8cd13000c)\n(cherry picked from commit 106376d3fd9486aee9fd60a35a504d10b436a754)\n(cherry picked from commit ffdcfb486c2ca916129b62fcb5e4f18880785317)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/bd1af1282a10d556a14aafd5ad44e529b9c7a3c9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/bd1af1282a10d556a14aafd5ad44e529b9c7a3c9"}]},"branch":"refs/heads/stable/xena","description":"Rebase"},"70cdc8770b2aa33db4cd6b8ba146e4c1824aaa61":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2023-10-18 15:57:34.000000000","uploader":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"ref":"refs/changes/92/888992/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/92/888992/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/92/888992/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/92/888992/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/92/888992/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/92/888992/3"}}},"commit":{"parents":[{"commit":"c08afbec81a1e9d263e104af67ad45521af307de","subject":"Merge \"Fix octavia-status with amphorav2\" into stable/xena","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/c08afbec81a1e9d263e104af67ad45521af307de"}]}],"author":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2023-02-17 00:21:19.000000000","tz":0},"committer":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2023-10-18 15:57:34.000000000","tz":0},"subject":"Fix text/plain mime type with healthcheck endpoint","message":"Fix text/plain mime type with healthcheck endpoint\n\nPreviously if you used an accept header of text/plain when querying the healthcheck endpoint, you would get a 406 Not Acceptable response.\nThis patch corrects that to allow requesting the healthcheck document in text format.\n\nChange-Id: I0e2d06b02c06b8c77fb1c2f94207d422066a0c6d\n(cherry picked from commit 8db655aed2e579ace3e778e100f1baa8cd13000c)\n(cherry picked from commit 106376d3fd9486aee9fd60a35a504d10b436a754)\n(cherry picked from commit ffdcfb486c2ca916129b62fcb5e4f18880785317)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/70cdc8770b2aa33db4cd6b8ba146e4c1824aaa61"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/70cdc8770b2aa33db4cd6b8ba146e4c1824aaa61"}]},"branch":"refs/heads/stable/xena","description":"Rebase"},"8219f39e21216c97b4c12defef6e0597bbfcf658":{"kind":"TRIVIAL_REBASE","_number":4,"created":"2023-10-23 12:54:06.000000000","uploader":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"ref":"refs/changes/92/888992/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/92/888992/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/92/888992/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/92/888992/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/92/888992/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/92/888992/4"}}},"commit":{"parents":[{"commit":"195383c0c2ecf5b35afcd6028b45b22b52bfc9aa","subject":"Merge \"Prevent incorrect reschedule of resumed tasks with jobboard\" into stable/xena","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/195383c0c2ecf5b35afcd6028b45b22b52bfc9aa"}]}],"author":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2023-02-17 00:21:19.000000000","tz":0},"committer":{"name":"Tom Weininger","email":"tweining@redhat.com","date":"2023-10-23 12:53:23.000000000","tz":120},"subject":"Fix text/plain mime type with healthcheck endpoint","message":"Fix text/plain mime type with healthcheck endpoint\n\nPreviously if you used an accept header of text/plain when querying the healthcheck endpoint, you would get a 406 Not Acceptable response.\nThis patch corrects that to allow requesting the healthcheck document in text format.\n\nChange-Id: I0e2d06b02c06b8c77fb1c2f94207d422066a0c6d\n(cherry picked from commit 8db655aed2e579ace3e778e100f1baa8cd13000c)\n(cherry picked from commit 106376d3fd9486aee9fd60a35a504d10b436a754)\n(cherry picked from commit ffdcfb486c2ca916129b62fcb5e4f18880785317)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/8219f39e21216c97b4c12defef6e0597bbfcf658"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/8219f39e21216c97b4c12defef6e0597bbfcf658"}]},"branch":"refs/heads/stable/xena"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
