)]}'
{"id":"openstack%2Foctavia~888993","triplet_id":"openstack%2Foctavia~stable%2Fwallaby~I0e2d06b02c06b8c77fb1c2f94207d422066a0c6d","project":"openstack/octavia","branch":"stable/wallaby","attention_set":{},"removed_from_attention_set":{"11628":{"account":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"last_update":"2023-10-18 09:15:25.000000000","reason":"Change was submitted"},"29244":{"account":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"last_update":"2023-10-18 09:15:25.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I0e2d06b02c06b8c77fb1c2f94207d422066a0c6d","subject":"Fix text/plain mime type with healthcheck endpoint","status":"MERGED","created":"2023-07-20 07:35:35.000000000","updated":"2023-10-18 09:16:44.000000000","submitted":"2023-10-18 09:15:25.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"888993","meta_rev_id":"af01c1d506b56984a9ed3a887098f074949b44f3","_number":888993,"virtual_id_number":888993,"owner":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-10-18 09:15:25.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"all":[{"value":2,"date":"2023-10-12 18:15:49.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2023-10-18 07:05:01.000000000","_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"all":[{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2023-10-18 07:05:01.000000000","_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-07-20 07:35:35.000000000","updated_by":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"reviewer":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"state":"CC"},{"updated":"2023-07-20 09:54:56.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-10-12 18:15:49.000000000","updated_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"reviewer":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"state":"REVIEWER"},{"updated":"2023-10-18 07:05:01.000000000","updated_by":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"reviewer":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"state":"REVIEWER"}],"messages":[{"id":"e1ad56111ffe7033b78f43b2dcac214ce5a6cb1e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2023-07-20 07:35:35.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"3992a7c250cb17b32112ffb25bd7c5b1437407d5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-20 09:54:56.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/0ba5e70547214659a9921dc5502ee8f0\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/324a21a62d534fba80585b9d50bad08e : SUCCESS in 10m 11s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/baacbd9f55f2436894a72e288f047a03 : FAILURE in 3m 56s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ea0dd474e541417cbc9962489ac37e73 : SUCCESS in 5m 34s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/62cb64ca3cba4c6ab329134dac27c560 : SUCCESS in 6m 36s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/c51585424cc749e28e87f4ba3dd931a4 : SUCCESS in 5m 21s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/ccd8684bde204042a47195e786e53afb : SUCCESS in 5m 21s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5c0b2ebf882e4d1698d632ed961a7e5d : SUCCESS in 12m 03s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/4799849ab99c445884d96470d95ee08b : SUCCESS in 5m 01s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/82ecd18bb574434498468e8741ed87f6 : SUCCESS in 6m 21s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/bb95f05d055f4e7f970715362cd32cbf : SUCCESS in 8m 44s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/e4f88e90d4fa4c3aa1b577fc634d5c07 : SUCCESS in 6m 16s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/dcef52dc14094d03b853a1ff6392923c : SUCCESS in 8m 12s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/d420a5b63dd94f0e8f865f53b22b5fb6 : SUCCESS in 1h 34m 08s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/2e5829b06bd64854a2cc7c42806a8f7d : SUCCESS in 1h 31m 12s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/77028d8bd60149e591852336a0c28b71 : SUCCESS in 40m 34s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/d946cb1f9d7a48f1b06bb3d99d6fa09c : SUCCESS in 34m 12s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/8311cb2948d445e8b5ae34f1b30d563d : SUCCESS in 29m 49s\n- octavia-v2-dsvm-tls-barbican-fips https://zuul.opendev.org/t/openstack/build/c6c2c35ab392495abaad36a162b272ca : FAILURE in 12m 01s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/071e5a97299145f6b4aab16f8e89fa05 : SUCCESS in 1h 33m 05s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/3de2984be73946ccb8912ea5a7f2d22f : FAILURE in 9m 44s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/15a88ed0544e49bea506a3b1d279c298 : FAILURE in 5m 28s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/ed230f5a989c464bab1818f2f65c5421 : SUCCESS in 2h 09m 11s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/4d1e2f07da704fddac47491f369cdbf1 : SUCCESS in 1h 55m 48s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8 https://zuul.opendev.org/t/openstack/build/9abba8127e34425e903e840ef21d9837 : NODE_FAILURE Node request 300-0021734116 failed in 0s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"90da2853c7bb16b8c7faa01e998ed7f1fd3824ed","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2023-10-12 18:14:44.000000000","message":"Patch Set 2: Patch Set 1 was rebased\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"4a3c03e831d598e8e4d7b84fbf0c99b41cf9c38a","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2023-10-12 18:15:49.000000000","message":"Patch Set 2: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"563ddf7770bfdd5794136336bdfef28a0e0daec4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-12 20:47:40.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/de78926c6e65407abe6ebab841526b27\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/024483e567324345a44fb021c4473488 : SUCCESS in 7m 50s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/454aa43f5c52451c938a26cbbb6b3636 : SUCCESS in 5m 06s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/1630e4e6be1d483e99e237ef9feb9123 : SUCCESS in 4m 48s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/a305f4c1f364461f934665b11546e665 : SUCCESS in 4m 36s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/428a836a86ee45ab9e68b58d01d143bc : SUCCESS in 5m 02s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/97426ea32ef7400785cfb62e6c2bcf28 : SUCCESS in 10m 05s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/e1446e91530f47d8842fb7c674dee665 : SUCCESS in 5m 54s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/54efcc56d0f34899a4d4edb8722432e1 : SUCCESS in 4m 41s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/b020493548c042f88a9d4fa23ce2ab3c : SUCCESS in 9m 39s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/0e1824acba2940358242b82444d9d101 : SUCCESS in 5m 22s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/7ca9a37136974925834c23bee4187a9e : SUCCESS in 7m 39s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/57c50cc347194f48b9da820f00f3efa5 : SUCCESS in 45m 30s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/3b34b6bea9374069a293164acab5ba10 : SUCCESS in 1h 29m 53s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/9a21398601b5498d87b499c9995a9c4e : SUCCESS in 35m 56s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/20d1c445f0674726ba18205f70b294ca : SUCCESS in 31m 47s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/6b1526bdc3f24de081172fa68b0fd85b : SUCCESS in 32m 26s\n- octavia-v2-dsvm-tls-barbican-fips https://zuul.opendev.org/t/openstack/build/bdef767be3d44d3a9e2c1d438cb8a201 : FAILURE in 10m 58s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/5e228e9f115a4758bc8ee7bc4bb812c1 : SUCCESS in 1h 24m 32s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/cd00edba458848cd96f23de6c45d1bad : FAILURE in 11m 12s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/1d1c36940fdc4b2f93028d5036bd685b : FAILURE in 5m 51s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/cc28f8cf973842ceb6cb2c23ed07d254 : SUCCESS in 2h 29m 37s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/826caff547774a45b894367b32477ed8 : SUCCESS in 2h 00m 56s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8 https://zuul.opendev.org/t/openstack/build/5ca3a91cad5e46d7a3c99b20bc70eb6e : NODE_FAILURE Node request 300-0022457323 failed in 0s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"a7ebb1d627d9db8aab6f4e18e14718fa6413dd5a","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2023-10-18 07:05:01.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"5f0ce610b047b6346441f47c9dc79382694e4530","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-18 07:05:36.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"e48c383867a79b7df208712064e0580e17506b7d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-18 09:15:25.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e559295bbe93403493eadeecf214a46c\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/33560e618fdc4057a3e2b4eb8054a931 : SUCCESS in 5m 13s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/e9cb228aeb7847b69968d647af1d6de4 : SUCCESS in 4m 47s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/68b1d2fa26db461a845ea21a44f31d98 : SUCCESS in 4m 19s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/336beba6a1c341eca195e64034f29bac : SUCCESS in 13m 08s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/aac3297cd4fe48189d012af612f8b406 : SUCCESS in 4m 30s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/24497db5f6064917846d68b43c5b7a1a : SUCCESS in 5m 11s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/707e69553cb747de854f9d3c47cc0d46 : SUCCESS in 6m 46s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/1e5fdeca91224ef8b0c202f006a40e3d : SUCCESS in 48m 46s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/b6e5a2997e004dd886d1542af25b96fe : SUCCESS in 1h 34m 44s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/115ca83f5f0847e8a4864fe9761fe1cc : SUCCESS in 33m 48s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/b23fe8c86ba3462196ed9d352920f553 : SUCCESS in 32m 06s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/baed1cabbbc0405a8ae7f2fd284e0301 : SUCCESS in 47m 37s","accounts_in_message":[],"_revision_number":2},{"id":"1202b97e380c5a869c1b93109997c3bea1e1410b","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-18 09:15:25.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"af01c1d506b56984a9ed3a887098f074949b44f3","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-18 09:16:44.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/733bce2aa49f4916b7762c2bbd4acbd7\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fc13f96d924b4580ab7aa2b2dfe17ecc : SUCCESS in 59s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/a876970b4bb54e9aa8a9985a566b0d78 : SUCCESS in 44s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"c13a867c4f4d50e7ebac2de9dcea39cf2185ee0b","revisions":{"56c00bbf8298d5b530e6e4fcd9de70e8b1aeec09":{"kind":"REWORK","_number":1,"created":"2023-07-20 07:35:35.000000000","uploader":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"ref":"refs/changes/93/888993/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/93/888993/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/93/888993/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/93/888993/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/93/888993/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/93/888993/1"}}},"commit":{"parents":[{"commit":"8af6306df63df55d47a56c2c222eb5a254b88574","subject":"Merge \"Fix hm operating status to ONLINE in single lb call\" into stable/wallaby","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/8af6306df63df55d47a56c2c222eb5a254b88574"}]}],"author":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2023-02-17 00:21:19.000000000","tz":0},"committer":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2023-07-20 07:35:32.000000000","tz":120},"subject":"Fix text/plain mime type with healthcheck endpoint","message":"Fix text/plain mime type with healthcheck endpoint\n\nPreviously if you used an accept header of text/plain when querying the healthcheck endpoint, you would get a 406 Not Acceptable response.\nThis patch corrects that to allow requesting the healthcheck document in text format.\n\nChange-Id: I0e2d06b02c06b8c77fb1c2f94207d422066a0c6d\n(cherry picked from commit 8db655aed2e579ace3e778e100f1baa8cd13000c)\n(cherry picked from commit 106376d3fd9486aee9fd60a35a504d10b436a754)\n(cherry picked from commit ffdcfb486c2ca916129b62fcb5e4f18880785317)\n(cherry picked from commit 972b5ad70050a2df2ebf53ca01ef8a23abccb4c0)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/56c00bbf8298d5b530e6e4fcd9de70e8b1aeec09"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/56c00bbf8298d5b530e6e4fcd9de70e8b1aeec09"}]},"branch":"refs/heads/stable/wallaby"},"c13a867c4f4d50e7ebac2de9dcea39cf2185ee0b":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2023-10-12 18:14:44.000000000","uploader":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"ref":"refs/changes/93/888993/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/93/888993/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/93/888993/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/93/888993/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/93/888993/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/93/888993/2"}}},"commit":{"parents":[{"commit":"77e14a8d790bfb0facaf2ee56c1f798e58269b13","subject":"Fix race condition in members batch update API call","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/77e14a8d790bfb0facaf2ee56c1f798e58269b13"}]}],"author":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2023-02-17 00:21:19.000000000","tz":0},"committer":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2023-10-12 18:14:44.000000000","tz":0},"subject":"Fix text/plain mime type with healthcheck endpoint","message":"Fix text/plain mime type with healthcheck endpoint\n\nPreviously if you used an accept header of text/plain when querying the healthcheck endpoint, you would get a 406 Not Acceptable response.\nThis patch corrects that to allow requesting the healthcheck document in text format.\n\nChange-Id: I0e2d06b02c06b8c77fb1c2f94207d422066a0c6d\n(cherry picked from commit 8db655aed2e579ace3e778e100f1baa8cd13000c)\n(cherry picked from commit 106376d3fd9486aee9fd60a35a504d10b436a754)\n(cherry picked from commit ffdcfb486c2ca916129b62fcb5e4f18880785317)\n(cherry picked from commit 972b5ad70050a2df2ebf53ca01ef8a23abccb4c0)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/c13a867c4f4d50e7ebac2de9dcea39cf2185ee0b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/c13a867c4f4d50e7ebac2de9dcea39cf2185ee0b"}]},"branch":"refs/heads/stable/wallaby","description":"Rebase"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
