)]}'
{"id":"openstack%2Foctavia~889000","triplet_id":"openstack%2Foctavia~stable%2F2023.1~I16430fa52db02e7445203994220673c1037d764c","project":"openstack/octavia","branch":"stable/2023.1","attention_set":{},"removed_from_attention_set":{"29244":{"account":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"last_update":"2024-01-25 12:49:16.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I16430fa52db02e7445203994220673c1037d764c","subject":"Fix upgrade check not working","status":"MERGED","created":"2023-07-20 07:56:29.000000000","updated":"2024-01-25 12:50:46.000000000","submitted":"2024-01-25 12:49:16.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"889000","meta_rev_id":"97adb6f7b3602c0c5b94312f389811973ae5ed40","_number":889000,"virtual_id_number":889000,"owner":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},{"value":0,"_account_id":27442,"name":"Quentin GROLLEAU","email":"quentin.grolleau@corp.ovh.com","username":"QG"},{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-01-25 12:49:15.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"recommended":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"all":[{"value":1,"date":"2023-07-21 12:42:30.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},{"value":1,"date":"2024-01-24 16:46:32.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":27442,"name":"Quentin GROLLEAU","email":"quentin.grolleau@corp.ovh.com","username":"QG"},{"value":1,"date":"2023-07-28 07:55:17.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":2,"date":"2023-10-12 18:00:57.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2024-01-25 10:24:38.000000000","_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"all":[{"value":0,"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},{"value":0,"_account_id":27442,"name":"Quentin GROLLEAU","email":"quentin.grolleau@corp.ovh.com","username":"QG"},{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2024-01-25 10:24:38.000000000","_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},{"value":0,"_account_id":27442,"name":"Quentin GROLLEAU","email":"quentin.grolleau@corp.ovh.com","username":"QG"},{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":27442,"name":"Quentin GROLLEAU","email":"quentin.grolleau@corp.ovh.com","username":"QG"},{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-07-20 07:56:29.000000000","updated_by":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"reviewer":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"state":"CC"},{"updated":"2023-07-20 11:09:19.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-07-21 12:42:30.000000000","updated_by":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"reviewer":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"state":"REVIEWER"},{"updated":"2023-07-28 07:55:17.000000000","updated_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"reviewer":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"state":"REVIEWER"},{"updated":"2023-10-12 18:00:57.000000000","updated_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"reviewer":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"state":"REVIEWER"},{"updated":"2024-01-24 16:46:32.000000000","updated_by":{"_account_id":27442,"name":"Quentin GROLLEAU","email":"quentin.grolleau@corp.ovh.com","username":"QG"},"reviewer":{"_account_id":27442,"name":"Quentin GROLLEAU","email":"quentin.grolleau@corp.ovh.com","username":"QG"},"state":"REVIEWER"},{"updated":"2024-01-25 10:24:38.000000000","updated_by":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"reviewer":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"state":"REVIEWER"}],"messages":[{"id":"9ec5d8c660ffba4556af45437e03ec8a96cd44c6","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2023-07-20 07:56:29.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"d2eb7f57533ed6d1784310a28281551f75716959","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-20 11:09:19.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/82b28c4e6db148cfaab94cd99bf56fb4\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/0430a910d36e4d76a287eb31cab21f5e : SUCCESS in 10m 20s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/76e0accc09b34788bb3e8d352c548ed3 : SUCCESS in 5m 37s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/b8f8cedc77c74674b88040f13d5fc664 : SUCCESS in 5m 10s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/033dcdea85234b1a806924341f4496d6 : SUCCESS in 4m 21s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f0e8ad65336545a9b6b58c91c373a474 : SUCCESS in 11m 21s\n- octavia-tox-py39-tips https://zuul.opendev.org/t/openstack/build/89feda4bd7a743008937ea9ddbab8a69 : SUCCESS in 4m 39s\n- octavia-tox-functional-py39-tips https://zuul.opendev.org/t/openstack/build/cd55192b8c684a0da324679b19dc59e7 : SUCCESS in 8m 32s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/5cfdfc8a1f63433b970324ed98768478 : SUCCESS in 4m 35s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/7e5e8530e5904a4fb25115797d88b749 : SUCCESS in 8m 13s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/fb43736282914dd79bc66e6176c7e6b9 : SUCCESS in 8m 24s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/cfa6d996613e4f3699687bd65b7c041b : SUCCESS in 8m 48s (non-voting)\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/a2528043939a4d70b372d001baca5c4b : SUCCESS in 1h 20m 46s\n- octavia-v2-dsvm-scenario-traffic-ops https://zuul.opendev.org/t/openstack/build/2597e48c4ceb4106be9d91541ca82330 : SUCCESS in 1h 49m 33s\n- octavia-v2-dsvm-scenario-non-traffic-ops https://zuul.opendev.org/t/openstack/build/17953c54416c482f98fce242a7e2aa36 : SUCCESS in 1h 15m 57s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/939f0a3ad7764ff9b30d4d836baf0766 : SUCCESS in 49m 29s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/7505be053b984814b0cf0e9588ffb734 : SUCCESS in 52m 37s\n- octavia-grenade-skip-level https://zuul.opendev.org/t/openstack/build/518c23df664a4d608ab21585fc0c7064 : SUCCESS in 57m 12s\n- octavia-v2-dsvm-tls-barbican-fips https://zuul.opendev.org/t/openstack/build/53879e8057284d9aa6cece481722db7d : FAILURE in 17m 55s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/3f1a97c4bcff42679d5535db04d2e6a0 : SUCCESS in 43m 41s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/541d62d51b8a4a52a2b99e33f67f071a : SUCCESS in 2h 25m 09s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/348394550082458fa1c6c6372eae604d : SUCCESS in 2h 46m 09s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/279e7a375cac4e3da851271157c1fa24 : SUCCESS in 2h 30m 27s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/9703216a0a6941be87ea798f8aac7561 : TIMED_OUT in 2h 34m 16s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/b12e66a8803144f194b28f666174aa46 : TIMED_OUT in 2h 32m 45s (non-voting)\n- octavia-v2-dsvm-scenario-centos-9-stream https://zuul.opendev.org/t/openstack/build/43978362215a4e1c9cafe07407a5b94a : SUCCESS in 2h 15m 38s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"ef742cf8b1809912233507e3d090db7434432c4f","author":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"date":"2023-07-21 12:42:30.000000000","message":"Patch Set 1: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"ca7659d43ef6ebfe48582f097e4ea6a93e6de8f5","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2023-07-28 07:55:17.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"691db85a354d67fcb238f7d14f9970604bfe51f5","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2023-10-12 18:00:57.000000000","message":"Patch Set 1: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"7552b9626949c883612bb8bbab43a454d7089183","author":{"_account_id":27442,"name":"Quentin GROLLEAU","email":"quentin.grolleau@corp.ovh.com","username":"QG"},"date":"2024-01-24 16:46:32.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"bc4b7f7bc599449990e3136d80a889b4bc9aadcd","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2024-01-25 10:24:38.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"c5c60402e6a452aaf9a16e44c4d2a67761adfe48","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-25 10:25:26.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"9cd828378499289c3be4798052b45225bd37b57a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-25 12:49:15.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7f4138b2cc3b4db084746bfafaca2e41\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1b18540e2b774330b40ed4a82db03902 : SUCCESS in 6m 30s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/0a13064e87d24f1098fc079dcbbfd028 : SUCCESS in 5m 27s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5fa13bc1f515462ca84bb89232b5d5a4 : SUCCESS in 13m 53s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/a4ec5d058beb4f3ca8ab7a2b8c06c629 : SUCCESS in 5m 11s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/ca48b0714bc94d27b46e714cc790a426 : SUCCESS in 9m 26s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/6c6a3ab5ac364f30a9737e92a5ce1a76 : SUCCESS in 1h 33m 59s\n- octavia-v2-dsvm-scenario-traffic-ops https://zuul.opendev.org/t/openstack/build/615d486c0af647319e466f0224f90243 : SUCCESS in 2h 00m 32s\n- octavia-v2-dsvm-scenario-non-traffic-ops https://zuul.opendev.org/t/openstack/build/830b97928dad48aa8a756308f9a30e6e : SUCCESS in 1h 23m 03s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/8233e52294124b0fbefd199772ee0585 : SUCCESS in 49m 08s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/8cdebdd76bcb4524bf6f355f57feca01 : SUCCESS in 37m 27s\n- octavia-grenade-skip-level https://zuul.opendev.org/t/openstack/build/ceb197dfd2c841319e5558949afda92d : SUCCESS in 1h 05m 13s","accounts_in_message":[],"_revision_number":1},{"id":"aded740aad3747d35aa3ad7f59bf3d067878ce97","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-25 12:49:16.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"97adb6f7b3602c0c5b94312f389811973ae5ed40","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-25 12:50:46.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/fb9470af3b5e42cf8d119953ca3ef902\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d8edab2e43c448f59bd669bde4927df0 : SUCCESS in 1m 09s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"81a396257d3453f409f905c8a9809fc5ca0c2634","revisions":{"81a396257d3453f409f905c8a9809fc5ca0c2634":{"kind":"REWORK","_number":1,"created":"2023-07-20 07:56:29.000000000","uploader":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"ref":"refs/changes/00/889000/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/00/889000/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/00/889000/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/00/889000/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/00/889000/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/00/889000/1"}}},"commit":{"parents":[{"commit":"8b196bb3bb05c9d87dae688750457dbb944d4d1b","subject":"Fix amphorav1 member deletion bug","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/8b196bb3bb05c9d87dae688750457dbb944d4d1b"}]}],"author":{"name":"Tobias Urdin","email":"tobias.urdin@binero.se","date":"2023-03-05 22:16:16.000000000","tz":0},"committer":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2023-07-20 07:55:29.000000000","tz":120},"subject":"Fix upgrade check not working","message":"Fix upgrade check not working\n\nThe octavia-status upgrade check command fails because\nthe initialization of Policy() requires oslo_config to\nbe initialized already which it isn\u0027t.\n\nThis can be reproduced in master by running:\n\n  tox -e venv -- octavia-status upgrade check\n\nI can observe this issue multiple relases back which\nprobably means this has been broken for a long time.\n\nWe shouldn\u0027t need to init Policy() because it just\nloads policy rules which is not checked in upgrade\ncheck anyway.\n\nChange-Id: I16430fa52db02e7445203994220673c1037d764c\n(cherry picked from commit 47de8fa3444cc7b1b68d486ad8c06d294a9ef9de)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/81a396257d3453f409f905c8a9809fc5ca0c2634"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/81a396257d3453f409f905c8a9809fc5ca0c2634"}]},"branch":"refs/heads/stable/2023.1"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
