)]}'
{"id":"openstack%2Foctavia~889075","triplet_id":"openstack%2Foctavia~stable%2Fxena~Ib929fbe28241bd6d3c3bd352b2d900289efd6d49","project":"openstack/octavia","branch":"stable/xena","attention_set":{},"removed_from_attention_set":{"29244":{"account":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"last_update":"2023-07-20 17:07:58.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Ib929fbe28241bd6d3c3bd352b2d900289efd6d49","subject":"Drop lower-constraints.txt and its testing","status":"MERGED","created":"2023-07-20 10:27:22.000000000","updated":"2023-07-20 17:11:35.000000000","submitted":"2023-07-20 17:07:58.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"889075","meta_rev_id":"d44aec17b90b32faf80eac5296435336c2665b10","_number":889075,"virtual_id_number":889075,"owner":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-07-20 17:07:58.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"all":[{"value":2,"date":"2023-07-20 15:09:38.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"all":[{"value":1,"date":"2023-07-20 15:09:38.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-07-20 14:06:20.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-07-20 15:09:38.000000000","updated_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"reviewer":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"state":"REVIEWER"}],"messages":[{"id":"233c4812a9a1b6540ec8fbd3a5a77f24d7a92d97","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2023-07-20 10:27:22.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"8474f3a4d6420bb852bbb1a6553decef3ec6deb0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-20 14:06:20.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/59561c0867994071b7cf7e1c502e9188\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/436eb4ed38454b93ada930c538587a61 : SUCCESS in 2m 11s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/a4e2a89dfe674102ac00551962b868f1 : SUCCESS in 11m 51s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/071687a2ace24cbba564649112e45885 : SUCCESS in 5m 08s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/116ce984a5b846259dcbf3dd5beb4230 : SUCCESS in 4m 50s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/04aace945d244e3aab353a1d2ef570fc : SUCCESS in 5m 41s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/f114acadc07f4d6ebe3012df11aa2a55 : SUCCESS in 5m 25s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ba3f59b277124702a9f834e56b956d55 : SUCCESS in 12m 40s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/2f34cc3c92464c718e44f6e3d1bbc2d9 : SUCCESS in 4m 47s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/76927e98a9f5419d8ed942502870bc54 : SUCCESS in 6m 10s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/9bb6c00436e04745b93e5c0a49a7efc1 : SUCCESS in 8m 49s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/c9ee6cbe361b42fdb00a9af89db916df : SUCCESS in 5m 00s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/4274abe572a245ddab59e7cd7bec54eb : SUCCESS in 7m 38s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/9f3e0984ffbf4f48b0c248fa226fba80 : SUCCESS in 1h 42m 44s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/c6e7ec5231d94ce8adb75d29f81f03fa : SUCCESS in 1h 37m 05s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/b669c83469e1480a8f8efade4832ddad : SUCCESS in 34m 40s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/850f53873e25410eb217f572ee2d3305 : SUCCESS in 39m 21s\n- octavia-v2-dsvm-tls-barbican-fips https://zuul.opendev.org/t/openstack/build/fe7ba0336415401d8a1e5a27d2b35b16 : FAILURE in 9m 04s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/672f42862b9644e6a34d80b993764972 : SUCCESS in 33m 41s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/4353bdc3486840fdb90d511c0fb09a90 : SUCCESS in 1h 48m 13s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/00db280c83fd41ca98326e34cfa19bed : FAILURE in 10m 08s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/fd11ed2093614bdba4be0ebf8a48b576 : FAILURE in 5m 42s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/df04a1fa4b2e43908420a3bfc1c16baa : SUCCESS in 1h 54m 32s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/2547bd5116af4b4996630d99660e701e : TIMED_OUT in 2h 34m 17s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8 https://zuul.opendev.org/t/openstack/build/4bc0d0251f1546c4bfb1e8df504ff40a : NODE_FAILURE Node request 300-0021736680 failed in 0s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"792deef0e75808eb2af29a4f19931137cb0c0067","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2023-07-20 15:09:38.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"79519c82d4e3cc6e0e2fe6406c00c197755f63d1","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-20 15:10:30.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"314d11cad4abd9213ca92175c76e08ecf50fc6ce","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-20 17:07:58.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d5d1a1e3178146c299734411630d507b\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/0c48d8c6827043f89fc922e18f61a94a : SUCCESS in 2m 57s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a7a30c144aab4924ab0f30a458212349 : SUCCESS in 6m 22s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/77cab5a695f7482aa6abad791542ab5b : SUCCESS in 5m 46s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/1ba5dd00bc12426ab965668a5996da4f : SUCCESS in 5m 19s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b21d67df3d6d405396747fc8bde59102 : SUCCESS in 10m 24s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/f33904dc38754fc796de828d2ede691e : SUCCESS in 5m 11s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/58dfe68629a74a30959e894e6b6ed2a3 : SUCCESS in 5m 32s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/56fb66cde10a428a8be7b4d0c316b040 : SUCCESS in 9m 30s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/97c7676179374ed195372f5c04b6e3c2 : SUCCESS in 1h 33m 55s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/bcfa6d2281c54dfe9a7a2775688a9323 : SUCCESS in 1h 43m 11s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/5f3d8bf872aa4d2facead3ce5bfec2ba : SUCCESS in 35m 20s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/a0e95956890b40c79b964e6818f90ccb : SUCCESS in 40m 12s","accounts_in_message":[],"_revision_number":1},{"id":"2d9e964dd232da79d1b50694a7cdf54586cbeba0","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-20 17:07:58.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"d44aec17b90b32faf80eac5296435336c2665b10","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-20 17:11:35.000000000","message":"Patch Set 1:\n\nBuild failed (promote pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/d95f6427021842c89cdca5cc341774fa\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b7d7414c65b54ea28368736b1af34759 : SUCCESS in 1m 02s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/0956250e701d40aca79c4da48ebf2712 : FAILURE in 43s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"774f92214390469dc6911c62d79b3f90f0c97a3f","revisions":{"774f92214390469dc6911c62d79b3f90f0c97a3f":{"kind":"REWORK","_number":1,"created":"2023-07-20 10:27:22.000000000","uploader":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"ref":"refs/changes/75/889075/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/75/889075/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/75/889075/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/75/889075/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/75/889075/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/75/889075/1"}}},"commit":{"parents":[{"commit":"c5b5bc98c1f6e780a342d03cccd8daa3a8108085","subject":"Merge \"Fix hm operating status to ONLINE in single lb call\" into stable/xena","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/c5b5bc98c1f6e780a342d03cccd8daa3a8108085"}]}],"author":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2022-07-06 16:44:21.000000000","tz":120},"committer":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2023-07-20 10:27:13.000000000","tz":120},"subject":"Drop lower-constraints.txt and its testing","message":"Drop lower-constraints.txt and its testing\n\nAs discussed in TC PTG[1] and TC resolution[2], we are\ndropping the lower-constraints.txt file and its testing.\nWe will keep lower bounds in the requirements.txt file but\nwith a note that these are not tested lower bounds and we\ntry our best to keep them updated.\n\n[1] https://etherpad.opendev.org/p/tc-zed-ptg#L326\n[2] https://governance.openstack.org/tc/resolutions/20220414-drop-lower-constraints.html#proposal\n\nChange-Id: Ib929fbe28241bd6d3c3bd352b2d900289efd6d49\n(cherry picked from commit 38cb7b6fe8e52eb29efdc8cbc08619af4ab7d4c1)\n(cherry picked from commit 5bf40c896a678cc11f4ab6aecdd8cc004fc809d9)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/774f92214390469dc6911c62d79b3f90f0c97a3f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/774f92214390469dc6911c62d79b3f90f0c97a3f"}]},"branch":"refs/heads/stable/xena"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
