)]}'
{"id":"openstack%2Foctavia~889076","triplet_id":"openstack%2Foctavia~stable%2Fwallaby~Ib929fbe28241bd6d3c3bd352b2d900289efd6d49","project":"openstack/octavia","branch":"stable/wallaby","attention_set":{},"removed_from_attention_set":{"29244":{"account":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"last_update":"2023-07-20 17:08:00.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Ib929fbe28241bd6d3c3bd352b2d900289efd6d49","subject":"Drop lower-constraints.txt and its testing","status":"MERGED","created":"2023-07-20 10:28:37.000000000","updated":"2023-07-20 17:11:30.000000000","submitted":"2023-07-20 17:08:00.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"889076","meta_rev_id":"4ac0fb8a8485430776ba45eb535a4bacf3809d83","_number":889076,"virtual_id_number":889076,"owner":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-07-20 17:08:00.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"all":[{"value":2,"date":"2023-07-20 15:09:49.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"all":[{"value":1,"date":"2023-07-20 15:09:49.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-07-20 14:14:41.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-07-20 15:09:49.000000000","updated_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"reviewer":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"state":"REVIEWER"}],"messages":[{"id":"f85a586d6570b5338ce4dc4ad4ebe4c3ea47b18e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2023-07-20 10:28:37.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"9e1abcc2114f3fa532b70c5148ac587e1fb2c527","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-20 14:14:41.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c10c3daa337e40d88ddf85564716f993\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/539ed384ccc3438dbbe4ab4fcfafb323 : SUCCESS in 2m 08s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/108929c82f1f40ca9d079e2fedfed414 : SUCCESS in 8m 37s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/63455712638c401da729f5009796c566 : SUCCESS in 5m 05s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/d61a2b19e4c74061afda2bf5973f9252 : SUCCESS in 4m 59s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/ab13ff26037d4a03a95c35390ba64deb : SUCCESS in 5m 03s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/7d0c03ab350b42c69672094e306438a7 : SUCCESS in 6m 15s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4d5d38428ef04102ac7d49e8ca6e70f1 : SUCCESS in 12m 35s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/bde04069a9554a5f86c95be13f152929 : SUCCESS in 5m 14s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/441ca181ad424d84a9f1b111d402ce04 : SUCCESS in 6m 10s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/aa54d732f7134e28b2a71ea6783bdcbe : SUCCESS in 7m 26s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/cf8d0fa43e9643cc816f64b1951f2ba9 : SUCCESS in 5m 00s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/d1edb426a8e34eab9a97dc791cac4101 : SUCCESS in 6m 40s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/73a74a97ed8a4a3489f4577b8799354d : SUCCESS in 1h 05m 05s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/9e82d70ab9864407911e8c7f4ebf539f : SUCCESS in 1h 32m 27s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/19b245857f8b479296c2a89f4b4a97d6 : SUCCESS in 34m 19s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/c3838318554942cca9355d0e9352acc5 : SUCCESS in 28m 45s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/447cf9ee92f343f696b50f550bc38a79 : SUCCESS in 30m 00s\n- octavia-v2-dsvm-tls-barbican-fips https://zuul.opendev.org/t/openstack/build/5ef135713fbe4ebab5c054ce524d6081 : FAILURE in 10m 46s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/f0deee1c3c7149fc9ba227c3a0a31ee2 : SUCCESS in 1h 21m 43s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/bb4005d4bee6426fa869764285a788e5 : FAILURE in 10m 48s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/2122e5c0fbfc4377a0be3a00fd2ace4b : FAILURE in 5m 57s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/1e329ae5d7ef46478f5ea98f7c459b9d : TIMED_OUT in 2h 35m 18s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/ea3d81e292d0497aa0b75b51ec1cc147 : SUCCESS in 2h 07m 38s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8 https://zuul.opendev.org/t/openstack/build/6bb118a7dca14b75b622686b81e891d2 : NODE_FAILURE Node request 300-0021736704 failed in 0s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"5b74d942b51b0dc1d4b97319ceb3f9fa98c58d48","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2023-07-20 15:09:49.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"2b697e3fa45b65d9d99f56e5bf5398e480069c9a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-20 15:10:47.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"b63a662e0d9a602a3380b194c348b4621a7f9588","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-20 17:08:00.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e46b0e3a16db4d6391dfa16a484ebb92\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/831c260e8526426781ebc517f0429b90 : SUCCESS in 2m 14s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/39404a3656424748b69e86e8a8fb266e : SUCCESS in 5m 27s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/334c867cfb89478d9c9835c0fb0aa5ae : SUCCESS in 5m 30s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/636aa115861b42b5b2ec82694fecb89b : SUCCESS in 4m 40s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e62ae698e62f456285b6300e990123e0 : SUCCESS in 11m 27s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/3cd8aee829c44e1d8a32bc3f93c8b32a : SUCCESS in 5m 35s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/b2b7964a9a91418199b1f06446038bc6 : SUCCESS in 5m 30s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/400109a52ac64e02b4a3fbcb16430f63 : SUCCESS in 7m 36s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/179199370c79472bb2adb5871fe8a4b4 : SUCCESS in 1h 20m 32s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/8e45bc53fb6d4ebbab0604bef5bc9e8a : SUCCESS in 1h 35m 03s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/0921ab17cd4149a0aa38d0c887778929 : SUCCESS in 33m 26s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/f203c31dcb454a5dbabede7fa51c3adf : SUCCESS in 30m 43s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/745875ee575b4ab8912cc9de13cc42d5 : SUCCESS in 34m 06s","accounts_in_message":[],"_revision_number":1},{"id":"4fa85845868a5e7c78cd986b1a3923aee7828bf5","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-20 17:08:00.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"4ac0fb8a8485430776ba45eb535a4bacf3809d83","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-20 17:11:30.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9a78c95aa68649c585cd660dc3660d2b\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5e75480fe03545e69ec7f9850bdd1014 : SUCCESS in 57s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/634f47d43ba442938968fe1ee450a8a9 : SUCCESS in 43s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"60ffac310dc7b520ec20c11f3c1fed1f1cb5af9e","revisions":{"60ffac310dc7b520ec20c11f3c1fed1f1cb5af9e":{"kind":"REWORK","_number":1,"created":"2023-07-20 10:28:37.000000000","uploader":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"ref":"refs/changes/76/889076/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/76/889076/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/76/889076/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/76/889076/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/76/889076/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/76/889076/1"}}},"commit":{"parents":[{"commit":"8af6306df63df55d47a56c2c222eb5a254b88574","subject":"Merge \"Fix hm operating status to ONLINE in single lb call\" into stable/wallaby","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/8af6306df63df55d47a56c2c222eb5a254b88574"}]}],"author":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2022-07-06 16:44:21.000000000","tz":120},"committer":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2023-07-20 10:27:49.000000000","tz":120},"subject":"Drop lower-constraints.txt and its testing","message":"Drop lower-constraints.txt and its testing\n\nAs discussed in TC PTG[1] and TC resolution[2], we are\ndropping the lower-constraints.txt file and its testing.\nWe will keep lower bounds in the requirements.txt file but\nwith a note that these are not tested lower bounds and we\ntry our best to keep them updated.\n\n[1] https://etherpad.opendev.org/p/tc-zed-ptg#L326\n[2] https://governance.openstack.org/tc/resolutions/20220414-drop-lower-constraints.html#proposal\n\nChange-Id: Ib929fbe28241bd6d3c3bd352b2d900289efd6d49\n(cherry picked from commit 38cb7b6fe8e52eb29efdc8cbc08619af4ab7d4c1)\n(cherry picked from commit 5bf40c896a678cc11f4ab6aecdd8cc004fc809d9)\n(cherry picked from commit 774f92214390469dc6911c62d79b3f90f0c97a3f)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/60ffac310dc7b520ec20c11f3c1fed1f1cb5af9e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/60ffac310dc7b520ec20c11f3c1fed1f1cb5af9e"}]},"branch":"refs/heads/stable/wallaby"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
