)]}'
{"id":"openstack%2Foctavia~889274","triplet_id":"openstack%2Foctavia~stable%2Fxena~I78cf4f2798adff5823ddf960ff0c1c6faeb7a9a4","project":"openstack/octavia","branch":"stable/xena","topic":"xena","attention_set":{},"removed_from_attention_set":{"33950":{"account":{"_account_id":33950,"name":"Liangyi Ma","email":"maliangyi@inspur.com","username":"maliangyi"},"last_update":"2023-08-09 07:41:00.000000000","reason":"Change was abandoned"}},"hashtags":[],"change_id":"I78cf4f2798adff5823ddf960ff0c1c6faeb7a9a4","subject":"Fix pep8 error","status":"ABANDONED","created":"2023-07-26 11:30:18.000000000","updated":"2023-08-09 07:41:00.000000000","total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"3fd9f4a0bca1ea43295a9e2f03139cfafffed0ba","_number":889274,"virtual_id_number":889274,"owner":{"_account_id":33950,"name":"Liangyi Ma","email":"maliangyi@inspur.com","username":"maliangyi"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":1,"date":"2023-07-26 14:17:02.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-07-26 11:42:50.000000000","updated_by":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"reviewer":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"state":"CC"},{"updated":"2023-07-26 14:17:02.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"170e3671bd072af934cfdf3035e8749999f5c90c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":33950,"name":"Liangyi Ma","email":"maliangyi@inspur.com","username":"maliangyi"},"date":"2023-07-26 11:30:18.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"34c7f26285b9cbb0a6c5887869dccd962ac5d26d","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2023-07-26 11:42:50.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"08cc9e526102615b9097fc2d30da5b327fb14590","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-26 14:17:02.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f54ef36ff075446fa870bda92638ccbd\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/d2b755d08d324f3a97f0d59a7f32b04f : SUCCESS in 9m 50s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/380c8e8df0d345da88a653705783d95c : SUCCESS in 5m 26s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/4d857f30449a4a1d923dbba98e308862 : SUCCESS in 6m 54s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/1052907142c549099204472937857f28 : SUCCESS in 4m 45s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/4d69bac9b60e47e991280ae6133c3a01 : SUCCESS in 4m 46s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/39aa65d2c0d94172b78113fd6ee7d1a9 : SUCCESS in 11m 00s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/7fffb62b8f704c3db7c1d10c1f288920 : SUCCESS in 6m 40s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/d678dc6bd6424081aa239b92083dde5d : SUCCESS in 10m 01s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/fa7a923f2e264891bec8a6a5ac383a11 : SUCCESS in 3m 58s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/bf0362b8752d4434bfb884b2681ae4ad : SUCCESS in 6m 37s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/b233c6ad6b4a41568511bf9b37784eb1 : SUCCESS in 55m 15s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/79dd3549217e47a39a6055ccc21c95ab : SUCCESS in 1h 44m 21s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/c587947f37f84613b34a1e799b24897a : SUCCESS in 36m 53s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/d51a978eb3454c24bd7ef8ddec109f95 : SUCCESS in 42m 22s\n- octavia-v2-dsvm-tls-barbican-fips https://zuul.opendev.org/t/openstack/build/ec0a73c7e53c42099bf6ee58ff513101 : FAILURE in 9m 30s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/5ae83a483926425392059fd780216291 : SUCCESS in 31m 36s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/8bf068e6496e482f9c5457a3c560bd5c : SUCCESS in 1h 34m 21s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/d3c135eddf0748ccb323e6ec9de60406 : FAILURE in 10m 57s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/1eef02845dde4f8ca1c3d40ad076af3d : FAILURE in 6m 58s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/2816d632a19d416ca248b0df7b740bdb : SUCCESS in 1h 52m 55s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/e1095df9dd6643339ab45b42fe758862 : SUCCESS in 2h 05m 02s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8 https://zuul.opendev.org/t/openstack/build/690ccc0612e644aea7dcc692c219aa3f : NODE_FAILURE Node request 300-0021782178 failed in 0s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"3fd9f4a0bca1ea43295a9e2f03139cfafffed0ba","tag":"autogenerated:gerrit:abandon","author":{"_account_id":33950,"name":"Liangyi Ma","email":"maliangyi@inspur.com","username":"maliangyi"},"date":"2023-08-09 07:41:00.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"ddc6c23b7261f200fd9ccae7f09f199f811444da","revisions":{"ddc6c23b7261f200fd9ccae7f09f199f811444da":{"kind":"REWORK","_number":1,"created":"2023-07-26 11:30:18.000000000","uploader":{"_account_id":33950,"name":"Liangyi Ma","email":"maliangyi@inspur.com","username":"maliangyi"},"ref":"refs/changes/74/889274/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/74/889274/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/74/889274/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/74/889274/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/74/889274/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/74/889274/1"}}},"commit":{"parents":[{"commit":"774f92214390469dc6911c62d79b3f90f0c97a3f","subject":"Drop lower-constraints.txt and its testing","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/774f92214390469dc6911c62d79b3f90f0c97a3f"}]}],"author":{"name":"maliangyi","email":"maliangyi@inspur.com","date":"2023-07-26 08:51:58.000000000","tz":480},"committer":{"name":"maliangyi","email":"maliangyi@inspur.com","date":"2023-07-26 11:27:50.000000000","tz":480},"subject":"Fix pep8 error","message":"Fix pep8 error\n\nnew flake8 issue\n - E275 missing whitespace after keyword\n\nChange-Id: I78cf4f2798adff5823ddf960ff0c1c6faeb7a9a4\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/ddc6c23b7261f200fd9ccae7f09f199f811444da"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/ddc6c23b7261f200fd9ccae7f09f199f811444da"}]},"branch":"refs/heads/stable/xena"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
