)]}'
{"id":"openstack%2Foctavia~890152","triplet_id":"openstack%2Foctavia~master~I88371364b88bec5f1fd42d6ade8b316be3130f3f","project":"openstack/octavia","branch":"master","attention_set":{},"removed_from_attention_set":{"11628":{"account":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"last_update":"2023-08-17 14:45:56.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I88371364b88bec5f1fd42d6ade8b316be3130f3f","subject":"Remove blockdiag from Octavia","status":"MERGED","created":"2023-07-31 16:16:33.000000000","updated":"2023-08-17 14:47:24.000000000","submitted":"2023-08-17 14:45:56.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"890152","meta_rev_id":"d7915f2bd3cd17abc45b172a54f4156f215fa36d","_number":890152,"virtual_id_number":890152,"owner":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},{"value":0,"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-08-17 14:45:55.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"recommended":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"all":[{"value":1,"date":"2023-08-07 14:52:03.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},{"value":2,"date":"2023-08-17 12:44:22.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2023-08-01 15:19:02.000000000","_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"all":[{"value":0,"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},{"value":1,"date":"2023-08-17 12:44:22.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},{"value":0,"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-07-31 19:09:08.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-08-01 15:19:02.000000000","updated_by":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"reviewer":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"state":"REVIEWER"},{"updated":"2023-08-07 14:52:03.000000000","updated_by":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"reviewer":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"state":"REVIEWER"},{"updated":"2023-08-17 12:44:22.000000000","updated_by":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"reviewer":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"state":"REVIEWER"}],"messages":[{"id":"6d20b07058d721f1d377576807d09185443f73a2","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2023-07-31 16:16:33.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"004fe8eb52c8cef774a09add51be44dba208a57d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-31 19:09:08.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3ef0149f049b4090bb73f7a23862a414\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/405bea1900ce4a1ab08bcdc500b425bb : SUCCESS in 4m 43s\n- requirements-check https://zuul.opendev.org/t/openstack/build/baed916e36cb4592969695ecb2b0a28a : SUCCESS in 2m 10s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/5f5f5827f560416b91f1fcc97fdcc931 : SUCCESS in 10m 19s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/62905015d1e34626a9deeed57bdd065a : SUCCESS in 5m 19s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/526a080d5a254468a762c5ca433a3ec8 : SUCCESS in 4m 59s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/0ba0438074cf4ac8b7b49821ffb279bf : SUCCESS in 4m 25s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/0992b24d840b49e585195af23e928ce5 : SUCCESS in 5m 04s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/93dafbe1db1c4bdb8369dd6b3e608e3f : SUCCESS in 10m 24s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/3190b4f6bccb4b169b34ae8c66ca897a : SUCCESS in 7m 54s\n- octavia-tox-py39-tips https://zuul.opendev.org/t/openstack/build/d75dadec6b2b40d3bccded4adabeb7ea : SUCCESS in 5m 14s\n- octavia-tox-functional-py39-tips https://zuul.opendev.org/t/openstack/build/ada9e5e3deea4c4083400e775046a9cb : SUCCESS in 7m 09s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/ad37add870b2454392b4a37a3960c9ab : SUCCESS in 4m 24s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/f41b3969cc0d4b90a928be066a742fd8 : SUCCESS in 7m 37s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/d8d124ee0524467ead78b345cc0e1938 : SUCCESS in 7m 30s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/9ab57ee5d18f4a3eaadf5bed766e6036 : SUCCESS in 11m 42s (non-voting)\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/3845a27e275d4757aa1413b7ee59ff19 : SUCCESS in 1h 23m 12s\n- octavia-v2-dsvm-scenario-traffic-ops https://zuul.opendev.org/t/openstack/build/2cee603d810b40549cb80c9a56db7822 : SUCCESS in 2h 00m 36s\n- octavia-v2-dsvm-scenario-non-traffic-ops https://zuul.opendev.org/t/openstack/build/0c42b4e743e14b9abdc4e8fae9520b0c : SUCCESS in 1h 20m 57s\n- octavia-v2-dsvm-scenario-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/9c5a1cd95e634b4e853debe040a97a08 : SUCCESS in 2h 08m 57s\n- octavia-v2-dsvm-scenario-non-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/3a6ff4c418f6488394bdf4307bc6052f : SUCCESS in 1h 34m 58s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/ce811ff86fe74f45a952ffb70d54e878 : SUCCESS in 38m 32s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/685b3035ee1440f28b465383d272e571 : SUCCESS in 48m 15s\n- octavia-v2-dsvm-tls-barbican-fips https://zuul.opendev.org/t/openstack/build/9dd4c5bc1c4f4e12972f3ea35f82bf40 : FAILURE in 40m 30s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/cf8cb1253984469eb5e139ef3e63fbff : SUCCESS in 33m 06s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/ec6f817dea4543d9897dab21124ccb46 : SUCCESS in 2h 23m 44s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/d0ce56d9e30d435d910a719addb441a3 : SUCCESS in 2h 43m 58s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/eedfdb3d205143e88673e3652f99074b : SUCCESS in 2h 21m 50s (non-voting)\n- octavia-v2-dsvm-scenario-centos-9-stream https://zuul.opendev.org/t/openstack/build/8dea11e2d9244afda2e8ac93cb18d743 : FAILURE in 33m 08s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"f3dddaa4f3fe3c212ad59383366d8961f79e4257","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2023-08-01 15:19:02.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"5e2fbbe0a9f814a525d702fe6cf166839662b40e","author":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"date":"2023-08-07 14:52:03.000000000","message":"Patch Set 1: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"75466fc55fb7ed02d9c2115dbe643a8461a6ee39","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2023-08-17 12:44:22.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"5a24cc74b57a75ddf86a9bbdf1314d4d2d3008d3","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-17 12:45:01.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"aa2a701e6161c5eb344f1dbac24b58fdc2f0cda8","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-17 14:45:55.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4b1ccb9565114cd1b599886b97783408\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/87c96eca737343fab096097d4bb6394e : SUCCESS in 6m 02s\n- requirements-check https://zuul.opendev.org/t/openstack/build/7ac73ac1a02247a186a31c7809636b62 : SUCCESS in 2m 13s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bf81af6064c6489fab5f4145a917ebdc : SUCCESS in 4m 50s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/931771e6578a4ccbb64850dcbb324568 : SUCCESS in 5m 53s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/3614230f4c0445e788a6a24c2ad7c494 : SUCCESS in 5m 11s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/999bd55ffe0e4f27b87463eb2d70f693 : SUCCESS in 4m 54s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/58c65a0bb4e94e6bb17bb4700e51c0ea : SUCCESS in 13m 43s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/ae5611f9e85d47c5a11759c6471c1be3 : SUCCESS in 5m 01s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/09828713a3ef4994bd3b1b7d542d4579 : SUCCESS in 4m 46s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/6524933a56ee4ca59a962f4672daae46 : SUCCESS in 6m 52s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/cf0c88f2f4894efc9ea5455f7ac648c3 : SUCCESS in 1h 30m 20s\n- octavia-v2-dsvm-scenario-traffic-ops https://zuul.opendev.org/t/openstack/build/6c2e659b182b48d0aa3c82cc355f85ef : SUCCESS in 2h 00m 21s\n- octavia-v2-dsvm-scenario-non-traffic-ops https://zuul.opendev.org/t/openstack/build/7f5d98990164479a83954129f70972f5 : SUCCESS in 1h 13m 14s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/3db19250bc02452da76d87977727297a : SUCCESS in 36m 40s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/930867bfee2f45f997b6e83935742aa6 : SUCCESS in 39m 56s","accounts_in_message":[],"_revision_number":1},{"id":"329fe7553b1e952be25ef01a39e134a343dff169","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-17 14:45:56.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"d7915f2bd3cd17abc45b172a54f4156f215fa36d","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-17 14:47:24.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4566770aa9b5437ca2e02c60a060a4d6\n\n- promote-api-ref https://zuul.opendev.org/t/openstack/build/0be08891b7ac47d2ad55bc8b1a28885e : SUCCESS in 53s\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/43424323adfe48f49ba6b21b446894a5 : SUCCESS in 1m 02s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/c546f8ae0a404b4b8b5f76540edaa131 : SUCCESS in 44s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"ff80e454ab3b0b52e03584af0bd985a2860ef02a","revisions":{"ff80e454ab3b0b52e03584af0bd985a2860ef02a":{"kind":"REWORK","_number":1,"created":"2023-07-31 16:16:33.000000000","uploader":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"ref":"refs/changes/52/890152/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/52/890152/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/52/890152/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/52/890152/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/52/890152/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/52/890152/1"}}},"commit":{"parents":[{"commit":"12d8e0de5de9ca754859542bcffd2311f5145d61","subject":"Merge \"Update doc for the OVN provider\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/12d8e0de5de9ca754859542bcffd2311f5145d61"}]}],"author":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2023-07-31 16:15:26.000000000","tz":0},"committer":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2023-07-31 16:16:32.000000000","tz":0},"subject":"Remove blockdiag from Octavia","message":"Remove blockdiag from Octavia\n\nThe python blockdiag module is unmaintained and will likely be removed from distros.[1]\nThis patch removes mention of blockdiag from Octavia as it is unused.\n\n[1] https://github.com/blockdiag/blockdiag/pull/171\n\nChange-Id: I88371364b88bec5f1fd42d6ade8b316be3130f3f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/ff80e454ab3b0b52e03584af0bd985a2860ef02a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/ff80e454ab3b0b52e03584af0bd985a2860ef02a"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
