)]}'
{"id":"openstack%2Foctavia~890814","triplet_id":"openstack%2Foctavia~master~Ib65e459bcd10a5ab877c0cf6f234d634d25d1e55","project":"openstack/octavia","branch":"master","topic":"add_noop_cert_manager","attention_set":{},"removed_from_attention_set":{"31664":{"account":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"last_update":"2024-01-10 19:16:14.000000000","reason":"Change was submitted"},"29244":{"account":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"last_update":"2024-01-10 19:16:14.000000000","reason":"Change was submitted"},"11628":{"account":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"last_update":"2024-01-10 17:36:21.000000000","reason":"\u003cGERRIT_ACCOUNT_11628\u003e replied on the change","reason_account":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}},"34429":{"account":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"last_update":"2023-11-07 08:44:30.000000000","reason":"\u003cGERRIT_ACCOUNT_34429\u003e replied on the change","reason_account":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}}},"hashtags":[],"change_id":"Ib65e459bcd10a5ab877c0cf6f234d634d25d1e55","subject":"Provide Amphora stats for Octavia no-op drivers","status":"MERGED","created":"2023-08-08 16:53:41.000000000","updated":"2024-01-10 19:17:28.000000000","submitted":"2024-01-10 19:16:14.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":16,"unresolved_comment_count":1,"has_review_started":true,"submission_id":"890814-add_noop_cert_manager","meta_rev_id":"cc7d6b1a466026ffc28cea27922604d1ffa9cc87","_number":890814,"virtual_id_number":890814,"owner":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-01-10 19:16:14.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":0,"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"all":[{"value":0,"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2024-01-10 17:36:21.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":2,"date":"2023-11-07 08:44:30.000000000","_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"all":[{"value":0,"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2024-01-10 17:36:21.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":0,"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":0,"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-08-08 19:58:04.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-08-09 13:00:24.000000000","updated_by":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"reviewer":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"state":"REVIEWER"},{"updated":"2023-08-09 13:00:24.000000000","updated_by":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"reviewer":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"state":"REVIEWER"},{"updated":"2023-08-09 13:00:24.000000000","updated_by":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"reviewer":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"state":"REVIEWER"}],"messages":[{"id":"59c62e255e65eb69cd9fe88a5e6bcc83aadaa556","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"date":"2023-08-08 16:53:41.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"1a11b9c834ee531e76cbc2b3d1b1c8c035ce783b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-08 19:58:04.000000000","message":"Patch Set 1: Verified-1\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/01dcb342074f4929bad32ec20c3ebb02\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/ac7a33c365194841be2246c54b98da09 : SUCCESS in 9m 55s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7907ec77340544e9bcd1e367af7e42a3 : FAILURE in 4m 32s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/d79cc4ace6da4f1aafe32db0b6792ea1 : SUCCESS in 7m 57s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/afce8af9ecde4b228c231c01a9602f43 : SUCCESS in 5m 14s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/1efaf7bcda1b45ba95802aa36a515d72 : SUCCESS in 4m 58s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/309f153413b04437ac3a1c9d5a97bd51 : SUCCESS in 13m 16s\n- octavia-tox-py39-tips https://zuul.opendev.org/t/openstack/build/f7235f90e7ad4877a980d03ea80c088e : SUCCESS in 6m 13s\n- octavia-tox-functional-py39-tips https://zuul.opendev.org/t/openstack/build/1bb8e377120444cf82581da188825427 : SUCCESS in 9m 13s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/8786134e88284697843682b10ca8b35b : SUCCESS in 4m 08s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/6aa15d3b69124740b3876659181c2041 : SUCCESS in 9m 42s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/a16765254ab446e2af7ebdf143fb9efc : SUCCESS in 8m 42s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/880698feb0284485a64c3b723c69af59 : SUCCESS in 8m 49s (non-voting)\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/c0d48985c48d4d398e55e3f3d28af826 : SUCCESS in 1h 22m 17s\n- octavia-v2-dsvm-scenario-traffic-ops https://zuul.opendev.org/t/openstack/build/1074ac46ffba473c8ffde25ae691769f : SUCCESS in 1h 24m 31s\n- octavia-v2-dsvm-scenario-non-traffic-ops https://zuul.opendev.org/t/openstack/build/8dbbafef4216478a94705b1e8450dc2b : SUCCESS in 1h 13m 30s\n- octavia-v2-dsvm-scenario-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/76c237c094914a948c234da333eba399 : SUCCESS in 1h 35m 36s\n- octavia-v2-dsvm-scenario-non-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/8fcea420f20d498a9da335203d89ae49 : SUCCESS in 1h 02m 29s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/ace66a9592dc410b865ad14128bd76c9 : SUCCESS in 37m 40s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/5aa97bc83533474e8a8df7d65f0ba6c6 : SUCCESS in 40m 06s\n- octavia-v2-dsvm-tls-barbican-fips https://zuul.opendev.org/t/openstack/build/aaa09e08f61842d88c04e3f34a0d0105 : SUCCESS in 1h 07m 12s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/252f49e1e5a04f228a49b9fde7b1d9cb : SUCCESS in 36m 45s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/bb9687bf37e84ace97e460602bf493a9 : TIMED_OUT in 2h 32m 51s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/4fe73818764e45febb908ac6222bbdd4 : TIMED_OUT in 3h 02m 55s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/0339be824d45430d96cf789b57adcb5c : TIMED_OUT in 2h 33m 04s (non-voting)\n- octavia-v2-dsvm-scenario-centos-9-stream https://zuul.opendev.org/t/openstack/build/8a8783dae1154a2691a0d6719febc831 : SUCCESS in 1h 54m 29s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"13d80f8a37658d278aeeb29e3109562b01cd5ab1","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"date":"2023-08-09 12:54:48.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"6b744a6728b207cf678df4d9644f75550d0ddc9d","author":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"date":"2023-08-09 12:58:16.000000000","message":"Patch Set 2:\n\n(1 comment)\n\nThis change is ready for review.","accounts_in_message":[],"_revision_number":2},{"id":"d1052cbf55acd5e22c6ae58e34b087f85e0fd662","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"date":"2023-08-09 12:59:55.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"108412306a21a543cb41983196a10977eea7d4d9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-09 16:06:47.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2174bcac7fe14bf5a8b6fc8202c0678a\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/3dfaf66f84fe4eeca516fb5077b93a64 : SUCCESS in 9m 41s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d082f81bd56549fdbcbf33b41ab94e5b : SUCCESS in 5m 33s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/b206af2acc1b4bf4b91ba4a8266144ac : SUCCESS in 4m 18s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/ed2d3e502e5e43948fe630dfc9ea8a02 : SUCCESS in 4m 55s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/c36466bf81374faf9ed9707903cb8887 : SUCCESS in 5m 12s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/88dcdba7308b4cceae75afb257781012 : SUCCESS in 12m 46s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/564cbd58eaad4858a79abe22841efb8e : SUCCESS in 4m 30s\n- octavia-tox-py39-tips https://zuul.opendev.org/t/openstack/build/db417d4fabbc4c499660e5a08706dbe2 : SUCCESS in 4m 31s\n- octavia-tox-functional-py39-tips https://zuul.opendev.org/t/openstack/build/8065c2519048498c8160393a142bf4ee : SUCCESS in 7m 26s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/4c2d7a42c05549358b5272ae4f3f6c81 : SUCCESS in 4m 40s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/2fab5ea4a5ce4c32849a48f2e036e1c5 : SUCCESS in 7m 30s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/0acabac042ea48ddb5ad75caf093e04c : SUCCESS in 8m 08s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/bf9aa00288ac4ac59563d4016d10c5b0 : SUCCESS in 6m 10s (non-voting)\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/61854f6376f64734a135c0f64989d84e : SUCCESS in 1h 17m 10s\n- octavia-v2-dsvm-scenario-traffic-ops https://zuul.opendev.org/t/openstack/build/08d292be4f0c43d39d43a607244dab5d : SUCCESS in 1h 40m 54s\n- octavia-v2-dsvm-scenario-non-traffic-ops https://zuul.opendev.org/t/openstack/build/dd4d164794b54cca9fd8629113ca2031 : SUCCESS in 1h 24m 42s\n- octavia-v2-dsvm-scenario-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/ac912855ff924b8e890dc67d7c12a74b : SUCCESS in 1h 58m 48s\n- octavia-v2-dsvm-scenario-non-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/26eac9853a4646f9bf305c2be3b9b0ab : SUCCESS in 1h 36m 11s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/63c6746916924664a35e08408f37f217 : SUCCESS in 35m 54s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/2e12a551bd8e434a94d0438b35ed6747 : SUCCESS in 48m 25s\n- octavia-v2-dsvm-tls-barbican-fips https://zuul.opendev.org/t/openstack/build/292fa3ecce7b4763adb815ebf58ed0d9 : SUCCESS in 1h 14m 39s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/0034f53d76d34c8596eaa8ba714d8ce9 : SUCCESS in 36m 45s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/5f62517b19bb489c9c59469feac606e1 : TIMED_OUT in 2h 32m 52s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/911c1a511a8743e0a4c19f0e5e0985af : TIMED_OUT in 3h 03m 29s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/7fe866724b7c49fc8bb3e87a435459aa : TIMED_OUT in 2h 32m 33s (non-voting)\n- octavia-v2-dsvm-scenario-centos-9-stream https://zuul.opendev.org/t/openstack/build/3fc7d6eb15624b41b58b1024aaa90671 : TIMED_OUT in 2h 34m 46s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"eb9b94da2e92cf39f5bcbef8d495432971e00e68","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2023-08-21 08:37:07.000000000","message":"Patch Set 3: Code-Review-1\n\n(2 comments)","accounts_in_message":[],"_revision_number":3},{"id":"6eb6e722c5817b23f622caec3ad9e4ab1dcc26c3","author":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"date":"2023-08-21 10:41:19.000000000","message":"Patch Set 3:\n\n(2 comments)","accounts_in_message":[],"_revision_number":3},{"id":"e5a1ecf35f59e8a7ba480ff1f1b67f6d024e6668","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"date":"2023-08-21 10:41:27.000000000","message":"Uploaded patch set 4.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":4},{"id":"d6f97cd6e18b8e9b2972bf6fa2f47012dd8bbd85","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2023-08-21 11:54:16.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"28d1a945f6500e45a41cf742136aec44a23a2ad6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-21 13:46:13.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/57699ea94ce240a6b0e6a5e885a6fbc5\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/ed6312b36475490abc85943da4778c26 : SUCCESS in 9m 21s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1cc7f25c5eb7428db621db6fb9ae8115 : SUCCESS in 5m 52s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/710677037c6f40e5b16065d8a2e3e71c : SUCCESS in 4m 01s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/e4af40ee2dc642749ecf7481da48f7bc : SUCCESS in 4m 29s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/20aa6cef6a6c4b68adff482aa621b230 : SUCCESS in 4m 56s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/7791e8a33a3b43cb8dc3f645f2eaa9fe : SUCCESS in 4m 42s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/26689ad583e64875938ea5bf2c8ba6f2 : SUCCESS in 8m 30s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/d35fbe0594ff4a9b88d9c1c25c2990a7 : SUCCESS in 5m 18s\n- octavia-tox-py39-tips https://zuul.opendev.org/t/openstack/build/359ff7697723458fbc5bde968f3b7450 : SUCCESS in 4m 47s\n- octavia-tox-functional-py39-tips https://zuul.opendev.org/t/openstack/build/ab9f67e85d794c0c8852f9c0418bf6f8 : SUCCESS in 7m 56s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/d70c0c4907ac401498582ffea27a4c76 : SUCCESS in 3m 45s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/353e73d1895d4066b611e10b1b36be14 : SUCCESS in 7m 25s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/e1d8def40b0d481a85d1f15eb7224805 : SUCCESS in 7m 47s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/d8a040d7d4ef42e6917fa4098090e072 : SUCCESS in 6m 20s (non-voting)\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/e185b2cd38964d189def77017c630c0e : SUCCESS in 1h 26m 50s\n- octavia-v2-dsvm-scenario-traffic-ops https://zuul.opendev.org/t/openstack/build/f20296b4987044bbb9bc72eaa2ab6df7 : SUCCESS in 1h 42m 18s\n- octavia-v2-dsvm-scenario-non-traffic-ops https://zuul.opendev.org/t/openstack/build/d1a1c4677d3946169a8cbcca85cb7d24 : SUCCESS in 1h 08m 39s\n- octavia-v2-dsvm-scenario-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/56855970c8984c79b4f6255ca625c1a5 : SUCCESS in 1h 45m 43s\n- octavia-v2-dsvm-scenario-non-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/5c5bef46176a4ac88b51d7d15b24bc42 : SUCCESS in 1h 18m 26s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/42898eb65a334d5b908b6caa88c8d6e6 : SUCCESS in 36m 16s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/a5f64e45239c48588f50a956e6a28feb : SUCCESS in 51m 36s\n- octavia-v2-dsvm-tls-barbican-fips https://zuul.opendev.org/t/openstack/build/f97ad1f3026f4e32a08ca0e7f27df613 : FAILURE in 25m 16s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/51174c3530cc43e9b59d826bb1e16ed0 : SUCCESS in 43m 05s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/cf82d49920ba44cf9459d69333984e34 : TIMED_OUT in 2h 33m 10s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/c75d7c7148074116b3404a8d6c466a22 : TIMED_OUT in 3h 03m 00s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/bea74e861432440a95953b9aa6199d38 : TIMED_OUT in 2h 32m 50s (non-voting)\n- octavia-v2-dsvm-scenario-centos-9-stream https://zuul.opendev.org/t/openstack/build/051db713696741a984f16411acb6e0df : FAILURE in 15m 45s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"9954863b28d16f8fd0e57189a71cd7d801831444","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"date":"2023-08-22 10:41:00.000000000","message":"Uploaded patch set 5.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":5},{"id":"228462c65cf2b1d5478c5b6865fec712f6bbaa2f","author":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"date":"2023-08-22 10:43:57.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"7c17e1d43af7b0fcde3ae0232bd941e07f2714b8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-22 13:07:59.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/fbd75930d30d4f629ddb6017ed47a71b\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/f97de7ca8ce54527bff76c275c7f96f5 : SUCCESS in 11m 36s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/db0d3d580be9453b8f102f33b1bd6152 : SUCCESS in 5m 42s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/1d98da433b2243eba743eaf8034ed86d : SUCCESS in 5m 21s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/aaad4ea2bfd547f4bd68af82aed52fca : SUCCESS in 6m 02s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/c1fcf8f9b0d54b2eb62c26a0f27a33cc : SUCCESS in 4m 57s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/443fd393f2684e598681d3d46c9ee2c9 : SUCCESS in 5m 13s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8b6595c31b004bfc960fb17f86a17290 : SUCCESS in 11m 01s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/d7c3fd6af9964bdca2b7463aef1f117e : SUCCESS in 5m 25s\n- octavia-tox-py39-tips https://zuul.opendev.org/t/openstack/build/468a1327ae5c4a53a7061cb41f37abaa : SUCCESS in 5m 22s\n- octavia-tox-functional-py39-tips https://zuul.opendev.org/t/openstack/build/4b8fd91417354d1fb6390183dbbf96f4 : SUCCESS in 8m 55s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/50945516173d476b81285661dbe99917 : SUCCESS in 5m 30s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/f70a56b6ca7442608e5dc635c963ab12 : SUCCESS in 7m 38s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/c4e1eb736467498685a2b5e095a5dc58 : SUCCESS in 7m 44s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/0aaf88f1170a471a88cebfaf990554f3 : SUCCESS in 8m 03s (non-voting)\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/ac6e6f8c9da04453adf853a1fe1e7eb8 : SUCCESS in 1h 40m 39s\n- octavia-v2-dsvm-scenario-traffic-ops https://zuul.opendev.org/t/openstack/build/4fad149fe27e495cb47ee4ac8e398f2a : SUCCESS in 2h 11m 37s\n- octavia-v2-dsvm-scenario-non-traffic-ops https://zuul.opendev.org/t/openstack/build/c062a7fc5ab54788b174ded2e39dacc2 : SUCCESS in 1h 30m 35s\n- octavia-v2-dsvm-scenario-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/828261c740254ef38f9b3d09f96c8d06 : SUCCESS in 2h 13m 08s\n- octavia-v2-dsvm-scenario-non-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/ff710d89e6604c6a9b8633b12ebb704a : SUCCESS in 1h 23m 42s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/9977488c6f1342b29ca42cd118a3ab3d : SUCCESS in 36m 02s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/24748e43b09140848e7aab5012b22603 : SUCCESS in 39m 07s\n- octavia-v2-dsvm-tls-barbican-fips https://zuul.opendev.org/t/openstack/build/9ae167cc07014234a251e38271f966a6 : FAILURE in 20m 54s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/a9e392f502ed404da026a8152cd976b9 : SUCCESS in 34m 25s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/a79ee6abc23b496795b20e1ca2c93bd3 : SUCCESS in 2h 23m 45s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/c7e64a4b365548628404190b60fbd61b : SUCCESS in 2h 23m 54s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/f068cfa9f7d24e37a869bbf7c68e64fe : SUCCESS in 1h 57m 39s (non-voting)\n- octavia-v2-dsvm-scenario-centos-9-stream https://zuul.opendev.org/t/openstack/build/eeae3fa0cdff41eaaf10dab98d2742e8 : FAILURE in 18m 48s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"08ae738388af1454defb83389a98809c6d225e6b","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2023-08-23 20:57:33.000000000","message":"Patch Set 5: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"95810d389af731abded05ab8115466e08603a61c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"date":"2023-09-01 13:28:41.000000000","message":"Uploaded patch set 6.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":6},{"id":"01ffe53de3f4d2cfbc99f225609aa6f475e14753","author":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"date":"2023-09-01 13:30:08.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"c117bb9f6b85898eb85958132938c4a5bf7060be","author":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"date":"2023-09-01 15:22:51.000000000","message":"Patch Set 6: Workflow-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"f0484a3fa2c0ac38c3192c2fccbdf1a97adbd668","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-01 16:23:10.000000000","message":"Patch Set 6: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/5709dbb6388d462b86ca83fa2277d5df\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/365acf23af1949d49c0e488732009591 : SUCCESS in 10m 36s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/443314d37f444def8b731c293ae9292f : SUCCESS in 5m 50s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/7a3a4797dae94aa191accae64be9ad62 : SUCCESS in 3m 59s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/d12584489ce74149bab3ed449c309a20 : SUCCESS in 5m 56s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/9586c1af34aa439bb337923dfec51f81 : SUCCESS in 5m 03s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/2359ec6abf1f4dbe9087d607c7099a29 : SUCCESS in 5m 25s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bbedaa7c79c74ea0bbfa8d43a2708514 : SUCCESS in 12m 42s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/1837ca7e171c46609b22f19313ef7df6 : SUCCESS in 5m 08s\n- octavia-tox-py39-tips https://zuul.opendev.org/t/openstack/build/d78d729418ff4f83aa249a0e5ee32a43 : SUCCESS in 5m 19s\n- octavia-tox-functional-py39-tips https://zuul.opendev.org/t/openstack/build/e405ec64e716418fa575d8b7d225be76 : SUCCESS in 6m 55s\n- octavia-tox-functional-py39-sqlalchemy-tips https://zuul.opendev.org/t/openstack/build/5285201c8f6945bfa67002f90fa0b084 : SUCCESS in 8m 47s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/08342c7bb7be42199cfb908332bf5475 : SUCCESS in 4m 12s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/22fb58c308c745eda2a573599c61a9f2 : SUCCESS in 8m 01s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/1533a04b4c3b4bcd9d46c34a18788dc9 : SUCCESS in 8m 22s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/09e1e6076e324914b6f3b8fb23743b25 : SUCCESS in 8m 17s (non-voting)\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/60d178b43a704b029bd4123f800606c5 : SUCCESS in 1h 08m 34s\n- octavia-v2-dsvm-scenario-traffic-ops https://zuul.opendev.org/t/openstack/build/f3ba6b190e294ff1afc8b7bf66d57c3d : SUCCESS in 2h 03m 50s\n- octavia-v2-dsvm-scenario-non-traffic-ops https://zuul.opendev.org/t/openstack/build/d42070726b8b4924825c932313c07f35 : SUCCESS in 1h 29m 23s\n- octavia-v2-dsvm-scenario-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/47bbad0a39a74f7bbe2b95083eef83ff : SUCCESS in 2h 04m 04s\n- octavia-v2-dsvm-scenario-non-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/c6f612f0f0d349efb1588a67c0f9b2f7 : SUCCESS in 1h 32m 51s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/a5a6683c6ae64d2f81c4ef06f6e31612 : SUCCESS in 51m 13s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/6c8faf1992754ebd8bad78b0afcdf03d : FAILURE in 41m 42s\n- octavia-v2-dsvm-tls-barbican-fips https://zuul.opendev.org/t/openstack/build/06d49772b225491ba3b66db813aae29f : SUCCESS in 1h 13m 25s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/d6911ddde224461582b58c3b4d069050 : SUCCESS in 36m 44s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/7521b5a7b0b64bb486fb41b1bb2b3a69 : TIMED_OUT in 2h 32m 48s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/e3b0ca6d649f4270a967fa6e79f7eaab : SUCCESS in 2h 51m 51s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/9bfb3168b1b74be3ae7170e6c2a47ee1 : TIMED_OUT in 2h 33m 11s (non-voting)\n- octavia-v2-dsvm-scenario-centos-9-stream https://zuul.opendev.org/t/openstack/build/f10364bb8bd24ad98b8d981408b03069 : SUCCESS in 2h 32m 29s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"42221ff655d913ec798d62c5c447bf6a96ba3f63","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"date":"2023-10-30 13:41:59.000000000","message":"Uploaded patch set 7: Patch Set 6 was rebased.\n\nOutdated Votes:\n* Verified-1\n* Workflow-1\n","accounts_in_message":[],"_revision_number":7},{"id":"3322085db0fc6bf16d6dbbd04f382d50c2efe5b7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-30 17:28:34.000000000","message":"Patch Set 7: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ccc9997b2b404d1a87cd98a07f69faab\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/18f8e438c3574a87bbd544cd33d9a2c3 : SUCCESS in 9m 12s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b81f70caf77e404ba86afee80c9d49ca : SUCCESS in 5m 55s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/abe1ddcd18a0457192e548accf2ebc08 : SUCCESS in 5m 41s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/3165b100c995403d8782f3aa04870cc9 : SUCCESS in 4m 54s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6b2344bd303443ee90b07fc80af43a20 : SUCCESS in 10m 41s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/ff84b52dcb8c4c1783067604b6de89f5 : SUCCESS in 5m 04s\n- octavia-tox-py39-tips https://zuul.opendev.org/t/openstack/build/03cbb3a2f6064f16a739671ff35d1cca : SUCCESS in 4m 40s\n- octavia-tox-functional-py39-tips https://zuul.opendev.org/t/openstack/build/e423b0017a69411fb16dc16180210b54 : SUCCESS in 8m 55s\n- octavia-tox-functional-py39-sqlalchemy-tips https://zuul.opendev.org/t/openstack/build/ead181941ee64407bcbaf716b5d02f10 : SUCCESS in 9m 37s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/a3d6961e4fed4424a664c9aa1d45f82f : SUCCESS in 4m 17s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/bbde05e80d704798ab3326fa22d9507a : SUCCESS in 7m 26s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/e6cc62a85d0343669d125a9f7e603234 : SUCCESS in 8m 58s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/58c011f3194f43ffb385145cb51178f1 : SUCCESS in 9m 12s (non-voting)\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/249f57177e0746b992c43a2d57b1c91d : SUCCESS in 1h 14m 05s\n- octavia-v2-dsvm-scenario-traffic-ops https://zuul.opendev.org/t/openstack/build/4e4a548df6a546ef9fc0a7d0d83b3bbb : SUCCESS in 1h 35m 47s\n- octavia-v2-dsvm-scenario-non-traffic-ops https://zuul.opendev.org/t/openstack/build/79cec5fe4f074f54ad13b6751ab5236a : SUCCESS in 59m 06s\n- octavia-v2-dsvm-scenario-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/f52ffd339b0f4c599b1cc6b25a5b6c07 : SUCCESS in 1h 41m 37s\n- octavia-v2-dsvm-scenario-non-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/741b2abdb05f402ab91bbb7cef18ed04 : SUCCESS in 1h 16m 44s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/4364d9eee2a342b09be3db6522bc18c3 : SUCCESS in 38m 33s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/f28bfcd0a19a477797ef84143ac04344 : SUCCESS in 38m 02s\n- octavia-grenade-skip-level https://zuul.opendev.org/t/openstack/build/4c67da4baf124daba1a2e44678ae90e7 : SUCCESS in 49m 37s (non-voting)\n- octavia-v2-dsvm-tls-barbican-fips https://zuul.opendev.org/t/openstack/build/eccd2671349d4ff6a0151e815394e0f1 : SUCCESS in 1h 03m 11s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/d4e4db27e43243f39dd35684ddbd8e35 : SUCCESS in 36m 58s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/22fd133bdb6a4e2e99655d7f5c638dc2 : TIMED_OUT in 2h 32m 57s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/612aba10ebaf47e68811446c061501b9 : SUCCESS in 2h 48m 08s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/6deeb83969284adcb37b4176d36392f8 : TIMED_OUT in 2h 33m 23s (non-voting)\n- octavia-v2-dsvm-scenario-centos-9-stream https://zuul.opendev.org/t/openstack/build/1b87c34ca9ae4b0597d7406cb11a5483 : TIMED_OUT in 2h 33m 33s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"3cf3eded749ec37b40969437c8fc096119c01ea6","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2023-10-31 08:05:36.000000000","message":"Patch Set 7: Code-Review-1\n\n(2 comments)","accounts_in_message":[],"_revision_number":7},{"id":"e7abf3674bf7d51e5341cb988b537eeae6b73679","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2023-10-31 08:06:34.000000000","message":"Patch Set 7:\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"9366f28af65c11e79b4e0824e9c94d783d20e1e8","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"date":"2023-10-31 09:39:34.000000000","message":"Uploaded patch set 8.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":8},{"id":"07f7632a9541a49c0c67eea2fc6a70053529f62f","author":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"date":"2023-10-31 09:39:56.000000000","message":"Patch Set 8:\n\n(2 comments)","accounts_in_message":[],"_revision_number":8},{"id":"4618b791b824cc540cd08a482b279582dff24664","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-31 12:23:50.000000000","message":"Patch Set 8: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a60b480a7f6f4a81bffc79789d8b9d52\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/71bd79f12f82417c85cb7b754618d7c1 : SUCCESS in 10m 12s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2a0027e68b714816be2bfb199553e3ef : SUCCESS in 5m 28s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/74d32d7c1e53433f849b45ddc85f1fcc : SUCCESS in 5m 30s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/d11072fc201245dda22182d5e33bcc50 : SUCCESS in 5m 32s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/89cf9e66e7d9473ca43e263b80d458b8 : SUCCESS in 11m 49s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/3761d547748f4d4294478a5f8bd85769 : SUCCESS in 5m 46s\n- octavia-tox-py39-tips https://zuul.opendev.org/t/openstack/build/3495b81a81d24ab4b5bd6cc75e4a508a : SUCCESS in 5m 52s\n- octavia-tox-functional-py39-tips https://zuul.opendev.org/t/openstack/build/cd39206eae034c998ce0e43f4f684f95 : SUCCESS in 8m 40s\n- octavia-tox-functional-py39-sqlalchemy-tips https://zuul.opendev.org/t/openstack/build/71b1d0dd86ad46d6a987589df74ddc25 : SUCCESS in 12m 20s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/3992f930574746588ed3764e9add588e : SUCCESS in 4m 58s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/aadd3be79eb341e9b063757d39c4d50b : SUCCESS in 7m 07s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/e98387d87d434de39fecacc83fa3d9d0 : SUCCESS in 7m 18s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/afcf0083d16e4e67b1fbbd568bb1ea1a : SUCCESS in 7m 27s (non-voting)\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/d57b3a1832ce47f7b2549c86f33f5ce7 : SUCCESS in 1h 09m 52s\n- octavia-v2-dsvm-scenario-traffic-ops https://zuul.opendev.org/t/openstack/build/c46a3a09de8d46c99472265b98bbfa2f : SUCCESS in 1h 56m 53s\n- octavia-v2-dsvm-scenario-non-traffic-ops https://zuul.opendev.org/t/openstack/build/57693289409749c0bb9c28e158863046 : SUCCESS in 55m 12s\n- octavia-v2-dsvm-scenario-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/94f7f5a402784541ac6c998ad5fcabbc : SUCCESS in 1h 41m 41s\n- octavia-v2-dsvm-scenario-non-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/49d522b02a364c26bb7699d68b0aacb7 : SUCCESS in 1h 18m 47s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/b1343c77c94845cc92863d2bdd0de1e3 : SUCCESS in 38m 05s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/472dc62f2d7b4d27ada6015fc8253363 : SUCCESS in 52m 02s\n- octavia-grenade-skip-level https://zuul.opendev.org/t/openstack/build/633e52606b5546399e5a0577b1c123bf : FAILURE in 34m 55s (non-voting)\n- octavia-v2-dsvm-tls-barbican-fips https://zuul.opendev.org/t/openstack/build/c8ac824b67064151bc0bf06e59ef7489 : SUCCESS in 1h 11m 56s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/14201e7753dc415eae5a5a026c5ed896 : SUCCESS in 45m 29s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/d02b515d8bde482fa6b14f5856e86bef : SUCCESS in 2h 28m 36s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/432367b49d3a49a890aae9caed4a8cb1 : SUCCESS in 2h 42m 49s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/c4268dc0774743ab9b834bf2952ecce7 : SUCCESS in 2h 28m 37s (non-voting)\n- octavia-v2-dsvm-scenario-centos-9-stream https://zuul.opendev.org/t/openstack/build/6e2016765b234fbba4cbdfa29eeb3d3f : SUCCESS in 2h 31m 00s (non-voting)","accounts_in_message":[],"_revision_number":8},{"id":"0dd3b9db6bfc495cfcb111ce366c4485f5e3d9bb","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"date":"2023-11-02 11:07:59.000000000","message":"Uploaded patch set 9: Patch Set 8 was rebased.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":9},{"id":"151ab778552440ceb46976b60aa3e6ea2bf45f3a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-02 13:44:44.000000000","message":"Patch Set 9: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/83a6e624d5454d5c897eb639f64e4869\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/f6534500bfa24131a87a2202cf7cd62a : SUCCESS in 8m 45s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5519305c6ba547f8bd72062794292e6b : SUCCESS in 4m 56s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/0681f83d6b204d01b671d8d3bcaa102c : SUCCESS in 4m 46s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/20a2e72896a649a8bc477d51754d7810 : SUCCESS in 8m 11s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/83cbf897e8ad4295b29f8304702b3ddb : SUCCESS in 12m 37s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/72b1c37252004e5f9c0f65b8be265446 : SUCCESS in 7m 26s\n- octavia-tox-py39-tips https://zuul.opendev.org/t/openstack/build/91df658abf3f4180946c47dfc3abc8f0 : SUCCESS in 4m 51s\n- octavia-tox-functional-py39-tips https://zuul.opendev.org/t/openstack/build/ad858cd29e304dbf87c0616af3835192 : SUCCESS in 9m 38s\n- octavia-tox-functional-py39-sqlalchemy-tips https://zuul.opendev.org/t/openstack/build/839016f08cc04f22a2d9f72dc9e74fa6 : SUCCESS in 10m 29s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/36ac703f84d7474baf56e5a3797e8e22 : SUCCESS in 4m 24s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/6d10c774bbe347b59e878fcdd732e4e3 : SUCCESS in 7m 57s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/3434f04702f341f7ae0a9d9835331d4e : SUCCESS in 9m 33s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/c3a556a6eff7497b9a53a195154aa684 : SUCCESS in 7m 27s (non-voting)\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/e4106bcf80f6467399eedd265e239a21 : SUCCESS in 1h 16m 18s\n- octavia-v2-dsvm-scenario-traffic-ops https://zuul.opendev.org/t/openstack/build/1abb87c69d7142c8807169c7a5b29319 : SUCCESS in 1h 56m 27s\n- octavia-v2-dsvm-scenario-non-traffic-ops https://zuul.opendev.org/t/openstack/build/483ed42e9f0747ab92ebf171f0e4c633 : SUCCESS in 1h 15m 12s\n- octavia-v2-dsvm-scenario-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/3219bf74c59343b6bad51e72af947aa8 : SUCCESS in 2h 03m 07s\n- octavia-v2-dsvm-scenario-non-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/ee5e9c8ce1214b578937984962782d5e : SUCCESS in 1h 32m 25s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/bcc716ccb3614309b8ca3be745b87a02 : SUCCESS in 52m 55s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/949b82504a894cfd8cfb8f2f1d262615 : SUCCESS in 55m 38s\n- octavia-grenade-skip-level https://zuul.opendev.org/t/openstack/build/15a3328da0dd4637b33ad38283836ed3 : SUCCESS in 56m 25s (non-voting)\n- octavia-v2-dsvm-tls-barbican-fips https://zuul.opendev.org/t/openstack/build/0de8563263d840b3bd46db09df9b32bf : SUCCESS in 1h 03m 59s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/eefc181da08c4f86becd67d61370eb27 : SUCCESS in 41m 23s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/661b46b410ab49d9b23ad0a7bd352f7c : SUCCESS in 2h 03m 27s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/872520ba2d8144e3b42b6e3f756e1d27 : SUCCESS in 2h 20m 48s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/06e8a7b8eb31426ea2490c8400f3f7d0 : SUCCESS in 2h 14m 00s (non-voting)\n- octavia-v2-dsvm-scenario-centos-9-stream https://zuul.opendev.org/t/openstack/build/d35e34c605c542ac947d01612805a8a3 : TIMED_OUT in 2h 34m 15s (non-voting)","accounts_in_message":[],"_revision_number":9},{"id":"4f361eb24f7de76f445ed6a2ded757cf8c1b88f9","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2023-11-07 08:44:30.000000000","message":"Patch Set 9: Code-Review+2","accounts_in_message":[],"_revision_number":9},{"id":"ab895f4e0320ea5207db912cc7579f1c894e5738","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2024-01-10 17:36:21.000000000","message":"Patch Set 9: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":9},{"id":"00f10908c1d2138e0d9ba3c81c46733420fa1126","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-10 17:36:58.000000000","message":"Patch Set 9: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":9},{"id":"19573e22cd360568a5f3ba8a80621e940b866211","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-10 19:16:14.000000000","message":"Patch Set 9: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/de7701a84cac477f8f3800ebc409ba47\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1aea490c31244001a8463b07c9fbaf8c : SUCCESS in 5m 39s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/d942dc5916a242269c4e09be1be7c419 : SUCCESS in 5m 02s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/0e3b22ee824e43beb4300ed52f8468f0 : SUCCESS in 5m 33s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/baa2a2641a3d40e7821fbd713663fabf : SUCCESS in 11m 39s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/8a9ee4614b914ee797d0c123e620e038 : SUCCESS in 5m 21s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/511f330f9db94d2884ac5ccc1150a7c6 : SUCCESS in 4m 49s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/344c68fb4b8646b9be44248c3202fcfc : SUCCESS in 8m 19s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/c6b069619ec9457381bc08e4b32f2613 : SUCCESS in 1h 22m 29s\n- octavia-v2-dsvm-scenario-traffic-ops https://zuul.opendev.org/t/openstack/build/edd7b89acb40475a8abd5ac1a48c283a : SUCCESS in 1h 38m 41s\n- octavia-v2-dsvm-scenario-non-traffic-ops https://zuul.opendev.org/t/openstack/build/4ea24620f4244c80a794ddb352a25d60 : SUCCESS in 1h 03m 48s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/0e2a431ae47b49a78e3b405c2aaa2301 : SUCCESS in 40m 04s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/57895e0f88b342ab85edd34f4d0de585 : SUCCESS in 43m 46s","accounts_in_message":[],"_revision_number":9},{"id":"b02b115c954b18a56b202abf8eff2b1cf5fece5c","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-10 19:16:14.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":9},{"id":"cc7d6b1a466026ffc28cea27922604d1ffa9cc87","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-10 19:17:28.000000000","message":"Patch Set 9:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/38455545d3164754a3b03d50209f0146\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a225fd89c4d947baaaa398d91bf745fb : SUCCESS in 54s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/9744d976be8445eb992831e08171e659 : SUCCESS in 43s","accounts_in_message":[],"_revision_number":9}],"current_revision_number":9,"current_revision":"0c367b4fedecbc0242d6bd3c217bc33e92ab5a1d","revisions":{"3cafa3ba6e1b0ed5abe9d8e15bdf04d74df9737a":{"kind":"REWORK","_number":1,"created":"2023-08-08 16:53:41.000000000","uploader":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"ref":"refs/changes/14/890814/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/14/890814/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/14/890814/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/14/890814/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/14/890814/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/14/890814/1"}}},"commit":{"parents":[{"commit":"12d8e0de5de9ca754859542bcffd2311f5145d61","subject":"Merge \"Update doc for the OVN provider\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/12d8e0de5de9ca754859542bcffd2311f5145d61"}]}],"author":{"name":"Omer","email":"oschwart@redhat.com","date":"2023-08-08 16:46:44.000000000","tz":120},"committer":{"name":"Omer","email":"oschwart@redhat.com","date":"2023-08-08 16:46:44.000000000","tz":120},"subject":"WIP Provide Amphora stats for Octavia no-op drivers","message":"WIP Provide Amphora stats for Octavia no-op drivers\n\nSo far, when Octavia was running with noop drivers, there were no\namphora statistics data provided.\n\nThis patch adds fake statistics to amphora when Octavia uses noop\ndriver.\n\nCloses-Bug: #2030774\nChange-Id: Ib65e459bcd10a5ab877c0cf6f234d634d25d1e55\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/3cafa3ba6e1b0ed5abe9d8e15bdf04d74df9737a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/3cafa3ba6e1b0ed5abe9d8e15bdf04d74df9737a"}]},"branch":"refs/heads/master"},"6e56dfac7851605f0a7af520d552ec322479d331":{"kind":"REWORK","_number":2,"created":"2023-08-09 12:54:48.000000000","uploader":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"ref":"refs/changes/14/890814/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/14/890814/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/14/890814/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/14/890814/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/14/890814/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/14/890814/2"}}},"commit":{"parents":[{"commit":"12d8e0de5de9ca754859542bcffd2311f5145d61","subject":"Merge \"Update doc for the OVN provider\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/12d8e0de5de9ca754859542bcffd2311f5145d61"}]}],"author":{"name":"Omer","email":"oschwart@redhat.com","date":"2023-08-08 16:46:44.000000000","tz":120},"committer":{"name":"Omer","email":"oschwart@redhat.com","date":"2023-08-09 12:54:27.000000000","tz":120},"subject":"Provide Amphora stats for Octavia no-op drivers","message":"Provide Amphora stats for Octavia no-op drivers\n\nSo far, when Octavia was running with noop drivers, there were no\namphora statistics data provided.\n\nThis patch adds fake statistics to amphora when Octavia uses noop\ndriver.\n\nCloses-Bug: #2030774\nChange-Id: Ib65e459bcd10a5ab877c0cf6f234d634d25d1e55\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/6e56dfac7851605f0a7af520d552ec322479d331"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/6e56dfac7851605f0a7af520d552ec322479d331"}]},"branch":"refs/heads/master"},"34a44c80bfa0341f12b9435aeb9607661f4a05d4":{"kind":"REWORK","_number":3,"created":"2023-08-09 12:59:55.000000000","uploader":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"ref":"refs/changes/14/890814/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/14/890814/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/14/890814/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/14/890814/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/14/890814/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/14/890814/3"}}},"commit":{"parents":[{"commit":"12d8e0de5de9ca754859542bcffd2311f5145d61","subject":"Merge \"Update doc for the OVN provider\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/12d8e0de5de9ca754859542bcffd2311f5145d61"}]}],"author":{"name":"Omer","email":"oschwart@redhat.com","date":"2023-08-08 16:46:44.000000000","tz":120},"committer":{"name":"Omer","email":"oschwart@redhat.com","date":"2023-08-09 12:59:47.000000000","tz":120},"subject":"Provide Amphora stats for Octavia no-op drivers","message":"Provide Amphora stats for Octavia no-op drivers\n\nSo far, when Octavia was running with noop drivers, there were no\namphora statistics data provided.\n\nThis patch adds fake statistics to amphora when Octavia uses noop\ndriver.\n\nCloses-Bug: #2030774\nChange-Id: Ib65e459bcd10a5ab877c0cf6f234d634d25d1e55\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/34a44c80bfa0341f12b9435aeb9607661f4a05d4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/34a44c80bfa0341f12b9435aeb9607661f4a05d4"}]},"branch":"refs/heads/master"},"9a6dfffc3297961411651085d3390293663aa714":{"kind":"REWORK","_number":4,"created":"2023-08-21 10:41:27.000000000","uploader":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"ref":"refs/changes/14/890814/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/14/890814/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/14/890814/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/14/890814/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/14/890814/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/14/890814/4"}}},"commit":{"parents":[{"commit":"2e563016ce77617e5416f27ff5800b8821561c25","subject":"Merge \"Remove blockdiag from Octavia\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/2e563016ce77617e5416f27ff5800b8821561c25"}]}],"author":{"name":"Omer","email":"oschwart@redhat.com","date":"2023-08-08 16:46:44.000000000","tz":120},"committer":{"name":"Omer","email":"oschwart@redhat.com","date":"2023-08-21 10:41:04.000000000","tz":120},"subject":"Provide Amphora stats for Octavia no-op drivers","message":"Provide Amphora stats for Octavia no-op drivers\n\nSo far, when Octavia was running with noop drivers, there were no\namphora statistics data provided.\n\nThis patch adds fake statistics to amphora when Octavia uses noop\ndriver.\n\nCloses-Bug: #2030774\nChange-Id: Ib65e459bcd10a5ab877c0cf6f234d634d25d1e55\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/9a6dfffc3297961411651085d3390293663aa714"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/9a6dfffc3297961411651085d3390293663aa714"}]},"branch":"refs/heads/master"},"832552544ce5ec8566b1170339236d00249e932c":{"kind":"REWORK","_number":5,"created":"2023-08-22 10:41:00.000000000","uploader":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"ref":"refs/changes/14/890814/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/14/890814/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/14/890814/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/14/890814/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/14/890814/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/14/890814/5"}}},"commit":{"parents":[{"commit":"2e563016ce77617e5416f27ff5800b8821561c25","subject":"Merge \"Remove blockdiag from Octavia\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/2e563016ce77617e5416f27ff5800b8821561c25"}]}],"author":{"name":"Omer","email":"oschwart@redhat.com","date":"2023-08-08 16:46:44.000000000","tz":120},"committer":{"name":"Omer","email":"oschwart@redhat.com","date":"2023-08-22 10:18:33.000000000","tz":120},"subject":"Provide Amphora stats for Octavia no-op drivers","message":"Provide Amphora stats for Octavia no-op drivers\n\nSo far, when Octavia was either running with noop drivers or deployed\nan LB (and amphora) without a listener, there were no amphora statistics\ndata provided.\n\nThis patch adds empty statistics to amphora for those scenarios.\n\nCloses-Bug: #2030774\nChange-Id: Ib65e459bcd10a5ab877c0cf6f234d634d25d1e55\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/832552544ce5ec8566b1170339236d00249e932c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/832552544ce5ec8566b1170339236d00249e932c"}]},"branch":"refs/heads/master"},"092042b7c9c1163bdcd5fadf435a1152f3f5e1e2":{"kind":"REWORK","_number":6,"created":"2023-09-01 13:28:41.000000000","uploader":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"ref":"refs/changes/14/890814/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/14/890814/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/14/890814/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/14/890814/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/14/890814/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/14/890814/6"}}},"commit":{"parents":[{"commit":"0426285a41464a5015494584f109eed35a0d44db","subject":"Merge \"Add octavia-grenade-slurp CI job\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/0426285a41464a5015494584f109eed35a0d44db"}]}],"author":{"name":"Omer","email":"oschwart@redhat.com","date":"2023-08-08 16:46:44.000000000","tz":120},"committer":{"name":"Omer","email":"oschwart@redhat.com","date":"2023-09-01 13:22:32.000000000","tz":120},"subject":"Provide Amphora stats for Octavia no-op drivers","message":"Provide Amphora stats for Octavia no-op drivers\n\nSo far, when Octavia was running with noop drivers, there were no\namphora statistics data provided and 404 was returned as the\nAmphoraStatistics object was not created, and therefore not found.\n\nThis patch adds fake statistics to amphora noop driver.\n\nCloses-Bug: #2030774\nChange-Id: Ib65e459bcd10a5ab877c0cf6f234d634d25d1e55\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/092042b7c9c1163bdcd5fadf435a1152f3f5e1e2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/092042b7c9c1163bdcd5fadf435a1152f3f5e1e2"}]},"branch":"refs/heads/master"},"73270b6fbad126a5fffb0d5fc49f6c471480710d":{"kind":"TRIVIAL_REBASE","_number":7,"created":"2023-10-30 13:41:59.000000000","uploader":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"ref":"refs/changes/14/890814/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/14/890814/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/14/890814/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/14/890814/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/14/890814/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/14/890814/7"}}},"commit":{"parents":[{"commit":"ed8c7a1e7e95650b4680630e39a84505dd4ac37f","subject":"Add Noop Certificate Manager","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/ed8c7a1e7e95650b4680630e39a84505dd4ac37f"}]}],"author":{"name":"Omer","email":"oschwart@redhat.com","date":"2023-08-08 16:46:44.000000000","tz":120},"committer":{"name":"Omer","email":"oschwart@redhat.com","date":"2023-10-30 13:41:19.000000000","tz":60},"subject":"Provide Amphora stats for Octavia no-op drivers","message":"Provide Amphora stats for Octavia no-op drivers\n\nSo far, when Octavia was running with noop drivers, there were no\namphora statistics data provided and 404 was returned as the\nAmphoraStatistics object was not created, and therefore not found.\n\nThis patch adds fake statistics to amphora noop driver.\n\nCloses-Bug: #2030774\nChange-Id: Ib65e459bcd10a5ab877c0cf6f234d634d25d1e55\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/73270b6fbad126a5fffb0d5fc49f6c471480710d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/73270b6fbad126a5fffb0d5fc49f6c471480710d"}]},"branch":"refs/heads/master"},"06a3b1d88168e21831a4d1a549891f66beb4215c":{"kind":"REWORK","_number":8,"created":"2023-10-31 09:39:34.000000000","uploader":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"ref":"refs/changes/14/890814/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/14/890814/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/14/890814/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/14/890814/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/14/890814/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/14/890814/8"}}},"commit":{"parents":[{"commit":"ed8c7a1e7e95650b4680630e39a84505dd4ac37f","subject":"Add Noop Certificate Manager","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/ed8c7a1e7e95650b4680630e39a84505dd4ac37f"}]}],"author":{"name":"Omer","email":"oschwart@redhat.com","date":"2023-08-08 16:46:44.000000000","tz":120},"committer":{"name":"Omer","email":"oschwart@redhat.com","date":"2023-10-31 09:39:00.000000000","tz":60},"subject":"Provide Amphora stats for Octavia no-op drivers","message":"Provide Amphora stats for Octavia no-op drivers\n\nSo far, when Octavia was running with noop drivers, there were no\namphora statistics data provided and 404 was returned as the\nAmphoraStatistics object was not created, and therefore not found.\n\nThis patch adds fake statistics to amphora noop driver.\n\nCloses-Bug: #2030774\nChange-Id: Ib65e459bcd10a5ab877c0cf6f234d634d25d1e55\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/06a3b1d88168e21831a4d1a549891f66beb4215c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/06a3b1d88168e21831a4d1a549891f66beb4215c"}]},"branch":"refs/heads/master"},"0c367b4fedecbc0242d6bd3c217bc33e92ab5a1d":{"kind":"TRIVIAL_REBASE","_number":9,"created":"2023-11-02 11:07:59.000000000","uploader":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"ref":"refs/changes/14/890814/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/14/890814/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/14/890814/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/14/890814/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/14/890814/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/14/890814/9"}}},"commit":{"parents":[{"commit":"7310986de9bf68ed86de90de0501a1bc46945526","subject":"Add Noop Certificate Manager","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/7310986de9bf68ed86de90de0501a1bc46945526"}]}],"author":{"name":"Omer","email":"oschwart@redhat.com","date":"2023-08-08 16:46:44.000000000","tz":120},"committer":{"name":"Omer","email":"oschwart@redhat.com","date":"2023-11-02 11:07:13.000000000","tz":60},"subject":"Provide Amphora stats for Octavia no-op drivers","message":"Provide Amphora stats for Octavia no-op drivers\n\nSo far, when Octavia was running with noop drivers, there were no\namphora statistics data provided and 404 was returned as the\nAmphoraStatistics object was not created, and therefore not found.\n\nThis patch adds fake statistics to amphora noop driver.\n\nCloses-Bug: #2030774\nChange-Id: Ib65e459bcd10a5ab877c0cf6f234d634d25d1e55\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/0c367b4fedecbc0242d6bd3c217bc33e92ab5a1d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/0c367b4fedecbc0242d6bd3c217bc33e92ab5a1d"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
