)]}'
{"id":"openstack%2Foctavia~890958","triplet_id":"openstack%2Foctavia~master~Ie0e865f9a12ad565112d7c458cea1c2704dee0f8","project":"openstack/octavia","branch":"master","attention_set":{},"removed_from_attention_set":{"11628":{"account":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"last_update":"2023-08-24 14:37:57.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Ie0e865f9a12ad565112d7c458cea1c2704dee0f8","subject":"Remove unused wait_for_port_detach code","status":"MERGED","created":"2023-08-09 19:51:26.000000000","updated":"2023-08-24 14:39:30.000000000","submitted":"2023-08-24 14:37:57.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"890958","meta_rev_id":"547e37d794f0cdd32036c2daaed2c5b1872a7ca6","_number":890958,"virtual_id_number":890958,"owner":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-08-24 14:37:57.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"all":[{"value":2,"date":"2023-08-17 12:43:37.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2023-08-24 12:36:46.000000000","_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"all":[{"value":0,"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2023-08-24 12:36:46.000000000","_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-08-09 22:53:55.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-08-17 12:43:37.000000000","updated_by":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"reviewer":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"state":"REVIEWER"},{"updated":"2023-08-24 12:36:46.000000000","updated_by":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"reviewer":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"state":"REVIEWER"}],"messages":[{"id":"f43ef995ddeb74910c477f5f382874548d2e2860","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2023-08-09 19:51:26.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"fb6f7625f40b695a930e519a32d2329c41b53b4c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-09 22:53:55.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f9a675a385244e539a1555503a0d93ba\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/68a15dbab4594f649157c3ab207a8642 : SUCCESS in 11m 15s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4bd33c6341e74335be4f638f3692a3f6 : SUCCESS in 5m 39s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/cc8be8bdd045485f98f5a644c3ab7525 : SUCCESS in 4m 54s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/39f33064a4e6415f9ced187234c000ac : SUCCESS in 5m 20s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/31278b3d6152420ab066fe3e93080b24 : SUCCESS in 5m 14s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/40f6eb6118854aa8bcd9674af7a52151 : SUCCESS in 10m 38s\n- octavia-tox-py39-tips https://zuul.opendev.org/t/openstack/build/1bbe9de6d6164353abe8ef6d4c647222 : SUCCESS in 5m 38s\n- octavia-tox-functional-py39-tips https://zuul.opendev.org/t/openstack/build/d942f76cc5f04728900bc69cb137aedd : SUCCESS in 7m 28s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/cdf1dd023de64d8eae839c3573873151 : SUCCESS in 4m 19s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/f25a357e44f24f5d946fa50ace187e6f : SUCCESS in 8m 06s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/537e566618124f07aa8ab01014bd1acd : SUCCESS in 8m 37s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/4a53932e137442a59e295e66a3814906 : SUCCESS in 8m 27s (non-voting)\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/43f32979f9c44b279f6db6b36fe8b6ac : SUCCESS in 1h 21m 49s\n- octavia-v2-dsvm-scenario-traffic-ops https://zuul.opendev.org/t/openstack/build/d88fb4a8d8394c7d8ba70b123bb64cfb : SUCCESS in 2h 07m 16s\n- octavia-v2-dsvm-scenario-non-traffic-ops https://zuul.opendev.org/t/openstack/build/b0bfaa8af6274fb99f938ca3a0c64424 : SUCCESS in 1h 22m 11s\n- octavia-v2-dsvm-scenario-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/ea6ca0593d6f48beabbe4e2905819551 : SUCCESS in 1h 41m 10s\n- octavia-v2-dsvm-scenario-non-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/de2cdc4a857c44ef9ec6df0b1d241a16 : SUCCESS in 1h 37m 22s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/f3c2ec52240e4f199c20570c51f938d9 : SUCCESS in 44m 01s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/63406a3b34104bd3bedbacf754adccd9 : SUCCESS in 50m 03s\n- octavia-v2-dsvm-tls-barbican-fips https://zuul.opendev.org/t/openstack/build/b86c358e80274cb9bafaddaa28f6d2ba : SUCCESS in 1h 04m 43s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/b404e5eadcf74e5193a6626c13edc538 : SUCCESS in 39m 04s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/e028f4d8defb4f72bd6b150b865795dd : SUCCESS in 2h 03m 31s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/211ccf87b12b47fea08dcaab7c248324 : SUCCESS in 2h 57m 59s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/36d37adb424647309f51a9adf8bcddec : SUCCESS in 1h 59m 57s (non-voting)\n- octavia-v2-dsvm-scenario-centos-9-stream https://zuul.opendev.org/t/openstack/build/80c444b51f734129950d937bd99ec6fe : SUCCESS in 2h 23m 59s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"642a6dd0b9d7a0568a32685c7d90dbcb987894d5","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2023-08-17 12:43:37.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"6677f29d9afe86b1e30c19ff95cbb1444f45dede","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2023-08-24 12:36:46.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"d2c73c4c2cb0c5c1468aa2878881b4c364102a9f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-24 12:37:21.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"243b317a50dc1831531061621a092d92963e00bd","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-24 14:37:57.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d1f919ac2d0b419e956e57d3e4996d93\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9b1c93462f634a5f8245858908c4c69c : SUCCESS in 5m 26s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/b78ffdc1cde74e84911fda5b229b2bf3 : SUCCESS in 5m 00s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/ca37884492a94aa38df613f0e1f686ae : SUCCESS in 4m 58s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/20ee06296f104b17aed7788455ceedd5 : SUCCESS in 5m 07s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0931287c64be416384794d18f3df0fc9 : SUCCESS in 11m 50s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/8bdaa7b5e59145a79564bd16a806274b : SUCCESS in 4m 42s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/74f17eeb703f4beaaf82696c100ddcf0 : SUCCESS in 6m 38s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/8e2e91fc983e4f0090c3cf5efdf9d02b : SUCCESS in 1h 24m 07s\n- octavia-v2-dsvm-scenario-traffic-ops https://zuul.opendev.org/t/openstack/build/eb123764811a456aa84e6124d9e2d73b : SUCCESS in 1h 59m 56s\n- octavia-v2-dsvm-scenario-non-traffic-ops https://zuul.opendev.org/t/openstack/build/68849413a377452bb306025828efbf42 : SUCCESS in 1h 13m 21s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/a623431728954b31beb44b0025df234c : SUCCESS in 38m 37s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/8181f3b319054c6fb60c73fc2bc848a1 : SUCCESS in 40m 09s","accounts_in_message":[],"_revision_number":1},{"id":"e11bd19bf924c73a058e201d0b409ef4630bfc26","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-24 14:37:57.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"547e37d794f0cdd32036c2daaed2c5b1872a7ca6","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-24 14:39:30.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c7800961e0a24335a7c41b8a46292dc5\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/87a524f8581044b281a9731f7151591c : SUCCESS in 1m 08s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"fd587500c6e9666a92701e53407e8420b0e6b466","revisions":{"fd587500c6e9666a92701e53407e8420b0e6b466":{"kind":"REWORK","_number":1,"created":"2023-08-09 19:51:26.000000000","uploader":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"ref":"refs/changes/58/890958/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/58/890958/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/58/890958/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/58/890958/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/58/890958/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/58/890958/1"}}},"commit":{"parents":[{"commit":"8ccc0d01764c234ce9d3dc0191b5a0952111e2ef","subject":"Don\u0027t update the management port when calculating delta","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/8ccc0d01764c234ce9d3dc0191b5a0952111e2ef"}]}],"author":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2023-08-09 19:50:19.000000000","tz":0},"committer":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2023-08-09 19:51:24.000000000","tz":0},"subject":"Remove unused wait_for_port_detach code","message":"Remove unused wait_for_port_detach code\n\nAfter the failover v2 patch, the wait_for_port_detach code is no longer used.\nThis patch removes those methods from the networking drivers and unit tests.\n\nChange-Id: Ie0e865f9a12ad565112d7c458cea1c2704dee0f8\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/fd587500c6e9666a92701e53407e8420b0e6b466"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/fd587500c6e9666a92701e53407e8420b0e6b466"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
