)]}'
{"id":"openstack%2Foctavia~892281","triplet_id":"openstack%2Foctavia~master~I4f858a71784ae640bb6b5eddf9347c7b70afa632","project":"openstack/octavia","branch":"master","topic":"fix-concurrent-delete","attention_set":{"11290":{"account":{"_account_id":11290,"name":"Gaudenz Steinlin","email":"gaudenz.steinlin@cloudscale.ch","username":"gaudenz"},"last_update":"2023-12-18 00:28:23.000000000","reason":"A robot voted negatively on a label"}},"removed_from_attention_set":{},"hashtags":[],"change_id":"I4f858a71784ae640bb6b5eddf9347c7b70afa632","subject":"Raise NotFound on update of inexistent resources","status":"NEW","created":"2023-08-22 08:55:56.000000000","updated":"2023-12-20 11:02:00.000000000","submit_type":"MERGE_IF_NECESSARY","mergeable":false,"submittable":false,"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"e95608dd80f58016f2d9e56b0f31b738273f11ed","_number":892281,"virtual_id_number":892281,"owner":{"_account_id":11290,"name":"Gaudenz Steinlin","email":"gaudenz.steinlin@cloudscale.ch","username":"gaudenz"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":-1,"date":"2023-12-20 11:02:00.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-08-22 12:00:55.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"7faf9ff7668c8c4384d51e08fd00f4b9e6dc152b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11290,"name":"Gaudenz Steinlin","email":"gaudenz.steinlin@cloudscale.ch","username":"gaudenz"},"date":"2023-08-22 08:55:56.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"0c41c29e86342bc7362d23971c75e299bdf2da3b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-22 12:00:55.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/59091b94644d4a09aeaa840da7d7a0ac\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/477c29ed845e4d9fb040238d1e74751a : SUCCESS in 9m 36s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d35cea6b9b324c46a9244243cc037cb6 : SUCCESS in 5m 53s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/571cf1b8db8d4efd999d2405b6c3007a : SUCCESS in 4m 44s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/09d77b59e7d2488e856d480b9a77d2eb : SUCCESS in 5m 02s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/6e5b12c21a6641679f1e4809d8c12e42 : SUCCESS in 5m 03s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/0158e68bf9324a2189735e030a94053a : SUCCESS in 5m 26s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/eae0464d5bd14ffab43beb8de6913521 : SUCCESS in 12m 55s\n- octavia-tox-py39-tips https://zuul.opendev.org/t/openstack/build/a46dff731a0c4606bb66cf57536f72cf : SUCCESS in 5m 22s\n- octavia-tox-functional-py39-tips https://zuul.opendev.org/t/openstack/build/d2963ba20de24aee9910fc562198985b : SUCCESS in 7m 52s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/5591f3dcf7cd46b8a762f4901656f4fd : SUCCESS in 4m 19s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/9b68e83363aa4fe689e8c4a98c681746 : SUCCESS in 8m 05s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/54d1f5f153df482ebc71cc6e5d20032f : SUCCESS in 7m 20s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/1a97b6fd89094fd2b8202f30ef2a97c0 : SUCCESS in 8m 45s (non-voting)\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/a64d8bf50dc14498830775d461c67dac : TIMED_OUT in 3h 03m 02s\n- octavia-v2-dsvm-scenario-traffic-ops https://zuul.opendev.org/t/openstack/build/7d657a97e924459fa7ed331e8e39c494 : SUCCESS in 1h 53m 52s\n- octavia-v2-dsvm-scenario-non-traffic-ops https://zuul.opendev.org/t/openstack/build/d56db29f9ef946eca033ea5a6c96e8df : SUCCESS in 1h 07m 56s\n- octavia-v2-dsvm-scenario-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/82c53057b2934eb39a36ee37718e1714 : SUCCESS in 2h 03m 40s\n- octavia-v2-dsvm-scenario-non-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/9c6ae00522ee4abb83017ca8b8cd2a8a : SUCCESS in 1h 19m 41s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/19920515cd2d4e7ea08186f10a8bbfed : SUCCESS in 47m 09s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/dc9704af57a64dd5a56414f82990c620 : SUCCESS in 42m 31s\n- octavia-v2-dsvm-tls-barbican-fips https://zuul.opendev.org/t/openstack/build/87a90c298e7b4c39b4f6e863c7d35db4 : FAILURE in 27m 29s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/6ddc4b845615404b90f4dc2dcc752990 : FAILURE in 42m 00s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/4a60a2464aa649ab9f5d6158e5e715cd : TIMED_OUT in 2h 40m 46s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/e96e9d653c044f3c9ac11593f03d8fb7 : TIMED_OUT in 3h 03m 41s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/74b17be7e9c7476b905f74189d884294 : TIMED_OUT in 2h 33m 03s (non-voting)\n- octavia-v2-dsvm-scenario-centos-9-stream https://zuul.opendev.org/t/openstack/build/b8a2adfc3e9c45a0bf397d8664ff0833 : FAILURE in 17m 30s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"8782aefc6358d915f803389b0bbf8cac18f7ede8","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11290,"name":"Gaudenz Steinlin","email":"gaudenz.steinlin@cloudscale.ch","username":"gaudenz"},"date":"2023-08-23 07:00:56.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"731369cc85c4fcd1576593b84fc34e085fb35b91","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-23 10:06:42.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/124dd19ff10042d1b3328e035748f41b\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/29efe8e89343428d936788714967cd42 : SUCCESS in 12m 37s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/284216cbd1794d36890e9a22341d717e : SUCCESS in 4m 32s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/045eced1bd1e470b9ec3203af732d883 : SUCCESS in 4m 33s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/26fd02a52c9046bdbcae293233f606f6 : SUCCESS in 4m 52s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/68948ee9a20c4048b854b238473d9ff3 : SUCCESS in 4m 23s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/5fd7065bc03a410f9c2b2a4e79f7be2e : SUCCESS in 5m 27s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3ecf7886a96443d681122971359ca3ff : SUCCESS in 10m 48s\n- octavia-tox-py39-tips https://zuul.opendev.org/t/openstack/build/efa4422002da4a78abe2b564d001c999 : SUCCESS in 5m 09s\n- octavia-tox-functional-py39-tips https://zuul.opendev.org/t/openstack/build/26db20b3c2ec4513a3535e1b9436e99a : SUCCESS in 9m 07s\n- octavia-tox-functional-py39-sqlalchemy-tips https://zuul.opendev.org/t/openstack/build/3844eb9008104cb7b615ef93137ecf55 : SUCCESS in 10m 22s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/cac769989a6d4beca3a6979dbd796842 : SUCCESS in 4m 29s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/4a0c108c82794c4aaaef3b9148b10399 : SUCCESS in 8m 57s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/7fc69867cd154e6aa689256ed719cd2b : SUCCESS in 8m 23s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/8bed085016d447bdb7218ce6cc1df51a : SUCCESS in 9m 06s (non-voting)\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/8d4cf03f15a840c6a510ea226e7d06de : TIMED_OUT in 3h 04m 05s\n- octavia-v2-dsvm-scenario-traffic-ops https://zuul.opendev.org/t/openstack/build/029dda335f6b41b8ab50a48ec4b0d7eb : SUCCESS in 1h 45m 05s\n- octavia-v2-dsvm-scenario-non-traffic-ops https://zuul.opendev.org/t/openstack/build/fbada0bc9ae14b97832eed9df801f662 : SUCCESS in 1h 12m 45s\n- octavia-v2-dsvm-scenario-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/01ea1c8da03f469fa2be80e59ec94e35 : SUCCESS in 1h 42m 17s\n- octavia-v2-dsvm-scenario-non-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/135b4bab7133415795e0c801110f75bd : SUCCESS in 1h 21m 13s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/2902b29725e6423ab571315f65b8b817 : SUCCESS in 43m 03s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/5f78a4fca38547739ee801e447433e48 : SUCCESS in 44m 43s\n- octavia-v2-dsvm-tls-barbican-fips https://zuul.opendev.org/t/openstack/build/98e2c06330ae4d7094ef5f9d81f27947 : FAILURE in 19m 41s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/c943e6a8cfda4f04b103b4add574aa31 : FAILURE in 43m 04s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/4458b7d5bece451fb11ddf21a5cb2319 : TIMED_OUT in 2h 33m 05s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/94001193721c4bc2b1fb725c98b7fc29 : SUCCESS in 2h 55m 45s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/1559f3c62eea4e62a49d29e1fdbea910 : SUCCESS in 2h 07m 51s (non-voting)\n- octavia-v2-dsvm-scenario-centos-9-stream https://zuul.opendev.org/t/openstack/build/ab18f5151a4440908eac6c9080c31257 : FAILURE in 14m 43s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"08046bcdc57e67d0bc9e786ff77cfcb729adbf7e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11290,"name":"Gaudenz Steinlin","email":"gaudenz.steinlin@cloudscale.ch","username":"gaudenz"},"date":"2023-08-23 14:18:31.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":3},{"id":"6e5e4c1999410fbd8f91c3271623416e45729dfc","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-23 17:16:10.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0975954ae5474bfb926154ccf2210921\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/72e3249de5604c7aa0b367542d729b5e : SUCCESS in 9m 19s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/148c0c4770264b81aa5a4b2bf545b5f0 : SUCCESS in 4m 34s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/924554759c0742d18b04516e1249840c : SUCCESS in 4m 58s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/cfb8c0184a5c48c7a86c7299c39ae4f3 : SUCCESS in 5m 49s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/5d694c7f65eb439683cfbaff362eae01 : SUCCESS in 5m 29s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/1618c7b42a6e4293a93724ccef3873eb : SUCCESS in 5m 15s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/10cb277a76fd4605a129d2987541cf05 : SUCCESS in 12m 22s\n- octavia-tox-py39-tips https://zuul.opendev.org/t/openstack/build/415919d8e11b4524a92953c27fc39500 : SUCCESS in 5m 24s\n- octavia-tox-functional-py39-tips https://zuul.opendev.org/t/openstack/build/23fb1aa1800a405a880a0193cbb2f3b3 : SUCCESS in 7m 51s\n- octavia-tox-functional-py39-sqlalchemy-tips https://zuul.opendev.org/t/openstack/build/a733080991e1419f8982de173e9dc11d : SUCCESS in 13m 15s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/db3c0261525741f3a11a3a35195395f2 : SUCCESS in 4m 29s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/6c5bd9e8e5954aae9c829239bd3ed084 : SUCCESS in 7m 15s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/7b517224b4324870965a9f2b22a86f2d : SUCCESS in 7m 05s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/2397c2d79ed64b0d88a6fac7f2ff7a80 : SUCCESS in 7m 36s (non-voting)\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/a17ed2cb5af14634ba0a57476b93baa1 : SUCCESS in 1h 17m 06s\n- octavia-v2-dsvm-scenario-traffic-ops https://zuul.opendev.org/t/openstack/build/41cbe63022c54eeeaf34f8b2cf025248 : SUCCESS in 1h 26m 37s\n- octavia-v2-dsvm-scenario-non-traffic-ops https://zuul.opendev.org/t/openstack/build/860d18be85934828948a231dcf83bf58 : SUCCESS in 1h 01m 34s\n- octavia-v2-dsvm-scenario-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/04238e6738484b37b2b1a9b74836fadd : SUCCESS in 1h 29m 11s\n- octavia-v2-dsvm-scenario-non-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/9328457cfb824e8da0c77f057f2b75b4 : SUCCESS in 1h 11m 30s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/de7eb9cb40db4713b470228c5b9dcea4 : SUCCESS in 46m 50s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/01395e992e52426d8670660059ca47bd : SUCCESS in 49m 48s\n- octavia-v2-dsvm-tls-barbican-fips https://zuul.opendev.org/t/openstack/build/89d23896c1a94f379f5224c94069f4ab : FAILURE in 24m 25s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/7cad6cce07ab467e9f4638b183641792 : SUCCESS in 40m 47s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/2d77780039a34f828b2a7a160fc23550 : TIMED_OUT in 2h 32m 59s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/a401d15294b54dabb10b419c1fd31507 : SUCCESS in 2h 56m 13s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/3764e58d5cf34b4a99f576f17a6187aa : SUCCESS in 2h 02m 19s (non-voting)\n- octavia-v2-dsvm-scenario-centos-9-stream https://zuul.opendev.org/t/openstack/build/5791df1a41a3417aa327d5d2f6f223e1 : FAILURE in 16m 46s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"739b813005e436b6c3249cd9d949c2a399e5fd93","author":{"_account_id":11290,"name":"Gaudenz Steinlin","email":"gaudenz.steinlin@cloudscale.ch","username":"gaudenz"},"date":"2023-10-31 14:39:48.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"8ef2f632dfb93033d7c94fdebade771a740bb29f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11290,"name":"Gaudenz Steinlin","email":"gaudenz.steinlin@cloudscale.ch","username":"gaudenz"},"date":"2023-12-17 20:36:47.000000000","message":"Uploaded patch set 4: Patch Set 3 was rebased.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":4},{"id":"c30b98fc981c7bcd4168f73de3c053e2c4c542d4","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11290,"name":"Gaudenz Steinlin","email":"gaudenz.steinlin@cloudscale.ch","username":"gaudenz"},"date":"2023-12-17 21:31:11.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"ac78e623834774b638ed8a70ed2192cb351c6502","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-12-18 00:28:23.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/efee491398014820a6f3a36632748707\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/9c341ca46ef14da095c4b0564b993c8f : SUCCESS in 12m 32s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/cb406826cb9e4441a4f18696d18b4d0b : FAILURE in 6m 03s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/99ba393f61784826b6fcdf11f27fb347 : SUCCESS in 6m 42s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/8bb94f5eb51f495e936c195a76871090 : SUCCESS in 4m 59s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6386661940894211a479823df2810a82 : SUCCESS in 11m 58s\n- octavia-tox-py39-tips https://zuul.opendev.org/t/openstack/build/691163bcefe74992bada351e500510eb : SUCCESS in 5m 32s\n- octavia-tox-functional-py39-tips https://zuul.opendev.org/t/openstack/build/789649b6ad8744b887b34966fd78d479 : SUCCESS in 9m 34s\n- octavia-tox-functional-py39-sqlalchemy-tips https://zuul.opendev.org/t/openstack/build/f8f8e6e39d9b4007a678cc8d7309d29a : SUCCESS in 10m 35s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/94b77109f4124a7ab7eb541bdd98bbb9 : SUCCESS in 4m 57s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/5728667951264196a7df020f5f46078c : SUCCESS in 8m 02s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/5e184789ba3f440ba5a39add51fea127 : SUCCESS in 7m 46s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/1d48c2ca41d544c18916338024423263 : SUCCESS in 8m 34s (non-voting)\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/66c3e93510f74bda8dc1b63c478d28c7 : SUCCESS in 1h 35m 52s\n- octavia-v2-dsvm-scenario-traffic-ops https://zuul.opendev.org/t/openstack/build/3245e349daea4b20be7128bfdfd5cca8 : SUCCESS in 1h 35m 13s\n- octavia-v2-dsvm-scenario-non-traffic-ops https://zuul.opendev.org/t/openstack/build/817be8b2ea17467e96c86b16e5849dba : SUCCESS in 1h 01m 07s\n- octavia-v2-dsvm-scenario-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/544dfa7579d044a2892debb58d09e20a : SUCCESS in 1h 44m 35s\n- octavia-v2-dsvm-scenario-non-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/283c3c72c56a4943944c83dcea42759d : SUCCESS in 1h 15m 50s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/72ddefdd693047b384fedb9ecbd9caf0 : SUCCESS in 37m 39s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/6b33d2708b9f4939a1a379910054142f : SUCCESS in 42m 21s\n- octavia-grenade-skip-level https://zuul.opendev.org/t/openstack/build/920ef51458ea4031b0318d745ed6fa14 : SUCCESS in 40m 02s (non-voting)\n- octavia-v2-dsvm-tls-barbican-fips https://zuul.opendev.org/t/openstack/build/4304d09ed8824974b6cf526fa7a02c2b : SUCCESS in 1h 18m 45s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/ba12e3f86ff54768974c8e3ce2a04cc8 : SUCCESS in 33m 49s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/76165e11cea7418abe6370046fba75ff : SUCCESS in 2h 11m 04s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/2cdb0ce0997845eb8b8749bba932d64c : SUCCESS in 2h 31m 50s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/48ab163d88b54d6084754d1da2beed3e : SUCCESS in 2h 29m 48s (non-voting)\n- octavia-v2-dsvm-scenario-centos-9-stream https://zuul.opendev.org/t/openstack/build/d2bcedbda66b41509fd895fe3797ee78 : TIMED_OUT in 2h 34m 36s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"9230e067713dc345bcd2adce643dfe91e8cebfc0","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11290,"name":"Gaudenz Steinlin","email":"gaudenz.steinlin@cloudscale.ch","username":"gaudenz"},"date":"2023-12-20 08:08:27.000000000","message":"Uploaded patch set 6: Patch Set 5 was rebased.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":6},{"id":"e95608dd80f58016f2d9e56b0f31b738273f11ed","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-12-20 11:02:00.000000000","message":"Patch Set 6: Verified-1\n\n(40 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/788cfc0787d74d149a93ff1dcaced59d\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/af88191dcb1b4eb5a8689a70fcdbd6cf : SUCCESS in 11m 40s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/baf12b7d018e4090af77b807c150dd4d : FAILURE in 5m 43s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/70183a88963e4f94ad308fabbfdf90dc : SUCCESS in 5m 19s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/2bdbf967712b453d8e5d5b13666850a9 : SUCCESS in 5m 23s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e2a3241d679243cc97aa403e65c852aa : SUCCESS in 12m 16s\n- octavia-tox-py39-tips https://zuul.opendev.org/t/openstack/build/bef4c58c0eef43bbb6185e0be9052eca : SUCCESS in 5m 53s\n- octavia-tox-functional-py39-tips https://zuul.opendev.org/t/openstack/build/831d3c794c89437eb904bf8867362850 : SUCCESS in 10m 05s\n- octavia-tox-functional-py39-sqlalchemy-tips https://zuul.opendev.org/t/openstack/build/e693b0857feb4417a4c7e5c3a45b1041 : SUCCESS in 10m 51s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/4ffb2abd199f4131b632ed66dba1f7c0 : SUCCESS in 4m 54s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/89503dca6bc24708bd53982b533f3a3d : SUCCESS in 10m 34s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/247d9741c8a34d95a893bdb3507c2fc4 : SUCCESS in 7m 34s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/780a41f8b90a45328b7125e672de3221 : SUCCESS in 8m 31s (non-voting)\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/ba67161973344e7486652e08e91baaf9 : SUCCESS in 1h 24m 44s\n- octavia-v2-dsvm-scenario-traffic-ops https://zuul.opendev.org/t/openstack/build/67376d589d4f41fa8d121a12ad068f78 : SUCCESS in 1h 33m 46s\n- octavia-v2-dsvm-scenario-non-traffic-ops https://zuul.opendev.org/t/openstack/build/6d2c0180e8614c7eb6895ba38ac4903f : SUCCESS in 57m 31s\n- octavia-v2-dsvm-scenario-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/e75ec4e90f4f45aa87945bf91c0dc511 : SUCCESS in 1h 43m 57s\n- octavia-v2-dsvm-scenario-non-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/311a7c3416d8489a806f4b9c618cb7b5 : SUCCESS in 1h 01m 14s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/4814514965ea4378894664f56e6a0a87 : SUCCESS in 36m 46s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/0eb7394444b1434ba3676ec9c306f8f1 : SUCCESS in 42m 02s\n- octavia-grenade-skip-level https://zuul.opendev.org/t/openstack/build/487cd115dc26455e86066f78b809d405 : SUCCESS in 37m 36s (non-voting)\n- octavia-v2-dsvm-tls-barbican-fips https://zuul.opendev.org/t/openstack/build/6528c363879240a69701267170be48ac : SUCCESS in 1h 16m 59s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/28f004b4118a499ca494ba6aef80245d : SUCCESS in 35m 19s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/c8459fda318a4874b69f2149fc078543 : SUCCESS in 2h 16m 09s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/4cea417f535640828563bb9d82fd5816 : SUCCESS in 2h 30m 04s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/e3d61ed8252c4540a992e3d74dd16fca : SUCCESS in 2h 13m 21s (non-voting)\n- octavia-v2-dsvm-scenario-centos-9-stream https://zuul.opendev.org/t/openstack/build/b3ba41901974415bae16a13ac9486713 : TIMED_OUT in 2h 35m 04s (non-voting)","accounts_in_message":[],"_revision_number":6}],"current_revision_number":6,"current_revision":"35a7ba3a79e3692c03235aab0455dc44515a825c","revisions":{"bd760c073d7501ab77951d5324ffb163a0da50b2":{"kind":"REWORK","_number":1,"created":"2023-08-22 08:55:56.000000000","uploader":{"_account_id":11290,"name":"Gaudenz Steinlin","email":"gaudenz.steinlin@cloudscale.ch","username":"gaudenz"},"ref":"refs/changes/81/892281/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/81/892281/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/81/892281/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/81/892281/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/81/892281/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/81/892281/1"}}},"commit":{"parents":[{"commit":"2e563016ce77617e5416f27ff5800b8821561c25","subject":"Merge \"Remove blockdiag from Octavia\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/2e563016ce77617e5416f27ff5800b8821561c25"}]}],"author":{"name":"Gaudenz Steinlin","email":"gaudenz.steinlin@cloudscale.ch","date":"2023-08-22 07:59:00.000000000","tz":120},"committer":{"name":"Gaudenz Steinlin","email":"gaudenz.steinlin@cloudscale.ch","date":"2023-08-22 07:59:00.000000000","tz":120},"subject":"Raise NotFound on update of inexistent resources","message":"Raise NotFound on update of inexistent resources\n\nSome DB repository classes already raised NotFound on update of an\ninexistent resources. Others just returned, but did not update anything.\nThis commit unifies the behavior by raising NotFound in all update\nmethods if no resource is found.\n\nThe missing check caused issues when two requests concurrently deleted\nthe same object. We encountered this for HealthMonitor resources, but\nother resources are probably affected by the same issue.\n\n1. The first request is received and the LB locked (\u003dPENDING_UPDATE)\n   and the HealthMonitor set to PENDING_DELETE. A 204 NO CONTENT\n   response is sent to the client.\n\n2. The second request is received while the health monitor still\n   exists. Thus no immediate 404 NOT FOUND response is sent.\n\n3. The health monitor deletion job finishes and the LB is unlocked\n   (\u003dACTIVE).\n\n4. The second request proceeds and is able to lock the LB\n   (\u003dPENDING_UPDATE). The health monitor repository update also\n   succeeds, but does not update any object. A 204 NO CONTENT\n   response is sent to the client.\n\n5. The deletion job issued by the second request fails because the\n   health monitor no longer exists in the database. Because this\n   happens during setup of the deletion taskflow in the worker, no\n   flow is started and the LB is stuck in the PENDING_UPDATE state.\n\nWith this change, the second health monitor DELETE request fails in step\n4 and returns NotFound.\n\nChange-Id: I4f858a71784ae640bb6b5eddf9347c7b70afa632\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/bd760c073d7501ab77951d5324ffb163a0da50b2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/bd760c073d7501ab77951d5324ffb163a0da50b2"}]},"branch":"refs/heads/master"},"594f6e0667592f833355d398e0055591421e07a9":{"kind":"REWORK","_number":2,"created":"2023-08-23 07:00:56.000000000","uploader":{"_account_id":11290,"name":"Gaudenz Steinlin","email":"gaudenz.steinlin@cloudscale.ch","username":"gaudenz"},"ref":"refs/changes/81/892281/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/81/892281/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/81/892281/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/81/892281/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/81/892281/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/81/892281/2"}}},"commit":{"parents":[{"commit":"9e22faba65a6a3556f61a2b33dd20d5f3eedfb1f","subject":"Merge \"Add job to test with SQLAlchemy master (2.x)\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/9e22faba65a6a3556f61a2b33dd20d5f3eedfb1f"}]}],"author":{"name":"Gaudenz Steinlin","email":"gaudenz.steinlin@cloudscale.ch","date":"2023-08-22 07:59:00.000000000","tz":120},"committer":{"name":"Gaudenz Steinlin","email":"gaudenz.steinlin@cloudscale.ch","date":"2023-08-23 06:49:30.000000000","tz":120},"subject":"Raise NotFound on update of inexistent resources","message":"Raise NotFound on update of inexistent resources\n\nSome DB repository classes already raised NotFound on update of an\ninexistent resources. Others just returned, but did not update anything.\nThis commit unifies the behavior by raising NotFound in all update\nmethods if no resource is found.\n\nThe missing check caused issues when two requests concurrently deleted\nthe same object. We encountered this for HealthMonitor resources, but\nother resources are probably affected by the same issue.\n\n1. The first request is received and the LB locked (\u003dPENDING_UPDATE)\n   and the HealthMonitor set to PENDING_DELETE. A 204 NO CONTENT\n   response is sent to the client.\n\n2. The second request is received while the health monitor still\n   exists. Thus no immediate 404 NOT FOUND response is sent.\n\n3. The health monitor deletion job finishes and the LB is unlocked\n   (\u003dACTIVE).\n\n4. The second request proceeds and is able to lock the LB\n   (\u003dPENDING_UPDATE). The health monitor repository update also\n   succeeds, but does not update any object. A 204 NO CONTENT\n   response is sent to the client.\n\n5. The deletion job issued by the second request fails because the\n   health monitor no longer exists in the database. Because this\n   happens during setup of the deletion taskflow in the worker, no\n   flow is started and the LB is stuck in the PENDING_UPDATE state.\n\nWith this change, the second health monitor DELETE request fails in step\n4 and returns NotFound.\n\nChange-Id: I4f858a71784ae640bb6b5eddf9347c7b70afa632\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/594f6e0667592f833355d398e0055591421e07a9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/594f6e0667592f833355d398e0055591421e07a9"}]},"branch":"refs/heads/master"},"b7f6866e69db10be746e73b2407e8d03b63a622e":{"kind":"REWORK","_number":3,"created":"2023-08-23 14:18:31.000000000","uploader":{"_account_id":11290,"name":"Gaudenz Steinlin","email":"gaudenz.steinlin@cloudscale.ch","username":"gaudenz"},"ref":"refs/changes/81/892281/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/81/892281/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/81/892281/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/81/892281/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/81/892281/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/81/892281/3"}}},"commit":{"parents":[{"commit":"9e22faba65a6a3556f61a2b33dd20d5f3eedfb1f","subject":"Merge \"Add job to test with SQLAlchemy master (2.x)\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/9e22faba65a6a3556f61a2b33dd20d5f3eedfb1f"}]}],"author":{"name":"Gaudenz Steinlin","email":"gaudenz.steinlin@cloudscale.ch","date":"2023-08-22 07:59:00.000000000","tz":120},"committer":{"name":"Gaudenz Steinlin","email":"gaudenz.steinlin@cloudscale.ch","date":"2023-08-23 14:10:56.000000000","tz":120},"subject":"Raise NotFound on update of inexistent resources","message":"Raise NotFound on update of inexistent resources\n\nSome DB repository classes already raised NotFound on update of an\ninexistent resources. Others just returned, but did not update anything.\nThis commit unifies the behavior by raising NotFound in all update\nmethods if no resource is found.\n\nThe missing check caused issues when two requests concurrently deleted\nthe same object. We encountered this for HealthMonitor resources, but\nother resources are probably affected by the same issue.\n\n1. The first request is received and the LB locked (\u003dPENDING_UPDATE)\n   and the HealthMonitor set to PENDING_DELETE. A 204 NO CONTENT\n   response is sent to the client.\n\n2. The second request is received while the health monitor still\n   exists. Thus no immediate 404 NOT FOUND response is sent.\n\n3. The health monitor deletion job finishes and the LB is unlocked\n   (\u003dACTIVE).\n\n4. The second request proceeds and is able to lock the LB\n   (\u003dPENDING_UPDATE). The health monitor repository update also\n   succeeds, but does not update any object. A 204 NO CONTENT\n   response is sent to the client.\n\n5. The deletion job issued by the second request fails because the\n   health monitor no longer exists in the database. Because this\n   happens during setup of the deletion taskflow in the worker, no\n   flow is started and the LB is stuck in the PENDING_UPDATE state.\n\nWith this change, the second health monitor DELETE request fails in step\n4 and returns NotFound.\n\nThis also fixes a hidden bug in Amphora health record update where a\nexception was expected which never occurred.\n\nChange-Id: I4f858a71784ae640bb6b5eddf9347c7b70afa632\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/b7f6866e69db10be746e73b2407e8d03b63a622e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/b7f6866e69db10be746e73b2407e8d03b63a622e"}]},"branch":"refs/heads/master"},"0f0bdd0676d7d3cf67b6efb959fc30ba8e86b05b":{"kind":"TRIVIAL_REBASE","_number":4,"created":"2023-12-17 20:36:47.000000000","uploader":{"_account_id":11290,"name":"Gaudenz Steinlin","email":"gaudenz.steinlin@cloudscale.ch","username":"gaudenz"},"ref":"refs/changes/81/892281/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/81/892281/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/81/892281/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/81/892281/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/81/892281/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/81/892281/4"}}},"commit":{"parents":[{"commit":"261ee7dcffb1fe634152c4f2ccf79264c1a653b5","subject":"Merge \"Add spec for SR-IOV in Octavia Amphora\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/261ee7dcffb1fe634152c4f2ccf79264c1a653b5"}]}],"author":{"name":"Gaudenz Steinlin","email":"gaudenz.steinlin@cloudscale.ch","date":"2023-08-22 07:59:00.000000000","tz":120},"committer":{"name":"Gaudenz Steinlin","email":"gaudenz.steinlin@cloudscale.ch","date":"2023-12-17 20:36:18.000000000","tz":60},"subject":"Raise NotFound on update of inexistent resources","message":"Raise NotFound on update of inexistent resources\n\nSome DB repository classes already raised NotFound on update of an\ninexistent resources. Others just returned, but did not update anything.\nThis commit unifies the behavior by raising NotFound in all update\nmethods if no resource is found.\n\nThe missing check caused issues when two requests concurrently deleted\nthe same object. We encountered this for HealthMonitor resources, but\nother resources are probably affected by the same issue.\n\n1. The first request is received and the LB locked (\u003dPENDING_UPDATE)\n   and the HealthMonitor set to PENDING_DELETE. A 204 NO CONTENT\n   response is sent to the client.\n\n2. The second request is received while the health monitor still\n   exists. Thus no immediate 404 NOT FOUND response is sent.\n\n3. The health monitor deletion job finishes and the LB is unlocked\n   (\u003dACTIVE).\n\n4. The second request proceeds and is able to lock the LB\n   (\u003dPENDING_UPDATE). The health monitor repository update also\n   succeeds, but does not update any object. A 204 NO CONTENT\n   response is sent to the client.\n\n5. The deletion job issued by the second request fails because the\n   health monitor no longer exists in the database. Because this\n   happens during setup of the deletion taskflow in the worker, no\n   flow is started and the LB is stuck in the PENDING_UPDATE state.\n\nWith this change, the second health monitor DELETE request fails in step\n4 and returns NotFound.\n\nThis also fixes a hidden bug in Amphora health record update where a\nexception was expected which never occurred.\n\nChange-Id: I4f858a71784ae640bb6b5eddf9347c7b70afa632\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/0f0bdd0676d7d3cf67b6efb959fc30ba8e86b05b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/0f0bdd0676d7d3cf67b6efb959fc30ba8e86b05b"}]},"branch":"refs/heads/master"},"cac7a2a2fcb31a0113e468b772e128a095084df8":{"kind":"REWORK","_number":5,"created":"2023-12-17 21:31:11.000000000","uploader":{"_account_id":11290,"name":"Gaudenz Steinlin","email":"gaudenz.steinlin@cloudscale.ch","username":"gaudenz"},"ref":"refs/changes/81/892281/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/81/892281/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/81/892281/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/81/892281/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/81/892281/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/81/892281/5"}}},"commit":{"parents":[{"commit":"d7b2e5f665dc5a56dc44ee454aaf9febee62c6c9","subject":"Lock SELECTed DB rows when updating","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/d7b2e5f665dc5a56dc44ee454aaf9febee62c6c9"}]}],"author":{"name":"Gaudenz Steinlin","email":"gaudenz.steinlin@cloudscale.ch","date":"2023-08-22 07:59:00.000000000","tz":120},"committer":{"name":"Gaudenz Steinlin","email":"gaudenz.steinlin@cloudscale.ch","date":"2023-12-17 21:29:59.000000000","tz":60},"subject":"Raise NotFound on update of inexistent resources","message":"Raise NotFound on update of inexistent resources\n\nSome DB repository classes already raised NotFound on update of an\ninexistent resources. Others just returned, but did not update anything.\nThis commit unifies the behavior by raising NotFound in all update\nmethods if no resource is found.\n\nThe missing check caused issues when two requests concurrently deleted\nthe same object. We encountered this for HealthMonitor resources, but\nother resources are probably affected by the same issue.\n\n1. The first request is received and the LB locked (\u003dPENDING_UPDATE)\n   and the HealthMonitor set to PENDING_DELETE. A 204 NO CONTENT\n   response is sent to the client.\n\n2. The second request is received while the health monitor still\n   exists. Thus no immediate 404 NOT FOUND response is sent.\n\n3. The health monitor deletion job finishes and the LB is unlocked\n   (\u003dACTIVE).\n\n4. The second request proceeds and is able to lock the LB\n   (\u003dPENDING_UPDATE). The health monitor repository update also\n   succeeds, but does not update any object. A 204 NO CONTENT\n   response is sent to the client.\n\n5. The deletion job issued by the second request fails because the\n   health monitor no longer exists in the database. Because this\n   happens during setup of the deletion taskflow in the worker, no\n   flow is started and the LB is stuck in the PENDING_UPDATE state.\n\nWith this change, the second health monitor DELETE request fails in step\n4 and returns NotFound.\n\nThis also fixes a hidden bug in Amphora health record update where a\nexception was expected which never occurred.\n\nChange-Id: I4f858a71784ae640bb6b5eddf9347c7b70afa632\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/cac7a2a2fcb31a0113e468b772e128a095084df8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/cac7a2a2fcb31a0113e468b772e128a095084df8"}]},"branch":"refs/heads/master"},"35a7ba3a79e3692c03235aab0455dc44515a825c":{"kind":"TRIVIAL_REBASE","_number":6,"created":"2023-12-20 08:08:27.000000000","uploader":{"_account_id":11290,"name":"Gaudenz Steinlin","email":"gaudenz.steinlin@cloudscale.ch","username":"gaudenz"},"ref":"refs/changes/81/892281/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/81/892281/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/81/892281/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/81/892281/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/81/892281/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/81/892281/6"}}},"commit":{"parents":[{"commit":"5381c5bf61879123fbdb1fccfc5e77df23121093","subject":"Lock SELECTed DB rows when updating","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/5381c5bf61879123fbdb1fccfc5e77df23121093"}]}],"author":{"name":"Gaudenz Steinlin","email":"gaudenz.steinlin@cloudscale.ch","date":"2023-08-22 07:59:00.000000000","tz":120},"committer":{"name":"Gaudenz Steinlin","email":"gaudenz.steinlin@cloudscale.ch","date":"2023-12-20 08:08:19.000000000","tz":60},"subject":"Raise NotFound on update of inexistent resources","message":"Raise NotFound on update of inexistent resources\n\nSome DB repository classes already raised NotFound on update of an\ninexistent resources. Others just returned, but did not update anything.\nThis commit unifies the behavior by raising NotFound in all update\nmethods if no resource is found.\n\nThe missing check caused issues when two requests concurrently deleted\nthe same object. We encountered this for HealthMonitor resources, but\nother resources are probably affected by the same issue.\n\n1. The first request is received and the LB locked (\u003dPENDING_UPDATE)\n   and the HealthMonitor set to PENDING_DELETE. A 204 NO CONTENT\n   response is sent to the client.\n\n2. The second request is received while the health monitor still\n   exists. Thus no immediate 404 NOT FOUND response is sent.\n\n3. The health monitor deletion job finishes and the LB is unlocked\n   (\u003dACTIVE).\n\n4. The second request proceeds and is able to lock the LB\n   (\u003dPENDING_UPDATE). The health monitor repository update also\n   succeeds, but does not update any object. A 204 NO CONTENT\n   response is sent to the client.\n\n5. The deletion job issued by the second request fails because the\n   health monitor no longer exists in the database. Because this\n   happens during setup of the deletion taskflow in the worker, no\n   flow is started and the LB is stuck in the PENDING_UPDATE state.\n\nWith this change, the second health monitor DELETE request fails in step\n4 and returns NotFound.\n\nThis also fixes a hidden bug in Amphora health record update where a\nexception was expected which never occurred.\n\nChange-Id: I4f858a71784ae640bb6b5eddf9347c7b70afa632\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/35a7ba3a79e3692c03235aab0455dc44515a825c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/35a7ba3a79e3692c03235aab0455dc44515a825c"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"OK","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY"},{"label":"Workflow","status":"MAY"},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
