)]}'
{"id":"openstack%2Foctavia~892460","triplet_id":"openstack%2Foctavia~master~I39f0a89dbb3df468c561abea750fe8b5bc079d87","project":"openstack/octavia","branch":"master","topic":"fix-concurrent-delete","attention_set":{"11290":{"account":{"_account_id":11290,"name":"Gaudenz Steinlin","email":"gaudenz.steinlin@cloudscale.ch","username":"gaudenz"},"last_update":"2023-12-17 23:27:30.000000000","reason":"A robot voted negatively on a label"}},"removed_from_attention_set":{},"hashtags":[],"change_id":"I39f0a89dbb3df468c561abea750fe8b5bc079d87","subject":"Lock SELECTed DB rows when updating","status":"NEW","created":"2023-08-23 07:23:15.000000000","updated":"2024-01-09 09:21:40.000000000","submit_type":"MERGE_IF_NECESSARY","mergeable":false,"submittable":false,"total_comment_count":2,"unresolved_comment_count":1,"has_review_started":true,"meta_rev_id":"4a783bf06ca54457fc57a95ab8811ff2a94d0cb4","_number":892460,"virtual_id_number":892460,"owner":{"_account_id":11290,"name":"Gaudenz Steinlin","email":"gaudenz.steinlin@cloudscale.ch","username":"gaudenz"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":1,"date":"2023-12-20 10:49:09.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"disliked":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":-1,"date":"2024-01-09 09:21:40.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":-1,"default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-08-23 10:28:22.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-01-09 09:21:40.000000000","updated_by":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"reviewer":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"state":"REVIEWER"}],"messages":[{"id":"e5d6da1ed2ecdaeaf29471fed5a6ed45f72e4a5c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11290,"name":"Gaudenz Steinlin","email":"gaudenz.steinlin@cloudscale.ch","username":"gaudenz"},"date":"2023-08-23 07:23:15.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"399f66a17329d319ceb1a508c79c94f51ba83ef6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-23 10:28:22.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/013305385e814de0add42c038cb078cf\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/c0391cbb1812430fbb3b1ada5e80358a : SUCCESS in 10m 12s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/657ad92cb7874bfeac79fe168cf8a973 : SUCCESS in 5m 13s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/52af4917bf1a431586f13d8191461d3d : SUCCESS in 5m 18s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/4ba91cf5e4f049df9d6367ccfdc5655e : SUCCESS in 5m 19s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/cd3e2b14810e46acbf8f2ddab85fbc62 : SUCCESS in 5m 38s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/858a4f3c3b9744979994ea8a695e6d46 : SUCCESS in 6m 50s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/06a658e4803c49a29ca8accf9c0d41da : SUCCESS in 11m 24s\n- octavia-tox-py39-tips https://zuul.opendev.org/t/openstack/build/a83989944b0c477b9b3179cb58b26009 : SUCCESS in 5m 49s\n- octavia-tox-functional-py39-tips https://zuul.opendev.org/t/openstack/build/d4d4cb400a374318bd84d36647229414 : SUCCESS in 8m 44s\n- octavia-tox-functional-py39-sqlalchemy-tips https://zuul.opendev.org/t/openstack/build/c4d8ca0eb24949a89cbc169ccefa43bb : SUCCESS in 10m 09s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/3f6b257c4bba443c8d08b0ba038122e8 : SUCCESS in 5m 06s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/7dc8dabbfbde47f0b798118cee40ef95 : SUCCESS in 6m 35s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/53fdd3d38f9940b0bd28c6e0c2cd53a0 : SUCCESS in 8m 59s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/ddacc7f0bbdd4cf29b95ada4e9b8e736 : SUCCESS in 8m 17s (non-voting)\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/9c7c3f95643e4a6597e403e007fac47b : SUCCESS in 1h 29m 29s\n- octavia-v2-dsvm-scenario-traffic-ops https://zuul.opendev.org/t/openstack/build/c768b00e0833408d96c9c0718539c36c : SUCCESS in 1h 56m 13s\n- octavia-v2-dsvm-scenario-non-traffic-ops https://zuul.opendev.org/t/openstack/build/2292ea110765448f907bbf31a7136577 : SUCCESS in 1h 16m 17s\n- octavia-v2-dsvm-scenario-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/7cd9eb2f01124ae2bd37db75e5b07674 : SUCCESS in 2h 16m 31s\n- octavia-v2-dsvm-scenario-non-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/ac6dcd2366f14ad095e651a30e88b8d0 : SUCCESS in 1h 25m 36s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/e35e86b1a0e24e40895c460815cf9bb3 : SUCCESS in 44m 59s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/5374c525c7e740a997ac386640e7d850 : SUCCESS in 47m 59s\n- octavia-v2-dsvm-tls-barbican-fips https://zuul.opendev.org/t/openstack/build/e4b86958160c4269bff359323e4f75cd : FAILURE in 20m 22s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/a98328958b764daa95b4429f4f25a03b : SUCCESS in 42m 29s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/bf7828beda0a45aeaaee8160d139e13d : SUCCESS in 2h 05m 19s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/e1014843e6bc452cbe26868ee6e17a63 : TIMED_OUT in 3h 02m 54s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/34bbb58b232b440692df8a19dbc0bd16 : SUCCESS in 1h 48m 36s (non-voting)\n- octavia-v2-dsvm-scenario-centos-9-stream https://zuul.opendev.org/t/openstack/build/524ca114ba044c61857da0101a143bbf : FAILURE in 21m 06s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"7e101e3bf4a7b9e68b8e2f640642a611fb5b14f8","author":{"_account_id":11290,"name":"Gaudenz Steinlin","email":"gaudenz.steinlin@cloudscale.ch","username":"gaudenz"},"date":"2023-10-31 14:40:34.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"93663d34187325fa9daa147eef1c40e1eeb8268c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11290,"name":"Gaudenz Steinlin","email":"gaudenz.steinlin@cloudscale.ch","username":"gaudenz"},"date":"2023-12-17 20:39:03.000000000","message":"Uploaded patch set 2: Patch Set 1 was rebased.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":2},{"id":"67e497047e6c30b88428b3308787017ec06b8e70","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-12-17 23:27:30.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/7bdc487a7dac4edebc3b28d3e49fcb37\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/ee86e2560ed04522bed9d2ca39a5f2df : SUCCESS in 10m 15s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b18e9d50600a4e539b719d6bf06106ae : FAILURE in 3m 54s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/203aec7913934d2f8a2fdd36ce557940 : SUCCESS in 6m 58s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/7227b25863fd4e33a464e7a1e5ef3528 : SUCCESS in 6m 26s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/dc072dc9f9384748be8c35782d3e3382 : SUCCESS in 12m 59s\n- octavia-tox-py39-tips https://zuul.opendev.org/t/openstack/build/e6c8106e51394c95a1227c5bf63d442f : SUCCESS in 6m 01s\n- octavia-tox-functional-py39-tips https://zuul.opendev.org/t/openstack/build/d90838f3755d460e99469d3891c51037 : SUCCESS in 7m 55s\n- octavia-tox-functional-py39-sqlalchemy-tips https://zuul.opendev.org/t/openstack/build/99be9eeadbf74cbf9f39341977963f79 : SUCCESS in 9m 53s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/6915cf5b646c41eb839d65cd82d6a6e5 : SUCCESS in 5m 46s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/ed19c89b2d9f4e8d88ab20c3f070d2a0 : SUCCESS in 8m 02s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/333318ecd9994395b144a8484e19a0c7 : SUCCESS in 8m 56s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/bff8b499a6a1456c9fe084443037ebea : SUCCESS in 10m 12s (non-voting)\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/f79812bd25264cf1a505d78c318482ae : SUCCESS in 1h 19m 27s\n- octavia-v2-dsvm-scenario-traffic-ops https://zuul.opendev.org/t/openstack/build/0549eea99a2a48f19422b6c998016da6 : SUCCESS in 1h 36m 27s\n- octavia-v2-dsvm-scenario-non-traffic-ops https://zuul.opendev.org/t/openstack/build/551f1038c6504846bd6c7976945ff496 : SUCCESS in 1h 06m 25s\n- octavia-v2-dsvm-scenario-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/47d6af61bec746928b4dc73334ee9dab : SUCCESS in 1h 38m 48s\n- octavia-v2-dsvm-scenario-non-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/422e769be2de4022be81d5a05fc37c1e : SUCCESS in 1h 03m 32s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/2d3858cee713406385634e6b1cdb5cf3 : SUCCESS in 37m 43s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/bd9af546764a4e00a4db7edfb8c73c23 : SUCCESS in 40m 08s\n- octavia-grenade-skip-level https://zuul.opendev.org/t/openstack/build/3f546d20575f4d339223bcde06097bf9 : SUCCESS in 38m 33s (non-voting)\n- octavia-v2-dsvm-tls-barbican-fips https://zuul.opendev.org/t/openstack/build/c510dba23f634ce9af8534ce05e1b725 : SUCCESS in 1h 21m 28s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/7c9da81ae1604cc89698683fc6830f75 : SUCCESS in 33m 45s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/980c92fee6d04d9d9ee207ff47defd73 : SUCCESS in 2h 03m 33s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/6db0fef1cbd54627b1a1bcd30365dcdc : SUCCESS in 2h 46m 00s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/80224003afac4a5398bf55ac130c7abb : SUCCESS in 2h 09m 48s (non-voting)\n- octavia-v2-dsvm-scenario-centos-9-stream https://zuul.opendev.org/t/openstack/build/a025e572fb274e8697c902d2794016ea : TIMED_OUT in 2h 35m 55s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"a4ec1c71706c5165877bed4f9c61c1699e4ce703","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11290,"name":"Gaudenz Steinlin","email":"gaudenz.steinlin@cloudscale.ch","username":"gaudenz"},"date":"2023-12-20 08:07:59.000000000","message":"Uploaded patch set 3: Patch Set 2 was rebased.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":3},{"id":"f88f0ab2305723c55e96f8b24f8e541ab306ba7e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11290,"name":"Gaudenz Steinlin","email":"gaudenz.steinlin@cloudscale.ch","username":"gaudenz"},"date":"2023-12-20 08:08:27.000000000","message":"Uploaded patch set 4: New patch set was added with same tree, parent tree, and commit message as Patch Set 3.","accounts_in_message":[],"_revision_number":4},{"id":"60d4cf01ad97476e4dbf296eb2323c3f54f3a106","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-12-20 10:49:09.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/14c7695e2a7441ee913f78f1b0687460\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/5a67998f79194971a17048826f1647bd : SUCCESS in 10m 17s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/02790a0026394f559ecc35c14c356a4f : SUCCESS in 5m 33s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/4260f4cf91504cea8292407569692e11 : SUCCESS in 5m 09s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/dc69837b48b14e7a9efaabb6c88e268d : SUCCESS in 5m 42s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fc76b55939e643dea3a2aea45713097d : SUCCESS in 12m 54s\n- octavia-tox-py39-tips https://zuul.opendev.org/t/openstack/build/4ca36340df3b4dd3b598c09e58db758e : SUCCESS in 6m 20s\n- octavia-tox-functional-py39-tips https://zuul.opendev.org/t/openstack/build/25bc03006c3c46958ef6d261b90f74ea : SUCCESS in 7m 32s\n- octavia-tox-functional-py39-sqlalchemy-tips https://zuul.opendev.org/t/openstack/build/5a5c650e67f8410589f207ae6141fa97 : SUCCESS in 11m 21s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/6dcb5a1b170441bf974cfd84a882188a : SUCCESS in 4m 52s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/a2caa842dd5f4748967a38263222635d : SUCCESS in 6m 38s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/da2a28b2e99940bf9d3a75e8739deb88 : SUCCESS in 8m 59s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/6868743ed9594bbfb9b94561d17af7a0 : SUCCESS in 8m 50s (non-voting)\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/61e56a3ed6d14086a3d509fe082e14d8 : SUCCESS in 1h 42m 17s\n- octavia-v2-dsvm-scenario-traffic-ops https://zuul.opendev.org/t/openstack/build/c834493e08d945c580ca3087ae5ad50c : SUCCESS in 1h 41m 02s\n- octavia-v2-dsvm-scenario-non-traffic-ops https://zuul.opendev.org/t/openstack/build/332527469f5141b5a56858a22c53975b : SUCCESS in 59m 29s\n- octavia-v2-dsvm-scenario-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/bffb317839eb48c281720a445cce09d7 : SUCCESS in 1h 41m 16s\n- octavia-v2-dsvm-scenario-non-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/738935cd4b3a4b518d5c64a7bfb69fd0 : SUCCESS in 54m 24s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/126f1ac58ef34f44bf3f487bc5f9c8b6 : SUCCESS in 37m 32s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/40773a9ede3b407e89fcfa5da9794221 : SUCCESS in 38m 09s\n- octavia-grenade-skip-level https://zuul.opendev.org/t/openstack/build/a99fd13065da4b098cbfd2f81df0bf32 : SUCCESS in 38m 05s (non-voting)\n- octavia-v2-dsvm-tls-barbican-fips https://zuul.opendev.org/t/openstack/build/9255a80fe7e942ef8d87209ea27cfd00 : SUCCESS in 1h 15m 38s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/b5913cf9a5ba4ffbbc7ac2f838127d27 : SUCCESS in 34m 28s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/d62b89d11e364f8dbb668645895fdcfd : SUCCESS in 2h 01m 33s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/778a70e26c1f482986ff5324c351bc39 : SUCCESS in 2h 22m 25s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/9e55368424414bdc8c71bb903e7f6125 : SUCCESS in 2h 06m 53s (non-voting)\n- octavia-v2-dsvm-scenario-centos-9-stream https://zuul.opendev.org/t/openstack/build/b6d502dfc345481ab68b4e9c96dc455b : TIMED_OUT in 2h 35m 00s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"4a783bf06ca54457fc57a95ab8811ff2a94d0cb4","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2024-01-09 09:21:40.000000000","message":"Patch Set 4: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"5381c5bf61879123fbdb1fccfc5e77df23121093","revisions":{"9e9f4fed08d7d9ac081e2632d5277ea297eff26b":{"kind":"REWORK","_number":1,"created":"2023-08-23 07:23:15.000000000","uploader":{"_account_id":11290,"name":"Gaudenz Steinlin","email":"gaudenz.steinlin@cloudscale.ch","username":"gaudenz"},"ref":"refs/changes/60/892460/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/60/892460/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/60/892460/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/60/892460/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/60/892460/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/60/892460/1"}}},"commit":{"parents":[{"commit":"9e22faba65a6a3556f61a2b33dd20d5f3eedfb1f","subject":"Merge \"Add job to test with SQLAlchemy master (2.x)\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/9e22faba65a6a3556f61a2b33dd20d5f3eedfb1f"}]}],"author":{"name":"Gaudenz Steinlin","email":"gaudenz.steinlin@cloudscale.ch","date":"2023-08-22 15:33:51.000000000","tz":120},"committer":{"name":"Gaudenz Steinlin","email":"gaudenz.steinlin@cloudscale.ch","date":"2023-08-23 07:12:07.000000000","tz":120},"subject":"Lock SELECTed DB rows when updating","message":"Lock SELECTed DB rows when updating\n\nThe use of with_for_update() when querying the DB for an update was\nonly applied for some updates and not for others. This commit adds\npopulate_existing() and with_for_update() to all places found where a\nquery result is later used in an update.\n\nThis fixes a concurrency issue when deleting the same pool member with\n2 requests at the same time. It probably also fixes other similar\nconcurrency issues.\n\nChange-Id: I39f0a89dbb3df468c561abea750fe8b5bc079d87\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/9e9f4fed08d7d9ac081e2632d5277ea297eff26b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/9e9f4fed08d7d9ac081e2632d5277ea297eff26b"}]},"branch":"refs/heads/master"},"d7b2e5f665dc5a56dc44ee454aaf9febee62c6c9":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2023-12-17 20:39:03.000000000","uploader":{"_account_id":11290,"name":"Gaudenz Steinlin","email":"gaudenz.steinlin@cloudscale.ch","username":"gaudenz"},"ref":"refs/changes/60/892460/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/60/892460/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/60/892460/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/60/892460/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/60/892460/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/60/892460/2"}}},"commit":{"parents":[{"commit":"261ee7dcffb1fe634152c4f2ccf79264c1a653b5","subject":"Merge \"Add spec for SR-IOV in Octavia Amphora\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/261ee7dcffb1fe634152c4f2ccf79264c1a653b5"}]}],"author":{"name":"Gaudenz Steinlin","email":"gaudenz.steinlin@cloudscale.ch","date":"2023-08-22 15:33:51.000000000","tz":120},"committer":{"name":"Gaudenz Steinlin","email":"gaudenz.steinlin@cloudscale.ch","date":"2023-12-17 20:38:55.000000000","tz":60},"subject":"Lock SELECTed DB rows when updating","message":"Lock SELECTed DB rows when updating\n\nThe use of with_for_update() when querying the DB for an update was\nonly applied for some updates and not for others. This commit adds\npopulate_existing() and with_for_update() to all places found where a\nquery result is later used in an update.\n\nThis fixes a concurrency issue when deleting the same pool member with\n2 requests at the same time. It probably also fixes other similar\nconcurrency issues.\n\nChange-Id: I39f0a89dbb3df468c561abea750fe8b5bc079d87\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/d7b2e5f665dc5a56dc44ee454aaf9febee62c6c9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/d7b2e5f665dc5a56dc44ee454aaf9febee62c6c9"}]},"branch":"refs/heads/master"},"485a5601306f9c33da757a6ca34fc72884989f6a":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2023-12-20 08:07:59.000000000","uploader":{"_account_id":11290,"name":"Gaudenz Steinlin","email":"gaudenz.steinlin@cloudscale.ch","username":"gaudenz"},"ref":"refs/changes/60/892460/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/60/892460/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/60/892460/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/60/892460/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/60/892460/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/60/892460/3"}}},"commit":{"parents":[{"commit":"240b0066a12fc81c9e42aade67435cbb0e65c2a5","subject":"Merge \"Use cryptography to load PKCS12 certificates\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/240b0066a12fc81c9e42aade67435cbb0e65c2a5"}]}],"author":{"name":"Gaudenz Steinlin","email":"gaudenz.steinlin@cloudscale.ch","date":"2023-08-22 15:33:51.000000000","tz":120},"committer":{"name":"Gaudenz Steinlin","email":"gaudenz.steinlin@cloudscale.ch","date":"2023-12-20 08:07:43.000000000","tz":60},"subject":"Lock SELECTed DB rows when updating","message":"Lock SELECTed DB rows when updating\n\nThe use of with_for_update() when querying the DB for an update was\nonly applied for some updates and not for others. This commit adds\npopulate_existing() and with_for_update() to all places found where a\nquery result is later used in an update.\n\nThis fixes a concurrency issue when deleting the same pool member with\n2 requests at the same time. It probably also fixes other similar\nconcurrency issues.\n\nChange-Id: I39f0a89dbb3df468c561abea750fe8b5bc079d87\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/485a5601306f9c33da757a6ca34fc72884989f6a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/485a5601306f9c33da757a6ca34fc72884989f6a"}]},"branch":"refs/heads/master"},"5381c5bf61879123fbdb1fccfc5e77df23121093":{"kind":"NO_CHANGE","_number":4,"created":"2023-12-20 08:08:27.000000000","uploader":{"_account_id":11290,"name":"Gaudenz Steinlin","email":"gaudenz.steinlin@cloudscale.ch","username":"gaudenz"},"ref":"refs/changes/60/892460/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/60/892460/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/60/892460/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/60/892460/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/60/892460/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/60/892460/4"}}},"commit":{"parents":[{"commit":"240b0066a12fc81c9e42aade67435cbb0e65c2a5","subject":"Merge \"Use cryptography to load PKCS12 certificates\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/240b0066a12fc81c9e42aade67435cbb0e65c2a5"}]}],"author":{"name":"Gaudenz Steinlin","email":"gaudenz.steinlin@cloudscale.ch","date":"2023-08-22 15:33:51.000000000","tz":120},"committer":{"name":"Gaudenz Steinlin","email":"gaudenz.steinlin@cloudscale.ch","date":"2023-12-20 08:08:19.000000000","tz":60},"subject":"Lock SELECTed DB rows when updating","message":"Lock SELECTed DB rows when updating\n\nThe use of with_for_update() when querying the DB for an update was\nonly applied for some updates and not for others. This commit adds\npopulate_existing() and with_for_update() to all places found where a\nquery result is later used in an update.\n\nThis fixes a concurrency issue when deleting the same pool member with\n2 requests at the same time. It probably also fixes other similar\nconcurrency issues.\n\nChange-Id: I39f0a89dbb3df468c561abea750fe8b5bc079d87\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/5381c5bf61879123fbdb1fccfc5e77df23121093"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/5381c5bf61879123fbdb1fccfc5e77df23121093"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"OK","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}},{"label":"Workflow","status":"MAY"},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
