)]}'
{"id":"openstack%2Foctavia~892679","triplet_id":"openstack%2Foctavia~master~I9309629aaf9f0e0860e4a28eddaa883262ea37c1","project":"openstack/octavia","branch":"master","hashtags":[],"change_id":"I9309629aaf9f0e0860e4a28eddaa883262ea37c1","subject":"WIP Remove device_id from the VIP port","status":"NEW","created":"2023-08-24 14:25:08.000000000","updated":"2023-08-24 17:30:20.000000000","submit_type":"MERGE_IF_NECESSARY","mergeable":false,"submittable":false,"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"4ca4901d9e582777385fe777cabcab96024acfaa","_number":892679,"virtual_id_number":892679,"owner":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":1,"date":"2023-08-24 17:30:20.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-08-24 17:30:20.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"6f3208a1ce2220fb3efb38a528ba595c93dd90f6","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2023-08-24 14:25:08.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"4ca4901d9e582777385fe777cabcab96024acfaa","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-24 17:30:20.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b40473e0cb4e40fc9ac7d68ddcd5db2a\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/f5617c13a0a14f24957779454afc1748 : SUCCESS in 11m 44s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a40487631bdf444aa14957bf9e6ec64e : SUCCESS in 5m 23s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/b1dfedf5ddd04572b8263a43a56c1dcf : SUCCESS in 5m 05s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/85889ec4965b4e2d9b027cdc2755684d : SUCCESS in 5m 08s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/345086398b5449b794fb55cc42ef8ecb : SUCCESS in 4m 53s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/1995ac661f2547b58d2f82f46ba31eb9 : SUCCESS in 5m 14s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/418fb9ce15554326a297663a1399c552 : SUCCESS in 11m 35s\n- octavia-tox-py39-tips https://zuul.opendev.org/t/openstack/build/61644e386adc4033afe2de641b54a2f6 : SUCCESS in 5m 25s\n- octavia-tox-functional-py39-tips https://zuul.opendev.org/t/openstack/build/913b6c63c0b74f4bb505b13a425da128 : SUCCESS in 8m 07s\n- octavia-tox-functional-py39-sqlalchemy-tips https://zuul.opendev.org/t/openstack/build/110c2ca41cd241afb22a6ed5d1a411b0 : SUCCESS in 9m 57s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/319ff9131b444da995be9df6fdca9cda : SUCCESS in 4m 45s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/2381c961e4af42ffb6d3e2e36b030a76 : SUCCESS in 7m 39s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/dadac9efc0ac47e584ac3a69969e6724 : SUCCESS in 7m 40s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/6f8e02daa5354aee8b5418dde82299fb : SUCCESS in 7m 44s (non-voting)\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/20dfbc00885a45cb85744dd6c8527467 : SUCCESS in 1h 26m 33s\n- octavia-v2-dsvm-scenario-traffic-ops https://zuul.opendev.org/t/openstack/build/8da9c200b8ba440eaab636e6eb00b740 : SUCCESS in 2h 14m 19s\n- octavia-v2-dsvm-scenario-non-traffic-ops https://zuul.opendev.org/t/openstack/build/f6208536432a4eaa9314623c4716297b : SUCCESS in 1h 07m 18s\n- octavia-v2-dsvm-scenario-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/624d9943183f492ea403fa21d0b20c17 : SUCCESS in 1h 59m 25s\n- octavia-v2-dsvm-scenario-non-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/76d1fcad236548b2a113b6c4de19f88c : SUCCESS in 1h 11m 27s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/1c7186509e2a41cca6863bb1eb3392d6 : SUCCESS in 47m 20s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/053f0c9d4e5e491ea8204f3631d43cfd : SUCCESS in 49m 26s\n- octavia-v2-dsvm-tls-barbican-fips https://zuul.opendev.org/t/openstack/build/06e79086ac0744aebc74d9ecda4b35ad : SUCCESS in 1h 06m 22s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/277fe1c432d94a8681dc4b38424cc56c : SUCCESS in 33m 46s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/f91264f7714441a1b29f1969c1d512ef : TIMED_OUT in 2h 33m 04s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/d8b8a3f54fb94eda9d58f93b626bdde3 : TIMED_OUT in 3h 03m 40s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/58ea1dc593d14aa981e25335e40cf636 : TIMED_OUT in 2h 33m 02s (non-voting)\n- octavia-v2-dsvm-scenario-centos-9-stream https://zuul.opendev.org/t/openstack/build/0d748657250c4c5d9d3dad58ee94a8ff : SUCCESS in 2h 11m 26s (non-voting)","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"9771ea8ce7b8c68c2412f1295a0019d5faf23565","revisions":{"9771ea8ce7b8c68c2412f1295a0019d5faf23565":{"kind":"REWORK","_number":1,"created":"2023-08-24 14:25:08.000000000","uploader":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"ref":"refs/changes/79/892679/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/79/892679/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/79/892679/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/79/892679/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/79/892679/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/79/892679/1"}}},"commit":{"parents":[{"commit":"a88aa18d6f2d5f787cfe1d74148a5a46b76cdb95","subject":"Merge \"Fix the issue of losing nf_conntrace* values after a reboot.\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/a88aa18d6f2d5f787cfe1d74148a5a46b76cdb95"}]}],"author":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2023-08-24 13:10:40.000000000","tz":-240},"committer":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2023-08-24 14:22:09.000000000","tz":-240},"subject":"WIP Remove device_id from the VIP port","message":"WIP Remove device_id from the VIP port\n\nChange-Id: I9309629aaf9f0e0860e4a28eddaa883262ea37c1\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/9771ea8ce7b8c68c2412f1295a0019d5faf23565"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/9771ea8ce7b8c68c2412f1295a0019d5faf23565"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"OK","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY"},{"label":"Workflow","status":"MAY"},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
