)]}'
{"id":"openstack%2Foctavia~893536","triplet_id":"openstack%2Foctavia~master~Ib33a0b8d2875e4ff97c65933fe9360bb06994d32","project":"openstack/octavia","branch":"master","attention_set":{},"removed_from_attention_set":{"27442":{"account":{"_account_id":27442,"name":"Quentin GROLLEAU","email":"quentin.grolleau@corp.ovh.com","username":"QG"},"last_update":"2023-10-09 15:35:27.000000000","reason":"Change was submitted"},"32761":{"account":{"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"},"last_update":"2023-10-09 15:35:27.000000000","reason":"Change was submitted"},"11628":{"account":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"last_update":"2023-10-09 15:35:27.000000000","reason":"Change was submitted"},"29244":{"account":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"last_update":"2023-10-09 15:35:27.000000000","reason":"Change was submitted"},"34429":{"account":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"last_update":"2023-10-09 11:46:21.000000000","reason":"\u003cGERRIT_ACCOUNT_34429\u003e replied on the change","reason_account":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}}},"hashtags":[],"change_id":"Ib33a0b8d2875e4ff97c65933fe9360bb06994d32","subject":"Reduce duration of failovers with amphora in ERROR","status":"MERGED","created":"2023-09-01 15:37:22.000000000","updated":"2023-10-09 15:36:45.000000000","submitted":"2023-10-09 15:35:27.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":3,"unresolved_comment_count":1,"has_review_started":true,"submission_id":"893536","meta_rev_id":"492d77ad496d1e0292666be95f94cde8eef476ca","_number":893536,"virtual_id_number":893536,"owner":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":27442,"name":"Quentin GROLLEAU","email":"quentin.grolleau@corp.ovh.com","username":"QG"},{"value":0,"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-10-09 15:35:26.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":0,"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"all":[{"value":0,"_account_id":27442,"name":"Quentin GROLLEAU","email":"quentin.grolleau@corp.ovh.com","username":"QG"},{"value":0,"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":2,"date":"2023-10-09 11:46:21.000000000","_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"all":[{"value":0,"_account_id":27442,"name":"Quentin GROLLEAU","email":"quentin.grolleau@corp.ovh.com","username":"QG"},{"value":0,"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":1,"date":"2023-10-09 11:46:21.000000000","_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":27442,"name":"Quentin GROLLEAU","email":"quentin.grolleau@corp.ovh.com","username":"QG"},{"value":0,"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":0,"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":27442,"name":"Quentin GROLLEAU","email":"quentin.grolleau@corp.ovh.com","username":"QG"},{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"},{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-09-01 18:29:53.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-09-14 16:16:16.000000000","updated_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"reviewer":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"state":"REVIEWER"},{"updated":"2023-09-25 06:51:46.000000000","updated_by":{"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"},"reviewer":{"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"},"state":"REVIEWER"},{"updated":"2023-09-25 13:13:07.000000000","updated_by":{"_account_id":27442,"name":"Quentin GROLLEAU","email":"quentin.grolleau@corp.ovh.com","username":"QG"},"reviewer":{"_account_id":27442,"name":"Quentin GROLLEAU","email":"quentin.grolleau@corp.ovh.com","username":"QG"},"state":"REVIEWER"},{"updated":"2023-10-09 11:14:23.000000000","updated_by":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"reviewer":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"state":"REVIEWER"}],"messages":[{"id":"132c697c0bdb65e4f56e2b5220a6056af3365f2a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2023-09-01 15:37:22.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"2e7289d5dc92c948e6fd529c785cd9e82c941672","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-01 18:29:53.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/afc4ed89540e4462ae50055a4bb99028\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/ce3c4f45aa15400aaf5e4d2f3b230928 : SUCCESS in 10m 07s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d1fd0e3ac1884565b5227edb3350a1cf : SUCCESS in 5m 34s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/d1ec36fb752f4efcafd81db7fe162504 : SUCCESS in 4m 59s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/120040a8cec74eb3b08a5d67750882a0 : SUCCESS in 4m 39s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/05003b50ec25421d9d4b7c956517fe96 : SUCCESS in 5m 49s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/dc0e4ed187754ad28582721b0c3d6410 : SUCCESS in 5m 50s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e029ee8b473f45299125b4c15ebefc3d : FAILURE in 6m 41s\n- octavia-tox-py39-tips https://zuul.opendev.org/t/openstack/build/5afd32a86d9c452095af52ee6778e6d2 : SUCCESS in 5m 12s\n- octavia-tox-functional-py39-tips https://zuul.opendev.org/t/openstack/build/887dd6c26afe40edb39061125e33b2f3 : SUCCESS in 7m 15s\n- octavia-tox-functional-py39-sqlalchemy-tips https://zuul.opendev.org/t/openstack/build/9d589d0f0f634d0fb74daba74efb8cde : SUCCESS in 10m 56s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/f74246e84ed243939985d9ee799b893a : SUCCESS in 5m 01s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/42b98ce919f34f3d8a7676c58f3343c0 : SUCCESS in 8m 10s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/b0a1158cbc6640089c92d297dc8bc4d1 : SUCCESS in 8m 06s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/ca1b8afaac4146428a35a460b41d8e8b : SUCCESS in 7m 50s (non-voting)\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/e0a98723715e47f3a8b27093d21c5459 : SUCCESS in 1h 28m 05s\n- octavia-v2-dsvm-scenario-traffic-ops https://zuul.opendev.org/t/openstack/build/1f69e32833fd4df4b249aea5d224eb56 : SUCCESS in 1h 53m 26s\n- octavia-v2-dsvm-scenario-non-traffic-ops https://zuul.opendev.org/t/openstack/build/caca3471f0e04ad798729d43378aeae5 : SUCCESS in 1h 12m 37s\n- octavia-v2-dsvm-scenario-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/3bb8f6879faa47b79851d470532e8220 : SUCCESS in 2h 02m 54s\n- octavia-v2-dsvm-scenario-non-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/ed09af9594a34e0997de2bfcb413d47a : SUCCESS in 1h 16m 54s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/ed94f646ed68410f9ece12ecc9fe9096 : SUCCESS in 39m 52s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/2e35af24ad2a4fb5b78a46b224502f67 : FAILURE in 38m 49s\n- octavia-v2-dsvm-tls-barbican-fips https://zuul.opendev.org/t/openstack/build/69b752fd6d7d4ab2947fa8759eb0fca2 : SUCCESS in 1h 02m 49s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/9537202fd301407e8e79fdd9076f3f26 : SUCCESS in 42m 35s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/45a79b7407ec4635ae3d732e34a421d1 : TIMED_OUT in 2h 32m 53s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/524cc68487874aab8f5e9cf4d7c5bf2f : FAILURE in 2h 50m 35s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/51ae463e7e5849ba83f13ba60a4ba0e4 : SUCCESS in 2h 15m 55s (non-voting)\n- octavia-v2-dsvm-scenario-centos-9-stream https://zuul.opendev.org/t/openstack/build/2311e03aff744bcea8a5bee682e249e0 : FAILURE in 2h 16m 47s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"bdfd85453fcb49370ef73aaf0fc18b6f920697d3","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2023-09-04 10:01:18.000000000","message":"Uploaded patch set 2: Patch Set 1 was rebased.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"a2e893ba88c636bc9ee6c55b4ecbe6d5a3dcab3b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2023-09-04 11:44:35.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"b3247c1bcdedded193d0255e88a5b6b4c9dc9809","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2023-09-04 12:04:18.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"5042bbadb6bf0e8c6adeabdf41b7cebed5126e39","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-04 14:56:55.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/1422d2c555fa49688462afde0af069ff\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/56768f2c4b3942638c8932305b259295 : SUCCESS in 12m 26s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2aae19109db4405096feb4323c8d34ac : SUCCESS in 5m 32s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/5cf1e859a06b4bfa9703047dc21fef7c : SUCCESS in 4m 15s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/c0d211c5cc524d93b0446b8bc4b52bd1 : SUCCESS in 5m 46s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/df6a2abf00ab442d8cbeb6ff38c8fe3e : SUCCESS in 5m 32s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/c4f1ffa1721d4b4bb2ead9e3c84ba6c4 : SUCCESS in 6m 05s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/19732e63b0434278a5122acde6443bce : SUCCESS in 11m 21s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/16fc5341f85a41dc95134357ede29cee : SUCCESS in 5m 07s\n- octavia-tox-py39-tips https://zuul.opendev.org/t/openstack/build/afcdbdb38ddc40e393bbd74d94eb29e2 : SUCCESS in 5m 59s\n- octavia-tox-functional-py39-tips https://zuul.opendev.org/t/openstack/build/3345ee99f75d4c718a7a693fd57fc6fd : SUCCESS in 7m 52s\n- octavia-tox-functional-py39-sqlalchemy-tips https://zuul.opendev.org/t/openstack/build/f287396c088e424fa4d751913691d9f7 : SUCCESS in 10m 44s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/11314da7cc3b41cc89f1c7040d5209a6 : SUCCESS in 4m 43s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/f89357d2c9314f8e8269f8a4d96631e6 : SUCCESS in 8m 57s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/1af8456729c0433691eedd20c6202b49 : SUCCESS in 9m 24s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/c84419d16fe44400a194bb94c4b948dd : SUCCESS in 10m 28s (non-voting)\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/0f6af78cb90a402c8b32afc5d3abcbd7 : SUCCESS in 1h 09m 33s\n- octavia-v2-dsvm-scenario-traffic-ops https://zuul.opendev.org/t/openstack/build/a437b6bea0d24b27830c15e829045020 : SUCCESS in 2h 05m 11s\n- octavia-v2-dsvm-scenario-non-traffic-ops https://zuul.opendev.org/t/openstack/build/682e8a24abb740d5b3ef2cd396d0d66e : SUCCESS in 1h 28m 36s\n- octavia-v2-dsvm-scenario-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/cf66ca3e5b1d42039620edc3eaa25852 : SUCCESS in 2h 12m 32s\n- octavia-v2-dsvm-scenario-non-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/04d145cfc0c245eaa365a8cde286231e : SUCCESS in 1h 33m 10s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/06a3c0bb38684dbb92f91087cb291a54 : SUCCESS in 40m 22s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/342e093111f34dc2b505d494329365bc : FAILURE in 40m 15s\n- octavia-v2-dsvm-tls-barbican-fips https://zuul.opendev.org/t/openstack/build/057ea17c88db4e6dad6b3641c40d54e6 : SUCCESS in 1h 06m 28s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/bb3b017242f9477c809adef3ad4f4479 : SUCCESS in 39m 53s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/fa3c5c5a4f684b7ea9e44eadc10c1413 : TIMED_OUT in 2h 34m 56s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/edf7807d8f2c43e5a9c72615760cfca6 : SUCCESS in 2h 49m 41s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/e1aa0ec0c04e43ffbeb58a39512ac79b : SUCCESS in 1h 56m 46s (non-voting)\n- octavia-v2-dsvm-scenario-centos-9-stream https://zuul.opendev.org/t/openstack/build/f8ee05fc54d1424cac6e5114ed82f310 : SUCCESS in 2h 18m 37s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"182ce6947c963a6ef20a53cf8ad4272761dd2402","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2023-09-06 14:39:25.000000000","message":"Patch Set 4: Workflow-1","accounts_in_message":[],"_revision_number":4},{"id":"b66f8d45effbf0857f56493f75fd92bbe11ff561","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2023-09-07 14:25:51.000000000","message":"Uploaded patch set 5.\n\nOutdated Votes:\n* Verified-1\n* Workflow-1\n","accounts_in_message":[],"_revision_number":5},{"id":"28c249f9f07005557819b3b7467bd2c4125efbb3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-07 17:14:16.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/1929eaa6cdd44590b67f430da9ee8ae3\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/6aa6b646135a4270899f023779f6022a : SUCCESS in 10m 27s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3498f9bf376e4e6c9223e2684946f2e0 : SUCCESS in 4m 50s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/da7d14320a08445195171767dddc0c60 : FAILURE in 4m 14s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/d50fde4e59904930aca33929362a32ce : SUCCESS in 5m 03s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/0a38195fe82745a1b062af4f4fba1e07 : SUCCESS in 5m 29s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/27ecf54ead81434bb3a687af06b2449e : SUCCESS in 5m 17s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6739792cd1bb4cefb8a219ade5af5745 : SUCCESS in 11m 13s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/8ce89026355e486486205df58f3c4405 : SUCCESS in 6m 31s\n- octavia-tox-py39-tips https://zuul.opendev.org/t/openstack/build/67aefa1c1e04466daef3848b77e91987 : SUCCESS in 6m 03s\n- octavia-tox-functional-py39-tips https://zuul.opendev.org/t/openstack/build/2160a9cef7bf4f71bedecb438354f437 : SUCCESS in 8m 36s\n- octavia-tox-functional-py39-sqlalchemy-tips https://zuul.opendev.org/t/openstack/build/67d6cab2df8741f5824667fdf7518036 : SUCCESS in 10m 13s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/9b2ad5a8e15640fa90d0f2c45dea712a : SUCCESS in 4m 54s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/bb550c6436a24fc6bdd62be18580a873 : SUCCESS in 8m 38s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/e014e688a3ce4b95b10ff0d7e122711d : SUCCESS in 7m 17s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/9472b0913a1f4c01bfb6c96fd7f1d59b : SUCCESS in 7m 16s (non-voting)\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/08499c4d4da44108906be249de254af1 : FAILURE in 1h 35m 38s\n- octavia-v2-dsvm-scenario-traffic-ops https://zuul.opendev.org/t/openstack/build/da007f54380546f4aa2cbe6f00286ffd : SUCCESS in 2h 05m 09s\n- octavia-v2-dsvm-scenario-non-traffic-ops https://zuul.opendev.org/t/openstack/build/eac5523bf41c45758a369fab077ccb11 : SUCCESS in 1h 35m 11s\n- octavia-v2-dsvm-scenario-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/238cd7fe93de435c83232e9e76a873a2 : SUCCESS in 2h 08m 41s\n- octavia-v2-dsvm-scenario-non-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/f0df6dd79ee44e60992ed68db250d2e8 : SUCCESS in 1h 41m 51s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/c1e5dce4b9a04316beecc91260670f2e : SUCCESS in 37m 03s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/c1421e63d1184e81869b45696a69be67 : SUCCESS in 39m 10s\n- octavia-v2-dsvm-tls-barbican-fips https://zuul.opendev.org/t/openstack/build/09f4267a73794f71a4370d3c656e70e1 : SUCCESS in 1h 16m 48s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/71bf77a641244f37b4507407b9b1c65c : SUCCESS in 45m 00s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/e2b20f7c15054ab8821c298f8e9544fc : TIMED_OUT in 2h 33m 14s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/05c2422abbbf4c1bb0a1804dc0869c6f : SUCCESS in 2h 46m 35s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/3af50f9b72cb4ea69ad46ff0a58a5ac8 : TIMED_OUT in 2h 33m 52s (non-voting)\n- octavia-v2-dsvm-scenario-centos-9-stream https://zuul.opendev.org/t/openstack/build/9df334399076469093a9c335ffd0a8e1 : TIMED_OUT in 2h 34m 25s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"3d56d6bc7c9e8fd51826bcfbb1bdc7c14821a9a0","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2023-09-08 06:51:13.000000000","message":"Uploaded patch set 6.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":6},{"id":"f953a4a10d795dfd33389721007c3302c4c73422","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-08 09:51:47.000000000","message":"Patch Set 6: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/9e5f80eb775b4f7ab6b70b4e33489537\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/713111fd82584a9880ad94f09cac7d17 : SUCCESS in 9m 43s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bff3a353c48a4f8fa0693cdbbfed8eef : SUCCESS in 5m 02s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/debdbbf1147e4cefa9d19c107f551c4e : SUCCESS in 5m 07s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/4dbf2e5870934591970f99d77e802582 : SUCCESS in 5m 00s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/2012fa92e13840749526692f209c5cfe : SUCCESS in 7m 19s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/bdc6358e0e274e168587a1c5c266c125 : SUCCESS in 6m 04s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f0c61fcfd96c4965b86cbf0685737989 : SUCCESS in 11m 52s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/1b72263d2feb489997103ea6d1d63c47 : SUCCESS in 4m 50s\n- octavia-tox-py39-tips https://zuul.opendev.org/t/openstack/build/0e3be7a673f746e796b42bb65103753a : SUCCESS in 5m 04s\n- octavia-tox-functional-py39-tips https://zuul.opendev.org/t/openstack/build/62daa27944b44f7c9bad4511def7f93e : SUCCESS in 9m 40s\n- octavia-tox-functional-py39-sqlalchemy-tips https://zuul.opendev.org/t/openstack/build/34ed79ce2e38468ebab78d243d143d0a : SUCCESS in 9m 45s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/3705ef3b9345430dba512ab4c326051a : SUCCESS in 4m 26s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/22aae6b3ec83459b870929efb2ff77ef : SUCCESS in 7m 50s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/e2b4abaf11704ac59a8133729cac2e3c : SUCCESS in 7m 16s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/38305f24044f431f9ed8e3095da61b5e : SUCCESS in 6m 54s (non-voting)\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/29c0f03f225a411091562e41c6075b01 : FAILURE in 1h 14m 02s\n- octavia-v2-dsvm-scenario-traffic-ops https://zuul.opendev.org/t/openstack/build/700d43f32895472fa8a7256d062703fc : SUCCESS in 1h 30m 19s\n- octavia-v2-dsvm-scenario-non-traffic-ops https://zuul.opendev.org/t/openstack/build/ba6978e42839449e85167e7f041630dc : SUCCESS in 1h 13m 28s\n- octavia-v2-dsvm-scenario-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/32cd10bedf314475ae18d953d1409ae3 : SUCCESS in 1h 33m 35s\n- octavia-v2-dsvm-scenario-non-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/5e4ffa721513444a8f695ad8fcd3e478 : SUCCESS in 1h 15m 21s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/fb56f8fa3346451bb55cc616138f3111 : SUCCESS in 40m 06s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/a4952a7907894295bc80902d2905f759 : SUCCESS in 42m 57s\n- octavia-v2-dsvm-tls-barbican-fips https://zuul.opendev.org/t/openstack/build/bb9874d947404abf872d9a3c033febe1 : SUCCESS in 1h 13m 37s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/ea3d06bbbbbf484aa9350b7c3948b448 : SUCCESS in 39m 42s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/b97a8a46f11d4d778a5d2bc5bbab3f38 : TIMED_OUT in 2h 33m 34s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/49674029cd56430face5b1e32bc3b550 : SUCCESS in 2h 57m 54s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/73cc827fd9314d6d80030642512a6d16 : TIMED_OUT in 2h 33m 04s (non-voting)\n- octavia-v2-dsvm-scenario-centos-9-stream https://zuul.opendev.org/t/openstack/build/186e767b5b524fdda93365896b7090c6 : FAILURE in 2h 30m 13s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"64896272d4f82cb163491e6a1ac95d59574d261c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2023-09-12 11:18:54.000000000","message":"Uploaded patch set 7.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":7},{"id":"fbf655a4d1be24e686aa693b55ef866d55ca65cb","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-12 14:24:26.000000000","message":"Patch Set 7: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0b84b6636e9b49c58e2b4a2ef33b2c4f\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/f10efad9ac3e451387ed18e9e3ed1ebd : SUCCESS in 11m 22s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1cb6128cad5542cf9f646d931aa17a1c : SUCCESS in 5m 25s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/73b50119b4ab450f82b51d8940e05b0c : SUCCESS in 4m 19s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/f41266e8864043a2afaba695fedf59b2 : SUCCESS in 4m 49s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/aed8245e12ac4dfdb027b7cb09378f5c : SUCCESS in 4m 53s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/c21c70b51ed64ef6bd7aee5f100c77c6 : SUCCESS in 5m 03s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2b1f122689cd4d7f9c2b5460d9e7d3f6 : SUCCESS in 10m 10s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/303bef793426436488eca41da4700100 : SUCCESS in 4m 48s\n- octavia-tox-py39-tips https://zuul.opendev.org/t/openstack/build/52c5e894af014533aa3b2021e2250697 : SUCCESS in 5m 13s\n- octavia-tox-functional-py39-tips https://zuul.opendev.org/t/openstack/build/e9aeb50ca7974fb4a88f9be746c70414 : SUCCESS in 7m 25s\n- octavia-tox-functional-py39-sqlalchemy-tips https://zuul.opendev.org/t/openstack/build/c98754df49d4429eb9e0e8ea6b567fab : SUCCESS in 9m 20s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/58b0c8ec8bea4975870195d6643c6a10 : SUCCESS in 5m 17s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/cc83664d3e3c4461ad21bb1320bc87a1 : SUCCESS in 7m 56s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/c227844c321c42239d9fd3f1aaf4e4b4 : SUCCESS in 7m 31s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/ebd788ded5b241c88c770e1fa2a8586e : SUCCESS in 7m 44s (non-voting)\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/3ed16650baa641a5ac70b510eefac81c : SUCCESS in 1h 17m 49s\n- octavia-v2-dsvm-scenario-traffic-ops https://zuul.opendev.org/t/openstack/build/baa713a351cf45348cfafbe795a17832 : SUCCESS in 1h 39m 03s\n- octavia-v2-dsvm-scenario-non-traffic-ops https://zuul.opendev.org/t/openstack/build/5e866ac2c6f741668a6deda56a1e8795 : SUCCESS in 1h 06m 59s\n- octavia-v2-dsvm-scenario-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/9b971ebeb91c41ccbdf21132667d5165 : SUCCESS in 1h 39m 28s\n- octavia-v2-dsvm-scenario-non-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/e5999e54770e4c8ebb2c1b7ca25cf724 : SUCCESS in 1h 11m 49s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/42aa11dc51a345fa811ee86458809cb5 : SUCCESS in 47m 36s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/d6f50f1117b347c28ff6f294cf3dbc56 : SUCCESS in 44m 23s\n- octavia-v2-dsvm-tls-barbican-fips https://zuul.opendev.org/t/openstack/build/fe63fc7e44494aa5b2d2ad6bde15dd1a : SUCCESS in 1h 08m 46s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/a854f3a7d6e140ed94a94cd5ccb46ac5 : SUCCESS in 30m 23s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/3bd8b1a00ce2410d89a4532e157a9483 : TIMED_OUT in 2h 32m 58s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/ab483560590d4317861cf03db62f062d : TIMED_OUT in 3h 03m 51s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/5f58d762a3f04241816e07edd7bbb26c : SUCCESS in 1h 45m 49s (non-voting)\n- octavia-v2-dsvm-scenario-centos-9-stream https://zuul.opendev.org/t/openstack/build/8ded1fa515bc48f492ae50d07143ab13 : SUCCESS in 2h 01m 15s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"e36f9722cf754fc0a3eaea0b655c6d3c71e4d441","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2023-09-14 16:16:16.000000000","message":"Patch Set 7: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"40bb130fe1b1105d27b5fc50b6e51d589438df76","author":{"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"},"date":"2023-09-25 06:51:46.000000000","message":"Patch Set 7: Code-Review+1","accounts_in_message":[],"_revision_number":7},{"id":"e325fd82ddf3b1742228f44b529d4a1edba160ad","author":{"_account_id":27442,"name":"Quentin GROLLEAU","email":"quentin.grolleau@corp.ovh.com","username":"QG"},"date":"2023-09-25 13:13:07.000000000","message":"Patch Set 7: Code-Review+1","accounts_in_message":[],"_revision_number":7},{"id":"16f6a5976d9bdd21fc425fe6bb00475e0be83c9e","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2023-10-09 11:14:23.000000000","message":"Patch Set 7: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"a5464d5a5c0648806fe828b71d8cc2349458b1f6","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2023-10-09 11:39:08.000000000","message":"Uploaded patch set 8.\n\nOutdated Votes:\n* Code-Review+1, Code-Review+2 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":8},{"id":"cd62583ef313ff6b02c6481374cd40e238eca7f6","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2023-10-09 11:46:21.000000000","message":"Patch Set 8: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":8},{"id":"04c0a7ea97a921cf42377f2cd25ab323c675ba84","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2023-10-09 11:47:11.000000000","message":"Patch Set 8:\n\n(1 comment)","accounts_in_message":[],"_revision_number":8},{"id":"74b7a756db911c0c6d37065e4e2fe2bd5f835e19","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-09 14:02:01.000000000","message":"Patch Set 8: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b931e0659d534c5a85d068c570d2fe3f\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/c020f30d48df446d86488e12a51cbd28 : SUCCESS in 12m 19s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a6b792a0f27d4364ba38cc444c383420 : SUCCESS in 4m 48s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/dca77ac4b31b4b1fb2319857ee3edbc3 : SUCCESS in 5m 45s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/1da4c3c7ef4d4675aa8b29746d37d920 : SUCCESS in 5m 00s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/9633e989a72d4aa7bb46daccabbf617d : SUCCESS in 5m 06s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/46357172e2f74be9aeda58914634ad27 : SUCCESS in 6m 46s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d8a9abe5967547ad8330bd6803c0cf9a : SUCCESS in 12m 31s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/3ec564a286544860852b935dc90fc94a : SUCCESS in 5m 13s\n- octavia-tox-py39-tips https://zuul.opendev.org/t/openstack/build/c504d154ea59409ba2ef9a56b4795dba : SUCCESS in 6m 38s\n- octavia-tox-functional-py39-tips https://zuul.opendev.org/t/openstack/build/ab5a1aa066984546a849c6f7a9ad5a8a : SUCCESS in 8m 05s\n- octavia-tox-functional-py39-sqlalchemy-tips https://zuul.opendev.org/t/openstack/build/41ceb5174c854fcc91701e856b064c7d : SUCCESS in 11m 09s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/124160b549034b16a27a629f901e34c7 : SUCCESS in 4m 58s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/3eccba53369846219f260737a50d310c : SUCCESS in 7m 50s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/3fdc08e4d261482184379da8f4b1d38d : SUCCESS in 7m 25s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/708c499282504110b41b2d59b753be72 : SUCCESS in 8m 36s (non-voting)\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/66cc3de300b84ad9a1e882991d2e84e1 : SUCCESS in 1h 09m 30s\n- octavia-v2-dsvm-scenario-traffic-ops https://zuul.opendev.org/t/openstack/build/4b29cebb80cd4a05bc13c7837fa3ed0c : SUCCESS in 1h 39m 08s\n- octavia-v2-dsvm-scenario-non-traffic-ops https://zuul.opendev.org/t/openstack/build/e00c589a2c044b34acf2c20b58764d0e : SUCCESS in 1h 05m 43s\n- octavia-v2-dsvm-scenario-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/36a61d59543b49f295fdb0c2a9849167 : SUCCESS in 1h 39m 46s\n- octavia-v2-dsvm-scenario-non-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/ac4f639485fe417fb8017dba7f180078 : SUCCESS in 1h 25m 39s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/9f3c97fa19a347afa8900d4a423443c1 : SUCCESS in 34m 14s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/7c1189bbaefb4c47bb69210f95d88beb : SUCCESS in 45m 14s\n- octavia-grenade-skip-level https://zuul.opendev.org/t/openstack/build/aa4d54ef83ce4e729f512c7f1249a26b : FAILURE in 38m 23s (non-voting)\n- octavia-v2-dsvm-tls-barbican-fips https://zuul.opendev.org/t/openstack/build/b1114dbfcba34dbd8b06444b64da2110 : SUCCESS in 1h 03m 33s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/78a0f6395830457da3afd2daeeea1951 : SUCCESS in 32m 19s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/5c344861efba40e79709237032477a9b : SUCCESS in 1h 46m 47s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/6f731c8d7dea4c68b46c364b35dc01bb : SUCCESS in 2h 21m 06s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/f6013501abc042aea4ccc38df7af7943 : SUCCESS in 1h 53m 28s (non-voting)\n- octavia-v2-dsvm-scenario-centos-9-stream https://zuul.opendev.org/t/openstack/build/cd7601018ec34179802a3c17331abcde : FAILURE in 27m 04s (non-voting)","accounts_in_message":[],"_revision_number":8},{"id":"82f99eb4c45ae41cf976eab32aec29d3f2132c8c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-09 14:02:39.000000000","message":"Patch Set 8: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":8},{"id":"c38321c5f5b085f56df84f64dea60b16bc891930","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-09 15:35:26.000000000","message":"Patch Set 8: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5ff99dc72999416ab720403b17627927\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/79b38d10342f404eb6987981ef43813a : SUCCESS in 6m 10s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/c81b7d01cfb04c4292ede7eb5a666abf : SUCCESS in 4m 52s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/fa6e9129dd75440f97639989428f88ee : SUCCESS in 5m 13s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/ff4a2c5c333a484fa038cdd8ba42ecb2 : SUCCESS in 4m 26s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e329e5a0afb840ffafd3e06c0bfb0b4c : SUCCESS in 12m 20s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/d63002379d234fa7a1be9cd117c297fc : SUCCESS in 5m 52s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/2b5e47803a81426e90cedd434e9020f5 : SUCCESS in 5m 41s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/1e8668e6b944471b88f6f72d5504c5b7 : SUCCESS in 7m 28s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/7d7856be77ac4f70a34c8bdac13bf289 : SUCCESS in 1h 27m 47s\n- octavia-v2-dsvm-scenario-traffic-ops https://zuul.opendev.org/t/openstack/build/b2cf235b6cbf4209b62a7c26d99387bd : SUCCESS in 1h 12m 38s\n- octavia-v2-dsvm-scenario-non-traffic-ops https://zuul.opendev.org/t/openstack/build/3e69ff96f0474be9907b97b4426535ee : SUCCESS in 54m 29s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/b138776c0b734e93ba47b2fc64aed595 : SUCCESS in 36m 04s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/4100a45aced44a23aa04ea1c3d02916b : SUCCESS in 41m 17s","accounts_in_message":[],"_revision_number":8},{"id":"ce9eef59b9b11889f131caa779010af65764dbf5","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-09 15:35:27.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":8},{"id":"492d77ad496d1e0292666be95f94cde8eef476ca","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-09 15:36:45.000000000","message":"Patch Set 8:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d59ae42a58ee4c7493876e3486b67c3b\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3d9f9988735f49ab8ad5a4442ca3c755 : SUCCESS in 57s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/9d60b76e9d4443bd9841ed58c434a705 : SUCCESS in 49s","accounts_in_message":[],"_revision_number":8}],"current_revision_number":8,"current_revision":"ca70587f0bcb4b2a05bd5c243f111140a5762114","revisions":{"5c4ca1a2ac295fbef9c61d6ef5478418b34ca94f":{"kind":"REWORK","_number":1,"created":"2023-09-01 15:37:22.000000000","uploader":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"ref":"refs/changes/36/893536/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/36/893536/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/36/893536/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/36/893536/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/36/893536/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/36/893536/1"}}},"commit":{"parents":[{"commit":"ee51a2a5b0870819c242da69216e8d318c1c6353","subject":"Merge \"Fix haproxy global maxconn with disabled listeners\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/ee51a2a5b0870819c242da69216e8d318c1c6353"}]}],"author":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2023-09-01 13:08:37.000000000","tz":-240},"committer":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2023-09-01 15:31:35.000000000","tz":-240},"subject":"Reduce duration of failovers with amphora in ERROR","message":"Reduce duration of failovers with amphora in ERROR\n\nIn the failover flow, there are multiple tasks for the configuration of\nVRRP for the other amphorae of the load balancer, but during outage the\nother amps may not be available. To prevent the tasks from attempting\nconnections to unreachable amphorae, we can detect in the first task\nthat an amp is unreachable and pass this information to the other tasks.\n\nThose connection attempts could have taken a lot of time, between 15 min\nand 40 min depending on the configuration of Octavia and the provider\ndriver (amphorav1 or amphorav2)\n\nCloses-Bug: #2033894\n\nChange-Id: Ib33a0b8d2875e4ff97c65933fe9360bb06994d32\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/5c4ca1a2ac295fbef9c61d6ef5478418b34ca94f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/5c4ca1a2ac295fbef9c61d6ef5478418b34ca94f"}]},"branch":"refs/heads/master"},"c4c4ce8aa4627146208db2976c826fa09dbdca59":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2023-09-04 10:01:18.000000000","uploader":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"ref":"refs/changes/36/893536/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/36/893536/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/36/893536/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/36/893536/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/36/893536/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/36/893536/2"}}},"commit":{"parents":[{"commit":"cce348040d43b92b59146adeb15a4f1486946a64","subject":"Fix timeout duration in start_vrrp_service during failovers","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/cce348040d43b92b59146adeb15a4f1486946a64"}]}],"author":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2023-09-01 13:08:37.000000000","tz":-240},"committer":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2023-09-04 09:57:19.000000000","tz":-240},"subject":"Reduce duration of failovers with amphora in ERROR","message":"Reduce duration of failovers with amphora in ERROR\n\nIn the failover flow, there are multiple tasks for the configuration of\nVRRP for the other amphorae of the load balancer, but during outage the\nother amps may not be available. To prevent the tasks from attempting\nconnections to unreachable amphorae, we can detect in the first task\nthat an amp is unreachable and pass this information to the other tasks.\n\nThose connection attempts could have taken a lot of time, between 15 min\nand 40 min depending on the configuration of Octavia and the provider\ndriver (amphorav1 or amphorav2)\n\nCloses-Bug: #2033894\n\nChange-Id: Ib33a0b8d2875e4ff97c65933fe9360bb06994d32\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/c4c4ce8aa4627146208db2976c826fa09dbdca59"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/c4c4ce8aa4627146208db2976c826fa09dbdca59"}]},"branch":"refs/heads/master"},"957c9bfe503de1d9d1dbca0d2b521deab26c2b98":{"kind":"REWORK","_number":3,"created":"2023-09-04 11:44:35.000000000","uploader":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"ref":"refs/changes/36/893536/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/36/893536/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/36/893536/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/36/893536/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/36/893536/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/36/893536/3"}}},"commit":{"parents":[{"commit":"cce348040d43b92b59146adeb15a4f1486946a64","subject":"Fix timeout duration in start_vrrp_service during failovers","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/cce348040d43b92b59146adeb15a4f1486946a64"}]}],"author":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2023-09-01 13:08:37.000000000","tz":-240},"committer":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2023-09-04 11:23:28.000000000","tz":-240},"subject":"Reduce duration of failovers with amphora in ERROR","message":"Reduce duration of failovers with amphora in ERROR\n\nIn the failover flow, there are multiple tasks for the configuration of\nVRRP for the other amphorae of the load balancer, but during outage the\nother amps may not be available. To prevent the tasks from attempting\nconnections to unreachable amphorae, we can detect in the first task\nthat an amp is unreachable and pass this information to the other tasks.\n\nThose connection attempts could have taken a lot of time, between 15 min\nand 40 min depending on the configuration of Octavia and the provider\ndriver (amphorav1 or amphorav2)\n\nCloses-Bug: #2033894\n\nChange-Id: Ib33a0b8d2875e4ff97c65933fe9360bb06994d32\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/957c9bfe503de1d9d1dbca0d2b521deab26c2b98"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/957c9bfe503de1d9d1dbca0d2b521deab26c2b98"}]},"branch":"refs/heads/master"},"c6faf8dcceed828360e026af1debe921fa008c79":{"kind":"REWORK","_number":4,"created":"2023-09-04 12:04:18.000000000","uploader":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"ref":"refs/changes/36/893536/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/36/893536/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/36/893536/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/36/893536/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/36/893536/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/36/893536/4"}}},"commit":{"parents":[{"commit":"cce348040d43b92b59146adeb15a4f1486946a64","subject":"Fix timeout duration in start_vrrp_service during failovers","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/cce348040d43b92b59146adeb15a4f1486946a64"}]}],"author":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2023-09-01 13:08:37.000000000","tz":-240},"committer":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2023-09-04 12:00:41.000000000","tz":-240},"subject":"Reduce duration of failovers with amphora in ERROR","message":"Reduce duration of failovers with amphora in ERROR\n\nIn the failover flow, there are multiple tasks for the configuration of\nVRRP for the other amphorae of the load balancer, but during outage the\nother amps may not be available. To prevent the tasks from attempting\nconnections to unreachable amphorae, we can detect in the first task\nthat an amp is unreachable and pass this information to the other tasks.\n\nThose connection attempts could have taken a lot of time, between 15 min\nand 40 min depending on the configuration of Octavia and the provider\ndriver (amphorav1 or amphorav2)\n\nCloses-Bug: #2033894\n\nChange-Id: Ib33a0b8d2875e4ff97c65933fe9360bb06994d32\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/c6faf8dcceed828360e026af1debe921fa008c79"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/c6faf8dcceed828360e026af1debe921fa008c79"}]},"branch":"refs/heads/master"},"a69d9ee734512cd3ca2752dafba272ab14bf8b70":{"kind":"REWORK","_number":5,"created":"2023-09-07 14:25:51.000000000","uploader":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"ref":"refs/changes/36/893536/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/36/893536/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/36/893536/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/36/893536/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/36/893536/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/36/893536/5"}}},"commit":{"parents":[{"commit":"cce348040d43b92b59146adeb15a4f1486946a64","subject":"Fix timeout duration in start_vrrp_service during failovers","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/cce348040d43b92b59146adeb15a4f1486946a64"}]}],"author":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2023-09-01 13:08:37.000000000","tz":-240},"committer":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2023-09-07 14:00:16.000000000","tz":-240},"subject":"Reduce duration of failovers with amphora in ERROR","message":"Reduce duration of failovers with amphora in ERROR\n\nIn the failover flow, there are multiple tasks for the configuration of\nVRRP for the other amphorae of the load balancer, but during outage the\nother amps may not be available. To prevent the tasks from attempting\nconnections to unreachable amphorae, we can detect in the first task\nthat an amp is unreachable and pass this information to the other tasks.\n\nThose connection attempts could have taken a lot of time, between 15 min\nand 40 min depending on the configuration of Octavia and the provider\ndriver (amphorav1 or amphorav2)\n\nCloses-Bug: #2033894\n\nChange-Id: Ib33a0b8d2875e4ff97c65933fe9360bb06994d32\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/a69d9ee734512cd3ca2752dafba272ab14bf8b70"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/a69d9ee734512cd3ca2752dafba272ab14bf8b70"}]},"branch":"refs/heads/master"},"35cfea5d140a78c74e1661db89901798b1d8ce32":{"kind":"REWORK","_number":6,"created":"2023-09-08 06:51:13.000000000","uploader":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"ref":"refs/changes/36/893536/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/36/893536/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/36/893536/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/36/893536/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/36/893536/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/36/893536/6"}}},"commit":{"parents":[{"commit":"cce348040d43b92b59146adeb15a4f1486946a64","subject":"Fix timeout duration in start_vrrp_service during failovers","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/cce348040d43b92b59146adeb15a4f1486946a64"}]}],"author":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2023-09-01 13:08:37.000000000","tz":-240},"committer":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2023-09-08 06:47:01.000000000","tz":-240},"subject":"Reduce duration of failovers with amphora in ERROR","message":"Reduce duration of failovers with amphora in ERROR\n\nIn the failover flow, there are multiple tasks for the configuration of\nVRRP for the other amphorae of the load balancer, but during outage the\nother amps may not be available. To prevent the tasks from attempting\nconnections to unreachable amphorae, we can detect in the first task\nthat an amp is unreachable and pass this information to the other tasks.\n\nThose connection attempts could have taken a lot of time, between 15 min\nand 40 min depending on the configuration of Octavia and the provider\ndriver (amphorav1 or amphorav2)\n\nCloses-Bug: #2033894\n\nChange-Id: Ib33a0b8d2875e4ff97c65933fe9360bb06994d32\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/35cfea5d140a78c74e1661db89901798b1d8ce32"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/35cfea5d140a78c74e1661db89901798b1d8ce32"}]},"branch":"refs/heads/master"},"61f580d8b19d0c49f48c72e6e3449d3a51d51210":{"kind":"REWORK","_number":7,"created":"2023-09-12 11:18:54.000000000","uploader":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"ref":"refs/changes/36/893536/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/36/893536/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/36/893536/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/36/893536/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/36/893536/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/36/893536/7"}}},"commit":{"parents":[{"commit":"cce348040d43b92b59146adeb15a4f1486946a64","subject":"Fix timeout duration in start_vrrp_service during failovers","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/cce348040d43b92b59146adeb15a4f1486946a64"}]}],"author":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2023-09-01 13:08:37.000000000","tz":-240},"committer":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2023-09-12 11:14:07.000000000","tz":-240},"subject":"Reduce duration of failovers with amphora in ERROR","message":"Reduce duration of failovers with amphora in ERROR\n\nIn the failover flow, there are multiple tasks for the configuration of\nVRRP for the other amphorae of the load balancer, but during outage the\nother amps may not be available. To prevent the tasks from attempting\nconnections to unreachable amphorae, we can detect in the first task\nthat an amp is unreachable and pass this information to the other tasks.\n\nThose connection attempts could have taken a lot of time, between 15 min\nand 40 min depending on the configuration of Octavia and the provider\ndriver (amphorav1 or amphorav2)\n\nCloses-Bug: #2033894\n\nChange-Id: Ib33a0b8d2875e4ff97c65933fe9360bb06994d32\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/61f580d8b19d0c49f48c72e6e3449d3a51d51210"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/61f580d8b19d0c49f48c72e6e3449d3a51d51210"}]},"branch":"refs/heads/master"},"ca70587f0bcb4b2a05bd5c243f111140a5762114":{"kind":"REWORK","_number":8,"created":"2023-10-09 11:39:08.000000000","uploader":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"ref":"refs/changes/36/893536/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/36/893536/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/36/893536/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/36/893536/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/36/893536/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/36/893536/8"}}},"commit":{"parents":[{"commit":"cce348040d43b92b59146adeb15a4f1486946a64","subject":"Fix timeout duration in start_vrrp_service during failovers","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/cce348040d43b92b59146adeb15a4f1486946a64"}]}],"author":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2023-09-01 13:08:37.000000000","tz":-240},"committer":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2023-10-09 11:37:18.000000000","tz":120},"subject":"Reduce duration of failovers with amphora in ERROR","message":"Reduce duration of failovers with amphora in ERROR\n\nIn the failover flow, there are multiple tasks for the configuration of\nVRRP for the other amphorae of the load balancer, but during outage the\nother amps may not be available. To prevent the tasks from attempting\nconnections to unreachable amphorae, we can detect in the first task\nthat an amp is unreachable and pass this information to the other tasks.\n\nThose connection attempts could have taken a lot of time, between 15 min\nand 40 min depending on the configuration of Octavia and the provider\ndriver (amphorav1 or amphorav2)\n\nCloses-Bug: #2033894\n\nChange-Id: Ib33a0b8d2875e4ff97c65933fe9360bb06994d32\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/ca70587f0bcb4b2a05bd5c243f111140a5762114"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/ca70587f0bcb4b2a05bd5c243f111140a5762114"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
