)]}'
{"id":"openstack%2Foctavia~893537","triplet_id":"openstack%2Foctavia~master~I497f2d8587e25eda0d169aa919cb6fbede7e9fc6","project":"openstack/octavia","branch":"master","attention_set":{},"removed_from_attention_set":{"29244":{"account":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"last_update":"2023-10-09 12:10:36.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I497f2d8587e25eda0d169aa919cb6fbede7e9fc6","subject":"Fix timeout duration in start_vrrp_service during failovers","status":"MERGED","created":"2023-09-01 15:42:36.000000000","updated":"2023-10-09 12:11:51.000000000","submitted":"2023-10-09 12:10:36.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"893537","meta_rev_id":"2b45eb8b4f98b3876e6ec99077250cde7195e187","_number":893537,"virtual_id_number":893537,"owner":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":27442,"name":"Quentin GROLLEAU","email":"quentin.grolleau@corp.ovh.com","username":"QG"},{"value":0,"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-10-09 12:10:36.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"recommended":{"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"},"all":[{"value":1,"date":"2023-09-25 11:48:59.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":27442,"name":"Quentin GROLLEAU","email":"quentin.grolleau@corp.ovh.com","username":"QG"},{"value":1,"date":"2023-09-25 06:51:50.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2023-09-14 16:08:30.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":2,"date":"2023-10-09 10:48:05.000000000","_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"all":[{"value":0,"_account_id":27442,"name":"Quentin GROLLEAU","email":"quentin.grolleau@corp.ovh.com","username":"QG"},{"value":0,"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":1,"date":"2023-10-09 10:48:05.000000000","_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":27442,"name":"Quentin GROLLEAU","email":"quentin.grolleau@corp.ovh.com","username":"QG"},{"value":0,"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":27442,"name":"Quentin GROLLEAU","email":"quentin.grolleau@corp.ovh.com","username":"QG"},{"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"},{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-09-01 18:29:56.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-09-14 16:08:30.000000000","updated_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"reviewer":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"state":"REVIEWER"},{"updated":"2023-09-25 06:51:50.000000000","updated_by":{"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"},"reviewer":{"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"},"state":"REVIEWER"},{"updated":"2023-09-25 11:48:59.000000000","updated_by":{"_account_id":27442,"name":"Quentin GROLLEAU","email":"quentin.grolleau@corp.ovh.com","username":"QG"},"reviewer":{"_account_id":27442,"name":"Quentin GROLLEAU","email":"quentin.grolleau@corp.ovh.com","username":"QG"},"state":"REVIEWER"},{"updated":"2023-10-09 10:48:05.000000000","updated_by":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"reviewer":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"state":"REVIEWER"}],"messages":[{"id":"6d938f587256b5a98b2cfe34b6bcf3f7a0c13acd","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2023-09-01 15:42:36.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"e778145d814b5e1a85532dc6d360fc3bf02bb07a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2023-09-01 15:44:39.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"25c2fccc477e1fe9b7121482abbbced803824d15","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-01 18:29:56.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/44fb0875d6cc4dcbb2e4ec4125c4225a\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/6b5fd899bf9349d7b0cfd1d801340975 : SUCCESS in 9m 14s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1281526f0f15418c8c8a1de0a0490089 : SUCCESS in 5m 28s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/605ffbb8091b440398bf5b78fd144513 : SUCCESS in 5m 19s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/2957da9c72f44ca29c55cb4180c8684f : SUCCESS in 4m 48s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/ed3ff7d9e97440b6a6c352e1086e985e : SUCCESS in 4m 34s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/aaa59ea8a9184e97bbb8b420c36339c9 : SUCCESS in 6m 01s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0f730d595af147839dad30f0e0754467 : FAILURE in 7m 21s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/67ddf2121cdd4eda9874d677741241c2 : SUCCESS in 5m 09s\n- octavia-tox-py39-tips https://zuul.opendev.org/t/openstack/build/2ad75e8fe703412da8a0f2b481c66062 : SUCCESS in 5m 59s\n- octavia-tox-functional-py39-tips https://zuul.opendev.org/t/openstack/build/e4d15fc214aa4c299a7c24abad65713e : SUCCESS in 8m 33s\n- octavia-tox-functional-py39-sqlalchemy-tips https://zuul.opendev.org/t/openstack/build/a673f75837244db69b2c060cc2c9cdcf : SUCCESS in 8m 29s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/e379274ed6e948fdaad9ae0698b69238 : SUCCESS in 5m 03s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/456565c01aa348aea598e7fc1870733f : SUCCESS in 7m 10s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/4ff53c9190174c8d856c4ea2e15d204d : SUCCESS in 8m 46s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/a704eb460e4a4a0e8d49bab3fa96bb43 : SUCCESS in 7m 20s (non-voting)\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/f1c6305faac847ceb839facb35dd80e9 : SUCCESS in 1h 39m 18s\n- octavia-v2-dsvm-scenario-traffic-ops https://zuul.opendev.org/t/openstack/build/40ee7173688a40e5942c77779d7a1ecb : SUCCESS in 1h 42m 43s\n- octavia-v2-dsvm-scenario-non-traffic-ops https://zuul.opendev.org/t/openstack/build/45f7e7f37892473bb0a9ca9e4a51a2bc : SUCCESS in 1h 06m 57s\n- octavia-v2-dsvm-scenario-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/41c3878aba13411186a1c1521e5eee87 : SUCCESS in 1h 48m 08s\n- octavia-v2-dsvm-scenario-non-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/0a5f92a106374362973cfea0eaf738f0 : SUCCESS in 1h 26m 11s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/4c84af4757a44c328cd58727f5071e36 : SUCCESS in 47m 50s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/53b886ea4f0141c3a145bd76c47b2ff8 : FAILURE in 44m 58s\n- octavia-v2-dsvm-tls-barbican-fips https://zuul.opendev.org/t/openstack/build/f2608e7c3f4a4c6986f53412e9968881 : SUCCESS in 59m 08s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/9592191b15f8474ba354b50bdaccec5a : SUCCESS in 35m 04s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/91ab8d0c122c4188b617ec1ea3aec6e0 : TIMED_OUT in 2h 33m 05s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/064b748ece334c1ebbab3ee413475d72 : SUCCESS in 2h 43m 25s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/dc9c68dfbb174e26b7badcefc13c080b : TIMED_OUT in 2h 32m 57s (non-voting)\n- octavia-v2-dsvm-scenario-centos-9-stream https://zuul.opendev.org/t/openstack/build/2cc081290fc741e8b9ea5f5d5421aaf4 : SUCCESS in 2h 01m 31s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"0fea7ac30befc625f80ca52de3b790f16a03e09f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2023-09-04 10:01:18.000000000","message":"Uploaded patch set 3: Patch Set 2 was rebased.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":3},{"id":"670f5123bf52df2748a22a94fa3a40e6b2822a3f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-04 12:58:56.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/1222ddbfccad4649a8a18f416585f249\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/785947b58bcd4c77bdd668226c339fcd : SUCCESS in 12m 29s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6c6df51a8e9a4ee08f6313b3e9a10884 : SUCCESS in 4m 51s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/0e414609e1ab4da79afaed638686315d : SUCCESS in 4m 45s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/5bec40cfccfb4914a79b1c08bd47a6d3 : SUCCESS in 4m 25s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/0e162276a66e43b49c2c8f4f966489ba : SUCCESS in 4m 55s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/1a458b03c0be44d0b8d533775146a717 : SUCCESS in 5m 17s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c099d140377b4fc2a2e9a42ac4276c7c : SUCCESS in 12m 00s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/58818931ae1544c5816861522b928bdd : SUCCESS in 4m 57s\n- octavia-tox-py39-tips https://zuul.opendev.org/t/openstack/build/5c70d8f7534d4c95a1abb613c6e32c2d : SUCCESS in 5m 57s\n- octavia-tox-functional-py39-tips https://zuul.opendev.org/t/openstack/build/bf3fa13f31004f818b679c6adf8b1e76 : SUCCESS in 8m 13s\n- octavia-tox-functional-py39-sqlalchemy-tips https://zuul.opendev.org/t/openstack/build/a9f42e98ddd44d93a0dd68f255f67ed5 : SUCCESS in 11m 57s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/9e457d65ecc4404c9763b7de92360d69 : SUCCESS in 4m 39s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/db5b4c48954845fe8eab1c8b40d40cf0 : SUCCESS in 8m 09s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/35347b2ab44b41d5b034ed269b758abe : SUCCESS in 7m 21s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/6f27570352d84459aee2702aa35c056c : SUCCESS in 9m 10s (non-voting)\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/d878b790d15c4fd883ca967026a953bb : SUCCESS in 1h 26m 41s\n- octavia-v2-dsvm-scenario-traffic-ops https://zuul.opendev.org/t/openstack/build/feb06fcd399148c2ae3eadc1eaf3b1a5 : SUCCESS in 1h 22m 43s\n- octavia-v2-dsvm-scenario-non-traffic-ops https://zuul.opendev.org/t/openstack/build/d5bbe8aee4894b53b7d845c95b61c2ed : SUCCESS in 1h 21m 35s\n- octavia-v2-dsvm-scenario-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/32426df78a3f4454a569a2db75d7fdf7 : SUCCESS in 2h 07m 21s\n- octavia-v2-dsvm-scenario-non-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/7e3424c458da4a82b06cf46e81484ae6 : FAILURE in 1h 31m 26s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/3d572ff1f2e74758ab0b6b09558b3e8b : SUCCESS in 41m 43s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/f1685f7951d04d458f77adf2b2b543d7 : FAILURE in 37m 14s\n- octavia-v2-dsvm-tls-barbican-fips https://zuul.opendev.org/t/openstack/build/c4f262221bcb41febf886841f5f30597 : SUCCESS in 1h 06m 14s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/dcfadf04e63644a9bcb68ec6b524960c : SUCCESS in 36m 42s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/9b8c911535df453d961af3e77ad6dd2e : TIMED_OUT in 2h 33m 02s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/6160aaefa520424eb83dee790f88392a : SUCCESS in 2h 55m 49s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/9035acf0b61543c787d9d42a1cf7d189 : TIMED_OUT in 2h 32m 54s (non-voting)\n- octavia-v2-dsvm-scenario-centos-9-stream https://zuul.opendev.org/t/openstack/build/53afcbe51ba4490e956f787ec3c1fd3e : SUCCESS in 2h 14m 17s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"b21dec056778697c34d252f9f6d90904fb97d574","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2023-09-06 16:15:51.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"4b87def186320e3a121a6f5728d93fbbb022cecd","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-06 19:18:15.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6ec873e1c79849a889263205f335a2bb\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/b9223a085f104c0e8e98a034fa71cffa : SUCCESS in 9m 13s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9f920f56fda846ab8504850b8447d4bb : SUCCESS in 5m 12s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/df74d61cc9db492aa2c99c31b01a3278 : SUCCESS in 4m 31s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/6ebc57f65bfd46bc8a37b258241a345b : SUCCESS in 5m 08s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/4633f430f47c49d8be0fb0844445312c : SUCCESS in 5m 07s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/7407dd8827d042d59b1415c697ff9f7e : SUCCESS in 5m 02s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/81166b60c42541e0b243df28027c5308 : SUCCESS in 13m 09s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/59288705bcc446e1b134887c0aba83e2 : SUCCESS in 4m 44s\n- octavia-tox-py39-tips https://zuul.opendev.org/t/openstack/build/c9933724d3bd4e2f83d399b9e380725d : SUCCESS in 4m 46s\n- octavia-tox-functional-py39-tips https://zuul.opendev.org/t/openstack/build/9f4020cd4c3e4adca4606b7105258d63 : SUCCESS in 9m 27s\n- octavia-tox-functional-py39-sqlalchemy-tips https://zuul.opendev.org/t/openstack/build/ff49ca3125cf4904af07f2c99fb80cef : SUCCESS in 10m 49s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/e4c9594945364844a930e27622c3da35 : SUCCESS in 5m 53s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/4e1344641fc542c686643ffc646f4492 : SUCCESS in 6m 54s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/1937e42deed9464d93d4b568b71ea0ef : SUCCESS in 7m 39s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/7a77615f64a044e8ab31414be2b40227 : SUCCESS in 7m 41s (non-voting)\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/d0d1c778959d4ed08a7e6a35c56bb033 : SUCCESS in 1h 17m 50s\n- octavia-v2-dsvm-scenario-traffic-ops https://zuul.opendev.org/t/openstack/build/702d39c2273a47149d2e8f0a630c5010 : SUCCESS in 1h 37m 32s\n- octavia-v2-dsvm-scenario-non-traffic-ops https://zuul.opendev.org/t/openstack/build/93166d087a9849659e208256183c9c71 : SUCCESS in 1h 13m 53s\n- octavia-v2-dsvm-scenario-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/5ae57594fe7c426cb15468d987b71e08 : SUCCESS in 1h 52m 15s\n- octavia-v2-dsvm-scenario-non-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/30ee76630e644fe28036750054440c8f : SUCCESS in 1h 27m 31s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/5cff5a6b7b1c487082269305fcebdb3f : SUCCESS in 44m 26s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/3d210f21c3c2458d9aed1e0dbe7b54e2 : SUCCESS in 39m 33s\n- octavia-v2-dsvm-tls-barbican-fips https://zuul.opendev.org/t/openstack/build/51cface6574b4ebfb1ea5dc7538989ac : SUCCESS in 1h 04m 26s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/d33fe44b3a3c455ba3d18b4badfba71a : SUCCESS in 31m 11s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/8a969b2e784a4a6daafd9a5b50ddc8f6 : SUCCESS in 1h 54m 44s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/1d776d9a908a4c5690bb48abed9d5aec : SUCCESS in 3h 00m 38s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/c47c755e14384dfdbf90456d3f1220af : SUCCESS in 1h 53m 56s (non-voting)\n- octavia-v2-dsvm-scenario-centos-9-stream https://zuul.opendev.org/t/openstack/build/c87004f1b09045f490f1515d50ddb62c : SUCCESS in 1h 50m 35s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"41cdeea3d47455e0d50e8a41719ab8067abd0af8","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2023-09-14 16:08:30.000000000","message":"Patch Set 3: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"45a6a643807bb3438ce3ae30c7c28d042d06bf13","author":{"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"},"date":"2023-09-25 06:51:50.000000000","message":"Patch Set 3: Code-Review+1","accounts_in_message":[],"_revision_number":3},{"id":"c4060d22e138aab7a5478551cfa4565274ae9dcd","author":{"_account_id":27442,"name":"Quentin GROLLEAU","email":"quentin.grolleau@corp.ovh.com","username":"QG"},"date":"2023-09-25 11:48:59.000000000","message":"Patch Set 3: Code-Review+1","accounts_in_message":[],"_revision_number":3},{"id":"986d186be77c4ad9369dd6a6ea7dcbc258fdd52a","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2023-10-09 10:48:05.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"9f47d0a43d4d6f579ab8f2138cd5f2adb9882ed7","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-09 10:48:50.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"2810f3106edab2bd13f284734b009f4aee8a6f1e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-09 12:10:36.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e182aeba30a541abafce4afa94fccaf8\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0277bcc5f1084bd8b9adaa6ca37e7ccd : SUCCESS in 5m 32s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/acce736f6bbd4ecca8e6c25990e865a3 : SUCCESS in 6m 39s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/716dc9987bd64862b558ee99b17b9427 : SUCCESS in 5m 29s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/936564ecc0f04e1f9a24e7faf95050ab : SUCCESS in 5m 42s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6bd7e3a2a7014654ba7e4765fca88287 : SUCCESS in 11m 54s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/0a3fb90e06894925b61fa693b4cc363a : SUCCESS in 5m 35s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/0773c0bf56b841729b94406b3b35353d : SUCCESS in 5m 06s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/c08672daf2ec4c1198e727b27e6ada67 : SUCCESS in 7m 56s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/878d61840d8b4df0b3a7adecb8e8de5c : SUCCESS in 1h 21m 37s\n- octavia-v2-dsvm-scenario-traffic-ops https://zuul.opendev.org/t/openstack/build/eb40fbe89f3f438382c2a41a3d24e19d : SUCCESS in 1h 15m 54s\n- octavia-v2-dsvm-scenario-non-traffic-ops https://zuul.opendev.org/t/openstack/build/e32f1bf46a89473ea923926f60e5ff28 : SUCCESS in 55m 27s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/a79f98c2bd734cfca081f2b83d08f471 : SUCCESS in 33m 13s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/fb8b91dad44e4d87b167986e4fa4cda2 : SUCCESS in 38m 33s","accounts_in_message":[],"_revision_number":3},{"id":"4fa725bd0daf761b2833216d28618dc1a8ecf20a","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-09 12:10:36.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":3},{"id":"2b45eb8b4f98b3876e6ec99077250cde7195e187","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-09 12:11:51.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/36b2c9c975c34116b76eaf27fa452089\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7e479663efbe416da1d270e09efd8873 : SUCCESS in 54s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/91c3e7bc874349dd92516db8a73b1baf : SUCCESS in 48s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"cce348040d43b92b59146adeb15a4f1486946a64","revisions":{"d4b79f0f73abbade492c2ac9c894d82a67431f1e":{"kind":"REWORK","_number":1,"created":"2023-09-01 15:42:36.000000000","uploader":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"ref":"refs/changes/37/893537/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/37/893537/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/37/893537/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/37/893537/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/37/893537/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/37/893537/1"}}},"commit":{"parents":[{"commit":"5c4ca1a2ac295fbef9c61d6ef5478418b34ca94f","subject":"Reduce duration of failovers with amphora in ERROR","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/5c4ca1a2ac295fbef9c61d6ef5478418b34ca94f"}]}],"author":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2023-09-01 15:35:34.000000000","tz":-240},"committer":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2023-09-01 15:37:38.000000000","tz":-240},"subject":"Fix timeout duration in start_vrrp_service during failovers","message":"Fix timeout duration in start_vrrp_service during failovers\n\nThe timeout_dict argument was not passed to the\n_populate_amphora_api_version, it could have made the failover of broken\namphora really long.\n\nRelated-Bug: #2033894\n\nChange-Id: I497f2d8587e25eda0d169aa919cb6fbede7e9fc6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/d4b79f0f73abbade492c2ac9c894d82a67431f1e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/d4b79f0f73abbade492c2ac9c894d82a67431f1e"}]},"branch":"refs/heads/master"},"efc4342166e26c580541456cc43585631797d29b":{"kind":"REWORK","_number":2,"created":"2023-09-01 15:44:39.000000000","uploader":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"ref":"refs/changes/37/893537/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/37/893537/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/37/893537/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/37/893537/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/37/893537/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/37/893537/2"}}},"commit":{"parents":[{"commit":"5c4ca1a2ac295fbef9c61d6ef5478418b34ca94f","subject":"Reduce duration of failovers with amphora in ERROR","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/5c4ca1a2ac295fbef9c61d6ef5478418b34ca94f"}]}],"author":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2023-09-01 15:35:34.000000000","tz":-240},"committer":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2023-09-01 15:41:11.000000000","tz":-240},"subject":"Fix timeout duration in start_vrrp_service during failovers","message":"Fix timeout duration in start_vrrp_service during failovers\n\nThe timeout_dict argument was not passed to the\n_populate_amphora_api_version, it could have made the failover of broken\namphora really long.\n\nRelated-Bug: #2033894\n\nChange-Id: I497f2d8587e25eda0d169aa919cb6fbede7e9fc6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/efc4342166e26c580541456cc43585631797d29b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/efc4342166e26c580541456cc43585631797d29b"}]},"branch":"refs/heads/master"},"cce348040d43b92b59146adeb15a4f1486946a64":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2023-09-04 10:01:18.000000000","uploader":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"ref":"refs/changes/37/893537/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/37/893537/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/37/893537/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/37/893537/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/37/893537/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/37/893537/3"}}},"commit":{"parents":[{"commit":"0426285a41464a5015494584f109eed35a0d44db","subject":"Merge \"Add octavia-grenade-slurp CI job\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/0426285a41464a5015494584f109eed35a0d44db"}]}],"author":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2023-09-01 15:35:34.000000000","tz":-240},"committer":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2023-09-04 09:57:19.000000000","tz":-240},"subject":"Fix timeout duration in start_vrrp_service during failovers","message":"Fix timeout duration in start_vrrp_service during failovers\n\nThe timeout_dict argument was not passed to the\n_populate_amphora_api_version, it could have made the failover of broken\namphora really long.\n\nRelated-Bug: #2033894\n\nChange-Id: I497f2d8587e25eda0d169aa919cb6fbede7e9fc6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/cce348040d43b92b59146adeb15a4f1486946a64"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/cce348040d43b92b59146adeb15a4f1486946a64"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
