)]}'
{"id":"openstack%2Foctavia~893612","triplet_id":"openstack%2Foctavia~master~I4bd027346c61b93b537ab53810c2ecb6160b6be2","project":"openstack/octavia","branch":"master","attention_set":{},"removed_from_attention_set":{"29244":{"account":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"last_update":"2023-10-09 14:32:43.000000000","reason":"\u003cGERRIT_ACCOUNT_29244\u003e replied on the change","reason_account":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"}}},"hashtags":[],"change_id":"I4bd027346c61b93b537ab53810c2ecb6160b6be2","subject":"Fix amphorae in ERROR during the failover","status":"MERGED","created":"2023-09-04 10:01:18.000000000","updated":"2023-10-09 18:06:59.000000000","submitted":"2023-10-09 18:05:58.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"893612","meta_rev_id":"7fd67406b831c9e450abd70e886af6a95b6f8994","_number":893612,"virtual_id_number":893612,"owner":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":27442,"name":"Quentin GROLLEAU","email":"quentin.grolleau@corp.ovh.com","username":"QG"},{"value":0,"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-10-09 18:05:58.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"recommended":{"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"},"all":[{"value":1,"date":"2023-10-09 11:39:17.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":27442,"name":"Quentin GROLLEAU","email":"quentin.grolleau@corp.ovh.com","username":"QG"},{"value":1,"date":"2023-10-09 11:39:17.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2023-10-09 11:39:17.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":2,"date":"2023-10-09 11:47:35.000000000","_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"all":[{"value":0,"_account_id":27442,"name":"Quentin GROLLEAU","email":"quentin.grolleau@corp.ovh.com","username":"QG"},{"value":0,"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":1,"date":"2023-10-09 11:47:35.000000000","_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":27442,"name":"Quentin GROLLEAU","email":"quentin.grolleau@corp.ovh.com","username":"QG"},{"value":0,"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":27442,"name":"Quentin GROLLEAU","email":"quentin.grolleau@corp.ovh.com","username":"QG"},{"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"},{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-09-04 14:39:33.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-09-14 16:19:55.000000000","updated_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"reviewer":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"state":"REVIEWER"},{"updated":"2023-09-25 06:51:43.000000000","updated_by":{"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"},"reviewer":{"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"},"state":"REVIEWER"},{"updated":"2023-10-02 14:57:13.000000000","updated_by":{"_account_id":27442,"name":"Quentin GROLLEAU","email":"quentin.grolleau@corp.ovh.com","username":"QG"},"reviewer":{"_account_id":27442,"name":"Quentin GROLLEAU","email":"quentin.grolleau@corp.ovh.com","username":"QG"},"state":"REVIEWER"},{"updated":"2023-10-09 11:47:35.000000000","updated_by":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"reviewer":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"state":"REVIEWER"}],"messages":[{"id":"ccb70d02f0ea78c13c8f70faedce68470f04dc76","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2023-09-04 10:01:18.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"c7d05d7d32fcb147e518f4b668bbb2a5f78d76e4","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2023-09-04 11:10:44.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"668d69b3452608c0c8ec102e2db4fd88bc4b32d3","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2023-09-04 11:44:35.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"0ccc8e6d648420f90c1bd32cbba13ab5c13a7e86","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2023-09-04 12:04:18.000000000","message":"Uploaded patch set 4: Patch Set 3 was rebased.","accounts_in_message":[],"_revision_number":4},{"id":"4a7d11d08e97b6cbd2ee9250dfa2480e52627784","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-04 14:39:33.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/3d813de6299246618f599a74b91ec89e\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/037bcc1fcabf44e0bc9ac2b7006e5c98 : SUCCESS in 10m 40s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3a7c1760876243de8ce59749019a5925 : SUCCESS in 5m 06s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/7f42e73aa0fb44db9bd04fb15bd8af27 : SUCCESS in 5m 01s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/cb992ce9c79c42c2b5d14d42b55a9d3d : SUCCESS in 5m 22s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/a6af47d0565d4f8f93eda74fa53a3a9a : SUCCESS in 5m 10s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/a78ae2ee2a8b48aea4495ed18f4bba7f : SUCCESS in 6m 00s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/beb44fa4c58541958ff3e20682b47a9a : SUCCESS in 12m 00s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/a448c6c8729d4c9197410921bb765a14 : SUCCESS in 4m 39s\n- octavia-tox-py39-tips https://zuul.opendev.org/t/openstack/build/9737fb97249c42988dcbe10cae976ded : SUCCESS in 5m 57s\n- octavia-tox-functional-py39-tips https://zuul.opendev.org/t/openstack/build/3bb9bcf116cd48aba0882414bbfea8c2 : SUCCESS in 8m 01s\n- octavia-tox-functional-py39-sqlalchemy-tips https://zuul.opendev.org/t/openstack/build/209df3056cc94f4da010bf7d59b0d561 : SUCCESS in 10m 24s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/f56a59c6ec344059ab43d29807d5147a : SUCCESS in 5m 21s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/c3503c420815496c9cf85a0292f8e752 : SUCCESS in 7m 49s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/9c66d1a24b494993982589b88ac66530 : SUCCESS in 8m 34s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/2124895944c9446eb9c533c955fe2c39 : SUCCESS in 9m 09s (non-voting)\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/ee9d80489ddb40848438c926512118d7 : SUCCESS in 1h 11m 38s\n- octavia-v2-dsvm-scenario-traffic-ops https://zuul.opendev.org/t/openstack/build/99ca15a842f54272bfd0109ea85420d1 : SUCCESS in 2h 05m 18s\n- octavia-v2-dsvm-scenario-non-traffic-ops https://zuul.opendev.org/t/openstack/build/44f81c22f4ac482987145b3f44f76b32 : SUCCESS in 1h 35m 55s\n- octavia-v2-dsvm-scenario-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/89305fec375b4a10b109b962d57eae87 : SUCCESS in 2h 12m 24s\n- octavia-v2-dsvm-scenario-non-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/49463b98b4994f168433c1019129a099 : SUCCESS in 1h 24m 47s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/1d73a02f981542b49afef57203c2b3fe : SUCCESS in 47m 19s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/3462aace6a204697b29a17755df36cae : FAILURE in 42m 51s\n- octavia-v2-dsvm-tls-barbican-fips https://zuul.opendev.org/t/openstack/build/d878b1a076be4903ade47098b67e8eb8 : SUCCESS in 1h 10m 04s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/cb691650030f4754b9203c1e4d726373 : SUCCESS in 39m 43s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/3342084578bd47d4b436ce6898e3f250 : TIMED_OUT in 2h 32m 58s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/a3c9b44eea6645f5a1c4e331414b662b : SUCCESS in 2h 32m 59s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/83620110b3f14e6fa1baa8a432d97789 : TIMED_OUT in 2h 32m 54s (non-voting)\n- octavia-v2-dsvm-scenario-centos-9-stream https://zuul.opendev.org/t/openstack/build/4945d68b56b0490c8fb8ca7943d41a6f : SUCCESS in 1h 49m 57s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"24b67ba72982b270e039b7c60892a443d3f8e25b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2023-09-07 14:25:51.000000000","message":"Uploaded patch set 5.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":5},{"id":"57807d5644745b175a6ac7b4c463b54b9e2d816b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2023-09-07 15:01:49.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"7687aebe7db3a5c51b33b2e764c512924f55c459","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-07 17:37:46.000000000","message":"Patch Set 6: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/6282c9305f414d4bb4a1880af2587a5a\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/c06ae444c7ec4f5c822ba646b735c9f7 : SUCCESS in 10m 59s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8becb4618494403c8d7ecbaa9f3fcd76 : SUCCESS in 5m 41s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/cf6685d0e12447ddacfdeb1ab104a1be : FAILURE in 4m 43s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/0bb913134b164539a83f97e62bc7cf6c : SUCCESS in 5m 30s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/626121779b564ee3b37c98ade9e558ca : SUCCESS in 5m 03s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/aea1f3df63f34cbd965aff4131e79d9c : SUCCESS in 8m 00s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/51f67b81ba374870a2565b788c8a5d5a : SUCCESS in 12m 24s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/3af4814420ac4dc9a6bb0089a573b23a : SUCCESS in 5m 58s\n- octavia-tox-py39-tips https://zuul.opendev.org/t/openstack/build/5acceab1d9d043349182575122941529 : SUCCESS in 5m 42s\n- octavia-tox-functional-py39-tips https://zuul.opendev.org/t/openstack/build/0000181c88ef422d9348e0ba0e5ef2df : SUCCESS in 7m 22s\n- octavia-tox-functional-py39-sqlalchemy-tips https://zuul.opendev.org/t/openstack/build/fd017c54a0144329905e8549f1c217e1 : SUCCESS in 9m 50s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/1e7f490c10cc4cab916a1599219434e5 : SUCCESS in 5m 05s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/7c5609a3f90e4c8b980d4a56f82db9de : SUCCESS in 7m 31s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/bf5850e85a444ce2bc10e0da336b4d4e : SUCCESS in 7m 20s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/94f46a1d168d4f3a80356c7b855f9aa4 : SUCCESS in 8m 57s (non-voting)\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/733ffe57e3a943de9674a31ec6a277f2 : FAILURE in 1h 20m 52s\n- octavia-v2-dsvm-scenario-traffic-ops https://zuul.opendev.org/t/openstack/build/e808d5b7e00b4a35bdc21346c2e85974 : SUCCESS in 1h 57m 56s\n- octavia-v2-dsvm-scenario-non-traffic-ops https://zuul.opendev.org/t/openstack/build/0331a348fea647fba9c91c8e33dfe6de : SUCCESS in 1h 29m 27s\n- octavia-v2-dsvm-scenario-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/b19dd2d611904660884035bde8685041 : SUCCESS in 2h 02m 14s\n- octavia-v2-dsvm-scenario-non-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/bf375984b08d4d598622a78e369e9da6 : SUCCESS in 1h 14m 39s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/4eab80452f2c4bf0a9356c36f0cf52b8 : SUCCESS in 39m 05s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/9feb9a7030d4473f9e302ef69930cb32 : SUCCESS in 45m 16s\n- octavia-v2-dsvm-tls-barbican-fips https://zuul.opendev.org/t/openstack/build/72ae41b18f9e4b57889bb63f55a827c5 : SUCCESS in 1h 09m 02s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/d8e134e306424978a898353eb7c3f000 : SUCCESS in 33m 07s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/0ee5c3fc6faa4901a5c7c060d734c294 : TIMED_OUT in 2h 33m 12s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/4c126c7961e34187bbcb657cfb884158 : SUCCESS in 2h 27m 24s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/77f2855937cf48b4ba0de7abcbe9f40f : TIMED_OUT in 2h 32m 55s (non-voting)\n- octavia-v2-dsvm-scenario-centos-9-stream https://zuul.opendev.org/t/openstack/build/419b0d6bb6ed414d85b3b59e1bc3041a : FAILURE in 2h 29m 18s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"478671bb67e3779ffdc36c69d41092d6ebb44dc2","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2023-09-08 06:51:13.000000000","message":"Uploaded patch set 7: Patch Set 6 was rebased.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":7},{"id":"f5cf588e300673b33663091f21a445f24e084444","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-08 09:47:33.000000000","message":"Patch Set 7: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/e434927ed6e84358b1503b847e0df2e2\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/d6c7202de6044ec488799e924184dac8 : SUCCESS in 12m 01s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/28582c23975444f9b66ed3920812f768 : SUCCESS in 4m 58s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/bdb68e728a0b450dac40f8fbe6b577b1 : SUCCESS in 5m 02s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/3bd67e2da1f544a2936520e6f6b0d7a6 : SUCCESS in 5m 02s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/30fbc3965b8a48ff98ff8c2e46c95bf0 : SUCCESS in 5m 14s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/baf19df9a55c4634ae3f0048604a7edc : SUCCESS in 5m 14s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f9504c59e82949829f1c96c0e3e654db : SUCCESS in 12m 52s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/c86ee31e367d401c861c61b1325cfa89 : SUCCESS in 5m 11s\n- octavia-tox-py39-tips https://zuul.opendev.org/t/openstack/build/f2f7a74d24c64b65ae6b3b3a321bd43a : SUCCESS in 4m 37s\n- octavia-tox-functional-py39-tips https://zuul.opendev.org/t/openstack/build/176cfdfb8c1d40dcaa550df380822e1a : SUCCESS in 8m 36s\n- octavia-tox-functional-py39-sqlalchemy-tips https://zuul.opendev.org/t/openstack/build/a3907d18cdd74bdf9d86115a93801878 : SUCCESS in 9m 33s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/159ed72b145d450cb5c7c8827f583428 : SUCCESS in 4m 42s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/dfc31577038941bbb0d066b64acfc533 : SUCCESS in 6m 18s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/5984da22c4f742578262ce5134cef59a : SUCCESS in 8m 27s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/f646af9b1fc14a209bf762863fe8b113 : SUCCESS in 7m 24s (non-voting)\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/f006b43c07894fa2b8ccfd855dd14ce0 : FAILURE in 1h 05m 54s\n- octavia-v2-dsvm-scenario-traffic-ops https://zuul.opendev.org/t/openstack/build/d12982d2462b42e8b61a4d4ee4be8ca3 : SUCCESS in 1h 42m 45s\n- octavia-v2-dsvm-scenario-non-traffic-ops https://zuul.opendev.org/t/openstack/build/f11046a6ed3f48c8a23fe89e50a4cab8 : SUCCESS in 1h 00m 17s\n- octavia-v2-dsvm-scenario-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/32b1a5112c204f24aa7575bad6eac996 : SUCCESS in 1h 53m 53s\n- octavia-v2-dsvm-scenario-non-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/be82be5792b34ed2873e05744481e5fd : SUCCESS in 1h 23m 25s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/c59823e38d6044689e446d7f7b2a7203 : SUCCESS in 50m 56s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/d720cb992a0944e1b2a72bda5dd2d192 : SUCCESS in 43m 20s\n- octavia-v2-dsvm-tls-barbican-fips https://zuul.opendev.org/t/openstack/build/afff524255734f62ad1653084ef76353 : SUCCESS in 1h 13m 47s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/1dd7c7306aa84468994cc02ced1b39e6 : SUCCESS in 40m 37s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/c69b957badf844148871feba5e9c1105 : TIMED_OUT in 2h 32m 52s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/64a52c46319f4035a34bdc4b4f9059d9 : SUCCESS in 2h 53m 46s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/c1122914e4064a1599d0d0ff2bf19662 : TIMED_OUT in 2h 33m 52s (non-voting)\n- octavia-v2-dsvm-scenario-centos-9-stream https://zuul.opendev.org/t/openstack/build/a8d894860a0c4eb295cf1893c6ec0f8a : SUCCESS in 2h 07m 53s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"3d40caae596cb41878d9c2a776d50046f8898764","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2023-09-12 11:18:54.000000000","message":"Uploaded patch set 8: Patch Set 7 was rebased.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":8},{"id":"742f797ab6c0278fa1901983ed1d9450b4efdfeb","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-12 14:03:24.000000000","message":"Patch Set 8: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f8fb13b322bb42bfb61d2eb8ee550695\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/22c0abc790024a7a9f97cb5100b61eb2 : SUCCESS in 8m 09s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/56eb15232a3d46508f965bda6f956290 : SUCCESS in 4m 58s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/c65e09061c124d1685bbddeb2f99d9a3 : SUCCESS in 4m 23s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/bf3a624269b34bb2a80dd097f5710dbd : SUCCESS in 5m 24s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/6cb2b9378d084e62a718fc7c91bd3dbd : SUCCESS in 4m 40s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/30d6d2eb5ad74f288b513ab10653d3d3 : SUCCESS in 5m 05s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e0c51f92a3a443e4ab94c4cf5b676e46 : SUCCESS in 13m 04s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/29ef99cf92b74b038bd5f2f6b3686df1 : SUCCESS in 5m 04s\n- octavia-tox-py39-tips https://zuul.opendev.org/t/openstack/build/aa4029cf68814f689bdf0e0c43dadbdd : SUCCESS in 5m 24s\n- octavia-tox-functional-py39-tips https://zuul.opendev.org/t/openstack/build/031363aa7586433d949f1dc2282b8047 : SUCCESS in 7m 52s\n- octavia-tox-functional-py39-sqlalchemy-tips https://zuul.opendev.org/t/openstack/build/aa1494a79e554a7d916bd0d29a6d2e06 : SUCCESS in 9m 57s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/fe7e0397640f4f2c8af5aef7d91176e2 : SUCCESS in 5m 11s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/f71cbb4824cd4d0ba28e14685fb50d4b : SUCCESS in 7m 35s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/5341fa30dbdf429e99b32aa1cc2aae98 : SUCCESS in 9m 26s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/936031fde7e94f17bfc70d221457b1aa : SUCCESS in 7m 35s (non-voting)\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/e500d07ff85e4193992fe90917f9f363 : SUCCESS in 1h 28m 29s\n- octavia-v2-dsvm-scenario-traffic-ops https://zuul.opendev.org/t/openstack/build/5132d2185d98489aa75b66dab32ac502 : SUCCESS in 1h 58m 55s\n- octavia-v2-dsvm-scenario-non-traffic-ops https://zuul.opendev.org/t/openstack/build/c576409282984f09bdfecaf3de6a466a : SUCCESS in 1h 18m 37s\n- octavia-v2-dsvm-scenario-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/0bfe801d8d23429f8044faab7ecdbb39 : SUCCESS in 2h 02m 38s\n- octavia-v2-dsvm-scenario-non-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/1a12e6825e3c4ff58d35b4403169bf0f : SUCCESS in 1h 19m 06s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/c87129756700466fb6b48c1368b670f0 : SUCCESS in 39m 46s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/ae91d71cab6246c4938ac70a89f7965d : SUCCESS in 41m 29s\n- octavia-v2-dsvm-tls-barbican-fips https://zuul.opendev.org/t/openstack/build/8e924d3ca3e34b83add1f26908387e79 : SUCCESS in 58m 11s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/8c5655cf20884f5d9fcc07a9cc379385 : FAILURE in 47m 20s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/09299d5abc2e49c9a2fe27c87c599e7f : SUCCESS in 2h 06m 46s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/cff1df6152b04d6e894b3184899353bc : SUCCESS in 2h 32m 23s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/285d1f91a20a45abaa45db234d4fa963 : SUCCESS in 1h 41m 31s (non-voting)\n- octavia-v2-dsvm-scenario-centos-9-stream https://zuul.opendev.org/t/openstack/build/f88839042b904083a1888a3865bc03bf : SUCCESS in 1h 40m 18s (non-voting)","accounts_in_message":[],"_revision_number":8},{"id":"4cc4f3aeb052875ce03f5e4b3e64412240781c63","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2023-09-14 16:19:55.000000000","message":"Patch Set 8: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":8},{"id":"7db0b7bcf54d054abc050dd0a342d4dca658e643","author":{"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"},"date":"2023-09-25 06:51:43.000000000","message":"Patch Set 8: Code-Review+1","accounts_in_message":[],"_revision_number":8},{"id":"0c33536cb469c86eb811d94ad3e4e5e736756055","author":{"_account_id":27442,"name":"Quentin GROLLEAU","email":"quentin.grolleau@corp.ovh.com","username":"QG"},"date":"2023-10-02 14:57:13.000000000","message":"Patch Set 8: Code-Review+1","accounts_in_message":[],"_revision_number":8},{"id":"3573d05c5ca5416453a41014cd99d6e63d5245e9","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2023-10-09 11:39:17.000000000","message":"Patch Set 9: Patch Set 8 was rebased\n\nCopied Votes:\n* Code-Review+1, Code-Review+2 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":9},{"id":"3afd8ddb58bc15e73cc502c42b0ae7cd3514eeb5","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2023-10-09 11:47:35.000000000","message":"Patch Set 9: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":9},{"id":"d1399ee587b0f471cb98af1790bcbf32a7cdead5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-09 14:07:31.000000000","message":"Patch Set 9: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/52cbb60d1a8243d995c0e64435eeb1dc\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/fdb03ecd5c1346bd99b9a4481d2b9287 : SUCCESS in 11m 36s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e7ff0bdae1e641fc95676cdf7f80c6fe : SUCCESS in 5m 13s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/7ef0178564b944c8b76a0d936a5eb4ad : SUCCESS in 4m 34s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/774c0d0840a34f7cadea3f97c8a2af10 : SUCCESS in 4m 57s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/1d5c811373124f60aae554344eb9d42a : SUCCESS in 5m 15s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/ecdb8f033aff423dbd1552f1f4cf7ba2 : SUCCESS in 4m 43s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b4e7ad32a79d4f87a54378bc60374c15 : SUCCESS in 11m 36s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/38bebaea0f5e4e0bb5df9e1254d8693c : SUCCESS in 6m 26s\n- octavia-tox-py39-tips https://zuul.opendev.org/t/openstack/build/46c31b34d6884fc89085189cb67f70a9 : SUCCESS in 6m 22s\n- octavia-tox-functional-py39-tips https://zuul.opendev.org/t/openstack/build/9b7ff24181d142b4b68247491b073502 : SUCCESS in 8m 03s\n- octavia-tox-functional-py39-sqlalchemy-tips https://zuul.opendev.org/t/openstack/build/4d3e7bbe79c2448c97041a287deb0351 : SUCCESS in 11m 40s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/138bfd58ca064fdebc89b2465083b629 : SUCCESS in 4m 55s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/e77504996c8e4f4c8975ca107750f212 : SUCCESS in 7m 10s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/ae5f3e0d51c84c1d95fd53bfd12c7222 : SUCCESS in 8m 00s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/ccd942ffb73e4c8fa8fdd53145ff16e7 : SUCCESS in 8m 00s (non-voting)\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/18784b8582ba4d71aa605273449c1fdd : SUCCESS in 1h 26m 43s\n- octavia-v2-dsvm-scenario-traffic-ops https://zuul.opendev.org/t/openstack/build/d5529da0e0504c458ae0bd1e9f3d4c9d : SUCCESS in 1h 19m 17s\n- octavia-v2-dsvm-scenario-non-traffic-ops https://zuul.opendev.org/t/openstack/build/b4006b1c0cdc4f77baa3825dec49eb57 : SUCCESS in 59m 03s\n- octavia-v2-dsvm-scenario-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/19b6adc8c0f44fae9e2656c763a1bb5c : SUCCESS in 1h 23m 03s\n- octavia-v2-dsvm-scenario-non-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/39a5392eb4e7437985f9590eb28ab5f7 : FAILURE in 1h 22m 35s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/aa834fd5def247b4b6b539ad7e895295 : SUCCESS in 39m 56s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/49979c47eeac4757a3604c6386f3b686 : SUCCESS in 43m 24s\n- octavia-grenade-skip-level https://zuul.opendev.org/t/openstack/build/05753324df83497781309d5292949563 : FAILURE in 37m 44s (non-voting)\n- octavia-v2-dsvm-tls-barbican-fips https://zuul.opendev.org/t/openstack/build/bed2c18039be453395d0ac5b820ee84f : SUCCESS in 1h 04m 08s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/53811db037534eb7b1532dc257ee3da1 : SUCCESS in 31m 25s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/75d1c2e958e744b28b9b47ec364491a0 : SUCCESS in 1h 58m 17s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/0f72c80c4156464e8e18a2e1f09f1d92 : SUCCESS in 2h 25m 24s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/099cc1e96f7e44828b9b2b60ccb8c21b : SUCCESS in 1h 45m 47s (non-voting)\n- octavia-v2-dsvm-scenario-centos-9-stream https://zuul.opendev.org/t/openstack/build/b2b1b39191f349e4aaa65fa8f4816cf4 : SUCCESS in 2h 11m 50s (non-voting)","accounts_in_message":[],"_revision_number":9},{"id":"e06a5de1fed3b3964d9e6ff0c5846b8f76f1e6d9","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2023-10-09 14:32:43.000000000","message":"Patch Set 9:\n\n(1 comment)","accounts_in_message":[],"_revision_number":9},{"id":"9f46d85c0b2fa65a3a0b1019d4ba53c9f0cbb290","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-09 16:34:00.000000000","message":"Patch Set 9: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/41cf0341bbe94573b9cee2b72dcc0e97\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/1ade9945daab47609e95d029c94ff438 : SUCCESS in 11m 19s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/70ca7e6f13254c409e4a1970fac32458 : SUCCESS in 4m 51s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/ca8a7bdf7bde48b88a178e49fe4cae7d : SUCCESS in 4m 26s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/50fd7b6aa9644d8c90e88182bf99fcee : SUCCESS in 6m 59s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/4e2fc250321b4b7b8b08e51d3c4c48d5 : SUCCESS in 5m 21s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/17c17440ee3648e18c1954ab31acff92 : SUCCESS in 5m 39s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7547a00baf4c43608c1636ab3f2f9db4 : SUCCESS in 11m 05s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/d7f6e028ecc849bc8bccc9f441202623 : SUCCESS in 5m 16s\n- octavia-tox-py39-tips https://zuul.opendev.org/t/openstack/build/9d800a301b0740b98c535c922730aeb9 : SUCCESS in 5m 37s\n- octavia-tox-functional-py39-tips https://zuul.opendev.org/t/openstack/build/f5716d4baef94a19b348cd8ee0ad835e : SUCCESS in 8m 09s\n- octavia-tox-functional-py39-sqlalchemy-tips https://zuul.opendev.org/t/openstack/build/bf8748c8464941bca2859d8bf14d7c3b : SUCCESS in 10m 55s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/aefa67e5a2064c1cbd52859e2c3c0e15 : SUCCESS in 4m 49s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/5a88cb036bf94941b60f856904888a90 : SUCCESS in 7m 46s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/bf9a50ff624b439da8584f347ac323a1 : SUCCESS in 8m 53s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/339bdb2852a9413a8d338857cddbbdfd : SUCCESS in 7m 45s (non-voting)\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/7352ce2adef547c990b6f7a65a07f9b8 : SUCCESS in 1h 21m 40s\n- octavia-v2-dsvm-scenario-traffic-ops https://zuul.opendev.org/t/openstack/build/bceddd71c81949e68dbb9b407cb0bfac : SUCCESS in 1h 15m 31s\n- octavia-v2-dsvm-scenario-non-traffic-ops https://zuul.opendev.org/t/openstack/build/b0d81500450f4b98899fe2988db8d100 : SUCCESS in 54m 39s\n- octavia-v2-dsvm-scenario-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/b4383fd1fdac431cac4b097e002ac41f : SUCCESS in 1h 20m 10s\n- octavia-v2-dsvm-scenario-non-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/8fc6ecdb17de401287d278dbd48a29e8 : SUCCESS in 1h 00m 53s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/832e358120e645758a9310709ed1408a : SUCCESS in 34m 16s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/6875d200abdd43b682d8ed424522b8d1 : SUCCESS in 38m 58s\n- octavia-grenade-skip-level https://zuul.opendev.org/t/openstack/build/66735ea7a35d4ac9b28d73cfaacdf19d : FAILURE in 39m 27s (non-voting)\n- octavia-v2-dsvm-tls-barbican-fips https://zuul.opendev.org/t/openstack/build/9770e0b98049477f8964642b6f630a48 : FAILURE in 10m 16s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/1d8fabf6ae744754bd32654d35208938 : SUCCESS in 33m 28s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/5233ec81667e472f8cefe5aade497a6b : SUCCESS in 1h 49m 02s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/95fa6f3cde534fd6b3cf1e8a4c6bc7d8 : SUCCESS in 1h 59m 20s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/2ef7b3b57a994d5db34cdcbce83e8aad : SUCCESS in 1h 44m 34s (non-voting)\n- octavia-v2-dsvm-scenario-centos-9-stream https://zuul.opendev.org/t/openstack/build/270385d6c64a442bb211cb9d5b8f9096 : FAILURE in 7m 50s (non-voting)","accounts_in_message":[],"_revision_number":9},{"id":"ef729b9503a090e77abf6f5fcb480667058a4fea","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-09 16:34:36.000000000","message":"Patch Set 9: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":9},{"id":"7d49af4a880c47d96d0ee004c2291dbe44630fad","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-09 18:05:58.000000000","message":"Patch Set 9: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/07ef3e6d0ff24177aff265e132a8637b\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/553313c2a58849d4aa881a8f0b15e5b2 : SUCCESS in 5m 33s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/179aa2dd01a742c38270b63e410f10d2 : SUCCESS in 4m 27s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/cad1330b0a7f4fa383c5c49ae19a2d1c : SUCCESS in 5m 24s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/90512bad969f41329f8c935346b30030 : SUCCESS in 4m 41s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0e5b57c8a808484f9019e5368d826553 : SUCCESS in 10m 03s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/cca6f840d7644b5d8223e43c7ac5db99 : SUCCESS in 4m 50s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/4683eb12748040118c24079c7eff44b7 : SUCCESS in 5m 08s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/2dc48f4984f9489eb7bb2642a0a074d5 : SUCCESS in 7m 18s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/7ab6c65df431422f8303c27b39237a68 : SUCCESS in 1h 26m 44s\n- octavia-v2-dsvm-scenario-traffic-ops https://zuul.opendev.org/t/openstack/build/80d9994c959c47198680ce72aef1df30 : SUCCESS in 1h 16m 37s\n- octavia-v2-dsvm-scenario-non-traffic-ops https://zuul.opendev.org/t/openstack/build/d8a48ae7bf394a0fb7e6f690babb83ed : SUCCESS in 57m 34s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/abcff55379564ebc83ea3f4a2f5cfafa : SUCCESS in 39m 17s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/56dad55c95664d9982998b0f8d146a22 : SUCCESS in 43m 28s","accounts_in_message":[],"_revision_number":9},{"id":"53ad4d7ca0ee03446867310e96f12f5e2d1e1141","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-09 18:05:58.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":9},{"id":"7fd67406b831c9e450abd70e886af6a95b6f8994","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-09 18:06:59.000000000","message":"Patch Set 9:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/77613e9564e446e6965c178aa55d44fb\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cda9214f76444c3293c185f720eadadb : SUCCESS in 44s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/e9bb6e3fc32d45a38a3a5248372098bd : SUCCESS in 39s","accounts_in_message":[],"_revision_number":9}],"current_revision_number":9,"current_revision":"248cf2893ea913a74d870a763802a51586e0f63b","revisions":{"85e60a7490df53c2fd2f3e2d33b72a61f2d15a5a":{"kind":"REWORK","_number":1,"created":"2023-09-04 10:01:18.000000000","uploader":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"ref":"refs/changes/12/893612/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/12/893612/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/12/893612/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/12/893612/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/12/893612/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/12/893612/1"}}},"commit":{"parents":[{"commit":"c4c4ce8aa4627146208db2976c826fa09dbdca59","subject":"Reduce duration of failovers with amphora in ERROR","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/c4c4ce8aa4627146208db2976c826fa09dbdca59"}]}],"author":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2023-09-04 09:56:25.000000000","tz":-240},"committer":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2023-09-04 09:57:19.000000000","tz":-240},"subject":"WIP Fix amphora in ERROR during the failover","message":"WIP Fix amphora in ERROR during the failover\n\nChange-Id: I4bd027346c61b93b537ab53810c2ecb6160b6be2\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/85e60a7490df53c2fd2f3e2d33b72a61f2d15a5a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/85e60a7490df53c2fd2f3e2d33b72a61f2d15a5a"}]},"branch":"refs/heads/master"},"121f43e0f35382e77331bbcf6bd9a995aa33c9b8":{"kind":"REWORK","_number":2,"created":"2023-09-04 11:10:44.000000000","uploader":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"ref":"refs/changes/12/893612/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/12/893612/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/12/893612/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/12/893612/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/12/893612/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/12/893612/2"}}},"commit":{"parents":[{"commit":"c4c4ce8aa4627146208db2976c826fa09dbdca59","subject":"Reduce duration of failovers with amphora in ERROR","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/c4c4ce8aa4627146208db2976c826fa09dbdca59"}]}],"author":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2023-09-04 09:56:25.000000000","tz":-240},"committer":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2023-09-04 11:05:00.000000000","tz":-240},"subject":"Fix amphora in ERROR during the failover","message":"Fix amphora in ERROR during the failover\n\nWhen 2 amps were down, the failover flow created the first one and\nneeded to update both amp to configure VRRP, but as the 2nd was missing,\nit was set to ERROR. Then the health-manager could not trigger a\nfailover becasue amphorae in ERROR are excluded from the automated\nfailover process.\n\nThis commit changes the tasks that must be run on both amphorae during a\nfailover of one amphora, it doesn\u0027t mark the secondary amphora in ERROR\nif it is not reachable.\n\nCloses-Bug: #2033734\n\nChange-Id: I4bd027346c61b93b537ab53810c2ecb6160b6be2\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/121f43e0f35382e77331bbcf6bd9a995aa33c9b8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/121f43e0f35382e77331bbcf6bd9a995aa33c9b8"}]},"branch":"refs/heads/master"},"e843d58e9f7bb6c64514aaca4b8b69fc31b71880":{"kind":"REWORK","_number":3,"created":"2023-09-04 11:44:35.000000000","uploader":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"ref":"refs/changes/12/893612/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/12/893612/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/12/893612/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/12/893612/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/12/893612/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/12/893612/3"}}},"commit":{"parents":[{"commit":"957c9bfe503de1d9d1dbca0d2b521deab26c2b98","subject":"Reduce duration of failovers with amphora in ERROR","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/957c9bfe503de1d9d1dbca0d2b521deab26c2b98"}]}],"author":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2023-09-04 09:56:25.000000000","tz":-240},"committer":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2023-09-04 11:29:06.000000000","tz":-240},"subject":"Fix amphora in ERROR during the failover","message":"Fix amphora in ERROR during the failover\n\nWhen 2 amps were down, the failover flow created the first one and\nneeded to update both amp to configure VRRP, but as the 2nd was missing,\nit was set to ERROR. Then the health-manager could not trigger a\nfailover becasue amphorae in ERROR are excluded from the automated\nfailover process.\n\nThis commit changes the tasks that must be run on both amphorae during a\nfailover of one amphora, it doesn\u0027t mark the secondary amphora in ERROR\nif it is not reachable.\n\nCloses-Bug: #2033734\n\nChange-Id: I4bd027346c61b93b537ab53810c2ecb6160b6be2\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/e843d58e9f7bb6c64514aaca4b8b69fc31b71880"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/e843d58e9f7bb6c64514aaca4b8b69fc31b71880"}]},"branch":"refs/heads/master"},"0baedb36bd28df5ea3edc369093e65b374e39fc6":{"kind":"TRIVIAL_REBASE","_number":4,"created":"2023-09-04 12:04:18.000000000","uploader":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"ref":"refs/changes/12/893612/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/12/893612/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/12/893612/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/12/893612/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/12/893612/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/12/893612/4"}}},"commit":{"parents":[{"commit":"c6faf8dcceed828360e026af1debe921fa008c79","subject":"Reduce duration of failovers with amphora in ERROR","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/c6faf8dcceed828360e026af1debe921fa008c79"}]}],"author":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2023-09-04 09:56:25.000000000","tz":-240},"committer":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2023-09-04 12:00:41.000000000","tz":-240},"subject":"Fix amphora in ERROR during the failover","message":"Fix amphora in ERROR during the failover\n\nWhen 2 amps were down, the failover flow created the first one and\nneeded to update both amp to configure VRRP, but as the 2nd was missing,\nit was set to ERROR. Then the health-manager could not trigger a\nfailover becasue amphorae in ERROR are excluded from the automated\nfailover process.\n\nThis commit changes the tasks that must be run on both amphorae during a\nfailover of one amphora, it doesn\u0027t mark the secondary amphora in ERROR\nif it is not reachable.\n\nCloses-Bug: #2033734\n\nChange-Id: I4bd027346c61b93b537ab53810c2ecb6160b6be2\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/0baedb36bd28df5ea3edc369093e65b374e39fc6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/0baedb36bd28df5ea3edc369093e65b374e39fc6"}]},"branch":"refs/heads/master"},"8f47654830a61c57bad6c6f68445a115d5c1047e":{"kind":"REWORK","_number":5,"created":"2023-09-07 14:25:51.000000000","uploader":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"ref":"refs/changes/12/893612/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/12/893612/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/12/893612/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/12/893612/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/12/893612/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/12/893612/5"}}},"commit":{"parents":[{"commit":"a69d9ee734512cd3ca2752dafba272ab14bf8b70","subject":"Reduce duration of failovers with amphora in ERROR","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/a69d9ee734512cd3ca2752dafba272ab14bf8b70"}]}],"author":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2023-09-04 09:56:25.000000000","tz":-240},"committer":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2023-09-07 14:19:49.000000000","tz":-240},"subject":"Fix amphorae in ERROR during the failover","message":"Fix amphorae in ERROR during the failover\n\nWhen 2 amps were down, the failover flow created the first one and\nneeded to update both amp to configure VRRP, but as the 2nd was missing,\nit was set to ERROR. Then the health-manager could not trigger a\nfailover becasue amphorae in ERROR are excluded from the automated\nfailover process.\n\nThis commit changes the tasks that must be run on both amphorae during a\nfailover of one amphora, it doesn\u0027t mark the secondary amphora in ERROR\nif it is not reachable.\n\nCloses-Bug: #2033734\n\nChange-Id: I4bd027346c61b93b537ab53810c2ecb6160b6be2\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/8f47654830a61c57bad6c6f68445a115d5c1047e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/8f47654830a61c57bad6c6f68445a115d5c1047e"}]},"branch":"refs/heads/master"},"2d37f81c7d9b2caf661213c9630f49d73124c271":{"kind":"REWORK","_number":6,"created":"2023-09-07 15:01:49.000000000","uploader":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"ref":"refs/changes/12/893612/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/12/893612/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/12/893612/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/12/893612/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/12/893612/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/12/893612/6"}}},"commit":{"parents":[{"commit":"a69d9ee734512cd3ca2752dafba272ab14bf8b70","subject":"Reduce duration of failovers with amphora in ERROR","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/a69d9ee734512cd3ca2752dafba272ab14bf8b70"}]}],"author":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2023-09-04 09:56:25.000000000","tz":-240},"committer":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2023-09-07 14:46:03.000000000","tz":-240},"subject":"Fix amphorae in ERROR during the failover","message":"Fix amphorae in ERROR during the failover\n\nWhen 2 amps were down, the failover flow created the first one and\nneeded to update both amp to configure VRRP, but as the 2nd was missing,\nit was set to ERROR. Then the health-manager could not trigger a\nfailover becasue amphorae in ERROR are excluded from the automated\nfailover process.\n\nThis commit changes the tasks that must be run on both amphorae during a\nfailover of one amphora, it doesn\u0027t mark the secondary amphora in ERROR\nif it is not reachable.\n\nCloses-Bug: #2033734\n\nChange-Id: I4bd027346c61b93b537ab53810c2ecb6160b6be2\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/2d37f81c7d9b2caf661213c9630f49d73124c271"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/2d37f81c7d9b2caf661213c9630f49d73124c271"}]},"branch":"refs/heads/master"},"53583372bf1f3332f643d0b9c8f3c7a4de080e1a":{"kind":"TRIVIAL_REBASE","_number":7,"created":"2023-09-08 06:51:13.000000000","uploader":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"ref":"refs/changes/12/893612/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/12/893612/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/12/893612/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/12/893612/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/12/893612/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/12/893612/7"}}},"commit":{"parents":[{"commit":"35cfea5d140a78c74e1661db89901798b1d8ce32","subject":"Reduce duration of failovers with amphora in ERROR","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/35cfea5d140a78c74e1661db89901798b1d8ce32"}]}],"author":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2023-09-04 09:56:25.000000000","tz":-240},"committer":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2023-09-08 06:47:01.000000000","tz":-240},"subject":"Fix amphorae in ERROR during the failover","message":"Fix amphorae in ERROR during the failover\n\nWhen 2 amps were down, the failover flow created the first one and\nneeded to update both amp to configure VRRP, but as the 2nd was missing,\nit was set to ERROR. Then the health-manager could not trigger a\nfailover becasue amphorae in ERROR are excluded from the automated\nfailover process.\n\nThis commit changes the tasks that must be run on both amphorae during a\nfailover of one amphora, it doesn\u0027t mark the secondary amphora in ERROR\nif it is not reachable.\n\nCloses-Bug: #2033734\n\nChange-Id: I4bd027346c61b93b537ab53810c2ecb6160b6be2\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/53583372bf1f3332f643d0b9c8f3c7a4de080e1a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/53583372bf1f3332f643d0b9c8f3c7a4de080e1a"}]},"branch":"refs/heads/master"},"abaf0212dbafbacb34e86d76cf3eea344e7438fb":{"kind":"TRIVIAL_REBASE","_number":8,"created":"2023-09-12 11:18:54.000000000","uploader":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"ref":"refs/changes/12/893612/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/12/893612/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/12/893612/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/12/893612/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/12/893612/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/12/893612/8"}}},"commit":{"parents":[{"commit":"61f580d8b19d0c49f48c72e6e3449d3a51d51210","subject":"Reduce duration of failovers with amphora in ERROR","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/61f580d8b19d0c49f48c72e6e3449d3a51d51210"}]}],"author":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2023-09-04 09:56:25.000000000","tz":-240},"committer":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2023-09-12 11:14:07.000000000","tz":-240},"subject":"Fix amphorae in ERROR during the failover","message":"Fix amphorae in ERROR during the failover\n\nWhen 2 amps were down, the failover flow created the first one and\nneeded to update both amp to configure VRRP, but as the 2nd was missing,\nit was set to ERROR. Then the health-manager could not trigger a\nfailover becasue amphorae in ERROR are excluded from the automated\nfailover process.\n\nThis commit changes the tasks that must be run on both amphorae during a\nfailover of one amphora, it doesn\u0027t mark the secondary amphora in ERROR\nif it is not reachable.\n\nCloses-Bug: #2033734\n\nChange-Id: I4bd027346c61b93b537ab53810c2ecb6160b6be2\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/abaf0212dbafbacb34e86d76cf3eea344e7438fb"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/abaf0212dbafbacb34e86d76cf3eea344e7438fb"}]},"branch":"refs/heads/master"},"248cf2893ea913a74d870a763802a51586e0f63b":{"kind":"TRIVIAL_REBASE","_number":9,"created":"2023-10-09 11:39:17.000000000","uploader":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"ref":"refs/changes/12/893612/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/12/893612/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/12/893612/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/12/893612/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/12/893612/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/12/893612/9"}}},"commit":{"parents":[{"commit":"ca70587f0bcb4b2a05bd5c243f111140a5762114","subject":"Reduce duration of failovers with amphora in ERROR","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/ca70587f0bcb4b2a05bd5c243f111140a5762114"}]}],"author":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2023-09-04 09:56:25.000000000","tz":-240},"committer":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2023-10-09 11:39:17.000000000","tz":0},"subject":"Fix amphorae in ERROR during the failover","message":"Fix amphorae in ERROR during the failover\n\nWhen 2 amps were down, the failover flow created the first one and\nneeded to update both amp to configure VRRP, but as the 2nd was missing,\nit was set to ERROR. Then the health-manager could not trigger a\nfailover becasue amphorae in ERROR are excluded from the automated\nfailover process.\n\nThis commit changes the tasks that must be run on both amphorae during a\nfailover of one amphora, it doesn\u0027t mark the secondary amphora in ERROR\nif it is not reachable.\n\nCloses-Bug: #2033734\n\nChange-Id: I4bd027346c61b93b537ab53810c2ecb6160b6be2\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/248cf2893ea913a74d870a763802a51586e0f63b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/248cf2893ea913a74d870a763802a51586e0f63b"}]},"branch":"refs/heads/master","description":"Rebase"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
