)]}'
{"id":"openstack%2Foctavia~893704","triplet_id":"openstack%2Foctavia~master~Id647cc71ec6fc982a152960e112fe5c63da01767","project":"openstack/octavia","branch":"master","attention_set":{},"removed_from_attention_set":{"29244":{"account":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"last_update":"2023-10-04 17:48:12.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Id647cc71ec6fc982a152960e112fe5c63da01767","subject":"Fix Zuul warnings with regexps","status":"MERGED","created":"2023-09-05 07:21:53.000000000","updated":"2023-10-04 17:50:42.000000000","submitted":"2023-10-04 17:48:12.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"893704","meta_rev_id":"06d23cbc206f7d5283bd6c8a56230326d136b87b","_number":893704,"virtual_id_number":893704,"owner":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-10-04 17:48:12.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},{"value":0,"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"recommended":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"all":[{"value":1,"date":"2023-09-06 16:07:39.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2023-09-06 16:06:49.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":1,"date":"2023-09-06 11:55:07.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},{"value":2,"date":"2023-10-04 16:09:34.000000000","_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"all":[{"value":0,"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},{"value":1,"date":"2023-10-04 16:09:34.000000000","_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},{"value":0,"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-09-05 10:23:17.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-09-06 11:55:07.000000000","updated_by":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"reviewer":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"state":"REVIEWER"},{"updated":"2023-09-06 16:06:49.000000000","updated_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"reviewer":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"state":"REVIEWER"},{"updated":"2023-09-06 16:07:39.000000000","updated_by":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"reviewer":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"state":"REVIEWER"},{"updated":"2023-10-04 16:09:34.000000000","updated_by":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"reviewer":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"state":"REVIEWER"}],"messages":[{"id":"8e288d50d8bbbe04ddf4b0ef372a9e0955b88498","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2023-09-05 07:21:53.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"8ae905f1b0ee2716f95f4631d09058f47efb7bbd","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-05 10:23:17.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/4020ef23d96c45d699e06eb0a3fafe1b\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/f811d9b814c7447ab9e4e35db4f997f5 : SUCCESS in 11m 39s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f8bf8a6c14394238a38fb02961e3f4aa : SUCCESS in 5m 32s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/8e378c0ae0a942ed97971fa7fa0653bf : SUCCESS in 5m 06s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/44ed88830655470289c5fa5203fa7440 : SUCCESS in 4m 38s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/491f511fc7da4b75a3cb62cf899fa246 : SUCCESS in 5m 02s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/6accb1d2daac4a9eb83238c50aa4db36 : SUCCESS in 5m 11s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4bcd648a0fdb436ab93493320bd0f816 : SUCCESS in 11m 26s\n- octavia-tox-py39-tips https://zuul.opendev.org/t/openstack/build/f32092bf32e745b7ba5879ea3cf87868 : SUCCESS in 5m 38s\n- octavia-tox-functional-py39-tips https://zuul.opendev.org/t/openstack/build/b96e67d7c5a040e69b809f7740db7ddc : SUCCESS in 7m 01s\n- octavia-tox-functional-py39-sqlalchemy-tips https://zuul.opendev.org/t/openstack/build/9af016c9a02a4fb2ac5e8511df5177b8 : SUCCESS in 11m 23s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/dab6eafe8f55461097ac979d1f3fed92 : SUCCESS in 5m 49s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/eb7c20474e7746c5b200001f1491c55c : SUCCESS in 8m 48s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/95b45129b9d9422fab8c39629c248405 : SUCCESS in 6m 46s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/0f680719bda34375bee95e95ece27200 : SUCCESS in 7m 27s (non-voting)\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/b7b683c6415243769dd6e35ab1a5b41a : SUCCESS in 1h 30m 01s\n- octavia-v2-dsvm-scenario-traffic-ops https://zuul.opendev.org/t/openstack/build/c001fdae15d8441d990949bcc1621a2c : SUCCESS in 1h 57m 53s\n- octavia-v2-dsvm-scenario-non-traffic-ops https://zuul.opendev.org/t/openstack/build/580005dbda75456ba665b36f25e2451f : SUCCESS in 1h 13m 18s\n- octavia-v2-dsvm-scenario-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/f6f631755a4741aba9ee8e47d61b80e0 : SUCCESS in 2h 02m 49s\n- octavia-v2-dsvm-scenario-non-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/c8c45bd019d2417a9c4ad68d33ad6b8f : SUCCESS in 1h 10m 01s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/ea138bcbf0d24dddae4982e6a82492a0 : SUCCESS in 35m 40s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/d5f431689bf94eedb319f3d0d1fa7c7c : FAILURE in 39m 51s\n- octavia-v2-dsvm-tls-barbican-fips https://zuul.opendev.org/t/openstack/build/5a7a710a14a44f04bb74eda475f2ad4d : SUCCESS in 1h 01m 34s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/7d46f3c9ab194005a823dc7d4f3223a0 : SUCCESS in 39m 06s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/ccec959c20ef4ddea05e40fb7a797fb3 : TIMED_OUT in 2h 32m 42s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/2aba429471404fcd9a607337a30f5658 : SUCCESS in 2h 58m 07s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/c817a5bc0240441e9e037fa591b9d473 : TIMED_OUT in 2h 32m 48s (non-voting)\n- octavia-v2-dsvm-scenario-centos-9-stream https://zuul.opendev.org/t/openstack/build/b2b8c7727b7644d3b3a115fbb9a2685f : SUCCESS in 2h 05m 00s (non-voting)\n\nWarning:\n  Zuul encountered a deprecated syntax while parsing its configuration\nin the repo openstack/octavia on branch master.  The problem was:\n\n  All regular expressions must conform to RE2 syntax, but an\n  expression using the deprecated Perl-style syntax has been detected.\n  Adjust the configuration to conform to RE2 syntax.\n  \n  The RE2 syntax error is: invalid perl operator: (?!\n\nThe problem appears in the following project stanza:\n\n  project:\n      templates:\n        - check-requirements\n        - periodic-stable-jobs-neutron\n        - openstack-cover-jobs\n        - openstack-python3-jobs\n        - publish-openstack-docs-pti\n        - release-notes-jobs-python3\n        - octavia-tox-tips\n        - octavia-tox-sqlalchemy-tips\n      check:\n        jobs:\n          - openstack-tox-pip-check-reqs:\n              irrelevant-files:\n                - ^.*\\.rst$\n                - ^api-ref/.*$\n                - ^doc/.*$\n                - ^etc/.*$\n                - ^octavia/tests/.*$\n                - ^releasenotes/.*$\n          - openstack-tox-functional-py38:\n              irrelevant-files:\n                - ^.*\\.rst$\n                - ^api-ref/.*$\n                - ^doc/.*$\n                - ^etc/.*$\n                - ^octavia/tests/unit/.*$\n                - ^releasenotes/.*$\n          - openstack-tox-functional-py39:\n              irrelevant-files:\n                - ^.*\\.rst$\n                - ^api-ref/.*$\n                - ^doc/.*$\n                - ^etc/.*$\n                - ^octavia/tests/unit/.*$\n                - ^releasenotes/.*$\n          - openstack-tox-functional-py310:\n              irrelevant-files:\n                - ^.*\\.rst$\n                - ^api-ref/.*$\n                - ^doc/.*$\n                - ^etc/.*$\n                - ^octavia/tests/unit/.*$\n                - ^releasenotes/.*$\n              voting: false\n          - octavia-v2-dsvm-noop-api:\n              irrelevant-files: \u0026irrelevant-files\n                - ^.*\\.rst$\n                - ^api-ref/.*$\n                - ^doc/.*$\n                - ^octavia/tests/.*$\n                - ^releasenotes/.*$\n          - octavia-v2-dsvm-scenario-traffic-ops:\n              irrelevant-files: *irrelevant-files\n          - octavia-v2-dsvm-scenario-non-traffic-ops:\n              irrelevant-files: *irrelevant-files\n          - octavia-v2-dsvm-scenario-traffic-ops-jobboard:\n              irrelevant-files: *irrelevant-files\n          - octavia-v2-dsvm-scenario-non-traffic-ops-jobboard:\n              irrelevant-files: *irrelevant-files\n          - octavia-v2-dsvm-tls-barbican:\n              irrelevant-files: *irrelevant-files\n          - octavia-grenade:\n              irrelevant-files: \u0026grenade-irrelevant-files\n                - ^.*\\.rst$\n                - ^api-ref/.*$\n                - ^doc/.*$\n                - ^octavia/tests/.*$\n                - ^releasenotes/.*$\n                - ^setup.cfg$\n                - ^tools/.*$\n                - ^(test-|)requirements.txt$\n                - ^tox.ini$\n          - octavia-grenade-skip-level:\n              irrelevant-files: *grenade-irrelevant-files\n          - octavia-v2-dsvm-tls-barbican-fips:\n              irrelevant-files: *irrelevant-files\n              voting: false\n          - octavia-v2-act-stdby-dsvm-scenario:\n              irrelevant-files: *irrelevant-files\n              voting: false\n          - octavia-v2-dsvm-cinder-amphora:\n              irrelevant-files: *irrelevant-files\n              voting: false\n          - octavia-v2-dsvm-scenario-two-node:\n              irrelevant-files: *irrelevant-files\n              voting: false\n          - octavia-v2-dsvm-scenario-ipv6-only:\n              irrelevant-files: *irrelevant-files\n              voting: false\n          - octavia-v2-dsvm-scenario-centos-9-stream:\n              irrelevant-files: *irrelevant-files\n              voting: false\n      queue: octavia\n      gate:\n        fail-fast: true\n        jobs:\n          - openstack-tox-pip-check-reqs:\n              irrelevant-files:\n                - ^.*\\.rst$\n                - ^api-ref/.*$\n                - ^doc/.*$\n                - ^etc/.*$\n                - ^octavia/tests/.*$\n                - ^releasenotes/.*$\n          - openstack-tox-functional-py38:\n              irrelevant-files:\n                - ^.*\\.rst$\n                - ^api-ref/.*$\n                - ^doc/.*$\n                - ^etc/.*$\n                - ^octavia/tests/unit/.*$\n                - ^releasenotes/.*$\n          - octavia-v2-dsvm-noop-api\n          - octavia-v2-dsvm-scenario-traffic-ops\n          - octavia-v2-dsvm-scenario-non-traffic-ops\n          - octavia-v2-dsvm-tls-barbican\n          - octavia-grenade\n          - octavia-grenade-skip-level\n      periodic:\n        jobs:\n          - publish-openstack-octavia-amphora-image-focal:\n              branches: ^(?!stable/.*).*$\n          - publish-openstack-octavia-amphora-image-centos-9-stream:\n              branches: ^(?!stable/.*).*$\n          - octavia-amphora-image-build\n  # Putting octavia-v2-dsvm-scenario-fips in periodic as centos 8 is too slow\n          - octavia-v2-dsvm-scenario-fips:\n              voting: false\n      experimental:\n        jobs:\n          - octavia-v2-dsvm-scenario-nftables\n\n  in \"openstack/octavia/zuul.d/projects.yaml@master\", line 3, column 3","accounts_in_message":[],"_revision_number":1},{"id":"c5bd0196266d2b157af3171667c126214556f599","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2023-09-05 10:35:03.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"c888298380f16a48808a5848e5efb83aff4f7c6a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-05 13:48:56.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/897a383803124f11b77e095f80e8da82\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/bcfd5f23fc0f42fd9eaa56847b640a81 : SUCCESS in 9m 34s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/48283996f07b4c5aa180561f78abd9d0 : SUCCESS in 4m 57s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/380c9bff1d004553bb89cba2d4a2b64e : SUCCESS in 4m 37s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/c543708edc7845849304827d4205c59f : SUCCESS in 4m 51s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/01e0156846a04729bb4d93818b5678e2 : SUCCESS in 5m 32s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/ea749c80c82642aaa4929d3ccf04a4de : SUCCESS in 5m 46s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/05d0182ee2fb487f8a779491e678ff27 : SUCCESS in 9m 15s\n- octavia-tox-py39-tips https://zuul.opendev.org/t/openstack/build/010f65805f3c487ba119167d33ef9aa2 : SUCCESS in 5m 58s\n- octavia-tox-functional-py39-tips https://zuul.opendev.org/t/openstack/build/091b269c4452480696f0b97ddbc6fe62 : SUCCESS in 8m 44s\n- octavia-tox-functional-py39-sqlalchemy-tips https://zuul.opendev.org/t/openstack/build/fbba5c4b44be4374896b1f50bd7e58ca : SUCCESS in 10m 38s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/bcf966e1a7da499092d883302b147ba8 : SUCCESS in 4m 45s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/f9a65fa825cb4445bc10711d741449d4 : SUCCESS in 6m 10s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/5d616418a7c24f5ba62f82164cdb239a : SUCCESS in 6m 54s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/022a67646170444cb94f16bb9d32ca21 : SUCCESS in 6m 46s (non-voting)\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/9502b2756d994e578a786aac521e8544 : SUCCESS in 59m 07s\n- octavia-v2-dsvm-scenario-traffic-ops https://zuul.opendev.org/t/openstack/build/c14d3559f2e2426ba149d20a1935cd23 : SUCCESS in 1h 35m 34s\n- octavia-v2-dsvm-scenario-non-traffic-ops https://zuul.opendev.org/t/openstack/build/7de52a5e5289469aa56021e62bafb445 : SUCCESS in 1h 14m 31s\n- octavia-v2-dsvm-scenario-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/38a224bb7eaa4d7ca14a53150aa0a84a : SUCCESS in 1h 46m 26s\n- octavia-v2-dsvm-scenario-non-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/6066cbcdb4d3442da2d25263a370d451 : SUCCESS in 1h 15m 24s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/0c0efba4a395452d8ac9c7eb6774d4d4 : SUCCESS in 46m 08s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/823ea49958cf4590a9666810cac7c852 : SUCCESS in 51m 55s\n- octavia-v2-dsvm-tls-barbican-fips https://zuul.opendev.org/t/openstack/build/2385f079d55a42ef910ed560718b902b : SUCCESS in 1h 05m 43s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/d6a2f22397e441789fb7e283cfd07201 : SUCCESS in 31m 33s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/4a2041a907df465ab9860dd699058431 : TIMED_OUT in 2h 33m 21s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/befd58fea62d4d859e8a2e0a1c2c9986 : SUCCESS in 3h 01m 10s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/d14d05337c924df38b03160d6dc33560 : SUCCESS in 1h 55m 30s (non-voting)\n- octavia-v2-dsvm-scenario-centos-9-stream https://zuul.opendev.org/t/openstack/build/34597fb24a3f48b5bfa8114da0997f7c : SUCCESS in 2h 12m 59s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"2c9d8b70fe9faf8087f68fca03a62a2316ebb53a","author":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"date":"2023-09-06 11:55:07.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"1fbf5a1c0c0789849c75ec43a563fa0c97fd4932","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2023-09-06 16:06:49.000000000","message":"Patch Set 2: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"1a3d48bdb00a1a0fa295bedefa429fac85e138f4","author":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"date":"2023-09-06 16:07:39.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"7b06b1396c255c4fe94c6c72b9a5d32111195f25","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2023-10-04 16:09:34.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"46d7857fc65bfa7e6c9c64de8524809367dd53ff","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-04 16:10:18.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"4b3f955e3ba5d2f521d7dbd01d38cb3b12080764","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-04 17:48:12.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/de986e5d9c19440f83a880d620b65786\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a1fb04393bcc45cb9ce3b7b8c13a14ed : SUCCESS in 6m 21s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/572f1f183ba14282b3c27cd346b3208a : SUCCESS in 6m 38s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/4059e1ee534f4fcb9e7dc0151be82248 : SUCCESS in 6m 56s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/e27a9b3d89154356879ed462306bf856 : SUCCESS in 5m 37s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b0347de132dd4ec8a499b293dab0f1fd : SUCCESS in 13m 34s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/24dd7bcfa2f9415c9466be6a0273f30e : SUCCESS in 6m 05s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/49c68bb3391247b19a1fb9a761660c6c : SUCCESS in 9m 15s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/f8343510a13d46b390df2f2dc8497958 : SUCCESS in 1h 32m 41s\n- octavia-v2-dsvm-scenario-traffic-ops https://zuul.opendev.org/t/openstack/build/7965a9d5ae644926a6984b37604ecb1c : SUCCESS in 1h 24m 42s\n- octavia-v2-dsvm-scenario-non-traffic-ops https://zuul.opendev.org/t/openstack/build/5e14ee6e654a4fb98d6b69e5732f939e : SUCCESS in 1h 07m 51s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/b17e08e1b96a4a2c8a76443debcddea6 : SUCCESS in 40m 04s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/de8b9f76aa884bf3a1a0503237bcc8c0 : SUCCESS in 54m 03s","accounts_in_message":[],"_revision_number":2},{"id":"93cf67e97d33548f03093b6c4678a35bbf3b47e9","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-04 17:48:12.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"06d23cbc206f7d5283bd6c8a56230326d136b87b","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-04 17:50:42.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/766c74867df24607a7663c4284fc6f42\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e2150aaae8904a0b87b0da8d14faa4b5 : SUCCESS in 1m 08s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"7cea604329cb1e81e7d44fb2833343e1f3bd3cb2","revisions":{"298eef382749f760f9f3af9fab3ad727e3506b4e":{"kind":"REWORK","_number":1,"created":"2023-09-05 07:21:53.000000000","uploader":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"ref":"refs/changes/04/893704/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/04/893704/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/04/893704/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/04/893704/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/04/893704/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/04/893704/1"}}},"commit":{"parents":[{"commit":"0426285a41464a5015494584f109eed35a0d44db","subject":"Merge \"Add octavia-grenade-slurp CI job\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/0426285a41464a5015494584f109eed35a0d44db"}]}],"author":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2023-09-05 07:20:02.000000000","tz":120},"committer":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2023-09-05 07:21:36.000000000","tz":120},"subject":"Fix Zuul warnings with regexps","message":"Fix Zuul warnings with regexps\n\nzuul now reports warnings when using perl-style regexp:\n| The RE2 syntax error is: invalid perl operator: (?!\n\nfor\n| branches: ^(?!stable)\n\nThis commit fixes it by using the correct RE2 syntax\n\nChange-Id: Id647cc71ec6fc982a152960e112fe5c63da01767\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/298eef382749f760f9f3af9fab3ad727e3506b4e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/298eef382749f760f9f3af9fab3ad727e3506b4e"}]},"branch":"refs/heads/master"},"7cea604329cb1e81e7d44fb2833343e1f3bd3cb2":{"kind":"REWORK","_number":2,"created":"2023-09-05 10:35:03.000000000","uploader":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"ref":"refs/changes/04/893704/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/04/893704/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/04/893704/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/04/893704/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/04/893704/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/04/893704/2"}}},"commit":{"parents":[{"commit":"0426285a41464a5015494584f109eed35a0d44db","subject":"Merge \"Add octavia-grenade-slurp CI job\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/0426285a41464a5015494584f109eed35a0d44db"}]}],"author":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2023-09-05 07:20:02.000000000","tz":120},"committer":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2023-09-05 10:34:53.000000000","tz":120},"subject":"Fix Zuul warnings with regexps","message":"Fix Zuul warnings with regexps\n\nzuul now reports warnings when using perl-style regexp:\n| The RE2 syntax error is: invalid perl operator: (?!\n\nfor\n| branches: ^(?!stable)\n\nThis commit fixes it by using the correct RE2 syntax\n\nChange-Id: Id647cc71ec6fc982a152960e112fe5c63da01767\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/7cea604329cb1e81e7d44fb2833343e1f3bd3cb2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/7cea604329cb1e81e7d44fb2833343e1f3bd3cb2"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
