)]}'
{"id":"openstack%2Foctavia~893780","triplet_id":"openstack%2Foctavia~stable%2F2023.1~I3fe9bb584c8522482e9d7289ae8ea1aab205dd4d","project":"openstack/octavia","branch":"stable/2023.1","attention_set":{},"removed_from_attention_set":{"31746":{"account":{"_account_id":31746,"name":"Oleksandr Kozachenko","email":"okozachenko1203@gmail.com","username":"okozachenko"},"last_update":"2023-11-15 14:56:47.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I3fe9bb584c8522482e9d7289ae8ea1aab205dd4d","subject":"Fix incorrect removal of IP rules in the amphora","status":"MERGED","created":"2023-09-06 15:28:52.000000000","updated":"2023-11-15 14:58:21.000000000","submitted":"2023-11-15 14:56:47.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"893780","cherry_pick_of_change":881728,"cherry_pick_of_patch_set":3,"meta_rev_id":"3d6fc17bb8cc96c8ddd62f5381c531218da6d7db","_number":893780,"virtual_id_number":893780,"owner":{"_account_id":31746,"name":"Oleksandr Kozachenko","email":"okozachenko1203@gmail.com","username":"okozachenko"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-11-15 14:56:47.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"recommended":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"all":[{"value":1,"date":"2023-09-06 21:10:42.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},{"value":2,"date":"2023-10-12 17:58:34.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":2,"date":"2023-11-15 08:11:51.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"all":[{"value":0,"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":1,"date":"2023-11-15 08:11:51.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-09-06 15:28:52.000000000","updated_by":{"_account_id":31746,"name":"Oleksandr Kozachenko","email":"okozachenko1203@gmail.com","username":"okozachenko"},"reviewer":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"state":"CC"},{"updated":"2023-09-06 18:33:52.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-09-06 21:10:42.000000000","updated_by":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"reviewer":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"state":"REVIEWER"},{"updated":"2023-10-12 17:58:34.000000000","updated_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"reviewer":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"state":"REVIEWER"},{"updated":"2023-11-15 08:11:51.000000000","updated_by":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"reviewer":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"state":"REVIEWER"}],"messages":[{"id":"f2948d2fc12fd804627ab3f564ec58276ba33fea","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":31746,"name":"Oleksandr Kozachenko","email":"okozachenko1203@gmail.com","username":"okozachenko"},"date":"2023-09-06 15:28:52.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"9ac4ab1a48ecdf5b48d2e6dfa34eb5dfa9a9bff4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-06 18:33:52.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a358b0b650824974ac5fd4459afc3dbd\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/63fcc3f3acfb4fa8b8d7ba3d12f13402 : SUCCESS in 10m 43s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5317fe3844a349b493f6e75e668bac29 : SUCCESS in 4m 56s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/872c4acc62a041528741c39f6cfafddd : SUCCESS in 4m 53s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/0af5de30358d44259f58abb9513276bb : SUCCESS in 5m 24s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b2083a9566564f73be39415ea71885e5 : SUCCESS in 14m 07s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/82c6dae372614f0693e8673d23d4b0e4 : SUCCESS in 5m 40s\n- octavia-tox-py39-tips https://zuul.opendev.org/t/openstack/build/3faf2d261ac84affae30427df87b81a4 : SUCCESS in 5m 21s\n- octavia-tox-functional-py39-tips https://zuul.opendev.org/t/openstack/build/62e77469fbc34d4085e7ca2f67056b9a : SUCCESS in 9m 15s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/a187c10e878044ab823398f68c0ae34e : SUCCESS in 5m 17s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/8f69778813644f4fa4ec4e651ee13de0 : SUCCESS in 7m 12s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/e153a96ee9404f6082af4b5faeb034f6 : SUCCESS in 8m 21s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/71631b315a8d4e1ca81695f63f4372b9 : SUCCESS in 8m 30s (non-voting)\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/ab24a17fb649454495fd9f7750e32f64 : SUCCESS in 1h 53m 16s\n- octavia-v2-dsvm-scenario-traffic-ops https://zuul.opendev.org/t/openstack/build/4e1d655fd41548db96d1bd8f79feff3c : SUCCESS in 1h 54m 22s\n- octavia-v2-dsvm-scenario-non-traffic-ops https://zuul.opendev.org/t/openstack/build/38adb1d2b87143feb1971a59ea62e15d : SUCCESS in 1h 23m 40s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/a35fad8bd5ea4933b41bf29fd5784cd4 : SUCCESS in 50m 36s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/4e4facc3ab2f4f29a7bac9f7775f7933 : SUCCESS in 49m 11s\n- octavia-grenade-skip-level https://zuul.opendev.org/t/openstack/build/732ab5158e854f50a9c23369f3110219 : SUCCESS in 59m 10s\n- octavia-v2-dsvm-tls-barbican-fips https://zuul.opendev.org/t/openstack/build/f29e865d64bf4433b766428bf5aebe2d : SUCCESS in 1h 02m 03s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/3db95984be06459cb9e3b2e67edc54a7 : SUCCESS in 47m 30s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/008d16ca0d134d529394eb93590aef02 : SUCCESS in 2h 31m 25s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/e91435c3c9b0432eb535656c08d2759e : SUCCESS in 3h 01m 26s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/f1eadafa7336410bb1f9c296a55e9cfc : SUCCESS in 1h 51m 39s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/0dbcdb027db3443591a7e97bd0b0c775 : TIMED_OUT in 2h 34m 46s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/060443fd2a3f4ec9ba3c982c2fae38dd : FAILURE in 2h 05m 40s (non-voting)\n- octavia-v2-dsvm-scenario-centos-9-stream https://zuul.opendev.org/t/openstack/build/f96bc7a562ea430792380bd138536e3b : SUCCESS in 2h 34m 34s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"245e88b1c3719e4dd3ad1e2c73e3cf309b07b46e","author":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"date":"2023-09-06 21:10:42.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"dc01f6b6196c4ebec8978cdbe8ef7ee78e0f568a","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2023-10-12 17:58:34.000000000","message":"Patch Set 1: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"38e63f7e908df5a69808088f0ef1cd4010a9aa23","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2023-11-15 08:11:51.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"ce9b4576b6ec8fe7e03026ee742e45c37b104c01","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-15 08:12:26.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"f601e0fd9caf6cc87afdbf53915dae63c58ab543","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-15 09:40:48.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/af39cfbe5f9a443fb1c490270e068a0b\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/417f232001d74fb1b6c828499b9201d8 : SUCCESS in 4m 37s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/40de0eb66b614677895ba88197d98352 : SUCCESS in 5m 02s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e795af9e1eba4609a0f7649bc50f6aac : SUCCESS in 11m 48s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/6777688e16c844b7988ed12ffcb300b5 : SUCCESS in 5m 44s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/ab607a6167d646f1b278b5e02657651f : SUCCESS in 4m 16s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/b81de7fef8fe4b64878b70edc6624924 : SUCCESS in 7m 12s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/f3aeebbc02484a28a14a0045d703362b : CANCELED\n- octavia-v2-dsvm-scenario-traffic-ops https://zuul.opendev.org/t/openstack/build/0312da3186964d3bafe9b420f42690b4 : FAILURE in 1h 25m 15s\n- octavia-v2-dsvm-scenario-non-traffic-ops https://zuul.opendev.org/t/openstack/build/19842f1c410b45bea17ca3b02762bfe3 : SUCCESS in 1h 08m 36s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/18c0b32e445a4e2bbc21989ac6133ac7 : SUCCESS in 41m 42s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/5c6a1f371b17431eaea8ee65ddf1f72d : SUCCESS in 44m 53s\n- octavia-grenade-skip-level https://zuul.opendev.org/t/openstack/build/a13d9ed6daf5492fb2854f2006837214 : SUCCESS in 46m 56s","accounts_in_message":[],"_revision_number":1},{"id":"4830172bc610e5652898f1c62311e77218d92c11","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2023-11-15 10:27:21.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"527cb8df4b2d03678d989d508f41d431a818b2e4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-15 13:04:12.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/01c562b081ad461b88c971ced18ed115\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/7b0f6d2e50e0471091c863fe80eb99e2 : SUCCESS in 10m 47s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/52a115d8bd2242ad9ca09cd438b0fe5c : SUCCESS in 5m 32s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/d1352dce7c9649a885369919933297ec : SUCCESS in 4m 48s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/424784b9bd554d72b5b81373bac71ff0 : SUCCESS in 12m 00s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/4af4a16f0e784e708a4485c277413a29 : SUCCESS in 4m 51s\n- octavia-tox-py39-tips https://zuul.opendev.org/t/openstack/build/5a216e24f75c443a92b0f6e8b827a7fa : SUCCESS in 4m 33s\n- octavia-tox-functional-py39-tips https://zuul.opendev.org/t/openstack/build/077324b125e746b59205b737a3509122 : SUCCESS in 6m 44s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/d7871f9bc46b429a8db5f39798c19d79 : SUCCESS in 4m 44s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/ba2ae10372d24c5eb6352b80e46a4c41 : SUCCESS in 6m 53s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/e484615075c5411287663a7474af162d : SUCCESS in 8m 36s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/0aab273ad219470683efee74ae40ae75 : SUCCESS in 6m 28s (non-voting)\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/e9e4611cae89418cbdf0d653927c8bda : SUCCESS in 1h 27m 12s\n- octavia-v2-dsvm-scenario-traffic-ops https://zuul.opendev.org/t/openstack/build/5b157e2a3f0c4563a510668300cb53c1 : SUCCESS in 1h 15m 40s\n- octavia-v2-dsvm-scenario-non-traffic-ops https://zuul.opendev.org/t/openstack/build/c70147317c164625bd173f1eb926e718 : SUCCESS in 58m 25s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/1a86da3ecf6e4f1ab9fab85283f8bb81 : SUCCESS in 34m 08s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/b2f48b4354644625b788a96b1a1b0ec4 : SUCCESS in 38m 33s\n- octavia-grenade-skip-level https://zuul.opendev.org/t/openstack/build/aee918285ea04d09995ae82e37f3b925 : SUCCESS in 45m 18s\n- octavia-v2-dsvm-tls-barbican-fips https://zuul.opendev.org/t/openstack/build/27fa6d37ddf24f89b9b69bdaeb6b5dc4 : SUCCESS in 1h 09m 05s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/8ff5ca27148f4ffcbdf121efa088b031 : SUCCESS in 29m 11s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/7b14184d1ee24c79b9d867b71fb373e3 : SUCCESS in 1h 52m 11s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/9a141d337f26410797f472875276a386 : SUCCESS in 2h 13m 42s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/8b64af8505574448806075ec88fd414f : SUCCESS in 2h 13m 02s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/f8569107d9c1447d85860bd05e153e78 : SUCCESS in 1h 57m 29s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/0902f07aa69f4881abf23f2c77fa3b58 : SUCCESS in 2h 09m 53s (non-voting)\n- octavia-v2-dsvm-scenario-centos-9-stream https://zuul.opendev.org/t/openstack/build/dcc365bb9baa4e55b471fbcc7d61c403 : SUCCESS in 2h 32m 45s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"752ce661db58411012198a606e89f0504bdaa608","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-15 13:04:47.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"adc03d8bb1f1ec5f635409cba3b0c921e81bde8b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-15 14:56:47.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/952f7aa5c7ad4738947106f85c409af6\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2b7bf1d44fbf4d939a32207761939fc0 : SUCCESS in 4m 49s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/b16f0582f0da4d02a072026168b09cdf : SUCCESS in 5m 15s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c31dea5433574061b53e5e2ba866179b : SUCCESS in 10m 56s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/d34e685fe4154a209c21633e931fb63a : SUCCESS in 5m 48s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/827926b3a0bc43e1a06d208515561d01 : SUCCESS in 4m 42s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/6f330c09e92942159aeeab068fa92fca : SUCCESS in 7m 28s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/f04eaf7caf9f493894627c55f797d8a1 : SUCCESS in 1h 14m 49s\n- octavia-v2-dsvm-scenario-traffic-ops https://zuul.opendev.org/t/openstack/build/1a99c4a100dc498591226ab806ba7d76 : SUCCESS in 1h 50m 40s\n- octavia-v2-dsvm-scenario-non-traffic-ops https://zuul.opendev.org/t/openstack/build/eb2c4520eb714cf09e46e3e0e94ec1a9 : SUCCESS in 1h 22m 33s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/6f878a0e7704459d9f0d8064403ff4cb : SUCCESS in 48m 02s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/99b563a1b46e4591b7a348d97690fb97 : SUCCESS in 50m 13s\n- octavia-grenade-skip-level https://zuul.opendev.org/t/openstack/build/36ded26a85ae47e29a087cbd6e715ca6 : SUCCESS in 43m 48s","accounts_in_message":[],"_revision_number":1},{"id":"b8131b785e2648ed2597640a6c85e712cc94d22f","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-15 14:56:47.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"3d6fc17bb8cc96c8ddd62f5381c531218da6d7db","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-15 14:58:21.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/eb0fb92297d9463a8e519db3438fbb3f\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/557b812bf21c4e21848b308f75745568 : SUCCESS in 1m 12s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/8a32d28c62b748b7819ea1ad24d8b7a8 : SUCCESS in 49s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"80c35d250f4146561ab3de1e14d1e93e29f72a31","revisions":{"80c35d250f4146561ab3de1e14d1e93e29f72a31":{"kind":"REWORK","_number":1,"created":"2023-09-06 15:28:52.000000000","uploader":{"_account_id":31746,"name":"Oleksandr Kozachenko","email":"okozachenko1203@gmail.com","username":"okozachenko"},"ref":"refs/changes/80/893780/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/80/893780/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/80/893780/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/80/893780/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/80/893780/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/80/893780/1"}}},"commit":{"parents":[{"commit":"8b196bb3bb05c9d87dae688750457dbb944d4d1b","subject":"Fix amphorav1 member deletion bug","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/8b196bb3bb05c9d87dae688750457dbb944d4d1b"}]}],"author":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2023-04-26 20:55:04.000000000","tz":-240},"committer":{"name":"Oleksandr Kozachenko","email":"okozachenko1203@gmail.com","date":"2023-09-06 15:28:52.000000000","tz":0},"subject":"Fix incorrect removal of IP rules in the amphora","message":"Fix incorrect removal of IP rules in the amphora\n\nWhen plugging a member, an IP rule was deleted from the amphora. It\ntriggered connectivity issue when a client on subnet1 sends a request to\nthe VIP (on subnet2) while subnet1 is also plugged to the amp as a\nmember subnet.\nThe IP rules should be updated only when plugging/updating the VIP\ninterface. The JSON files that store the configuration of the interfaces\nnow contain the type of the interface (\u0027vip\u0027, \u0027lo\u0027 or \u0027member\u0027) and the\nrules update step is skipped when the type is not \u0027vip\u0027.\n\nCloses-Bug: #2016475\n\nChange-Id: I3fe9bb584c8522482e9d7289ae8ea1aab205dd4d\n(cherry picked from commit b3bf1779929eebdcf66d38cdd386fb5d785698f5)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/80c35d250f4146561ab3de1e14d1e93e29f72a31"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/80c35d250f4146561ab3de1e14d1e93e29f72a31"}]},"branch":"refs/heads/stable/2023.1"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
