)]}'
{"id":"openstack%2Foctavia~893942","triplet_id":"openstack%2Foctavia~stable%2Fyoga~I3fe9bb584c8522482e9d7289ae8ea1aab205dd4d","project":"openstack/octavia","branch":"stable/yoga","attention_set":{},"removed_from_attention_set":{"31746":{"account":{"_account_id":31746,"name":"Oleksandr Kozachenko","email":"okozachenko1203@gmail.com","username":"okozachenko"},"last_update":"2023-11-15 11:39:11.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I3fe9bb584c8522482e9d7289ae8ea1aab205dd4d","subject":"Fix incorrect removal of IP rules in the amphora","status":"MERGED","created":"2023-09-06 15:30:03.000000000","updated":"2023-11-15 11:40:41.000000000","submitted":"2023-11-15 11:39:11.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"893942","cherry_pick_of_change":881728,"cherry_pick_of_patch_set":3,"meta_rev_id":"1413e82c91d809e7c8b63d9be5f9ae0560125b98","_number":893942,"virtual_id_number":893942,"owner":{"_account_id":31746,"name":"Oleksandr Kozachenko","email":"okozachenko1203@gmail.com","username":"okozachenko"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-11-15 11:39:11.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"recommended":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"all":[{"value":1,"date":"2023-09-06 21:11:23.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},{"value":2,"date":"2023-10-12 17:59:27.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":2,"date":"2023-11-15 08:12:57.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"all":[{"value":0,"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":1,"date":"2023-11-15 08:12:57.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-09-06 15:30:03.000000000","updated_by":{"_account_id":31746,"name":"Oleksandr Kozachenko","email":"okozachenko1203@gmail.com","username":"okozachenko"},"reviewer":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"state":"CC"},{"updated":"2023-09-06 17:58:02.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-09-06 21:11:23.000000000","updated_by":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"reviewer":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"state":"REVIEWER"},{"updated":"2023-10-12 17:59:27.000000000","updated_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"reviewer":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"state":"REVIEWER"},{"updated":"2023-11-15 08:12:57.000000000","updated_by":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"reviewer":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"state":"REVIEWER"}],"messages":[{"id":"99255c88f4877b2d004a68ec13bb88befa68dc94","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":31746,"name":"Oleksandr Kozachenko","email":"okozachenko1203@gmail.com","username":"okozachenko"},"date":"2023-09-06 15:30:03.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"1b9d1482d98c9d579d93fb6604c8113f9f9c8dab","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-06 17:58:02.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/60c9364f34c24e4995059bc59fe15584\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/cf1b972b49e543eaa5039d6e71772af7 : SUCCESS in 10m 42s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/45ee2aab878648b389926d7c2b9dd9b2 : SUCCESS in 6m 44s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/556b2d5dd7cf49daa00873479c16cb95 : SUCCESS in 5m 09s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/b9832b2051ec4c7a894d97669baf60a4 : SUCCESS in 4m 57s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/42cd5d42a205497db1e369eb6d2aa45e : SUCCESS in 12m 23s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/125cc64be6854f349e91b9bd86d5ef87 : SUCCESS in 5m 12s\n- octavia-tox-py39-tips https://zuul.opendev.org/t/openstack/build/f5f4f5e321a04f149906ffd8745cc50a : SUCCESS in 6m 32s\n- octavia-tox-functional-py39-tips https://zuul.opendev.org/t/openstack/build/971615fb2d7745bbaf9e7689fc9b084b : SUCCESS in 6m 24s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/6253dd4e34d7407d9453e65631493216 : SUCCESS in 4m 57s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/1036f848d8f446599c1c641ab4691988 : SUCCESS in 9m 31s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/466327d617a543c093009b26ee5e03aa : SUCCESS in 1h 15m 21s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/420410067c3c4feb8493c66da10edd2f : SUCCESS in 1h 45m 36s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/10ff7779e87f4020a75c60fe43cd09f2 : SUCCESS in 35m 48s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/4b19d9b8239341c9b2644c82aaa18dda : SUCCESS in 37m 05s\n- octavia-v2-dsvm-tls-barbican-fips https://zuul.opendev.org/t/openstack/build/4b3ea0a9a8e5446798c78bdb41a404b9 : FAILURE in 50m 49s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/30b3b50b34774d91bf2326493274c296 : SUCCESS in 30m 52s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/a1cd619893794e97811c4560fa45fd47 : SUCCESS in 1h 41m 47s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/8cc0a2a7ceab403cb4d01826e4c3d444 : FAILURE in 10m 48s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/0f9f0230bbfe46c685b1e79ae6b5a15c : FAILURE in 6m 16s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/098cb9e876804a7db20df92db49fe944 : SUCCESS in 1h 49m 52s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/a02fde82010646808e0d15c50afc1b27 : SUCCESS in 2h 25m 35s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8-stream https://zuul.opendev.org/t/openstack/build/ce74ad30455a4cedb4d7f063f35b96cf : FAILURE in 47m 23s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"befef9584011b1d286b8b2f2ddde107e4149c76e","author":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"date":"2023-09-06 21:11:23.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"78ca799e6ececdc4eb237cba41d68c4782ba871e","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2023-10-12 17:59:27.000000000","message":"Patch Set 1: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"90d202d67613f5a246228f9bb567840091b02773","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2023-11-15 08:12:57.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"007f5a7b382030666d62ca8b3dd925954db32d37","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-15 08:13:36.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"d5ab304b2a2bfc3ee1b1b3327abd2eaa83a01c27","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-15 11:39:11.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/66bf5972cd634e29a27d6f668e45611a\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9a1aa2d8729845938fb1d96ecd0c6865 : SUCCESS in 6m 39s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/5f170110138b4922a04d1ff5618f84f4 : SUCCESS in 13m 17s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/c698152bc63f4de3a438997924a06c8f : SUCCESS in 4m 42s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cd36052b17b64acf8c3fee26d4dcddea : SUCCESS in 12m 03s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/86e3ca0257764dbe83eefc8c12d45f22 : SUCCESS in 10m 17s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/35f6bf6856c140fbb3d95b83f0b19e38 : SUCCESS in 4m 34s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/f6c4407681eb4914912ee30169fa77de : SUCCESS in 7m 18s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/b423aa111c1343688447729a83ae9937 : SUCCESS in 1h 18m 47s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/a96436c8ba8b4217b1e879db99137bb6 : SUCCESS in 2h 00m 06s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/32dfcaa313ee4f32b75fe99a22ea328a : SUCCESS in 37m 21s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/d0f83157bb2548c5a086d252e1462cde : SUCCESS in 49m 46s","accounts_in_message":[],"_revision_number":1},{"id":"79798224c12b37e8c69f634be109f3803bebb2c9","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-15 11:39:11.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"1413e82c91d809e7c8b63d9be5f9ae0560125b98","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-15 11:40:41.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2fad2d0f27e64218959ac1f6bf571432\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5c94a1fd8691410ebbe706406e543db5 : SUCCESS in 1m 05s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/0fac6e071f644f7cb96527f4cbc7bff8 : SUCCESS in 1m 09s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"676499f2502595500f3c9cf8591c3ffa2a031e9d","revisions":{"676499f2502595500f3c9cf8591c3ffa2a031e9d":{"kind":"REWORK","_number":1,"created":"2023-09-06 15:30:03.000000000","uploader":{"_account_id":31746,"name":"Oleksandr Kozachenko","email":"okozachenko1203@gmail.com","username":"okozachenko"},"ref":"refs/changes/42/893942/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/42/893942/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/42/893942/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/42/893942/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/42/893942/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/42/893942/1"}}},"commit":{"parents":[{"commit":"5bf40c896a678cc11f4ab6aecdd8cc004fc809d9","subject":"Drop lower-constraints.txt and its testing","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/5bf40c896a678cc11f4ab6aecdd8cc004fc809d9"}]}],"author":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2023-04-26 20:55:04.000000000","tz":-240},"committer":{"name":"Oleksandr Kozachenko","email":"okozachenko1203@gmail.com","date":"2023-09-06 15:30:03.000000000","tz":0},"subject":"Fix incorrect removal of IP rules in the amphora","message":"Fix incorrect removal of IP rules in the amphora\n\nWhen plugging a member, an IP rule was deleted from the amphora. It\ntriggered connectivity issue when a client on subnet1 sends a request to\nthe VIP (on subnet2) while subnet1 is also plugged to the amp as a\nmember subnet.\nThe IP rules should be updated only when plugging/updating the VIP\ninterface. The JSON files that store the configuration of the interfaces\nnow contain the type of the interface (\u0027vip\u0027, \u0027lo\u0027 or \u0027member\u0027) and the\nrules update step is skipped when the type is not \u0027vip\u0027.\n\nCloses-Bug: #2016475\n\nChange-Id: I3fe9bb584c8522482e9d7289ae8ea1aab205dd4d\n(cherry picked from commit b3bf1779929eebdcf66d38cdd386fb5d785698f5)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/676499f2502595500f3c9cf8591c3ffa2a031e9d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/676499f2502595500f3c9cf8591c3ffa2a031e9d"}]},"branch":"refs/heads/stable/yoga"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
