)]}'
{"id":"openstack%2Foctavia~893943","triplet_id":"openstack%2Foctavia~stable%2Fxena~I3fe9bb584c8522482e9d7289ae8ea1aab205dd4d","project":"openstack/octavia","branch":"stable/xena","attention_set":{},"removed_from_attention_set":{"31746":{"account":{"_account_id":31746,"name":"Oleksandr Kozachenko","email":"okozachenko1203@gmail.com","username":"okozachenko"},"last_update":"2023-11-15 11:39:13.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I3fe9bb584c8522482e9d7289ae8ea1aab205dd4d","subject":"Fix incorrect removal of IP rules in the amphora","status":"MERGED","created":"2023-09-06 15:30:24.000000000","updated":"2023-11-15 11:40:44.000000000","submitted":"2023-11-15 11:39:13.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"893943","cherry_pick_of_change":881728,"cherry_pick_of_patch_set":3,"meta_rev_id":"e245b8169c4d06701177074376d0d0ae6a4d77f5","_number":893943,"virtual_id_number":893943,"owner":{"_account_id":31746,"name":"Oleksandr Kozachenko","email":"okozachenko1203@gmail.com","username":"okozachenko"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-11-15 11:39:13.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"recommended":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"all":[{"value":1,"date":"2023-09-06 21:11:26.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},{"value":2,"date":"2023-10-12 17:59:18.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":2,"date":"2023-11-15 08:13:19.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"all":[{"value":0,"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":1,"date":"2023-11-15 08:13:19.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-09-06 15:30:24.000000000","updated_by":{"_account_id":31746,"name":"Oleksandr Kozachenko","email":"okozachenko1203@gmail.com","username":"okozachenko"},"reviewer":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"state":"CC"},{"updated":"2023-09-06 17:21:20.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-09-06 21:11:26.000000000","updated_by":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"reviewer":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"state":"REVIEWER"},{"updated":"2023-10-12 17:59:18.000000000","updated_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"reviewer":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"state":"REVIEWER"},{"updated":"2023-11-15 08:13:19.000000000","updated_by":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"reviewer":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"state":"REVIEWER"}],"messages":[{"id":"ea3d65c08fba1502f1c0df29209ac986082dd839","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":31746,"name":"Oleksandr Kozachenko","email":"okozachenko1203@gmail.com","username":"okozachenko"},"date":"2023-09-06 15:30:24.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"488eff863a403377fbc738bae342369b69efc27f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-06 17:21:20.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/042a5db585d14ba7a3baec186c78f591\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/79fe75ff06ed401a81f2af57c23e029d : SUCCESS in 11m 13s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0c39c23669ca4307ad56d4192c4b2fde : SUCCESS in 5m 28s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/b4bfd2ad971b4e83b36a73dc6f06b629 : SUCCESS in 6m 00s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/62b9368a38fa4005a8ffa4c63e1a091c : SUCCESS in 5m 05s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/af91ffabee934a80a8490ea22ba0a953 : SUCCESS in 5m 06s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/38acc6afaadd466c909eb8bb42f16553 : SUCCESS in 14m 33s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/82e12d317ff544aa9614600246bb8fc4 : SUCCESS in 4m 49s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/92fcf832a1fb499c893852e78237c612 : SUCCESS in 7m 20s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/18ec685de35b4c918bfc4bcdd49ff43b : SUCCESS in 8m 18s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/25fba826d885484681f91617ea92c9f9 : SUCCESS in 4m 54s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/8f4b142f4d7241c79e68dc5ad3a31a3b : SUCCESS in 9m 07s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/064c6bc7c91c4c23859a22010d6e5b6c : SUCCESS in 1h 20m 00s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/08c586c2a29a40afa178ae3702febb11 : SUCCESS in 1h 45m 22s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/1f030f74096843b1a571cecadd7dc252 : SUCCESS in 34m 06s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/f2c57333abd148afb75af48fe58604fe : SUCCESS in 41m 56s\n- octavia-v2-dsvm-tls-barbican-fips https://zuul.opendev.org/t/openstack/build/0fefaad8654b4ec2a6599fde0d6a6a35 : FAILURE in 10m 09s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/bb1307f070c4413083ae97371ce84d55 : SUCCESS in 32m 12s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/a357f75984d6474c9e2e09fed3315378 : SUCCESS in 1h 48m 40s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/850d5c323c2b4165b035f16e34b54851 : FAILURE in 11m 20s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/24a4d65585854ff19ce8f9d69473b31c : FAILURE in 9m 33s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/5830edc710414a6787a1a30af3dd1c6c : SUCCESS in 1h 37m 11s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/57eb00ee156c4453bd4c4587c9752685 : SUCCESS in 1h 43m 39s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8 https://zuul.opendev.org/t/openstack/build/e3642db7346d455d97956104f9e9b04e : NODE_FAILURE Node request 300-0022160238 failed in 0s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"c54f47b9bc3613ddd24c95417804457ef8e1910b","author":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"date":"2023-09-06 21:11:26.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"7d47b95909995b7e18ec762e25bb246284fd89f2","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2023-10-12 17:59:18.000000000","message":"Patch Set 1: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"93ac75442cd58f347e980313e7e4a38e3df4a5ab","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2023-11-15 08:13:19.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"07ec99e8b6fd27fa4aaeef0d79f941cd5ad5e64a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-15 08:13:59.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"1aa7d43c283f9fe7a3772869758d018c00b74770","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-15 11:39:13.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/87145f85f82149e78a94e4ca907399f2\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e6f0d2a56f84473b910c3620df7273dd : SUCCESS in 4m 38s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/50bbacf9084a4df3b73d8ccd8695ef1b : SUCCESS in 4m 36s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/d8e6283b226349bea05c89263cea98c3 : SUCCESS in 5m 17s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/63f9125742c448c3bb34d0e395b64023 : SUCCESS in 12m 50s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/cf65905ca7b64cc59e858e01ffa73ff2 : SUCCESS in 9m 58s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/947f2e0c34524b51bf84ffaeede2bbad : SUCCESS in 6m 32s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/4c4caa104ef44685aa68d4a888e4a767 : SUCCESS in 7m 04s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/cc11ca59e69644fb8f674e6a6ee9024c : SUCCESS in 1h 22m 56s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/485fef7206f24784b8a1afcbf2a98089 : SUCCESS in 1h 52m 48s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/fd80a4794588448196fdf38f35d9f424 : SUCCESS in 38m 13s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/1114a40a36e44b77a69f97e87ab42def : SUCCESS in 43m 25s","accounts_in_message":[],"_revision_number":1},{"id":"ef80e6f7721d19890545bcf14ff44abc14e43d15","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-15 11:39:13.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"e245b8169c4d06701177074376d0d0ae6a4d77f5","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-15 11:40:44.000000000","message":"Patch Set 1:\n\nBuild failed (promote pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/78650b5dd5124402a6e73cd2d2906b13\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ff32078a926744f5ba283ff793274587 : SUCCESS in 1m 01s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/ec40d323347f42d9b4424433eb29fcb8 : FAILURE in 53s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"c943ed78d1b755cecd97721f7a25beb0e585130f","revisions":{"c943ed78d1b755cecd97721f7a25beb0e585130f":{"kind":"REWORK","_number":1,"created":"2023-09-06 15:30:24.000000000","uploader":{"_account_id":31746,"name":"Oleksandr Kozachenko","email":"okozachenko1203@gmail.com","username":"okozachenko"},"ref":"refs/changes/43/893943/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/43/893943/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/43/893943/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/43/893943/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/43/893943/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/43/893943/1"}}},"commit":{"parents":[{"commit":"774f92214390469dc6911c62d79b3f90f0c97a3f","subject":"Drop lower-constraints.txt and its testing","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/774f92214390469dc6911c62d79b3f90f0c97a3f"}]}],"author":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2023-04-26 20:55:04.000000000","tz":-240},"committer":{"name":"Oleksandr Kozachenko","email":"okozachenko1203@gmail.com","date":"2023-09-06 15:30:24.000000000","tz":0},"subject":"Fix incorrect removal of IP rules in the amphora","message":"Fix incorrect removal of IP rules in the amphora\n\nWhen plugging a member, an IP rule was deleted from the amphora. It\ntriggered connectivity issue when a client on subnet1 sends a request to\nthe VIP (on subnet2) while subnet1 is also plugged to the amp as a\nmember subnet.\nThe IP rules should be updated only when plugging/updating the VIP\ninterface. The JSON files that store the configuration of the interfaces\nnow contain the type of the interface (\u0027vip\u0027, \u0027lo\u0027 or \u0027member\u0027) and the\nrules update step is skipped when the type is not \u0027vip\u0027.\n\nCloses-Bug: #2016475\n\nChange-Id: I3fe9bb584c8522482e9d7289ae8ea1aab205dd4d\n(cherry picked from commit b3bf1779929eebdcf66d38cdd386fb5d785698f5)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/c943ed78d1b755cecd97721f7a25beb0e585130f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/c943ed78d1b755cecd97721f7a25beb0e585130f"}]},"branch":"refs/heads/stable/xena"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
