)]}'
{"id":"openstack%2Foctavia~894196","triplet_id":"openstack%2Foctavia~master~I700c65fb17bad28b2b922e03d9c94c4716de9cbe","project":"openstack/octavia","branch":"master","topic":"add_noop_cert_manager","attention_set":{},"removed_from_attention_set":{"31664":{"account":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"last_update":"2023-11-06 11:36:10.000000000","reason":"\u003cGERRIT_ACCOUNT_31664\u003e replied on the change","reason_account":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"}},"32761":{"account":{"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"},"last_update":"2023-11-06 16:05:38.000000000","reason":"Change was submitted"},"11628":{"account":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"last_update":"2023-11-03 18:13:00.000000000","reason":"\u003cGERRIT_ACCOUNT_11628\u003e replied on the change","reason_account":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}},"34429":{"account":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"last_update":"2023-11-06 09:52:58.000000000","reason":"\u003cGERRIT_ACCOUNT_34429\u003e replied on the change","reason_account":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}}},"hashtags":[],"change_id":"I700c65fb17bad28b2b922e03d9c94c4716de9cbe","subject":"Add Noop Certificate Manager","status":"MERGED","created":"2023-09-07 14:10:59.000000000","updated":"2023-11-06 16:06:52.000000000","submitted":"2023-11-06 16:05:38.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":27,"unresolved_comment_count":1,"has_review_started":true,"submission_id":"894196-add_noop_cert_manager","meta_rev_id":"d847fff9c70ede9d2221374692d5c185fe90c34b","_number":894196,"virtual_id_number":894196,"owner":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},{"value":0,"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-11-06 16:05:38.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"all":[{"value":0,"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},{"value":0,"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"},{"value":2,"date":"2023-11-03 18:13:00.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2023-11-06 09:52:58.000000000","_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"all":[{"value":0,"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},{"value":0,"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2023-11-06 09:52:58.000000000","_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},{"value":0,"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},{"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"},{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-09-07 17:15:11.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-09-19 16:16:40.000000000","updated_by":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"reviewer":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"state":"REVIEWER"},{"updated":"2023-10-12 17:44:56.000000000","updated_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"reviewer":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"state":"REVIEWER"},{"updated":"2023-10-30 01:24:14.000000000","updated_by":{"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"},"reviewer":{"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"},"state":"REVIEWER"}],"messages":[{"id":"51b7f7449823f14348623d1c6aa535886e14a169","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"date":"2023-09-07 14:10:59.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"d182dcc6e101d9ff2570033e09bd91753a3a7230","author":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"date":"2023-09-07 16:26:27.000000000","message":"Patch Set 1: Workflow-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"0600577bcf5130e647c68e34c0bdf20157cf1b2d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-07 17:15:11.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5c67217e95de4b77b49ede08fe60248e\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/68c2b2519e1943e5b191aaca92463693 : SUCCESS in 2m 10s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/4d3c196c73b14b0d9224f0b2f6a54294 : SUCCESS in 11m 23s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/db6c4a1a734f426185852c57a99e2d75 : SUCCESS in 5m 24s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/ca9ee5620273464aae21b632eec943b7 : SUCCESS in 4m 41s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/1c3d1849bc814eb2878f6e0d9c33e3d8 : SUCCESS in 5m 19s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/23f258851a2b4d98bfc3f05c0f5233b8 : SUCCESS in 5m 11s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/b07a8a1fde904c34a86e3d117255eb3d : SUCCESS in 5m 51s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8535e9d769754ba58ff34bc805092e3c : SUCCESS in 11m 40s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/b8744342fbd143d09f6ed2e1d8607bf9 : SUCCESS in 4m 55s\n- octavia-tox-py39-tips https://zuul.opendev.org/t/openstack/build/30354e1fac1341b1a5e4d8e98c42a0ce : SUCCESS in 5m 21s\n- octavia-tox-functional-py39-tips https://zuul.opendev.org/t/openstack/build/089581e4ed404835ac2f72ca28f12e7f : SUCCESS in 9m 15s\n- octavia-tox-functional-py39-sqlalchemy-tips https://zuul.opendev.org/t/openstack/build/849b7988a1654db299f7af46d32eef0d : SUCCESS in 9m 56s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/6b781b2f0bf7449b9ab0445c53002f99 : SUCCESS in 5m 19s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/a16b3d28ba9f43e78ee58dceb30cd861 : SUCCESS in 8m 41s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/522e82237d6a481eb0a95f3f4ef215f3 : SUCCESS in 7m 55s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/5045e60fe34d47f7bf63d1ab7291017e : SUCCESS in 7m 04s (non-voting)\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/ad6ec04e4cce41bd85ab126f0da5ee98 : SUCCESS in 1h 13m 40s\n- octavia-v2-dsvm-scenario-traffic-ops https://zuul.opendev.org/t/openstack/build/55966b8929e2459994b25c685fcd506e : SUCCESS in 2h 00m 11s\n- octavia-v2-dsvm-scenario-non-traffic-ops https://zuul.opendev.org/t/openstack/build/ee871010fcad4527ae04b0d86fe78165 : SUCCESS in 1h 23m 39s\n- octavia-v2-dsvm-scenario-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/df14ee3e4aa44faa9e7e581a59c9d2bc : SUCCESS in 2h 01m 17s\n- octavia-v2-dsvm-scenario-non-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/e17719b30d314ec18e60f2e83fccafd1 : SUCCESS in 1h 32m 09s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/2cb28a7be5484ae38b2f2e00c89a2f52 : SUCCESS in 37m 51s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/8a16f49b780d4f809429436952c2f24f : SUCCESS in 40m 36s\n- octavia-v2-dsvm-tls-barbican-fips https://zuul.opendev.org/t/openstack/build/16417c356f7746d6901afcb7732b450a : SUCCESS in 1h 16m 51s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/7b5f7803653a41d0a119007ab0cf6535 : SUCCESS in 32m 35s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/bc5fdd6a01eb45b584e7cad136396f1e : TIMED_OUT in 2h 32m 55s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/0f6de8b4ba504e029e72b881ae720b83 : SUCCESS in 3h 02m 35s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/e76bdcc9bfa54c8dbcba2a16ffb8fa69 : SUCCESS in 2h 35m 33s (non-voting)\n- octavia-v2-dsvm-scenario-centos-9-stream https://zuul.opendev.org/t/openstack/build/783b86e66b474f508f91eb5312d306f0 : SUCCESS in 2h 34m 38s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"554c32b32223c2e80b0584de0592849dad15e92d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"date":"2023-09-11 13:03:09.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified+1\n* Workflow-1\n","accounts_in_message":[],"_revision_number":2},{"id":"ba7e7a9e9cd7871537dc658e647cfd24585f71b4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-11 15:13:28.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8128233b097346dc8352255e30ab2942\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/29249a5983c84d32b0dc8dab283b621c : SUCCESS in 2m 06s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/d11c764c510440db8684b55249efd31c : SUCCESS in 8m 26s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/57f1d40c6bd5452ba1f60dfe214de5eb : SUCCESS in 4m 35s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/a00f8f5192304daf82841e118a0ac4a3 : SUCCESS in 5m 23s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/b7fb1c0f68cc44baa38483c737af67cc : SUCCESS in 5m 09s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/f57e889925e844eb90a00fd5e76165ca : SUCCESS in 4m 26s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/5f920aaeee2c4b31a23c08d7ff1a3594 : SUCCESS in 4m 30s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c82b1420e5b14ed78cdb4022e451e95c : SUCCESS in 12m 32s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/dd5512e319e5445fb7dfc40188fe5b52 : SUCCESS in 5m 19s\n- octavia-tox-py39-tips https://zuul.opendev.org/t/openstack/build/60344f838b274135ad28f2f528938356 : SUCCESS in 5m 32s\n- octavia-tox-functional-py39-tips https://zuul.opendev.org/t/openstack/build/0a64824d357245678a6df4d2f0184c16 : SUCCESS in 8m 42s\n- octavia-tox-functional-py39-sqlalchemy-tips https://zuul.opendev.org/t/openstack/build/57ac7500b23d4eb08f1325998a4bbebe : SUCCESS in 10m 15s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/b28721939c6c466dbc2d647324110023 : SUCCESS in 5m 28s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/5b4efbce49f54d0b84e182f5f0716c33 : SUCCESS in 7m 23s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/df11c24df8294f41bb4074549899558e : SUCCESS in 9m 42s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/ac2c8c6f6dc9499bb81c47c58ed5ff17 : SUCCESS in 7m 38s (non-voting)\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/3c92eee924824f458d21c2bb467ee3be : SUCCESS in 1h 45m 25s\n- octavia-v2-dsvm-scenario-traffic-ops https://zuul.opendev.org/t/openstack/build/c2598a2be52c40618b98a2f73ed6648d : SUCCESS in 1h 47m 43s\n- octavia-v2-dsvm-scenario-non-traffic-ops https://zuul.opendev.org/t/openstack/build/e072cef99c8c4216aecc8abb7fce54a3 : SUCCESS in 1h 17m 50s\n- octavia-v2-dsvm-scenario-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/db78e53221014c728a748cb79698cee1 : SUCCESS in 2h 05m 45s\n- octavia-v2-dsvm-scenario-non-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/91279b8ad7fb47ff9b6ddc90751f412f : SUCCESS in 1h 21m 29s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/968be21e1b424449b5c7cf1d09386dbe : SUCCESS in 44m 54s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/5116b5315a614d2a9f84553fe6eb3779 : SUCCESS in 57m 05s\n- octavia-v2-dsvm-tls-barbican-fips https://zuul.opendev.org/t/openstack/build/e4dadb1a9e8f4f6f967cb45194163b29 : SUCCESS in 1h 12m 46s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/3c25e338bea94221822afd93d6929a07 : SUCCESS in 44m 11s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/3f5996e8702f40248b29b6471400c3fa : SUCCESS in 1h 59m 10s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/45ca9601d2e94fd0a6a28f586a9270db : SUCCESS in 2h 08m 26s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/3b0e236d04ed4a179fdf1bc531556f97 : SUCCESS in 1h 33m 34s (non-voting)\n- octavia-v2-dsvm-scenario-centos-9-stream https://zuul.opendev.org/t/openstack/build/d1ef089b5a974064905d6b86b46c1107 : FAILURE in 2h 07m 24s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"6b1bbdd977a6b69d961e82bfb79e9f91b5367a57","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2023-09-19 16:16:40.000000000","message":"Patch Set 2: Code-Review-1\n\n(3 comments)","accounts_in_message":[],"_revision_number":2},{"id":"32176912759ef2dc2625d946e4036d23167dc5c8","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"date":"2023-09-25 13:19:15.000000000","message":"Patch Set 3: Published edit on patch set 2.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":3},{"id":"159ea0c8d32715f0520849ce63901897c3e4e488","author":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"date":"2023-09-25 13:19:37.000000000","message":"Patch Set 3:\n\n(2 comments)","accounts_in_message":[],"_revision_number":3},{"id":"82767f0745e5bc03675e76f84f92e02e937aafd7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-25 15:51:07.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/bc2ab51cd4f5495d99940d4a90fe2c69\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/7e62ba0ed2d64ea3a4c8e8159df6e89b : SUCCESS in 1m 45s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/af9f0725fc2542359ca195f0e77c94cc : SUCCESS in 10m 19s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/185e1c7f375241ed93d2d73b974edef9 : SUCCESS in 5m 34s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/fa0ecdeadafc49a680b4dc202d449742 : SUCCESS in 4m 51s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/035313030ed6408489cc349082289081 : SUCCESS in 4m 43s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/2500006c1f7140e7b485b69d90821975 : SUCCESS in 4m 35s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/e9acb40e9af1435dbd8ee907ca335085 : SUCCESS in 5m 48s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/47676616e55a4eb988eab7e050de8865 : SUCCESS in 12m 51s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/73fae571c2d64db0a55ba38ac3a8a0d2 : SUCCESS in 4m 42s\n- octavia-tox-py39-tips https://zuul.opendev.org/t/openstack/build/8e806d7e04ca413898211cf2014a8b37 : SUCCESS in 5m 44s\n- octavia-tox-functional-py39-tips https://zuul.opendev.org/t/openstack/build/601489898f9346b8b61a92dd9c045d3a : SUCCESS in 7m 12s\n- octavia-tox-functional-py39-sqlalchemy-tips https://zuul.opendev.org/t/openstack/build/31711df957d64f7c972d2181ecbcc591 : SUCCESS in 9m 32s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/1e059f65575d4cc18d08da7c3bc1f25e : SUCCESS in 4m 53s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/318a68de061f4b5b96e4c1d39517cdaf : SUCCESS in 7m 24s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/810c2c4094b941f1b9dbfe6ef5211fe4 : SUCCESS in 7m 59s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/f4f32672dfdd43aba8732e10c1619966 : SUCCESS in 8m 29s (non-voting)\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/09e415b4671f40ba855ed3395654f558 : SUCCESS in 1h 31m 28s\n- octavia-v2-dsvm-scenario-traffic-ops https://zuul.opendev.org/t/openstack/build/a7d006efca1544288a7d7754c638b2ed : SUCCESS in 2h 08m 02s\n- octavia-v2-dsvm-scenario-non-traffic-ops https://zuul.opendev.org/t/openstack/build/84b1396dc2d94eaa9f9d657cd8071352 : SUCCESS in 1h 17m 32s\n- octavia-v2-dsvm-scenario-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/2a3c3a754fe34ea5b6db6ec7439e5a9a : SUCCESS in 2h 13m 41s\n- octavia-v2-dsvm-scenario-non-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/4cb3b17605c041e79fdd2798b22e31d6 : SUCCESS in 1h 19m 30s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/bc37117557b14f2db928a34a8fa41b75 : SUCCESS in 39m 35s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/43a12f5a01c640169eced17715cf5184 : SUCCESS in 41m 52s\n- octavia-v2-dsvm-tls-barbican-fips https://zuul.opendev.org/t/openstack/build/ae705106b18b4d39ac3f85e1e64b9555 : SUCCESS in 59m 50s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/1710e81ce15840ffad69b896b7342c59 : SUCCESS in 35m 08s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/2c1fdce5560545af8a733fd602d21d86 : SUCCESS in 2h 23m 49s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/efdc41370aa1480cbab1f26c337b4de1 : SUCCESS in 2h 28m 31s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/b23f22a32b5b43c1983c2c2c1a44aea1 : SUCCESS in 2h 27m 16s (non-voting)\n- octavia-v2-dsvm-scenario-centos-9-stream https://zuul.opendev.org/t/openstack/build/d0df396047384610b28523638e0ce105 : SUCCESS in 2h 14m 18s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"68a29213cc2474f9f6f1a544d67e1741cc82179c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"date":"2023-10-04 16:21:43.000000000","message":"Patch Set 4: Published edit on patch set 3.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":4},{"id":"71950db4a41b14bf15af13b9e9de0b8d31535bc4","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2023-10-04 16:23:28.000000000","message":"Patch Set 4: Code-Review+2","accounts_in_message":[],"_revision_number":4},{"id":"f9d970b466b1566ecf7d8f0a228b148811c6a1b1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-04 19:38:41.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/aadb589fdd194616ae63e1d0f7839cd3\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/ce4d5a3eefc54028ba3cfaeec9da54e3 : SUCCESS in 3m 11s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/67454b731abe417498842db5c7c24bb4 : SUCCESS in 12m 57s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d7ca80c92dbd4290b7f22b8d7fc67a7e : SUCCESS in 5m 29s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/5bfe99a55a124357aec9f3688fda1995 : SUCCESS in 6m 26s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/4bc64b1b7124443bb2a1cd28d129001d : SUCCESS in 6m 05s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/edecd76c03f143aba4de414cddacd929 : SUCCESS in 5m 23s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/0c093f6b25e84abd93bcbb9f47d43faf : SUCCESS in 5m 55s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3ff1e49ea2814c679ad9941f0aae3b25 : SUCCESS in 11m 36s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/ff2efca3b7e448bd8216ac797564c63e : SUCCESS in 5m 47s\n- octavia-tox-py39-tips https://zuul.opendev.org/t/openstack/build/c085bb0d5fdd4a04b2258db4b5641928 : SUCCESS in 5m 37s\n- octavia-tox-functional-py39-tips https://zuul.opendev.org/t/openstack/build/c567cea463ac459fbfed111e2f850456 : SUCCESS in 10m 04s\n- octavia-tox-functional-py39-sqlalchemy-tips https://zuul.opendev.org/t/openstack/build/a11c59aa54eb4fcf847ab7119cee1da5 : SUCCESS in 10m 40s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/e55cc9e3e73f4a77a2f993c54898186e : SUCCESS in 6m 05s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/6b9d1754aa97414499db136541d392e3 : SUCCESS in 8m 06s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/4ceef92b5b944b55b3a7bfb9059fe92c : SUCCESS in 6m 44s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/94a16f7950dc45e595720bbcd391c29c : SUCCESS in 9m 21s (non-voting)\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/ff1de9f410c149a5bcec95e127c126fe : SUCCESS in 1h 35m 34s\n- octavia-v2-dsvm-scenario-traffic-ops https://zuul.opendev.org/t/openstack/build/60aac0beab624cec895bb83ae989a28a : SUCCESS in 1h 32m 33s\n- octavia-v2-dsvm-scenario-non-traffic-ops https://zuul.opendev.org/t/openstack/build/1da56763c8a64605bd8e1d1f70c0ef93 : SUCCESS in 1h 14m 42s\n- octavia-v2-dsvm-scenario-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/cad54a6da2524470baa92d86e5e0714b : SUCCESS in 1h 42m 32s\n- octavia-v2-dsvm-scenario-non-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/c008b3d2d1814cf2af156c9f32b4054d : SUCCESS in 1h 17m 51s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/01d8e51ade4c491bba709b241b92408d : SUCCESS in 46m 41s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/063f95356498479cb7cb6c0b365f5425 : SUCCESS in 41m 18s\n- octavia-v2-dsvm-tls-barbican-fips https://zuul.opendev.org/t/openstack/build/953dd9685dca4034a509f02332fdb699 : SUCCESS in 1h 07m 41s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/062df1a09da44402840481dec1490af5 : SUCCESS in 35m 21s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/40416d5982d941218e5714bdf68e1fff : TIMED_OUT in 2h 33m 25s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/71e1cfde570d41869914286399617b93 : TIMED_OUT in 3h 03m 19s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/a8af85c3b7e144e2b5552e2b3c9505d5 : TIMED_OUT in 2h 33m 04s (non-voting)\n- octavia-v2-dsvm-scenario-centos-9-stream https://zuul.opendev.org/t/openstack/build/78b11cff0bce4b8081422d24a82209d3 : SUCCESS in 2h 33m 49s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"1fe4660044ce9a54fd38d26c47e86ed007bd2cf5","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2023-10-12 17:44:56.000000000","message":"Patch Set 4: Code-Review-1\n\n(5 comments)","accounts_in_message":[],"_revision_number":4},{"id":"e9931ef9fe47c82ad8ea2e4cfd51a5071a368b94","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"date":"2023-10-18 14:27:31.000000000","message":"Uploaded patch set 5.\n\nOutdated Votes:\n* Code-Review+2, Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":5},{"id":"e7010395505419c84d246b94f605b170a197ad94","author":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"date":"2023-10-18 14:28:30.000000000","message":"Patch Set 5: Workflow-1","accounts_in_message":[],"_revision_number":5},{"id":"f6f514e220c9d7af19b760a7ee7245a57ea662f5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-18 17:03:12.000000000","message":"Patch Set 5: Verified-1\n\n(7 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/39a29f0df8584cf1bcd63acc654059d1\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/e8b25e186917401f827f035370567264 : SUCCESS in 2m 07s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/f67a8c06ac4d4b179faacec652aae1db : SUCCESS in 11m 30s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4b0b7ef7d0ef4e6fb00474ac6da435ee : FAILURE in 4m 40s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/fc808f64bd994704b11e18f5a9d7e367 : SUCCESS in 4m 55s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/3a0bc754523f40d6a7278b42e3826378 : SUCCESS in 4m 53s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/f8c9c61b8248472f85f5eb9461b0e20c : SUCCESS in 4m 56s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/dfd69e4d26944726a3b7d81078bcab61 : SUCCESS in 5m 21s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/763cfc38cc0446ae9c9e207fd833acc2 : SUCCESS in 11m 30s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/818d115b3ad044d5a39cc310be7608d7 : SUCCESS in 5m 18s\n- octavia-tox-py39-tips https://zuul.opendev.org/t/openstack/build/0f36361f01b9428382fc61e3300f76ea : SUCCESS in 5m 21s\n- octavia-tox-functional-py39-tips https://zuul.opendev.org/t/openstack/build/f4dd7a50214e4e2c992a233017ba0a66 : SUCCESS in 8m 42s\n- octavia-tox-functional-py39-sqlalchemy-tips https://zuul.opendev.org/t/openstack/build/8a651e48151e495881aac39b256db326 : SUCCESS in 11m 39s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/4677cea92bd240ec93296bf473a1445e : SUCCESS in 6m 13s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/acbd187d16da408999d51346ded8500b : SUCCESS in 7m 58s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/817faad905044e52b8bfa33e65aa2897 : SUCCESS in 8m 21s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/aa64acad8395405eb19c46baab75321c : SUCCESS in 8m 08s (non-voting)\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/50548e558c1842f6a11789ed569e73fb : SUCCESS in 1h 26m 58s\n- octavia-v2-dsvm-scenario-traffic-ops https://zuul.opendev.org/t/openstack/build/9983f728cd5e4fbcaad854a258f88ce1 : SUCCESS in 2h 09m 56s\n- octavia-v2-dsvm-scenario-non-traffic-ops https://zuul.opendev.org/t/openstack/build/cc28328b6acb4ad0903dcd746d197cfd : SUCCESS in 1h 18m 08s\n- octavia-v2-dsvm-scenario-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/87af078ea69940fca443cf2107d61123 : SUCCESS in 1h 57m 58s\n- octavia-v2-dsvm-scenario-non-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/c07b52f27baf4c35a482cc119b5958f0 : SUCCESS in 1h 14m 48s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/a16f5b88192d45ed9360d316669b58c7 : SUCCESS in 39m 43s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/5d5afe48d7b94e4db5ead8825a01c5d5 : SUCCESS in 55m 10s\n- octavia-grenade-skip-level https://zuul.opendev.org/t/openstack/build/94c84bbca23c41bdaa42cce2819f52e7 : SUCCESS in 55m 59s (non-voting)\n- octavia-v2-dsvm-tls-barbican-fips https://zuul.opendev.org/t/openstack/build/823c9367c5c245a3ac8a3388f0138a4d : SUCCESS in 1h 17m 57s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/0d247d71350c40a494b8c1ebd2f87815 : SUCCESS in 33m 48s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/ad11d537f38743d29b043dc4f17a43f5 : FAILURE in 2h 28m 03s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/347bc9f195774eeda2f93cde25061265 : SUCCESS in 2h 17m 56s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/3c939f4f39aa4d9c897dce776a841df9 : SUCCESS in 2h 03m 17s (non-voting)\n- octavia-v2-dsvm-scenario-centos-9-stream https://zuul.opendev.org/t/openstack/build/252212b22d924a81b8f0d097d1391c87 : TIMED_OUT in 2h 33m 35s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"c778cccd2a9702c79d9a588f16085226cfafc21f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"date":"2023-10-19 08:54:00.000000000","message":"Uploaded patch set 6.\n\nOutdated Votes:\n* Verified-1\n* Workflow-1\n","accounts_in_message":[],"_revision_number":6},{"id":"a142e976d8f0635dfd70667082f9113c8d489469","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-19 11:59:24.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/46493db316ad46189e59ba5d835c01f6\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/359c788a5fae490ea8a54a096a030723 : SUCCESS in 2m 39s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/5d54c0ea96e542b5b8a72a9a7ac87386 : SUCCESS in 8m 38s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/69331951f6374dedb29b340131145621 : SUCCESS in 4m 35s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/013e5fd1efc647e28e430d0d2dc30d6e : SUCCESS in 4m 52s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/fbc90f77941341f3b21f25da7e49eb0c : SUCCESS in 5m 12s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ab731f9115d244be863ae23decc39222 : SUCCESS in 13m 47s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/c8b0965b693345a597d13be9872bbf59 : SUCCESS in 5m 04s\n- octavia-tox-py39-tips https://zuul.opendev.org/t/openstack/build/8429d3c7079b48b4bad5019885a34ccc : SUCCESS in 6m 01s\n- octavia-tox-functional-py39-tips https://zuul.opendev.org/t/openstack/build/5185cd832542497d8974e2246b8a868d : SUCCESS in 8m 26s\n- octavia-tox-functional-py39-sqlalchemy-tips https://zuul.opendev.org/t/openstack/build/6bd10a371ee441788c5bfbcfd7180313 : SUCCESS in 10m 46s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/3db99658281345c780827d448438e676 : SUCCESS in 5m 27s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/ae45b7d262c24d7993adf7b9f7312990 : SUCCESS in 10m 01s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/19b7149df37d4e7381539213984cb096 : SUCCESS in 8m 17s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/56049e9382444c7f8c148eca9cfae974 : SUCCESS in 8m 07s (non-voting)\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/5cb9168b8cb84e3b85ede21167e7a68f : SUCCESS in 1h 23m 41s\n- octavia-v2-dsvm-scenario-traffic-ops https://zuul.opendev.org/t/openstack/build/3d9b4aeab5eb4e28be824eca3dda38ce : SUCCESS in 1h 49m 55s\n- octavia-v2-dsvm-scenario-non-traffic-ops https://zuul.opendev.org/t/openstack/build/0319db1ec30e42c68de87e8457c78d44 : SUCCESS in 1h 13m 38s\n- octavia-v2-dsvm-scenario-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/12103a03b57443a295c8a3024b71fbbb : SUCCESS in 1h 45m 19s\n- octavia-v2-dsvm-scenario-non-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/91eec21b22284f9d950b92e43ab5fae3 : SUCCESS in 1h 25m 22s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/2698daeeb6224686b067ecb51fc391ae : SUCCESS in 41m 45s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/a09a67052665467f9b7c06c09445f712 : SUCCESS in 53m 15s\n- octavia-grenade-skip-level https://zuul.opendev.org/t/openstack/build/7f31ff6fb6ca4fea8394700b4e7c5bfe : SUCCESS in 53m 16s (non-voting)\n- octavia-v2-dsvm-tls-barbican-fips https://zuul.opendev.org/t/openstack/build/0a9b851b5bf844a19d5a81da9dbd7afe : SUCCESS in 1h 08m 34s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/854fa2207240481691833838bc03425c : SUCCESS in 48m 41s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/645d919b64f14198821943168226421b : SUCCESS in 2h 20m 06s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/944d14ce66dc4dac827b447ef9a409e8 : TIMED_OUT in 3h 03m 29s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/a0395ac89a704b3c8739be39748fb79e : SUCCESS in 2h 26m 05s (non-voting)\n- octavia-v2-dsvm-scenario-centos-9-stream https://zuul.opendev.org/t/openstack/build/8cc86a1e86104a11b8d2ee98b6914529 : TIMED_OUT in 2h 33m 41s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"13ffa94df90d39e53f2e934a3e5cc88ef9d3b4c6","author":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"date":"2023-10-19 12:15:07.000000000","message":"Patch Set 6:\n\n(6 comments)","accounts_in_message":[],"_revision_number":6},{"id":"7e506f8e81be2c70591a651f4144b0e2bcd61266","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"date":"2023-10-20 16:33:30.000000000","message":"Uploaded patch set 7.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":7},{"id":"622115f81f107ec53ca50de00d59a7113fb5fe08","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-20 19:25:12.000000000","message":"Patch Set 7: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/7dadd01128734f47be10966a0c4fae77\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/8dc962f1ab814dbda483dc591ed670fd : SUCCESS in 3m 10s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/892251d0e09749e8a0d98808ac292d1a : FAILURE in 6m 29s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/176ab4417354405a9261ef04805449b0 : SUCCESS in 6m 04s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/8c3f9ca06ba84693a4a2365cedcb484b : FAILURE in 6m 36s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/102a1a2a166142dda8665efbb81cb58f : FAILURE in 4m 45s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2f458f3cf7124acd81174d6fcce4cbd7 : SUCCESS in 14m 16s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/525aea2091234e4d8a77372e563d4eb9 : SUCCESS in 7m 27s\n- octavia-tox-py39-tips https://zuul.opendev.org/t/openstack/build/21bee1c69b87425390234de7f5e465b3 : FAILURE in 4m 51s\n- octavia-tox-functional-py39-tips https://zuul.opendev.org/t/openstack/build/b8de110fdd0f47248109d6e2ee28e96d : SUCCESS in 7m 15s\n- octavia-tox-functional-py39-sqlalchemy-tips https://zuul.opendev.org/t/openstack/build/ace0714710ec4632bf572e1115e3d2af : SUCCESS in 11m 05s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/46b2487abe7f478ab7d34af5c734f572 : SUCCESS in 4m 24s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/9441b8ddcb4a4e9b9387fc9bb679bd26 : SUCCESS in 6m 17s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/67dca5daa566474084203f8cf127e1fb : SUCCESS in 6m 48s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/d4512e3571f04f628a55d02f405c18fd : SUCCESS in 6m 43s (non-voting)\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/634d8965a7654c98892951f5d59ade1e : SUCCESS in 1h 12m 12s\n- octavia-v2-dsvm-scenario-traffic-ops https://zuul.opendev.org/t/openstack/build/b239d8dcad95418b8011f835af286f56 : FAILURE in 22m 32s\n- octavia-v2-dsvm-scenario-non-traffic-ops https://zuul.opendev.org/t/openstack/build/2ef3995540434b3b9f7524c5efa5c237 : SUCCESS in 1h 05m 28s\n- octavia-v2-dsvm-scenario-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/366775649b8543788a4981bc3503a91f : SUCCESS in 1h 44m 01s\n- octavia-v2-dsvm-scenario-non-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/c5da42e219644b628ef303cd0c64c656 : FAILURE in 24m 36s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/b6f50f4915674f3a9abf7c9ce00d2daa : SUCCESS in 36m 54s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/9709b29f631e4fc092c85e60ed2cba28 : SUCCESS in 47m 20s\n- octavia-grenade-skip-level https://zuul.opendev.org/t/openstack/build/aef5f038599a44bc96b95aa3e43fdfe5 : SUCCESS in 51m 15s (non-voting)\n- octavia-v2-dsvm-tls-barbican-fips https://zuul.opendev.org/t/openstack/build/96829a0e8b8e4aa990b1320ea2d70263 : SUCCESS in 1h 10m 24s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/2bff9970df404535b4cd8e36387cdcc5 : SUCCESS in 40m 32s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/bdabba405d2b4c91ba13ae2c6712dbca : TIMED_OUT in 2h 33m 41s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/bbb70d1a77b740558cb501abdbdb369d : SUCCESS in 2h 47m 30s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/e82ec5af7eb242ccac75c9d06f30c9ca : SUCCESS in 2h 28m 42s (non-voting)\n- octavia-v2-dsvm-scenario-centos-9-stream https://zuul.opendev.org/t/openstack/build/e3bf72176020406cba52858b1df51237 : SUCCESS in 2h 27m 45s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"4ace3e528eba2714c9181cdb0e45a212d5abd6c6","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"date":"2023-10-23 09:10:15.000000000","message":"Uploaded patch set 8.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":8},{"id":"9c068ef38f980c1e7639f2210dc3c7fee1ac26c4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-23 12:11:25.000000000","message":"Patch Set 8: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6b6fbe0bd8aa49fa9c444af673500c91\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/28d9e43a33fe4c4aa5b70aec91168744 : SUCCESS in 1m 44s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/ced4ec093e414c6fa76becaa5b29a8c5 : SUCCESS in 10m 19s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ea8d49f920d4476b824302adb80ae55f : SUCCESS in 5m 14s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/3cd0dd647a6e4dcfb6442649e7e1f81f : SUCCESS in 5m 01s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/72cd7fa7f6f641968696adc20bbd588d : SUCCESS in 6m 39s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ea123615656140ccbba7cb725c852a5c : SUCCESS in 12m 43s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/3fd7c5e7fda9454ea89547d5d71f84e0 : SUCCESS in 4m 59s\n- octavia-tox-py39-tips https://zuul.opendev.org/t/openstack/build/d40990b8196c4eef9bb4a1970f7c79f8 : SUCCESS in 5m 22s\n- octavia-tox-functional-py39-tips https://zuul.opendev.org/t/openstack/build/9f068521c6014872a130c64a8be81403 : SUCCESS in 7m 15s\n- octavia-tox-functional-py39-sqlalchemy-tips https://zuul.opendev.org/t/openstack/build/dbbc1ec81197410abc7df144a842ba1d : SUCCESS in 9m 53s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/8e9b2aa78b4a44be9e6729029ddd0a6f : SUCCESS in 4m 29s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/3a83a0886e6e4a658e18a9c985a6fdf8 : SUCCESS in 5m 33s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/aca58d87bdec41ec926a3cf22f6cd67f : SUCCESS in 7m 47s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/34eb238c3a854aa899441604e8a31db1 : SUCCESS in 6m 21s (non-voting)\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/bbd812e79414416e970a47196bf530b2 : SUCCESS in 1h 06m 09s\n- octavia-v2-dsvm-scenario-traffic-ops https://zuul.opendev.org/t/openstack/build/564af56e32974e49b7acffa5e77ac971 : SUCCESS in 1h 40m 30s\n- octavia-v2-dsvm-scenario-non-traffic-ops https://zuul.opendev.org/t/openstack/build/64efd1dc1f814e32a465c09c62d6d90b : SUCCESS in 58m 14s\n- octavia-v2-dsvm-scenario-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/15f3727d6b7b413298cdabfd08b9104f : SUCCESS in 1h 41m 12s\n- octavia-v2-dsvm-scenario-non-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/9dc463afecfa426f9f07b6bf9d5f6336 : SUCCESS in 1h 13m 00s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/1bbf455f6e09454a87be7f517db927c7 : SUCCESS in 36m 47s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/9243a835116144e1b9ec49fffee50060 : SUCCESS in 39m 43s\n- octavia-grenade-skip-level https://zuul.opendev.org/t/openstack/build/9d9d2e8efc09411db3ea3601669b0022 : SUCCESS in 39m 09s (non-voting)\n- octavia-v2-dsvm-tls-barbican-fips https://zuul.opendev.org/t/openstack/build/03dc7daef820430b8e941c7c2b05fdc4 : SUCCESS in 1h 08m 23s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/7102574ea41241b4811282d268a2473f : SUCCESS in 38m 44s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/07ff1fc9501940ee9b1558d5a6129b0d : SUCCESS in 2h 03m 09s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/f260852579394515884535f62f4da122 : SUCCESS in 2h 44m 38s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/0cb810ad8da94406919346afc0563f43 : SUCCESS in 1h 55m 49s (non-voting)\n- octavia-v2-dsvm-scenario-centos-9-stream https://zuul.opendev.org/t/openstack/build/f14e52c978b246389b7e3041e53b9ea3 : TIMED_OUT in 2h 34m 49s (non-voting)","accounts_in_message":[],"_revision_number":8},{"id":"e726f8d448d7aebc922f5e936a535cc00ff47bde","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2023-10-23 13:35:51.000000000","message":"Patch Set 8: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":8},{"id":"b3d0f79ac1371814e530b7be492205b259f7af3c","author":{"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"},"date":"2023-10-30 01:24:14.000000000","message":"Patch Set 8: Code-Review+1","accounts_in_message":[],"_revision_number":8},{"id":"7970c6a6bc460703cd27fb5a73df786cd1a62db7","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2023-11-01 18:08:59.000000000","message":"Patch Set 8: Code-Review+2\n\n(3 comments)","accounts_in_message":[],"_revision_number":8},{"id":"908138d909e4bc2e898d9a89e812a1ebf3557c9c","author":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"date":"2023-11-02 10:59:16.000000000","message":"Patch Set 8:\n\n(1 comment)","accounts_in_message":[],"_revision_number":8},{"id":"488564265159227b12658573ebb417b11236a5b0","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"date":"2023-11-02 11:07:59.000000000","message":"Uploaded patch set 9.\n\nOutdated Votes:\n* Code-Review+1, Code-Review+2 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":9},{"id":"6c4c55cff3be60f5e784207f737153b14652065f","author":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"date":"2023-11-02 11:09:12.000000000","message":"Patch Set 9:\n\n(2 comments)","accounts_in_message":[],"_revision_number":9},{"id":"efa95ead0507a1bb1b1ea7fe747557f90ee8d8c9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-02 13:44:16.000000000","message":"Patch Set 9: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/fe08d5bb3d2449a893704910637aabad\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/7f9554c2b6764d6f9ccb8aed8d5a0407 : SUCCESS in 2m 32s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/6c880c1d511c4c788d5187b47a7b1067 : SUCCESS in 10m 54s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/64f7f304ee3549ff8396b32e803ac951 : SUCCESS in 5m 04s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/25647f9a899245b2873b65bf4827d6db : SUCCESS in 4m 52s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/46083d1b4cba46e2bb89e926c9770814 : SUCCESS in 4m 44s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a7be43e755e64a0195c21ab30c77dbd6 : SUCCESS in 11m 18s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/bddd4a8a60b44fffb91ee9231d7b724e : SUCCESS in 5m 13s\n- octavia-tox-py39-tips https://zuul.opendev.org/t/openstack/build/d0d583ae95ca4d33809c95fb52dea52b : SUCCESS in 5m 25s\n- octavia-tox-functional-py39-tips https://zuul.opendev.org/t/openstack/build/3137a0c4f54c4abfb8a684e31439e972 : SUCCESS in 7m 33s\n- octavia-tox-functional-py39-sqlalchemy-tips https://zuul.opendev.org/t/openstack/build/c2a33d46ec414c20b2d6af37ca77a4d4 : SUCCESS in 9m 30s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/72f6085a42024d30af3c30a7b17b075e : SUCCESS in 4m 25s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/dc7a058ffa584ed38ecea898ba60b4b0 : SUCCESS in 7m 20s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/c6b80e07dc2f455ba25c875c17bc306c : SUCCESS in 8m 15s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/1e31aed440b04140a3f483d1d611925f : SUCCESS in 7m 18s (non-voting)\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/3e0f5b719b2e410795ec5a2a547e187e : SUCCESS in 1h 02m 32s\n- octavia-v2-dsvm-scenario-traffic-ops https://zuul.opendev.org/t/openstack/build/5252b80ce367470aaa918bccff2e1e0b : SUCCESS in 1h 55m 21s\n- octavia-v2-dsvm-scenario-non-traffic-ops https://zuul.opendev.org/t/openstack/build/c20c3a9d7e3d4389b2886ca1406ff1d4 : SUCCESS in 1h 07m 07s\n- octavia-v2-dsvm-scenario-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/4fcd4e9bd48d4c2d9b50377714194a5c : SUCCESS in 2h 01m 52s\n- octavia-v2-dsvm-scenario-non-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/5e9dcdc54d0147c7bc32ac6200a54bba : SUCCESS in 1h 09m 46s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/3e43151ece1c48f48ac810fe2dd12d65 : SUCCESS in 41m 16s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/dd1bbd9d9bbc4dd8ac48a3622f1abc1b : SUCCESS in 46m 09s\n- octavia-grenade-skip-level https://zuul.opendev.org/t/openstack/build/855b8ffadc9c4cf384729f07fa18e54d : SUCCESS in 43m 50s (non-voting)\n- octavia-v2-dsvm-tls-barbican-fips https://zuul.opendev.org/t/openstack/build/68eb4d31fee74596a8a4aae6f52dd703 : SUCCESS in 1h 15m 16s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/b005a88e34124a9e89b8e591202930c5 : SUCCESS in 43m 50s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/27ae3865e7fe457cb1e8ca76a30af349 : SUCCESS in 2h 32m 12s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/56029b33dc9e469c8d81ba51b218fd29 : SUCCESS in 2h 23m 51s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/6968819bec4e45e68eef593b3e350314 : SUCCESS in 2h 27m 41s (non-voting)\n- octavia-v2-dsvm-scenario-centos-9-stream https://zuul.opendev.org/t/openstack/build/e3097286fc894725a50c0c364c1af184 : TIMED_OUT in 2h 33m 32s (non-voting)","accounts_in_message":[],"_revision_number":9},{"id":"653d27dd27c7517c54f2fb36353b347cb9c1bc1c","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2023-11-03 18:13:00.000000000","message":"Patch Set 9: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":9},{"id":"5e7524aff10e2ad86a74644dfd370a0b1d4e3ba2","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2023-11-06 09:52:58.000000000","message":"Patch Set 9: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":9},{"id":"bd8e8506f0ff8cce02b17f572f8f40b19bcd0a54","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-06 09:53:34.000000000","message":"Patch Set 9: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":9},{"id":"eab93ccf1853cd832a8009b790c17b0e947ab8d2","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-06 10:50:07.000000000","message":"Patch Set 9: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/ad72acd3309e40ed8ed00654c91b3030\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/4427814811964df6af277e1c37bb5df8 : SUCCESS in 1m 48s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ade97b84eed8447ca7e1461270b754ee : SUCCESS in 4m 33s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/e7504c7c41544e7086056336e1ef39c2 : SUCCESS in 5m 40s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/50ab7c40100a4405a986073b8fb89468 : SUCCESS in 6m 06s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e20aac3fe6df47c4bbcc418ba7e854d7 : SUCCESS in 12m 18s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/814bf938dbbe41ec8bec983cfcf5c986 : SUCCESS in 5m 53s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/98c7e70134d4478fbe7fe604b674ac2d : SUCCESS in 4m 50s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/f9cd5bcb2c1749de8a9b8515210220eb : SUCCESS in 8m 26s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/d76d7cc6b79c41d28e85e53d911a3d8c : CANCELED\n- octavia-v2-dsvm-scenario-traffic-ops https://zuul.opendev.org/t/openstack/build/c903fd7be4cc4bab82080cab763d6354 : CANCELED\n- octavia-v2-dsvm-scenario-non-traffic-ops https://zuul.opendev.org/t/openstack/build/be2ada02e82241c899b9a334b933eebc : CANCELED\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/ddf5bb1d922f42e89ba8dc2f160f99fc : SUCCESS in 33m 58s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/259b6abb566645a79ae102df341129c7 : FAILURE in 55m 16s","accounts_in_message":[],"_revision_number":9},{"id":"c7b66df25eec693e9e0424bf9ffe4930e447e672","author":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"date":"2023-11-06 11:36:10.000000000","message":"Patch Set 9:\n\n(1 comment)","accounts_in_message":[],"_revision_number":9},{"id":"7a29a7832186aa256b78ed3c74135c1dc7d06c9d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-06 14:16:13.000000000","message":"Patch Set 9: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/cdf8fa0d2c564f2d97b7afc44d86e684\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/5accdc3fd57b4f35a0616f8f6a879605 : SUCCESS in 2m 06s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/7f73de115ae148bcaff87634d3816641 : SUCCESS in 10m 31s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5e52274e79ec40d98ca1f1e7a3e62636 : SUCCESS in 4m 57s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/114bccb744f242e1bc87453d322e00fd : SUCCESS in 4m 35s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/b7db648feb68494a90f9e5d8a663940a : SUCCESS in 4m 42s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6ec6d8cda92042348a926a652dddc8c4 : SUCCESS in 13m 23s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/8545c0f306d44dddb0ecf6c96dfd57dc : SUCCESS in 4m 35s\n- octavia-tox-py39-tips https://zuul.opendev.org/t/openstack/build/eb282c2ecd1b470388c3158258205cb0 : SUCCESS in 6m 06s\n- octavia-tox-functional-py39-tips https://zuul.opendev.org/t/openstack/build/62a6997eebf748e4ad0e1f5311794126 : SUCCESS in 8m 06s\n- octavia-tox-functional-py39-sqlalchemy-tips https://zuul.opendev.org/t/openstack/build/d870f4fbff0d4b3089e4bf2b9722be5f : SUCCESS in 9m 33s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/cf2a89f2bd3e44a3be56c39bf7ed4c11 : SUCCESS in 4m 40s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/868fd3d9eee54e6ebdd81d878310586d : SUCCESS in 6m 50s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/21d1303eeb0d44e0902c4c6afe7ba8be : SUCCESS in 8m 25s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/09debe276f6d45b9940ae72dc30bb4de : SUCCESS in 7m 13s (non-voting)\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/98efb4b4ae2843319e97ee739a191105 : SUCCESS in 1h 09m 31s\n- octavia-v2-dsvm-scenario-traffic-ops https://zuul.opendev.org/t/openstack/build/fdb43093b6c04dd08ca7e4d6c7a5a5ab : SUCCESS in 2h 01m 41s\n- octavia-v2-dsvm-scenario-non-traffic-ops https://zuul.opendev.org/t/openstack/build/81fd6f17db2243ca86872d5c6c975062 : SUCCESS in 1h 02m 30s\n- octavia-v2-dsvm-scenario-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/00d60f1323de4ac2a768d21a090044d3 : SUCCESS in 1h 46m 32s\n- octavia-v2-dsvm-scenario-non-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/46f21ecbf5a243838458694d90fe368f : SUCCESS in 1h 21m 01s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/81f064053a924d78b84732839b4e6687 : SUCCESS in 41m 54s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/b56780ad2773470699d570a4fedc03d6 : SUCCESS in 42m 21s\n- octavia-grenade-skip-level https://zuul.opendev.org/t/openstack/build/44554d27e14a4877be598f2853367c11 : SUCCESS in 36m 43s (non-voting)\n- octavia-v2-dsvm-tls-barbican-fips https://zuul.opendev.org/t/openstack/build/6309c49304ec450d957ebc920ef825c9 : SUCCESS in 1h 06m 17s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/c2dc5e4619be4026bff61329ff85e66e : SUCCESS in 31m 39s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/a7bc95e64e1e472aabd003576e170800 : SUCCESS in 2h 15m 57s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/b7d310e2b8cc460a907f31fff87a5dff : SUCCESS in 2h 36m 58s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/eb4cde302fce4e5787c7e5f9f3864bc2 : SUCCESS in 2h 14m 56s (non-voting)\n- octavia-v2-dsvm-scenario-centos-9-stream https://zuul.opendev.org/t/openstack/build/16ca99808fd742b2acb2e8197b8e1fc7 : FAILURE in 2h 32m 46s (non-voting)","accounts_in_message":[],"_revision_number":9},{"id":"d4b4ed76f3beb70029de4ccfeb4f38186de4a1b8","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-06 14:16:48.000000000","message":"Patch Set 9: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":9},{"id":"0a6db8070b63126ed773b2897c3c5745a2bab154","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-06 16:05:38.000000000","message":"Patch Set 9: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/44324bb66e8c4501af1d85e789d5f2af\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/6c0fa2e6733744b78c9f047f1636939c : SUCCESS in 2m 08s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/699174d757204f149b1a80a719e67fe8 : SUCCESS in 4m 21s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/6dfe9fa19f424448bdedb94ad1c710c6 : SUCCESS in 4m 31s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/2b69e4db044e404d9012c9bd05e1d824 : SUCCESS in 4m 44s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b8bacc9a102140af98b739574ca09e2f : SUCCESS in 11m 06s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/50d97ead67794f82897bfe9cb526bcf9 : SUCCESS in 4m 52s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/e29f029e8cad4492bb8cf9ee1a0b791c : SUCCESS in 4m 01s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/de0ac3e8b94b45aebdbf06cda7ead76e : SUCCESS in 6m 38s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/d8047d248dc2479a987e6ee697b750de : SUCCESS in 1h 09m 18s\n- octavia-v2-dsvm-scenario-traffic-ops https://zuul.opendev.org/t/openstack/build/3cd41ad4dbaf41b1b5611728cccec42d : SUCCESS in 1h 48m 08s\n- octavia-v2-dsvm-scenario-non-traffic-ops https://zuul.opendev.org/t/openstack/build/161e51a9e6e04ca3b729d9b18b9a87e8 : SUCCESS in 1h 09m 57s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/dd64ecc2e19b4dc3a596a856ed4aeb92 : SUCCESS in 45m 48s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/550c1d80904e4c2091f784bb6c5bde6e : SUCCESS in 49m 46s","accounts_in_message":[],"_revision_number":9},{"id":"bd1328487da169e5dc74248a8eda23ec2c6b826e","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-06 16:05:38.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":9},{"id":"d847fff9c70ede9d2221374692d5c185fe90c34b","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-06 16:06:52.000000000","message":"Patch Set 9:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/fb13856344684579b32116307fd8f34d\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1aec5dfc4ad04b5dbc33ccc9622cc647 : SUCCESS in 56s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/7556e84133fe40b0a8a2fc765c7cb6d0 : SUCCESS in 43s","accounts_in_message":[],"_revision_number":9}],"current_revision_number":9,"current_revision":"7310986de9bf68ed86de90de0501a1bc46945526","revisions":{"88b256ff484f7ba5dcfdef7da8d9d9cdef7be37e":{"kind":"REWORK","_number":1,"created":"2023-09-07 14:10:59.000000000","uploader":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"ref":"refs/changes/96/894196/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/96/894196/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/96/894196/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/96/894196/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/96/894196/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/96/894196/1"}}},"commit":{"parents":[{"commit":"0426285a41464a5015494584f109eed35a0d44db","subject":"Merge \"Add octavia-grenade-slurp CI job\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/0426285a41464a5015494584f109eed35a0d44db"}]}],"author":{"name":"Omer","email":"oschwart@redhat.com","date":"2023-09-07 13:42:36.000000000","tz":120},"committer":{"name":"Omer","email":"oschwart@redhat.com","date":"2023-09-07 14:10:46.000000000","tz":120},"subject":"Add Noop Certificate Manager","message":"Add Noop Certificate Manager\n\nSo far, Octavia noop drivers were using real certificate managers, which\nhave validated the cerifitates for every certificate required operation,\nsometimes without any need.\n\nOctavia should have a Noop Certificate Manager for faster testing\npurposes.\n\nThis patch adds it.\n\nCloses-Bug: #2034711\n\nChange-Id: I700c65fb17bad28b2b922e03d9c94c4716de9cbe\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/88b256ff484f7ba5dcfdef7da8d9d9cdef7be37e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/88b256ff484f7ba5dcfdef7da8d9d9cdef7be37e"}]},"branch":"refs/heads/master"},"58822d2ab6725af37130afffc72e1d8b8f38b641":{"kind":"REWORK","_number":2,"created":"2023-09-11 13:03:09.000000000","uploader":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"ref":"refs/changes/96/894196/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/96/894196/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/96/894196/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/96/894196/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/96/894196/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/96/894196/2"}}},"commit":{"parents":[{"commit":"0426285a41464a5015494584f109eed35a0d44db","subject":"Merge \"Add octavia-grenade-slurp CI job\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/0426285a41464a5015494584f109eed35a0d44db"}]}],"author":{"name":"Omer","email":"oschwart@redhat.com","date":"2023-09-07 13:42:36.000000000","tz":120},"committer":{"name":"Omer","email":"oschwart@redhat.com","date":"2023-09-11 12:56:41.000000000","tz":120},"subject":"Add Noop Certificate Manager","message":"Add Noop Certificate Manager\n\nSo far, Octavia noop drivers were using real certificate managers, which\nhave validated the cerifitates for every certificate required operation,\nsometimes without any need.\n\nOctavia should have a Noop Certificate Manager for faster testing\npurposes.\n\nThis patch adds it.\n\nCloses-Bug: #2034711\n\nChange-Id: I700c65fb17bad28b2b922e03d9c94c4716de9cbe\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/58822d2ab6725af37130afffc72e1d8b8f38b641"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/58822d2ab6725af37130afffc72e1d8b8f38b641"}]},"branch":"refs/heads/master"},"77edbef1a0a8432e7028564ad9dc4b958003b0a3":{"kind":"REWORK","_number":3,"created":"2023-09-25 13:19:15.000000000","uploader":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"ref":"refs/changes/96/894196/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/96/894196/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/96/894196/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/96/894196/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/96/894196/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/96/894196/3"}}},"commit":{"parents":[{"commit":"0426285a41464a5015494584f109eed35a0d44db","subject":"Merge \"Add octavia-grenade-slurp CI job\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/0426285a41464a5015494584f109eed35a0d44db"}]}],"author":{"name":"Omer","email":"oschwart@redhat.com","date":"2023-09-07 13:42:36.000000000","tz":120},"committer":{"name":"Omer Schwartz","email":"oschwart@redhat.com","date":"2023-09-25 13:19:15.000000000","tz":0},"subject":"Add Noop Certificate Manager","message":"Add Noop Certificate Manager\n\nSo far, Octavia noop drivers were using real certificate managers, which\nhave validated the cerifitates for every certificate required operation,\nsometimes without any need.\n\nOctavia should have a Noop Certificate Manager for faster testing\npurposes.\n\nThis patch adds it.\n\nCloses-Bug: #2034711\n\nChange-Id: I700c65fb17bad28b2b922e03d9c94c4716de9cbe\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/77edbef1a0a8432e7028564ad9dc4b958003b0a3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/77edbef1a0a8432e7028564ad9dc4b958003b0a3"}]},"branch":"refs/heads/master"},"15c0e4773bdc3ce9761594a77bdde03a793b283f":{"kind":"REWORK","_number":4,"created":"2023-10-04 16:21:43.000000000","uploader":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"ref":"refs/changes/96/894196/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/96/894196/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/96/894196/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/96/894196/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/96/894196/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/96/894196/4"}}},"commit":{"parents":[{"commit":"0426285a41464a5015494584f109eed35a0d44db","subject":"Merge \"Add octavia-grenade-slurp CI job\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/0426285a41464a5015494584f109eed35a0d44db"}]}],"author":{"name":"Omer","email":"oschwart@redhat.com","date":"2023-09-07 13:42:36.000000000","tz":120},"committer":{"name":"Omer Schwartz","email":"oschwart@redhat.com","date":"2023-10-04 16:21:43.000000000","tz":0},"subject":"Add Noop Certificate Manager","message":"Add Noop Certificate Manager\n\nSo far, Octavia noop drivers were using real certificate managers, which\nhave validated the cerifitates for every certificate required operation,\nsometimes without any need.\n\nOctavia should have a Noop Certificate Manager for faster testing\npurposes.\n\nThis patch adds it.\n\nCloses-Bug: #2034711\n\nChange-Id: I700c65fb17bad28b2b922e03d9c94c4716de9cbe\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/15c0e4773bdc3ce9761594a77bdde03a793b283f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/15c0e4773bdc3ce9761594a77bdde03a793b283f"}]},"branch":"refs/heads/master"},"02d4fdc9fdcde5d8aaaedcef7bd5ec0d4bec9a79":{"kind":"REWORK","_number":5,"created":"2023-10-18 14:27:31.000000000","uploader":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"ref":"refs/changes/96/894196/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/96/894196/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/96/894196/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/96/894196/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/96/894196/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/96/894196/5"}}},"commit":{"parents":[{"commit":"c91ffe0e497a7333e283c4b6e90282c23053e2c9","subject":"Merge \"Fix Amphora Configure API call\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/c91ffe0e497a7333e283c4b6e90282c23053e2c9"}]}],"author":{"name":"Omer","email":"oschwart@redhat.com","date":"2023-09-07 13:42:36.000000000","tz":120},"committer":{"name":"Omer","email":"oschwart@redhat.com","date":"2023-10-18 14:02:08.000000000","tz":120},"subject":"Add Noop Certificate Manager","message":"Add Noop Certificate Manager\n\nSo far, Octavia noop drivers were using real certificate managers, which\nhave validated the cerifitates for every certificate required operation,\nsometimes without any need.\n\nOctavia should have a Noop Certificate Manager for faster testing\npurposes.\n\nThis patch adds it.\n\nCloses-Bug: #2034711\n\nChange-Id: I700c65fb17bad28b2b922e03d9c94c4716de9cbe\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/02d4fdc9fdcde5d8aaaedcef7bd5ec0d4bec9a79"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/02d4fdc9fdcde5d8aaaedcef7bd5ec0d4bec9a79"}]},"branch":"refs/heads/master"},"5da58b82cc555fab9f2350cf98bc858abee16053":{"kind":"REWORK","_number":6,"created":"2023-10-19 08:54:00.000000000","uploader":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"ref":"refs/changes/96/894196/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/96/894196/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/96/894196/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/96/894196/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/96/894196/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/96/894196/6"}}},"commit":{"parents":[{"commit":"c91ffe0e497a7333e283c4b6e90282c23053e2c9","subject":"Merge \"Fix Amphora Configure API call\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/c91ffe0e497a7333e283c4b6e90282c23053e2c9"}]}],"author":{"name":"Omer","email":"oschwart@redhat.com","date":"2023-09-07 13:42:36.000000000","tz":120},"committer":{"name":"Omer","email":"oschwart@redhat.com","date":"2023-10-19 08:53:53.000000000","tz":120},"subject":"Add Noop Certificate Manager","message":"Add Noop Certificate Manager\n\nSo far, Octavia noop drivers were using real certificate managers, which\nhave validated the cerifitates for every certificate required operation,\nsometimes without any need.\n\nOctavia should have a Noop Certificate Manager for faster testing\npurposes.\n\nThis patch adds it.\n\nCloses-Bug: #2034711\n\nChange-Id: I700c65fb17bad28b2b922e03d9c94c4716de9cbe\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/5da58b82cc555fab9f2350cf98bc858abee16053"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/5da58b82cc555fab9f2350cf98bc858abee16053"}]},"branch":"refs/heads/master"},"b3b78cfa1cbec214e5524825d2079d15391a6342":{"kind":"REWORK","_number":7,"created":"2023-10-20 16:33:30.000000000","uploader":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"ref":"refs/changes/96/894196/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/96/894196/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/96/894196/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/96/894196/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/96/894196/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/96/894196/7"}}},"commit":{"parents":[{"commit":"c91ffe0e497a7333e283c4b6e90282c23053e2c9","subject":"Merge \"Fix Amphora Configure API call\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/c91ffe0e497a7333e283c4b6e90282c23053e2c9"}]}],"author":{"name":"Omer","email":"oschwart@redhat.com","date":"2023-09-07 13:42:36.000000000","tz":120},"committer":{"name":"Omer","email":"oschwart@redhat.com","date":"2023-10-20 16:32:20.000000000","tz":120},"subject":"Add Noop Certificate Manager","message":"Add Noop Certificate Manager\n\nSo far, Octavia noop drivers were using real certificate managers, which\nhave validated the cerifitates for every certificate required operation,\nsometimes without any need.\n\nOctavia should have a Noop Certificate Manager for faster testing\npurposes.\n\nThis patch adds it.\n\nCloses-Bug: #2034711\n\nChange-Id: I700c65fb17bad28b2b922e03d9c94c4716de9cbe\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/b3b78cfa1cbec214e5524825d2079d15391a6342"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/b3b78cfa1cbec214e5524825d2079d15391a6342"}]},"branch":"refs/heads/master"},"ed8c7a1e7e95650b4680630e39a84505dd4ac37f":{"kind":"REWORK","_number":8,"created":"2023-10-23 09:10:15.000000000","uploader":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"ref":"refs/changes/96/894196/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/96/894196/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/96/894196/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/96/894196/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/96/894196/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/96/894196/8"}}},"commit":{"parents":[{"commit":"c91ffe0e497a7333e283c4b6e90282c23053e2c9","subject":"Merge \"Fix Amphora Configure API call\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/c91ffe0e497a7333e283c4b6e90282c23053e2c9"}]}],"author":{"name":"Omer","email":"oschwart@redhat.com","date":"2023-09-07 13:42:36.000000000","tz":120},"committer":{"name":"Omer","email":"oschwart@redhat.com","date":"2023-10-23 09:10:04.000000000","tz":120},"subject":"Add Noop Certificate Manager","message":"Add Noop Certificate Manager\n\nSo far, Octavia noop drivers were using real certificate managers, which\nhave validated the cerifitates for every certificate required operation,\nsometimes without any need.\n\nOctavia should have a Noop Certificate Manager for faster testing\npurposes.\n\nThis patch adds it.\n\nCloses-Bug: #2034711\n\nChange-Id: I700c65fb17bad28b2b922e03d9c94c4716de9cbe\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/ed8c7a1e7e95650b4680630e39a84505dd4ac37f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/ed8c7a1e7e95650b4680630e39a84505dd4ac37f"}]},"branch":"refs/heads/master"},"7310986de9bf68ed86de90de0501a1bc46945526":{"kind":"REWORK","_number":9,"created":"2023-11-02 11:07:59.000000000","uploader":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"ref":"refs/changes/96/894196/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/96/894196/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/96/894196/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/96/894196/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/96/894196/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/96/894196/9"}}},"commit":{"parents":[{"commit":"51ae11032e962c4bce957be8750d6bc6f9a9f1a5","subject":"Merge \"Run codespell on doc directory\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/51ae11032e962c4bce957be8750d6bc6f9a9f1a5"}]}],"author":{"name":"Omer","email":"oschwart@redhat.com","date":"2023-09-07 13:42:36.000000000","tz":120},"committer":{"name":"Omer","email":"oschwart@redhat.com","date":"2023-11-02 11:07:11.000000000","tz":60},"subject":"Add Noop Certificate Manager","message":"Add Noop Certificate Manager\n\nSo far, Octavia noop drivers were using real certificate managers, which\nhave validated the cerifitates for every certificate required operation,\nsometimes without any need.\n\nOctavia should have a Noop Certificate Manager for faster testing\npurposes.\n\nThis patch adds it.\n\nCloses-Bug: #2034711\n\nChange-Id: I700c65fb17bad28b2b922e03d9c94c4716de9cbe\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/7310986de9bf68ed86de90de0501a1bc46945526"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/7310986de9bf68ed86de90de0501a1bc46945526"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
