)]}'
{"id":"openstack%2Foctavia~894303","triplet_id":"openstack%2Foctavia~master~I605964757c4931de0a5999a8cf9c7c4ec936faff","project":"openstack/octavia","branch":"master","hashtags":[],"change_id":"I605964757c4931de0a5999a8cf9c7c4ec936faff","subject":"Remove invalid network driver entry point","status":"ABANDONED","created":"2023-09-08 14:00:31.000000000","updated":"2024-07-25 13:58:17.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"b52755db27a6c4ae13c8365e4473249b692120dd","_number":894303,"virtual_id_number":894303,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":1,"date":"2023-09-08 16:31:22.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-09-08 16:31:22.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"72220f04763fab8d27be555e59eb815aced28e9e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2023-09-08 14:00:31.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"d71fbf0fad98afbcfbe39bd1ac3b6ef209e06232","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-08 16:31:22.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/13fbb7f361d6417fb2696a093c302ade\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/dc526537b05d4aeb997057d676a74556 : SUCCESS in 2m 12s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/23be4310c44e4cd29edc39987471be1c : SUCCESS in 12m 09s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/65e0df4d0d684629a49530f212186b88 : SUCCESS in 5m 47s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/aca1e754a40141838b7353212fb30658 : SUCCESS in 5m 50s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/098ac4193b4441ca80ad65a3063b3e7d : SUCCESS in 4m 49s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/358ca4c7eb0b4147986506915ba1cb21 : SUCCESS in 5m 38s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/73368a51c2a64570a97aaa2f4336b477 : SUCCESS in 6m 38s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8cb4a74f09a54bf385cb80cd71657cf8 : SUCCESS in 12m 14s\n- octavia-tox-py39-tips https://zuul.opendev.org/t/openstack/build/6d9883b5c5d74c37962ad082ef8d0e2d : SUCCESS in 7m 58s\n- octavia-tox-functional-py39-tips https://zuul.opendev.org/t/openstack/build/d6748dd0c3d544a79db0b56801b2f1df : SUCCESS in 8m 44s\n- octavia-tox-functional-py39-sqlalchemy-tips https://zuul.opendev.org/t/openstack/build/c550793349dd4ebebae571e3ecdde604 : SUCCESS in 10m 15s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/be41af15737a481586df49429edaabbb : SUCCESS in 6m 13s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/8630a2869ac6462e806649e8b119c385 : SUCCESS in 8m 12s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/edc79cd441b64e6eb9bcf5fe4c8b1cf7 : SUCCESS in 8m 48s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/3a855359bc8f42ad90f19eccdb820676 : SUCCESS in 7m 25s (non-voting)\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/851c6b73cd624984bd251bb6018b30fe : SUCCESS in 1h 19m 34s\n- octavia-v2-dsvm-scenario-traffic-ops https://zuul.opendev.org/t/openstack/build/e943f7d98f814eb884facca57172f33d : SUCCESS in 1h 56m 56s\n- octavia-v2-dsvm-scenario-non-traffic-ops https://zuul.opendev.org/t/openstack/build/940940c7985f497fa816c3def93eb2c7 : SUCCESS in 1h 16m 29s\n- octavia-v2-dsvm-scenario-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/80e5731fda9547ecb53bc7c024db343f : SUCCESS in 1h 58m 48s\n- octavia-v2-dsvm-scenario-non-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/f9b451ef8d464bfda44289fbac179b0b : SUCCESS in 1h 23m 52s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/7a8f2b77d5324fbabda31c335ce407ca : SUCCESS in 42m 44s\n- octavia-v2-dsvm-tls-barbican-fips https://zuul.opendev.org/t/openstack/build/ae16ff595e104af3864e99bd223ebcaf : SUCCESS in 1h 19m 56s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/afdc035750714a5a8cfe62f7455186db : SUCCESS in 44m 44s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/6adcb202bef048bf9bc85cf913691fc1 : SUCCESS in 2h 18m 39s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/5b5f61385b6448a8a6f467d591e255b9 : SUCCESS in 2h 28m 44s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/b6b3bf9cf64c4d80a603f85a2702fb0e : FAILURE in 2h 16m 29s (non-voting)\n- octavia-v2-dsvm-scenario-centos-9-stream https://zuul.opendev.org/t/openstack/build/d2c29e2dcbc744e2975f194bb0d56acc : SUCCESS in 2h 23m 57s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"b52755db27a6c4ae13c8365e4473249b692120dd","tag":"autogenerated:gerrit:abandon","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-07-25 13:58:17.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"2717f3464e7612999d9dc86565472c72daa07c59","revisions":{"2717f3464e7612999d9dc86565472c72daa07c59":{"kind":"REWORK","_number":1,"created":"2023-09-08 14:00:31.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/03/894303/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/03/894303/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/03/894303/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/03/894303/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/03/894303/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/03/894303/1"}}},"commit":{"parents":[{"commit":"e2bc07222f40ce5c6ba73a0e4bf79af9f5ebde11","subject":"Merge \"Fix hm operating status to ONLINE in single lb call\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/e2bc07222f40ce5c6ba73a0e4bf79af9f5ebde11"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2023-09-08 13:57:35.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2023-09-08 13:57:35.000000000","tz":540},"subject":"Remove invalid network driver entry point","message":"Remove invalid network driver entry point\n\nThe entry point of the container network driver was added by [1], but\nit has never been implemented.\n\n[1] fc427679bd9033dd5a97c785fc61b9d6ed4c7335\n\nChange-Id: I605964757c4931de0a5999a8cf9c7c4ec936faff\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/2717f3464e7612999d9dc86565472c72daa07c59"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/2717f3464e7612999d9dc86565472c72daa07c59"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
