)]}'
{"id":"openstack%2Foctavia~894547","triplet_id":"openstack%2Foctavia~stable%2Fvictoria~Id77ce36f59b71d309f153e5c1d44059f162ee440","project":"openstack/octavia","branch":"stable/victoria","topic":"story/2007619","attention_set":{},"removed_from_attention_set":{"32371":{"account":{"_account_id":32371,"name":"David Negreira","email":"david.negreira@canonical.com","username":"dnegreira"},"last_update":"2023-10-09 13:32:46.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Id77ce36f59b71d309f153e5c1d44059f162ee440","subject":"Fix barbican client with application credentials/trusts","status":"MERGED","created":"2023-09-11 15:18:07.000000000","updated":"2023-10-09 13:33:54.000000000","submitted":"2023-10-09 13:32:46.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":1,"has_review_started":true,"submission_id":"894547-story/2007619","meta_rev_id":"76b40342e36660e8ba0a995f438200d9e10161a4","_number":894547,"virtual_id_number":894547,"owner":{"_account_id":32371,"name":"David Negreira","email":"david.negreira@canonical.com","username":"dnegreira"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-10-09 13:32:46.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"all":[{"value":2,"date":"2023-10-09 09:30:21.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2023-10-09 12:04:26.000000000","_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"all":[{"value":0,"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2023-10-09 12:04:26.000000000","_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-09-11 15:18:07.000000000","updated_by":{"_account_id":32371,"name":"David Negreira","email":"david.negreira@canonical.com","username":"dnegreira"},"reviewer":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"state":"CC"},{"updated":"2023-09-11 17:42:38.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-10-09 09:30:21.000000000","updated_by":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"reviewer":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"state":"REVIEWER"},{"updated":"2023-10-09 12:04:26.000000000","updated_by":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"reviewer":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"state":"REVIEWER"}],"messages":[{"id":"2d7612612cd8f939a4bc1e7afe10cd40a25a1061","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":32371,"name":"David Negreira","email":"david.negreira@canonical.com","username":"dnegreira"},"date":"2023-09-11 15:18:07.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"d6a3500e8ffe5d7f7b5d27b380171d7680eeca6a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-11 17:42:38.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/f0b1cf876309455d86d6dbb19d6aff12\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/015cda02255f4db0989eea71e0b36f4a : FAILURE in 3m 50s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9483c00d973f4d3abe0f9b2c3b46c7d3 : FAILURE in 3m 43s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/a73037897d10468e85d907b124dec253 : SUCCESS in 5m 42s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/5957f38ff81743e3a09a5e76190479d7 : FAILURE in 3m 42s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3984b4c7530c424b9ee630446faaf9a2 : FAILURE in 5m 39s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/cd1b3452d8c0489ba5859782acef8bb7 : SUCCESS in 5m 40s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/0b4a6b6222d24605aa6a4b325bc467d3 : SUCCESS in 6m 31s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/d9375a8af1ef49b9884d9615a5cf69a2 : SUCCESS in 8m 49s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/d7b6e5a5f3d641f5bf6b9f351a28a74b : SUCCESS in 5m 12s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/7d1820fddbf34837ac85728ddc0b30e8 : SUCCESS in 8m 02s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/a30340ce7c40432c854018b97075267f : SUCCESS in 1h 13m 15s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/2bf91324dd264c498cc7b42ec9c057f5 : SUCCESS in 2h 11m 38s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/3c411979f89f40339cb3b8a482386e36 : SUCCESS in 37m 42s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/fe5430fb1ad34f549d4aa755546f2670 : SUCCESS in 41m 36s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/d60ff0a20d9944bba2d84f1ad8f37a68 : SUCCESS in 33m 16s\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/ffc7c0c99637474e836a27a574251836 : SUCCESS in 2h 10m 34s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/5db9781c51f348fb9df5c8f0f152b6c3 : FAILURE in 10m 23s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/c8461b43d05d417ea93bce3404248ac0 : FAILURE in 9m 25s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/b6d832575d614c008482f87c06619250 : FAILURE in 2h 17m 54s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/cd542315fd7e48fa8fa882d040149d1d : FAILURE in 1h 28m 09s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8 https://zuul.opendev.org/t/openstack/build/b1b28f6f8ae44df380bbd7db76e53365 : NODE_FAILURE Node request 300-0022199272 failed in 0s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"ce6a4fefe0d452dae9668a6713ba1100428da445","author":{"_account_id":32371,"name":"David Negreira","email":"david.negreira@canonical.com","username":"dnegreira"},"date":"2023-09-26 08:29:09.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"182f16926f62f233f07656b1e5d68374718252b6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-26 10:34:22.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/f2bce744b17e424ab856efa78c7b048f\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/820cf5baef9c4be79fb0d0bf91ca4f99 : FAILURE in 3m 10s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2412f14e50704d78872ff0dff7c3bbcc : FAILURE in 2m 55s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/e64de13514494fff88f01d1931dc7eb9 : SUCCESS in 4m 10s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/2aaa3804d4234b1599fe165d1dd43f42 : FAILURE in 3m 16s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/28463adbe17e428092098a0a3793c9a8 : FAILURE in 4m 58s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/72ab31065e174f418d11a660ec3f143e : SUCCESS in 4m 45s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/3ac0bf4fdb7b4bbe9154962267ce39ce : SUCCESS in 4m 45s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/1c84c6d6e4934d90ae421851eb3387fe : SUCCESS in 6m 25s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/4c91899e0bba463a86811b9afa171394 : SUCCESS in 4m 45s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/936b764017c2426998040a527573e5ef : SUCCESS in 8m 02s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/4f73d73ab3314be18c6d63a4d7b0d98c : SUCCESS in 59m 46s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/c97fcebd76c94515b1448fce4bd6d6ae : SUCCESS in 1h 43m 11s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/7715d042115e457f873f3f5b00c7848e : SUCCESS in 35m 50s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/22cd987b31134d8195fb26764d4fc2b0 : SUCCESS in 32m 44s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/9203e1b33cc34338a8c7d38a42b559b4 : SUCCESS in 33m 05s\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/660d78bf7b7b4a91bb592ba4bdc033f0 : SUCCESS in 1h 44m 15s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/01b18672c2554eb186de9daed657dc81 : FAILURE in 10m 12s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/53bcacb333004ce48418747dee3af769 : FAILURE in 5m 15s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/a7b6b028d13f4a05b96eb9e7589b3088 : SUCCESS in 2h 00m 20s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/fb27d11d97db429d92b2172e415d09cb : SUCCESS in 1h 59m 17s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8 https://zuul.opendev.org/t/openstack/build/c59c99ac9b7549ae8ff55dec45c170f4 : NODE_FAILURE Node request 300-0022320938 failed in 0s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"a9278d3e3ab5584df224ff338d32d411b31e0711","author":{"_account_id":32371,"name":"David Negreira","email":"david.negreira@canonical.com","username":"dnegreira"},"date":"2023-10-09 09:28:56.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"daac9c68a3c888a2f976d6976499839ef1af0908","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2023-10-09 09:30:21.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"8d6d2d40519453c2544d6d9de3a3565b9c6e20c1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-09 11:49:49.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/76b86a795e7143acbae4de8c972602f0\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/ecd7ce9ba85c4810a9d0d3a57035fedc : SUCCESS in 8m 19s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/cf794a43046341ac94c84ea3a94d7e9c : SUCCESS in 6m 08s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/7a3980c54f3d4508a706c51c695c608b : SUCCESS in 6m 02s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/c2e37e5e5a8b45d099214b8e1502f4b5 : SUCCESS in 4m 54s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/72a9486590164707b5567fa159dd0074 : SUCCESS in 13m 23s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/9d9a1253b0bb4294909bcc521f02b906 : SUCCESS in 5m 06s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/7e5a06e26d5140a7a86decde4e323e87 : SUCCESS in 5m 19s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/8a1ce47356b2423abefb97f7030c849e : SUCCESS in 10m 05s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/9fb08d406f1d4b1bbb02ed674ebaabe2 : SUCCESS in 4m 59s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/f0e08f458917499183b8d240d9a213be : SUCCESS in 7m 38s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/01b4f509d0784d4db38cc6f973ca307a : SUCCESS in 1h 15m 40s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/12f234d1773a4a0ea0c7282f75456b49 : SUCCESS in 1h 40m 33s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/eed12d44dd9d442aa8246731f54bceaf : SUCCESS in 34m 20s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/f06cab226809479ca0a9b64da726ddc0 : SUCCESS in 31m 58s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/144ed2fb7c1942e294587c14c860aae0 : SUCCESS in 39m 21s\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/bcaf341975794e0da978a731bd3714e1 : SUCCESS in 2h 10m 55s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/0e23ea7b50b24660b7a92315fc694f7e : FAILURE in 10m 13s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/f4fecbd755eb4e89910d7bda64d531f6 : FAILURE in 8m 01s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/cd9fdb26a53447b79785e9cd97d9dd61 : FAILURE in 2h 13m 56s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/28a7cbe490eb4996a819ca8a507dce1b : FAILURE in 1h 56m 00s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8 https://zuul.opendev.org/t/openstack/build/15e9a612e1de4d48bf68da4a58c82de0 : NODE_FAILURE Node request 300-0022422148 failed in 0s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"bb3ed81887e68b3430b28ad04d791a3d84993e0e","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2023-10-09 12:04:26.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"39adc8ee18262c9937ecc4859fcd02faf6e4ef37","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-09 12:05:02.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"e42cd0abc70912bf75cfc24f93325e7a2e3eaa29","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-09 13:32:46.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c01858c29dc04909835f7cb502ee3e61\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/62c91b4b145741c4890b17e00bf94365 : SUCCESS in 4m 28s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/75f384d8670d402e906ceed0ebb6d42c : SUCCESS in 4m 51s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/b8c8176f9995407aa713ec2c5ca02b68 : SUCCESS in 4m 49s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4c6b86d0f4ea4e7d9d83336d809f8c48 : SUCCESS in 14m 58s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/0db70e4d84fa44e0aa2b3f2d3887a100 : SUCCESS in 5m 08s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/a516e444448745f9bfe146c0a7122fea : SUCCESS in 4m 14s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/9b35c4ecf10943508aec66404126aec2 : SUCCESS in 8m 46s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/91d8037280c24fb3bd4452ef804eefde : SUCCESS in 1h 20m 33s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/e9f327842e4d4f35812a493e67d669ae : SUCCESS in 1h 27m 01s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/0b620be81aba4f58818b6321b2c14001 : SUCCESS in 35m 35s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/c0cbef3876714989aa5c94bbe32d04b1 : SUCCESS in 32m 48s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/62bceccae2ca438abef0a1ae612fd6ff : SUCCESS in 33m 03s","accounts_in_message":[],"_revision_number":1},{"id":"e5439f3f58592c189f19b2122cffae14fdca8c05","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-09 13:32:46.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"76b40342e36660e8ba0a995f438200d9e10161a4","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-09 13:33:54.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/bda32f91eaf64e7dbabe5587febb55ff\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/314327530f684fd984c558dc1c68d919 : SUCCESS in 50s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/f7999c6d4098497c868c96f2111e0786 : SUCCESS in 36s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"b45a1b655a07f286e415e2d84e56c8874568b3ae","revisions":{"b45a1b655a07f286e415e2d84e56c8874568b3ae":{"kind":"REWORK","_number":1,"created":"2023-09-11 15:18:07.000000000","uploader":{"_account_id":32371,"name":"David Negreira","email":"david.negreira@canonical.com","username":"dnegreira"},"ref":"refs/changes/47/894547/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/47/894547/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/47/894547/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/47/894547/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/47/894547/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/47/894547/1"}}},"commit":{"parents":[{"commit":"fe0b3fa4a9517f8772ec8e6bd6591555c5bd3ded","subject":"Avoid interface name collisions in the amphora","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/fe0b3fa4a9517f8772ec8e6bd6591555c5bd3ded"}]}],"author":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2022-08-25 08:33:21.000000000","tz":120},"committer":{"name":"David Negreira","email":"david.negreira@canonical.com","date":"2023-09-11 15:18:06.000000000","tz":120},"subject":"Fix barbican client with application credentials/trusts","message":"Fix barbican client with application credentials/trusts\n\nIt seems that keystoneauth1.identity.generic.token doesn\u0027t handle\nproperly the application credential/trust tokens passed by the context\nof the requests. When using app credentials, Octavia failed to retrieve\nthe certificates from barbican.\nSwitching to keystoneauth1.token_endpoint fixes the issue, the\nauth tokens are correctly passed to the barbican client.\n\nStory: 2007619\nTask: 39737\n\nChange-Id: Id77ce36f59b71d309f153e5c1d44059f162ee440\n(cherry picked from commit ce7f27e3b7ef6a94501ce975fb0e9dadcffb822b)\n(cherry picked from commit 851510f4c74658b128790da1aab410e245355a90)\n(cherry picked from commit b20bb2eb23d236421cca5a621b25ada0da3297f6)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/b45a1b655a07f286e415e2d84e56c8874568b3ae"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/b45a1b655a07f286e415e2d84e56c8874568b3ae"}]},"branch":"refs/heads/stable/victoria"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
