)]}'
{"id":"openstack%2Foctavia~895798","triplet_id":"openstack%2Foctavia~stable%2F2023.2~Icf607cc9b00926b437c51c426ab5c994d79b6e45","project":"openstack/octavia","branch":"stable/2023.2","attention_set":{},"removed_from_attention_set":{"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2023-10-17 10:06:31.000000000","reason":"Change was submitted"},"11628":{"account":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"last_update":"2023-10-17 10:06:31.000000000","reason":"Change was submitted"},"29244":{"account":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"last_update":"2023-10-17 06:33:02.000000000","reason":"\u003cGERRIT_ACCOUNT_29244\u003e replied on the change","reason_account":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"}},"34429":{"account":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"last_update":"2023-10-17 08:42:51.000000000","reason":"\u003cGERRIT_ACCOUNT_34429\u003e replied on the change","reason_account":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}}},"hashtags":[],"change_id":"Icf607cc9b00926b437c51c426ab5c994d79b6e45","subject":"Fix remaining usage of [neutron] endpoint_type","status":"MERGED","created":"2023-09-20 03:57:37.000000000","updated":"2023-10-17 10:07:48.000000000","submitted":"2023-10-17 10:06:31.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":7,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"895798","cherry_pick_of_change":894017,"cherry_pick_of_patch_set":1,"meta_rev_id":"0cf67d093f9a19c6da027fd9346ebdb338dfc5de","_number":895798,"virtual_id_number":895798,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-10-17 10:06:30.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"all":[{"value":2,"date":"2023-10-17 06:33:02.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2023-10-17 08:42:51.000000000","_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"all":[{"value":0,"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2023-10-17 08:42:51.000000000","_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}],"reviewers":{"REVIEWER":[{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-09-20 03:57:57.000000000","updated_by":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"reviewer":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"state":"REVIEWER"},{"updated":"2023-09-20 03:57:57.000000000","updated_by":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"reviewer":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"state":"REVIEWER"},{"updated":"2023-09-20 03:57:57.000000000","updated_by":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"reviewer":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"state":"REVIEWER"},{"updated":"2023-09-20 04:12:40.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"a2476c47f886cd935f0ab1cba003d503fd0adfae","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2023-09-20 03:57:37.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"ba5abed459c713b3156bedbb00ae3a543e130503","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-20 04:12:40.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/fef7bb5887a94549ad16e7cd39148222\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/57d373bfc5ab45b996359007c69c5932 : SUCCESS in 10m 11s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/076ddf51a8494343ada9c1e49ad67e16 : SUCCESS in 5m 10s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/1ae0c0c4ed85457b8a760862549f744f : SUCCESS in 5m 29s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/e626c214eb9d4dfca2cd2fd3aee2bb39 : SUCCESS in 4m 59s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/54020aa33e794b57a13aebb18ba227d9 : SUCCESS in 5m 25s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/1017aeb533d84a929a0767d031cff9de : SUCCESS in 5m 53s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f81d1b56bfbf46cdab89069f29f7fa94 : SUCCESS in 11m 15s\n- octavia-tox-py39-tips https://zuul.opendev.org/t/openstack/build/1b21027be2ca4015b78f2ab4cd6bc5f5 : SUCCESS in 5m 35s\n- octavia-tox-functional-py39-tips https://zuul.opendev.org/t/openstack/build/3aebcf387afa4e498a8ccf1fdde20ab0 : SUCCESS in 9m 33s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/56973bd07fd14beab3f3a853c5167915 : SUCCESS in 4m 21s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/5c0a588c10074025bb54310312fe0713 : SUCCESS in 7m 38s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/dd2c6b0cdd944be99e730620ff273f42 : SUCCESS in 8m 03s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/46de90c5a9e84c439ff8c517af5268ac : SUCCESS in 6m 21s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"7d19bd6ec40bbc4e63bd9f25396193b77026127d","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2023-09-25 12:59:03.000000000","message":"Patch Set 1: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"2215676798fb915afb17c843dc40ce02b02fc003","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2023-09-26 10:43:38.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"e25585628faac8d1d0875444a53956cad67ce7b8","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2023-09-26 11:42:35.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"4c5b7b051944c4d75dbde824da05b68a1aeefdac","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2023-10-04 01:37:21.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"a3db984050ae65db939c2336e465a411e8922d45","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2023-10-04 01:37:58.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"5c33c4fc472073fe2f3d25acc8ff6eacbca884be","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-04 04:42:37.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/1daac8da152c4cf7b41661ef6ecec3c0\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/e26fc5e68a6e404d88d90c829789cd6a : SUCCESS in 10m 14s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/537fa849a6c64485b71a5eaee4fb6cd0 : SUCCESS in 4m 07s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/5d8c3faad137436e84751acbe2ad9d7d : SUCCESS in 5m 13s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/c599a2c274bb480288697f357b15dc0e : SUCCESS in 4m 27s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/34e8f4cd4ae94690838735169b7919fa : SUCCESS in 4m 22s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/7160d54ef2ab453d9184d0be48df1f5f : SUCCESS in 4m 50s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/582f64eabc284e38b2d409c25f983db7 : SUCCESS in 12m 18s\n- octavia-tox-py39-tips https://zuul.opendev.org/t/openstack/build/2b07276eef67403f943750da464a47d2 : SUCCESS in 4m 40s\n- octavia-tox-functional-py39-tips https://zuul.opendev.org/t/openstack/build/3d7e9cc1a865485994e0390344627554 : SUCCESS in 5m 51s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/8f289ffa78d94f0880bdf8561c35c6c5 : SUCCESS in 5m 22s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/0a7c266481bd4753808cf8c0db8133d9 : SUCCESS in 6m 43s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/67462263687b4e2b86b4da2443baa239 : SUCCESS in 6m 20s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/7f665829635544159265608b9e75069f : SUCCESS in 8m 56s (non-voting)\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/26de4c036b1346ea927dc52a4de85c3a : SUCCESS in 1h 37m 39s\n- octavia-v2-dsvm-scenario-traffic-ops https://zuul.opendev.org/t/openstack/build/55100e69f6b24c53b124257a4be58440 : SUCCESS in 2h 01m 58s\n- octavia-v2-dsvm-scenario-non-traffic-ops https://zuul.opendev.org/t/openstack/build/33902889dafd4e058aac034d56095871 : SUCCESS in 1h 28m 02s\n- octavia-v2-dsvm-scenario-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/99b1cc02c32445d9b7dc1e4848d5a551 : SUCCESS in 2h 07m 07s\n- octavia-v2-dsvm-scenario-non-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/3d767208f61d4fecb1e676b4e142a10d : SUCCESS in 1h 33m 46s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/d7b62e084eb0413c923b7568edf33ef3 : SUCCESS in 49m 48s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/39b8a0e52ee54cc59f161ecdaf8fa683 : SUCCESS in 53m 04s\n- octavia-grenade-skip-level https://zuul.opendev.org/t/openstack/build/679de54db8f24983a6d74433f54cdbf1 : FAILURE in 24m 47s\n- octavia-v2-dsvm-tls-barbican-fips https://zuul.opendev.org/t/openstack/build/3ed8931476644a529d15884efd03a867 : SUCCESS in 1h 04m 04s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/52e854f866f44e84a5c842bed26bfc17 : SUCCESS in 47m 17s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/a1985809cc1749119a0eab05029a5823 : TIMED_OUT in 2h 32m 44s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/c3a83881f3824e3dabceb1f1d037e4ec : TIMED_OUT in 3h 03m 51s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/5e27185ce19641408f60ed6e781ac09c : TIMED_OUT in 2h 32m 56s (non-voting)\n- octavia-v2-dsvm-scenario-centos-9-stream https://zuul.opendev.org/t/openstack/build/b13f58bfa98c469b8e1829ac78c21500 : SUCCESS in 1h 55m 50s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"4451e0f7db93de5d45a45bff6db7912e908c32e8","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2023-10-14 19:11:38.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"4d3867a5122d3e645f97615629ca8b315df41f51","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2023-10-14 19:12:00.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"9bdba7595b889066aab1fbea950c5f2fbacbe2ba","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-14 21:33:33.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/381bb74701644a978f3fad2be246de08\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/199bc5f816a84fbf870d87a11034cacc : SUCCESS in 9m 04s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2373f927db9f41e48337a0681e2aa2e8 : SUCCESS in 4m 50s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/72173b706192413399205c9b531b6d5d : SUCCESS in 5m 06s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/a5205005e1ae4410b9c94742d124157f : SUCCESS in 4m 32s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/61b00486ebac49138976d55eb3bba790 : SUCCESS in 4m 29s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/f4924adba49d431ca49d49a16ddfb059 : SUCCESS in 4m 37s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2cbbe45786d64809a7f4d3f91865634f : SUCCESS in 11m 09s\n- octavia-tox-py39-tips https://zuul.opendev.org/t/openstack/build/8888f4eb33044daca85fa2cc81074614 : SUCCESS in 5m 00s\n- octavia-tox-functional-py39-tips https://zuul.opendev.org/t/openstack/build/247002cfcb6d410087c358501ea6c580 : SUCCESS in 7m 47s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/4089f188b2ee4407a52088f85ef62073 : SUCCESS in 4m 51s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/2ce6968bcb47478fae676594be15a58f : SUCCESS in 8m 34s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/282a4d9aa123470c8d0a3593b12c9b50 : SUCCESS in 7m 51s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/4893c5b3e13f484e976d31e14beab92f : SUCCESS in 8m 12s (non-voting)\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/9853323d71b04fbcaf600deb6ee4c722 : SUCCESS in 1h 32m 27s\n- octavia-v2-dsvm-scenario-traffic-ops https://zuul.opendev.org/t/openstack/build/4e394af33c1f4bcfa335a79fed082503 : SUCCESS in 1h 38m 24s\n- octavia-v2-dsvm-scenario-non-traffic-ops https://zuul.opendev.org/t/openstack/build/35fd7a87b6f14b27a83fc9c76d2af3b3 : SUCCESS in 57m 08s\n- octavia-v2-dsvm-scenario-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/a3e7d64a35d14d29a013818b50472775 : SUCCESS in 1h 43m 26s\n- octavia-v2-dsvm-scenario-non-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/a1921be554c24513bf1c90dde1126207 : SUCCESS in 1h 06m 39s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/fa52a5576dc94bef814d4cf73ff8ff1c : SUCCESS in 42m 00s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/58b80260bb2e4c46a6db1cbe88ea2e71 : SUCCESS in 52m 55s\n- octavia-v2-dsvm-tls-barbican-fips https://zuul.opendev.org/t/openstack/build/d7641fd94fb744048865ef8641ace8c1 : FAILURE in 11m 28s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/067761e3ab8c45e69c7d7f7acbf7f380 : SUCCESS in 41m 41s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/6148bf1dfd9f4aa483ecb600e1809697 : SUCCESS in 1h 40m 16s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/4aa19d94b69b45dfa3faa2b8f853e906 : SUCCESS in 2h 11m 27s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/c225766952f04139aed20385e3e1eda9 : SUCCESS in 1h 45m 48s (non-voting)\n- octavia-v2-dsvm-scenario-centos-9-stream https://zuul.opendev.org/t/openstack/build/bd1f35637ac54d37a69116ee770a8164 : FAILURE in 9m 07s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"992d02a35ffe573ac8a79742190b2b7e66b71751","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2023-10-17 06:33:02.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"acc252449906c3b3cc59975670e059081d0ceec3","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2023-10-17 08:42:51.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"7c329a6d8624714a05582b46ac2aaad6650a4294","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-17 08:43:26.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"d5b99ec69de6b72ec0ab72b7eb6e18d4282619b9","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-17 10:06:30.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/915a4ae9c50a4b34a84e914278e9432b\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8ddaed91b45e485fb0e6e8c51957a2be : SUCCESS in 5m 43s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/26090f38591c48cc94419a59b831df38 : SUCCESS in 5m 21s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/8687fb3eed2241d8b71c77b0027421ce : SUCCESS in 5m 41s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/07ef4104b366440b89ca780beead86d2 : SUCCESS in 5m 14s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4875c4cf33a74a7bb4dd1d4ceb11d29b : SUCCESS in 12m 55s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/65c7d635cf77433492119ae3da419c9f : SUCCESS in 4m 06s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/7a315a242e234193986a799700acd1aa : SUCCESS in 8m 14s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/23078114247e4b9d962453411705662f : SUCCESS in 1h 18m 13s\n- octavia-v2-dsvm-scenario-traffic-ops https://zuul.opendev.org/t/openstack/build/6c8fcf41e3df496da1d032576243380e : SUCCESS in 1h 16m 57s\n- octavia-v2-dsvm-scenario-non-traffic-ops https://zuul.opendev.org/t/openstack/build/74c1c3f5444b4a25a48d673c5c9999ae : SUCCESS in 1h 05m 43s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/cbd65572fce44fb896c2b753464a9cca : SUCCESS in 56m 37s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/a5eb2b03481c4305b2a28975f23dcfbf : SUCCESS in 42m 13s","accounts_in_message":[],"_revision_number":1},{"id":"9c72d854f87a00e5c8c594d6faf2071c70f2e796","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-17 10:06:31.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"0cf67d093f9a19c6da027fd9346ebdb338dfc5de","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-17 10:07:48.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/aa2a8d28dfa146f99b9541d3ad4d85f1\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d2b5139909ba4f329bd4e4bce8f4edd4 : SUCCESS in 57s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"4f827c4a7f8c21b1e08a8ea08a31e811b1af36f9","revisions":{"4f827c4a7f8c21b1e08a8ea08a31e811b1af36f9":{"kind":"REWORK","_number":1,"created":"2023-09-20 03:57:37.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/98/895798/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/98/895798/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/98/895798/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/98/895798/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/98/895798/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/98/895798/1"}}},"commit":{"parents":[{"commit":"d7bba6c5af3c0f6d99ea8063b54ec3e895a5eea4","subject":"Move octavia grenade job on jammy","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/d7bba6c5af3c0f6d99ea8063b54ec3e895a5eea4"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2023-09-07 08:15:44.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2023-09-20 03:57:37.000000000","tz":0},"subject":"Fix remaining usage of [neutron] endpoint_type","message":"Fix remaining usage of [neutron] endpoint_type\n\nThis option was deprecated in favor of the valid_interfaces option.\n\nChange-Id: Icf607cc9b00926b437c51c426ab5c994d79b6e45\n(cherry picked from commit 8b33b521153000e10223204fef269068b79ec088)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/4f827c4a7f8c21b1e08a8ea08a31e811b1af36f9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/4f827c4a7f8c21b1e08a8ea08a31e811b1af36f9"}]},"branch":"refs/heads/stable/2023.2"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
