)]}'
{"id":"openstack%2Foctavia~896387","triplet_id":"openstack%2Foctavia~stable%2Fyoga~I453ff385620f488a3e43d4fda01634cdce3be5d4","project":"openstack/octavia","branch":"stable/yoga","attention_set":{},"removed_from_attention_set":{"29244":{"account":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"last_update":"2023-09-26 10:38:55.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I453ff385620f488a3e43d4fda01634cdce3be5d4","subject":"Fix race condition in members batch update API call","status":"MERGED","created":"2023-09-25 12:07:00.000000000","updated":"2023-09-26 10:40:12.000000000","submitted":"2023-09-26 10:38:55.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"896387","meta_rev_id":"a14fabe0227ce02a2cd668a51bdda8e6dae7696b","_number":896387,"virtual_id_number":896387,"owner":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-09-26 10:38:55.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"all":[{"value":2,"date":"2023-09-25 15:52:40.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2023-09-26 08:06:19.000000000","_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"all":[{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2023-09-26 08:06:19.000000000","_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-09-25 14:07:44.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-09-25 15:52:40.000000000","updated_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"reviewer":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"state":"REVIEWER"},{"updated":"2023-09-26 08:06:19.000000000","updated_by":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"reviewer":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"state":"REVIEWER"}],"messages":[{"id":"3ee3f259cc8afc692040633a741c6943e4f7580c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2023-09-25 12:07:00.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"ea4d73ac68b31f5ed5d9bd7f3389a89d4d8b589a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-25 14:07:44.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4f8825aad0744ea49bd6918654123092\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/e05ac57d8059436386e673b9edb54f79 : SUCCESS in 8m 43s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7f0353f48d184d19b4dbef3750be60a9 : SUCCESS in 7m 11s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/ebf68935d41c41d8bcaeba6f78813198 : SUCCESS in 5m 40s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/49d09d3704ee4962adce4156d046babd : SUCCESS in 5m 14s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6493a63a8db748ffbee44823b4fa4762 : SUCCESS in 11m 19s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/a8a5ea41ac9d472180cb172cc228ec66 : SUCCESS in 5m 37s\n- octavia-tox-py39-tips https://zuul.opendev.org/t/openstack/build/0f71aa2e58ff47688332f83e0ff9f68e : SUCCESS in 5m 36s\n- octavia-tox-functional-py39-tips https://zuul.opendev.org/t/openstack/build/0b9b18c6e38b438ea9e8f88a4103bcae : SUCCESS in 5m 29s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/54de6c1550b84f399ae15028991a4abc : SUCCESS in 5m 45s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/2d372af8e7b944bcb30ffcf76c930491 : SUCCESS in 7m 14s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/fc3c46641f734fc1b497977cd7c7baac : SUCCESS in 1h 14m 38s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/c154684e4e3c410688f24438436c9b32 : SUCCESS in 1h 54m 03s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/af4d0ef05e5441c9ae513cf8b880698d : SUCCESS in 34m 15s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/bd37350767d741a8bcf04b601b2a65e4 : SUCCESS in 51m 28s\n- octavia-v2-dsvm-tls-barbican-fips https://zuul.opendev.org/t/openstack/build/048fac490a9746e8a3b35dd64ca7cd3c : FAILURE in 1h 10m 18s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/98350991b6044ccca5a0612eb741eceb : SUCCESS in 32m 39s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/e20ff1c4809147f48eae0190c4e42a94 : SUCCESS in 1h 43m 17s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/e89814524f8b4cecaee371525f0c38f3 : FAILURE in 10m 09s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/eaf0047f909741e7a48347639a8b2f7b : FAILURE in 5m 39s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/14df7674db6a49e1bfaf93ff6e458825 : SUCCESS in 1h 58m 14s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/b98b6c30247041aea27b0cd9d9472dd0 : SUCCESS in 1h 41m 23s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8-stream https://zuul.opendev.org/t/openstack/build/60289c768726450c944656d3bf5a1bf5 : FAILURE in 44m 31s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"9b47e6d8ff7128ce00dc180b4cdb56b2c21de344","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2023-09-25 15:52:40.000000000","message":"Patch Set 1: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"3c87ca7fc3c003c1aae7aaedd8c081038866778c","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2023-09-26 08:06:19.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"d83cef55a018ecc494b36d7e409ae710239133f3","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-26 08:06:57.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"bff7e771e0b85393165b927714cd33fc0daab463","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-26 10:38:55.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d898475d0f94408ea034309c1e017a1e\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/adae766d532b450b9cfbe180c87a2940 : SUCCESS in 6m 35s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/1654e2982a8149da90433a3fd3edeb78 : SUCCESS in 5m 50s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/8c9040eaac984275832aeb4b335b623a : SUCCESS in 5m 41s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4e8e853788eb4153b90eb24401a048bd : SUCCESS in 12m 59s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/3a3f3e45090c4089bdd74356e8f8ad1d : SUCCESS in 4m 23s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/546d4ae039224f1785406a2f9728e7e6 : SUCCESS in 5m 05s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/9fd2ec7796ad48ea8b898cff02d82567 : SUCCESS in 9m 51s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/a7e9c8d9ce5c4235a2a73c990c72bc2b : SUCCESS in 1h 29m 22s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/5a67b5b1ddc54d4f929b08647c1efdb8 : SUCCESS in 2h 09m 25s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/78b99cb139bd486fbf92d6c9969f1edc : SUCCESS in 33m 13s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/4e4526d48afa46cd90f1b3a557cd5f19 : SUCCESS in 36m 16s","accounts_in_message":[],"_revision_number":1},{"id":"0407e743b5e86f16a76388b8f2b30ce980091e26","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-26 10:38:55.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"a14fabe0227ce02a2cd668a51bdda8e6dae7696b","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-26 10:40:12.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/94f56d26185f4ca2a34e6fbfe6aa7019\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9fad106bf35943249f285d4cfdc841bc : SUCCESS in 58s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/a4ce721fa37d4c8a8e82d1332c02abf9 : SUCCESS in 39s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"231398ca5fee7b825646eeed84523dbfcb414438","revisions":{"231398ca5fee7b825646eeed84523dbfcb414438":{"kind":"REWORK","_number":1,"created":"2023-09-25 12:07:00.000000000","uploader":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"ref":"refs/changes/87/896387/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/87/896387/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/87/896387/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/87/896387/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/87/896387/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/87/896387/1"}}},"commit":{"parents":[{"commit":"5bf40c896a678cc11f4ab6aecdd8cc004fc809d9","subject":"Drop lower-constraints.txt and its testing","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/5bf40c896a678cc11f4ab6aecdd8cc004fc809d9"}]}],"author":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2023-09-15 07:42:26.000000000","tz":-240},"committer":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2023-09-25 12:06:57.000000000","tz":120},"subject":"Fix race condition in members batch update API call","message":"Fix race condition in members batch update API call\n\nFix a potential race condition, the member batch update API call\ncomputes a delta between the current state of the members in the DB and\nthe members in the requests. But it fetches the members from the DB\nbefore locking the load balancer, it means that another Octavia service\nmay still update the list of members, then the computation of the\nchanges may be incorrect.\n\nThe list of members is now queried in the locked section.\n\nCloses-Bug: #2036156\nChange-Id: I453ff385620f488a3e43d4fda01634cdce3be5d4\n(cherry picked from commit 8a47577b7668e6dc82e048d937d1344c667e5124)\n(cherry picked from commit 8a5bed4cebd70c1c721a3916dc79f2b19343e445)\n(cherry picked from commit 64d9bdb97556652aa05a34276778260b541ad9c6)\n(cherry picked from commit 4c97b585ce6e519a5ff090c2cd97fab0e801be0c)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/231398ca5fee7b825646eeed84523dbfcb414438"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/231398ca5fee7b825646eeed84523dbfcb414438"}]},"branch":"refs/heads/stable/yoga"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
