)]}'
{"id":"openstack%2Foctavia~896388","triplet_id":"openstack%2Foctavia~stable%2Fxena~I453ff385620f488a3e43d4fda01634cdce3be5d4","project":"openstack/octavia","branch":"stable/xena","attention_set":{},"removed_from_attention_set":{"29244":{"account":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"last_update":"2023-09-26 10:38:57.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I453ff385620f488a3e43d4fda01634cdce3be5d4","subject":"Fix race condition in members batch update API call","status":"MERGED","created":"2023-09-25 12:07:10.000000000","updated":"2023-09-26 10:40:17.000000000","submitted":"2023-09-26 10:38:57.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"896388","meta_rev_id":"3c043bf232872c073f5a7460419cd5be2268ba3b","_number":896388,"virtual_id_number":896388,"owner":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-09-26 10:38:57.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"all":[{"value":2,"date":"2023-09-25 15:52:47.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2023-09-26 08:06:31.000000000","_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"all":[{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2023-09-26 08:06:31.000000000","_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-09-25 14:40:25.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-09-25 15:52:47.000000000","updated_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"reviewer":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"state":"REVIEWER"},{"updated":"2023-09-26 08:06:31.000000000","updated_by":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"reviewer":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"state":"REVIEWER"}],"messages":[{"id":"fd8bc5dec73e9dd9f3941e054b8ae810890ad1a7","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2023-09-25 12:07:10.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"96b5e7abe925e9f33db37551992fa45fc188c416","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-25 14:40:25.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/accbf44e1fc34e8fb4b5ce4ed6ad2c85\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/e19ab5605a074be4997ca4dd8158964a : SUCCESS in 10m 39s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c8e7b9edc3324392bce4eb5dd527d9b3 : SUCCESS in 6m 13s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/38f62038e8774110bc16865d49dfd872 : SUCCESS in 5m 11s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/70172922b3384946a2ef1b25d308be7a : SUCCESS in 4m 53s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/8b61daeb811647148c4a801aa298fc1a : SUCCESS in 5m 08s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8e7874bed741483fb49d330303c8ea41 : SUCCESS in 11m 03s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/1c4fef4a17924ad6be2614b0eaea89aa : SUCCESS in 5m 07s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/1bc8f540e0c743fc855376fda7330287 : SUCCESS in 5m 30s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/e548ccda7e25444b8e0a320931bd2663 : SUCCESS in 11m 36s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/87eeeb93ac2e4c9a9ec69cc4e3094daf : SUCCESS in 5m 05s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/ca1f247e5d294646bc3f172d4913cb2d : SUCCESS in 9m 23s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/d9ad2ffeb403426b8249b0e60248ad98 : SUCCESS in 1h 47m 40s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/7566e8ab973d4f5891e2add4962d2889 : SUCCESS in 2h 27m 35s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/1b61da7c64114a7ea5052533b1339e0c : SUCCESS in 33m 56s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/2fd4671fd69c47a6a37aa6b34cffb55f : SUCCESS in 41m 20s\n- octavia-v2-dsvm-tls-barbican-fips https://zuul.opendev.org/t/openstack/build/a4c33ef3dd7d4e45a615c91a87ba2a6d : FAILURE in 10m 25s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/cc0f642a003e4ee98ad7e8bfdbb61261 : SUCCESS in 31m 58s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/7c3792aba24045f8b543f132c0e553c8 : SUCCESS in 1h 53m 06s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/8843cd0f370449bf89ffdb55da6d62c9 : FAILURE in 10m 47s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/46a20364c7fc4302b5fc0ddbcbf398f9 : FAILURE in 8m 32s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/63dd64cc9e4e4c17adbb6f2385d9884e : SUCCESS in 2h 06m 24s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/99a87c36989c4580a0a1509102460d06 : SUCCESS in 1h 36m 25s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8 https://zuul.opendev.org/t/openstack/build/7dc9114dbfa34be89c2513318772b2cd : NODE_FAILURE Node request 300-0022313607 failed in 0s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"340f194319c06b406c9b01ac87524506f1ff44b2","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2023-09-25 15:52:47.000000000","message":"Patch Set 1: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"bfb355e69bbe8438efd57f854f89496c4b211e97","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2023-09-26 08:06:31.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"7afedcf9a8fc133edbda58d2b192b9d1a62abab1","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-26 08:07:11.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"fd93ca752deb312967a30aa10213ac28ec8320ec","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-26 10:38:57.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9845c59b14734b0c85ebab0536c4260c\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1fca7a989f0f45589132a9b4bd389f27 : SUCCESS in 6m 16s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/8bd4dde557c9410181a9917f0400a405 : SUCCESS in 5m 01s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/7c23360d81cf4549b34438c2f3d7714d : SUCCESS in 5m 43s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2d7cca243ad742a7b90b94485c9bf723 : SUCCESS in 11m 18s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/55e756dddf914a468d4b084a272a2960 : SUCCESS in 5m 24s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/1b1e5feb2ec74187ae67bb2835957649 : SUCCESS in 6m 16s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/ef20b89fc8054035a785baf62c9059a1 : SUCCESS in 8m 30s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/0b913e3b688e493b8623a25cfbbfe0ab : SUCCESS in 1h 28m 39s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/30c27e3a78804f76b90006e608af5dba : SUCCESS in 2h 17m 17s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/b38b6d369ca94fe788183dfee941157f : SUCCESS in 31m 21s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/af0dc488989b40b397e5a72fda565de6 : SUCCESS in 38m 58s","accounts_in_message":[],"_revision_number":1},{"id":"c20949b36693473a4adfab53ac9de785f8bb95d2","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-26 10:38:57.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"3c043bf232872c073f5a7460419cd5be2268ba3b","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-26 10:40:17.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5717fa27a68e487181fd39277b4082f1\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e2cb246dc6a04e288d57acf574ba29fb : SUCCESS in 59s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/2f798113caab4257afa5fb6fc9c48d28 : SUCCESS in 55s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"c6dd946b996fb6e5720892db25c2ad0a217b8bb2","revisions":{"c6dd946b996fb6e5720892db25c2ad0a217b8bb2":{"kind":"REWORK","_number":1,"created":"2023-09-25 12:07:10.000000000","uploader":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"ref":"refs/changes/88/896388/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/88/896388/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/88/896388/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/88/896388/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/88/896388/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/88/896388/1"}}},"commit":{"parents":[{"commit":"774f92214390469dc6911c62d79b3f90f0c97a3f","subject":"Drop lower-constraints.txt and its testing","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/774f92214390469dc6911c62d79b3f90f0c97a3f"}]}],"author":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2023-09-15 07:42:26.000000000","tz":-240},"committer":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2023-09-25 12:07:07.000000000","tz":120},"subject":"Fix race condition in members batch update API call","message":"Fix race condition in members batch update API call\n\nFix a potential race condition, the member batch update API call\ncomputes a delta between the current state of the members in the DB and\nthe members in the requests. But it fetches the members from the DB\nbefore locking the load balancer, it means that another Octavia service\nmay still update the list of members, then the computation of the\nchanges may be incorrect.\n\nThe list of members is now queried in the locked section.\n\nCloses-Bug: #2036156\nChange-Id: I453ff385620f488a3e43d4fda01634cdce3be5d4\n(cherry picked from commit 8a47577b7668e6dc82e048d937d1344c667e5124)\n(cherry picked from commit 8a5bed4cebd70c1c721a3916dc79f2b19343e445)\n(cherry picked from commit 64d9bdb97556652aa05a34276778260b541ad9c6)\n(cherry picked from commit 4c97b585ce6e519a5ff090c2cd97fab0e801be0c)\n(cherry picked from commit 231398ca5fee7b825646eeed84523dbfcb414438)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/c6dd946b996fb6e5720892db25c2ad0a217b8bb2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/c6dd946b996fb6e5720892db25c2ad0a217b8bb2"}]},"branch":"refs/heads/stable/xena"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
